)]}'
{"id":"openstack%2Ftripleo-docs~673719","triplet_id":"openstack%2Ftripleo-docs~master~If42152e768b590f14716e38f65d8f9ce2a94c46f","project":"openstack/tripleo-docs","branch":"master","topic":"fix-acheive-spell","hashtags":[],"change_id":"If42152e768b590f14716e38f65d8f9ce2a94c46f","subject":"Correcting typo in achieve spelling","status":"MERGED","created":"2019-07-31 03:08:13.000000000","updated":"2019-08-01 07:17:12.000000000","submitted":"2019-08-01 07:17:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673719-1564643832216-cf50056f","meta_rev_id":"dda14878d8cd7879826a9c19fabe0bce6c4cb7d2","_number":673719,"virtual_id_number":673719,"owner":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-08-01 07:17:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-01 07:07:34.000000000","_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-01 07:07:34.000000000","_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-01 07:07:34.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2019-08-01 07:17:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f6134da5d3f01de9b758c99c517fe024a7a0ae60","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2019-07-31 03:08:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"27b666d92634b297ad44f9a96e9216634f6c3b84","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 03:23:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/19/673719/1/check/openstack-tox-docs/d3bcc7f/html/ : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":1},{"id":"27e3624d035ac16c08fb52392c83f6676d67c3ff","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-08-01 07:07:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"90b7911e2cc2b83d50d7f84b1334e0fa8ba74492","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 07:07:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3e38746cb69f17f50dff4ff549005b8f6fa6404b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 07:17:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/19/673719/1/gate/openstack-tox-docs/a976c34/html/ : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":1},{"id":"d54623e6006628d57e507b4aee834c8a717512b5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 07:17:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"130aa09fdba2e8dd99ef192e70fd6bf3146bbd65","revisions":{"130aa09fdba2e8dd99ef192e70fd6bf3146bbd65":{"kind":"REWORK","_number":1,"created":"2019-07-31 03:08:13.000000000","uploader":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"ref":"refs/changes/19/673719/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-docs","ref":"refs/changes/19/673719/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/19/673719/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/19/673719/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/19/673719/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-docs refs/changes/19/673719/1"}}},"commit":{"parents":[{"commit":"a8849950e3df62df3ae1e727ac976212cc40174b","subject":"Merge \"[tripleo-ci][docs] Update Grafana link after recent migration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/a8849950e3df62df3ae1e727ac976212cc40174b"}]}],"author":{"name":"Sandeep Yadav","email":"sandyada@redhat.com","date":"2019-07-30 21:38:11.000000000","tz":330},"committer":{"name":"Sandeep Yadav","email":"sandyada@redhat.com","date":"2019-07-30 21:39:07.000000000","tz":330},"subject":"Correcting typo in achieve spelling","message":"Correcting typo in achieve spelling\n\nThere is a typographical errors in baremetal_overcloud.rst. Correcting\nspelling from acheive to achieve.\n\nChange-Id: If42152e768b590f14716e38f65d8f9ce2a94c46f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/130aa09fdba2e8dd99ef192e70fd6bf3146bbd65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/130aa09fdba2e8dd99ef192e70fd6bf3146bbd65"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
