)]}'
{"id":"openstack%2Ftripleo-docs~689530","triplet_id":"openstack%2Ftripleo-docs~master~I37fe3146d6d5d33dcfa2a09b3d0620f137c7efd5","project":"openstack/tripleo-docs","branch":"master","topic":"I37fe3146d6d5d33dcfa2a09b3d0620f137c7efd5","hashtags":[],"change_id":"I37fe3146d6d5d33dcfa2a09b3d0620f137c7efd5","subject":"tripleo-ci ruck|rover and promotions docs","status":"MERGED","created":"2019-10-18 15:49:21.000000000","updated":"2019-11-08 16:15:10.000000000","submitted":"2019-11-08 16:13:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":18,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"689530-1573229605349-e7a1f455","meta_rev_id":"69e959c259d5eb11e946fa8d8efc45d0c532bd54","_number":689530,"virtual_id_number":689530,"owner":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-11-05 06:46:38.000000000","_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-08 16:13:25.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":2,"date":"2019-11-07 16:18:11.000000000","_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":2,"date":"2019-11-05 22:10:28.000000000","_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-11-07 16:21:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":1,"date":"2019-11-08 16:02:40.000000000","_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-30 13:37:46.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2019-10-30 13:38:09.000000000","updated_by":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"reviewer":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"state":"REVIEWER"},{"updated":"2019-10-30 13:39:34.000000000","updated_by":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"reviewer":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"state":"REVIEWER"},{"updated":"2019-10-30 13:47:53.000000000","updated_by":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"reviewer":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"state":"REVIEWER"},{"updated":"2019-11-01 16:17:38.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2019-11-07 16:18:11.000000000","updated_by":{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},"reviewer":{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},"state":"REVIEWER"},{"updated":"2019-11-07 16:21:50.000000000","updated_by":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"reviewer":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"state":"REVIEWER"},{"updated":"2019-11-08 16:02:40.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"REVIEWER"},{"updated":"2019-11-08 16:13:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c884d2d0c77d0f3c65d3f88d16680dcf3a242789","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2019-10-18 15:49:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"81df95edc2d109f7c4b3dda81e3a1e84826b81ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 16:01:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae2bab12fdc848ce99e9ad81fcf75ccd : SUCCESS in 6m 12s","accounts_in_message":[],"_revision_number":1},{"id":"8232a89f07563e3a9b16b8406ca1a037ccf65bd0","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2019-10-30 13:33:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f57821f797aba5b5397a777e06cdb053b5ff56f9","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2019-10-30 13:36:09.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3aa1a2680b508c8f43799133ac1fa74d9ca91c25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 13:44:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01b917da19534fec9e54ef07ced802b5 : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":3},{"id":"43e834629916efba9e8fc6e4aad04d6cfd6f3746","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2019-10-30 13:46:42.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"490fc6bebdc6176a9b4d2c48e8ff88e335799b71","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2019-10-30 13:47:53.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"d3a4dc3d51b1e3c377530a7f1f0d6dce8f51ff1f","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2019-10-30 16:10:25.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2af94126d9ce9cee90ba4569b9e2087b5b76002d","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2019-10-31 10:47:53.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"53a601a0f503ff97c509121f2be073b61b769acd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 10:55:07.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74f7f0f6320b4193b3a88a010af96c27 : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":4},{"id":"f540e842cf8671b03bebcbc96d3ba11fae79c23f","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2019-10-31 12:32:58.000000000","message":"Patch Set 4: Code-Review+1\n\n(4 comments)\n\nThanks for writing this up. A few comments in line","accounts_in_message":[],"_revision_number":4},{"id":"a96defb2e02e67b13a05c1f3ac6e93460bc31332","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2019-10-31 15:17:17.000000000","message":"Patch Set 4: Code-Review-1\n\n(4 comments)\n\nthanks for review incoming fixes","accounts_in_message":[],"_revision_number":4},{"id":"cfebd87124c00a17841ea0822f663893bfabc076","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2019-10-31 15:17:28.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"5ae29f95c7bcdf70a8f34602c4606dfcd2f2617c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 15:32:57.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acbb0df5c98546a7ac6936adb8af08e2 : FAILURE in 6m 47s","accounts_in_message":[],"_revision_number":5},{"id":"b8d654e2679154f3f82c0423ba2cbdc66679f4e8","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2019-11-01 10:24:49.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"2fcd861ff270fbe51b9e1d96a840cd7776de9143","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2019-11-01 10:24:59.000000000","message":"Patch Set 5:\n\nincoming fix for build fail sorry forgotten spaces","accounts_in_message":[],"_revision_number":5},{"id":"8b44ce2719509cda32309eb933111cf4b1504595","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 10:31:54.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9849cef49f2148ef9b0d56bf4d1f5645 : SUCCESS in 5m 54s","accounts_in_message":[],"_revision_number":6},{"id":"45d4ffb19bb6879c4e59ca1ba8c5b6958f7d57e4","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2019-11-01 14:20:06.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"1c6c4b8816c5664bd6e2028b36d384fb260ced1c","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2019-11-01 16:17:38.000000000","message":"Patch Set 6: Code-Review-1\n\n(3 comments)\n\nnits","accounts_in_message":[],"_revision_number":6},{"id":"0c2e803a85d1119f0c0e7f0f2840cbc70eb82611","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2019-11-04 15:51:25.000000000","message":"Patch Set 6: Code-Review-1\n\n(3 comments)\n\nupdate incoming thanks for reviews","accounts_in_message":[],"_revision_number":6},{"id":"55fdf8faf005f968f77cd9c55bda3f2017b4c7c3","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2019-11-04 15:51:37.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"422513f1f7bedab8234fb2befaf71f9017964270","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-04 15:58:53.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29d4bb3c1cef410590d2d5e75ef198f2 : SUCCESS in 6m 19s","accounts_in_message":[],"_revision_number":7},{"id":"1863a1fe2e2a03dd0d825bb759687c996b34fada","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2019-11-05 06:46:38.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"c6b614e9199a3bbd2dcc6091a3d2cba1e24fbbab","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2019-11-05 22:10:28.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"3c9e773f436d8fc234dfb22729c2e923eced8dd6","author":{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},"date":"2019-11-07 16:18:11.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"6ace344c142f02b897c253ceb5c9ec2b790c6bbd","author":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"date":"2019-11-07 16:21:50.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"50845f0aea95e0bf73430692e6f5f730179e1353","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2019-11-08 16:02:40.000000000","message":"Patch Set 7: Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"3d5d135abc27c39aad46e5bd6dad4477ead5d7bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-08 16:02:53.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"774582568cbad01ea9b10c09574ea4db3318d6db","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-08 16:13:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"e43b51813e713db9753cd20da02d478cf7474c30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-08 16:13:25.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be176f4e1e504502b6ac993aab0819e2 : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":7},{"id":"4470a6c3674c8a29d5b79244792745537a7e6c4b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-08 16:15:10.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a81b06790b3948edb1de4cb8c57b24cf : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"508ffb72645feeaeba8a49c45c4654c186a3cb8f","revisions":{"bda834e0414a9ec7250cddbeb7b7627c7a20f334":{"kind":"REWORK","_number":1,"created":"2019-10-18 15:49:21.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/30/689530/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-docs","ref":"refs/changes/30/689530/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/30/689530/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/30/689530/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/30/689530/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-docs refs/changes/30/689530/1"}}},"commit":{"parents":[{"commit":"f164ec87bb4b9cf0e84c7fa11bedb9da52e39e35","subject":"Merge \"Update DCN docs.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/f164ec87bb4b9cf0e84c7fa11bedb9da52e39e35"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2019-10-18 15:48:35.000000000","tz":180},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2019-10-18 15:48:35.000000000","tz":180},"subject":"WIP: tripleo-ci ruck|rover and promotions docs","message":"WIP: tripleo-ci ruck|rover and promotions docs\n\nWIP needs more work posting what i have for first pass\n\nChange-Id: I37fe3146d6d5d33dcfa2a09b3d0620f137c7efd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/bda834e0414a9ec7250cddbeb7b7627c7a20f334"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/bda834e0414a9ec7250cddbeb7b7627c7a20f334"}]},"branch":"refs/heads/master"},"a60766a583115336db59908b1181e63956bbee70":{"kind":"REWORK","_number":2,"created":"2019-10-30 13:33:27.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/30/689530/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-docs","ref":"refs/changes/30/689530/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/30/689530/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/30/689530/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/30/689530/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-docs refs/changes/30/689530/2"}}},"commit":{"parents":[{"commit":"ec51d4f83eae652a2d17643fd4c92b4db6aaeffa","subject":"Merge \"Remove docs for deployed-server bootstrap files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/ec51d4f83eae652a2d17643fd4c92b4db6aaeffa"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2019-10-18 15:48:35.000000000","tz":180},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2019-10-30 13:33:01.000000000","tz":120},"subject":"tripleo-ci ruck|rover and promotions docs","message":"tripleo-ci ruck|rover and promotions docs\n\nSome docs about the promotion process from the ruck|rover\nperspective and note about testproject and periodics\n\nChange-Id: I37fe3146d6d5d33dcfa2a09b3d0620f137c7efd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/a60766a583115336db59908b1181e63956bbee70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/a60766a583115336db59908b1181e63956bbee70"}]},"branch":"refs/heads/master"},"43a4d6402353c8543a4bc529dc470b9a364c630e":{"kind":"REWORK","_number":3,"created":"2019-10-30 13:36:09.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/30/689530/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-docs","ref":"refs/changes/30/689530/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/30/689530/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/30/689530/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/30/689530/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-docs refs/changes/30/689530/3"}}},"commit":{"parents":[{"commit":"ec51d4f83eae652a2d17643fd4c92b4db6aaeffa","subject":"Merge \"Remove docs for deployed-server bootstrap files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/ec51d4f83eae652a2d17643fd4c92b4db6aaeffa"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2019-10-18 15:48:35.000000000","tz":180},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2019-10-30 13:35:59.000000000","tz":120},"subject":"tripleo-ci ruck|rover and promotions docs","message":"tripleo-ci ruck|rover and promotions docs\n\nSome docs about the promotion process from the ruck|rover\nperspective and note about testproject and periodics\n\nChange-Id: I37fe3146d6d5d33dcfa2a09b3d0620f137c7efd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/43a4d6402353c8543a4bc529dc470b9a364c630e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/43a4d6402353c8543a4bc529dc470b9a364c630e"}]},"branch":"refs/heads/master"},"a96dfe432cd8ded9237f93961c78becc72b8c44e":{"kind":"REWORK","_number":4,"created":"2019-10-31 10:47:53.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/30/689530/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-docs","ref":"refs/changes/30/689530/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/30/689530/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/30/689530/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/30/689530/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-docs refs/changes/30/689530/4"}}},"commit":{"parents":[{"commit":"ec51d4f83eae652a2d17643fd4c92b4db6aaeffa","subject":"Merge \"Remove docs for deployed-server bootstrap files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/ec51d4f83eae652a2d17643fd4c92b4db6aaeffa"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2019-10-18 15:48:35.000000000","tz":180},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2019-10-31 10:47:35.000000000","tz":120},"subject":"tripleo-ci ruck|rover and promotions docs","message":"tripleo-ci ruck|rover and promotions docs\n\nSome docs about the promotion process from the ruck|rover\nperspective and note about testproject and periodics\n\nChange-Id: I37fe3146d6d5d33dcfa2a09b3d0620f137c7efd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/a96dfe432cd8ded9237f93961c78becc72b8c44e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/a96dfe432cd8ded9237f93961c78becc72b8c44e"}]},"branch":"refs/heads/master"},"a632abec0640c3e86e79648349d1be6e85a8c18e":{"kind":"REWORK","_number":5,"created":"2019-10-31 15:17:28.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/30/689530/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-docs","ref":"refs/changes/30/689530/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/30/689530/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/30/689530/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/30/689530/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-docs refs/changes/30/689530/5"}}},"commit":{"parents":[{"commit":"ec51d4f83eae652a2d17643fd4c92b4db6aaeffa","subject":"Merge \"Remove docs for deployed-server bootstrap files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/ec51d4f83eae652a2d17643fd4c92b4db6aaeffa"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2019-10-18 15:48:35.000000000","tz":180},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2019-10-31 14:57:30.000000000","tz":120},"subject":"tripleo-ci ruck|rover and promotions docs","message":"tripleo-ci ruck|rover and promotions docs\n\nSome docs about the promotion process from the ruck|rover\nperspective and note about testproject and periodics\n\nChange-Id: I37fe3146d6d5d33dcfa2a09b3d0620f137c7efd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/a632abec0640c3e86e79648349d1be6e85a8c18e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/a632abec0640c3e86e79648349d1be6e85a8c18e"}]},"branch":"refs/heads/master"},"b8496166caa0a4f5e35694fbfc52ac63d8481289":{"kind":"REWORK","_number":6,"created":"2019-11-01 10:24:49.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/30/689530/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-docs","ref":"refs/changes/30/689530/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/30/689530/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/30/689530/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/30/689530/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-docs refs/changes/30/689530/6"}}},"commit":{"parents":[{"commit":"ec51d4f83eae652a2d17643fd4c92b4db6aaeffa","subject":"Merge \"Remove docs for deployed-server bootstrap files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/ec51d4f83eae652a2d17643fd4c92b4db6aaeffa"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2019-10-18 15:48:35.000000000","tz":180},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2019-11-01 10:24:39.000000000","tz":120},"subject":"tripleo-ci ruck|rover and promotions docs","message":"tripleo-ci ruck|rover and promotions docs\n\nSome docs about the promotion process from the ruck|rover\nperspective and note about testproject and periodics\n\nChange-Id: I37fe3146d6d5d33dcfa2a09b3d0620f137c7efd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/b8496166caa0a4f5e35694fbfc52ac63d8481289"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/b8496166caa0a4f5e35694fbfc52ac63d8481289"}]},"branch":"refs/heads/master"},"508ffb72645feeaeba8a49c45c4654c186a3cb8f":{"kind":"REWORK","_number":7,"created":"2019-11-04 15:51:37.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/30/689530/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-docs","ref":"refs/changes/30/689530/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/30/689530/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/30/689530/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/30/689530/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-docs refs/changes/30/689530/7"}}},"commit":{"parents":[{"commit":"ec51d4f83eae652a2d17643fd4c92b4db6aaeffa","subject":"Merge \"Remove docs for deployed-server bootstrap files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/ec51d4f83eae652a2d17643fd4c92b4db6aaeffa"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2019-10-18 15:48:35.000000000","tz":180},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2019-11-04 15:50:34.000000000","tz":120},"subject":"tripleo-ci ruck|rover and promotions docs","message":"tripleo-ci ruck|rover and promotions docs\n\nSome docs about the promotion process from the ruck|rover\nperspective and note about testproject and periodics\n\nChange-Id: I37fe3146d6d5d33dcfa2a09b3d0620f137c7efd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/508ffb72645feeaeba8a49c45c4654c186a3cb8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/508ffb72645feeaeba8a49c45c4654c186a3cb8f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
