)]}'
{"id":"openstack%2Ftripleo-docs~690449","triplet_id":"openstack%2Ftripleo-docs~master~I55cb5fc4a069dc3829eb7e0b4ae6c9bc37094740","project":"openstack/tripleo-docs","branch":"master","hashtags":[],"change_id":"I55cb5fc4a069dc3829eb7e0b4ae6c9bc37094740","subject":"Update docs to not use deployed-server-roles-data.yaml","status":"MERGED","created":"2019-10-23 00:00:08.000000000","updated":"2019-10-23 14:05:10.000000000","submitted":"2019-10-23 14:03:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"690449-1571839430215-9e252c24","meta_rev_id":"cb6438e372070ea28c05c6c37d50b3aed8cad13e","_number":690449,"virtual_id_number":690449,"owner":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-23 14:03:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11166,"name":"Lukas Bezdicka","email":"social@v3.sk","username":"xbezdick"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},{"value":2,"date":"2019-10-23 03:49:39.000000000","_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11166,"name":"Lukas Bezdicka","email":"social@v3.sk","username":"xbezdick"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-10-23 13:54:21.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2019-10-23 09:32:29.000000000","_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-10-23 09:32:34.000000000","_account_id":11166,"name":"Lukas Bezdicka","email":"social@v3.sk","username":"xbezdick"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},{"_account_id":11166,"name":"Lukas Bezdicka","email":"social@v3.sk","username":"xbezdick"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-23 03:49:39.000000000","updated_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"reviewer":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"state":"REVIEWER"},{"updated":"2019-10-23 09:32:29.000000000","updated_by":{"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},"reviewer":{"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},"state":"REVIEWER"},{"updated":"2019-10-23 09:32:34.000000000","updated_by":{"_account_id":11166,"name":"Lukas Bezdicka","email":"social@v3.sk","username":"xbezdick"},"reviewer":{"_account_id":11166,"name":"Lukas Bezdicka","email":"social@v3.sk","username":"xbezdick"},"state":"REVIEWER"},{"updated":"2019-10-23 13:54:21.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-10-23 14:03:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bd012b33ec3970dac64a7a633bbf1290e5a00529","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2019-10-23 00:00:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4707a29b64937775f44780fff8559f66f528d040","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 00:06:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76bb9130a8e443cc8a36ac493490185e : SUCCESS in 4m 58s\n- build-openstack-deploy-guide https://zuul.opendev.org/t/openstack/build/b5f7b116fe2b4c71be7cbc3f7b9aeb77 : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":1},{"id":"1451ffde940f61e8de40665771e397652826e0ed","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2019-10-23 03:49:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d15cd5d44fac83cfb9d1a07cad4eba9a634951f3","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-10-23 13:54:21.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"09d8e97c7646d807ec5bf03bd836672fa2538038","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 13:54:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"483e7dc7e14caecedc825a7d144a594c364b0751","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 14:03:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d1d2838532c815ac1b62a2f113e0a40a7ed3faaa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 14:03:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f20ae029f5d34c5984d6e1dfc2ca955d : SUCCESS in 4m 25s\n- build-openstack-deploy-guide https://zuul.opendev.org/t/openstack/build/a8081e7d7265478c95a76baad345e27a : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":1},{"id":"cb6438e372070ea28c05c6c37d50b3aed8cad13e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 14:05:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cfc443ee865414083cae2c107c0ca4f : SUCCESS in 1m 02s\n- promote-deploy-guide https://zuul.opendev.org/t/openstack/build/211ecd9af02746ae9fa31e473a95b3e6 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"daa9eaf129f94173db11923722dabacd976c70cc","revisions":{"daa9eaf129f94173db11923722dabacd976c70cc":{"kind":"REWORK","_number":1,"created":"2019-10-23 00:00:08.000000000","uploader":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"ref":"refs/changes/49/690449/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-docs","ref":"refs/changes/49/690449/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/49/690449/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/49/690449/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/49/690449/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-docs refs/changes/49/690449/1"}}},"commit":{"parents":[{"commit":"f164ec87bb4b9cf0e84c7fa11bedb9da52e39e35","subject":"Merge \"Update DCN docs.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/f164ec87bb4b9cf0e84c7fa11bedb9da52e39e35"}]}],"author":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2019-10-22 23:59:22.000000000","tz":-240},"committer":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2019-10-22 23:59:22.000000000","tz":-240},"subject":"Update docs to not use deployed-server-roles-data.yaml","message":"Update docs to not use deployed-server-roles-data.yaml\n\nUsing the custom roles file at deployed-server-roles-data.yaml is no\nlonger necessary when using deployed-server. This patch updates the\ndocumentation accordingly.\n\nChange-Id: I55cb5fc4a069dc3829eb7e0b4ae6c9bc37094740\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/daa9eaf129f94173db11923722dabacd976c70cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/daa9eaf129f94173db11923722dabacd976c70cc"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"f164ec87bb4b9cf0e84c7fa11bedb9da52e39e35","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
