)]}'
{"id":"openstack%2Ftripleo-docs~690768","triplet_id":"openstack%2Ftripleo-docs~master~I95f43da44c869f5de2fa4cdaa3d07fcef4ef11b9","project":"openstack/tripleo-docs","branch":"master","hashtags":[],"change_id":"I95f43da44c869f5de2fa4cdaa3d07fcef4ef11b9","subject":"Remove docs for deployed-server bootstrap files","status":"MERGED","created":"2019-10-23 23:53:35.000000000","updated":"2019-10-29 20:10:11.000000000","submitted":"2019-10-29 20:08:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"690768-1572379732088-91d4b73c","meta_rev_id":"828232d8f816d10832830d05e847f3a9fe98f258","_number":690768,"virtual_id_number":690768,"owner":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-29 20:08:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-10-25 12:32:44.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-10-29 20:02:07.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2019-10-29 20:02:07.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-25 12:32:44.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-10-29 20:02:07.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-10-29 20:08:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b133fac029140fa37071ce5d87d904483e757b13","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2019-10-23 23:53:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1c4e21834cab314215bd626d569b200ab61d583f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 00:05:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b79a7451276e49af9e3505dbea0ef6dd : SUCCESS in 5m 05s\n- build-openstack-deploy-guide https://zuul.opendev.org/t/openstack/build/bd59d982d0c2464da83c6ff1dc295d10 : SUCCESS in 5m 37s","accounts_in_message":[],"_revision_number":1},{"id":"dae82272c7fd4fbf3800ebd8e19582bdaab992db","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-10-25 12:32:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"adf68921ea3bb6d00e14707d18641e376d7e924a","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-10-29 20:02:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7f678f244ca3bbdf24706daf45742db4c4d62435","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 20:02:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d11515fc01e6f84ac4caeb4f3db95f01eecc26ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 20:08:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01e7c7b8d2b34140a8245f4885e39d34 : SUCCESS in 5m 38s\n- build-openstack-deploy-guide https://zuul.opendev.org/t/openstack/build/f0d36fc79dc8440cbc6c29ff2f05386a : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":1},{"id":"735baf059e1ec523d593f0c3fd477e142c37bc7c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 20:08:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"828232d8f816d10832830d05e847f3a9fe98f258","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 20:10:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef5d367cb48e46679851070b9b0ce101 : SUCCESS in 1m 01s\n- promote-deploy-guide https://zuul.opendev.org/t/openstack/build/d20dd30b09dd4d37841cfc7736ab5463 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"129877f17e962311ad3fe6ba560caf395edbbe7b","revisions":{"129877f17e962311ad3fe6ba560caf395edbbe7b":{"kind":"REWORK","_number":1,"created":"2019-10-23 23:53:35.000000000","uploader":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"ref":"refs/changes/68/690768/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-docs","ref":"refs/changes/68/690768/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/68/690768/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/68/690768/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/68/690768/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-docs refs/changes/68/690768/1"}}},"commit":{"parents":[{"commit":"daa9eaf129f94173db11923722dabacd976c70cc","subject":"Update docs to not use deployed-server-roles-data.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/daa9eaf129f94173db11923722dabacd976c70cc"}]}],"author":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2019-10-23 23:53:04.000000000","tz":-240},"committer":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2019-10-23 23:53:04.000000000","tz":-240},"subject":"Remove docs for deployed-server bootstrap files","message":"Remove docs for deployed-server bootstrap files\n\nThese envrionments were deprecated in\n198cfb6f8d6c4384118aee94cd404a8a54432691 and are no longer needed.\n\nChange-Id: I95f43da44c869f5de2fa4cdaa3d07fcef4ef11b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/129877f17e962311ad3fe6ba560caf395edbbe7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/129877f17e962311ad3fe6ba560caf395edbbe7b"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"daa9eaf129f94173db11923722dabacd976c70cc","is_merged_in_target_branch":true,"change_id":"I55cb5fc4a069dc3829eb7e0b4ae6c9bc37094740","change_number":690449,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
