)]}'
{"id":"openstack%2Ftripleo-docs~721209","triplet_id":"openstack%2Ftripleo-docs~master~Id98f49fd314ae4521fb73c26fae5bfaeea1abc4d","project":"openstack/tripleo-docs","branch":"master","hashtags":[],"change_id":"Id98f49fd314ae4521fb73c26fae5bfaeea1abc4d","subject":"Document ceph dashboard Safe mode feature","status":"MERGED","created":"2020-04-20 09:12:27.000000000","updated":"2020-04-20 16:09:50.000000000","submitted":"2020-04-20 16:02:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"721209-1587398572883-653b66dd","meta_rev_id":"01e0407fd1a1091baaee1df5b7c9c7a661c9f91a","_number":721209,"virtual_id_number":721209,"owner":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":0,"_account_id":15205,"name":"Guillaume Abrioux","email":"gabrioux@redhat.com"},{"value":0,"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-20 16:02:52.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-04-20 15:53:06.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":0,"_account_id":15205,"name":"Guillaume Abrioux","email":"gabrioux@redhat.com"},{"value":2,"date":"2020-04-20 10:13:54.000000000","_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-04-20 15:53:06.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2020-04-20 10:11:31.000000000","_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":0,"date":"2020-04-20 10:11:35.000000000","_account_id":15205,"name":"Guillaume Abrioux","email":"gabrioux@redhat.com"},{"value":0,"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"_account_id":15205,"name":"Guillaume Abrioux","email":"gabrioux@redhat.com"},{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-20 10:11:31.000000000","updated_by":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"reviewer":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"state":"REVIEWER"},{"updated":"2020-04-20 10:11:35.000000000","updated_by":{"_account_id":15205,"name":"Guillaume Abrioux","email":"gabrioux@redhat.com"},"reviewer":{"_account_id":15205,"name":"Guillaume Abrioux","email":"gabrioux@redhat.com"},"state":"REVIEWER"},{"updated":"2020-04-20 10:13:54.000000000","updated_by":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"reviewer":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"state":"REVIEWER"},{"updated":"2020-04-20 15:53:06.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-04-20 16:02:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9d04fe3e7ecd908bf9e58cb53e5f5ebe046a8859","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2020-04-20 09:12:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6ea92f8d2c50c0650eab1c34d3c4a6c45f328447","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 09:26:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25450eabb27245979b322aa73cfe2fa0 : POST_FAILURE in 12m 22s\n- build-openstack-deploy-guide https://zuul.opendev.org/t/openstack/build/79e33ab094994c1c89acbf4290d58e69 : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":1},{"id":"20b57a54b0b191a84299d28ed0ed4da9fc082a40","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2020-04-20 09:32:49.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"2bee85e03c6d1c69f7c2cfe78ff22e39753d80a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 10:01:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad5671f40eb147b8827e5df7c822bb0b : SUCCESS in 6m 09s\n- build-openstack-deploy-guide https://zuul.opendev.org/t/openstack/build/fed422155f2745829568593bc114b04f : SUCCESS in 6m 19s","accounts_in_message":[],"_revision_number":2},{"id":"9feec00493c3316ea6b0756a0c07805a58f0fbaa","author":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"date":"2020-04-20 10:13:54.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"baaff339df16da77b140c17e54948a7fb0120935","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-04-20 15:53:06.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1359088ad7b196d7baa098bff84bcadea70bc189","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 15:53:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3700c7ae879072ce2b0da7a32d8ce7c4fa514d1f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 16:02:52.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed105b6f78d043818537eeea212a7b30 : SUCCESS in 7m 13s\n- build-openstack-deploy-guide https://zuul.opendev.org/t/openstack/build/cf213a382bb04bb79ea0df7bd6f9c4bf : SUCCESS in 6m 03s","accounts_in_message":[],"_revision_number":2},{"id":"953c6754de1e031fe478f5602deb40c7ff13a357","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 16:02:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"01e0407fd1a1091baaee1df5b7c9c7a661c9f91a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 16:09:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ca471e88219416391e5d4bb47a5a152 : SUCCESS in 1m 18s\n- promote-deploy-guide https://zuul.opendev.org/t/openstack/build/5a55f9b65cc14ec9bdc98aacae91fa8e : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f907d0af58e60ea070f233a43826307c4809c4ea","revisions":{"3c5e109a20789cd1dcaa2e287a38bf70e97b21bf":{"kind":"REWORK","_number":1,"created":"2020-04-20 09:12:27.000000000","uploader":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"ref":"refs/changes/09/721209/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-docs","ref":"refs/changes/09/721209/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/09/721209/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/09/721209/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/09/721209/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-docs refs/changes/09/721209/1"}}},"commit":{"parents":[{"commit":"9814a1a90668b2c73f0b903a3f4ffcee1e2a3822","subject":"Cleanup py27 support and docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/9814a1a90668b2c73f0b903a3f4ffcee1e2a3822"}]}],"author":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2020-04-20 09:04:00.000000000","tz":120},"committer":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2020-04-20 09:04:00.000000000","tz":120},"subject":"Document ceph dashboard Safe mode feature","message":"Document ceph dashboard Safe mode feature\n\nThe Ceph dashboard is deployed using read-only privileges\nfor admin user, so no ceph cluster changes can be made\n(e.g., pool creation).\nThis new paragraph document the Safe mode feature and how\nto alter the behaviour of the admin user, providing fully\nadministrative privileges.\n\nChange-Id: Id98f49fd314ae4521fb73c26fae5bfaeea1abc4d\nSigned-off-by: Francesco Pantano \u003cfpantano@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/3c5e109a20789cd1dcaa2e287a38bf70e97b21bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/3c5e109a20789cd1dcaa2e287a38bf70e97b21bf"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"9814a1a90668b2c73f0b903a3f4ffcee1e2a3822","is_merged_in_target_branch":true,"change_id":"I2659000c905b3de2f37bd88ab348d2f45e0968cc","change_number":719184,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"f907d0af58e60ea070f233a43826307c4809c4ea":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-04-20 09:32:49.000000000","uploader":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"ref":"refs/changes/09/721209/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-docs","ref":"refs/changes/09/721209/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/09/721209/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/09/721209/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/09/721209/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-docs refs/changes/09/721209/2"}}},"commit":{"parents":[{"commit":"9814a1a90668b2c73f0b903a3f4ffcee1e2a3822","subject":"Cleanup py27 support and docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/9814a1a90668b2c73f0b903a3f4ffcee1e2a3822"}]}],"author":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2020-04-20 09:04:00.000000000","tz":120},"committer":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2020-04-20 09:31:56.000000000","tz":120},"subject":"Document ceph dashboard Safe mode feature","message":"Document ceph dashboard Safe mode feature\n\nThe Ceph dashboard is deployed using read-only privileges\nfor admin user, so no ceph cluster changes can be made\n(e.g., pool creation).\nThe purpose of this new paragraph is to document the Safe\nmode feature and how to alter the behaviour of the admin\nuser, providing fully administrative privileges.\n\nChange-Id: Id98f49fd314ae4521fb73c26fae5bfaeea1abc4d\nSigned-off-by: Francesco Pantano \u003cfpantano@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/f907d0af58e60ea070f233a43826307c4809c4ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/f907d0af58e60ea070f233a43826307c4809c4ea"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"9814a1a90668b2c73f0b903a3f4ffcee1e2a3822","is_merged_in_target_branch":true,"change_id":"I2659000c905b3de2f37bd88ab348d2f45e0968cc","change_number":719184,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
