)]}'
{"id":"openstack%2Ftripleo-docs~723793","triplet_id":"openstack%2Ftripleo-docs~master~Ifd671623b053789a99ce1e4c49c994e8e4d07241","project":"openstack/tripleo-docs","branch":"master","topic":"cent8repo","hashtags":[],"change_id":"Ifd671623b053789a99ce1e4c49c994e8e4d07241","subject":"Add centos 8 repository info","status":"MERGED","created":"2020-04-28 04:40:22.000000000","updated":"2020-05-28 13:28:06.000000000","submitted":"2020-05-28 13:24:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"723793-1590672262383-2d12829e","meta_rev_id":"699425f394afbb0cda0c6a48c8fe4534158d3bba","_number":723793,"virtual_id_number":723793,"owner":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2020-04-30 01:54:10.000000000","_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-28 13:24:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-05-28 13:12:34.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},{"value":1,"date":"2020-05-28 10:51:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-29 12:56:38.000000000","_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-05-28 13:12:34.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-29 12:56:38.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-05-28 10:51:08.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2020-05-28 13:12:34.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-05-28 13:24:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7c953837d4ec3c751ffdc31f59d32ce7b92707a7","author":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"date":"2020-04-28 04:40:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a8588281721fa66441d4acf60334ce4529137a1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 04:47:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/370baf2b6a9049619c4596be70ca4919 : SUCCESS in 5m 32s\n- build-openstack-deploy-guide https://zuul.opendev.org/t/openstack/build/1ad3622a47494c5883d043a8f2c31039 : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":1},{"id":"e466fc974c478cb028c4e0f56c326ca1587a7736","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-04-28 13:02:57.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"68ca4e89680f68c164ee296d6dad4c5bdb25af7d","author":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"date":"2020-04-29 03:03:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"92a4a51fd22273380ac8537b696f09b7c2ddb21d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 03:11:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90a4e7bda3d746e080f5f92f2852767e : SUCCESS in 6m 49s\n- build-openstack-deploy-guide https://zuul.opendev.org/t/openstack/build/708f24ca396a4997a898d11ec91e16a2 : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":2},{"id":"6707f763d81743e3aeedf7b1f5b6b3339cbe8afc","author":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"date":"2020-04-29 03:13:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"879d37dfd5a9963c8ac1d815839f287eacb5d597","author":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"date":"2020-04-29 03:18:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e1b96b1edda4da7de3a88ef3de0c19a11a541696","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 03:28:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2640d9c87f334d94995c2349ccfa87a3 : SUCCESS in 6m 06s\n- build-openstack-deploy-guide https://zuul.opendev.org/t/openstack/build/adf2180e092e47ef95f9d0b77ae3e838 : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":3},{"id":"5caed68f4a964fa076b3a8517435a42c5322e4c0","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-04-29 12:44:07.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"15f05df64491c4c91637a7f646748a0091777b57","author":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"date":"2020-04-29 12:54:47.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7bb42ce173a770aa34859cdeb275ba892e850038","author":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"date":"2020-04-29 12:55:08.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3bd9970e1310401a01fbf660b094ca78dd9944fa","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-04-29 12:56:38.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"1d2e2951ac493e4232e10733fe4183d76d3a082b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 13:04:42.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c2da0d299584ca1a1c205780c336d00 : SUCCESS in 8m 04s\n- build-openstack-deploy-guide https://zuul.opendev.org/t/openstack/build/382d92ee320c46f4840442663b07e1d3 : SUCCESS in 5m 53s","accounts_in_message":[],"_revision_number":4},{"id":"ad7ac42a86afd47218738b6c5322624078b0bc36","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-04-29 13:54:37.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"335c35363d4b48268ae49cc90d199d1ce353799f","author":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"date":"2020-04-30 01:54:10.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f4d325f665be3f767b366432482c80767ba79e7e","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-05-28 10:51:08.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6a0e72c705411e531a2b46f24cb7a159b318cfc7","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-05-28 13:12:34.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"eb66ecf90aeee32f3f84f6655cd2a4d9e7ad68c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 13:12:45.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a2f65015b157db9ea62170a2d1623bc4dbf17cde","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 13:24:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"9b066f0afc041d4589dd49229e143ec194e34fc8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 13:24:22.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58eacd2eb7cd4d0398aba7de430ec11b : SUCCESS in 6m 39s\n- build-openstack-deploy-guide https://zuul.opendev.org/t/openstack/build/3e862b7033df498396611cb9268c4cdd : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":4},{"id":"2e5cb1bf65df244371021f14f17b9f7de7e26f7a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 13:28:06.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5351d00950974ca5ba727a3e698929b1 : SUCCESS in 1m 21s\n- promote-deploy-guide https://zuul.opendev.org/t/openstack/build/b5c2045322cd4b5bb7aed29d577a6378 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"1775ace3811ebe8c3508f8bfba0c6ecd64f51bd6","revisions":{"0e0bb46257e1f250aeae6c8380b299cf9920fecf":{"kind":"REWORK","_number":1,"created":"2020-04-28 04:40:22.000000000","uploader":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"ref":"refs/changes/93/723793/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-docs","ref":"refs/changes/93/723793/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/93/723793/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/93/723793/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/93/723793/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-docs refs/changes/93/723793/1"}}},"commit":{"parents":[{"commit":"03335434612f0b81c9de2479914146a705cb0bb4","subject":"Merge \"Update documentation for changed config-download location\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/03335434612f0b81c9de2479914146a705cb0bb4"}]}],"author":{"name":"michaeltchapman","email":"woppin@gmail.com","date":"2020-04-28 04:36:03.000000000","tz":600},"committer":{"name":"michaeltchapman","email":"woppin@gmail.com","date":"2020-04-28 04:40:16.000000000","tz":600},"subject":"Add centos 8 repository info","message":"Add centos 8 repository info\n\nThe tripleo repos rpm to be installed has changed its python version\nand its path - now residing under components/tripleo and needs its\nlocation updated.\n\nChange-Id: Ifd671623b053789a99ce1e4c49c994e8e4d07241\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/0e0bb46257e1f250aeae6c8380b299cf9920fecf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/0e0bb46257e1f250aeae6c8380b299cf9920fecf"}]},"branch":"refs/heads/master"},"3ee7ead0cef4b9d9e6874ccbb7cf0166e5cb8a76":{"kind":"REWORK","_number":2,"created":"2020-04-29 03:03:54.000000000","uploader":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"ref":"refs/changes/93/723793/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-docs","ref":"refs/changes/93/723793/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/93/723793/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/93/723793/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/93/723793/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-docs refs/changes/93/723793/2"}}},"commit":{"parents":[{"commit":"03335434612f0b81c9de2479914146a705cb0bb4","subject":"Merge \"Update documentation for changed config-download location\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/03335434612f0b81c9de2479914146a705cb0bb4"}]}],"author":{"name":"michaeltchapman","email":"woppin@gmail.com","date":"2020-04-28 04:36:03.000000000","tz":600},"committer":{"name":"michaeltchapman","email":"woppin@gmail.com","date":"2020-04-29 03:03:34.000000000","tz":600},"subject":"Add centos 8 repository info","message":"Add centos 8 repository info\n\nThe tripleo repos rpm to be installed has changed its python version\nand its path - now residing under components/tripleo and needs its\nlocation updated.\n\nChange-Id: Ifd671623b053789a99ce1e4c49c994e8e4d07241\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/3ee7ead0cef4b9d9e6874ccbb7cf0166e5cb8a76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/3ee7ead0cef4b9d9e6874ccbb7cf0166e5cb8a76"}]},"branch":"refs/heads/master"},"3ac6d035edf0076c57ef4c7dede973eb12705166":{"kind":"REWORK","_number":3,"created":"2020-04-29 03:18:26.000000000","uploader":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"ref":"refs/changes/93/723793/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-docs","ref":"refs/changes/93/723793/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/93/723793/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/93/723793/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/93/723793/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-docs refs/changes/93/723793/3"}}},"commit":{"parents":[{"commit":"03335434612f0b81c9de2479914146a705cb0bb4","subject":"Merge \"Update documentation for changed config-download location\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/03335434612f0b81c9de2479914146a705cb0bb4"}]}],"author":{"name":"michaeltchapman","email":"woppin@gmail.com","date":"2020-04-28 04:36:03.000000000","tz":600},"committer":{"name":"michaeltchapman","email":"woppin@gmail.com","date":"2020-04-29 03:18:13.000000000","tz":600},"subject":"Add centos 8 repository info","message":"Add centos 8 repository info\n\nThe tripleo repos rpm to be installed has changed its python version\nand its path - now residing under components/tripleo and needs its\nlocation updated.\n\nChange-Id: Ifd671623b053789a99ce1e4c49c994e8e4d07241\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/3ac6d035edf0076c57ef4c7dede973eb12705166"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/3ac6d035edf0076c57ef4c7dede973eb12705166"}]},"branch":"refs/heads/master"},"1775ace3811ebe8c3508f8bfba0c6ecd64f51bd6":{"kind":"REWORK","_number":4,"created":"2020-04-29 12:54:47.000000000","uploader":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"ref":"refs/changes/93/723793/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-docs","ref":"refs/changes/93/723793/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/93/723793/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/93/723793/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/93/723793/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-docs refs/changes/93/723793/4"}}},"commit":{"parents":[{"commit":"03335434612f0b81c9de2479914146a705cb0bb4","subject":"Merge \"Update documentation for changed config-download location\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/03335434612f0b81c9de2479914146a705cb0bb4"}]}],"author":{"name":"michaeltchapman","email":"woppin@gmail.com","date":"2020-04-28 04:36:03.000000000","tz":600},"committer":{"name":"michaeltchapman","email":"woppin@gmail.com","date":"2020-04-29 12:54:28.000000000","tz":600},"subject":"Add centos 8 repository info","message":"Add centos 8 repository info\n\nThe tripleo repos rpm to be installed has changed its python version\nand its path - now residing under components/tripleo and needs its\nlocation updated.\n\nChange-Id: Ifd671623b053789a99ce1e4c49c994e8e4d07241\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/1775ace3811ebe8c3508f8bfba0c6ecd64f51bd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/1775ace3811ebe8c3508f8bfba0c6ecd64f51bd6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
