)]}'
{"id":"openstack%2Ftripleo-docs~746403","triplet_id":"openstack%2Ftripleo-docs~master~I1bcf9e0f60f26003b92ab926e0c31a7c54ac30f8","project":"openstack/tripleo-docs","branch":"master","topic":"typo","hashtags":[],"change_id":"I1bcf9e0f60f26003b92ab926e0c31a7c54ac30f8","subject":"Fix a typo in the document","status":"MERGED","created":"2020-08-16 13:06:26.000000000","updated":"2020-08-16 16:41:30.000000000","submitted":"2020-08-16 16:39:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"746403-1597595944120-76a6aaaf","meta_rev_id":"ae3b285f50e6ba6eb14d5b8bee4528b78c107e95","_number":746403,"virtual_id_number":746403,"owner":{"_account_id":30696,"name":"zhaoleilc","email":"15247232416@163.com","username":"zhaoleilc"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-16 16:39:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-16 16:30:48.000000000","_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-16 16:30:48.000000000","_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-16 16:30:48.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-08-16 16:39:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b6dbe5de4b35d6fb849bb7d833fe32bdc0a2a82e","author":{"_account_id":30696,"name":"zhaoleilc","email":"15247232416@163.com","username":"zhaoleilc"},"date":"2020-08-16 13:06:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1b6f754a8d35b21bf5243bd62c61001a1dbc18fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-16 13:12:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e46220bd6b84d87b48551bebb0f35e1 : SUCCESS in 4m 15s\n- build-openstack-deploy-guide https://zuul.opendev.org/t/openstack/build/75f94906ca634b17b2c25123d8f69b40 : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":1},{"id":"c4a3bf348d85e2bb13f89e1d1b182c74eddaa837","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-08-16 16:30:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ad857a8c760aa78b96a5000f365b35996f517354","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-16 16:31:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8659492bd4fc634b71191713747aa0c675d723cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-16 16:39:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e0e19b1857b4893a9206536ca569f20 : SUCCESS in 6m 00s\n- build-openstack-deploy-guide https://zuul.opendev.org/t/openstack/build/421d32fb6a774f16aa9217c204780931 : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":1},{"id":"5f4d02d31edd3ecdcaabb3f4ddb03235a73cb3cf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-16 16:39:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9615f2b6ad40bfe6cc7f6c1425792becf9ea7fd7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-16 16:41:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cad08ee70ec840e9b79cf93b2ac2514c : SUCCESS in 54s\n- promote-deploy-guide https://zuul.opendev.org/t/openstack/build/a92754077f26488ba300384a0e14267f : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"996d8613bd61a4b74ae7ddcaa8b5a6895ed826e4","revisions":{"996d8613bd61a4b74ae7ddcaa8b5a6895ed826e4":{"kind":"REWORK","_number":1,"created":"2020-08-16 13:06:26.000000000","uploader":{"_account_id":30696,"name":"zhaoleilc","email":"15247232416@163.com","username":"zhaoleilc"},"ref":"refs/changes/03/746403/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-docs","ref":"refs/changes/03/746403/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/03/746403/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/03/746403/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/03/746403/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-docs refs/changes/03/746403/1"}}},"commit":{"parents":[{"commit":"64aee82ad841bc33c07d3a96156e71495e5f6037","subject":"Rename and update links related to tripleo-ansible roles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/64aee82ad841bc33c07d3a96156e71495e5f6037"}]}],"author":{"name":"zhaoleilc","email":"15247232416@163.com","date":"2020-08-16 13:04:12.000000000","tz":480},"committer":{"name":"zhaoleilc","email":"15247232416@163.com","date":"2020-08-16 13:04:12.000000000","tz":480},"subject":"Fix a typo in the document","message":"Fix a typo in the document\n\nThis patch changes \u0027seperate\u0027 to\n\u0027separate\u0027 in the 80 line of deploy-guide\n/source/features/multiple_overclouds.rst\n\nChange-Id: I1bcf9e0f60f26003b92ab926e0c31a7c54ac30f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/996d8613bd61a4b74ae7ddcaa8b5a6895ed826e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/996d8613bd61a4b74ae7ddcaa8b5a6895ed826e4"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"64aee82ad841bc33c07d3a96156e71495e5f6037","is_merged_in_target_branch":true,"change_id":"Ib62963bef59062ac9d81d0cf1f3972a08335af9d","change_number":745811,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
