)]}'
{"id":"openstack%2Ftripleo-heat-templates~789708","triplet_id":"openstack%2Ftripleo-heat-templates~stable%2Fqueens~Icfaa6ad999d59bfde1abbbef7e9c9598c3053c8e","project":"openstack/tripleo-heat-templates","branch":"stable/queens","topic":"bug-1956712","hashtags":[],"change_id":"Icfaa6ad999d59bfde1abbbef7e9c9598c3053c8e","subject":"[stein and before] Fix service_hostnames output","status":"ABANDONED","created":"2021-05-05 06:43:45.000000000","updated":"2021-07-08 15:19:46.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a654bdb6d2e477451d6bcd1ef40000d30ef408bc","_number":789708,"virtual_id_number":789708,"owner":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-05-05 07:04:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2021-05-05 07:45:21.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"all":[{"value":2,"date":"2021-05-11 22:25:30.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-05 07:04:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-05 07:45:21.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-11 22:25:30.000000000","updated_by":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"reviewer":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"state":"REVIEWER"}],"messages":[{"id":"bd98ca4de942b8fff6338519fb9cdd39c18c9bf1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-05-05 06:43:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c290c533b3efc8e7ef88eee3ce004304da0310c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-05-05 06:48:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"61c819f2ebc18ded0bea4de897c8baf30bf53de4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 07:04:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/9a57cb5f898e4c40aedfcd663fbfd652 : FAILURE in 10m 11s\n- tripleo-ci-centos-7-undercloud-oooq https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tox-py35 https://zuul.opendev.org/t/openstack/build/423d539855f04bb3b2e2ef8f582cf2b4 : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbdecd717e374847aa8174452aef692b : SUCCESS in 6m 02s","accounts_in_message":[],"_revision_number":2},{"id":"e739d3573d5869b56db7795c011b25191bed80cf","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-05-05 07:45:21.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/b5a2fb4fdc3541e58d4ee7943220afc5 : FAILURE in 17m 49s","accounts_in_message":[],"_revision_number":2},{"id":"6a926ed185372c8cbd9b48ee2eb268674e0f4162","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-05-06 03:41:36.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"1a20aadc923ec69f4288e23c787e336d8d228148","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-05-06 05:00:24.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/b6222c80f2744d548dd4d62ec112b2c1 : FAILURE in 1h 16m 54s","accounts_in_message":[],"_revision_number":2},{"id":"2dfcc22329b956b4f4d478e642e4fce203890f3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 06:09:23.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/ca933bec0f954fd9901721ec9db4a53a : SUCCESS in 2h 24m 40s\n- tripleo-ci-centos-7-undercloud-oooq https://zuul.opendev.org/t/openstack/build/beba9b645f5b45ebaa3701230889c216 : SUCCESS in 1h 12m 19s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/002d511734b440318cd928b1aba4f853 : FAILURE in 51m 39s\n- tox-py35 https://zuul.opendev.org/t/openstack/build/6ea1e7bdd45d47978aad549c480d0694 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdc06455a3984666aeb2dfa6ababe566 : SUCCESS in 10m 52s","accounts_in_message":[],"_revision_number":2},{"id":"4e7e07a4155dc981c4119bb3a1079036c2d20b4d","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-05-07 05:15:25.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"129baacd36daa4f5652ceb3175fca7ddf40b09ed","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-05-07 06:45:15.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/31612221ad404bffb69e8574ba037fca : FAILURE in 1h 28m 50s","accounts_in_message":[],"_revision_number":2},{"id":"d9ab3101e1197f7c8b46f95c1c660df06b62fa8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-07 08:17:04.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/aefb971b7ef442e29b59d1f0c31d34d3 : SUCCESS in 2h 56m 24s\n- tripleo-ci-centos-7-undercloud-oooq https://zuul.opendev.org/t/openstack/build/7b5cf0fed74e4609b302816daa8c9e2b : SUCCESS in 1h 27m 02s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/834a9cd552424ee1bb0c365d8e06332d : FAILURE in 1h 01m 20s\n- tox-py35 https://zuul.opendev.org/t/openstack/build/9bc29ef1d6094b0492a41503f7caf50a : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd80d952a1584defa617253e3594e3b3 : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":2},{"id":"e54f29e3037b969c81ffe6843c6b15f865c4d708","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-05-11 22:25:30.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b7f29efc87d4208647477dc1abd69df44a51f251","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-05-14 04:57:02.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"599add864bf3801cfb7697083ef2217917ae4e80","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-05-14 06:25:50.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/161e4e3486894420aeb3a879109cc9bb : FAILURE in 1h 26m 40s","accounts_in_message":[],"_revision_number":2},{"id":"eaa20585c92386c01d5286b7fe5e6a6534ac627e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 07:43:00.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/2c8799df674f4a63b8cd4d7dc38ccb6b : SUCCESS in 2h 45m 01s\n- tripleo-ci-centos-7-undercloud-oooq https://zuul.opendev.org/t/openstack/build/3a5cb103d7fb4b90860334f8d7e93722 : SUCCESS in 1h 03m 12s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/3dce5e4ce21442e9a67c3f34020900ae : FAILURE in 50m 16s\n- tox-py35 https://zuul.opendev.org/t/openstack/build/87af57b0192449049ecfcf3aade85a4b : SUCCESS in 10m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39f843b0a0f24e08b33e5f7f4dea3346 : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":2},{"id":"a654bdb6d2e477451d6bcd1ef40000d30ef408bc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-07-08 15:19:46.000000000","message":"Abandoned\n\nbranch is going EOL see http://lists.openstack.org/pipermail/openstack-discuss/2021-July/023537.html","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"34769f173feb62b7c409b1624a152a42af2b4942","revisions":{"6e14725e6a2ff5dd288e958ef82db33ce2f912b1":{"kind":"REWORK","_number":1,"created":"2021-05-05 06:43:45.000000000","uploader":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"ref":"refs/changes/08/789708/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-heat-templates","ref":"refs/changes/08/789708/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-heat-templates refs/changes/08/789708/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-heat-templates refs/changes/08/789708/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-heat-templates refs/changes/08/789708/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-heat-templates refs/changes/08/789708/1"}}},"commit":{"parents":[{"commit":"0f47a4de9b5e64023ddc594d6106fff2c053bcfb","subject":"Ensure old installation match new images and get python-docker.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-heat-templates/commit/0f47a4de9b5e64023ddc594d6106fff2c053bcfb"}]}],"author":{"name":"ramishra","email":"ramishra@redhat.com","date":"2021-05-05 05:55:56.000000000","tz":330},"committer":{"name":"ramishra","email":"ramishra@redhat.com","date":"2021-05-05 06:43:07.000000000","tz":330},"subject":"[stein and before] Fix service_hostnames output","message":"[stein and before] Fix service_hostnames output\n\nCheck if NetworkHostnameMap is empty map before\nbuilding service_hostnames. Else map_replace would\nleave with the network names.\n\nFrom stable/train ansible is used for all_nodes\nhieradata.\n\nRelated: https://bugzilla.redhat.com/show_bug.cgi?id\u003d1956712\nChange-Id: Icfaa6ad999d59bfde1abbbef7e9c9598c3053c8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-heat-templates/commit/6e14725e6a2ff5dd288e958ef82db33ce2f912b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-heat-templates/commit/6e14725e6a2ff5dd288e958ef82db33ce2f912b1"}]},"branch":"refs/heads/stable/queens"},"34769f173feb62b7c409b1624a152a42af2b4942":{"kind":"REWORK","_number":2,"created":"2021-05-05 06:48:01.000000000","uploader":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"ref":"refs/changes/08/789708/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-heat-templates","ref":"refs/changes/08/789708/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-heat-templates refs/changes/08/789708/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-heat-templates refs/changes/08/789708/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-heat-templates refs/changes/08/789708/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-heat-templates refs/changes/08/789708/2"}}},"commit":{"parents":[{"commit":"0f47a4de9b5e64023ddc594d6106fff2c053bcfb","subject":"Ensure old installation match new images and get python-docker.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-heat-templates/commit/0f47a4de9b5e64023ddc594d6106fff2c053bcfb"}]}],"author":{"name":"ramishra","email":"ramishra@redhat.com","date":"2021-05-05 05:55:56.000000000","tz":330},"committer":{"name":"ramishra","email":"ramishra@redhat.com","date":"2021-05-05 06:47:24.000000000","tz":330},"subject":"[stein and before] Fix service_hostnames output","message":"[stein and before] Fix service_hostnames output\n\nCheck if NetworkHostnameMap is empty map before\nbuilding service_hostnames. Else map_replace would\nleave with the network names.\n\nFrom stable/train ansible is used for all_nodes\nhieradata.\n\nRelated: https://bugzilla.redhat.com/show_bug.cgi?id\u003d1956712\nChange-Id: Icfaa6ad999d59bfde1abbbef7e9c9598c3053c8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-heat-templates/commit/34769f173feb62b7c409b1624a152a42af2b4942"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-heat-templates/commit/34769f173feb62b7c409b1624a152a42af2b4942"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
