)]}'
{"id":"openstack%2Ftripleo-heat-templates~822708","triplet_id":"openstack%2Ftripleo-heat-templates~stable%2Fussuri~Id04d1c45e37b6bd1e890d1cf34f17b12b52b034a","project":"openstack/tripleo-heat-templates","branch":"stable/ussuri","topic":"lp1946659","hashtags":[],"change_id":"Id04d1c45e37b6bd1e890d1cf34f17b12b52b034a","subject":"DNM testing tripleo-ci/+/822482","status":"ABANDONED","created":"2021-12-22 14:16:12.000000000","updated":"2022-03-18 15:48:25.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f981b30254e48e4c2d2a95ddba85291a0f2c1ad0","_number":822708,"virtual_id_number":822708,"owner":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-12-22 18:23:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-12-22 18:14:03.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-22 18:14:03.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-22 18:23:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3ae0ad76053e5c452c904c4588c51cc43ed01063","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-12-22 14:16:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2140ec371b2c73f33309699edaf91e203dbc15ec","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-12-22 14:30:36.000000000","message":"Topic set to lp1946659","accounts_in_message":[],"_revision_number":1},{"id":"54234ede1f0e471d3396a796a5f75c49d01a79ef","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-12-22 18:14:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/4e4d1bedc130405a90db3438dbf2cd3d : SUCCESS in 3h 13m 36s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp_1supp-featureset039 https://review.rdoproject.org/zuul/build/a9c9d49932c24456bbc45381c294a203 : SUCCESS in 3h 56m 43s","accounts_in_message":[],"_revision_number":1},{"id":"f848c770e5d85a832c731086c5cc68236c2beb3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 18:23:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02704f4aa4b44e8fa63434cef7a4058e : SUCCESS in 6m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cde069e400344b19881a7b0cdcb6e7ac : SUCCESS in 4m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0416ea182f1b49e192ff14fae63df63d : SUCCESS in 6m 13s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/f0752ae7b3074dac85f1201c399a79d9 : SUCCESS in 3h 52m 49s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/9a2bacf6bbd8413cbccaf16e98cf76e9 : SUCCESS in 4h 03m 51s\n- tripleo-ci-centos-8-standalone-upgrade-ussuri https://zuul.opendev.org/t/openstack/build/caf12e49a0f9414bac4ab99f01c7bc0c : SUCCESS in 2h 29m 13s\n- tripleo-ci-centos-8-undercloud-upgrade-ussuri https://zuul.opendev.org/t/openstack/build/a7781ece2fbe4d48b7e4e0b5bf720546 : SUCCESS in 2h 28m 05s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-ussuri https://zuul.opendev.org/t/openstack/build/578375c253614c488e797a6a70f5a5a4 : SUCCESS in 2h 06m 43s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades-ussuri https://zuul.opendev.org/t/openstack/build/a8af810f4f284f7aa274f9b745295659 : SUCCESS in 2h 01m 06s\n- openstack-tox-tht https://zuul.opendev.org/t/openstack/build/d0da4b0e919e4b4782672c55e229a1cd : SUCCESS in 10m 11s","accounts_in_message":[],"_revision_number":1},{"id":"f981b30254e48e4c2d2a95ddba85291a0f2c1ad0","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-03-18 15:48:25.000000000","message":"Abandoned\n\ndone testing","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2a4cd65231ace79473494a9f8b1a66783c7771d0","revisions":{"2a4cd65231ace79473494a9f8b1a66783c7771d0":{"kind":"REWORK","_number":1,"created":"2021-12-22 14:16:12.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/08/822708/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-heat-templates","ref":"refs/changes/08/822708/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-heat-templates refs/changes/08/822708/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-heat-templates refs/changes/08/822708/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-heat-templates refs/changes/08/822708/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-heat-templates refs/changes/08/822708/1"}}},"commit":{"parents":[{"commit":"0816be84ac4fc90d653b18e6c6dda248c5ae4d49","subject":"Merge \"Remove tripleo_ovn_mac_port_name port tag\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-heat-templates/commit/0816be84ac4fc90d653b18e6c6dda248c5ae4d49"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-12-21 11:07:28.000000000","tz":120},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-12-22 14:07:35.000000000","tz":120},"subject":"DNM testing tripleo-ci/+/822482","message":"DNM testing tripleo-ci/+/822482\n\nDepends-On: https://review.opendev.org/c/openstack/tripleo-ci/+/822482\nChange-Id: Id04d1c45e37b6bd1e890d1cf34f17b12b52b034a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-heat-templates/commit/2a4cd65231ace79473494a9f8b1a66783c7771d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-heat-templates/commit/2a4cd65231ace79473494a9f8b1a66783c7771d0"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
