)]}'
{"id":"openstack%2Ftripleo-heat-templates~829708","triplet_id":"openstack%2Ftripleo-heat-templates~stable%2Fvictoria~I4cc8837b52e6b2356b6bc4436c340b2cb83879fe","project":"openstack/tripleo-heat-templates","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"14250":{"account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"last_update":"2022-02-23 09:08:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4cc8837b52e6b2356b6bc4436c340b2cb83879fe","subject":"IDM server registration is fixed to one server only","status":"MERGED","created":"2022-02-17 16:41:45.000000000","updated":"2022-02-23 09:08:26.000000000","submitted":"2022-02-23 09:08:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"829708","cherry_pick_of_change":829707,"cherry_pick_of_patch_set":1,"meta_rev_id":"a0df09dfcd0fff95f0fdd2f09062d357da5ca528","_number":829708,"virtual_id_number":829708,"owner":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-23 09:08:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"all":[{"value":2,"date":"2022-02-21 03:53:47.000000000","_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":2,"date":"2022-02-17 22:10:17.000000000","_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"all":[{"value":1,"date":"2022-02-23 04:12:54.000000000","_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":16718,"name":"donghwi.cha","email":"donghwi.cha@gmail.com","username":"donghwi.cha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-17 16:41:45.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":16718,"name":"donghwi.cha","email":"donghwi.cha@gmail.com","username":"donghwi.cha"},"state":"CC"},{"updated":"2022-02-17 21:07:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-17 22:10:17.000000000","updated_by":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"reviewer":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"state":"REVIEWER"},{"updated":"2022-02-21 03:53:47.000000000","updated_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"REVIEWER"}],"messages":[{"id":"991d727faff7fc6d2181e9edfee20fe00415123b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2022-02-17 16:41:45.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.","accounts_in_message":[],"_revision_number":1},{"id":"106ccbb6c50f9b531e670b99e1d494b1f1dd0ac2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 21:07:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/d92d4d3a2e8f4eacafc0dc140db4b515 : SUCCESS in 4h 06m 32s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/49126662db144c0f962b4f6d09b7c00d : SUCCESS in 1h 29m 07s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/9e372e1c5aad45748233abef8aad6daf : SUCCESS in 2h 04m 57s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/8ccbf4b1de7742c0a7981803ff1f1f9c : SUCCESS in 1h 56m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/612ae75ca20c4a69a0719276f146f3ce : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/23f9f6f4716b409bacf32404947a6a34 : SUCCESS in 6m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b076e0429ee14b6099b4a3f95d29b2fc : SUCCESS in 5m 28s\n- tripleo-ci-centos-8-standalone-on-multinode-ipa https://zuul.opendev.org/t/openstack/build/6519df4e66684e5b826e7ff816dae831 : SUCCESS in 2h 06m 12s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/09874e8e79c84750a12f0375df18958c : SUCCESS in 4h 19m 11s\n- tripleo-ci-centos-8-standalone-upgrade-victoria https://zuul.opendev.org/t/openstack/build/8099596cf5404ec4ace2c295ba1c3cba : SUCCESS in 2h 28m 55s\n- tripleo-ci-centos-8-undercloud-upgrade-victoria https://zuul.opendev.org/t/openstack/build/cbd9bf2b3a854ebf8b8b336fc31b050d : SUCCESS in 1h 42m 23s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-victoria https://zuul.opendev.org/t/openstack/build/38c45e7c41ad4f48bbc76ad7bb2b9da3 : SUCCESS in 2h 15m 59s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades-victoria https://zuul.opendev.org/t/openstack/build/a1c083cb552d48719a28a1abe762ccfd : SUCCESS in 1h 56m 32s\n- openstack-tox-tht https://zuul.opendev.org/t/openstack/build/471d536c36554bc18d41f5a595abf56c : SUCCESS in 8m 09s","accounts_in_message":[],"_revision_number":1},{"id":"1ec7d423b603569d678590656d143c97b8eaa1fb","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2022-02-17 22:10:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ceeb0a3f0c185e270e77043ebb1e201d6290d3fc","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2022-02-21 03:53:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"33d574adebe70f6fbd8ce732c859e9e8e413b61d","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2022-02-23 04:12:54.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"91639aadeef688cf71acc3675dae4920e06967b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 04:14:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dbd187e5c7a0d53b1ab03356454794027a99a9c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 09:08:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/6cd67d8206a1492bb5573b76c52c6c8e : SUCCESS in 4h 54m 00s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/8cd1fe87732b404f823659eba5eaa397 : SUCCESS in 1h 33m 16s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/97fcf25956214fe7871216967a412d5e : SUCCESS in 2h 11m 56s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/1c8fc6737fb34d458dc7f72a8a2bf2dd : SUCCESS in 1h 50m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bf60d81eacb43ca89569ee6bb542970 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a0e13a884bb472481208362901ec850 : SUCCESS in 5m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0899c419cd3746458445c732b799c362 : SUCCESS in 4m 42s\n- tripleo-ci-centos-8-standalone-on-multinode-ipa https://zuul.opendev.org/t/openstack/build/164399a79a5a4060a9a4ac12eed53c4b : SUCCESS in 2h 06m 02s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/5e37450a5ea145799dc8c67f60461310 : SUCCESS in 4h 48m 34s\n- tripleo-ci-centos-8-standalone-upgrade-victoria https://zuul.opendev.org/t/openstack/build/144e5cbf0e1a4b5da1385ecf00fcc4b1 : SUCCESS in 3h 09m 13s\n- tripleo-ci-centos-8-undercloud-upgrade-victoria https://zuul.opendev.org/t/openstack/build/9f89466fec564ff4a168eece7c028716 : SUCCESS in 1h 41m 47s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-victoria https://zuul.opendev.org/t/openstack/build/467d8b51192f461786a6ce021a3f2fbb : SUCCESS in 2h 10m 57s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades-victoria https://zuul.opendev.org/t/openstack/build/e220ff15a1244cd6898795abf85e47fc : SUCCESS in 1h 55m 34s\n- openstack-tox-tht https://zuul.opendev.org/t/openstack/build/1dea17518d1340c086bc84d658a722c3 : SUCCESS in 8m 05s","accounts_in_message":[],"_revision_number":1},{"id":"a0df09dfcd0fff95f0fdd2f09062d357da5ca528","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 09:08:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c5b2e36d43aac5d33212cdc1a2b209f411232870","revisions":{"c5b2e36d43aac5d33212cdc1a2b209f411232870":{"kind":"REWORK","_number":1,"created":"2022-02-17 16:41:45.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/08/829708/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-heat-templates","ref":"refs/changes/08/829708/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-heat-templates refs/changes/08/829708/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-heat-templates refs/changes/08/829708/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-heat-templates refs/changes/08/829708/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-heat-templates refs/changes/08/829708/1"}}},"commit":{"parents":[{"commit":"371bb6528fa25935770ba6e4023048bda9f2110e","subject":"Add parameter IronicIPXEUefiSnpOnly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-heat-templates/commit/371bb6528fa25935770ba6e4023048bda9f2110e"}]}],"author":{"name":"donghwi.cha","email":"donghwi.cha@gmail.com","date":"2021-12-19 00:21:40.000000000","tz":540},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2022-02-17 16:41:45.000000000","tz":0},"subject":"IDM server registration is fixed to one server only","message":"IDM server registration is fixed to one server only\n\nHeat IdMServer can not convey multiple values of IPA servers,\nand it is set to a string, limited to one IPA server only.\n\nSince IPA ansible playbook can accept multiple values of servers,\nIdMServer needs update to be changed from string to array.\n\nCloses-Bug: #1955319\nChange-Id: I4cc8837b52e6b2356b6bc4436c340b2cb83879fe\n(cherry picked from commit 6d142165f12a03bd3d0b27ed3b377e48b6a8ad5b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-heat-templates/commit/c5b2e36d43aac5d33212cdc1a2b209f411232870"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-heat-templates/commit/c5b2e36d43aac5d33212cdc1a2b209f411232870"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
