)]}'
{"id":"openstack%2Ftripleo-heat-templates~833708","triplet_id":"openstack%2Ftripleo-heat-templates~master~I755cc96116a664f01622fcd30cdd0d82d184f0c7","project":"openstack/tripleo-heat-templates","branch":"master","topic":"bug/1964824","attention_set":{},"removed_from_attention_set":{"24245":{"account":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"last_update":"2022-03-15 09:10:03.000000000","reason":"Harald Jensås replied on the change"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-03-17 11:15:13.000000000","reason":"Change was submitted"},"20778":{"account":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},"last_update":"2022-03-17 11:15:13.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2022-03-17 11:15:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I755cc96116a664f01622fcd30cdd0d82d184f0c7","subject":"Memcached: Allow puppet to be aware IPv6 is used","status":"MERGED","created":"2022-03-15 02:48:38.000000000","updated":"2022-03-17 11:15:13.000000000","submitted":"2022-03-17 11:15:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"833708-bug/1964824","meta_rev_id":"8452b7f43712b1f062b8f4ae4cde553b0fd4057c","_number":833708,"virtual_id_number":833708,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-17 11:15:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2022-03-15 04:04:33.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"all":[{"value":2,"date":"2022-03-15 09:10:03.000000000","_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-15 07:45:40.000000000","_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"all":[{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2022-03-15 09:40:46.000000000","_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"reviewers":{"REVIEWER":[{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-15 04:04:33.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-15 05:50:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-15 07:44:43.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"CC"},{"updated":"2022-03-15 07:44:43.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-03-15 07:44:43.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},"state":"REVIEWER"},{"updated":"2022-03-15 07:44:43.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"state":"REVIEWER"},{"updated":"2022-03-15 07:45:40.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"}],"messages":[{"id":"bead701bab65379dbc3633c76a77d381a88ddd6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-15 02:48:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f1c4b27ff8bddd661ae90c0c09552530d1af6571","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-15 02:49:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"eda33a7b835b6482c3aad80ec09bef9b74cd2392","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-03-15 04:04:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/4b05454b78c74956a1d88f8dec2b4428 : FAILURE in 51m 00s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp_1supp-featureset039 https://review.rdoproject.org/zuul/build/037947d672c24b4d92e1ae8f8dcd7ecb : POST_FAILURE in 1h 13m 10s","accounts_in_message":[],"_revision_number":2},{"id":"541064b526954666eb09757987ccd96d8276eb68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 05:50:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/5c97d89a37ee4186b4fe39a289acb2c4 : SUCCESS in 2h 33m 10s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/13ac2c0d8bc84ad7886ab69094a385e0 : SUCCESS in 44m 34s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/f422c9f0be624868b5c296b0cf343578 : SUCCESS in 1h 31m 09s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/5900d29f21584a6885947c57a62ce014 : SUCCESS in 1h 21m 23s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/fd3576d2d7cc4777a9eb726452bd9890 : SUCCESS in 1h 16m 48s\n- tripleo-ci-centos-9-scenario012-standalone https://zuul.opendev.org/t/openstack/build/3b33ee7825454bd28509dfee7b0d82ab : SUCCESS in 1h 05m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd13ddd2f9834426b9eadc5f510e9f2f : SUCCESS in 10m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/146d9063d54e42c3b18bbe6f58ef4412 : SUCCESS in 8m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ee3d2358f2cb4cebab356752be4f9ff4 : SUCCESS in 8m 25s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/626d81f9be4047fcb2e79647d55ffe04 : SUCCESS in 2h 03m 05s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/b91bebcffd60497d912737a1b8997c74 : SUCCESS in 1h 13m 28s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/bf8d871bc63e4e92b8652f54e4d4a8aa : SUCCESS in 1h 08m 07s (non-voting)\n- openstack-tox-tht https://zuul.opendev.org/t/openstack/build/c24f5c2ac0564aee85ae6eb416e22069 : SUCCESS in 9m 30s","accounts_in_message":[],"_revision_number":2},{"id":"42fc8ead333f9d01b6a8558b2551e223f5bb87bb","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2022-03-15 07:45:40.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c959c59b15b942130955d967518baee97e4cce94","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2022-03-15 09:10:03.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b5b87ea05eafbd4977b1d9e45da073b438b487b8","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2022-03-15 09:40:46.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"329d2bc12456b549f982db11549cb9d5b4d6f58b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 09:41:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"046b5d97f50703ecc1b4eb2c3da0f28e9dd8a346","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 12:49:24.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/b0cb60766b9a4d5e9002d1a85238bd2d : SUCCESS in 2h 29m 35s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/467b8ea93306441c8bec5355d3b48b98 : SUCCESS in 41m 53s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/4f5a78bb9e4c420ba4a59542c7116d61 : SUCCESS in 1h 31m 18s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/32d37d528b4446e19d13d8381f302e56 : SUCCESS in 1h 34m 47s\n- tripleo-ci-centos-9-scenario012-standalone https://zuul.opendev.org/t/openstack/build/b0e79a57bfbf479a96538f364ea9638c : SUCCESS in 1h 11m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cc74bf729934f72b3fa66c1d861e708 : FAILURE in 8m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/adce31e8b2a04eee84b61ef69213c98c : SUCCESS in 6m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/808a4465d6ee4a139f654256e4007ccf : SUCCESS in 5m 53s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/788582cd99dd4f8684e745ba7467eeeb : SUCCESS in 1h 11m 46s\n- openstack-tox-tht https://zuul.opendev.org/t/openstack/build/22dd16c98db1439f9ebef580d1703cca : SUCCESS in 7m 47s","accounts_in_message":[],"_revision_number":2},{"id":"ed571a0eef8a45207424bdb6474cdb8bac10470a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-15 23:25:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"152a881831cad0af3ad7f7c73f387a172e73e035","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-03-16 01:47:06.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/669f15261f974a31ac7f61257ae96ab0 : FAILURE in 1h 44m 18s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp_1supp-featureset039 https://review.rdoproject.org/zuul/build/a7b1ca396ab34a69a866db1ccdf69269 : FAILURE in 2h 18m 47s","accounts_in_message":[],"_revision_number":2},{"id":"090647ffef9e5fd67e586233fa22bf6311d47b65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 02:36:18.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/ee740899d69d4f2e93b9b9d2ae0d6259 : SUCCESS in 2h 57m 20s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/26e717e8656e44e6b17475c2ee38d4a8 : SUCCESS in 40m 32s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/bf3fa326456549d6827669b949360b38 : FAILURE in 1h 47m 35s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/0a944ed6bcb641d7a1cec2475bf99f04 : SUCCESS in 1h 18m 24s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/f7b919bf9a014b34aeb13d0683c81810 : SUCCESS in 1h 08m 40s\n- tripleo-ci-centos-9-scenario012-standalone https://zuul.opendev.org/t/openstack/build/76f9b0d2fd9545689375e4583022cb40 : SUCCESS in 1h 04m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83152507fa53468aa68d097aa6e7d490 : SUCCESS in 7m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/62b9a65e9e444a44be585246c1c18a6f : SUCCESS in 6m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2e0ba459065f4fc0b0c63db1972c3a9a : SUCCESS in 6m 14s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/4bc621ad9d144c3da6a37b2da28a0e1f : SUCCESS in 2h 01m 45s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/a54c815015fb4cb0b2ae9d193b3ff29f : SUCCESS in 1h 10m 08s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/d9dcb28702a64f26965e1921f21f18db : SUCCESS in 1h 04m 01s (non-voting)\n- openstack-tox-tht https://zuul.opendev.org/t/openstack/build/bf31e92572be45469de9ce5b7580640f : SUCCESS in 8m 12s","accounts_in_message":[],"_revision_number":2},{"id":"c74f47e65f3fa6b30fe9c0ae881c5040c55ee2c7","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2022-03-17 05:24:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d46ccaddc680c13778359483cbaeededfb479120","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-03-17 08:33:44.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/e5166183a6484e0b83c87b54567917f0 : FAILURE in 3h 07m 32s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp_1supp-featureset039 https://review.rdoproject.org/zuul/build/c42bdf8c4a064a37a3c4a0a2a85836a6 : SUCCESS in 2h 55m 52s","accounts_in_message":[],"_revision_number":2},{"id":"10c5e39ea05526941344971360ee6094c6e3eadc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-17 08:41:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/2b41db43e4734ecb8a35f9190a18c0f3 : SUCCESS in 2h 51m 36s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/5e6f6aa2ef224ae3869c5202e15f6629 : SUCCESS in 47m 26s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/d68c35a0072e4716b5b3f041d2948d10 : SUCCESS in 1h 30m 56s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/172e01d9c7a44510abb7a1ced0644cbd : SUCCESS in 1h 29m 40s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/e4e20a863fab41fe8ae8b1d30900e94a : SUCCESS in 1h 10m 58s\n- tripleo-ci-centos-9-scenario012-standalone https://zuul.opendev.org/t/openstack/build/59030e83e8664a14991b895ace23c923 : SUCCESS in 1h 12m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59b38722c3e9437ba8ff5896987fc09d : SUCCESS in 6m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8a61047837744399542823a529a9b00 : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7dc776cafdef44558a876b1ad6d2e5c9 : SUCCESS in 6m 04s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/ee69d9d1af254d5f882e5c3f309f994b : SUCCESS in 2h 02m 21s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/37dc0e64f5974f168eb2bf56cf77d3d0 : SUCCESS in 1h 13m 08s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/0f002ac93be84d67a47add93dbaac5d4 : SUCCESS in 1h 09m 37s (non-voting)\n- openstack-tox-tht https://zuul.opendev.org/t/openstack/build/ade5eb3b6cb0427cbb0cce76b129ecf9 : SUCCESS in 7m 18s","accounts_in_message":[],"_revision_number":2},{"id":"e7cc8998666611032e71e0ab5f87843bd439fe39","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-17 08:42:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d38a83f73f42512ceaef0cd987e2072d95b58103","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-17 11:15:13.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/358af4c6744f4177ba2bb9d651e3622b : SUCCESS in 2h 29m 44s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/a45f1c8f61434f019fe1a00a45cc42b0 : SUCCESS in 41m 22s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/7451dbdbc6024d06ac2f0b63b75a9e45 : SUCCESS in 1h 26m 13s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/cbfab19714b94909a6b8d0cc66438b12 : SUCCESS in 1h 15m 04s\n- tripleo-ci-centos-9-scenario012-standalone https://zuul.opendev.org/t/openstack/build/5aec5003e4bb45358e8c05b71577907b : SUCCESS in 1h 06m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ede25e06bfbf4c4e9aeb2598038a102b : SUCCESS in 8m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5451b3d7e7684ac48e467e994959ef44 : SUCCESS in 7m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9815a925fe1d49a8b957fe98d8180463 : SUCCESS in 7m 02s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/cfa040f864004e61bb76eb28c9a968c2 : SUCCESS in 1h 05m 10s\n- openstack-tox-tht https://zuul.opendev.org/t/openstack/build/177359eda06444feb85756acf58eb3a0 : SUCCESS in 9m 47s","accounts_in_message":[],"_revision_number":2},{"id":"8452b7f43712b1f062b8f4ae4cde553b0fd4057c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-17 11:15:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6f8a5bc1ef16f9df5653c36bcf351026a49ea913","revisions":{"8cc69942efdc405c5a36d963ee604ac64bf4ac8b":{"kind":"REWORK","_number":1,"created":"2022-03-15 02:48:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/08/833708/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-heat-templates","ref":"refs/changes/08/833708/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-heat-templates refs/changes/08/833708/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-heat-templates refs/changes/08/833708/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-heat-templates refs/changes/08/833708/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-heat-templates refs/changes/08/833708/1"}}},"commit":{"parents":[{"commit":"a799241f0d842ad495b2f9efac5cae16e2f153a2","subject":"Merge \"Ironic: Use generated dnsmasq conf file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-heat-templates/commit/a799241f0d842ad495b2f9efac5cae16e2f153a2"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-03-15 01:53:27.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-03-15 01:53:27.000000000","tz":540},"subject":"Memcached: Allow puppet to be aware IPv6 is used","message":"Memcached: Allow puppet to be aware IPv6 is used\n\nWhen memcached is using IPv6 network, some parameters like\n[keystone_authtoke] memcached_servers require each server name has\ninet6 prefix, as is described in the following example.\n\n inet6:[\u003chostname\u003e]:\u003cport\u003e\n\nThis change introduces the global parameter so that puppet can detect\nwhether memcached is using IPv6 network.\n\nChange-Id: I755cc96116a664f01622fcd30cdd0d82d184f0c7\nRelated-Bug: #1964824\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-heat-templates/commit/8cc69942efdc405c5a36d963ee604ac64bf4ac8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-heat-templates/commit/8cc69942efdc405c5a36d963ee604ac64bf4ac8b"}]},"branch":"refs/heads/master"},"6f8a5bc1ef16f9df5653c36bcf351026a49ea913":{"kind":"REWORK","_number":2,"created":"2022-03-15 02:49:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/08/833708/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-heat-templates","ref":"refs/changes/08/833708/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-heat-templates refs/changes/08/833708/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-heat-templates refs/changes/08/833708/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-heat-templates refs/changes/08/833708/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-heat-templates refs/changes/08/833708/2"}}},"commit":{"parents":[{"commit":"a799241f0d842ad495b2f9efac5cae16e2f153a2","subject":"Merge \"Ironic: Use generated dnsmasq conf file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-heat-templates/commit/a799241f0d842ad495b2f9efac5cae16e2f153a2"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-03-15 01:53:27.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-03-15 02:49:26.000000000","tz":540},"subject":"Memcached: Allow puppet to be aware IPv6 is used","message":"Memcached: Allow puppet to be aware IPv6 is used\n\nWhen memcached is using IPv6 network, some parameters like\n[keystone_authtoke] memcached_servers require each server name has\ninet6 prefix, as is described in the following example.\n\n inet6:[\u003chostname\u003e]:\u003cport\u003e\n\nThis change introduces the global parameter so that puppet can detect\nwhether memcached is using IPv6 network.\n\nRelated-Bug: #1964824\nChange-Id: I755cc96116a664f01622fcd30cdd0d82d184f0c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-heat-templates/commit/6f8a5bc1ef16f9df5653c36bcf351026a49ea913"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-heat-templates/commit/6f8a5bc1ef16f9df5653c36bcf351026a49ea913"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
