)]}'
{"id":"openstack%2Ftripleo-heat-templates~848708","triplet_id":"openstack%2Ftripleo-heat-templates~master~Id812bfbf9fd27e9b9d6f7d671faf9e2a0bc88b72","project":"openstack/tripleo-heat-templates","branch":"master","topic":"bug/1980726","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-03-13 09:44:33.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Id812bfbf9fd27e9b9d6f7d671faf9e2a0bc88b72","subject":"Fix ignored OvnMetadataAgentLoggingSource","status":"ABANDONED","created":"2022-07-05 09:07:59.000000000","updated":"2023-03-13 09:44:33.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"69a421a128cdf945a108fb582764ecd7b0ac0531","_number":848708,"virtual_id_number":848708,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"disliked":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-07-05 17:46:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2022-07-05 09:36:14.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-05 09:36:14.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-05 12:04:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a1bc5f438cf0f5560473082dd2daa34b1e0208cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-05 09:07:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a3c1841615977f6992ca1bdf1c3136fd223a83d2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-07-05 09:36:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/8c5c8846ce524753acafdeacf4feefa0 : FAILURE in 25m 47s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp_1supp-featureset039 https://review.rdoproject.org/zuul/build/43eb66dbbaed44ec9508c73a8537f6e1 : FAILURE in 27m 12s","accounts_in_message":[],"_revision_number":1},{"id":"8c4959cf5522fc75570538ecfac58474874c74f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 12:04:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/15c33295c33a429d9c9047bccc33d8ac\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/40253f7d5aa64f78a205df4335b834c5 : SUCCESS in 2h 42m 39s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/2ab500d6cc18425db24b53e17977d0eb : SUCCESS in 41m 54s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/941cb1579af047c4a63158c7d3172b0d : SUCCESS in 1h 33m 38s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/a37f8ab6de934eada903b2a13ef0587a : SUCCESS in 1h 30m 29s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/b1094e2ff4d546fe89eb86163822ce48 : SUCCESS in 1h 12m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adf7b9a4f96543079a986b673079c4ba : SUCCESS in 6m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc3b86d1ec584039b6e2211f15998f42 : SUCCESS in 3m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eceafdec86c14d799b004f72ab1d6c6f : SUCCESS in 3m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/987f0ea25e6e4274b300a9a6035f528a : SUCCESS in 4m 00s (non-voting)\n- tripleo-ci-centos-9-standalone-on-multinode-ipa https://zuul.opendev.org/t/openstack/build/b6cf2f0205fe47e799ba4febc1b50c25 : FAILURE in 1h 42m 19s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/5f7202f60fd044b1b2bbcb38b5be5328 : SUCCESS in 2h 17m 42s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/d9d4e73c4bb64dfa98600f872b5983d0 : SUCCESS in 1h 28m 36s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/fb25b3cd3cba4d16b5689a53348f8d89 : SUCCESS in 1h 07m 17s\n- openstack-tox-tht https://zuul.opendev.org/t/openstack/build/36907b7ff285422192eeb3ae6afb46d9 : SUCCESS in 7m 03s","accounts_in_message":[],"_revision_number":1},{"id":"b8a0aaf6fd4d1fe4d8737dcf7a8eb13e498ac3f9","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-05 14:24:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0e21fb740dcda26a5e2a9a0fab28262998be92c0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-07-05 17:22:59.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/ae6d79db195f4854a78b044ffcf962e9 : SUCCESS in 2h 21m 43s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp_1supp-featureset039 https://review.rdoproject.org/zuul/build/587e8418df7f4667a73f04fafd655056 : FAILURE in 2h 30m 50s","accounts_in_message":[],"_revision_number":1},{"id":"bd12e62228a70e36a290200b2bd70eb8538a0e78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 17:46:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac01a98052c542f0becf4fd7faa156e5\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/d80698bfa27f4666b48b53b4219a2126 : SUCCESS in 3h 07m 38s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/8c1f8e8cdfa345b9a373e4612ee0fa43 : SUCCESS in 48m 25s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/efa123549bca45d686cf2975aa0dfde0 : SUCCESS in 1h 41m 56s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/27b9ba6f637e4150b1878b8106029112 : SUCCESS in 1h 35m 54s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/3cc52ae14dbe462c80d933acc5d9d17c : SUCCESS in 1h 23m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f18625ce95384c189cf2f39a31412c3c : SUCCESS in 7m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7716c8a74e0249af9551774e73825b4d : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df5b25ff659c4b2c867705354134ccf5 : SUCCESS in 4m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a3322fd1ff414ce3beb5535969ddc62c : SUCCESS in 5m 59s (non-voting)\n- tripleo-ci-centos-9-standalone-on-multinode-ipa https://zuul.opendev.org/t/openstack/build/434b53c62a74455686b2b7a303e497da : SUCCESS in 1h 51m 43s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/54265eee96eb4cf3b8898c8a584f8677 : SUCCESS in 2h 41m 23s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/6400dd23340a4576b87111d28a3fa663 : SUCCESS in 1h 33m 38s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/75dda81310774e569dae44e3988475f8 : SUCCESS in 1h 13m 31s\n- openstack-tox-tht https://zuul.opendev.org/t/openstack/build/c3fb71cdc87a41eb8b5ffc27ba8190e1 : SUCCESS in 6m 38s","accounts_in_message":[],"_revision_number":1},{"id":"69a421a128cdf945a108fb582764ecd7b0ac0531","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-13 09:44:33.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6db97096ddfcfd7c5d849b677a1535b472a3684a","revisions":{"6db97096ddfcfd7c5d849b677a1535b472a3684a":{"kind":"REWORK","_number":1,"created":"2022-07-05 09:07:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/08/848708/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-heat-templates","ref":"refs/changes/08/848708/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-heat-templates refs/changes/08/848708/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-heat-templates refs/changes/08/848708/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-heat-templates refs/changes/08/848708/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-heat-templates refs/changes/08/848708/1"}}},"commit":{"parents":[{"commit":"5a741ba973aa2742a7fe7d3c85b6ce1d353965aa","subject":"Merge \"ironic tftp: Do not format IPv6 address for bind_host\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-heat-templates/commit/5a741ba973aa2742a7fe7d3c85b6ce1d353965aa"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-05 09:07:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-05 09:07:45.000000000","tz":540},"subject":"Fix ignored OvnMetadataAgentLoggingSource","message":"Fix ignored OvnMetadataAgentLoggingSource\n\n... and ensure the log file of ovn-metadata is captured by rsyslog.\n\nCloses-Bug: #1980726\nChange-Id: Id812bfbf9fd27e9b9d6f7d671faf9e2a0bc88b72\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-heat-templates/commit/6db97096ddfcfd7c5d849b677a1535b472a3684a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-heat-templates/commit/6db97096ddfcfd7c5d849b677a1535b472a3684a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
