)]}'
{"id":"openstack%2Ftripleo-heat-templates~856310","triplet_id":"openstack%2Ftripleo-heat-templates~master~I25e0cbeed80fc77ee2d5c37d73a95d844b8f428d","project":"openstack/tripleo-heat-templates","branch":"master","topic":"bug/1981150","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-10-14 03:05:56.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I25e0cbeed80fc77ee2d5c37d73a95d844b8f428d","subject":"Fix ignored EnableAodhExpirer","status":"MERGED","created":"2022-09-07 17:37:27.000000000","updated":"2022-10-14 09:31:38.000000000","submitted":"2022-10-14 09:31:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"856310-bug/1981150","meta_rev_id":"ab23eb40c1aa740a0442a46e80a1d6eda32535e8","_number":856310,"virtual_id_number":856310,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-14 09:31:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-10-14 06:17:30.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"all":[{"value":2,"date":"2022-10-12 05:25:31.000000000","_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":1,"date":"2022-10-12 10:46:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":2,"date":"2022-09-16 15:29:47.000000000","_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"all":[{"value":1,"date":"2022-10-12 05:25:31.000000000","_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-07 19:14:45.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-07 21:41:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-16 15:29:47.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"},{"updated":"2022-10-12 05:25:31.000000000","updated_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"REVIEWER"},{"updated":"2022-10-12 10:46:34.000000000","updated_by":{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},"reviewer":{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},"state":"REVIEWER"}],"messages":[{"id":"fb80953cf1f6b2e43538a27b45a22b74c34650f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-09-07 17:37:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5f406be3a5f5738cb8d76c9baebf8dec6ee99b36","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-09-07 19:14:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/b42b61cef4084bdbb6910a6c01acf396 : FAILURE in 1h 36m 14s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp_1supp-featureset039 https://review.rdoproject.org/zuul/build/6d371badc8644de4bdd05c477fdfaa9a : FAILURE in 1h 15m 57s","accounts_in_message":[],"_revision_number":1},{"id":"0824714dfacd1efd7e9d3595f468521d6ba91fcb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-07 21:41:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ccb436550e944219f892c46440b1a1e\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/4446220e35394067ac0e78229057769d : SUCCESS in 3h 47m 45s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/38bd02a0ec1e4e53b4fdab2c7f2d0a82 : SUCCESS in 51m 33s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/8caa67e0acef43f29fc407932cd6539b : SUCCESS in 1h 56m 31s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/fea1086f43a74f108e887c627aa6dea5 : SUCCESS in 2h 05m 42s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/7331e61add814fecbc999a6f49c4dfa4 : SUCCESS in 1h 26m 33s\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/111fee07031044348458f0a358584dd5 : SUCCESS in 1h 47m 12s\n- tripleo-ci-centos-9-scenario002-standalone https://zuul.opendev.org/t/openstack/build/59fc8829a07b4733a51ae4351cc48166 : SUCCESS in 2h 00m 44s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/98ac5e6816b24687b44c59da767b4ed3 : SUCCESS in 1h 59m 52s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbb4f9dfdcbe4a19ab6524642c7b1095 : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5563c42a0cda4aeda11bb6e8f84c5f8c : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/08c942618729432c9be53ca15718b35b : SUCCESS in 7m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/035bb246afdb4c58a9d24d02b76be67c : SUCCESS in 4m 29s (non-voting)\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/c5f1bd27166e4fdf984402f45ab2ea2b : SUCCESS in 3h 12m 05s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/4aec1412d97844cdaabcbbe2fa11c44a : SUCCESS in 2h 02m 59s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/61727e9e063a41c4b4fed142321e5d45 : SUCCESS in 1h 19m 34s\n- openstack-tox-tht https://zuul.opendev.org/t/openstack/build/5a97a38edc904acab402ed4ce74e90d0 : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":1},{"id":"cc59ce604fa3fdcd2ab3040b85ee4625c2c6c40e","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2022-09-16 15:29:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"08d70e3c7aa687cf3300dbe8b320bbfb401f7bf8","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2022-10-12 05:25:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ea7b30b737d83ceae46d96929ae4888c2345d814","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 05:26:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"96b5a01bf75a4cb691ae67b0e69564c4ef87ef8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 08:58:35.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d98c0720ea1046aba9c74aa276f77c3f\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/900657bf33884355b44d37535cf960fc : SUCCESS in 3h 28m 22s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/f835df5293bd473fabf70d2397ae2445 : SUCCESS in 56m 27s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/cf82ca54d7cd4ecf99bdd08eb6559bfb : SUCCESS in 2h 14m 00s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/48d275ff52d44a69a9dbbc47f296f130 : SUCCESS in 1h 30m 42s\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/c7e3944f1450401d990d13b7a3905c14 : SUCCESS in 1h 52m 23s\n- tripleo-ci-centos-9-scenario002-standalone https://zuul.opendev.org/t/openstack/build/a82d433d705f46b388ed9333e880e1c0 : SUCCESS in 1h 31m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58c2c783a954416e88def594ea3b7d38 : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ca2c6f7c1f447ecaa97bfa1667f6ab0 : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e833e64337b3421ca16e329778097738 : SUCCESS in 4m 44s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/4c511212ab27437db2cc6aa49f885a6c : SUCCESS in 1h 27m 04s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/e160a3542770495abd2f88c3213236da : FAILURE in 14m 52s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/cb23aa74650c4d67b2dd659a109cb619 : SUCCESS in 1h 50m 30s\n- openstack-tox-tht https://zuul.opendev.org/t/openstack/build/54db6651a65548abbc683d2eff6e3393 : SUCCESS in 6m 07s","accounts_in_message":[],"_revision_number":1},{"id":"7e6200cc5570bafcd8de9c322ba300f0874005a8","author":{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},"date":"2022-10-12 10:46:34.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4d0a6635c70f034cb3ddfc22a510700861bbe457","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-10-14 03:05:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"527c9ec7ed8a5762ed99752e9d44ffbdf31ae386","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-14 06:02:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9dd2af86e7a24b3ea9a274c2013e6ac6\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/3f7ee6eaf402462b913f9320bc896bb2 : SUCCESS in 2h 45m 38s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/e52f33fe17a344dba270ba3f12bc77e3 : SUCCESS in 44m 34s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/1cc6c1aed8ee44b49eda32b79b48ce21 : SUCCESS in 1h 42m 01s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/df7e49c73a984955ba90190812c09bca : SUCCESS in 1h 35m 00s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/605682a4d385457ea2558c9779677c99 : SUCCESS in 1h 15m 43s\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/599b8f3ec89b4341b9736d74d38d3242 : SUCCESS in 1h 42m 14s\n- tripleo-ci-centos-9-scenario002-standalone https://zuul.opendev.org/t/openstack/build/45ea351e6ea44b7f947d21883290fd71 : SUCCESS in 1h 25m 16s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/e9990d717a4e40fd8be0a382c2f3318f : FAILURE in 1h 30m 31s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0a9170a8aec42db965274cbacbe0fc1 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/beb7a8cbaa484608a38aeae9bec4bf57 : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/58c18925a4344da68ea7004cbba9ebc6 : SUCCESS in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/94a9ca4a906740e58a379de2e18c976f : SUCCESS in 4m 23s (non-voting)\n- tripleo-ci-centos-9-content-provider-zed https://zuul.opendev.org/t/openstack/build/d56f39e53df24bd2a5540c1a9d63fc8b : SUCCESS in 2h 01m 18s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/78420d62a383439ba1c0b297a060bb1f : SUCCESS in 1h 25m 03s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/e1aa8778889843a4b80a0e04318690ce : SUCCESS in 50m 59s\n- openstack-tox-tht https://zuul.opendev.org/t/openstack/build/29896c8e52d04a9a801a845ac3816c48 : SUCCESS in 7m 11s","accounts_in_message":[],"_revision_number":1},{"id":"4e6f9f6db05728cfb277f19410ae63e30ef1ac86","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-14 06:02:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1bf28826c578b3093bf3b80a30206af7933f5a4f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-10-14 06:17:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/ef25d1a6c63c4556bb1b3ca1d81b762e : SUCCESS in 2h 13m 53s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp_1supp-featureset039 https://review.rdoproject.org/zuul/build/1da8f3fb9b7d40a5a8754af9a397eed7 : SUCCESS in 2h 49m 06s","accounts_in_message":[],"_revision_number":1},{"id":"f61fac95c4b2969cd9d006c07c84a44ecb0b562d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-14 09:31:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5a417a595944e2bbf5068e982ea7385\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/ac493d4c3faf479d8e5cea5e15e3a0bc : SUCCESS in 3h 24m 51s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/46e46d5880044dfbb0e3f25ce2a089ef : SUCCESS in 56m 47s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/067d30e75fdb415ca74fccb442dbbc5d : SUCCESS in 2h 05m 56s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/42dbc58b809b435ba62ead4a07440ea2 : SUCCESS in 1h 25m 44s\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/25a3340973e54526950f7bc27605ccda : SUCCESS in 1h 57m 50s\n- tripleo-ci-centos-9-scenario002-standalone https://zuul.opendev.org/t/openstack/build/d7083c6eddf041688bd9d63971dbcbbc : SUCCESS in 1h 30m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eda261db5ef94d5ab2f3ae78ab4d82c1 : SUCCESS in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/213c2637b35a4381ba7a655572aa81e4 : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fb3fb0b6126b453bae79ee61513c0b01 : SUCCESS in 3m 26s\n- tripleo-ci-centos-9-content-provider-zed https://zuul.opendev.org/t/openstack/build/6540023c09e9441495be654d5308ff4f : SUCCESS in 2h 22m 07s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/27b483c7ea66472686d0df7255f85e18 : SUCCESS in 1h 06m 41s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/e7623636001a4e9cae229e33fa9216c5 : SUCCESS in 1h 38m 24s\n- openstack-tox-tht https://zuul.opendev.org/t/openstack/build/ff725b73b59f432d8170af0bfc4917ce : SUCCESS in 7m 05s","accounts_in_message":[],"_revision_number":1},{"id":"ab23eb40c1aa740a0442a46e80a1d6eda32535e8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-14 09:31:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2a2fd0c4763bcfbccc391359ea307518f24e3972","revisions":{"2a2fd0c4763bcfbccc391359ea307518f24e3972":{"kind":"REWORK","_number":1,"created":"2022-09-07 17:37:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/10/856310/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-heat-templates","ref":"refs/changes/10/856310/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-heat-templates refs/changes/10/856310/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-heat-templates refs/changes/10/856310/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-heat-templates refs/changes/10/856310/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-heat-templates refs/changes/10/856310/1"}}},"commit":{"parents":[{"commit":"7bf981f4b79cf1e739ae3f27d1a59dbf9a27c70c","subject":"Merge \"Remove leftover of the tripleo::firewall class\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-heat-templates/commit/7bf981f4b79cf1e739ae3f27d1a59dbf9a27c70c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-09-07 17:35:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-09-07 17:37:04.000000000","tz":540},"subject":"Fix ignored EnableAodhExpirer","message":"Fix ignored EnableAodhExpirer\n\nThe enable_aodh_expirer hieradata key is not implemented in\npuppet-tripleo. This replaces that wrong key by the class parameter\nso that the defined value is properly picked up by puppet.\n\nCloses-Bug: #1981150\nChange-Id: I25e0cbeed80fc77ee2d5c37d73a95d844b8f428d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-heat-templates/commit/2a2fd0c4763bcfbccc391359ea307518f24e3972"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-heat-templates/commit/2a2fd0c4763bcfbccc391359ea307518f24e3972"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}}]}],"submit_requirements":[]}
