)]}'
{"id":"openstack%2Ftripleo-image-elements~131158","triplet_id":"openstack%2Ftripleo-image-elements~master~I83abd75a159b14465eb9aee615bb50d2bc7d930c","project":"openstack/tripleo-image-elements","branch":"master","topic":"bug/1334314","hashtags":[],"change_id":"I83abd75a159b14465eb9aee615bb50d2bc7d930c","subject":"Use oac to create .erlang.cookie","status":"MERGED","created":"2014-10-27 14:16:00.000000000","updated":"2015-01-03 23:54:29.000000000","submitted":"2015-01-03 23:54:29.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"868e524fdc2c1ab0d83d76a07ca503a41f780659","_number":131158,"virtual_id_number":131158,"owner":{"_account_id":8688,"name":"Alexis Lee","email":"openstack@lxsli.co.uk","username":"lxsli"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"date":"2014-11-10 13:40:57.000000000","_account_id":8688,"name":"Alexis Lee","email":"openstack@lxsli.co.uk","username":"lxsli"},{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":12385,"name":"Motohiro/Yuanying Otsuka","email":"yuanying@fraction.jp","username":"yuanying"},{"value":2,"date":"2015-01-03 23:54:27.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":10035,"name":"greghaynes","email":"greg@greghaynes.net","username":"greghaynes"},{"value":0,"_account_id":10373,"name":"Nicholas Randon","email":"nicholas.randon@hp.com","username":"randonn"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":9369,"name":"Steve Kowalik","email":"steven@wedontsleep.org","username":"stevenk"},{"value":0,"_account_id":12767,"name":"Craige McWhirter","email":"craige@mcwhirter.io","username":"craige"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":0,"_account_id":8688,"name":"Alexis Lee","email":"openstack@lxsli.co.uk","username":"lxsli"},{"value":2,"date":"2014-12-23 22:31:24.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":1,"date":"2014-11-07 12:47:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12385,"name":"Motohiro/Yuanying Otsuka","email":"yuanying@fraction.jp","username":"yuanying"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2015-01-03 23:54:26.000000000","_account_id":10035,"name":"greghaynes","email":"greg@greghaynes.net","username":"greghaynes"},{"value":1,"date":"2014-11-07 12:47:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10373,"name":"Nicholas Randon","email":"nicholas.randon@hp.com","username":"randonn"},{"value":2,"date":"2014-12-18 12:07:04.000000000","_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":9369,"name":"Steve Kowalik","email":"steven@wedontsleep.org","username":"stevenk"},{"value":1,"date":"2014-11-07 14:18:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12767,"name":"Craige McWhirter","email":"craige@mcwhirter.io","username":"craige"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10035,"name":"greghaynes","email":"greg@greghaynes.net","username":"greghaynes"},"all":[{"value":0,"_account_id":8688,"name":"Alexis Lee","email":"openstack@lxsli.co.uk","username":"lxsli"},{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":12385,"name":"Motohiro/Yuanying Otsuka","email":"yuanying@fraction.jp","username":"yuanying"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2015-01-03 23:54:26.000000000","_account_id":10035,"name":"greghaynes","email":"greg@greghaynes.net","username":"greghaynes"},{"value":0,"_account_id":10373,"name":"Nicholas Randon","email":"nicholas.randon@hp.com","username":"randonn"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":9369,"name":"Steve Kowalik","email":"steven@wedontsleep.org","username":"stevenk"},{"value":0,"_account_id":12767,"name":"Craige McWhirter","email":"craige@mcwhirter.io","username":"craige"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":8688,"name":"Alexis Lee","email":"openstack@lxsli.co.uk","username":"lxsli"},{"_account_id":9369,"name":"Steve Kowalik","email":"steven@wedontsleep.org","username":"stevenk"},{"_account_id":10035,"name":"greghaynes","email":"greg@greghaynes.net","username":"greghaynes"},{"_account_id":10373,"name":"Nicholas Randon","email":"nicholas.randon@hp.com","username":"randonn"},{"_account_id":12385,"name":"Motohiro/Yuanying Otsuka","email":"yuanying@fraction.jp","username":"yuanying"},{"_account_id":12767,"name":"Craige McWhirter","email":"craige@mcwhirter.io","username":"craige"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2014-10-27 14:17:52.000000000","updated_by":{"_account_id":10373,"name":"Nicholas Randon","email":"nicholas.randon@hp.com","username":"randonn"},"reviewer":{"_account_id":10373,"name":"Nicholas Randon","email":"nicholas.randon@hp.com","username":"randonn"},"state":"REVIEWER"},{"updated":"2014-10-28 04:14:14.000000000","updated_by":{"_account_id":9369,"name":"Steve Kowalik","email":"steven@wedontsleep.org","username":"stevenk"},"reviewer":{"_account_id":9369,"name":"Steve Kowalik","email":"steven@wedontsleep.org","username":"stevenk"},"state":"REVIEWER"},{"updated":"2014-10-28 04:56:50.000000000","updated_by":{"_account_id":12385,"name":"Motohiro/Yuanying Otsuka","email":"yuanying@fraction.jp","username":"yuanying"},"reviewer":{"_account_id":12385,"name":"Motohiro/Yuanying Otsuka","email":"yuanying@fraction.jp","username":"yuanying"},"state":"REVIEWER"},{"updated":"2014-11-07 14:18:43.000000000","updated_by":{"_account_id":12767,"name":"Craige McWhirter","email":"craige@mcwhirter.io","username":"craige"},"reviewer":{"_account_id":12767,"name":"Craige McWhirter","email":"craige@mcwhirter.io","username":"craige"},"state":"REVIEWER"},{"updated":"2014-12-18 12:07:04.000000000","updated_by":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"reviewer":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"state":"REVIEWER"},{"updated":"2014-12-23 22:31:24.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2015-01-03 23:54:26.000000000","updated_by":{"_account_id":10035,"name":"greghaynes","email":"greg@greghaynes.net","username":"greghaynes"},"reviewer":{"_account_id":10035,"name":"greghaynes","email":"greg@greghaynes.net","username":"greghaynes"},"state":"REVIEWER"},{"updated":"2015-01-03 23:54:28.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"1e3282e7837718c9811b72a942d2b2132fffe38b","author":{"_account_id":8688,"name":"Alexis Lee","email":"openstack@lxsli.co.uk","username":"lxsli"},"date":"2014-10-27 14:16:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ce16d8322fb41237862703ee907ea28a45609b83","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-27 14:16:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop http://logs.openstack.org/58/131158/1/check/noop/aaf67c2 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"e6e164e982541bd44e8edd2f0c52b150e9e43cc3","author":{"_account_id":10373,"name":"Nicholas Randon","email":"nicholas.randon@hp.com","username":"randonn"},"date":"2014-10-27 14:17:52.000000000","message":"Patch Set 1: Code-Review+1\n\nThis is a grate change need CI system to catch-up for it work work","accounts_in_message":[],"_revision_number":1},{"id":"f0f904a13dfafd89bfd2867fb87bef6c9d5349f4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-27 14:57:02.000000000","message":"Patch Set 1:\n\nBuild failed (check-tripleo pipeline).  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n\n- check-tripleo-ironic-undercloud-precise-nonha http://logs.openstack.org/58/131158/1/check-tripleo/check-tripleo-ironic-undercloud-precise-nonha/a441ce9 : FAILURE in 11m 51s\n- check-tripleo-ironic-overcloud-f20-nonha http://logs.openstack.org/58/131158/1/check-tripleo/check-tripleo-ironic-overcloud-f20-nonha/acecb7d : FAILURE in 40m 59s\n- check-tripleo-ironic-overcloud-precise-nonha http://logs.openstack.org/58/131158/1/check-tripleo/check-tripleo-ironic-overcloud-precise-nonha/8aca3a9 : FAILURE in 37m 01s\n- check-tripleo-ironic-overcloud-precise-ha http://logs.openstack.org/58/131158/1/check-tripleo/check-tripleo-ironic-overcloud-precise-ha/a75306f : FAILURE in 37m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5ee1ea17ab4b66e3c9981687333cc8921082baf1","author":{"_account_id":9369,"name":"Steve Kowalik","email":"steven@wedontsleep.org","username":"stevenk"},"date":"2014-10-28 04:14:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ff629ec75673e8d99b62efb44f2df4be8d62fca2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-28 04:14:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- noop http://logs.openstack.org/58/131158/1/check/noop/574790f : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"029bd0b0df65a44b1f474c62666b52c510fcafd9","author":{"_account_id":12385,"name":"Motohiro/Yuanying Otsuka","email":"yuanying@fraction.jp","username":"yuanying"},"date":"2014-10-28 04:56:50.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d432f345e4ff91b70c7a464703c1dc3c4dfc411b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-28 05:00:33.000000000","message":"Patch Set 1:\n\nBuild failed (check-tripleo pipeline).  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n\n- check-tripleo-ironic-undercloud-precise-nonha http://logs.openstack.org/58/131158/1/check-tripleo/check-tripleo-ironic-undercloud-precise-nonha/c5a8210 : FAILURE in 38m 31s\n- check-tripleo-ironic-overcloud-f20-nonha http://logs.openstack.org/58/131158/1/check-tripleo/check-tripleo-ironic-overcloud-f20-nonha/34c0415 : FAILURE in 46m 16s\n- check-tripleo-ironic-overcloud-precise-nonha http://logs.openstack.org/58/131158/1/check-tripleo/check-tripleo-ironic-overcloud-precise-nonha/ee90655 : FAILURE in 37m 06s\n- check-tripleo-ironic-overcloud-precise-ha http://logs.openstack.org/58/131158/1/check-tripleo/check-tripleo-ironic-overcloud-precise-ha/2a331e8 : FAILURE in 38m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d0b7cc7f3e95f80fefb83073d85f997e145490e1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-29 11:00:22.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"9fba892b8370877b64f0226f950a33453e722ccf","author":{"_account_id":8688,"name":"Alexis Lee","email":"openstack@lxsli.co.uk","username":"lxsli"},"date":"2014-10-31 09:11:59.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":1},{"id":"ce190a922cd20d9b6a2e3db29f96773bacdfb195","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-31 09:12:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop http://logs.openstack.org/58/131158/2/check/noop/bf3b554 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"46885bf5da30ab3b265b3add359cedd7c7566b91","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-31 09:52:30.000000000","message":"Patch Set 2:\n\nBuild failed (check-tripleo pipeline).  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n\n- check-tripleo-ironic-undercloud-precise-nonha http://logs.openstack.org/58/131158/2/check-tripleo/check-tripleo-ironic-undercloud-precise-nonha/9a4a0f1 : FAILURE in 37m 25s\n- check-tripleo-ironic-overcloud-f20-nonha http://logs.openstack.org/58/131158/2/check-tripleo/check-tripleo-ironic-overcloud-f20-nonha/6a18c65 : FAILURE in 40m 27s\n- check-tripleo-ironic-overcloud-precise-nonha http://logs.openstack.org/58/131158/2/check-tripleo/check-tripleo-ironic-overcloud-precise-nonha/ab19f43 : FAILURE in 35m 29s\n- check-tripleo-ironic-overcloud-precise-ha http://logs.openstack.org/58/131158/2/check-tripleo/check-tripleo-ironic-overcloud-precise-ha/e9b7b19 : FAILURE in 35m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"39f01f171d74ef2fff633e64e2f7bd7cd39e602d","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2014-10-31 14:28:53.000000000","message":"Patch Set 2: Code-Review-1\n\nSeeing this in the seed logs from the failed CI run:\n\n 31 09:39:08 AM] [ERROR] (\"unrecognised file control key \u0027owner\u0027\", \u0027/usr/libexec/os-apply-config/templates/var/lib/rabbitmq /.erlang.cookie\n 31 09:39:08,028] (os-refresh-config) [ERROR] during configure phase. [Command \u0027[\u0027dib-run-parts\u0027, \u0027/opt/stack/os-config- refresh/configure.\nrefresh-config:during configure phase. [Command \u0027[\u0027dib-run-parts\u0027, \u0027/opt/stack/os-config-refresh/configure.d\u0027]\u0027 returned  non-zero exit st\n 31 09:39:08,029] (os-refresh-config) [ERROR] Aborting...\n\nIs this going to require a new os-apply-config release before the CI will pass?","accounts_in_message":[],"_revision_number":2},{"id":"dd37d44260d0e79774baf244a34ae42fba29cd40","author":{"_account_id":8688,"name":"Alexis Lee","email":"openstack@lxsli.co.uk","username":"lxsli"},"date":"2014-10-31 17:07:14.000000000","message":"Patch Set 2:\n\nYep","accounts_in_message":[],"_revision_number":2},{"id":"3286447d1db949afcf6782afc5b04a4f0f5af999","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-11-05 16:13:57.000000000","message":"Patch Set 2: Verified-1\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"85c5da1afe7cf3d3da04428bdedf855f7540244b","author":{"_account_id":8688,"name":"Alexis Lee","email":"openstack@lxsli.co.uk","username":"lxsli"},"date":"2014-11-07 12:47:20.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"080148eb981655c91c2c87eecef4ff297fe00394","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-11-07 12:47:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop http://logs.openstack.org/58/131158/3/check/noop/971720b : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"869b6bb9f7e22b7a68232541f7222088cf847341","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-11-07 13:29:35.000000000","message":"Patch Set 3:\n\nBuild failed (check-tripleo pipeline).  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n\n- check-tripleo-ironic-undercloud-precise-nonha http://logs.openstack.org/58/131158/3/check-tripleo/check-tripleo-ironic-undercloud-precise-nonha/a30d3ed : FAILURE in 35m 04s\n- check-tripleo-ironic-overcloud-f20-nonha http://logs.openstack.org/58/131158/3/check-tripleo/check-tripleo-ironic-overcloud-f20-nonha/adac843 : FAILURE in 42m 12s\n- check-tripleo-ironic-overcloud-precise-nonha http://logs.openstack.org/58/131158/3/check-tripleo/check-tripleo-ironic-overcloud-precise-nonha/9ee6c2e : FAILURE in 35m 35s\n- check-tripleo-ironic-overcloud-precise-ha http://logs.openstack.org/58/131158/3/check-tripleo/check-tripleo-ironic-overcloud-precise-ha/03c81b9 : FAILURE in 35m 29s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4b5a2b1435df5f0c1a80c61551fa3e94bb67e84b","author":{"_account_id":12767,"name":"Craige McWhirter","email":"craige@mcwhirter.io","username":"craige"},"date":"2014-11-07 14:18:43.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"358562ef2201632eb652f0fbea86b61a48782f5b","author":{"_account_id":8688,"name":"Alexis Lee","email":"openstack@lxsli.co.uk","username":"lxsli"},"date":"2014-11-10 13:40:57.000000000","message":"Patch Set 3:\n\nrecheck\n\nGreg released oac 0.1.26 for me so this should work now, lets see.","accounts_in_message":[],"_revision_number":3},{"id":"7af854bd2e12bf05d6608beab8384d99435b8788","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-11-10 13:40:58.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- noop http://logs.openstack.org/58/131158/3/check/noop/21ca737 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"84e77848b8315477dcd5e3d8f0b0dae511cf463b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-11-10 15:46:50.000000000","message":"Patch Set 3:\n\nBuild failed (check-tripleo pipeline).  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n\n- check-tripleo-ironic-undercloud-precise-nonha http://logs.openstack.org/58/131158/3/check-tripleo/check-tripleo-ironic-undercloud-precise-nonha/d0c86fc : SUCCESS in 52m 27s\n- check-tripleo-ironic-overcloud-f20-nonha http://logs.openstack.org/58/131158/3/check-tripleo/check-tripleo-ironic-overcloud-f20-nonha/17a424b : FAILURE in 18m 00s\n- check-tripleo-ironic-overcloud-precise-nonha http://logs.openstack.org/58/131158/3/check-tripleo/check-tripleo-ironic-overcloud-precise-nonha/18da929 : SUCCESS in 1h 21m 31s\n- check-tripleo-ironic-overcloud-precise-ha http://logs.openstack.org/58/131158/3/check-tripleo/check-tripleo-ironic-overcloud-precise-ha/7090f75 : FAILURE in 2h 05m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cfda05d7143bdf362185132d8161b2782a5c80b8","author":{"_account_id":8688,"name":"Alexis Lee","email":"openstack@lxsli.co.uk","username":"lxsli"},"date":"2014-11-10 17:49:23.000000000","message":"Patch Set 3:\n\nrecheck\n\nCI was broken due to bug 1390427","accounts_in_message":[],"_revision_number":3},{"id":"1f1740d0c2ce54948a5b21e237ec38894b37caac","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-11-10 17:49:27.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- noop http://logs.openstack.org/58/131158/3/check/noop/0ddf19c : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"929d74ccc904f9fe08b2fd7f41bb91265ca6bf00","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-11-10 19:14:10.000000000","message":"Patch Set 3:\n\nBuild succeeded (check-tripleo pipeline).\n\n- check-tripleo-ironic-undercloud-precise-nonha http://logs.openstack.org/58/131158/3/check-tripleo/check-tripleo-ironic-undercloud-precise-nonha/c61deda : SUCCESS in 56m 12s\n- check-tripleo-ironic-overcloud-f20-nonha http://logs.openstack.org/58/131158/3/check-tripleo/check-tripleo-ironic-overcloud-f20-nonha/33c0a51 : SUCCESS in 1h 24m 43s\n- check-tripleo-ironic-overcloud-precise-nonha http://logs.openstack.org/58/131158/3/check-tripleo/check-tripleo-ironic-overcloud-precise-nonha/08b3bca : SUCCESS in 1h 05m 03s\n- check-tripleo-ironic-overcloud-precise-ha http://logs.openstack.org/58/131158/3/check-tripleo/check-tripleo-ironic-overcloud-precise-ha/6c6b350 : FAILURE in 32m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a22c8944f74c6403e1287c67fb57d56043ccedb1","author":{"_account_id":10373,"name":"Nicholas Randon","email":"nicholas.randon@hp.com","username":"randonn"},"date":"2014-11-12 09:34:00.000000000","message":"Patch Set 3:\n\nthis is now working and should land as it a simplification.  James Slagle was just pointing out that oac needed updating and it has how been done ;)","accounts_in_message":[],"_revision_number":3},{"id":"a02d0a19ee4412085074d7291abbac5f236cc590","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2014-12-18 12:07:04.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"40e47fa95b672fc5052988b4bbd66db59ea46b94","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2014-12-23 22:31:24.000000000","message":"Patch Set 3: Code-Review+2\n\nrecheck\n\nMonth and a half old CI results.  Just want to make sure we didn\u0027t break this functionality in the meantime. :-)","accounts_in_message":[],"_revision_number":3},{"id":"e834ed4607f66dab4c0c79cee75d67329a2442fb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-12-23 22:31:25.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- noop http://logs.openstack.org/58/131158/3/check/noop/fd5bdd6/ : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"04dd3a1d69a8f5972ca933142596a3383a79bf00","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-12-24 01:22:08.000000000","message":"Patch Set 3:\n\nBuild failed (check-tripleo pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- check-tripleo-ironic-undercloud-precise-nonha http://logs.openstack.org/58/131158/3/check-tripleo/check-tripleo-ironic-undercloud-precise-nonha/75e6739/ : SUCCESS in 58m 44s\n- check-tripleo-ironic-overcloud-f20-nonha http://logs.openstack.org/58/131158/3/check-tripleo/check-tripleo-ironic-overcloud-f20-nonha/40056a7/ : FAILURE in 2h 50m 42s\n- check-tripleo-ironic-overcloud-precise-nonha http://logs.openstack.org/58/131158/3/check-tripleo/check-tripleo-ironic-overcloud-precise-nonha/7e9ce1b/ : SUCCESS in 1h 21m 58s\n- check-tripleo-ironic-overcloud-precise-ha http://logs.openstack.org/58/131158/3/check-tripleo/check-tripleo-ironic-overcloud-precise-ha/5ff4623/ : FAILURE in 1h 02m 16s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6283497dd6cf600fd2b44174fcf4b94ae0d94f1a","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2015-01-02 17:58:47.000000000","message":"Patch Set 3:\n\nrecheck\n\nTimed out, no logs","accounts_in_message":[],"_revision_number":3},{"id":"e8cb5ce654e1a9f5131af38068a34648454c32e9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-01-02 17:58:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- noop http://logs.openstack.org/58/131158/3/check/noop/b57aad1/ : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"a7b86679f2d74787e0851f39efbfc07f1f9fe8c5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-01-02 19:43:58.000000000","message":"Patch Set 3:\n\nBuild failed (check-tripleo pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- check-tripleo-ironic-undercloud-precise-nonha http://logs.openstack.org/58/131158/3/check-tripleo/check-tripleo-ironic-undercloud-precise-nonha/d317ae2/ : SUCCESS in 1h 10m 56s\n- check-tripleo-ironic-overcloud-f20-nonha http://logs.openstack.org/58/131158/3/check-tripleo/check-tripleo-ironic-overcloud-f20-nonha/b52eae0/ : SUCCESS in 1h 35m 40s\n- check-tripleo-ironic-overcloud-precise-nonha http://logs.openstack.org/58/131158/3/check-tripleo/check-tripleo-ironic-overcloud-precise-nonha/5e51150/ : FAILURE in 1h 04m 30s\n- check-tripleo-ironic-overcloud-precise-ha http://logs.openstack.org/58/131158/3/check-tripleo/check-tripleo-ironic-overcloud-precise-ha/46c8bc8/ : SUCCESS in 1h 45m 09s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1b00e3262854c7940eec03af8acc27f6f8c0dd5a","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2015-01-02 22:45:26.000000000","message":"Patch Set 3:\n\nrecheck bug 1380782","accounts_in_message":[],"_revision_number":3},{"id":"83b86db359d00419785ead2a12619f404881079f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-01-02 22:45:27.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- noop http://logs.openstack.org/58/131158/3/check/noop/ec996ae/ : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"1c66d82d478f90fbfc3415314dfa855766e95d88","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-01-03 01:40:35.000000000","message":"Patch Set 3:\n\nBuild failed (check-tripleo pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- check-tripleo-ironic-undercloud-precise-nonha http://logs.openstack.org/58/131158/3/check-tripleo/check-tripleo-ironic-undercloud-precise-nonha/e05f916/ : SUCCESS in 1h 03m 08s\n- check-tripleo-ironic-overcloud-f20-nonha http://logs.openstack.org/58/131158/3/check-tripleo/check-tripleo-ironic-overcloud-f20-nonha/1a612a8/ : FAILURE in 2h 55m 07s\n- check-tripleo-ironic-overcloud-precise-nonha http://logs.openstack.org/58/131158/3/check-tripleo/check-tripleo-ironic-overcloud-precise-nonha/74bd5e2/ : SUCCESS in 1h 06m 57s\n- check-tripleo-ironic-overcloud-precise-ha http://logs.openstack.org/58/131158/3/check-tripleo/check-tripleo-ironic-overcloud-precise-ha/a07909a/ : SUCCESS in 1h 37m 34s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"978d6d13dcdd64613b077fac52b0595f976d5024","author":{"_account_id":10035,"name":"greghaynes","email":"greg@greghaynes.net","username":"greghaynes"},"date":"2015-01-03 23:54:26.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nBetween the past two test runs this has passed all the tests, so +Aing.","accounts_in_message":[],"_revision_number":3},{"id":"e1bab08cd40c164aab60a2760a7b17e9243c1a6c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-01-03 23:54:27.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"239147bc7bb1e84353409f0328e7285b32348440","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-01-03 23:54:28.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- noop http://logs.openstack.org/58/131158/3/gate/noop/6fd5ef3/ : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"34f9d0068cdeeb113de1dd4064861c90838a870b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-01-03 23:54:29.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7bc694aa7255760081c55e5a694eef8ee5052519","revisions":{"d66c10c89f77b43d93178c5df22e6ebda5480f29":{"kind":"REWORK","_number":1,"created":"2014-10-27 14:16:00.000000000","uploader":{"_account_id":8688,"name":"Alexis Lee","email":"openstack@lxsli.co.uk","username":"lxsli"},"ref":"refs/changes/58/131158/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-image-elements","ref":"refs/changes/58/131158/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-image-elements refs/changes/58/131158/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-image-elements refs/changes/58/131158/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-image-elements refs/changes/58/131158/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-image-elements refs/changes/58/131158/1"}}},"commit":{"parents":[{"commit":"29b3a5a1e94a08ec22992eb9dd41a8c0bb595a58","subject":"Fix RabbitMQ element clustering start and stop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-image-elements/commit/29b3a5a1e94a08ec22992eb9dd41a8c0bb595a58"}]}],"author":{"name":"Nicholas Randon","email":"nicholas.randon@hp.com","date":"2014-08-18 18:18:31.000000000","tz":60},"committer":{"name":"Alexis Lee","email":"alexisl@hp.com","date":"2014-10-27 14:13:15.000000000","tz":0},"subject":"Use oac to create .erlang.cookie","message":"Use oac to create .erlang.cookie\n\nThis is slightly neater than using a pre-configure script, now we have\nthe tools to do it.\n\nChange-Id: I83abd75a159b14465eb9aee615bb50d2bc7d930c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-image-elements/commit/d66c10c89f77b43d93178c5df22e6ebda5480f29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-image-elements/commit/d66c10c89f77b43d93178c5df22e6ebda5480f29"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"29b3a5a1e94a08ec22992eb9dd41a8c0bb595a58","is_merged_in_target_branch":false,"change_id":"Ic758256481fdd31d10f4e4a341ae93cb372a0766","change_number":115524,"patch_set_number":38,"change_status":"MERGED"}],"branch":"refs/heads/master"},"1bd6abeec72e6abb92c7a4dae4a048dc9fd1ab24":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2014-10-31 09:11:59.000000000","uploader":{"_account_id":8688,"name":"Alexis Lee","email":"openstack@lxsli.co.uk","username":"lxsli"},"ref":"refs/changes/58/131158/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-image-elements","ref":"refs/changes/58/131158/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-image-elements refs/changes/58/131158/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-image-elements refs/changes/58/131158/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-image-elements refs/changes/58/131158/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-image-elements refs/changes/58/131158/2"}}},"commit":{"parents":[{"commit":"bf0c4650247c60605115e8cf6fa1639ca51875de","subject":"Fix RabbitMQ element clustering start and stop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-image-elements/commit/bf0c4650247c60605115e8cf6fa1639ca51875de"}]}],"author":{"name":"Nicholas Randon","email":"nicholas.randon@hp.com","date":"2014-08-18 18:18:31.000000000","tz":60},"committer":{"name":"Alexis Lee","email":"lxsli@hp.com","date":"2014-10-31 09:11:59.000000000","tz":0},"subject":"Use oac to create .erlang.cookie","message":"Use oac to create .erlang.cookie\n\nThis is slightly neater than using a pre-configure script, now we have\nthe tools to do it.\n\nChange-Id: I83abd75a159b14465eb9aee615bb50d2bc7d930c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-image-elements/commit/1bd6abeec72e6abb92c7a4dae4a048dc9fd1ab24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-image-elements/commit/1bd6abeec72e6abb92c7a4dae4a048dc9fd1ab24"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"bf0c4650247c60605115e8cf6fa1639ca51875de","is_merged_in_target_branch":false,"change_id":"Ic758256481fdd31d10f4e4a341ae93cb372a0766","change_number":115524,"patch_set_number":40,"change_status":"MERGED"}],"branch":"refs/heads/master"},"7bc694aa7255760081c55e5a694eef8ee5052519":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2014-11-07 12:47:20.000000000","uploader":{"_account_id":8688,"name":"Alexis Lee","email":"openstack@lxsli.co.uk","username":"lxsli"},"ref":"refs/changes/58/131158/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-image-elements","ref":"refs/changes/58/131158/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-image-elements refs/changes/58/131158/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-image-elements refs/changes/58/131158/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-image-elements refs/changes/58/131158/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-image-elements refs/changes/58/131158/3"}}},"commit":{"parents":[{"commit":"d99cf49c9dd86ef0280a15be8857ef341bcb4117","subject":"Merge \"Add package support to the Ansible element.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-image-elements/commit/d99cf49c9dd86ef0280a15be8857ef341bcb4117"}]}],"author":{"name":"Nicholas Randon","email":"nicholas.randon@hp.com","date":"2014-08-18 18:18:31.000000000","tz":60},"committer":{"name":"Alexis Lee","email":"lxsli@hp.com","date":"2014-11-07 12:47:20.000000000","tz":0},"subject":"Use oac to create .erlang.cookie","message":"Use oac to create .erlang.cookie\n\nThis is slightly neater than using a pre-configure script, now we have\nthe tools to do it.\n\nChange-Id: I83abd75a159b14465eb9aee615bb50d2bc7d930c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-image-elements/commit/7bc694aa7255760081c55e5a694eef8ee5052519"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-image-elements/commit/7bc694aa7255760081c55e5a694eef8ee5052519"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d99cf49c9dd86ef0280a15be8857ef341bcb4117","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
