)]}'
{"id":"openstack%2Ftripleo-image-elements~837430","triplet_id":"openstack%2Ftripleo-image-elements~master~I354aeabed6d2993ae600b86381a7b625ca7f89dc","project":"openstack/tripleo-image-elements","branch":"master","topic":"first-boot","attention_set":{},"removed_from_attention_set":{"24245":{"account":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"last_update":"2022-04-26 06:04:30.000000000","reason":"Harald Jensås replied on the change"},"7144":{"account":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"last_update":"2022-04-13 11:28:25.000000000","reason":"James Slagle replied on the change"},"4571":{"account":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"last_update":"2022-05-03 19:21:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I354aeabed6d2993ae600b86381a7b625ca7f89dc","subject":"Add element reset-bls-entries","status":"MERGED","created":"2022-04-11 22:11:36.000000000","updated":"2022-05-03 19:21:36.000000000","submitted":"2022-05-03 19:21:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"837430-first-boot","meta_rev_id":"5f2b4665defd53e1539fa2af8dde243008c109a1","_number":837430,"virtual_id_number":837430,"owner":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-03 19:21:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-26 05:24:14.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"all":[{"value":2,"date":"2022-04-26 06:04:30.000000000","_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":2,"date":"2022-05-03 16:21:34.000000000","_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"all":[{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":1,"date":"2022-05-03 16:21:34.000000000","_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-11 23:07:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-12 00:45:49.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-12 04:53:20.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2022-04-12 20:31:31.000000000","updated_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"reviewer":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"state":"REVIEWER"},{"updated":"2022-04-12 20:31:31.000000000","updated_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"reviewer":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"state":"REVIEWER"},{"updated":"2022-04-13 07:38:45.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"}],"messages":[{"id":"e07ac97d7fc0e12969fd50c0baef59831b563b6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2022-04-11 22:11:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0cdd2e9d06387bfe2f61958d7e012de5cc1d4374","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 23:07:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-buildimage-overcloud-full-centos-9 https://zuul.opendev.org/t/openstack/build/58a38fe78822481689b5c6c6d960746b : SUCCESS in 50m 39s\n- tripleo-buildimage-ironic-python-agent-centos-9 https://zuul.opendev.org/t/openstack/build/6b72850592e646a4a696479a4c1ff0fb : SUCCESS in 49m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f77956cfb9c473fa0dcb64e085203e2 : SUCCESS in 3m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c2f8cae741d4da48509424d2f13a8c7 : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d226abf5610840c69eecab97d391cbf5 : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":1},{"id":"dc1d90d6fadce73448ccc1d51e7fc67446d48a8e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-04-12 00:45:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/6553634dbd804e1d99cec8b422a75ab6 : SUCCESS in 2h 33m 07s","accounts_in_message":[],"_revision_number":1},{"id":"ea69541848be89dd92a7df92d06c97710ba989d1","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2022-04-12 04:53:20.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9b3f69a3db96c20f18966112c5700bdea731e002","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2022-04-12 20:31:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4ab1857e7d9ad6e7ebc760aa7c0b9bff09cd67bd","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2022-04-13 07:38:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"759b74f84b61eaed2e0bd1aab8389000cc2d8d3c","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2022-04-13 11:28:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7e75077110c76f60aade2c83c25544a5ee3aa620","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2022-04-18 12:37:02.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5067d71ec21f913967463dbbd1dae7f35b384226","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-18 12:37:16.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n\nWarning:\n  Change 837251 in project openstack/diskimage-builder does not share a change queue with 837430 in project openstack/tripleo-image-elements","accounts_in_message":[],"_revision_number":1},{"id":"445969b827fbfb7fc517c945d28b25d5feeed72b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-18 12:37:22.000000000","message":"Patch Set 1:\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n\nWarning:\n  Change 837251 in project openstack/diskimage-builder does not share a change queue with 837430 in project openstack/tripleo-image-elements","accounts_in_message":[],"_revision_number":1},{"id":"058cb7aec59d724abf0806652cb59627cd7a13c8","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2022-04-25 21:56:46.000000000","message":"Patch Set 1: Code-Review-2 Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9d595aaae937a9a70c9cd013d5a152d1dc11de37","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2022-04-25 21:56:56.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"5b2fdb797295b9e63283e0ad291d4080ecc3bd89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2022-04-25 22:28:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"702f43df96be4da13e1345cd910b5fe443a9e67c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 23:10:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-buildimage-overcloud-full-centos-9 https://zuul.opendev.org/t/openstack/build/d0f90aec6f3249139e121f21a0fc0986 : SUCCESS in 36m 41s\n- tripleo-buildimage-ironic-python-agent-centos-9 https://zuul.opendev.org/t/openstack/build/fd128bd2688c47a4805838ce712c23f0 : SUCCESS in 39m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/980cca9830974ca99732acd30c9e0fd9 : SUCCESS in 3m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ec9d75be48824d4da8087c002d27d12c : SUCCESS in 3m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e775f2421634493eb5145de3f10012ab : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":2},{"id":"bac87945609e06972ba628184dcae757bfd48cea","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-04-26 01:13:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/600177e9aaec4772b196b4bc78378047 : SUCCESS in 2h 42m 57s","accounts_in_message":[],"_revision_number":2},{"id":"dd706fd99aa786b19914008c8843c787d47f73b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2022-04-26 02:58:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0a0e9b2efb0e6c91b30f19a2577c9dd4cbc892d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 04:28:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-buildimage-overcloud-full-centos-9 https://zuul.opendev.org/t/openstack/build/1693fefa92504ee6b6ed3f307c3da4bd : SUCCESS in 34m 27s\n- tripleo-buildimage-ironic-python-agent-centos-9 https://zuul.opendev.org/t/openstack/build/bef395c86315465d960d166b61cea1ef : SUCCESS in 33m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61872511123240daa2e2af6dca6c461f : SUCCESS in 3m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af6143f536414f7987e2eedc24e62897 : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c3b120f840c849758a1170ab7d7973bf : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":3},{"id":"57b38a14f55403cb1f4a74ab690116a8825588db","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-04-26 05:24:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/b0ec5fff0a7c498b93829452a07bc8ba : SUCCESS in 2h 24m 20s","accounts_in_message":[],"_revision_number":3},{"id":"42ab15b193b7833591be3d0a1f549511b080bced","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2022-04-26 06:04:30.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"861777fbffd24d5516faf8eb9a150e3c0a75c24c","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2022-05-03 16:21:34.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"936ec826eb4ccd2b47c0ab5bed5b031724cdef01","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-03 16:22:37.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8da869ec7f4348ed9a8cd6cf3fc63cbda87b76c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-03 19:21:35.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-buildimage-overcloud-full-centos-9 https://zuul.opendev.org/t/openstack/build/635e100cc0694173aad8bf3b1d62939a : SUCCESS in 33m 55s\n- tripleo-buildimage-ironic-python-agent-centos-9 https://zuul.opendev.org/t/openstack/build/367511ccd34b4972aff37284fa02f439 : SUCCESS in 33m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ff7dea63ac644f7bc4fad5942eaba9b : SUCCESS in 3m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8740254aed294466bcfb9e85a752b009 : SUCCESS in 3m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d5e2a0177e3e4247823db6dfe527682e : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":3},{"id":"5f2b4665defd53e1539fa2af8dde243008c109a1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-03 19:21:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7fd9cbef64548e2f982d69f3a9ab277da63cb4f9","revisions":{"a8db49eb9c846642a05ead78ab46afa862a2b88e":{"kind":"REWORK","_number":1,"created":"2022-04-11 22:11:36.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/30/837430/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-image-elements","ref":"refs/changes/30/837430/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-image-elements refs/changes/30/837430/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-image-elements refs/changes/30/837430/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-image-elements refs/changes/30/837430/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-image-elements refs/changes/30/837430/1"}}},"commit":{"parents":[{"commit":"2156463006e0cd5278b486db0c91bc9c5ccce7cd","subject":"Merge \"remove unicode literal from code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-image-elements/commit/2156463006e0cd5278b486db0c91bc9c5ccce7cd"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2022-04-11 22:09:18.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2022-04-11 22:10:57.000000000","tz":720},"subject":"Add element reset-bls-entries","message":"Add element reset-bls-entries\n\nThis is an element which will run a script on first boot to rename\n`/boot/loader/entries` filenames to match the new value of\n`/etc/machine-id`. This is required for `grub2-mkconfig` calls to\nincorporate existing values in the boot loader entry files.\n\nIt only runs when first-boot criteria is met, such as when\n`uninitialized` is written to `/etc/machine-id`\n\nrhbz#2073855\n\nChange-Id: I354aeabed6d2993ae600b86381a7b625ca7f89dc\nDepends-On: https://review.opendev.org/c/openstack/diskimage-builder/+/837251\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-image-elements/commit/a8db49eb9c846642a05ead78ab46afa862a2b88e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-image-elements/commit/a8db49eb9c846642a05ead78ab46afa862a2b88e"}]},"branch":"refs/heads/master"},"bb9c114e617897511555725a48cc861f24f8545b":{"kind":"REWORK","_number":2,"created":"2022-04-25 22:28:58.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/30/837430/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-image-elements","ref":"refs/changes/30/837430/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-image-elements refs/changes/30/837430/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-image-elements refs/changes/30/837430/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-image-elements refs/changes/30/837430/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-image-elements refs/changes/30/837430/2"}}},"commit":{"parents":[{"commit":"2156463006e0cd5278b486db0c91bc9c5ccce7cd","subject":"Merge \"remove unicode literal from code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-image-elements/commit/2156463006e0cd5278b486db0c91bc9c5ccce7cd"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2022-04-11 22:09:18.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2022-04-25 22:28:47.000000000","tz":720},"subject":"Add element reset-bls-entries","message":"Add element reset-bls-entries\n\nThis is an element which will run a script on first boot to rename\n`/boot/loader/entries` filenames to match the new value of\n`/etc/machine-id`. This is required for `grub2-mkconfig` calls to\nincorporate existing values in the boot loader entry files.\n\nIt only runs when first-boot criteria is met, such as when\n`uninitialized` is written to `/etc/machine-id`\n\nrhbz#2073855\n\nChange-Id: I354aeabed6d2993ae600b86381a7b625ca7f89dc\nDepends-On: https://review.opendev.org/c/openstack/diskimage-builder/+/837251\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-image-elements/commit/bb9c114e617897511555725a48cc861f24f8545b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-image-elements/commit/bb9c114e617897511555725a48cc861f24f8545b"}]},"branch":"refs/heads/master"},"7fd9cbef64548e2f982d69f3a9ab277da63cb4f9":{"kind":"REWORK","_number":3,"created":"2022-04-26 02:58:23.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/30/837430/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-image-elements","ref":"refs/changes/30/837430/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-image-elements refs/changes/30/837430/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-image-elements refs/changes/30/837430/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-image-elements refs/changes/30/837430/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-image-elements refs/changes/30/837430/3"}}},"commit":{"parents":[{"commit":"2156463006e0cd5278b486db0c91bc9c5ccce7cd","subject":"Merge \"remove unicode literal from code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-image-elements/commit/2156463006e0cd5278b486db0c91bc9c5ccce7cd"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2022-04-11 22:09:18.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2022-04-26 02:57:59.000000000","tz":720},"subject":"Add element reset-bls-entries","message":"Add element reset-bls-entries\n\nThis is an element which will run a script on first boot to rename\n`/boot/loader/entries` filenames to match the new value of\n`/etc/machine-id`. This is required for `grub2-mkconfig` calls to\nincorporate existing values in the boot loader entry files.\n\nIt only runs when first-boot criteria is met, such as when\n`uninitialized` is written to `/etc/machine-id`\n\nrhbz#2073855\n\nChange-Id: I354aeabed6d2993ae600b86381a7b625ca7f89dc\nDepends-On: https://review.opendev.org/c/openstack/diskimage-builder/+/837251\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-image-elements/commit/7fd9cbef64548e2f982d69f3a9ab277da63cb4f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-image-elements/commit/7fd9cbef64548e2f982d69f3a9ab277da63cb4f9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
