)]}'
{"id":"openstack%2Ftripleo-puppet-elements~660153","triplet_id":"openstack%2Ftripleo-puppet-elements~master~I551d6ac2d064bff05862b3ffa025240657fafd36","project":"openstack/tripleo-puppet-elements","branch":"master","topic":"bug/1829769","hashtags":[],"change_id":"I551d6ac2d064bff05862b3ffa025240657fafd36","subject":"Add element to install all openstack clients","status":"MERGED","created":"2019-05-20 15:45:21.000000000","updated":"2019-06-01 15:03:02.000000000","submitted":"2019-05-31 22:38:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660153-1559342299114-11ac1cc1","meta_rev_id":"b64aa4aa681889b62423a4ae747dbde031d33977","_number":660153,"virtual_id_number":660153,"owner":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-05-31 22:38:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":1,"date":"2019-05-23 17:13:57.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-05-28 21:07:38.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-31 16:59:17.000000000","_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2019-05-31 16:41:52.000000000","_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":1,"date":"2019-05-31 17:43:02.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-22 13:04:52.000000000","updated_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2019-05-23 17:13:57.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-28 21:07:38.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-05-31 16:41:52.000000000","updated_by":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"reviewer":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"state":"REVIEWER"},{"updated":"2019-05-31 16:59:17.000000000","updated_by":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"reviewer":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"state":"REVIEWER"},{"updated":"2019-05-31 22:38:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d28cc0a8ac130850b68728350d61910c70f82998","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-05-20 15:45:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4fab2ef9f314930e80f8701f3658d345e04a918d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-05-20 15:50:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5a762ed6912016bd3fb02d83ac37b2f78a90893b","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-05-20 15:58:58.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"5df55e5f616471d568901b4ba973bb02b57bb8b4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-20 17:15:19.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 : NODE_FAILURE in 0s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 : NODE_FAILURE in 0s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/53/660153/3/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/4b3e8f9/ : SUCCESS in 1h 13m 58s","accounts_in_message":[],"_revision_number":3},{"id":"66d7d5ed1445b85846b91c6a2a7f86ba3ce25ce6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 19:20:48.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/53/660153/3/check/tripleo-ci-centos-7-containers-multinode/5d4ae6f/ : SUCCESS in 2h 20m 13s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/53/660153/3/check/tripleo-ci-centos-7-undercloud-containers/8893183/ : SUCCESS in 1h 27m 10s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/53/660153/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/7c7e10d/ : SUCCESS in 2h 30m 34s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/53/660153/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/1e884ec/ : FAILURE in 2h 20m 30s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/53/660153/3/check/tripleo-ci-centos-7-standalone/a2b15e3/ : FAILURE in 2h 23m 02s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/53/660153/3/check/tripleo-ci-fedora-28-standalone/a5d95c5/ : SUCCESS in 1h 52m 26s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/53/660153/3/check/tripleo-ci-centos-7-standalone-upgrade/565aab6/ : SUCCESS in 2h 23m 40s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/53/660153/3/check/tripleo-ci-centos-7-standalone-upgrade-stein/9e48311/ : SUCCESS in 2h 15m 25s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/53/660153/3/check/tripleo-buildimage-overcloud-full-centos-7/03b5800/ : SUCCESS in 21m 28s\n- openstack-tox-pep8 http://logs.openstack.org/53/660153/3/check/openstack-tox-pep8/4f36e6f/ : SUCCESS in 3m 45s\n- openstack-tox-py27 http://logs.openstack.org/53/660153/3/check/openstack-tox-py27/9ae05e0/ : FAILURE in 3m 07s\n- openstack-tox-py35 http://logs.openstack.org/53/660153/3/check/openstack-tox-py35/3f7eef6/ : SUCCESS in 11m 34s\n- openstack-tox-py36 http://logs.openstack.org/53/660153/3/check/openstack-tox-py36/258238c/ : SUCCESS in 4m 18s\n- openstack-tox-py37 http://logs.openstack.org/53/660153/3/check/openstack-tox-py37/8cf9960/ : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":3},{"id":"242f99d57b78812254641bdb2a7ff7ba04f8d9db","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-05-20 19:27:40.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"57e578e61f316b24263c18642bd229d4ab0777b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 22:02:12.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/53/660153/4/check/tripleo-ci-centos-7-containers-multinode/65bba6b/ : SUCCESS in 2h 01m 44s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/53/660153/4/check/tripleo-ci-centos-7-undercloud-containers/556387e/ : SUCCESS in 1h 42m 05s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/53/660153/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/d37f5bd/ : SUCCESS in 2h 06m 56s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/53/660153/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/c65e4b6/ : SUCCESS in 2h 19m 20s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/53/660153/4/check/tripleo-ci-centos-7-standalone/a08908b/ : SUCCESS in 1h 55m 04s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/53/660153/4/check/tripleo-ci-fedora-28-standalone/5e3dba9/ : SUCCESS in 1h 40m 38s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/53/660153/4/check/tripleo-ci-centos-7-standalone-upgrade/a698176/ : SUCCESS in 2h 09m 58s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/53/660153/4/check/tripleo-ci-centos-7-standalone-upgrade-stein/c94564b/ : SUCCESS in 2h 02m 58s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/53/660153/4/check/tripleo-buildimage-overcloud-full-centos-7/dea564e/ : SUCCESS in 21m 17s\n- openstack-tox-pep8 http://logs.openstack.org/53/660153/4/check/openstack-tox-pep8/908baab/ : SUCCESS in 4m 36s\n- openstack-tox-py27 http://logs.openstack.org/53/660153/4/check/openstack-tox-py27/e22f02b/ : SUCCESS in 3m 55s\n- openstack-tox-py35 http://logs.openstack.org/53/660153/4/check/openstack-tox-py35/7e612e6/ : SUCCESS in 4m 02s\n- openstack-tox-py36 http://logs.openstack.org/53/660153/4/check/openstack-tox-py36/46aa142/ : SUCCESS in 4m 27s\n- openstack-tox-py37 http://logs.openstack.org/53/660153/4/check/openstack-tox-py37/49e0120/ : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":4},{"id":"346bcdff15c534070e31b3f8b705b55a40221575","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-20 22:08:44.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/53/660153/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/8f5434b/ : SUCCESS in 2h 29m 15s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/53/660153/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/ec393e6/ : SUCCESS in 2h 39m 06s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/53/660153/4/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/dfe738e/ : SUCCESS in 1h 11m 42s","accounts_in_message":[],"_revision_number":4},{"id":"11773b332f7f9db8cc9966abf4e3ec03698ba033","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-05-21 15:46:56.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"0057d656d4beddb10100fdc740446304e5fb2575","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2019-05-22 13:04:52.000000000","message":"Patch Set 4: Code-Review-1\n\n(4 comments)\n\nSmall typo, other than that LGTM, thanks Alex","accounts_in_message":[],"_revision_number":4},{"id":"4958c0a084636a852bf8dbed4865ee9c6ef1e19e","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-05-22 15:25:16.000000000","message":"Patch Set 4: Workflow-1\n\nat least i\u0027m consistent with my typos","accounts_in_message":[],"_revision_number":4},{"id":"4d7e59f4bc969b8654d2c0fae24cde5b8a68a718","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-05-22 15:27:30.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"75a60707c8f3620d155bf4d558ac85e262511f7b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-22 18:08:23.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/53/660153/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/ae21614/ : FAILURE in 2h 39m 08s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/53/660153/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/6c7fddc/ : FAILURE in 2h 28m 37s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/53/660153/5/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/86e52e4/ : SUCCESS in 1h 29m 00s","accounts_in_message":[],"_revision_number":5},{"id":"c272e91a868379a5946b5d78bacdd44a8362d477","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 01:58:31.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/53/660153/5/check/tripleo-ci-centos-7-containers-multinode/a854b0e/ : SUCCESS in 2h 06m 07s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/53/660153/5/check/tripleo-ci-centos-7-undercloud-containers/7036915/ : SUCCESS in 1h 34m 22s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/53/660153/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/b2ec0ac/ : SUCCESS in 2h 09m 53s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/53/660153/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/36951c5/ : SUCCESS in 2h 28m 00s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/53/660153/5/check/tripleo-ci-centos-7-standalone/fc0dc0b/ : SUCCESS in 2h 03m 44s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/53/660153/5/check/tripleo-ci-fedora-28-standalone/c7eab96/ : SUCCESS in 1h 42m 59s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/53/660153/5/check/tripleo-ci-centos-7-standalone-upgrade/b152bd0/ : FAILURE in 1h 27m 39s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/53/660153/5/check/tripleo-ci-centos-7-standalone-upgrade-stein/e3b4e0f/ : SUCCESS in 1h 54m 31s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/53/660153/5/check/tripleo-buildimage-overcloud-full-centos-7/b9e0c29/ : SUCCESS in 27m 17s\n- openstack-tox-pep8 http://logs.openstack.org/53/660153/5/check/openstack-tox-pep8/0845a02/ : SUCCESS in 4m 17s\n- openstack-tox-py27 http://logs.openstack.org/53/660153/5/check/openstack-tox-py27/c88296e/ : SUCCESS in 4m 25s\n- openstack-tox-py35 http://logs.openstack.org/53/660153/5/check/openstack-tox-py35/313e5ed/ : SUCCESS in 5m 46s\n- openstack-tox-py36 http://logs.openstack.org/53/660153/5/check/openstack-tox-py36/12d8f5a/ : SUCCESS in 4m 12s\n- openstack-tox-py37 http://logs.openstack.org/53/660153/5/check/openstack-tox-py37/0e8e754/ : SUCCESS in 6m 15s","accounts_in_message":[],"_revision_number":5},{"id":"67021fe578289105e7e26fbdd111f3c572a46da9","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-05-23 14:15:17.000000000","message":"Patch Set 5:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":5},{"id":"1d282035dee501cba5dfac2180532d855a9587f7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-23 17:13:57.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/53/660153/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/ef2e186/ : SUCCESS in 2h 31m 31s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/53/660153/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/2c4e83e/ : SUCCESS in 2h 33m 24s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/53/660153/5/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/a693e69/ : SUCCESS in 1h 17m 52s","accounts_in_message":[],"_revision_number":5},{"id":"29fcff54363d7d02008b6f01649cd2c1cc4be4d1","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-05-28 21:07:38.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"afb2260eb8c3c430a4c4d791236ae59336f04f0b","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-05-31 16:59:17.000000000","message":"Patch Set 5: Code-Review+2\n\nlgtm","accounts_in_message":[],"_revision_number":5},{"id":"93260834daac54e4867cf028dc4f741895393d8b","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-05-31 17:43:02.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"a9aac90af96f885a753e41e50d21a56bb80d7104","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 17:43:23.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"6146c3e17b109cdedda0f468763d1c4dad3f6d28","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 22:38:18.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/53/660153/5/gate/tripleo-ci-centos-7-containers-multinode/0b7460f/ : SUCCESS in 2h 15m 13s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/53/660153/5/gate/tripleo-ci-centos-7-standalone/3ff2d52/ : SUCCESS in 1h 45m 42s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/53/660153/5/gate/tripleo-ci-fedora-28-standalone/03bebef/ : SUCCESS in 1h 46m 16s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/53/660153/5/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/d22908c/ : SUCCESS in 2h 09m 51s\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/53/660153/5/gate/tripleo-ci-centos-7-standalone-upgrade-stein/80badef/ : SUCCESS in 2h 24m 40s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/53/660153/5/gate/tripleo-buildimage-overcloud-full-centos-7/4a99f0a/ : SUCCESS in 22m 35s\n- openstack-tox-pep8 http://logs.openstack.org/53/660153/5/gate/openstack-tox-pep8/dec68a8/ : SUCCESS in 3m 47s\n- openstack-tox-py27 http://logs.openstack.org/53/660153/5/gate/openstack-tox-py27/c39542b/ : SUCCESS in 4m 12s\n- openstack-tox-py35 http://logs.openstack.org/53/660153/5/gate/openstack-tox-py35/5ac3c00/ : SUCCESS in 3m 53s\n- openstack-tox-py36 http://logs.openstack.org/53/660153/5/gate/openstack-tox-py36/7abda72/ : SUCCESS in 3m 49s\n- openstack-tox-py37 http://logs.openstack.org/53/660153/5/gate/openstack-tox-py37/5f536b4/ : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":5},{"id":"292cfe89287d2fd7f65af1d7c07144fceb4f28f6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 22:38:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"82fba19b357385cf8203bb4ae0eabc4c3b9dc8c2","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-01 15:03:02.000000000","message":"Patch Set 5: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 60ee624c610f6069490af69a9a9c699c9e54131c","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"6a3c2b3e61923c82a3b95b6a1ec65f154021e42a","revisions":{"06af007d254d0b03d4e0f6c475d4d634b115e684":{"kind":"REWORK","_number":1,"created":"2019-05-20 15:45:21.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/53/660153/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-puppet-elements","ref":"refs/changes/53/660153/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/53/660153/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/53/660153/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/53/660153/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/53/660153/1"}}},"commit":{"parents":[{"commit":"95c93003998ed93c4f23939e0fda2e3415cf338f","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-puppet-elements/commit/95c93003998ed93c4f23939e0fda2e3415cf338f"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-05-20 15:43:12.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-05-20 15:45:17.000000000","tz":-360},"subject":"Add element to install all openstack clients","message":"Add element to install all openstack clients\n\nLet\u0027s add an element to ensure all the python openstack clients are\nproperly installed on the controllers. These were previously being\npulled in because we installed the actual services. Since we moved to\ncontainers, these clients are no longer installed by default.\n\nChange-Id: I551d6ac2d064bff05862b3ffa025240657fafd36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-puppet-elements/commit/06af007d254d0b03d4e0f6c475d4d634b115e684"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-puppet-elements/commit/06af007d254d0b03d4e0f6c475d4d634b115e684"}]},"branch":"refs/heads/master"},"dc6a49fc9cd7db9cd3cca896ed7f4c52969b7396":{"kind":"REWORK","_number":2,"created":"2019-05-20 15:50:16.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/53/660153/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-puppet-elements","ref":"refs/changes/53/660153/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/53/660153/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/53/660153/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/53/660153/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/53/660153/2"}}},"commit":{"parents":[{"commit":"95c93003998ed93c4f23939e0fda2e3415cf338f","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-puppet-elements/commit/95c93003998ed93c4f23939e0fda2e3415cf338f"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-05-20 15:43:12.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-05-20 15:49:02.000000000","tz":-360},"subject":"Add element to install all openstack clients","message":"Add element to install all openstack clients\n\nLet\u0027s add an element to ensure all the python openstack clients are\nproperly installed on the controllers. These were previously being\npulled in because we installed the actual services. Since we moved to\ncontainers, these clients are no longer installed by default.\n\nChange-Id: I551d6ac2d064bff05862b3ffa025240657fafd36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-puppet-elements/commit/dc6a49fc9cd7db9cd3cca896ed7f4c52969b7396"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-puppet-elements/commit/dc6a49fc9cd7db9cd3cca896ed7f4c52969b7396"}]},"branch":"refs/heads/master"},"d697918c834460c545fd3faa8729f30769310ddd":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-05-20 15:58:58.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/53/660153/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-puppet-elements","ref":"refs/changes/53/660153/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/53/660153/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/53/660153/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/53/660153/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/53/660153/3"}}},"commit":{"parents":[{"commit":"95c93003998ed93c4f23939e0fda2e3415cf338f","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-puppet-elements/commit/95c93003998ed93c4f23939e0fda2e3415cf338f"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-05-20 15:43:12.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-05-20 15:58:50.000000000","tz":-360},"subject":"Add element to install all openstack clients","message":"Add element to install all openstack clients\n\nLet\u0027s add an element to ensure all the python openstack clients are\nproperly installed on the controllers. These were previously being\npulled in because we installed the actual services. Since we moved to\ncontainers, these clients are no longer installed by default.\n\nCloses-Bug: #1829769\nChange-Id: I551d6ac2d064bff05862b3ffa025240657fafd36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-puppet-elements/commit/d697918c834460c545fd3faa8729f30769310ddd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-puppet-elements/commit/d697918c834460c545fd3faa8729f30769310ddd"}]},"branch":"refs/heads/master"},"b3decf1b4d3b5d7f30428dae0e3cac6b53815d94":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-05-20 19:27:40.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/53/660153/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-puppet-elements","ref":"refs/changes/53/660153/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/53/660153/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/53/660153/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/53/660153/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/53/660153/4"}}},"commit":{"parents":[{"commit":"98368f114f9a96c7dbb1284c82042362ea26fc50","subject":"Add upper constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-puppet-elements/commit/98368f114f9a96c7dbb1284c82042362ea26fc50"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-05-20 15:43:12.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-05-20 19:27:40.000000000","tz":0},"subject":"Add element to install all openstack clients","message":"Add element to install all openstack clients\n\nLet\u0027s add an element to ensure all the python openstack clients are\nproperly installed on the controllers. These were previously being\npulled in because we installed the actual services. Since we moved to\ncontainers, these clients are no longer installed by default.\n\nCloses-Bug: #1829769\nChange-Id: I551d6ac2d064bff05862b3ffa025240657fafd36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-puppet-elements/commit/b3decf1b4d3b5d7f30428dae0e3cac6b53815d94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-puppet-elements/commit/b3decf1b4d3b5d7f30428dae0e3cac6b53815d94"}]},"branch":"refs/heads/master"},"6a3c2b3e61923c82a3b95b6a1ec65f154021e42a":{"kind":"REWORK","_number":5,"created":"2019-05-22 15:27:30.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/53/660153/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-puppet-elements","ref":"refs/changes/53/660153/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/53/660153/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/53/660153/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/53/660153/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/53/660153/5"}}},"commit":{"parents":[{"commit":"98368f114f9a96c7dbb1284c82042362ea26fc50","subject":"Add upper constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-puppet-elements/commit/98368f114f9a96c7dbb1284c82042362ea26fc50"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-05-20 15:43:12.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-05-22 15:27:16.000000000","tz":-360},"subject":"Add element to install all openstack clients","message":"Add element to install all openstack clients\n\nLet\u0027s add an element to ensure all the python openstack clients are\nproperly installed on the controllers. These were previously being\npulled in because we installed the actual services. Since we moved to\ncontainers, these clients are no longer installed by default.\n\nCloses-Bug: #1829769\nChange-Id: I551d6ac2d064bff05862b3ffa025240657fafd36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-puppet-elements/commit/6a3c2b3e61923c82a3b95b6a1ec65f154021e42a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-puppet-elements/commit/6a3c2b3e61923c82a3b95b6a1ec65f154021e42a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
