)]}'
{"id":"openstack%2Ftripleo-puppet-elements~662609","triplet_id":"openstack%2Ftripleo-puppet-elements~stable%2Fstein~I551d6ac2d064bff05862b3ffa025240657fafd36","project":"openstack/tripleo-puppet-elements","branch":"stable/stein","topic":"bug/1829769-stable/stein","hashtags":[],"change_id":"I551d6ac2d064bff05862b3ffa025240657fafd36","subject":"Add element to install all openstack clients","status":"MERGED","created":"2019-06-01 15:03:02.000000000","updated":"2019-06-08 05:47:44.000000000","submitted":"2019-06-08 05:47:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"662609-1559972864249-95369c51","meta_rev_id":"42a49a20f6c2e573e876f7393154b8cab748b34a","_number":662609,"virtual_id_number":662609,"owner":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-06-08 05:47:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":1,"date":"2019-06-07 15:56:51.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-06-07 19:39:25.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-06-07 19:39:25.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-03 05:36:15.000000000","updated_by":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"reviewer":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"state":"REVIEWER"},{"updated":"2019-06-07 15:56:51.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-07 19:39:25.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-06-08 05:47:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0cc3cb96dd7e5c5c378f94f0cd2fc1121caa2362","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-01 15:03:02.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"34e33a31ec31b6d1cc824601a3a479b31c2f1be5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-01 17:34:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/09/662609/1/check/tripleo-ci-centos-7-containers-multinode/2019a15/ : SUCCESS in 2h 00m 39s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/09/662609/1/check/tripleo-ci-centos-7-undercloud-containers/f71b5e5/ : SUCCESS in 1h 31m 39s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/09/662609/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/ba394ed/ : SUCCESS in 2h 15m 38s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/09/662609/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/6de2810/ : SUCCESS in 2h 18m 30s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/09/662609/1/check/tripleo-ci-centos-7-standalone/6832717/ : SUCCESS in 1h 31m 10s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/09/662609/1/check/tripleo-ci-fedora-28-standalone/7104959/ : SUCCESS in 1h 42m 32s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/09/662609/1/check/tripleo-buildimage-overcloud-full-centos-7/2e891fe/ : SUCCESS in 21m 53s\n- openstack-tox-pep8 http://logs.openstack.org/09/662609/1/check/openstack-tox-pep8/eec1c1c/ : SUCCESS in 3m 52s\n- openstack-tox-py27 http://logs.openstack.org/09/662609/1/check/openstack-tox-py27/4b5cedf/ : FAILURE in 3m 22s\n- openstack-tox-py35 http://logs.openstack.org/09/662609/1/check/openstack-tox-py35/c7a5169/ : SUCCESS in 4m 30s\n- openstack-tox-py36 http://logs.openstack.org/09/662609/1/check/openstack-tox-py36/30307fd/ : SUCCESS in 3m 56s\n- openstack-tox-py37 http://logs.openstack.org/09/662609/1/check/openstack-tox-py37/4c50254/ : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":1},{"id":"7685b8b7a83c2cbf4ab944501553fab267bbd88f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-01 17:43:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/09/662609/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/38a7a4d/ : SUCCESS in 2h 17m 23s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/09/662609/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/4801360/ : FAILURE in 2h 38m 39s","accounts_in_message":[],"_revision_number":1},{"id":"6d2178dc7fa3431f126633813457be039f084207","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-06-03 05:36:15.000000000","message":"Patch Set 1:\n\ndo we need to cap the sphinx version? 2019-06-01 15:10:35.964381 | ubuntu-bionic | ERROR: Sphinx requires Python \u0027\u003e\u003d3.5\u0027 but the running Python is 2.7.15","accounts_in_message":[],"_revision_number":1},{"id":"adae078b630a63a5b1fc484ecfd8060d768ea390","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-03 13:34:57.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"df8cef3f8c650a8792a06dc2dfe9bac820055b99","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 15:55:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/09/662609/2/check/tripleo-ci-centos-7-containers-multinode/a1d20df/ : SUCCESS in 1h 53m 58s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/09/662609/2/check/tripleo-ci-centos-7-undercloud-containers/14f16aa/ : SUCCESS in 1h 42m 55s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/09/662609/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/1e34a1f/ : FAILURE in 20m 08s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/09/662609/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/cd8cb2a/ : SUCCESS in 2h 09m 36s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/09/662609/2/check/tripleo-ci-centos-7-standalone/5f9301f/ : FAILURE in 2h 04m 05s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/09/662609/2/check/tripleo-ci-fedora-28-standalone/6479178/ : FAILURE in 2h 07m 25s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/09/662609/2/check/tripleo-buildimage-overcloud-full-centos-7/c308beb/ : FAILURE in 10m 50s\n- openstack-tox-pep8 http://logs.openstack.org/09/662609/2/check/openstack-tox-pep8/f3b0e7d/ : SUCCESS in 4m 15s\n- openstack-tox-py27 http://logs.openstack.org/09/662609/2/check/openstack-tox-py27/8551f0f/ : SUCCESS in 4m 02s\n- openstack-tox-py35 http://logs.openstack.org/09/662609/2/check/openstack-tox-py35/803808c/ : SUCCESS in 5m 31s\n- openstack-tox-py36 http://logs.openstack.org/09/662609/2/check/openstack-tox-py36/3360219/ : SUCCESS in 4m 47s\n- openstack-tox-py37 http://logs.openstack.org/09/662609/2/check/openstack-tox-py37/29d3b53/ : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":2},{"id":"1a39713d8cf68be611949fd49104000ae0643271","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-03 16:15:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/09/662609/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/0d8888c/ : SUCCESS in 2h 39m 18s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/09/662609/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/0197cd5/ : SUCCESS in 2h 23m 34s","accounts_in_message":[],"_revision_number":2},{"id":"33c63b4e9bb7007573162417ae2d29500aa133be","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-03 19:45:13.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"3f7d743a21144b03b09c0445f3de1699a2504654","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-03 22:26:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/09/662609/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/1985e78/ : SUCCESS in 2h 24m 19s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/09/662609/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/378bc6b/ : SUCCESS in 2h 39m 07s","accounts_in_message":[],"_revision_number":2},{"id":"e3a2bfdac3c338b2b2a0519e05bdf6ccc2317ab1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 00:47:26.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/09/662609/2/check/tripleo-ci-centos-7-containers-multinode/67ffba0/ : SUCCESS in 1h 59m 11s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/09/662609/2/check/tripleo-ci-centos-7-undercloud-containers/c8e4058/ : SUCCESS in 1h 36m 16s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/09/662609/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/4b01f6c/ : SUCCESS in 2h 08m 41s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/09/662609/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/5885a73/ : SUCCESS in 2h 19m 01s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/09/662609/2/check/tripleo-ci-centos-7-standalone/9caf9a3/ : SUCCESS in 1h 43m 11s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/09/662609/2/check/tripleo-ci-fedora-28-standalone/876d88e/ : FAILURE in 1h 53m 22s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/09/662609/2/check/tripleo-buildimage-overcloud-full-centos-7/f661889/ : SUCCESS in 25m 30s\n- openstack-tox-pep8 http://logs.openstack.org/09/662609/2/check/openstack-tox-pep8/246ca6a/ : SUCCESS in 5m 37s\n- openstack-tox-py27 http://logs.openstack.org/09/662609/2/check/openstack-tox-py27/14352c3/ : SUCCESS in 4m 16s\n- openstack-tox-py35 http://logs.openstack.org/09/662609/2/check/openstack-tox-py35/fc42572/ : SUCCESS in 6m 22s\n- openstack-tox-py36 http://logs.openstack.org/09/662609/2/check/openstack-tox-py36/00ffa04/ : SUCCESS in 4m 03s\n- openstack-tox-py37 http://logs.openstack.org/09/662609/2/check/openstack-tox-py37/88886d7/ : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":2},{"id":"cc5220a91a23ca259417a41773a4823b0f7bbb6b","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-04 22:52:47.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"e038fb1ed9162d2e4b19afaf0dd89a7fb6cd6238","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-05 01:18:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/09/662609/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/609db7f/ : FAILURE in 1h 43m 42s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/09/662609/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/3428456/ : SUCCESS in 2h 24m 03s","accounts_in_message":[],"_revision_number":2},{"id":"31002cf3b4904f0dbe1ac26c46c5583df4c2098a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 03:19:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/09/662609/2/check/tripleo-ci-centos-7-containers-multinode/4a463cc/ : SUCCESS in 1h 49m 21s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/09/662609/2/check/tripleo-ci-centos-7-undercloud-containers/2cac40f/ : SUCCESS in 1h 30m 34s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/09/662609/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/4a9137b/ : SUCCESS in 2h 17m 27s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/09/662609/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/f1b23a2/ : SUCCESS in 2h 33m 56s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/09/662609/2/check/tripleo-ci-centos-7-standalone/4e32a21/ : SUCCESS in 1h 47m 22s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/09/662609/2/check/tripleo-ci-fedora-28-standalone/bb56b09/ : SUCCESS in 1h 37m 24s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/09/662609/2/check/tripleo-buildimage-overcloud-full-centos-7/042a72f/ : SUCCESS in 21m 40s\n- openstack-tox-pep8 http://logs.openstack.org/09/662609/2/check/openstack-tox-pep8/6d52ad8/ : SUCCESS in 4m 07s\n- openstack-tox-py27 http://logs.openstack.org/09/662609/2/check/openstack-tox-py27/3ca700e/ : SUCCESS in 4m 16s\n- openstack-tox-py35 http://logs.openstack.org/09/662609/2/check/openstack-tox-py35/1b1f0bf/ : SUCCESS in 4m 39s\n- openstack-tox-py36 http://logs.openstack.org/09/662609/2/check/openstack-tox-py36/cdd0f76/ : SUCCESS in 4m 26s\n- openstack-tox-py37 http://logs.openstack.org/09/662609/2/check/openstack-tox-py37/84294c2/ : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":2},{"id":"90e8de89423dde645c16268140674e9f41ddf360","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-07 13:21:24.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"505338971bf8df9f16f5a05e6e6de573886d8f7b","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-07 13:21:52.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3054589e09872c995f2f90844b8a9d690ebf9d16","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 13:22:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5d3998c93655392c57f2b3e356a5baeeedc308fb","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-07 13:28:17.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"5958913ad43b7b19587e845ef2a8bca0d6d8e387","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-07 15:56:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/09/662609/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/e24843d/ : SUCCESS in 2h 18m 10s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/09/662609/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/bdafaba/ : SUCCESS in 2h 27m 04s","accounts_in_message":[],"_revision_number":3},{"id":"9478e144276136b06323f81187a4fb2b303f5eef","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-07 19:39:25.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"1d48e9357e2a427b5438d12b6e60b84dc8946a5d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 03:13:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/09/662609/3/check/tripleo-ci-centos-7-containers-multinode/912f70d/ : SUCCESS in 2h 00m 32s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/09/662609/3/check/tripleo-ci-centos-7-undercloud-containers/96c70df/ : SUCCESS in 1h 33m 59s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/09/662609/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/01e42e7/ : SUCCESS in 2h 19m 37s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/09/662609/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/1c4e580/ : SUCCESS in 2h 25m 10s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/09/662609/3/check/tripleo-ci-centos-7-standalone/81316af/ : SUCCESS in 1h 54m 43s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/09/662609/3/check/tripleo-ci-fedora-28-standalone/0836922/ : SUCCESS in 2h 09m 23s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/09/662609/3/check/tripleo-buildimage-overcloud-full-centos-7/d185140/ : SUCCESS in 27m 43s\n- openstack-tox-pep8 http://logs.openstack.org/09/662609/3/check/openstack-tox-pep8/eb5afdb/ : SUCCESS in 6m 25s\n- openstack-tox-py27 http://logs.openstack.org/09/662609/3/check/openstack-tox-py27/643f166/ : SUCCESS in 6m 27s\n- openstack-tox-py35 http://logs.openstack.org/09/662609/3/check/openstack-tox-py35/0b48230/ : SUCCESS in 6m 50s\n- openstack-tox-py36 http://logs.openstack.org/09/662609/3/check/openstack-tox-py36/21d801f/ : SUCCESS in 4m 11s\n- openstack-tox-py37 http://logs.openstack.org/09/662609/3/check/openstack-tox-py37/6976ec2/ : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":3},{"id":"78470ba75b85b6eee8a6a60d56533a01547752f6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 03:13:36.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"526c5d9ac9e7cb0a2bcd7ee5855eddf38833dffe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 05:47:43.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/09/662609/3/gate/tripleo-ci-centos-7-containers-multinode/2e9c72d/ : SUCCESS in 2h 11m 19s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/09/662609/3/gate/tripleo-ci-centos-7-standalone/5419430/ : SUCCESS in 1h 39m 25s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/09/662609/3/gate/tripleo-ci-fedora-28-standalone/5c10ad5/ : SUCCESS in 1h 55m 35s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/09/662609/3/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/9f1293e/ : SUCCESS in 2h 17m 50s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/09/662609/3/gate/tripleo-buildimage-overcloud-full-centos-7/4dfc2be/ : SUCCESS in 25m 03s\n- openstack-tox-pep8 http://logs.openstack.org/09/662609/3/gate/openstack-tox-pep8/6589f52/ : SUCCESS in 4m 29s\n- openstack-tox-py27 http://logs.openstack.org/09/662609/3/gate/openstack-tox-py27/2896298/ : SUCCESS in 4m 22s\n- openstack-tox-py35 http://logs.openstack.org/09/662609/3/gate/openstack-tox-py35/27f4acc/ : SUCCESS in 4m 25s\n- openstack-tox-py36 http://logs.openstack.org/09/662609/3/gate/openstack-tox-py36/e4618e4/ : SUCCESS in 4m 29s\n- openstack-tox-py37 http://logs.openstack.org/09/662609/3/gate/openstack-tox-py37/4cc4c10/ : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":3},{"id":"b6b4438367a771a2d29422669429a1eccad5cba2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 05:47:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"79c0c760e89721899aba0de47721ae92a7ddeec0","revisions":{"60ee624c610f6069490af69a9a9c699c9e54131c":{"kind":"REWORK","_number":1,"created":"2019-06-01 15:03:02.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/09/662609/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-puppet-elements","ref":"refs/changes/09/662609/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/09/662609/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/09/662609/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/09/662609/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/09/662609/1"}}},"commit":{"parents":[{"commit":"a3593011054825ac7587d1a23ed15caef0f65301","subject":"Merge \"Update .gitreview for stable/stein\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-puppet-elements/commit/a3593011054825ac7587d1a23ed15caef0f65301"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-05-20 15:43:12.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-06-01 15:03:02.000000000","tz":0},"subject":"Add element to install all openstack clients","message":"Add element to install all openstack clients\n\nLet\u0027s add an element to ensure all the python openstack clients are\nproperly installed on the controllers. These were previously being\npulled in because we installed the actual services. Since we moved to\ncontainers, these clients are no longer installed by default.\n\nCloses-Bug: #1829769\nChange-Id: I551d6ac2d064bff05862b3ffa025240657fafd36\n(cherry picked from commit 6a3c2b3e61923c82a3b95b6a1ec65f154021e42a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-puppet-elements/commit/60ee624c610f6069490af69a9a9c699c9e54131c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-puppet-elements/commit/60ee624c610f6069490af69a9a9c699c9e54131c"}]},"branch":"refs/heads/stable/stein"},"3d2007388c8a19f1f8681cca2a92235946f4dd97":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-06-03 13:34:57.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/09/662609/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-puppet-elements","ref":"refs/changes/09/662609/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/09/662609/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/09/662609/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/09/662609/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/09/662609/2"}}},"commit":{"parents":[{"commit":"f18333d5c070bfe2a5fab43964a711f56f65bc81","subject":"Add upper constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-puppet-elements/commit/f18333d5c070bfe2a5fab43964a711f56f65bc81"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-05-20 15:43:12.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-06-03 13:34:57.000000000","tz":0},"subject":"Add element to install all openstack clients","message":"Add element to install all openstack clients\n\nLet\u0027s add an element to ensure all the python openstack clients are\nproperly installed on the controllers. These were previously being\npulled in because we installed the actual services. Since we moved to\ncontainers, these clients are no longer installed by default.\n\nCloses-Bug: #1829769\nChange-Id: I551d6ac2d064bff05862b3ffa025240657fafd36\n(cherry picked from commit 6a3c2b3e61923c82a3b95b6a1ec65f154021e42a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-puppet-elements/commit/3d2007388c8a19f1f8681cca2a92235946f4dd97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-puppet-elements/commit/3d2007388c8a19f1f8681cca2a92235946f4dd97"}]},"branch":"refs/heads/stable/stein"},"79c0c760e89721899aba0de47721ae92a7ddeec0":{"kind":"REWORK","_number":3,"created":"2019-06-07 13:28:17.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/09/662609/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-puppet-elements","ref":"refs/changes/09/662609/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/09/662609/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/09/662609/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/09/662609/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/09/662609/3"}}},"commit":{"parents":[{"commit":"f18333d5c070bfe2a5fab43964a711f56f65bc81","subject":"Add upper constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-puppet-elements/commit/f18333d5c070bfe2a5fab43964a711f56f65bc81"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-05-20 15:43:12.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-06-07 13:28:16.000000000","tz":0},"subject":"Add element to install all openstack clients","message":"Add element to install all openstack clients\n\nLet\u0027s add an element to ensure all the python openstack clients are\nproperly installed on the controllers. These were previously being\npulled in because we installed the actual services. Since we moved to\ncontainers, these clients are no longer installed by default.\n\nNOTE: includes fix for pankoclient typo\n\nCloses-Bug: #1829769\nChange-Id: I551d6ac2d064bff05862b3ffa025240657fafd36\n(cherry picked from commit 6a3c2b3e61923c82a3b95b6a1ec65f154021e42a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-puppet-elements/commit/79c0c760e89721899aba0de47721ae92a7ddeec0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-puppet-elements/commit/79c0c760e89721899aba0de47721ae92a7ddeec0"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
