)]}'
{"id":"openstack%2Ftripleo-puppet-elements~797767","triplet_id":"openstack%2Ftripleo-puppet-elements~stable%2Fwallaby~I6fb2243b64ff457a5c338d7e510acfdbf65f5585","project":"openstack/tripleo-puppet-elements","branch":"stable/wallaby","topic":"retire_panko-stable/wallaby","hashtags":[],"change_id":"I6fb2243b64ff457a5c338d7e510acfdbf65f5585","subject":"Remove leftover of pankoclient","status":"MERGED","created":"2021-06-24 09:59:37.000000000","updated":"2021-07-05 08:57:41.000000000","submitted":"2021-07-05 08:09:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"797767-retire_panko-stable/wallaby","cherry_pick_of_change":794974,"cherry_pick_of_patch_set":1,"meta_rev_id":"9f1f95a70a2421b77f9c32cccaa03d27ed949c0b","_number":797767,"virtual_id_number":797767,"owner":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8161,"name":"Lon Hohberger","email":"lhh@redhat.com","username":"lon"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-05 08:09:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2021-06-28 09:08:35.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"all":[{"value":0,"_account_id":8161,"name":"Lon Hohberger","email":"lhh@redhat.com","username":"lon"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2021-06-28 21:50:48.000000000","_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":1,"date":"2021-07-05 05:36:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":2,"date":"2021-07-01 14:40:42.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":8161,"name":"Lon Hohberger","email":"lhh@redhat.com","username":"lon"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":1,"date":"2021-07-01 14:40:42.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"}],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":8161,"name":"Lon Hohberger","email":"lhh@redhat.com","username":"lon"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-24 09:59:37.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2021-06-24 10:46:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-24 11:18:07.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-24 11:29:02.000000000","updated_by":{"_account_id":8161,"name":"Lon Hohberger","email":"lhh@redhat.com","username":"lon"},"reviewer":{"_account_id":8161,"name":"Lon Hohberger","email":"lhh@redhat.com","username":"lon"},"state":"REVIEWER"},{"updated":"2021-06-28 05:59:19.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2021-06-28 05:59:19.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2021-06-28 05:59:19.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-06-28 21:50:48.000000000","updated_by":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"reviewer":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"state":"REVIEWER"}],"messages":[{"id":"067f2d2edbb02df2a1ee65e45bf2316e73f3880a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-06-24 09:59:37.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"65ffbd8dab45a62d49ed1f5e7bd76ae2f9546f4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 10:46:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/975238c611da44c78068efc2f008da6a : FAILURE in 34m 11s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/5bcdfd0fb74c4e6dbf89dc6bc282f58e : SUCCESS in 40m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed816a9d8d374e9d9cd67494db80786a : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/daf26cd6d7ad487ca6c533b4982a98ef : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bbc7d557fce840e49ddf7ac976843c5d : SUCCESS in 3m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/06b38f2db7bf49bfabd2229731d1fe4f : SUCCESS in 4m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e8daf64c6bc6fe33a99b3a5c42a6a74703ce9433","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-06-24 11:18:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/611682b648c44333a1a80c111dfc37c1 : FAILURE in 1h 05m 54s","accounts_in_message":[],"_revision_number":1},{"id":"31e5d158b6408f8e0552dec09c31582c3931c477","author":{"_account_id":8161,"name":"Lon Hohberger","email":"lhh@redhat.com","username":"lon"},"date":"2021-06-24 11:29:02.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5d19d664894255c11c32bc8cd34258402f451f8e","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-06-24 12:14:32.000000000","message":"Patch Set 1:\n\nThe valid failure here is\n```\n2021-06-24 11:05:59.335 | Package gdisk-1.0.3-6.el8.x86_64 is already installed.\n\n2021-06-24 11:05:59.355 | No match for argument: pankoclient_package\n\n2021-06-24 11:05:59.377 | Error: Unable to find a match: pankoclient_package\n```\nApparently, this upstream change needs more patches if we want to backport it.","accounts_in_message":[],"_revision_number":1},{"id":"3d8659995d9e25ed7390f26c1fe204fcb7dc26c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-06-24 12:17:00.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"36150df663bb1ea112cd1eefc544ab955a6f3e31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 13:31:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/3c9daab5501442c2b037057cfd496a2a : SUCCESS in 1h 11m 07s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/279fd788855640b68432c64c85e88e52 : SUCCESS in 34m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a24ec2803fcf45f0bde0c23d6fd0fe33 : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1add1e16541a4123b7034c24ecac31b1 : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/010f5712680443dbaefd8dbaccaccb24 : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bad16af89c3349d09ce29101056c5217 : SUCCESS in 5m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7730a2784bac1e7da17e292864e7af49b4316876","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-06-24 14:11:57.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/6c8c8874291146f599654546dc8205e2 : LOST in 7m 11s","accounts_in_message":[],"_revision_number":2},{"id":"51e3f2b9ff85c23292c52422284b08ccad396bcb","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-06-28 05:59:19.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"7e3ee72e01f0c3c0f1030a175790491c5edc4037","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-28 06:21:24.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"939483a6627ef9242a9ba2fcd8ee9839538504de","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-28 06:21:56.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"70a54c49b4087923028911a6432cc0997e0f2a49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 07:50:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/1f565861f2664eb19391b454e15b4375 : SUCCESS in 1h 04m 37s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/fa3430624b1a415097a89d6e64ba3850 : SUCCESS in 36m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c81e494a33f41d2b5cc0483552af625 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a8c870594534d319f060e30d03b4452 : SUCCESS in 7m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/447d79df842c45fc93512ee59a97c9a4 : SUCCESS in 6m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0763eddf95b34766b052069a82f72878 : SUCCESS in 6m 35s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f94b4473e3de7668e2e6a1d41a645b1c11d09471","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-28 08:23:29.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"631030c530ea7b540841a9162aa986c6250ad7ba","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-06-28 09:08:35.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/5607f1e1a2f244d49864bf442e477354 : FAILURE in 2h 45m 47s","accounts_in_message":[],"_revision_number":3},{"id":"328c96789cf382aa657ec47c634ba184380e5b8d","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-06-28 21:50:48.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"51aca04ae3f1fc0b5022e6d629c70b811a0c4d5f","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-07-01 14:40:42.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"1ffac069ce06409341d0f25a988843021089826a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 17:32:47.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ff11bcc3d19c30f89b5da3e15622f46456ad5349","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 21:37:30.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/d8ed6ff2253a49c5bc9b19c62b87a8c6 : SUCCESS in 55m 47s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/eb53b77751364a04a8521e0be17aae82 : SUCCESS in 41m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8c901883ba74f5aba36b6289b235ad4 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c6e440a824a54d6d8bf880f7463a6b03 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc59a134b79745e49d79b5dd0941cee7 : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":3},{"id":"fba20bde63c8a5e560bc5323c8b8c1d22d33279b","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-07-05 05:36:48.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)\n\nThank you very much.\n\nUnfortunately, the gate did not submit this patch for whatever reason.","accounts_in_message":[],"_revision_number":3},{"id":"c3ca89b8b7de8921b362fbac5a6acc2eef90942e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-05 05:45:14.000000000","message":"Patch Set 3:\n\nLet me try recheck. It seems the dependent change was merged after this change received +2 from zuul","accounts_in_message":[],"_revision_number":3},{"id":"070f8b906494bdcfaef62fb95e238cdd66dc061d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-05 05:45:22.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"44bd688506f34f21c2871f927e8d61bc25fe0038","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-05 05:49:53.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cffed62bec274eeedf3d9ea149db85349d2e338b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-05 07:04:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/3c5a66d46cc044599291c8f1ae2b8f54 : SUCCESS in 59m 54s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/f729e29e2582447ea4d396c2eca959b3 : SUCCESS in 31m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba22008a67774e3ea89ed6005fd73053 : SUCCESS in 3m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1612a43eb91f4883aed731ca7c3605de : SUCCESS in 3m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8996070af4d4ac0923f40290d67876f : SUCCESS in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/24405505a7a143a394cfc967a1980440 : SUCCESS in 3m 40s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"63b8b0da046f37d844707cedf41ebdcdccea619a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-05 07:04:59.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"90dadb8e3e2dad605ccf4e741ee7e28d60a1e8be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-05 08:09:44.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/634c2fdb0e9649eb9f53bf955f6a96cd : SUCCESS in 1h 03m 27s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/00ed6a7a4c5a4ae9ae9a02dcc151a6fd : SUCCESS in 37m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24c8c69486dd4a0dac04cb4de8395112 : SUCCESS in 3m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/087934b87cd94e36a8fb07e25bf941e9 : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49a41e37b4484d45842fe3aa3faf1cec : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":3},{"id":"9d0615a059911b608be79be11132b26f3add9879","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-05 08:09:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"9f1f95a70a2421b77f9c32cccaa03d27ed949c0b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-07-05 08:57:41.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/c3b94a64022d4962a028f69649b2aff3 : FAILURE in 2h 22m 53s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ffb4b4cfbf844f4136a89dcb6bc1c1b61cf91e58","revisions":{"45b65afe9da8bfbfad792d76db05286ddc1cd20b":{"kind":"REWORK","_number":1,"created":"2021-06-24 09:59:37.000000000","uploader":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"ref":"refs/changes/67/797767/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-puppet-elements","ref":"refs/changes/67/797767/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/67/797767/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/67/797767/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/67/797767/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/67/797767/1"}}},"commit":{"parents":[{"commit":"c53c2bd2df7865035fd0a53eb94a7e3f3c11cc82","subject":"Update TOX_CONSTRAINTS_FILE for stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-puppet-elements/commit/c53c2bd2df7865035fd0a53eb94a7e3f3c11cc82"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-06 07:49:27.000000000","tz":540},"committer":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-06-24 09:59:37.000000000","tz":0},"subject":"Remove leftover of pankoclient","message":"Remove leftover of pankoclient\n\nThis change removes reference to pankoclient which was left when\nremoving pankoclient[1].\n\n[1] 0d0067a51a13f1e084255851328a872fbc10df3c\n\nChange-Id: I6fb2243b64ff457a5c338d7e510acfdbf65f5585\n(cherry picked from commit 27864d5e2935fe46a3bfed9eea2155cff4c463bc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-puppet-elements/commit/45b65afe9da8bfbfad792d76db05286ddc1cd20b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-puppet-elements/commit/45b65afe9da8bfbfad792d76db05286ddc1cd20b"}]},"branch":"refs/heads/stable/wallaby"},"2cc788b03b12df8f7d5a9d90ea14a085187e7035":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-06-24 12:17:00.000000000","uploader":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"ref":"refs/changes/67/797767/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-puppet-elements","ref":"refs/changes/67/797767/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/67/797767/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/67/797767/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/67/797767/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/67/797767/2"}}},"commit":{"parents":[{"commit":"c53c2bd2df7865035fd0a53eb94a7e3f3c11cc82","subject":"Update TOX_CONSTRAINTS_FILE for stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-puppet-elements/commit/c53c2bd2df7865035fd0a53eb94a7e3f3c11cc82"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-06 07:49:27.000000000","tz":540},"committer":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-06-24 12:16:48.000000000","tz":0},"subject":"Remove leftover of pankoclient","message":"Remove leftover of pankoclient\n\nThis change removes reference to pankoclient which was left when\nremoving pankoclient[1].\n\n[1] 0d0067a51a13f1e084255851328a872fbc10df3c\n\nDepends-on: https://review.opendev.org/c/openstack/tripleo-puppet-elements/+/797766\n\nChange-Id: I6fb2243b64ff457a5c338d7e510acfdbf65f5585\n(cherry picked from commit 27864d5e2935fe46a3bfed9eea2155cff4c463bc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-puppet-elements/commit/2cc788b03b12df8f7d5a9d90ea14a085187e7035"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-puppet-elements/commit/2cc788b03b12df8f7d5a9d90ea14a085187e7035"}]},"branch":"refs/heads/stable/wallaby","description":"Edit commit message"},"ffb4b4cfbf844f4136a89dcb6bc1c1b61cf91e58":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":3,"created":"2021-06-28 06:21:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/67/797767/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-puppet-elements","ref":"refs/changes/67/797767/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/67/797767/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/67/797767/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/67/797767/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/67/797767/3"}}},"commit":{"parents":[{"commit":"665911bd87273dd4e96ad2dc36eddd0c04b935f0","subject":"Remove pankoclient","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-puppet-elements/commit/665911bd87273dd4e96ad2dc36eddd0c04b935f0"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-06 07:49:27.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-28 06:21:24.000000000","tz":0},"subject":"Remove leftover of pankoclient","message":"Remove leftover of pankoclient\n\nThis change removes reference to pankoclient which was left when\nremoving pankoclient[1].\n\n[1] 0d0067a51a13f1e084255851328a872fbc10df3c\n\nChange-Id: I6fb2243b64ff457a5c338d7e510acfdbf65f5585\n(cherry picked from commit 27864d5e2935fe46a3bfed9eea2155cff4c463bc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-puppet-elements/commit/ffb4b4cfbf844f4136a89dcb6bc1c1b61cf91e58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-puppet-elements/commit/ffb4b4cfbf844f4136a89dcb6bc1c1b61cf91e58"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
