)]}'
{"id":"openstack%2Ftripleo-puppet-elements~825259","triplet_id":"openstack%2Ftripleo-puppet-elements~master~Ica02244aedf4f4192f8aedfab82e41fdbca47e77","project":"openstack/tripleo-puppet-elements","branch":"master","topic":"cs9","attention_set":{},"removed_from_attention_set":{"24245":{"account":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"last_update":"2022-01-21 14:33:16.000000000","reason":"Harald Jensås replied on the change"},"9976":{"account":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"last_update":"2022-01-21 16:18:22.000000000","reason":"Change was submitted"},"12393":{"account":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"last_update":"2022-01-21 16:18:22.000000000","reason":"Change was submitted"},"14985":{"account":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"last_update":"2022-01-20 15:06:35.000000000","reason":"Alex Schultz replied on the change"},"4571":{"account":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"last_update":"2022-01-21 00:29:37.000000000","reason":"Steve Baker replied on the change"},"29775":{"account":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"last_update":"2022-01-21 16:18:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ica02244aedf4f4192f8aedfab82e41fdbca47e77","subject":"Update overcloud-base element pkg-map","status":"MERGED","created":"2022-01-19 07:00:22.000000000","updated":"2022-01-21 16:18:22.000000000","submitted":"2022-01-21 16:18:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"825259-cs9","meta_rev_id":"81361e1cf728cf070943fece2b28a1f86164f759","_number":825259,"virtual_id_number":825259,"owner":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-21 16:18:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"all":[{"value":2,"date":"2022-01-21 14:33:16.000000000","_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-01-21 00:43:49.000000000","_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"all":[{"value":1,"date":"2022-01-21 14:33:16.000000000","_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"}],"reviewers":{"REVIEWER":[{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-19 07:02:25.000000000","updated_by":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"REVIEWER"},{"updated":"2022-01-19 07:02:25.000000000","updated_by":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2022-01-19 07:02:25.000000000","updated_by":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-01-19 07:02:25.000000000","updated_by":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"reviewer":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"state":"REVIEWER"},{"updated":"2022-01-19 07:02:25.000000000","updated_by":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"reviewer":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"state":"REVIEWER"},{"updated":"2022-01-19 07:28:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-19 08:24:24.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2d78ede1a7ca3ee9d93206ab80a3d7d57b1e2e85","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2022-01-19 07:00:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e6c033de40293213ddaf4f0c7506522cc312c2cf","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2022-01-19 07:01:42.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":1},{"id":"3780c8d7076e4881cab5dcd30b4776c9eeca10ed","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2022-01-19 07:24:36.000000000","message":"Topic set to rhel9_image_build","accounts_in_message":[],"_revision_number":1},{"id":"ab3929aada363839a20c59a13f30630fe5385473","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2022-01-19 07:26:14.000000000","message":"Topic rhel9_image_build removed","accounts_in_message":[],"_revision_number":1},{"id":"3933bc04c3ef32eba0cbb8b404a8458634e08abd","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2022-01-19 07:26:17.000000000","message":"Topic set to cs9","accounts_in_message":[],"_revision_number":1},{"id":"c2292821c46cead1e7c007b84de45ab197a6425f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 07:28:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/436475f33ae141988d4c656d4856490b : FAILURE in 23m 49s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/cbc228796b9a4afbaa78f8d2329f3066 : SUCCESS in 26m 24s\n- tripleo-buildimage-overcloud-full-centos-9 https://zuul.opendev.org/t/openstack/build/01e698b2afc24e2ebb8ba1cebdbc9829 : FAILURE in 19m 03s\n- tripleo-buildimage-ironic-python-agent-centos-9 https://zuul.opendev.org/t/openstack/build/1889cb16d1f94dd58500983bf403cbf3 : SUCCESS in 21m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b73638ff0bde441784ccbcc3892dd525 : SUCCESS in 2m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fdf90d2a9f38438d86a7da2c3bc1ac13 : SUCCESS in 2m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a5cac25868604504afa663aa402bd6bd : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":1},{"id":"50125f23323fa052204c51f6a42aefcb15db0707","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-01-19 08:24:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/698e1f5554b04b85a9f77ab8b6c99d96 : FAILURE in 1h 05m 20s","accounts_in_message":[],"_revision_number":1},{"id":"6ce27fbd0c9365acb4487d9d3e1728fef6c31347","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2022-01-19 10:11:28.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"5c62a5df1214b24feafb998f3276aad5bd6a6fc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 11:14:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/95bc7a79d12f450ba6c20b630b8fd3b7 : SUCCESS in 49m 11s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/124f185f59d1403c84f964ebd1ff1b95 : SUCCESS in 28m 34s\n- tripleo-buildimage-overcloud-full-centos-9 https://zuul.opendev.org/t/openstack/build/85de3601709b4e12bab0c084b0f6b033 : SUCCESS in 30m 00s\n- tripleo-buildimage-ironic-python-agent-centos-9 https://zuul.opendev.org/t/openstack/build/e6af4575f3d24be99727bc751be326ca : SUCCESS in 24m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbc665b882eb4a8b933e012f33df71bf : SUCCESS in 2m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ce2a3bdf031e4c42b64b00c00d381e90 : SUCCESS in 3m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/44dc89b6988d48e3a6204f96935ca934 : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":2},{"id":"65b87bae622bff3609268a86105073e52bcde730","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-01-19 13:18:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/b27ee7f466b2485480ebca391988c380 : SUCCESS in 3h 00m 28s","accounts_in_message":[],"_revision_number":2},{"id":"9aebdba02d5fd2bf26181c4e45191c63124120ca","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-01-19 14:47:51.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3bf99b928cce666154abcaa88aee89e1fcd4d665","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2022-01-19 19:47:16.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7abe0d290050bebf293474f34f56a6e275a24205","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2022-01-20 08:29:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f4fa146f0e54a17f6530ec5afaaea0e33240be93","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2022-01-20 08:32:47.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"142178c990fb1e60a45d9cf9d9fc1481e65a4544","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 09:16:08.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/99d4d5ebc58744cb889d0d3f79754f1c : RETRY_LIMIT in 5m 07s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/1a53d0f521464ea8bdd0516cff29ddf6 : RETRY_LIMIT in 5m 58s\n- tripleo-buildimage-overcloud-full-centos-9 https://zuul.opendev.org/t/openstack/build/6c552fd3297d4744b49cad513868e488 : SUCCESS in 41m 36s\n- tripleo-buildimage-ironic-python-agent-centos-9 https://zuul.opendev.org/t/openstack/build/7cf0b41a0f5a48cfbe980a86854884d1 : SUCCESS in 36m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1db944255a374315b0dbef7f62887278 : SUCCESS in 3m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed23d7ea56d74766b383422211fba1f6 : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e5a9a2fe7c647faaf0f3b5afae69c5e : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":3},{"id":"29efd415db00bb786f834e740fd1fc7dbd88dec5","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2022-01-20 15:04:31.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"caa12d1a3fc55c582f857f99ef1e653c0cfe3f12","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-01-20 15:06:35.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"afc21953b12a65a18d47775f818fd2d09b603889","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 16:26:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/fd21484d151d43faac45c089a23311a6 : SUCCESS in 1h 12m 26s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/fb806086dd33484085225759888ba792 : SUCCESS in 42m 13s\n- tripleo-buildimage-overcloud-full-centos-9 https://zuul.opendev.org/t/openstack/build/f83a570404ce4e2e8259e8cb152b09c5 : SUCCESS in 43m 53s\n- tripleo-buildimage-ironic-python-agent-centos-9 https://zuul.opendev.org/t/openstack/build/cc385dd860b148bb839a9dc964e78fd7 : SUCCESS in 37m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c12b8cecf9c4d9ba187827250c56011 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0d2bb5423970411f9f1501f6600d2be5 : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4b4a7ee9758640c8a1ce7b6f75257e03 : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":3},{"id":"fccbd09af4ae204cfe38b3c5c78209d9e9463bc2","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2022-01-20 20:36:26.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6ae8c1d62308aa97522bc2dddd068bacd7ea4995","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2022-01-20 23:52:20.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"23d56fbb999dfea63dd8b030a13b766d1d3819cb","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2022-01-20 23:58:53.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"184372b76f5f60913982dc57186f4676a9d02fea","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2022-01-21 00:10:29.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3c3a943f60aebde9c717347df47ec60155e14c9c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2022-01-21 00:25:16.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"483c58d1a84d51eeecfdb7c02526ae87b2ec68e1","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2022-01-21 00:29:37.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"86afe2948f49730e617d0fc584700aa0972d0c91","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2022-01-21 00:42:56.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"6c49879b4fb2895eeb22b656e79b294295cd20e9","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2022-01-21 00:43:49.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"77762a840d5801593d3776504a77118a79b96e05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 01:21:48.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/86f67696a9354dbf9f3e2eb3c6c5f278 : RETRY_LIMIT in 8m 17s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/c94645e36d52475f98b9bd20f08e0cb2 : RETRY_LIMIT in 6m 55s\n- tripleo-buildimage-overcloud-full-centos-9 https://zuul.opendev.org/t/openstack/build/aa8131889d14436e80f3f79ff327cb34 : SUCCESS in 35m 38s\n- tripleo-buildimage-ironic-python-agent-centos-9 https://zuul.opendev.org/t/openstack/build/13df8b62c52b4ebcbae70f2bd659bbdf : SUCCESS in 30m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f357bf445bf4e91bb076906bca8b87c : SUCCESS in 3m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6817632248ad4bfe9acdf4290cbc9e5f : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a457c44ad00b487abbaf164fa67017d4 : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":5},{"id":"85d9d1829edfeed97296657723b4dba0e6a9f8f2","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2022-01-21 11:34:22.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6c8e23d8a7acddb1b21bd18a242511f292caee29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 12:33:46.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/aaeec09b932843efbf2fea983953862d : SUCCESS in 56m 13s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/2893043e5ffb4420b241d24f3c93951b : SUCCESS in 30m 32s\n- tripleo-buildimage-overcloud-full-centos-9 https://zuul.opendev.org/t/openstack/build/4f09d5b1102040a498d7042eefbff12d : SUCCESS in 35m 19s\n- tripleo-buildimage-ironic-python-agent-centos-9 https://zuul.opendev.org/t/openstack/build/4e81de5548cc4166bab0ea582339b49a : SUCCESS in 24m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51500b7fbcad4d98811f986ba072e044 : SUCCESS in 2m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/71f46349413d4118a86337376c2ea49e : SUCCESS in 3m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e4c4a7101ff4090bf0059e3592b0dca : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":5},{"id":"4f85f3d394098f2fb2e9e10c04709397d9fbd13a","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2022-01-21 14:33:16.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"ac1a5596a0930e7b617ff99928093e6db36c64bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 14:34:15.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"5c2719f469e50847161a515a3e685a8cbb74f949","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 16:18:21.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/3ee2d73a86124bf4adba60947dceb408 : SUCCESS in 57m 43s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/dfe2c3c94cf84a2d85574199f2479d31 : SUCCESS in 41m 54s\n- tripleo-buildimage-overcloud-full-centos-9 https://zuul.opendev.org/t/openstack/build/ffd093b453c341f591f49d17212c51a7 : SUCCESS in 33m 05s\n- tripleo-buildimage-ironic-python-agent-centos-9 https://zuul.opendev.org/t/openstack/build/e27ba079bd4b490382d2c0bb5830149a : SUCCESS in 29m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf99b5039b6445cf8d71bafffb096a79 : SUCCESS in 3m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/68b72db5ad724e4f8f2baecc22cd8901 : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2fac5a7f17294680af1ec57b1766f473 : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":5},{"id":"81361e1cf728cf070943fece2b28a1f86164f759","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 16:18:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"fce5d94e021a948efc420d8f41a1ac3a97359e86","revisions":{"450a169571bb11aa893e9623124eadd09c1a57b3":{"kind":"REWORK","_number":1,"created":"2022-01-19 07:00:22.000000000","uploader":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"ref":"refs/changes/59/825259/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-puppet-elements","ref":"refs/changes/59/825259/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/59/825259/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/59/825259/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/59/825259/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/59/825259/1"}}},"commit":{"parents":[{"commit":"a60be92dd64b00e6b798e4589aee17ce3d6b5b88","subject":"Revert \"Add remove-machine-id element\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-puppet-elements/commit/a60be92dd64b00e6b798e4589aee17ce3d6b5b88"}]}],"author":{"name":"Sandeep Yadav","email":"sandyada@redhat.com","date":"2022-01-19 06:57:44.000000000","tz":330},"committer":{"name":"Sandeep Yadav","email":"sandyada@redhat.com","date":"2022-01-19 07:00:17.000000000","tz":330},"subject":"Add config for RHEL9 in overcloud-base element","message":"Add config for RHEL9 in overcloud-base element\n\nSimiliar change was proposed for Centos9[1]\n\n[1] https://review.opendev.org/c/openstack/tripleo-puppet-elements/+/806019\n\nChange-Id: Ica02244aedf4f4192f8aedfab82e41fdbca47e77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-puppet-elements/commit/450a169571bb11aa893e9623124eadd09c1a57b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-puppet-elements/commit/450a169571bb11aa893e9623124eadd09c1a57b3"}]},"branch":"refs/heads/master"},"74ee7b163ea820122b268f156c96e0cb33841b05":{"kind":"REWORK","_number":2,"created":"2022-01-19 10:11:28.000000000","uploader":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"ref":"refs/changes/59/825259/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-puppet-elements","ref":"refs/changes/59/825259/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/59/825259/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/59/825259/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/59/825259/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/59/825259/2"}}},"commit":{"parents":[{"commit":"a60be92dd64b00e6b798e4589aee17ce3d6b5b88","subject":"Revert \"Add remove-machine-id element\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-puppet-elements/commit/a60be92dd64b00e6b798e4589aee17ce3d6b5b88"}]}],"author":{"name":"Sandeep Yadav","email":"sandyada@redhat.com","date":"2022-01-19 06:57:44.000000000","tz":330},"committer":{"name":"Sandeep Yadav","email":"sandyada@redhat.com","date":"2022-01-19 10:11:28.000000000","tz":0},"subject":"Add config for RHEL9 in overcloud-base element","message":"Add config for RHEL9 in overcloud-base element\n\nSimiliar change was proposed for Centos9[1]\n\n[1] https://review.opendev.org/c/openstack/tripleo-puppet-elements/+/806019\n\nChange-Id: Ica02244aedf4f4192f8aedfab82e41fdbca47e77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-puppet-elements/commit/74ee7b163ea820122b268f156c96e0cb33841b05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-puppet-elements/commit/74ee7b163ea820122b268f156c96e0cb33841b05"}]},"branch":"refs/heads/master"},"05b8095da9b93e5f02f793d558c59a55b4997b03":{"kind":"REWORK","_number":3,"created":"2022-01-20 08:29:01.000000000","uploader":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"ref":"refs/changes/59/825259/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-puppet-elements","ref":"refs/changes/59/825259/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/59/825259/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/59/825259/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/59/825259/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/59/825259/3"}}},"commit":{"parents":[{"commit":"a60be92dd64b00e6b798e4589aee17ce3d6b5b88","subject":"Revert \"Add remove-machine-id element\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-puppet-elements/commit/a60be92dd64b00e6b798e4589aee17ce3d6b5b88"}]}],"author":{"name":"Sandeep Yadav","email":"sandyada@redhat.com","date":"2022-01-19 06:57:44.000000000","tz":330},"committer":{"name":"Sandeep Yadav","email":"sandyada@redhat.com","date":"2022-01-20 08:25:02.000000000","tz":330},"subject":"Update overcloud-base element pkg-map","message":"Update overcloud-base element pkg-map\n\nUpdating defaults to reflect latest release and removing\nthe \"release\" section as we don\u0027t need to handle exceptions\nseparetely now.\n\nChange-Id: Ica02244aedf4f4192f8aedfab82e41fdbca47e77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-puppet-elements/commit/05b8095da9b93e5f02f793d558c59a55b4997b03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-puppet-elements/commit/05b8095da9b93e5f02f793d558c59a55b4997b03"}]},"branch":"refs/heads/master"},"a9b1934bbf0c0d5c5a20db0b120595ad37a6ffcd":{"kind":"REWORK","_number":4,"created":"2022-01-21 00:25:16.000000000","uploader":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"ref":"refs/changes/59/825259/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-puppet-elements","ref":"refs/changes/59/825259/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/59/825259/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/59/825259/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/59/825259/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/59/825259/4"}}},"commit":{"parents":[{"commit":"a60be92dd64b00e6b798e4589aee17ce3d6b5b88","subject":"Revert \"Add remove-machine-id element\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-puppet-elements/commit/a60be92dd64b00e6b798e4589aee17ce3d6b5b88"}]}],"author":{"name":"Sandeep Yadav","email":"sandyada@redhat.com","date":"2022-01-19 06:57:44.000000000","tz":330},"committer":{"name":"Sandeep Yadav","email":"sandyada@redhat.com","date":"2022-01-21 00:22:04.000000000","tz":330},"subject":"Update overcloud-base element pkg-map","message":"Update overcloud-base element pkg-map\n\nWith this patch, we are removing yum_plugin_priorities_package and\npython_ipaddr from overcloud-base element as a part of refactoring\nto only support centos/rhel 8/9.\n\nThis repo have stable branches, so train builds should be using\nstable/train of tripleo-puppet-elements.\n\nChange-Id: Ica02244aedf4f4192f8aedfab82e41fdbca47e77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-puppet-elements/commit/a9b1934bbf0c0d5c5a20db0b120595ad37a6ffcd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-puppet-elements/commit/a9b1934bbf0c0d5c5a20db0b120595ad37a6ffcd"}]},"branch":"refs/heads/master"},"fce5d94e021a948efc420d8f41a1ac3a97359e86":{"kind":"REWORK","_number":5,"created":"2022-01-21 00:42:56.000000000","uploader":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"ref":"refs/changes/59/825259/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-puppet-elements","ref":"refs/changes/59/825259/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/59/825259/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/59/825259/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/59/825259/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-puppet-elements refs/changes/59/825259/5"}}},"commit":{"parents":[{"commit":"a60be92dd64b00e6b798e4589aee17ce3d6b5b88","subject":"Revert \"Add remove-machine-id element\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-puppet-elements/commit/a60be92dd64b00e6b798e4589aee17ce3d6b5b88"}]}],"author":{"name":"Sandeep Yadav","email":"sandyada@redhat.com","date":"2022-01-19 06:57:44.000000000","tz":330},"committer":{"name":"Sandeep Yadav","email":"sandyada@redhat.com","date":"2022-01-21 00:42:38.000000000","tz":330},"subject":"Update overcloud-base element pkg-map","message":"Update overcloud-base element pkg-map\n\nWith this patch, we are removing yum_plugin_priorities_package and\npython_ipaddr from overcloud-base element as a part of refactoring\nto only support centos/rhel 8/9.\n\nThis repo have stable branches, so train builds should be using\nstable/train of tripleo-puppet-elements.\n\nChange-Id: Ica02244aedf4f4192f8aedfab82e41fdbca47e77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-puppet-elements/commit/fce5d94e021a948efc420d8f41a1ac3a97359e86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-puppet-elements/commit/fce5d94e021a948efc420d8f41a1ac3a97359e86"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
