)]}'
{"id":"openstack%2Ftripleo-quickstart-extras~587745","triplet_id":"openstack%2Ftripleo-quickstart-extras~master~I1399f163eaf711e1603838086a3cd1c231553598","project":"openstack/tripleo-quickstart-extras","branch":"master","hashtags":[],"change_id":"I1399f163eaf711e1603838086a3cd1c231553598","subject":"[DNM] To test various toci vars refactorings","status":"ABANDONED","created":"2018-08-01 10:00:46.000000000","updated":"2018-08-07 07:21:07.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9c88190b3e586ef980e805f75a86bdd1c10b5ab8","_number":587745,"virtual_id_number":587745,"owner":{"_account_id":27898,"name":"Quique Llorente","email":"ellorent@redhat.com","username":"quiquell"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2018-08-01 10:40:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2018-08-01 15:22:41.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"date":"2018-08-01 10:42:16.000000000","_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-08-01 10:40:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-08-01 10:42:16.000000000","updated_by":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-08-01 15:22:41.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b23ecebdbc1c59ea2020c654edd02b5d0c6936ec","author":{"_account_id":27898,"name":"Quique Llorente","email":"ellorent@redhat.com","username":"quiquell"},"date":"2018-08-01 10:00:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9f9410f481a7b80ab30ec54f44392bda747a7383","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-01 10:00:58.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"e9abe3640ddac695d3768c59949d786d41eb009c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2018-08-01 10:01:00.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"6c3efc271517830d2a77834b694067b9481c7ab9","author":{"_account_id":27898,"name":"Quique Llorente","email":"ellorent@redhat.com","username":"quiquell"},"date":"2018-08-01 10:02:52.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"6d5b204f19cc3f2e784657c9b4a998b91fc05150","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-01 10:03:04.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"56262ce1694631ab079842935da88941f88fec6d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2018-08-01 10:03:06.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"d1b007162c0e6e69ee20481514f9d2ca009b7da8","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2018-08-01 10:05:00.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-newton-delorean-full-minimal https://ci.centos.org/job/tripleo-quickstart-extras-gate-newton-delorean-full-minimal/6780/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"b1a5d12da69f4e6d2ca8765a94722e267f0fbfec","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2018-08-01 10:21:05.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-newton-delorean-full-minimal https://ci.centos.org/job/tripleo-quickstart-extras-gate-newton-delorean-full-minimal/6781/ : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"1fb42b864ea3d48ae1264d722999f7b05310f80d","author":{"_account_id":27898,"name":"Quique Llorente","email":"ellorent@redhat.com","username":"quiquell"},"date":"2018-08-01 10:24:26.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"76b5b26b79d5428631fa4aa245e4faccbe669e32","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-01 10:40:10.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-linters http://logs.openstack.org/45/587745/3/check/openstack-tox-linters/04e82db/ : SUCCESS in 4m 42s\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/45/587745/3/check/tripleo-ci-centos-7-undercloud-oooq/b76ff94/ : FAILURE in 5m 39s\n- tripleo-ci-centos-7-undercloud-upgrades http://logs.openstack.org/45/587745/3/check/tripleo-ci-centos-7-undercloud-upgrades/2bf73c7/ : FAILURE in 5m 38s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/45/587745/3/check/tripleo-ci-centos-7-undercloud-containers/a0380cc/ : FAILURE in 5m 35s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/45/587745/3/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/1a8fa75/ : FAILURE in 8m 06s\n- tripleo-ci-centos-7-3nodes-multinode http://logs.openstack.org/45/587745/3/check/tripleo-ci-centos-7-3nodes-multinode/ce7933b/ : FAILURE in 5m 26s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq http://logs.openstack.org/45/587745/3/check/tripleo-ci-centos-7-scenario009-multinode-oooq/c3fad5f/ : FAILURE in 5m 02s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/45/587745/3/check/tripleo-ci-centos-7-containers-multinode/289840f/ : FAILURE in 9m 09s\n- tripleo-ci-centos-7-scenario001-multinode-oooq-container http://logs.openstack.org/45/587745/3/check/tripleo-ci-centos-7-scenario001-multinode-oooq-container/c220420/ : FAILURE in 8m 33s\n- tripleo-ci-centos-7-scenario002-multinode-oooq-container http://logs.openstack.org/45/587745/3/check/tripleo-ci-centos-7-scenario002-multinode-oooq-container/eebcc58/ : FAILURE in 9m 56s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/45/587745/3/check/tripleo-ci-centos-7-scenario003-multinode-oooq-container/0b25c1c/ : FAILURE in 6m 11s\n- tripleo-ci-centos-7-scenario004-multinode-oooq-container http://logs.openstack.org/45/587745/3/check/tripleo-ci-centos-7-scenario004-multinode-oooq-container/a9ebb0b/ : FAILURE in 4m 58s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/45/587745/3/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/b3579e4/ : FAILURE in 6m 13s\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container http://logs.openstack.org/45/587745/3/check/tripleo-ci-centos-7-scenario008-multinode-oooq-container/4cce9d3/ : FAILURE in 9m 00s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/45/587745/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/39261d2/ : FAILURE in 8m 42s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/45/587745/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/886b922/ : FAILURE in 5m 00s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/45/587745/3/check/tripleo-ci-centos-7-standalone/4fe87cd/ : FAILURE in 8m 20s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/45/587745/3/check/tripleo-ci-centos-7-containers-multinode-queens/6535c4e/ : FAILURE in 6m 19s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/45/587745/3/check/tripleo-ci-centos-7-containers-multinode-pike/08c1c42/ : FAILURE in 5m 13s","accounts_in_message":[],"_revision_number":3},{"id":"fd91413bded5a3e1e4e1ed144aac44ddb896e8f8","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2018-08-01 10:42:16.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-newton-delorean-full-minimal https://ci.centos.org/job/tripleo-quickstart-extras-gate-newton-delorean-full-minimal/6784/ : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"2dc4ac638f0d055706126419d86975587fc89a3f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2018-08-01 15:22:41.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-tripleo-ci-centos-7-containers-multinode-upgrades-pike-branch http://logs.rdoproject.org/45/587745/3/openstack-check/legacy-tripleo-ci-centos-7-containers-multinode-upgrades-pike-branch/2974e5f/ : SUCCESS in 2h 30m 03s (non-voting)\n- legacy-tripleo-ci-centos-7-multinode-1ctlr-featureset037-updates-master http://logs.rdoproject.org/45/587745/3/openstack-check/legacy-tripleo-ci-centos-7-multinode-1ctlr-featureset037-updates-master/5e8c06a/ : SUCCESS in 1h 53m 36s\n- legacy-tripleo-ci-centos-7-container-to-container-upgrades-master http://logs.rdoproject.org/45/587745/3/openstack-check/legacy-tripleo-ci-centos-7-container-to-container-upgrades-master/215c8df/ : FAILURE in 28m 50s (non-voting)\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-master http://logs.rdoproject.org/45/587745/3/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-master/c7df9dc/ : SUCCESS in 2h 23m 11s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035-master http://logs.rdoproject.org/45/587745/3/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035-master/49f4aca/ : SUCCESS in 2h 12m 52s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch http://logs.rdoproject.org/45/587745/3/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch/98fa299/ : SUCCESS in 1h 46m 41s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-ocata-branch http://logs.rdoproject.org/45/587745/3/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-ocata-branch/c6bb364/ : FAILURE in 2h 14m 14s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/45/587745/3/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/78a936f/ : FAILURE in 21m 02s","accounts_in_message":[],"_revision_number":3},{"id":"9c88190b3e586ef980e805f75a86bdd1c10b5ab8","author":{"_account_id":27898,"name":"Quique Llorente","email":"ellorent@redhat.com","username":"quiquell"},"date":"2018-08-07 07:21:07.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"086b9a61299afe3b6a1b4a788f6a1b22966a8ac4","revisions":{"36c1b1ff582a27ccfb5190d6abb09cc06eecd876":{"kind":"REWORK","_number":1,"created":"2018-08-01 10:00:46.000000000","uploader":{"_account_id":27898,"name":"Quique Llorente","email":"ellorent@redhat.com","username":"quiquell"},"ref":"refs/changes/45/587745/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/45/587745/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/45/587745/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/45/587745/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/45/587745/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/45/587745/1"}}},"commit":{"parents":[{"commit":"7479b75da5823433a5a9f28eb923d1d71ab29146","subject":"Merge \"Remove single quote from docker inspect command in tempest run\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7479b75da5823433a5a9f28eb923d1d71ab29146"}]}],"author":{"name":"Quique Llorente","email":"ellorent@redhat.com","date":"2018-07-25 07:07:41.000000000","tz":120},"committer":{"name":"Quique Llorente","email":"ellorent@redhat.com","date":"2018-08-01 10:00:08.000000000","tz":120},"subject":"[DNM] To test various toci refactor","message":"[DNM] To test various toci refactor\n\nDepends-On: Iaf0ba7e06c809859f8ad60070285941ec436b5e9\nChange-Id: I1399f163eaf711e1603838086a3cd1c231553598\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/36c1b1ff582a27ccfb5190d6abb09cc06eecd876"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/36c1b1ff582a27ccfb5190d6abb09cc06eecd876"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"7479b75da5823433a5a9f28eb923d1d71ab29146","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"e407b62b864795fbeb7ef51b93fc7498e6538748":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2018-08-01 10:02:52.000000000","uploader":{"_account_id":27898,"name":"Quique Llorente","email":"ellorent@redhat.com","username":"quiquell"},"ref":"refs/changes/45/587745/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/45/587745/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/45/587745/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/45/587745/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/45/587745/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/45/587745/2"}}},"commit":{"parents":[{"commit":"7479b75da5823433a5a9f28eb923d1d71ab29146","subject":"Merge \"Remove single quote from docker inspect command in tempest run\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7479b75da5823433a5a9f28eb923d1d71ab29146"}]}],"author":{"name":"Quique Llorente","email":"ellorent@redhat.com","date":"2018-07-25 07:07:41.000000000","tz":120},"committer":{"name":"Quique Llorente","email":"ellorent@redhat.com","date":"2018-08-01 10:02:28.000000000","tz":120},"subject":"[DNM] To test various toci vars refactorings","message":"[DNM] To test various toci vars refactorings\n\nDepends-On: Iaf0ba7e06c809859f8ad60070285941ec436b5e9\nChange-Id: I1399f163eaf711e1603838086a3cd1c231553598\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/e407b62b864795fbeb7ef51b93fc7498e6538748"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/e407b62b864795fbeb7ef51b93fc7498e6538748"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"7479b75da5823433a5a9f28eb923d1d71ab29146","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"086b9a61299afe3b6a1b4a788f6a1b22966a8ac4":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2018-08-01 10:24:26.000000000","uploader":{"_account_id":27898,"name":"Quique Llorente","email":"ellorent@redhat.com","username":"quiquell"},"ref":"refs/changes/45/587745/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/45/587745/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/45/587745/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/45/587745/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/45/587745/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/45/587745/3"}}},"commit":{"parents":[{"commit":"7479b75da5823433a5a9f28eb923d1d71ab29146","subject":"Merge \"Remove single quote from docker inspect command in tempest run\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7479b75da5823433a5a9f28eb923d1d71ab29146"}]}],"author":{"name":"Quique Llorente","email":"ellorent@redhat.com","date":"2018-07-25 07:07:41.000000000","tz":120},"committer":{"name":"Quique Llorente","email":"ellorent@redhat.com","date":"2018-08-01 10:24:16.000000000","tz":120},"subject":"[DNM] To test various toci vars refactorings","message":"[DNM] To test various toci vars refactorings\n\nDepends-On: Iaf0ba7e06c809859f8ad60070285941ec436b5e9\n\nChange-Id: I1399f163eaf711e1603838086a3cd1c231553598\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/086b9a61299afe3b6a1b4a788f6a1b22966a8ac4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/086b9a61299afe3b6a1b4a788f6a1b22966a8ac4"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"7479b75da5823433a5a9f28eb923d1d71ab29146","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
