)]}'
{"id":"openstack%2Ftripleo-quickstart-extras~621473","triplet_id":"openstack%2Ftripleo-quickstart-extras~master~I4367bd554bfc06b2416fee792b5b2a91ff8f7ff8","project":"openstack/tripleo-quickstart-extras","branch":"master","topic":"bug/1806328","hashtags":[],"change_id":"I4367bd554bfc06b2416fee792b5b2a91ff8f7ff8","subject":"Move \u0027extra_tht_configs\u0027 tasks to overcloud-deploy role","status":"MERGED","created":"2018-12-03 04:54:58.000000000","updated":"2019-06-11 13:57:38.000000000","submitted":"2019-06-11 13:57:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"621473-1560261458684-b5d80b2d","meta_rev_id":"4e52ef9205b8ce5f16723fd8e49383c8151a3e93","_number":621473,"virtual_id_number":621473,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-06-11 13:57:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-07 13:51:15.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-07 11:00:58.000000000","_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":2,"date":"2019-06-11 07:06:03.000000000","_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":2,"date":"2019-06-11 09:32:42.000000000","_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":1,"date":"2019-06-11 09:32:42.000000000","_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-02-19 13:24:56.000000000","updated_by":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"reviewer":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"state":"REVIEWER"},{"updated":"2019-02-28 05:18:44.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2019-03-05 13:12:36.000000000","updated_by":{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},"reviewer":{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},"state":"REVIEWER"},{"updated":"2019-06-04 13:35:45.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-06-07 11:00:58.000000000","updated_by":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-07 13:51:15.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-11 07:06:03.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2019-06-11 09:32:42.000000000","updated_by":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"reviewer":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"state":"REVIEWER"},{"updated":"2019-06-11 13:57:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bb873d5d7cc80be78f7c4062630773a8faeb86d8","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2018-12-03 04:54:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d85d335159d089013ff56a7831ee831b02102780","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2018-12-03 05:59:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cda911a6834ca551a62917aa0beab67b24723266","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2018-12-03 06:00:00.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-quick-basic/1983/ : FAILURE\n\n* tripleo-quickstart-extras-gate-ocata-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-extras-gate-ocata-delorean-quick-basic/1978/ : FAILURE\n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/1941/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"ae130a1160005662ea933c7ef409e616c31c96c5","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2018-12-03 08:05:24.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/1942/ : FAILURE\n\n* tripleo-quickstart-extras-gate-master-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-quick-basic/1984/ : SUCCESS\n\n* tripleo-quickstart-extras-gate-ocata-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-extras-gate-ocata-delorean-quick-basic/1979/ : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"813d5a4b239f8748042c25ab23dbf3fd482bb78d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-03 08:58:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/73/621473/2/check/tripleo-ci-centos-7-undercloud-containers/51d4285/ : SUCCESS in 1h 48m 16s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/73/621473/2/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/d1e1eaf/ : SUCCESS in 2h 05m 01s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq http://logs.openstack.org/73/621473/2/check/tripleo-ci-centos-7-scenario009-multinode-oooq/bc61a3c/ : SUCCESS in 2h 24m 20s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/73/621473/2/check/tripleo-ci-centos-7-containers-multinode/66bd747/ : SUCCESS in 2h 31m 52s\n- tripleo-ci-centos-7-scenario001-multinode-oooq-container http://logs.openstack.org/73/621473/2/check/tripleo-ci-centos-7-scenario001-multinode-oooq-container/209d4ff/ : SUCCESS in 2h 45m 26s (non-voting)\n- tripleo-ci-centos-7-scenario002-multinode-oooq-container http://logs.openstack.org/73/621473/2/check/tripleo-ci-centos-7-scenario002-multinode-oooq-container/818d80e/ : SUCCESS in 2h 42m 26s (non-voting)\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/73/621473/2/check/tripleo-ci-centos-7-scenario003-multinode-oooq-container/2190807/ : SUCCESS in 2h 18m 41s (non-voting)\n- tripleo-ci-centos-7-scenario004-multinode-oooq-container http://logs.openstack.org/73/621473/2/check/tripleo-ci-centos-7-scenario004-multinode-oooq-container/073941d/ : SUCCESS in 2h 32m 04s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/73/621473/2/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/e446cd7/ : SUCCESS in 2h 40m 23s (non-voting)\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container http://logs.openstack.org/73/621473/2/check/tripleo-ci-centos-7-scenario008-multinode-oooq-container/36cb1a5/ : SUCCESS in 2h 07m 04s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/73/621473/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/3b3f6b2/ : SUCCESS in 2h 27m 42s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/73/621473/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/b0505a8/ : SUCCESS in 2h 29m 48s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/73/621473/2/check/tripleo-ci-centos-7-standalone/675ad36/ : SUCCESS in 1h 45m 06s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/73/621473/2/check/tripleo-ci-fedora-28-standalone/6f511ef/ : POST_FAILURE in 1h 55m 16s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/73/621473/2/check/tripleo-ci-centos-7-standalone-upgrade/8751990/ : SUCCESS in 2h 16m 39s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/73/621473/2/check/tripleo-ci-centos-7-containers-multinode-rocky/b971e96/ : SUCCESS in 2h 17m 53s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/73/621473/2/check/tripleo-ci-centos-7-containers-multinode-queens/50008bb/ : SUCCESS in 2h 04m 17s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/73/621473/2/check/tripleo-ci-centos-7-containers-multinode-pike/84bc3b7/ : SUCCESS in 2h 14m 49s\n- openstack-tox-linters http://logs.openstack.org/73/621473/2/check/openstack-tox-linters/918d0d3/ : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":2},{"id":"1b2221d77a7fac1123bf04c249aa074730929a6f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2018-12-03 12:03:03.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/73/621473/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/2681bea/ : FAILURE in 26m 35s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/73/621473/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/ec5901b/ : FAILURE in 22m 10s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/73/621473/2/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/fd40a99/ : SUCCESS in 1h 53m 42s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch http://logs.rdoproject.org/73/621473/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch/e07bab8/ : FAILURE in 30m 03s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/73/621473/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/0654c58/ : FAILURE in 22m 53s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/73/621473/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/c45e39b/ : FAILURE in 31m 51s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/73/621473/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/8fa1edc/ : FAILURE in 11m 01s","accounts_in_message":[],"_revision_number":2},{"id":"e2ca4221971f47294a9075266093890bf3a962b5","author":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"date":"2019-02-19 13:17:05.000000000","message":"Patch Set 2:\n\nstill planning on working on this? or should this be abandoned?","accounts_in_message":[],"_revision_number":2},{"id":"64aa475058d66c501b52ddf2e2fa4c8bb018dc13","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-02-19 13:19:01.000000000","message":"Patch Set 2:\n\n\u003e still planning on working on this? or should this be abandoned?\n\nohh i thought this was merged, will rebase it and is open for reviews.","accounts_in_message":[],"_revision_number":2},{"id":"4e3e89402de860b83079fa3a13d0ebe031faa607","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-02-19 13:23:28.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6d158b7a8cb6c731fb100ebe4b9214a66de4346a","author":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"date":"2019-02-19 13:24:56.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"7f8116d5cb200158bac03d173d6c34bb2be4c5e4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-02-19 16:14:58.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/73/621473/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/6bf0ab4/ : FAILURE in 2h 50m 24s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/73/621473/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/e2401f9/ : RETRY_LIMIT in 8m 39s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/73/621473/3/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/a9146c1/ : SUCCESS in 1h 38m 50s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch : NODE_FAILURE in 0s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/73/621473/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/3b3e3c2/ : RETRY_LIMIT in 6m 30s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch : NODE_FAILURE in 0s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/73/621473/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/bb6fbbc/ : RETRY_LIMIT in 8m 39s","accounts_in_message":[],"_revision_number":3},{"id":"6c27575cda49ec14d9f38bc58f09067618ce2343","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-19 16:19:14.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/73/621473/3/check/tripleo-ci-centos-7-undercloud-containers/cd79d12/ : POST_FAILURE in 2h 24m 44s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/73/621473/3/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/21355b7/ : FAILURE in 1h 26m 23s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/73/621473/3/check/tripleo-ci-centos-7-containers-multinode/e89c194/ : SUCCESS in 2h 31m 25s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/73/621473/3/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/1c4761b/ : SUCCESS in 2h 45m 49s (non-voting)\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container http://logs.openstack.org/73/621473/3/check/tripleo-ci-centos-7-scenario008-multinode-oooq-container/8a6d760/ : FAILURE in 2h 17m 37s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container http://logs.openstack.org/73/621473/3/check/tripleo-ci-centos-7-scenario012-multinode-oooq-container/e586e9c/ : FAILURE in 2h 49m 11s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/73/621473/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/8de07bf/ : SUCCESS in 2h 17m 27s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/73/621473/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/4f3133c/ : SUCCESS in 2h 38m 19s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/73/621473/3/check/tripleo-ci-centos-7-standalone/8bfba9f/ : SUCCESS in 2h 19m 09s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/73/621473/3/check/tripleo-ci-fedora-28-standalone/edcb0b2/ : SUCCESS in 1h 29m 29s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/73/621473/3/check/tripleo-ci-centos-7-standalone-upgrade/6d52c3a/ : FAILURE in 1h 21m 37s (non-voting)\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/73/621473/3/check/tripleo-ci-centos-7-standalone-os-tempest/9476b2d/ : FAILURE in 1h 15m 26s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/73/621473/3/check/tripleo-ci-centos-7-containers-multinode-rocky/8c7234d/ : SUCCESS in 2h 49m 27s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/73/621473/3/check/tripleo-ci-centos-7-containers-multinode-queens/5341d24/ : SUCCESS in 1h 57m 00s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/73/621473/3/check/tripleo-ci-centos-7-containers-multinode-pike/b0dfc48/ : SUCCESS in 2h 17m 35s\n- openstack-tox-linters http://logs.openstack.org/73/621473/3/check/openstack-tox-linters/e1ea308/ : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":3},{"id":"848f694a36d9b0c3fff505210409206ac263e483","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-02-19 17:23:45.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/2607/ : FAILURE\n\n* tripleo-quickstart-extras-gate-master-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-quick-basic/2648/ : FAILURE\n\n* tripleo-quickstart-extras-gate-ocata-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-extras-gate-ocata-delorean-quick-basic/2642/ : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"759eb08dba2c8da4c783701eef6a1db846852538","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-02-26 04:46:04.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"29579c9b9459d4d0509c5d054520825b89640df1","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-02-26 07:23:38.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-quick-basic/2705/ : FAILURE\n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/2664/ : FAILURE\n\n* tripleo-quickstart-extras-gate-ocata-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-extras-gate-ocata-delorean-quick-basic/2699/ : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"7e30659dc96501d775b25ea293a8fe4eeb5bf6ad","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-02-26 07:50:20.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/73/621473/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/6a0456b/ : SUCCESS in 2h 26m 53s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/73/621473/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/1d81547/ : SUCCESS in 2h 41m 58s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/73/621473/4/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/e839cc0/ : SUCCESS in 1h 50m 08s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch http://logs.rdoproject.org/73/621473/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch/0e508cf/ : SUCCESS in 1h 48m 44s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/73/621473/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/f130bc2/ : SUCCESS in 2h 09m 49s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/73/621473/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/93ff8c4/ : SUCCESS in 2h 33m 52s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/73/621473/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/903af00/ : SUCCESS in 3h 02m 14s","accounts_in_message":[],"_revision_number":4},{"id":"ba5a9e894b80156b5545d44dcd450f6b4eb13f15","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-26 08:02:25.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/73/621473/4/check/tripleo-ci-centos-7-undercloud-containers/0e79779/ : SUCCESS in 1h 37m 23s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/73/621473/4/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/3a97d30/ : SUCCESS in 2h 00m 13s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/73/621473/4/check/tripleo-ci-centos-7-containers-multinode/04bd5df/ : SUCCESS in 2h 03m 24s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/73/621473/4/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/f3249be/ : SUCCESS in 2h 11m 46s (non-voting)\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container http://logs.openstack.org/73/621473/4/check/tripleo-ci-centos-7-scenario008-multinode-oooq-container/c917513/ : FAILURE in 1h 30m 36s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container http://logs.openstack.org/73/621473/4/check/tripleo-ci-centos-7-scenario012-multinode-oooq-container/1b10b4e/ : FAILURE in 2h 01m 20s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/73/621473/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/a6a0aec/ : SUCCESS in 2h 15m 19s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/73/621473/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/d6f261b/ : SUCCESS in 2h 22m 48s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/73/621473/4/check/tripleo-ci-centos-7-standalone/ce07984/ : SUCCESS in 1h 46m 53s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/73/621473/4/check/tripleo-ci-fedora-28-standalone/5d8d698/ : SUCCESS in 1h 33m 29s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/73/621473/4/check/tripleo-ci-centos-7-standalone-upgrade/bd657a8/ : SUCCESS in 1h 59m 45s (non-voting)\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/73/621473/4/check/tripleo-ci-centos-7-standalone-os-tempest/e654b45/ : TIMED_OUT in 3h 13m 20s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/73/621473/4/check/tripleo-ci-centos-7-containers-multinode-rocky/1c842dc/ : SUCCESS in 2h 14m 57s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/73/621473/4/check/tripleo-ci-centos-7-containers-multinode-queens/df60ace/ : SUCCESS in 1h 52m 18s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/73/621473/4/check/tripleo-ci-centos-7-containers-multinode-pike/54b0eb6/ : SUCCESS in 1h 58m 21s\n- openstack-tox-linters http://logs.openstack.org/73/621473/4/check/openstack-tox-linters/5aeefef/ : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":4},{"id":"b80a706f0510b33021b241ee2e45191b21286057","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-03-07 10:16:27.000000000","message":"Patch Set 4: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"97d4246624072d9e4f960492243420fe589f6db9","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-03 20:40:26.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ea219d04a5cdd272459e167af0d9f8670a6913e5","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-06-03 20:40:45.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/3458/ : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"c551622a91ce6a47b95ee8f0a08786cf8ee8483d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-03 23:21:00.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/73/621473/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/3a7a31a/ : SUCCESS in 2h 11m 33s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/73/621473/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/bddca36/ : SUCCESS in 2h 22m 30s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/73/621473/5/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/db13ffe/ : SUCCESS in 1h 11m 42s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/73/621473/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/865e764/ : SUCCESS in 2h 04m 36s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/73/621473/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/861752c/ : SUCCESS in 2h 08m 17s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/73/621473/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/8729b43/ : SUCCESS in 2h 13m 44s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/73/621473/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/3347358/ : SUCCESS in 2h 38m 45s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"4b7fcd1f548c363393f15e27151483381beb1299","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 01:01:09.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/73/621473/5/check/tripleo-ci-centos-7-undercloud-containers/5c95e3b/ : SUCCESS in 1h 25m 20s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/73/621473/5/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/b138d44/ : FAILURE in 1h 06m 13s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/73/621473/5/check/tripleo-ci-centos-7-containers-multinode/fb24f51/ : SUCCESS in 2h 06m 23s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/73/621473/5/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/6f93273/ : SUCCESS in 2h 06m 39s (non-voting)\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container http://logs.openstack.org/73/621473/5/check/tripleo-ci-centos-7-scenario008-multinode-oooq-container/e4bff5e/ : FAILURE in 1h 20m 16s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container http://logs.openstack.org/73/621473/5/check/tripleo-ci-centos-7-scenario012-multinode-oooq-container/12abbe6/ : FAILURE in 2h 17m 04s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/73/621473/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/90ead71/ : SUCCESS in 2h 01m 15s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/73/621473/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/1da8fe8/ : SUCCESS in 2h 17m 27s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/73/621473/5/check/tripleo-ci-centos-7-standalone/812dbbc/ : SUCCESS in 1h 39m 54s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/73/621473/5/check/tripleo-ci-fedora-28-standalone/b55ccc0/ : SUCCESS in 1h 37m 35s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/73/621473/5/check/tripleo-ci-centos-7-standalone-upgrade/94958db/ : SUCCESS in 2h 23m 24s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/73/621473/5/check/tripleo-ci-centos-7-standalone-upgrade-stein/c1972ef/ : SUCCESS in 2h 14m 44s\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/73/621473/5/check/tripleo-ci-centos-7-standalone-os-tempest/a693039/ : SUCCESS in 1h 08m 02s\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/73/621473/5/check/tripleo-ci-centos-7-containers-multinode-stein/7147cd1/ : SUCCESS in 2h 00m 48s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/73/621473/5/check/tripleo-ci-centos-7-containers-multinode-rocky/a3f4f3e/ : SUCCESS in 2h 01m 19s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/73/621473/5/check/tripleo-ci-centos-7-containers-multinode-queens/a7f6ca5/ : SUCCESS in 2h 10m 51s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/73/621473/5/check/tripleo-ci-centos-7-containers-multinode-pike/d2ddc2d/ : SUCCESS in 2h 01m 54s\n- openstack-tox-linters http://logs.openstack.org/73/621473/5/check/openstack-tox-linters/5649a35/ : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":5},{"id":"bf83723c6835476a3ff27329347743ebf47bc7e7","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-06-04 09:57:26.000000000","message":"Patch Set 5:\n\n\u003e Uploaded patch set 5.\n\nThanks Alex for revising it, it just went down in my review queue","accounts_in_message":[],"_revision_number":5},{"id":"fbd983f55d08dc225c119274a60ed01e75a9b529","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-04 13:35:45.000000000","message":"Patch Set 5:\n\nnp was reviewing untriaged bugs and wandered across this one :D","accounts_in_message":[],"_revision_number":5},{"id":"fbd40a229de21215b1303d489ebfc1276effea9f","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2019-06-05 06:34:40.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"091b75a5b7ea84c1ecc154d6ced1304b34062ea5","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-06-07 10:51:09.000000000","message":"Patch Set 5: Workflow-1\n\ni see some issues, will revise it.","accounts_in_message":[],"_revision_number":5},{"id":"e67440fe9a15046c3f230c12206ffb17f1dda801","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-06-07 11:00:44.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"754dcde85f6ffb485d6ffb4c12567644d1711775","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-06-07 11:00:58.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/3479/ : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"672063fd7de00e136b567e395a50218efd2abaaf","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-07 13:51:15.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/73/621473/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/f307553/ : SUCCESS in 2h 14m 36s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/73/621473/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/f74261f/ : SUCCESS in 2h 19m 50s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/73/621473/6/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/2f0d850/ : SUCCESS in 1h 11m 30s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/73/621473/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/78a593c/ : SUCCESS in 1h 59m 44s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/73/621473/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/6a30fb1/ : SUCCESS in 2h 07m 01s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/73/621473/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/c1c82c6/ : SUCCESS in 2h 27m 02s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/73/621473/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/79a0a53/ : SUCCESS in 2h 48m 39s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"31505f8129d215fd3e6b9fdb4e7628679584dd00","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 17:42:51.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/73/621473/6/check/tripleo-ci-centos-7-undercloud-containers/06e7a18/ : SUCCESS in 1h 24m 52s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/73/621473/6/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/09ad1fe/ : FAILURE in 1h 14m 43s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/73/621473/6/check/tripleo-ci-centos-7-containers-multinode/860e3fa/ : SUCCESS in 2h 02m 38s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/73/621473/6/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/cc6bae4/ : SUCCESS in 2h 11m 39s (non-voting)\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container http://logs.openstack.org/73/621473/6/check/tripleo-ci-centos-7-scenario008-multinode-oooq-container/2c916e8/ : FAILURE in 1h 15m 56s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container http://logs.openstack.org/73/621473/6/check/tripleo-ci-centos-7-scenario012-multinode-oooq-container/06af752/ : FAILURE in 2h 15m 33s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/73/621473/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/47d4794/ : SUCCESS in 2h 08m 23s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/73/621473/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/fee7ade/ : SUCCESS in 2h 11m 01s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/73/621473/6/check/tripleo-ci-centos-7-standalone/c7f898c/ : SUCCESS in 1h 46m 11s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/73/621473/6/check/tripleo-ci-fedora-28-standalone/b08d167/ : SUCCESS in 1h 42m 26s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/73/621473/6/check/tripleo-ci-centos-7-standalone-upgrade/0110742/ : SUCCESS in 2h 03m 29s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/73/621473/6/check/tripleo-ci-centos-7-standalone-upgrade-stein/9d938c5/ : SUCCESS in 2h 05m 12s\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/73/621473/6/check/tripleo-ci-centos-7-standalone-os-tempest/197070c/ : SUCCESS in 1h 18m 32s\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/73/621473/6/check/tripleo-ci-centos-7-containers-multinode-stein/a32b023/ : SUCCESS in 1h 59m 33s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/73/621473/6/check/tripleo-ci-centos-7-containers-multinode-rocky/d712a5e/ : SUCCESS in 2h 11m 23s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/73/621473/6/check/tripleo-ci-centos-7-containers-multinode-queens/010c75c/ : SUCCESS in 1h 57m 54s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/73/621473/6/check/tripleo-ci-centos-7-containers-multinode-pike/0a50834/ : SUCCESS in 2h 11m 12s\n- openstack-tox-linters http://logs.openstack.org/73/621473/6/check/openstack-tox-linters/49ae6b1/ : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":6},{"id":"69e8f07c8bfd2fbe4cbb2c3975ae94e392bbd93b","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2019-06-11 07:06:03.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"f3e95d838f1036fcbf882649462ef80e76a36827","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-06-11 09:32:42.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"32359633f80131e58611eb77c38c6b8c01d8e6b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 09:33:07.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"c281a199eb2c769c51640d2d8cc477793301908c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 13:57:38.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/73/621473/6/gate/tripleo-ci-centos-7-undercloud-containers/4292817/ : SUCCESS in 1h 37m 42s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/73/621473/6/gate/tripleo-ci-centos-7-containers-multinode/007735d/ : SUCCESS in 2h 18m 39s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/73/621473/6/gate/tripleo-ci-centos-7-standalone/e373d0a/ : SUCCESS in 2h 16m 21s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/73/621473/6/gate/tripleo-ci-fedora-28-standalone/a8f4906/ : SUCCESS in 1h 50m 08s\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/73/621473/6/gate/tripleo-ci-centos-7-standalone-os-tempest/9b4f943/ : SUCCESS in 1h 11m 41s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/73/621473/6/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/e6be14c/ : SUCCESS in 2h 29m 57s\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/73/621473/6/gate/tripleo-ci-centos-7-standalone-upgrade-stein/6929778/ : SUCCESS in 2h 03m 11s\n- openstack-tox-linters http://logs.openstack.org/73/621473/6/gate/openstack-tox-linters/81de1a3/ : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":6},{"id":"49ba4d552c344eb963fb1b36070343477eeb1612","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 13:57:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"4666dc02a0051e9eb7116e630986e7408480f927","revisions":{"e78879aeecc8de856733305bb7555bf2af71b085":{"kind":"REWORK","_number":1,"created":"2018-12-03 04:54:58.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/73/621473/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/73/621473/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/73/621473/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/73/621473/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/73/621473/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/73/621473/1"}}},"commit":{"parents":[{"commit":"47d0c0e8ddc8078c3df924646f8359988057c1fe","subject":"Add TestNetworkBasicOps.test_hotplug_nic to master skip list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/47d0c0e8ddc8078c3df924646f8359988057c1fe"}]}],"author":{"name":"yatin","email":"ykarel@redhat.com","date":"2018-12-03 04:50:26.000000000","tz":330},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2018-12-03 04:50:28.000000000","tz":330},"subject":"Move \u0027extra_tht_configs\u0027 tasks to overcloud-deploy role","message":"Move \u0027extra_tht_configs\u0027 tasks to overcloud-deploy role\n\n\u0027extra_tht_configs\u0027 parameter is used to pass extra tht\nconfigs to overcloud deploy command(using \u0027extra_tht_config_args\u0027\nfact), this parameter is used to modify overcloud-deploy command,\nso this patch moves the tasks associated with it to overcloud-deploy\nrole.\n\nCloses-Bug: #1806328\nChange-Id: I4367bd554bfc06b2416fee792b5b2a91ff8f7ff8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/e78879aeecc8de856733305bb7555bf2af71b085"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/e78879aeecc8de856733305bb7555bf2af71b085"}]},"branch":"refs/heads/master"},"d3915953a6af5fc448fc9a11e4800014dfdf10ad":{"kind":"REWORK","_number":2,"created":"2018-12-03 05:59:57.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/73/621473/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/73/621473/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/73/621473/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/73/621473/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/73/621473/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/73/621473/2"}}},"commit":{"parents":[{"commit":"47d0c0e8ddc8078c3df924646f8359988057c1fe","subject":"Add TestNetworkBasicOps.test_hotplug_nic to master skip list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/47d0c0e8ddc8078c3df924646f8359988057c1fe"}]}],"author":{"name":"yatin","email":"ykarel@redhat.com","date":"2018-12-03 04:50:26.000000000","tz":330},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2018-12-03 05:57:43.000000000","tz":330},"subject":"Move \u0027extra_tht_configs\u0027 tasks to overcloud-deploy role","message":"Move \u0027extra_tht_configs\u0027 tasks to overcloud-deploy role\n\n\u0027extra_tht_configs\u0027 parameter is used to pass extra tht\nconfigs to overcloud deploy command(using \u0027extra_tht_config_args\u0027\nfact), this parameter is used to modify overcloud-deploy command,\nso this patch moves the tasks associated with it to overcloud-deploy\nrole.\n\nCloses-Bug: #1806328\nChange-Id: I4367bd554bfc06b2416fee792b5b2a91ff8f7ff8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/d3915953a6af5fc448fc9a11e4800014dfdf10ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/d3915953a6af5fc448fc9a11e4800014dfdf10ad"}]},"branch":"refs/heads/master"},"5ba0964cea20564fe6944b3b89ea51e3bbbb8fb5":{"kind":"REWORK","_number":3,"created":"2019-02-19 13:23:28.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/73/621473/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/73/621473/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/73/621473/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/73/621473/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/73/621473/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/73/621473/3"}}},"commit":{"parents":[{"commit":"da62d19416c6c0c442af967c17107f24faa1133c","subject":"Merge \"Set network interface in extra node\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/da62d19416c6c0c442af967c17107f24faa1133c"}]}],"author":{"name":"yatin","email":"ykarel@redhat.com","date":"2018-12-03 04:50:26.000000000","tz":330},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2019-02-19 13:22:26.000000000","tz":330},"subject":"Move \u0027extra_tht_configs\u0027 tasks to overcloud-deploy role","message":"Move \u0027extra_tht_configs\u0027 tasks to overcloud-deploy role\n\n\u0027extra_tht_configs\u0027 parameter is used to pass extra tht\nconfigs to overcloud deploy command(using \u0027extra_tht_config_args\u0027\nfact), this parameter is used to modify overcloud-deploy command,\nso this patch moves the tasks associated with it to overcloud-deploy\nrole.\n\nCloses-Bug: #1806328\nChange-Id: I4367bd554bfc06b2416fee792b5b2a91ff8f7ff8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/5ba0964cea20564fe6944b3b89ea51e3bbbb8fb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/5ba0964cea20564fe6944b3b89ea51e3bbbb8fb5"}]},"branch":"refs/heads/master"},"3620329819a61ebff615619e712b0fe13a54d8ba":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-02-26 04:46:04.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/73/621473/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/73/621473/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/73/621473/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/73/621473/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/73/621473/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/73/621473/4"}}},"commit":{"parents":[{"commit":"8a6b1ae25126b771c448322a7ad0c9e20a1f4fab","subject":"Rename errors file if it\u0027s big","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/8a6b1ae25126b771c448322a7ad0c9e20a1f4fab"}]}],"author":{"name":"yatin","email":"ykarel@redhat.com","date":"2018-12-03 04:50:26.000000000","tz":330},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2019-02-26 04:46:04.000000000","tz":0},"subject":"Move \u0027extra_tht_configs\u0027 tasks to overcloud-deploy role","message":"Move \u0027extra_tht_configs\u0027 tasks to overcloud-deploy role\n\n\u0027extra_tht_configs\u0027 parameter is used to pass extra tht\nconfigs to overcloud deploy command(using \u0027extra_tht_config_args\u0027\nfact), this parameter is used to modify overcloud-deploy command,\nso this patch moves the tasks associated with it to overcloud-deploy\nrole.\n\nCloses-Bug: #1806328\nChange-Id: I4367bd554bfc06b2416fee792b5b2a91ff8f7ff8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/3620329819a61ebff615619e712b0fe13a54d8ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/3620329819a61ebff615619e712b0fe13a54d8ba"}]},"branch":"refs/heads/master"},"216942979f9a9d45d467a32be8a9948d5a8f96c3":{"kind":"REWORK","_number":5,"created":"2019-06-03 20:40:26.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/73/621473/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/73/621473/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/73/621473/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/73/621473/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/73/621473/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/73/621473/5"}}},"commit":{"parents":[{"commit":"6262e1f7cfe29d24c26c083afc166cf85f6a3251","subject":"Merge \"Update skip list for stein\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/6262e1f7cfe29d24c26c083afc166cf85f6a3251"}]}],"author":{"name":"yatin","email":"ykarel@redhat.com","date":"2018-12-03 04:50:26.000000000","tz":330},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-06-03 20:40:21.000000000","tz":-360},"subject":"Move \u0027extra_tht_configs\u0027 tasks to overcloud-deploy role","message":"Move \u0027extra_tht_configs\u0027 tasks to overcloud-deploy role\n\n\u0027extra_tht_configs\u0027 parameter is used to pass extra tht\nconfigs to overcloud deploy command(using \u0027extra_tht_config_args\u0027\nfact), this parameter is used to modify overcloud-deploy command,\nso this patch moves the tasks associated with it to overcloud-deploy\nrole.\n\nCloses-Bug: #1806328\nChange-Id: I4367bd554bfc06b2416fee792b5b2a91ff8f7ff8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/216942979f9a9d45d467a32be8a9948d5a8f96c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/216942979f9a9d45d467a32be8a9948d5a8f96c3"}]},"branch":"refs/heads/master"},"4666dc02a0051e9eb7116e630986e7408480f927":{"kind":"REWORK","_number":6,"created":"2019-06-07 11:00:44.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/73/621473/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/73/621473/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/73/621473/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/73/621473/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/73/621473/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/73/621473/6"}}},"commit":{"parents":[{"commit":"366634808a86e67a7ee51a048e3adcd4550f4cd3","subject":"Add resize_server_revert and rebuild_list_server to skip list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/366634808a86e67a7ee51a048e3adcd4550f4cd3"}]}],"author":{"name":"yatin","email":"ykarel@redhat.com","date":"2018-12-03 04:50:26.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2019-06-07 10:59:27.000000000","tz":330},"subject":"Move \u0027extra_tht_configs\u0027 tasks to overcloud-deploy role","message":"Move \u0027extra_tht_configs\u0027 tasks to overcloud-deploy role\n\n\u0027extra_tht_configs\u0027 parameter is used to pass extra tht\nconfigs to overcloud deploy command(using \u0027extra_tht_config_args\u0027\nfact), this parameter is used to modify overcloud-deploy command,\nso this patch moves the tasks associated with it to overcloud-deploy\nrole.\n\nCloses-Bug: #1806328\nChange-Id: I4367bd554bfc06b2416fee792b5b2a91ff8f7ff8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/4666dc02a0051e9eb7116e630986e7408480f927"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/4666dc02a0051e9eb7116e630986e7408480f927"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
