)]}'
{"id":"openstack%2Ftripleo-quickstart-extras~664170","triplet_id":"openstack%2Ftripleo-quickstart-extras~master~I1d62928b9572e9dae6c1f0eff064be29d187c620","project":"openstack/tripleo-quickstart-extras","branch":"master","topic":"bp/nova-less-deploy","hashtags":[],"change_id":"I1d62928b9572e9dae6c1f0eff064be29d187c620","subject":"DNM: undercloud_enable_nova: false by default for Train onwards","status":"ABANDONED","created":"2019-06-09 23:18:51.000000000","updated":"2019-09-06 04:14:32.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"920efaba15b7c5382b2bfd7977d7bcfbe2e4698b","_number":664170,"virtual_id_number":664170,"owner":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"actions":{},"labels":{"Verified":{"all":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-19 20:30:29.000000000","updated_by":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-19 21:49:13.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2019-08-21 13:49:59.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2019-08-27 06:51:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-27 07:14:51.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0e0b1b3a77770ee8b9c8d00d1cf054fd4b566211","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2019-06-09 23:18:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4eb81d187b15fce007e6dce146c019cd60b599d1","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-06-09 23:19:08.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/3492/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"496701477ea4554244c21ae15fe8f05ba04c4758","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-10 00:33:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/70/664170/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/8f8dac0/ : FAILURE in 1h 00m 04s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/70/664170/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/c658992/ : FAILURE in 58m 53s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/70/664170/1/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/9c248de/ : FAILURE in 42m 37s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/70/664170/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/18da4e2/ : FAILURE in 1h 07m 54s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/70/664170/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/2504e8c/ : FAILURE in 1h 11m 23s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/70/664170/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/d5ecf61/ : FAILURE in 1h 03m 00s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/70/664170/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/8eb8cb6/ : FAILURE in 59m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b721365d0c7915bb20009298ee6a2a631eb369b3","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2019-06-10 01:48:12.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"d819e97f0c98f621c0ed12d356d342ed00b6fb52","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-06-10 01:48:27.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/3495/ : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"cb7fc0d83e5d3495da107b7dd9d6f68668a2a176","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 04:15:21.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/70/664170/2/check/tripleo-ci-centos-7-undercloud-containers/78a0934/ : SUCCESS in 1h 19m 30s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/70/664170/2/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/d99f606/ : FAILURE in 1h 17m 31s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/70/664170/2/check/tripleo-ci-centos-7-containers-multinode/c5a801a/ : FAILURE in 1h 17m 26s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/70/664170/2/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/026fb15/ : FAILURE in 1h 19m 32s (non-voting)\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container http://logs.openstack.org/70/664170/2/check/tripleo-ci-centos-7-scenario008-multinode-oooq-container/11d0044/ : FAILURE in 1h 13m 13s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container http://logs.openstack.org/70/664170/2/check/tripleo-ci-centos-7-scenario012-multinode-oooq-container/c9ecf9e/ : FAILURE in 1h 14m 50s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/70/664170/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/9b6c38c/ : FAILURE in 1h 12m 31s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/70/664170/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/31c9a69/ : FAILURE in 1h 08m 20s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/70/664170/2/check/tripleo-ci-centos-7-standalone/2652984/ : SUCCESS in 1h 43m 45s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/70/664170/2/check/tripleo-ci-fedora-28-standalone/7c0c603/ : FAILURE in 1h 42m 10s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/70/664170/2/check/tripleo-ci-centos-7-standalone-upgrade/83418fe/ : SUCCESS in 1h 50m 24s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/70/664170/2/check/tripleo-ci-centos-7-standalone-upgrade-stein/ca80a95/ : SUCCESS in 1h 57m 20s\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/70/664170/2/check/tripleo-ci-centos-7-standalone-os-tempest/cdbaaba/ : SUCCESS in 1h 09m 58s\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/70/664170/2/check/tripleo-ci-centos-7-containers-multinode-stein/a3f914e/ : FAILURE in 1h 10m 59s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/70/664170/2/check/tripleo-ci-centos-7-containers-multinode-rocky/e46404a/ : FAILURE in 1h 14m 03s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/70/664170/2/check/tripleo-ci-centos-7-containers-multinode-queens/9bf1d07/ : FAILURE in 1h 10m 18s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/70/664170/2/check/tripleo-ci-centos-7-containers-multinode-pike/fbb7935/ : SUCCESS in 2h 00m 03s\n- openstack-tox-linters http://logs.openstack.org/70/664170/2/check/openstack-tox-linters/452b73a/ : FAILURE in 3m 21s","accounts_in_message":[],"_revision_number":2},{"id":"0d1ccae6c2ef5a6d4d80dc83f83d08950073fcb1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-10 05:09:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/70/664170/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/d0728b5/ : SUCCESS in 2h 27m 33s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/70/664170/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/2fe67e9/ : SUCCESS in 2h 14m 53s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/70/664170/2/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/bc8a82c/ : FAILURE in 42m 04s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/70/664170/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/daa688f/ : SUCCESS in 1h 54m 48s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/70/664170/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/6002ada/ : SUCCESS in 2h 06m 50s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/70/664170/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/b4665b0/ : SUCCESS in 2h 15m 26s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/70/664170/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/fe610e8/ : SUCCESS in 3h 18m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"43aeb1eadc26ccb77f8e21a5bd5e4c0e62dca84f","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2019-06-10 22:40:38.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5a7fab8c8fc74ad83443cf97b64f9ddae0689440","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-06-10 22:40:58.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/3505/ : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"2171f686f1944fa6eb03cdc50bb2a18cbaf7a658","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-11 00:59:44.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/70/664170/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/c0add5f/ : FAILURE in 45m 22s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/70/664170/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/d57f70a/ : FAILURE in 51m 20s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/70/664170/3/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/a09aa4e/ : FAILURE in 36m 19s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/70/664170/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/d5b616b/ : SUCCESS in 2h 15m 10s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/70/664170/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/5fabede/ : FAILURE in 1h 14m 53s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/70/664170/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/54c07d5/ : FAILURE in 1h 12m 42s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/70/664170/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/e47c6fa/ : FAILURE in 40m 16s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3ec3c738246679145b2b78f50508c9e00cf27410","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2019-06-11 04:11:58.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"72dab028683dca23090253f21cf563300ca772c9","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-06-11 04:12:13.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/3508/ : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"8599150f3d38556f21d6e810c97f3490eb2ea4b0","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2019-06-11 04:12:27.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"2fa31aadeff8fbc8a9cb40c0d1cf53b3d8a6e4a7","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-06-11 04:12:43.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/3511/ : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"69419c3e90688c76174645a5da0eb5dbb2b8a412","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2019-06-11 04:21:46.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"a99fed492bd4b9ed8ca50a823d1ef9ebc1dbaa77","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-06-11 04:22:00.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/3514/ : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"e92a19635a992bf64508ce9f71d6dbdc6eb78977","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 06:32:11.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/70/664170/6/check/tripleo-ci-centos-7-undercloud-containers/34eca1b/ : FAILURE in 16m 23s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/70/664170/6/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/77dbc99/ : FAILURE in 15m 44s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/70/664170/6/check/tripleo-ci-centos-7-containers-multinode/90929cb/ : FAILURE in 1h 02m 22s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/70/664170/6/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/e019784/ : FAILURE in 1h 01m 45s (non-voting)\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container http://logs.openstack.org/70/664170/6/check/tripleo-ci-centos-7-scenario008-multinode-oooq-container/3537fa2/ : FAILURE in 1h 06m 52s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container http://logs.openstack.org/70/664170/6/check/tripleo-ci-centos-7-scenario012-multinode-oooq-container/445f43b/ : FAILURE in 1h 07m 11s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/70/664170/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/7418125/ : FAILURE in 1h 10m 03s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/70/664170/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/565dd08/ : FAILURE in 1h 03m 35s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/70/664170/6/check/tripleo-ci-centos-7-standalone/0ebcfbc/ : FAILURE in 16m 13s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/70/664170/6/check/tripleo-ci-fedora-28-standalone/6dfeb19/ : SUCCESS in 2h 03m 00s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/70/664170/6/check/tripleo-ci-centos-7-standalone-upgrade/2027d0a/ : FAILURE in 14m 48s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/70/664170/6/check/tripleo-ci-centos-7-standalone-upgrade-stein/1aceeb3/ : FAILURE in 15m 06s\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/70/664170/6/check/tripleo-ci-centos-7-standalone-os-tempest/89b70f6/ : FAILURE in 16m 50s\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/70/664170/6/check/tripleo-ci-centos-7-containers-multinode-stein/e997acb/ : FAILURE in 20m 05s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/70/664170/6/check/tripleo-ci-centos-7-containers-multinode-rocky/29452cc/ : FAILURE in 21m 37s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/70/664170/6/check/tripleo-ci-centos-7-containers-multinode-queens/c649ac0/ : FAILURE in 19m 48s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/70/664170/6/check/tripleo-ci-centos-7-containers-multinode-pike/ba52bd4/ : SUCCESS in 2h 03m 19s\n- openstack-tox-linters http://logs.openstack.org/70/664170/6/check/openstack-tox-linters/deba5e5/ : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":6},{"id":"d82b6f5e10b7663170ec0004461234458c19c38f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-11 08:04:52.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/70/664170/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/588d0bc/ : FAILURE in 44m 03s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/70/664170/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/744eb9f/ : FAILURE in 46m 14s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/70/664170/6/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/6fa0cf4/ : FAILURE in 35m 58s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/70/664170/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/bb47a82/ : SUCCESS in 2h 16m 03s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/70/664170/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/a159e76/ : TIMED_OUT in 3h 39m 50s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/70/664170/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/c67105d/ : SUCCESS in 2h 27m 08s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/70/664170/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/e939653/ : FAILURE in 51m 15s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"a643d001567c00828ea6f4133306d500629304cf","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2019-06-16 21:32:29.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"7e2013d972b4115729f70e32cdf88c353779a1da","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-06-16 21:32:45.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/3548/ : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"950d713301343ea09d78467a5842baddf5847094","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-16 23:45:06.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/70/664170/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/545effa/ : FAILURE in 38m 43s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/70/664170/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/58c5ae1/ : FAILURE in 39m 54s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/70/664170/7/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/fa6a7f0/ : FAILURE in 28m 17s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/70/664170/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/04032c5/ : SUCCESS in 1h 57m 00s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/70/664170/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/3d3964c/ : SUCCESS in 2h 01m 41s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/70/664170/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/f7c8106/ : SUCCESS in 2h 10m 29s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/70/664170/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/3bf1ec4/ : FAILURE in 39m 19s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"310950294eef9815856055563b7fe4a6f7c979df","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 00:04:01.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/70/664170/7/check/tripleo-ci-centos-7-undercloud-containers/a55308c/ : FAILURE in 47m 28s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/70/664170/7/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/a80262d/ : FAILURE in 1h 26m 47s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/70/664170/7/check/tripleo-ci-centos-7-containers-multinode/1475349/ : FAILURE in 48m 30s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/70/664170/7/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/4cb6c94/ : FAILURE in 46m 31s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container http://logs.openstack.org/70/664170/7/check/tripleo-ci-centos-7-scenario012-multinode-oooq-container/22f2263/ : FAILURE in 58m 38s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/70/664170/7/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/bc671ef/ : FAILURE in 54m 30s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/70/664170/7/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/3889172/ : FAILURE in 52m 41s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/70/664170/7/check/tripleo-ci-centos-7-standalone/09e105a/ : SUCCESS in 2h 00m 33s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/70/664170/7/check/tripleo-ci-fedora-28-standalone/b22822d/ : SUCCESS in 1h 40m 06s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/70/664170/7/check/tripleo-ci-centos-7-standalone-upgrade/d091f7a/ : SUCCESS in 2h 05m 06s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/70/664170/7/check/tripleo-ci-centos-7-standalone-upgrade-stein/a7ace17/ : SUCCESS in 2h 01m 16s\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/70/664170/7/check/tripleo-ci-centos-7-standalone-os-tempest/b0d5baa/ : SUCCESS in 1h 05m 33s\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/70/664170/7/check/tripleo-ci-centos-7-containers-multinode-stein/f7b5b91/ : SUCCESS in 2h 01m 40s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/70/664170/7/check/tripleo-ci-centos-7-containers-multinode-rocky/7906769/ : SUCCESS in 2h 14m 00s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/70/664170/7/check/tripleo-ci-centos-7-containers-multinode-queens/8683d04/ : SUCCESS in 2h 04m 21s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/70/664170/7/check/tripleo-ci-centos-7-containers-multinode-pike/224bfef/ : SUCCESS in 2h 25m 07s\n- openstack-tox-linters http://logs.openstack.org/70/664170/7/check/openstack-tox-linters/fb92903/ : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":7},{"id":"06ef8e81e44fef5d299acb2f2076bca0972d5f19","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2019-08-15 02:33:39.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"ddd0be98537af3b17a2edc6e1167111b0133eab1","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-08-15 02:33:57.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/3791/ : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"2de2b4cb08142b64574e1f5b732b4a6740bb089c","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2019-08-15 03:31:30.000000000","message":"Uploaded patch set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"c020d8097e84a0520c83b3e63b55db892dd3d779","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-08-15 03:31:48.000000000","message":"Patch Set 9:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/3792/ : FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"529866cec9622b1e1536503f0223b9642009ff3e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 05:53:11.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/53ba0f4523034e28a5ca48205d888001 : FAILURE in 53m 56s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/c7cca3cc7ae049e393cd5fbc0b1fd2d1 : FAILURE in 1h 37m 50s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/b0469c92229940e4b327d9d9e0286621 : FAILURE in 55m 57s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/eeb6f3eeeaad4a3e9b67553456acd396 : FAILURE in 52m 13s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/0694109333094a7084c0504c58c2e9b8 : FAILURE in 52m 23s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/88d7979fa00848e7b5f6493e98918a4e : FAILURE in 54m 14s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/1dd299256911455b9668f67ef72d3ebd : FAILURE in 54m 53s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/2377454313ae4cfcbd8f4d8a4839dd85 : SUCCESS in 1h 41m 34s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/c594849b205a41b5a873cd32ad116aa6 : SUCCESS in 2h 16m 13s (non-voting)\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/b743820b0f1549dca88d044de937e3ad : SUCCESS in 1h 15m 04s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/f9fe795bba244569ac5f0833e6c46c01 : SUCCESS in 2h 02m 03s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/fa22231182084c819873fdfee179c7da : SUCCESS in 2h 11m 24s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/94516674b21741c7b3971af7cc849445 : SUCCESS in 1h 56m 16s\n- tripleo-ci-centos-7-containers-multinode-pike https://zuul.opendev.org/t/openstack/build/4d81e219f5804163a82b6efcecf33bd6 : SUCCESS in 2h 12m 07s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/006b5d7cc79448dbb75634d12cfe528e : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":9},{"id":"59e8d1f758562510cd07c30f665ac580b5232f33","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-15 06:41:10.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/70/664170/9/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/285d432/ : FAILURE in 40m 23s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/70/664170/9/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/a9760e9/ : FAILURE in 35m 05s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/70/664170/9/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/33de02a/ : FAILURE in 2h 23m 43s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/70/664170/9/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/5cdd411/ : SUCCESS in 2h 24m 11s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/70/664170/9/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/2cba198/ : FAILURE in 3h 07m 38s","accounts_in_message":[],"_revision_number":9},{"id":"d20cb80db3a30aaea2bf7eab54d03923e32b420b","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2019-08-18 21:11:10.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"6e6abffa6de9b8cda1dc511c7d1ba291162430d8","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-08-18 21:11:32.000000000","message":"Patch Set 10:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/3802/ : FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"789650e48f2e22432addaaae621e41482ea842a3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-18 23:31:31.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/52d3d637f2d1417c84ecbdaeb18d5775 : FAILURE in 1h 00m 14s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/2dcaf726d7514d54b0c1f5a24f93766a : FAILURE in 1h 44m 01s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/af184f4145dd495c89f827fe3dc0b19d : FAILURE in 1h 04m 24s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/50e75b880e87486f97b7f1991452288e : FAILURE in 1h 03m 04s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/6d2b543ccca04fdc96ab6bb08ed5c33a : FAILURE in 1h 03m 21s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/415dcd1031324cf380b5b0b756dc234a : FAILURE in 1h 01m 29s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/5191baaa0ef0452aab5aae3d49ca8b37 : FAILURE in 1h 05m 13s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/6045635d5ef24b3e9be10929778cb702 : SUCCESS in 1h 51m 38s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/81271ac874df4c57b1e19e3489cec8f0 : SUCCESS in 2h 14m 47s (non-voting)\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/fd07db3a8f5e4e33b11ee126419f5ca7 : SUCCESS in 1h 19m 34s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/86bb6f8bf2c748169b64116963c16895 : SUCCESS in 1h 59m 52s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/0f4147f78bef4d64ba50d77d57ec0b5a : SUCCESS in 2h 05m 06s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/bec60c10fc064851bbc6b9df0f5d82ee : SUCCESS in 1h 54m 25s\n- tripleo-ci-centos-7-containers-multinode-pike https://zuul.opendev.org/t/openstack/build/f7ab0cd3721e42b7adcb8be193d0bb4f : SUCCESS in 2h 04m 46s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6ca6d5587a06421a8b0d99819ff139a9 : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":10},{"id":"9d91f9e1cf92c8ab18b74d81c04c5f744985f9cf","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-19 00:10:13.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/70/664170/10/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/e563b7e/ : FAILURE in 48m 06s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/70/664170/10/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/95a040c/ : FAILURE in 47m 50s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/70/664170/10/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/28b8b4f/ : SUCCESS in 2h 24m 16s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/70/664170/10/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/8328158/ : SUCCESS in 2h 47m 38s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/70/664170/10/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/920f6bd/ : SUCCESS in 2h 57m 16s","accounts_in_message":[],"_revision_number":10},{"id":"4acbfeaa5d4b64561b00cbfb71413fd37a58463c","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2019-08-19 01:42:05.000000000","message":"Patch Set 11: Commit message was updated.","accounts_in_message":[],"_revision_number":11},{"id":"b2ee82ceff161e80fd2041c1a763955e4022fee5","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-08-19 01:42:29.000000000","message":"Patch Set 11:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/3803/ : FAILURE","accounts_in_message":[],"_revision_number":11},{"id":"817bcd22590e21f875db6066d2f0ac61fb438c58","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 04:14:26.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/0fd2736804554b6c91f69a663cd0131b : FAILURE in 1h 20m 11s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/6767eb82c5ba46ed90d482aec91dde87 : FAILURE in 2h 03m 03s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/4cdad8bf78124f19ab3f776f38d5725a : FAILURE in 1h 22m 31s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/def74e589d21440d97fe1d1a2f296c86 : FAILURE in 1h 03m 58s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/01d84ea1eda343489a491a0c4fee8313 : FAILURE in 1h 05m 02s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/b16512edad93481e97df5398912f11c4 : FAILURE in 1h 04m 29s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/4ef8834548d5469a83dfbbfef235dae3 : FAILURE in 1h 19m 27s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/d521668654004896b9e08839bd2f3ee3 : FAILURE in 1h 08m 19s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/491b565743764eaaa15e9c43c313fa58 : SUCCESS in 2h 11m 20s (non-voting)\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/85058bc9f65546f4bc78b61e5abe4a28 : FAILURE in 1h 07m 15s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/7c8cb01bdd564f41b0c9b738959eb76b : SUCCESS in 2h 07m 55s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/b405ebcad9394eea99dada6d7e0d1ae0 : SUCCESS in 2h 30m 39s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/329e3f2d5ac04066890cd8334e079c88 : FAILURE in 7m 27s\n- tripleo-ci-centos-7-containers-multinode-pike https://zuul.opendev.org/t/openstack/build/043725bd52a04087ab48195202b64c97 : SUCCESS in 2h 09m 13s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6c9048503a0844e68b91ae1933a7efb8 : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":11},{"id":"a396b5a65ae1d6e94b6b5b5c800845c8ed3aab42","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-19 04:34:17.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/70/664170/11/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/f78df68/ : FAILURE in 46m 06s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/70/664170/11/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/c27749e/ : FAILURE in 48m 16s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/70/664170/11/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/ce65e1b/ : FAILURE in 2h 22m 20s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/70/664170/11/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/510b52a/ : SUCCESS in 2h 18m 14s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/70/664170/11/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/5a75c2a/ : SUCCESS in 2h 50m 15s","accounts_in_message":[],"_revision_number":11},{"id":"2439d59d1e6f03f0bd4a64e8ee61adadda24cd02","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2019-08-19 20:30:12.000000000","message":"Patch Set 12: Commit message was updated.","accounts_in_message":[],"_revision_number":12},{"id":"b62850857ea21547e1a7df49a95abf2e5e45c4fd","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-08-19 20:30:29.000000000","message":"Patch Set 12:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/3809/ : FAILURE","accounts_in_message":[],"_revision_number":12},{"id":"ddbcaab3a09ea183c1b2560313eda10c7dee2d17","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 21:45:17.000000000","message":"Patch Set 12: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":12},{"id":"346385ec120f15ab7878b94deb2e16ccea5ddaab","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-19 21:45:17.000000000","message":"Patch Set 12: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":12},{"id":"6cc1f3ad8f245d89e40b68af41896b9d4346e924","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2019-08-19 21:47:07.000000000","message":"Patch Set 13: Commit message was updated.","accounts_in_message":[],"_revision_number":13},{"id":"2587d6de4a2a614a7f0c3d8428e02d815381888b","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2019-08-19 21:49:13.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"dd59e0bcebfc0cfded8908377f8abadffc7d148f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 01:09:29.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/595504aa720c4068bce02f80a149582f : FAILURE in 1h 14m 39s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/15e231a6efe048449047970dd90ce868 : FAILURE in 2h 12m 31s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/16b774ce30a747dbbf75b772c00c2b5e : FAILURE in 1h 15m 11s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/70c8b79c764a4d1db36f126c2c23fb3a : FAILURE in 1h 13m 33s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/f69c8f16d2ca4a2d8757c6b2f1265c20 : FAILURE in 1h 13m 50s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/8c669507613b4e408653d8d720ffec0d : FAILURE in 1h 16m 33s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/ca3d42d2c56e46979555432037a984cc : FAILURE in 1h 07m 10s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/ff5ac1edb09749b1a588c5533fc9f5ff : SUCCESS in 1h 59m 43s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/62be88cb5942463c9eefa71267f603ac : SUCCESS in 2h 33m 10s (non-voting)\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/679774e1dc934527a034b86c0d958dfc : SUCCESS in 1h 28m 13s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/5e34dd6054fa4ab5ad954c6906860a7b : SUCCESS in 2h 12m 14s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/77c8575c0c3a45e68791c74273f1aa1e : SUCCESS in 2h 19m 26s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/83f28941c7124bef9c8158f4fd66e528 : SUCCESS in 1h 46m 26s\n- tripleo-ci-centos-7-containers-multinode-pike https://zuul.opendev.org/t/openstack/build/3ce73bd347b2472391a715f2518e10ba : SUCCESS in 2h 25m 30s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/258b736036e04a03bc96e796df4c9fb7 : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":13},{"id":"7466ae86ab01cae703bfdd8e8dc519a6493cc7ba","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-20 01:34:38.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/70/664170/13/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/3f7a8ab/ : FAILURE in 48m 46s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/70/664170/13/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/0e0b8a8/ : FAILURE in 1h 02m 41s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/70/664170/13/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/288a555/ : SUCCESS in 2h 43m 56s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/70/664170/13/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/cb0a22d/ : SUCCESS in 3h 07m 30s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/70/664170/13/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/ef73f71/ : FAILURE in 3h 45m 37s","accounts_in_message":[],"_revision_number":13},{"id":"4b945c744ba42cb03a0f2073f502435da67217c6","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2019-08-21 04:36:49.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"910e4eb4dd048b8e0fd055b7cb6f7caafef37e57","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 05:17:15.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/c47acc473d35423e9495d21dd3160ca6 : FAILURE in 9m 56s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/495921e20d564b7fbd81db16c98546d6 : FAILURE in 9m 34s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/de7bed1dfa40480683b56d7613af0fbd : FAILURE in 12m 06s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/02b6232986824fd095ad161a74110d41 : FAILURE in 12m 18s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/c59c8dcfa0144a11a5366176cac7bc2c : FAILURE in 12m 33s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/1129aadac0cf4bb19594ceda23967508 : FAILURE in 13m 44s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/1dd1c16bf7a14caba4539ad5c3d95d0c : FAILURE in 12m 44s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/fd074c7b5b4e4704b2bbc99314e59d42 : FAILURE in 8m 25s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/49e7d8de524d4424b1708a54112f06b2 : FAILURE in 9m 22s (non-voting)\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/0213f74b90604600b2bd7979b5e2610a : FAILURE in 8m 37s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/dfd3ed601e1c43b297bcd5f0967c52e5 : FAILURE in 11m 43s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/611bafe65a8049788d6df08d71304c47 : FAILURE in 12m 29s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/a22759f8d98f44ab89947f5e3bca7d2e : FAILURE in 10m 14s\n- tripleo-ci-centos-7-containers-multinode-pike https://zuul.opendev.org/t/openstack/build/9ab9c5f398b6428bb814888b2cac2d90 : FAILURE in 13m 05s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/223211480ed346b797f7b7cfa1e5030f : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":13},{"id":"40e6d77b5629217a116448435da6e6684c3cba36","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-21 07:41:04.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/70/664170/13/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/123293a/ : FAILURE in 1h 03m 29s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/70/664170/13/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/5bcbc4d/ : FAILURE in 1h 03m 30s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/70/664170/13/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/9b3081d/ : FAILURE in 2h 25m 26s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/70/664170/13/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/08b3f73/ : SUCCESS in 2h 20m 03s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/70/664170/13/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/06aaa64/ : SUCCESS in 3h 03m 02s","accounts_in_message":[],"_revision_number":13},{"id":"dbca3c2c3c0ae2cd543b92bad3621ab126cb2841","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-08-21 13:49:59.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"52d6a7a0b7cfb8b27ba0d52256fc3153a4f293cf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 16:37:37.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/49483b74940b4de8b31f50b3db3d13ae : FAILURE in 1h 29m 39s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/29add850cab24913b5e4d1fbcf2161a6 : FAILURE in 1h 39m 04s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/2bed89d0b39b438485857145e1b3efc1 : FAILURE in 1h 38m 58s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/55c315ed651a44c7ab0c824c30d16e4d : FAILURE in 1h 28m 33s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/2e9ed8769db64daca5a51b9e664c6dd2 : FAILURE in 1h 34m 11s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/d190b15c75d54a28adb1650f5a6c9fe7 : FAILURE in 1h 28m 15s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/41e39735fab54421b168e6b9498d0440 : FAILURE in 1h 20m 45s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/c4fb02e9c2274fec9bf92c757bfa0cf8 : SUCCESS in 2h 02m 38s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/2867f0c10cfa4c24acfc064e494006fc : SUCCESS in 2h 32m 24s (non-voting)\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/672f901d9cf045c2a33b8249c57b8062 : SUCCESS in 1h 27m 24s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/df85e78c09e04f678bd6b1933b79d58f : SUCCESS in 2h 19m 31s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/f679807af0194b64af1a82ddb207c6e9 : SUCCESS in 2h 37m 04s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/a7755d8bee2b4816a9272cdd476f72b9 : SUCCESS in 2h 20m 09s\n- tripleo-ci-centos-7-containers-multinode-pike https://zuul.opendev.org/t/openstack/build/9d6fe7dc09ec4458a23026ebd3ec5223 : SUCCESS in 2h 27m 34s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/30be9ef850c04c679c5ec1c2caaa4b58 : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":13},{"id":"81aeb81d11142d0d5a128b17a1d92b5e2bc27e8b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-21 17:01:20.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/70/664170/13/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/c494d8d/ : FAILURE in 1h 41m 02s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/70/664170/13/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/4e8a58b/ : FAILURE in 1h 41m 47s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/70/664170/13/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/ab25375/ : FAILURE in 2h 32m 41s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/70/664170/13/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/87d7b15/ : SUCCESS in 2h 50m 07s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/70/664170/13/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/815d751/ : SUCCESS in 3h 09m 12s","accounts_in_message":[],"_revision_number":13},{"id":"d42821d43042eb131b176968155d4071e4fb3cfc","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2019-08-21 22:18:40.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"ba1765caa4aeea64b91fa73fbba9992734d35939","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-21 23:32:16.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/70/664170/14/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/55af5a3/ : FAILURE in 56m 38s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/70/664170/14/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/1ccaec7/ : FAILURE in 57m 32s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/70/664170/14/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/6eba01b/ : FAILURE in 45m 58s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/70/664170/14/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/2350e91/ : FAILURE in 1h 10m 40s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/70/664170/14/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/7868f5c/ : FAILURE in 1h 07m 13s","accounts_in_message":[],"_revision_number":14},{"id":"f76ac86a1e488c452cb0b77e41b289876cb1cd2b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 00:44:14.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/6b7292577192412599c1ab0574e7b675 : FAILURE in 1h 40m 47s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/68a09108ce8541b5a9ba84fc824f3009 : FAILURE in 1h 11m 17s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/abe35d5e075a450b9436ba8c3546ce38 : FAILURE in 1h 21m 59s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/bb9da7a7447646108d2421c9002b0ac7 : FAILURE in 1h 31m 31s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/720b2c00c7b94f168387de0d0a48cbf5 : FAILURE in 1h 34m 29s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/1a70b081499043c997855ad8bef3ebe4 : FAILURE in 1h 34m 31s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/dbdedd028aa84e83a12711adbac69a9a : FAILURE in 1h 38m 03s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/f1164a852455410fa178a2ed0c241951 : SUCCESS in 1h 53m 13s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/8ac25dd679474b5a9be8f98a888231d4 : SUCCESS in 2h 06m 59s (non-voting)\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/f60f41d2365e4526b4221d6403304563 : SUCCESS in 1h 27m 19s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/9b348f0384e5405fa22695fea77c2824 : FAILURE in 1h 22m 28s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/28c2eb585f814cd289164c0eff594295 : FAILURE in 1h 27m 37s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/f7240c6d433c4e62992861cc571ffd25 : FAILURE in 52m 28s\n- tripleo-ci-centos-7-containers-multinode-pike https://zuul.opendev.org/t/openstack/build/8e2b6192113f47c89fe7b0eb4512c202 : FAILURE in 52m 22s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2b533551e0f54ce491dd9d0302bcba5f : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":14},{"id":"277e7adddb0a8dc2c22f25a1fac6ab065266086c","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2019-08-22 22:38:22.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"6254dde86a265b758cb292affa5f856c897a4fd1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-23 00:23:38.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/70/664170/15/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/5198fd3/ : FAILURE in 1h 13m 37s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/70/664170/15/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/d81e6a4/ : FAILURE in 1h 05m 46s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/70/664170/15/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/8dc4ef2/ : FAILURE in 1h 34m 04s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/70/664170/15/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/aa3f1f0/ : FAILURE in 1h 27m 54s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/70/664170/15/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/27a40e0/ : FAILURE in 1h 43m 10s","accounts_in_message":[],"_revision_number":15},{"id":"40e6ebb6d774acb360838b7e7630878f974f548a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 01:14:41.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/a8d3f1f1d46849228c6b8e135563499d : FAILURE in 1h 29m 21s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/0e079ef221554d9c8c83f6da1562da93 : FAILURE in 2h 07m 42s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/58cc2f99cee9489ca8fb8877315ee4d0 : FAILURE in 1h 26m 37s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/280bb1a8c28d4482b42e8452a2667619 : FAILURE in 1h 19m 00s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/1d69243b2e8d4ee9bb07a2b29cf7c6d6 : FAILURE in 1h 22m 12s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/3aecbafa40ac4ae2ad0593d387f13bd8 : FAILURE in 1h 30m 50s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/22e8594df6fe40fc894b50eebbd03e69 : FAILURE in 1h 26m 11s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/89fc15416360426aa5f6b7f0eb4a08b8 : SUCCESS in 2h 15m 53s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/32af1a63633748758c84d1deed8cf161 : SUCCESS in 2h 18m 34s (non-voting)\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/374fbd735f4f4ccf8358198eb1452f70 : SUCCESS in 1h 43m 39s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/aa1a2bda3cb848eeafd47d59e69a0f09 : FAILURE in 1h 42m 53s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/1b8ee80571384b6983fec33445278416 : FAILURE in 1h 37m 29s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/bca796f089fd49f7a4372be42afb0a2d : FAILURE in 1h 15m 11s\n- tripleo-ci-centos-7-containers-multinode-pike https://zuul.opendev.org/t/openstack/build/92b94eba910c41c0a346272f07eb4a8f : FAILURE in 1h 09m 03s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/514bf1209ba642b4ae50ec6bd3d564b1 : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":15},{"id":"661b3ab3b5ffa2ba880159a3fb5350baf5ee24f2","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2019-08-27 03:53:51.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"425a01c171e995a4f0467f6eccf08dfedfa6ea29","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 06:51:38.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/d0f44225af8a492e8485c659c350ac20 : FAILURE in 1h 18m 26s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/664f3818e2e54d5b9c9174c4a4ddbeb5 : FAILURE in 1h 54m 21s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/1791cafea597400b86838e5adde914f0 : FAILURE in 1h 22m 03s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/4dc7eacbf82945198757ccb0bf7ba5b6 : FAILURE in 1h 24m 38s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/99ea9bd2cd1e4e02be24d3faa895c1c8 : FAILURE in 1h 10m 31s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/25b36d33bfef46a5a56cebb633c96b11 : FAILURE in 1h 25m 28s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/5786f241618242cc986cc8959d86b0b2 : FAILURE in 1h 16m 24s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/785fc8ba45244193b20ad6009cbe9a70 : SUCCESS in 1h 48m 26s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/f5f59404315c4cdeb41396c2d44400b4 : SUCCESS in 2h 21m 26s (non-voting)\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/c75b309658c44a78a0fdfa0ff125ede9 : SUCCESS in 1h 29m 01s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/63fc36a29d7f45cfa2f5ae5cc957fe0f : SUCCESS in 2h 11m 21s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/57a335b04ed942e4a8ab0258f0e62f68 : SUCCESS in 2h 30m 50s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/588b69908ba942c88b6921ef8fdeda51 : SUCCESS in 2h 04m 17s\n- tripleo-ci-centos-7-containers-multinode-pike https://zuul.opendev.org/t/openstack/build/6ac1abac999b4a74a29b8d4e0cec4993 : SUCCESS in 2h 25m 48s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/908066aee7a64334be3c23d0dba7bd87 : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":16},{"id":"c89834b3a91922b825b984871dd87da2c3c65732","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-27 07:14:51.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/70/664170/16/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/ede70aa/ : FAILURE in 1h 10m 00s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/70/664170/16/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/59a1f92/ : FAILURE in 1h 06m 13s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/70/664170/16/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/790eec7/ : SUCCESS in 2h 17m 35s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/70/664170/16/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/98e6a93/ : SUCCESS in 2h 24m 47s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/70/664170/16/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/eab3772/ : SUCCESS in 3h 19m 25s","accounts_in_message":[],"_revision_number":16},{"id":"e8da982e18d36be802e58f54fab0cf97104bb895","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2019-09-06 01:33:49.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"bc3ac46789cb644e4229615aef517e58ae759b46","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2019-09-06 04:14:32.000000000","message":"Abandoned\n\nmoving to featureset files","accounts_in_message":[],"_revision_number":17}],"current_revision_number":17,"current_revision":"05f2f57bbafbf7d10072fa75ad73e8fa54b3c83f","revisions":{"01d7849bb0ec5846e94f3a69f441750cfeedb4d4":{"kind":"REWORK","_number":1,"created":"2019-06-09 23:18:51.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/70/664170/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/70/664170/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/1"}}},"commit":{"parents":[{"commit":"0081df5ccfbb75fdbf5b004ea8b33dc26d716208","subject":"Write out overcloud_baremetal_deploy.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/0081df5ccfbb75fdbf5b004ea8b33dc26d716208"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-06-09 23:14:55.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-06-09 23:14:55.000000000","tz":720},"subject":"DNM: undercloud_enable_nova: false by default","message":"DNM: undercloud_enable_nova: false by default\n\nChange-Id: I1d62928b9572e9dae6c1f0eff064be29d187c620\nBlueprint: nova-less-deploy\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/01d7849bb0ec5846e94f3a69f441750cfeedb4d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/01d7849bb0ec5846e94f3a69f441750cfeedb4d4"}]},"branch":"refs/heads/master"},"c2e73898c82f496f11fe4307e231eb16d5b1cdc5":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-06-10 01:48:12.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/70/664170/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/70/664170/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/2"}}},"commit":{"parents":[{"commit":"89e501b785175d1b2da3b57bc94333919e454f6e","subject":"Write out overcloud_baremetal_deploy.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/89e501b785175d1b2da3b57bc94333919e454f6e"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-06-09 23:14:55.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-06-10 01:47:53.000000000","tz":720},"subject":"DNM: undercloud_enable_nova: false by default","message":"DNM: undercloud_enable_nova: false by default\n\nChange-Id: I1d62928b9572e9dae6c1f0eff064be29d187c620\nBlueprint: nova-less-deploy\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/c2e73898c82f496f11fe4307e231eb16d5b1cdc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/c2e73898c82f496f11fe4307e231eb16d5b1cdc5"}]},"branch":"refs/heads/master"},"0013b0f01068c97180ffaa2b61f3c0e464bdf5f2":{"kind":"REWORK","_number":3,"created":"2019-06-10 22:40:38.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/70/664170/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/70/664170/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/3"}}},"commit":{"parents":[{"commit":"57107e520c9f29ce1acebca9ba2369f7e90da1d3","subject":"Write out overcloud_baremetal_deploy.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/57107e520c9f29ce1acebca9ba2369f7e90da1d3"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-06-10 21:27:29.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-06-10 22:38:47.000000000","tz":720},"subject":"DNM: undercloud_enable_nova: false by default","message":"DNM: undercloud_enable_nova: false by default\n\nChange-Id: I1d62928b9572e9dae6c1f0eff064be29d187c620\nBlueprint: nova-less-deploy\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/0013b0f01068c97180ffaa2b61f3c0e464bdf5f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/0013b0f01068c97180ffaa2b61f3c0e464bdf5f2"}]},"branch":"refs/heads/master"},"8f6731aa35465c77ac73cea11281e2c6a45e6ac5":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-06-11 04:11:58.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/70/664170/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/70/664170/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/4"}}},"commit":{"parents":[{"commit":"157e989100ed767388f201ad508ff6b3c611a7b5","subject":"Support direct provisioning of bare metal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/157e989100ed767388f201ad508ff6b3c611a7b5"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-06-10 21:27:29.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-06-11 04:08:56.000000000","tz":720},"subject":"DNM: undercloud_enable_nova: false by default","message":"DNM: undercloud_enable_nova: false by default\n\nChange-Id: I1d62928b9572e9dae6c1f0eff064be29d187c620\nBlueprint: nova-less-deploy\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/8f6731aa35465c77ac73cea11281e2c6a45e6ac5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/8f6731aa35465c77ac73cea11281e2c6a45e6ac5"}]},"branch":"refs/heads/master"},"22fc9f973fb1556552ee171d8ddbaed96a28e3c7":{"kind":"NO_CHANGE","_number":5,"created":"2019-06-11 04:12:27.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/70/664170/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/70/664170/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/5"}}},"commit":{"parents":[{"commit":"def76bc2016ce3678f136ac95703e5ea822a88d4","subject":"Define var undercloud_enable_nova","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/def76bc2016ce3678f136ac95703e5ea822a88d4"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-06-10 21:27:29.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-06-11 04:12:16.000000000","tz":720},"subject":"DNM: undercloud_enable_nova: false by default","message":"DNM: undercloud_enable_nova: false by default\n\nChange-Id: I1d62928b9572e9dae6c1f0eff064be29d187c620\nBlueprint: nova-less-deploy\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/22fc9f973fb1556552ee171d8ddbaed96a28e3c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/22fc9f973fb1556552ee171d8ddbaed96a28e3c7"}]},"branch":"refs/heads/master"},"1296ee3f2a146a0721472738e18b86d4cd4ef588":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-06-11 04:21:46.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/70/664170/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/70/664170/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/6"}}},"commit":{"parents":[{"commit":"6fb6deb14a52c510dd242f0f0b04966280da8649","subject":"Define var undercloud_enable_nova","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/6fb6deb14a52c510dd242f0f0b04966280da8649"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-06-10 21:27:29.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-06-11 04:21:34.000000000","tz":720},"subject":"DNM: undercloud_enable_nova: false by default","message":"DNM: undercloud_enable_nova: false by default\n\nChange-Id: I1d62928b9572e9dae6c1f0eff064be29d187c620\nBlueprint: nova-less-deploy\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/1296ee3f2a146a0721472738e18b86d4cd4ef588"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/1296ee3f2a146a0721472738e18b86d4cd4ef588"}]},"branch":"refs/heads/master"},"66942467988f2511bd700520f66c9ca3d735e1fd":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-06-16 21:32:29.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/70/664170/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/70/664170/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/7"}}},"commit":{"parents":[{"commit":"a22b36a57e2779c51ceab31e8c0ba4a6a9eb6316","subject":"Define var undercloud_enable_nova","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/a22b36a57e2779c51ceab31e8c0ba4a6a9eb6316"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-06-10 21:27:29.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-06-16 21:31:58.000000000","tz":720},"subject":"DNM: undercloud_enable_nova: false by default","message":"DNM: undercloud_enable_nova: false by default\n\nChange-Id: I1d62928b9572e9dae6c1f0eff064be29d187c620\nBlueprint: nova-less-deploy\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/66942467988f2511bd700520f66c9ca3d735e1fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/66942467988f2511bd700520f66c9ca3d735e1fd"}]},"branch":"refs/heads/master"},"22149917d01d817fac860c4582918b418118bc28":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2019-08-15 02:33:39.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/70/664170/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/70/664170/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/8"}}},"commit":{"parents":[{"commit":"fdafa8ad1d62efb26b570fa8416f5cc800eaf073","subject":"Define var undercloud_enable_nova","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/fdafa8ad1d62efb26b570fa8416f5cc800eaf073"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-06-10 21:27:29.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-08-15 02:33:10.000000000","tz":720},"subject":"DNM: undercloud_enable_nova: false by default","message":"DNM: undercloud_enable_nova: false by default\n\nChange-Id: I1d62928b9572e9dae6c1f0eff064be29d187c620\nBlueprint: nova-less-deploy\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/22149917d01d817fac860c4582918b418118bc28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/22149917d01d817fac860c4582918b418118bc28"}]},"branch":"refs/heads/master"},"2aa9b4d314b8359c913475135ce5f177478d5a48":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2019-08-15 03:31:30.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/70/664170/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/70/664170/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/9"}}},"commit":{"parents":[{"commit":"fdafa8ad1d62efb26b570fa8416f5cc800eaf073","subject":"Define var undercloud_enable_nova","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/fdafa8ad1d62efb26b570fa8416f5cc800eaf073"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-06-10 21:27:29.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-08-15 03:24:20.000000000","tz":720},"subject":"DNM: undercloud_enable_nova: false by default","message":"DNM: undercloud_enable_nova: false by default\n\nChange-Id: I1d62928b9572e9dae6c1f0eff064be29d187c620\nBlueprint: nova-less-deploy\nDepends-On: https://review.opendev.org/#/c/675204/\nDepends-On: https://review.opendev.org/#/c/675497/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/2aa9b4d314b8359c913475135ce5f177478d5a48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/2aa9b4d314b8359c913475135ce5f177478d5a48"}]},"branch":"refs/heads/master"},"033f07e374db68e6c54d125d173f21f431cbfb3e":{"kind":"NO_CHANGE","_number":10,"created":"2019-08-18 21:11:10.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/70/664170/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/70/664170/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/10"}}},"commit":{"parents":[{"commit":"b4e628e2c83a6ce1af1ec407340dd3a5f03b2314","subject":"Support direct provisioning of bare metal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/b4e628e2c83a6ce1af1ec407340dd3a5f03b2314"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-06-10 21:27:29.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-08-18 21:10:38.000000000","tz":720},"subject":"DNM: undercloud_enable_nova: false by default","message":"DNM: undercloud_enable_nova: false by default\n\nChange-Id: I1d62928b9572e9dae6c1f0eff064be29d187c620\nBlueprint: nova-less-deploy\nDepends-On: https://review.opendev.org/#/c/675204/\nDepends-On: https://review.opendev.org/#/c/675497/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/033f07e374db68e6c54d125d173f21f431cbfb3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/033f07e374db68e6c54d125d173f21f431cbfb3e"}]},"branch":"refs/heads/master"},"f0bcf5915da743d205ea82e20fc36712b654ec37":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2019-08-19 01:42:05.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/70/664170/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/70/664170/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/11"}}},"commit":{"parents":[{"commit":"b4e628e2c83a6ce1af1ec407340dd3a5f03b2314","subject":"Support direct provisioning of bare metal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/b4e628e2c83a6ce1af1ec407340dd3a5f03b2314"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-06-10 21:27:29.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-08-19 01:42:02.000000000","tz":0},"subject":"DNM: undercloud_enable_nova: false by default","message":"DNM: undercloud_enable_nova: false by default\n\nChange-Id: I1d62928b9572e9dae6c1f0eff064be29d187c620\nBlueprint: nova-less-deploy\nDepends-On: https://review.opendev.org/675204\nDepends-On: https://review.opendev.org/675497\nDepends-On: https://review.opendev.org/677099\nDepends-On: https://review.opendev.org/677100\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/f0bcf5915da743d205ea82e20fc36712b654ec37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/f0bcf5915da743d205ea82e20fc36712b654ec37"}]},"branch":"refs/heads/master"},"4acd20028eb0849cdebbb86cf937421accf0c808":{"kind":"NO_CODE_CHANGE","_number":12,"created":"2019-08-19 20:30:12.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/70/664170/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/70/664170/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/12"}}},"commit":{"parents":[{"commit":"b4e628e2c83a6ce1af1ec407340dd3a5f03b2314","subject":"Support direct provisioning of bare metal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/b4e628e2c83a6ce1af1ec407340dd3a5f03b2314"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-06-10 21:27:29.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-08-19 04:17:35.000000000","tz":0},"subject":"DNM: undercloud_enable_nova: false by default","message":"DNM: undercloud_enable_nova: false by default\n\nChange-Id: I1d62928b9572e9dae6c1f0eff064be29d187c620\nBlueprint: nova-less-deploy\nDepends-On: https://review.opendev.org/675204\nDepends-On: https://review.opendev.org/675497\nDepends-On: https://review.opendev.org/677099\nDepends-On: https://review.opendev.org/677100\nDepends-On: https://review.opendev.org/677106\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/4acd20028eb0849cdebbb86cf937421accf0c808"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/4acd20028eb0849cdebbb86cf937421accf0c808"}]},"branch":"refs/heads/master"},"acc124aca907355e615754485a9a949166ca486f":{"kind":"NO_CODE_CHANGE","_number":13,"created":"2019-08-19 21:47:07.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/70/664170/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/70/664170/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/13"}}},"commit":{"parents":[{"commit":"b4e628e2c83a6ce1af1ec407340dd3a5f03b2314","subject":"Support direct provisioning of bare metal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/b4e628e2c83a6ce1af1ec407340dd3a5f03b2314"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-06-10 21:27:29.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-08-19 21:46:58.000000000","tz":0},"subject":"DNM: undercloud_enable_nova: false by default","message":"DNM: undercloud_enable_nova: false by default\n\nChange-Id: I1d62928b9572e9dae6c1f0eff064be29d187c620\nBlueprint: nova-less-deploy\nDepends-On: https://review.opendev.org/675204\nDepends-On: https://review.opendev.org/675497\nDepends-On: https://review.opendev.org/677099\nDepends-On: https://review.opendev.org/677100\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/acc124aca907355e615754485a9a949166ca486f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/acc124aca907355e615754485a9a949166ca486f"}]},"branch":"refs/heads/master"},"e43e0f50d995f27601544361ed4bec510888198f":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":14,"created":"2019-08-21 22:18:40.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/70/664170/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/70/664170/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/14"}}},"commit":{"parents":[{"commit":"0f687fe25d1af5022a5e2e417f30c90b87c4145a","subject":"Support direct provisioning of bare metal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/0f687fe25d1af5022a5e2e417f30c90b87c4145a"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-06-10 21:27:29.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-08-21 22:12:02.000000000","tz":720},"subject":"DNM: undercloud_enable_nova: false by default","message":"DNM: undercloud_enable_nova: false by default\n\nChange-Id: I1d62928b9572e9dae6c1f0eff064be29d187c620\nBlueprint: nova-less-deploy\nDepends-On: https://review.opendev.org/675204\nDepends-On: https://review.opendev.org/675497\nDepends-On: https://review.opendev.org/677099\nDepends-On: https://review.opendev.org/677100\nDepends-On: https://review.opendev.org/677106\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/e43e0f50d995f27601544361ed4bec510888198f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/e43e0f50d995f27601544361ed4bec510888198f"}]},"branch":"refs/heads/master"},"39fba35c8c9a90abc38ee89bd820ddc06876c950":{"kind":"REWORK","_number":15,"created":"2019-08-22 22:38:22.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/70/664170/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/70/664170/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/15"}}},"commit":{"parents":[{"commit":"3439ac61b383e66ac26793373affb0c9ecf3e653","subject":"DNM: baremetal_provision:true by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/3439ac61b383e66ac26793373affb0c9ecf3e653"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-08-22 22:12:02.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-08-22 22:36:19.000000000","tz":720},"subject":"DNM: undercloud_enable_nova: false by default","message":"DNM: undercloud_enable_nova: false by default\n\nChange-Id: I1d62928b9572e9dae6c1f0eff064be29d187c620\nBlueprint: nova-less-deploy\nDepends-On: https://review.opendev.org/675497\nDepends-On: https://review.opendev.org/677100\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/39fba35c8c9a90abc38ee89bd820ddc06876c950"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/39fba35c8c9a90abc38ee89bd820ddc06876c950"}]},"branch":"refs/heads/master"},"6381414e7fedf551eba2698430ebe5f98c21601e":{"kind":"REWORK","_number":16,"created":"2019-08-27 03:53:51.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/70/664170/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/70/664170/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/16"}}},"commit":{"parents":[{"commit":"51acc09d6e18d950f94db205d96da4df620a66d0","subject":"DNM: baremetal_provision:true by default for Train onwards","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/51acc09d6e18d950f94db205d96da4df620a66d0"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-08-22 22:12:02.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-08-27 03:42:09.000000000","tz":720},"subject":"DNM: undercloud_enable_nova: false by default for Train onwards","message":"DNM: undercloud_enable_nova: false by default for Train onwards\n\nChange-Id: I1d62928b9572e9dae6c1f0eff064be29d187c620\nBlueprint: nova-less-deploy\nDepends-On: https://review.opendev.org/677100\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/6381414e7fedf551eba2698430ebe5f98c21601e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/6381414e7fedf551eba2698430ebe5f98c21601e"}]},"branch":"refs/heads/master"},"05f2f57bbafbf7d10072fa75ad73e8fa54b3c83f":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":17,"created":"2019-09-06 01:33:49.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/70/664170/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/70/664170/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/664170/17"}}},"commit":{"parents":[{"commit":"1d8c6afc113742b55d50e5fdf49ce4c3f7c1c344","subject":"DNM: baremetal_provision:true by default for Train onwards","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/1d8c6afc113742b55d50e5fdf49ce4c3f7c1c344"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-08-22 22:12:02.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-09-06 01:33:10.000000000","tz":720},"subject":"DNM: undercloud_enable_nova: false by default for Train onwards","message":"DNM: undercloud_enable_nova: false by default for Train onwards\n\nChange-Id: I1d62928b9572e9dae6c1f0eff064be29d187c620\nBlueprint: nova-less-deploy\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/05f2f57bbafbf7d10072fa75ad73e8fa54b3c83f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/05f2f57bbafbf7d10072fa75ad73e8fa54b3c83f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
