)]}'
{"id":"openstack%2Ftripleo-quickstart-extras~664456","triplet_id":"openstack%2Ftripleo-quickstart-extras~master~I57893a64f4c3224ffe8bd12330f12bd9d3cbbf89","project":"openstack/tripleo-quickstart-extras","branch":"master","topic":"bp/nova-less-deploy","hashtags":[],"change_id":"I57893a64f4c3224ffe8bd12330f12bd9d3cbbf89","subject":"Support direct provisioning of bare metal","status":"MERGED","created":"2019-06-10 22:40:38.000000000","updated":"2019-09-19 17:34:33.000000000","submitted":"2019-09-19 17:34:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664456-1568914473918-24b88b4b","meta_rev_id":"45c009471c89bf4c740f95ecabb25e587ed0a17b","_number":664456,"virtual_id_number":664456,"owner":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":2,"date":"2019-09-19 17:34:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-09-15 16:47:38.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":2,"date":"2019-09-15 20:47:17.000000000","_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-09-16 22:16:48.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"date":"2019-09-15 22:31:57.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-11 04:26:14.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2019-06-11 08:16:13.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2019-06-11 18:35:55.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"REVIEWER"},{"updated":"2019-08-18 21:11:32.000000000","updated_by":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-15 16:47:38.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-15 20:47:17.000000000","updated_by":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"reviewer":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"state":"REVIEWER"},{"updated":"2019-09-15 22:31:57.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-09-16 22:16:48.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-09-19 17:34:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"25160bbc9219bdcf062f2b6d3624d46ee5477c2f","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2019-06-10 22:40:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"272a7153d96c5c25225a90fb2b2fae460a167d05","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-06-10 22:40:58.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/3506/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"5167b16f6b1c9acc177d309fc07516856efeec70","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-11 01:12:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/56/664456/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/523c328/ : FAILURE in 50m 36s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/56/664456/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/2f3ce5e/ : FAILURE in 50m 43s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/56/664456/1/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/0736d9c/ : FAILURE in 33m 22s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/56/664456/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/8b3de2b/ : SUCCESS in 2h 25m 33s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/56/664456/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/0644341/ : SUCCESS in 2h 19m 43s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/56/664456/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/7ff9cd5/ : SUCCESS in 2h 26m 54s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/56/664456/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/631d87c/ : FAILURE in 48m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c8adaf0f1e7d2cfbd5633f9b40f74838c852bc20","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2019-06-11 04:11:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6642347221ea9f538a955f97a80c9dbfdeb77ecb","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-06-11 04:12:13.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/3507/ : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"74a80389e3556c4385aa077828ae3772126f1ac1","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2019-06-11 04:12:27.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"7197325f895c42757019f0c6648fd66f6cbda676","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-06-11 04:12:43.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/3509/ : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"1689491e5c9d17f20f086e02cdedae7dfbeadc49","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2019-06-11 04:18:57.000000000","message":"Patch Set 3:\n\nThis is exercised in https://review.opendev.org/#/c/664457/","accounts_in_message":[],"_revision_number":3},{"id":"fbb365d7ec47d2b4ab9aa8d54462cf8be1f3e0fe","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2019-06-11 04:21:46.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2456acbf628d81734badc815a7cfdf1de5327cf2","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-06-11 04:22:01.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/3512/ : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"8dfc6a2f32c0c3538d988647c59e0deb60ae3dcd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 06:54:29.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/56/664456/4/check/tripleo-ci-centos-7-undercloud-containers/07915d8/ : SUCCESS in 1h 38m 21s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/56/664456/4/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/8ed7d2d/ : FAILURE in 1h 22m 35s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/56/664456/4/check/tripleo-ci-centos-7-containers-multinode/8ef37ea/ : SUCCESS in 2h 00m 42s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/56/664456/4/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/bdb03e8/ : SUCCESS in 2h 21m 59s (non-voting)\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container http://logs.openstack.org/56/664456/4/check/tripleo-ci-centos-7-scenario008-multinode-oooq-container/deb81c1/ : FAILURE in 16m 48s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container http://logs.openstack.org/56/664456/4/check/tripleo-ci-centos-7-scenario012-multinode-oooq-container/3314f1d/ : FAILURE in 2h 10m 40s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/56/664456/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/3a8e9cd/ : SUCCESS in 2h 22m 55s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/56/664456/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/40efae6/ : FAILURE in 18m 34s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/56/664456/4/check/tripleo-ci-centos-7-standalone/2cec45a/ : FAILURE in 16m 15s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/56/664456/4/check/tripleo-ci-fedora-28-standalone/9af0252/ : SUCCESS in 1h 41m 57s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/56/664456/4/check/tripleo-ci-centos-7-standalone-upgrade/1cf5a73/ : FAILURE in 13m 52s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/56/664456/4/check/tripleo-ci-centos-7-standalone-upgrade-stein/ed99811/ : FAILURE in 15m 14s\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/56/664456/4/check/tripleo-ci-centos-7-standalone-os-tempest/cd6269e/ : FAILURE in 15m 43s\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/56/664456/4/check/tripleo-ci-centos-7-containers-multinode-stein/5162b73/ : SUCCESS in 2h 00m 49s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/56/664456/4/check/tripleo-ci-centos-7-containers-multinode-rocky/f430e87/ : SUCCESS in 2h 10m 56s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/56/664456/4/check/tripleo-ci-centos-7-containers-multinode-queens/6f00806/ : SUCCESS in 2h 06m 34s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/56/664456/4/check/tripleo-ci-centos-7-containers-multinode-pike/3755639/ : SUCCESS in 2h 17m 40s\n- openstack-tox-linters http://logs.openstack.org/56/664456/4/check/openstack-tox-linters/9536022/ : FAILURE in 4m 17s","accounts_in_message":[],"_revision_number":4},{"id":"eaca62f9b18bd6ac768b911a9ff8351ad38bde34","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-11 07:14:55.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/56/664456/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/5fd8ae1/ : SUCCESS in 2h 36m 38s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/56/664456/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/bcc985a/ : SUCCESS in 2h 51m 41s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/56/664456/4/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/c5633ac/ : SUCCESS in 1h 14m 04s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/56/664456/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/1c9fc0c/ : SUCCESS in 2h 03m 03s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/56/664456/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/187b2b0/ : SUCCESS in 2h 08m 49s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/56/664456/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/899d74c/ : SUCCESS in 2h 18m 50s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/56/664456/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/d71aa6c/ : SUCCESS in 2h 37m 46s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5ac362a9b8dbafe9ed394f344d6ae53f5e5f6256","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-06-11 08:16:13.000000000","message":"Patch Set 4: Code-Review+1\n\nSeems okay, thanks!","accounts_in_message":[],"_revision_number":4},{"id":"3ea4d21a5aeed7e4d41af3032569a9f011541b96","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2019-06-16 21:32:29.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"73dfb53e46c4147471af171f178de11324f4b93f","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-06-16 21:32:45.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/3546/ : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"8bc8c29fd1a499a1722db7e69bafce6773994c5d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 00:14:13.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/56/664456/5/check/tripleo-ci-centos-7-undercloud-containers/97c4a4e/ : SUCCESS in 1h 26m 53s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/56/664456/5/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/05827e5/ : FAILURE in 1h 32m 13s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/56/664456/5/check/tripleo-ci-centos-7-containers-multinode/727baa0/ : SUCCESS in 1h 51m 02s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/56/664456/5/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/803f615/ : SUCCESS in 2h 35m 12s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container http://logs.openstack.org/56/664456/5/check/tripleo-ci-centos-7-scenario012-multinode-oooq-container/82180ce/ : FAILURE in 2h 18m 00s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/56/664456/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/214aefb/ : SUCCESS in 1h 57m 51s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/56/664456/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/0838363/ : SUCCESS in 2h 14m 27s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/56/664456/5/check/tripleo-ci-centos-7-standalone/434a876/ : SUCCESS in 1h 50m 50s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/56/664456/5/check/tripleo-ci-fedora-28-standalone/93eb284/ : SUCCESS in 1h 32m 08s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/56/664456/5/check/tripleo-ci-centos-7-standalone-upgrade/04a4c63/ : FAILURE in 1h 33m 51s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/56/664456/5/check/tripleo-ci-centos-7-standalone-upgrade-stein/1e2c064/ : SUCCESS in 2h 25m 08s\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/56/664456/5/check/tripleo-ci-centos-7-standalone-os-tempest/2cfbef7/ : SUCCESS in 1h 09m 07s\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/56/664456/5/check/tripleo-ci-centos-7-containers-multinode-stein/afcfeb1/ : SUCCESS in 1h 57m 53s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/56/664456/5/check/tripleo-ci-centos-7-containers-multinode-rocky/2b1117a/ : SUCCESS in 2h 00m 06s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/56/664456/5/check/tripleo-ci-centos-7-containers-multinode-queens/b3ca0a9/ : SUCCESS in 2h 18m 55s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/56/664456/5/check/tripleo-ci-centos-7-containers-multinode-pike/0223ceb/ : SUCCESS in 2h 13m 11s\n- openstack-tox-linters http://logs.openstack.org/56/664456/5/check/openstack-tox-linters/bba86d7/ : FAILURE in 4m 00s","accounts_in_message":[],"_revision_number":5},{"id":"03a3b130e238ffef8e293cddc75c3f8459da1fb8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-17 00:18:54.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/56/664456/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/f336df4/ : SUCCESS in 2h 10m 28s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/56/664456/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/265e85b/ : SUCCESS in 2h 20m 53s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/56/664456/5/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/91b303f/ : SUCCESS in 1h 13m 23s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/56/664456/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/7c23986/ : SUCCESS in 2h 05m 52s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/56/664456/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/0a6c9df/ : SUCCESS in 2h 18m 38s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/56/664456/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/92cffaf/ : SUCCESS in 2h 17m 33s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/56/664456/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/73bd860/ : SUCCESS in 2h 44m 30s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d3b47d6793248aea35d12187ad6908fe8c24f083","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-08-08 08:33:36.000000000","message":"Patch Set 5:\n\nrecheck stale results","accounts_in_message":[],"_revision_number":5},{"id":"570ac7d0223701cbb51f7aaf7e1647294c383f74","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-08-08 08:33:54.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/3745/ : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"c49992856e80488df9773a123885239e61d8de15","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-08 12:00:10.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/56/664456/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/3d87474/ : FAILURE in 2h 14m 35s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/56/664456/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/589962f/ : FAILURE in 3h 19m 19s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/56/664456/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/4727368/ : FAILURE in 3h 02m 42s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch finger://ze04.softwarefactory-project.io:7979/3cba737cf223449abcbdfdc3980a2c9f : POST_FAILURE in 3h 11m 39s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/56/664456/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/33527b3/ : SUCCESS in 3h 24m 06s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/56/664456/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/be96ffd/ : SUCCESS in 3h 24m 38s (non-voting)\n- tripleo-ci-rhel-8-standalone-rdo http://logs.rdoproject.org/56/664456/5/openstack-check/tripleo-ci-rhel-8-standalone-rdo/a57ba7b/ : FAILURE in 1h 04m 14s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2e76722a08cc11f4ba9a3bc34e6c88406f72b58e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 12:54:31.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/56/664456/5/check/tripleo-ci-centos-7-undercloud-containers/6424b95/ : SUCCESS in 1h 54m 39s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/56/664456/5/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/100bb74/ : FAILURE in 1h 50m 31s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/56/664456/5/check/tripleo-ci-centos-7-containers-multinode/b1c3fe9/ : SUCCESS in 2h 32m 26s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://logs.opendev.org/56/664456/5/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/24d52e4/ : SUCCESS in 2h 14m 38s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://logs.opendev.org/56/664456/5/check/tripleo-ci-centos-7-scenario012-multinode-oooq-container/dcad84c/ : FAILURE in 2h 26m 57s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/56/664456/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/fa8829a/ : SUCCESS in 2h 11m 34s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/56/664456/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/a4ff115/ : SUCCESS in 2h 16m 22s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/56/664456/5/check/tripleo-ci-centos-7-standalone/a9d93ac/ : SUCCESS in 1h 52m 46s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/56/664456/5/check/tripleo-ci-fedora-28-standalone/2b61c5c/ : FAILURE in 1h 39m 01s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/56/664456/5/check/tripleo-ci-centos-7-standalone-upgrade/033c944/ : SUCCESS in 2h 15m 15s (non-voting)\n- tripleo-ci-centos-7-standalone-os-tempest https://logs.opendev.org/56/664456/5/check/tripleo-ci-centos-7-standalone-os-tempest/5aac330/ : SUCCESS in 1h 20m 47s\n- tripleo-ci-centos-7-containers-multinode-stein https://logs.opendev.org/56/664456/5/check/tripleo-ci-centos-7-containers-multinode-stein/10ad78d/ : SUCCESS in 2h 27m 37s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky https://logs.opendev.org/56/664456/5/check/tripleo-ci-centos-7-containers-multinode-rocky/9b05dbe/ : SUCCESS in 2h 04m 51s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens https://logs.opendev.org/56/664456/5/check/tripleo-ci-centos-7-containers-multinode-queens/ce3549c/ : SUCCESS in 1h 55m 53s\n- tripleo-ci-centos-7-containers-multinode-pike https://logs.opendev.org/56/664456/5/check/tripleo-ci-centos-7-containers-multinode-pike/43c90c4/ : SUCCESS in 2h 08m 21s\n- openstack-tox-linters https://logs.opendev.org/56/664456/5/check/openstack-tox-linters/11cb485/ : FAILURE in 6m 09s","accounts_in_message":[],"_revision_number":5},{"id":"afdb409ed4e05cd0918c9b431262e2a7d5701ac8","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-08-08 13:31:57.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"cc5e718614c4e12e312e138ed843d9f3a8b90317","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2019-08-15 02:33:24.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2acabf57d369f1518b96e9ed88a546b4e72e642d","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2019-08-15 02:33:39.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"aa0e37adff64f26a1a22191acc02446c38758c7e","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-08-15 02:33:57.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/3789/ : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"efa5a9c7edd23d7aec53ea7132024658d15b3948","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 05:02:39.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/ab0bb582217c4dea90a79627c4bdff79 : SUCCESS in 1h 24m 29s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/02f03143fe05407294e50141d635cade : FAILURE in 1h 55m 43s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/c57ca11764e646c2b1c9dcc25e2ea257 : SUCCESS in 2h 14m 29s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/a31a4d395bde4d3e9baad9cd87911705 : SUCCESS in 2h 11m 22s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/b9e3070183e44325ba140bc25b9fc444 : FAILURE in 2h 15m 51s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/c12d5d70990b49c1a5d972c9c04ed3ee : SUCCESS in 2h 06m 58s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/00f53dc7ceb04da587026d243769174b : SUCCESS in 2h 13m 01s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/1f7de5b9ed5a474e8056715f51e230c5 : SUCCESS in 1h 39m 32s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/c48674362d0a4afa891dd88dd7075048 : SUCCESS in 2h 04m 38s (non-voting)\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/75af376c445642b5b903ee293af2e181 : SUCCESS in 1h 03m 15s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/c6d7f423b2c4483a9ab6cc8197eb2d04 : SUCCESS in 2h 09m 58s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/ac9a1d0ecdde4864aabf54dc8a1e4cd7 : SUCCESS in 2h 13m 03s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/eff9eca735754b469e9ddb33fb478625 : SUCCESS in 2h 08m 23s\n- tripleo-ci-centos-7-containers-multinode-pike https://zuul.opendev.org/t/openstack/build/8a5031f9efdb4fe4bd221ab1466dbfd5 : SUCCESS in 2h 21m 19s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/da7a4be55b3042e0bbc406b5f23ad342 : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":6},{"id":"4e46ebda10dcd22da1f0f71cbc6cf8cd7e4069e2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-15 06:14:26.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/56/664456/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/c5dca30/ : TIMED_OUT in 3h 39m 59s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/56/664456/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/cff5bf3/ : SUCCESS in 2h 28m 33s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/56/664456/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/2db8585/ : SUCCESS in 2h 07m 29s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/56/664456/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/c76e3a8/ : FAILURE in 2h 04m 52s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/56/664456/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/4c6af4d/ : SUCCESS in 2h 21m 44s","accounts_in_message":[],"_revision_number":6},{"id":"b5ed78b942ad21c86208f19142314be5159b2910","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2019-08-18 21:11:10.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"caa44e852155faa9e66dca07bd9f3ecc70d8a1fc","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-08-18 21:11:32.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/3801/ : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"f6bef1c26c52341337cee33777b01c2c6bf75ff6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-18 23:58:19.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/7758b52b7c214eeda71a05527cceadf4 : SUCCESS in 1h 22m 28s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/46f1f0a0e15e4fc4933fe1949a9b27b8 : FAILURE in 2h 05m 07s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/e296f572d9cb45cdb57cb32de09668ee : FAILURE in 7m 01s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/4a88010690bd4361a67720f8367e46ce : FAILURE in 13m 12s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/e0b53f17843d48e0a9b91bc4cca4e6bd : FAILURE in 2h 41m 05s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/543930f1754d4954a6435b6f3ab9b900 : SUCCESS in 2h 28m 01s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/e3f22fcc8473415d80c2f572d9238a2b : SUCCESS in 2h 26m 06s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/b6fc3ce05df74344aee41f6517560cb0 : FAILURE in 15m 01s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/773814d93eff4b86b31ef76a4930a114 : FAILURE in 5m 17s (non-voting)\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/552459ab81e64fff8d505a72f276b1f8 : SUCCESS in 1h 13m 13s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/11bad883a6ea49bca31d3a627c221631 : SUCCESS in 2h 09m 37s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/fe604e09e5d445ccb1870f924ed8413e : SUCCESS in 2h 11m 33s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/3acfd35edc5a4aae83694910b280e301 : SUCCESS in 2h 10m 33s\n- tripleo-ci-centos-7-containers-multinode-pike https://zuul.opendev.org/t/openstack/build/b54b024a7fe6459886fd9b295839dd69 : SUCCESS in 2h 19m 34s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3d39a4309686406cb3109e05a8e99f0d : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":7},{"id":"2c325a029301c33a460975c90940917b432e6b43","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-19 00:10:29.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/56/664456/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/c78de10/ : SUCCESS in 2h 52m 54s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/56/664456/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/c4689fe/ : SUCCESS in 2h 57m 57s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/56/664456/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/db7a75a/ : SUCCESS in 2h 14m 54s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/56/664456/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/7307e56/ : FAILURE in 2h 57m 19s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/56/664456/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/bd6d0f3/ : FAILURE in 1h 46m 19s","accounts_in_message":[],"_revision_number":7},{"id":"b29f581e5ae69903877f4c5644bfa1c6841f8f92","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2019-08-21 22:18:40.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"8e38427d47224ad4f5ad34293d7a3bb73acec1c0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-21 23:30:14.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/56/664456/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/f27f6fc/ : FAILURE in 1h 08m 41s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/56/664456/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/fd41a04/ : FAILURE in 1h 08m 41s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/56/664456/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/fcf2c3c/ : FAILURE in 43m 00s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/56/664456/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/a666ef5/ : FAILURE in 1h 00m 28s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/56/664456/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/a6b2542/ : FAILURE in 1h 09m 04s","accounts_in_message":[],"_revision_number":8},{"id":"2845336b1c58a6a6b8eb977b7a00c93d7cef189e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 00:33:48.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/aaa887c0d5514bf5bb62c560ed7650e0 : FAILURE in 1h 20m 01s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/5f5276c8298943f281aa1527e14a6e33 : FAILURE in 1h 15m 49s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/b33482366396404985d6d29eea0bce34 : FAILURE in 1h 13m 06s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/9b1d4ba609844be680c2eaaa755fcc69 : FAILURE in 1h 24m 10s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/6179d664ac9a46f3951aaa042ffa2de9 : FAILURE in 1h 25m 30s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/c451fcf728b346e586237a6b4c215616 : FAILURE in 1h 29m 36s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/c191d51ec33d4e1f8ede94f32e7f90ad : FAILURE in 1h 28m 44s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/821311d45ea948a592a2ae92bffae291 : SUCCESS in 1h 49m 47s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/aeb5174b52c04b288f73b785f9884190 : SUCCESS in 2h 08m 52s (non-voting)\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/46de1c6ecd3d4e27892d3b7cc6437516 : SUCCESS in 1h 15m 12s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/1879e695a28b4dc29b0c3f1c8b476088 : FAILURE in 1h 29m 26s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/59a5ab0a815c4af0b16dabeb03676b03 : FAILURE in 1h 27m 18s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/2c2a2b3407e644208d590e906a533954 : FAILURE in 55m 40s\n- tripleo-ci-centos-7-containers-multinode-pike https://zuul.opendev.org/t/openstack/build/dcd594b015804ecf91c31b304b72c63a : FAILURE in 55m 31s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2b94995ca4be4cf1923c1ccebb7f7fca : SUCCESS in 6m 01s","accounts_in_message":[],"_revision_number":8},{"id":"471d31e5f5d737e0332a814d4cf56fb9a9cd2be3","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2019-08-22 22:38:22.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"82ef28eb2695cdf2f9d7dd6ba5a93e243584602e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-23 01:35:13.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/56/664456/9/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/7ffaa83/ : SUCCESS in 2h 37m 16s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/56/664456/9/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/b70c773/ : SUCCESS in 2h 40m 57s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/56/664456/9/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/21e99f3/ : SUCCESS in 2h 19m 17s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/56/664456/9/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/0db6ab8/ : FAILURE in 2h 55m 25s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/56/664456/9/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/5cee847/ : FAILURE in 1h 42m 12s","accounts_in_message":[],"_revision_number":9},{"id":"e64f7c758164096b92798e28010b2ed845f9c58d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 01:57:24.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/17ddfe0caa264f0cbd87aff8700fd594 : SUCCESS in 1h 38m 51s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/ef31c27ca6614acd9ce82d46f586d8bc : FAILURE in 2h 01m 06s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/10bc16afea5d4b9991dacd9e8c1fe009 : SUCCESS in 2h 26m 25s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/b13cd42407194f17937a6c3b18f23517 : SUCCESS in 2h 35m 34s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/9d89fd04a1614b8cb67a291fb6b7f1d2 : FAILURE in 2h 59m 30s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/32ecf24301204a519e6662d6454fee39 : SUCCESS in 2h 45m 20s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/b135de1198de4decb648f82d1f0991d3 : SUCCESS in 2h 51m 46s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/38c036ea60624a8aaf32949edbe432b1 : SUCCESS in 2h 10m 57s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/92909b3142224aeea5d0f4189c711531 : SUCCESS in 2h 23m 41s (non-voting)\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/906bafb5fef8483ab7fadfe8138b3040 : SUCCESS in 1h 20m 05s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/28f383a9dce446d08134ac1307d611c7 : SUCCESS in 2h 00m 04s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/ef1bdc53826f4cd19b99d5d208e86fdc : SUCCESS in 2h 45m 15s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/5487e6d2d1fc4df7a443d77a29a63075 : SUCCESS in 2h 07m 09s\n- tripleo-ci-centos-7-containers-multinode-pike https://zuul.opendev.org/t/openstack/build/f882e87476004bd093062ae309a3d7a9 : SUCCESS in 2h 11m 38s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e67a3c714e5c458a8458d595743512d9 : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":9},{"id":"268502c1bcd7e4925547ff0a9de861bf031670e2","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-08-25 09:36:41.000000000","message":"Patch Set 9: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"09fd97ae6507aed827c1eb59a57f58159818f38c","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2019-08-27 03:40:30.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"b6e8498f7445720255584f395c835e317f94d200","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2019-08-27 03:53:51.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"641f8c6073aea895b754001ec831d342336d1494","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 06:42:22.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/7d89b48f3c4a4ab9ada759a2e2556aa7 : SUCCESS in 1h 42m 32s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/56a8aa4fe04d4ac2ac668383d4687460 : FAILURE in 1h 56m 08s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/4b41cff7da964630b40a9b717177964e : SUCCESS in 2h 31m 05s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/27908cebdd644f8294badcc35569778c : SUCCESS in 2h 43m 36s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/7bb5547eebe24373ad29ad64a3913cb6 : FAILURE in 2h 39m 31s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/90e38f7c1ac04aad85f87ac63668e97f : SUCCESS in 2h 37m 52s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/dec2b190f03e416885bf7f149ed30e21 : SUCCESS in 2h 37m 25s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/443769eec1e14fe98f53f974e57aeb2c : SUCCESS in 1h 53m 18s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/33195da97bfc4f6eb73dfe732a7141b2 : SUCCESS in 2h 02m 59s (non-voting)\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/13b23ac79dbc4ddc85511a5262dc7457 : SUCCESS in 1h 27m 40s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/5645afe80de848c1ad17a6f8abe20ebd : SUCCESS in 2h 27m 02s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/3e07b34854a747399e75c8b5a2932fe3 : SUCCESS in 2h 10m 26s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/f2ca3c1f49c14afa92648f5cc538b8f9 : SUCCESS in 2h 09m 03s\n- tripleo-ci-centos-7-containers-multinode-pike https://zuul.opendev.org/t/openstack/build/e9e6466275bb4563a4f8c5bf34a5e7eb : SUCCESS in 2h 20m 05s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/97238ceb36594e95ab4e453ed86c258f : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":10},{"id":"edc85aacff7b10798301c2a18f205d83f3ab90fe","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-27 07:14:27.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/56/664456/10/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/d5b35be/ : SUCCESS in 2h 49m 24s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/56/664456/10/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/b45c042/ : SUCCESS in 3h 18m 16s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/56/664456/10/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/aa206ab/ : SUCCESS in 2h 15m 49s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/56/664456/10/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/8112add/ : SUCCESS in 2h 26m 51s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/56/664456/10/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/0f58cce/ : FAILURE in 3h 19m 23s","accounts_in_message":[],"_revision_number":10},{"id":"c447c0d2031071a5f346d19590e21585c994396b","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-08-28 13:27:44.000000000","message":"Patch Set 10: Code-Review+1\n\n(1 comment)\n\nCan we run some dummy patch basing on that which runs baremetal_provision: true?","accounts_in_message":[],"_revision_number":10},{"id":"823e148d68e4e8ca49c51a139f31a015d38b8d0c","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2019-09-06 01:33:49.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"f5ff94e05399894177d1aed4712973f81592617e","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2019-09-06 04:14:10.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"0c1e38ff14552656b90db7f3ddf49890a8216d7c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-06 07:02:37.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/56/664456/12/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/fb38362/ : SUCCESS in 2h 42m 58s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/56/664456/12/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/a77bb89/ : FAILURE in 2h 04m 19s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/56/664456/12/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/4af3bfd/ : SUCCESS in 2h 13m 32s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/56/664456/12/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/33c63fd/ : SUCCESS in 2h 47m 05s","accounts_in_message":[],"_revision_number":12},{"id":"d7f86a575b28895e34ba5103f6f8da148b45d4dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 07:22:15.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/b0aa9ebac3814fe48ee21e04a64ecbb0 : SUCCESS in 1h 50m 31s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/806c8089ffd24ad88d3769e1edaa2169 : FAILURE in 2h 06m 52s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/7ce54a6a18e94ef19148e8c3dbd179ce : SUCCESS in 3h 01m 49s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/b27509b829b945de8913ab354a71efe5 : SUCCESS in 2h 49m 15s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/6fddbf8abef841d689e4cb18df854878 : FAILURE in 2h 46m 17s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/52b2c04048c0472095dffdb1caf5ad53 : SUCCESS in 2h 23m 26s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/d0f17cc707724bb890b64e86ba5ec1d7 : SUCCESS in 2h 34m 39s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/7133c7aef9de43f6aa94301e2b91143f : SUCCESS in 1h 58m 19s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/8d85fa07729b4d0786c62f7a9d234453 : SUCCESS in 2h 00m 09s (non-voting)\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/10a17b2a4e1446128795c15432fa61b1 : SUCCESS in 1h 22m 33s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/5f0df64723054b62af1b7f06e50aba35 : SUCCESS in 2h 22m 36s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/44535ff6be884b07ba734e512cd35c03 : SUCCESS in 2h 23m 38s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/bf58851d7ee74318b3cb441f72aafb62 : SUCCESS in 2h 17m 06s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2e764bbc8e2c4650b3b945073d141c90 : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":12},{"id":"3e125d58e28694ece5ceb2ef66469bc1af870d1a","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2019-09-12 02:28:03.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"69cd0eeca7b2bc978270496cf99b8596de47c7bd","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-12 05:19:10.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/56/664456/13/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/840bf60/ : SUCCESS in 2h 38m 41s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/56/664456/13/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/2fe0adb/ : SUCCESS in 2h 02m 07s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/56/664456/13/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/590a974/ : SUCCESS in 2h 19m 14s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/56/664456/13/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/deaec6f/ : SUCCESS in 2h 47m 54s","accounts_in_message":[],"_revision_number":13},{"id":"6107c62e904f231165ec7edcd3c96247ac738341","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 05:53:09.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/23dd484e06684a9593360bfbd9e3a016 : SUCCESS in 1h 32m 37s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/76dca519b21648e19aa87cd4ab771b05 : FAILURE in 1h 52m 17s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/1337e1df992e43e6b12447b298ac6ce2 : SUCCESS in 2h 53m 54s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/184ed6324b6e4c8abd081832bc9b3a41 : SUCCESS in 3h 09m 45s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/9e08bbdb8e38434daa213371b73c0e2a : FAILURE in 2h 17m 24s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/6e920272d3914fdda3ea35ee00e911c0 : SUCCESS in 2h 23m 09s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/4ee9bfdae81b46a1b7055d330b21f700 : SUCCESS in 2h 22m 23s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/23f2d3ee3d4a4221b9399b255731cde4 : SUCCESS in 2h 02m 05s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/908b30036d1e499ab54c30a02a96a497 : FAILURE in 1h 54m 27s (non-voting)\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/c68b5f956dcb45058052d6dbd4efc36c : SUCCESS in 1h 27m 40s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/89b0211c81fa44f88baef7ee390283d1 : SUCCESS in 2h 50m 51s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/82e724467e8541dd8a73c3b24ce028ab : SUCCESS in 2h 22m 05s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/9c13d7e2f639438d972910ae8605414a : SUCCESS in 2h 22m 47s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2aaccfc447604576aefe54b7c3be8a15 : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":13},{"id":"49b9ab9641136988f1e086a90942e0ce1c051432","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2019-09-13 02:07:09.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"b5f1007f117f7ec1f75bb751aed77ee8b0115d32","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2019-09-13 04:51:16.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"32851066e2b9d67d10e190fdfedb1f516119db8a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-13 07:49:07.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/56/664456/15/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/f1321de/ : SUCCESS in 2h 56m 07s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/56/664456/15/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/a1cc9b1/ : SUCCESS in 2h 10m 30s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/56/664456/15/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/2cdd298/ : SUCCESS in 2h 15m 53s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/56/664456/15/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/57dad46/ : SUCCESS in 2h 43m 24s","accounts_in_message":[],"_revision_number":15},{"id":"50fdb86959d250885ae41a3007b561b131190dec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 08:35:47.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/97798ab989b9441f9d9221871620ead6 : SUCCESS in 2h 17m 13s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/1b4eb972e2e14ce7b642f65cb36cf37d : FAILURE in 2h 34m 28s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/2a536c9f7eef4e29bd58a8ad0a08962d : SUCCESS in 3h 07m 31s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/3e376988f35b4b43a1fe4950fc944d24 : TIMED_OUT in 3h 28m 41s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/b1e6262798a64a9ca8ce056d60d8885f : FAILURE in 1h 48m 39s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/091b3541814f4c1891a85fc4669e63f8 : SUCCESS in 2h 37m 35s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/df4f868403064e90beb544c01fb33c2b : SUCCESS in 2h 33m 39s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/7f0846207b754bc59ac9156f492e025e : SUCCESS in 2h 04m 42s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/9613c965eb69495c94e19d52a986f38f : FAILURE in 2h 08m 57s (non-voting)\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/1d02c05005fb47d08694dff6f11ebd82 : SUCCESS in 1h 27m 32s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/e7a12c19208f4454a3d1f19a23878afc : SUCCESS in 2h 49m 31s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/befa72895c7e4f4299bbdbeaa4768b3f : SUCCESS in 2h 49m 38s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/ac28e7bde19645d29d974713dbe9859e : SUCCESS in 2h 23m 21s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7a7fc3cd3b3b488a8138ffecc41a1e4c : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":15},{"id":"ca9ec6f39c5fb4fc994876aac56ead5f04b52919","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-09-15 07:38:38.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"63beddeb46c10cf68a63fe9a58504ebd086902f0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-15 10:54:25.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/56/664456/15/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/d8492d4/ : SUCCESS in 2h 40m 40s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/56/664456/15/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/1b9545c/ : SUCCESS in 3h 14m 58s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/56/664456/15/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/0f191ee/ : SUCCESS in 2h 33m 16s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/56/664456/15/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/f2ef739/ : FAILURE in 1h 55m 21s","accounts_in_message":[],"_revision_number":15},{"id":"d1646e5c19198dbfe29035b649ab508f29128f05","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 11:08:59.000000000","message":"Patch Set 15:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/ceb64558ee6c42228c27d9525007ea1b : SUCCESS in 2h 08m 49s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/7c1afd3d42424eb7a8ad7152f66788a6 : FAILURE in 1h 57m 56s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/aeeb814fdbdc41c2b9c5e9f40aa30a29 : SUCCESS in 3h 18m 16s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/5b56231beeb744fbb1cbace8dc2aa078 : SUCCESS in 3h 02m 45s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/7137c18572f24475ba4ef58e3df1b46e : FAILURE in 2h 15m 09s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/b14a84fa1de548168d2f7700e4c57e37 : SUCCESS in 2h 34m 34s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/45653ac8f6e640ce944c1cdd5c8c75cd : SUCCESS in 2h 43m 35s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/ee709f75f7b4412b9b290ec0ebbbe6fb : SUCCESS in 2h 11m 55s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/c18c7e1b51ad4d05ba575b460a816208 : FAILURE in 2h 02m 04s (non-voting)\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/c1842baa64ac408d9b212777913dcef7 : SUCCESS in 1h 31m 38s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/1356b71be6b9495fab7d2a5a6950fdaa : SUCCESS in 2h 37m 04s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/ba8bf1f2b15b433c89e887d00f71428d : SUCCESS in 2h 16m 54s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/deb7def896bd4285a75a7391293ff3f0 : SUCCESS in 2h 04m 48s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d11b00fba2204646a8750c7283f3b109 : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":15},{"id":"41c07e8c01f00551b6e1bf53fd5ed1fd9a851255","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-09-15 13:58:24.000000000","message":"Patch Set 16: Published edit on patch set 15.","accounts_in_message":[],"_revision_number":16},{"id":"b417027bfbc31b467bee58ae065bca6848822159","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-09-15 13:58:30.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"1371856a9d629034f6d8020a6f9d9ef8ed54ae52","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-15 16:47:38.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/56/664456/16/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/b205bce/ : SUCCESS in 2h 47m 50s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/56/664456/16/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/0bc8dff/ : SUCCESS in 2h 31m 09s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/56/664456/16/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/e700912/ : SUCCESS in 2h 13m 00s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/56/664456/16/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/22ac5f8/ : FAILURE in 2h 45m 41s","accounts_in_message":[],"_revision_number":16},{"id":"d118510c622196513b8e188942a8498e42fb340c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 17:18:13.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/9704f688499847d6afa0cc58dc022f7c : SUCCESS in 1h 44m 37s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/3eb8e6e794db4d1baa2ef01ac70fb4a6 : FAILURE in 1h 41m 36s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/55317064cb3d4bd3b6cc5410c134d615 : SUCCESS in 3h 09m 59s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/ae1a6ea3cb614b7eb8ad28522aff9bcd : SUCCESS in 3h 01m 12s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/2da019f01bd94a3398a87a8a25bfe980 : FAILURE in 2h 12m 45s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/cf224bf7119748c0b78fd2211740565a : SUCCESS in 2h 38m 37s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/dd8151cc093d402d80d13bfde55ccf3d : SUCCESS in 2h 32m 32s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/964be67ff509400eb682454be339c44d : SUCCESS in 2h 08m 09s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/7a127e9d26a64cd8b95ce5d71002db53 : SUCCESS in 2h 42m 48s (non-voting)\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/c81c06258ec84f099e71dd5ace65e334 : SUCCESS in 1h 29m 17s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/ec388a66019e4d959ec97921d4f062dd : SUCCESS in 2h 20m 58s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/21f63265749c4abbae3ca0590b075c9e : SUCCESS in 2h 20m 59s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/7c7ff8fd5b4e45c8b82e6f790aae930b : SUCCESS in 1h 54m 19s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/70fcc75bee9d40e7a7f5e82898d3891e : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":16},{"id":"433b3926d35eafd150f373c0d36013cd1f0483fd","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-09-15 20:47:17.000000000","message":"Patch Set 16: Code-Review+2","accounts_in_message":[],"_revision_number":16},{"id":"545f548ffef0f95fd37d111f390a14610c926e8c","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-09-16 22:16:48.000000000","message":"Patch Set 16: Workflow+1","accounts_in_message":[],"_revision_number":16},{"id":"33d0a20a28f529a3862b0d45fb0016fb369d3d07","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 22:17:07.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"51e20a6916cbf261f6f61903c462073fac16ec8e","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-09-17 20:12:43.000000000","message":"Abandoned\n\nWe are facing gate issue: https://bugs.launchpad.net/tripleo/+bug/1844446\n\nTo clear the gate we need to abandon this patch and I will restore once the gate is ready again to land patches in TripleO. Please don\u0027t touch this patch, and ask on #tripleo Wes or Emilien for any question. Thanks for your patience.","accounts_in_message":[],"_revision_number":16},{"id":"e5069982395e1ef89bfbd4a6e9561bba35919248","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-09-19 02:26:07.000000000","message":"Restored\n\nrecheck","accounts_in_message":[],"_revision_number":16},{"id":"e6ee528459356e501d713bc51721b1c9b2ee642f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 05:42:19.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/ce3e6346f8e84fd1816c30ea9402059a : SUCCESS in 1h 53m 50s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/871b02f7ab97458b95861b15627e39ae : FAILURE in 2h 31m 58s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/a2fb76c49cdf455788ebb0934804e842 : SUCCESS in 2h 56m 30s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/259b24a837e34a8a97f653db370526e1 : SUCCESS in 3h 09m 25s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/15d9f657c2dc4a9f9049b115e2e5fb8b : FAILURE in 2h 14m 03s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/4e63050c00ba4281943db21b937b6109 : SUCCESS in 2h 43m 24s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/e2077003352e4d67be37c5546c374600 : SUCCESS in 2h 17m 33s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/d4dcca3f4ce34f3b802e4c6650ab3f94 : SUCCESS in 2h 11m 36s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/c703d189116f41d4a6b5eb1608095ee7 : SUCCESS in 2h 54m 25s (non-voting)\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/ba0bcb61a2034f3f9ed6f922a89cda90 : SUCCESS in 1h 49m 33s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/d1d9ded127074eb98f60e8f302655d98 : SUCCESS in 2h 56m 12s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/49d4b15e8ca94908b92920409615275e : SUCCESS in 2h 42m 14s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/6e07619a6de04857b49cf94bcd0b514e : SUCCESS in 2h 17m 03s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3c6ea35ac8734caebb25cf02d3e02c90 : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":16},{"id":"8428f55157ed8af6b73ffe9726ae51a73db99959","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 05:42:38.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"034719c2c4ef809986536f3df4548b5d124f758b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-19 10:21:17.000000000","message":"Patch Set 16:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 finger://ze03.softwarefactory-project.io:7979/c454233c9bc3419da0b3396a16abda2f : POST_FAILURE in 4h 00m 21s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/56/664456/16/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/44fa2eb/ : FAILURE in 2h 00m 32s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/56/664456/16/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/6e82221/ : TIMED_OUT in 3h 41m 33s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/56/664456/16/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/00a9221/ : FAILURE in 3h 18m 19s","accounts_in_message":[],"_revision_number":16},{"id":"b0834124c3175fd28b0239c5525b939d160ac423","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 17:34:33.000000000","message":"Patch Set 16: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/021a150ae866445bb7c697169fe0cf4c : SUCCESS in 2h 32m 39s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/a35418862bc84b82a5bf448592839b0f : SUCCESS in 3h 07m 55s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/e85be180f7f149ae9698921a3ebc5326 : SUCCESS in 2h 53m 00s\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/de0bea68ab864d219cb8274d633e1db9 : SUCCESS in 1h 55m 41s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/61d30f7f515e4749a2d5dfddb9a6753a : SUCCESS in 2h 57m 58s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/f073da3f4c4541f8a41022b1f550bff4 : SUCCESS in 3h 08m 45s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/60dda2f8190f44eba78182fa2176fcb8 : SUCCESS in 3h 13m 47s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/33729f43dbfe44ea94cc6ae280e1c64a : SUCCESS in 2h 42m 06s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f790e52980a94a508a5b50147df863eb : SUCCESS in 6m 38s","accounts_in_message":[],"_revision_number":16},{"id":"59ce56554298f428881624c53cf1a4903faf2caa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 17:34:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"7a9676908fa9e6ec814f8cbe12f86162864ec436","revisions":{"07f038b4718fa0c58ec8493b69c68f54c833b7c9":{"kind":"REWORK","_number":1,"created":"2019-06-10 22:40:38.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/56/664456/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/56/664456/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/1"}}},"commit":{"parents":[{"commit":"c72f25ea31ffbf8f64b36ab8959b2c276b0ca2b1","subject":"Define var undercloud_enable_nova","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/c72f25ea31ffbf8f64b36ab8959b2c276b0ca2b1"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-06-10 22:38:42.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-06-10 22:38:42.000000000","tz":720},"subject":"Define variable baremetal_provision","message":"Define variable baremetal_provision\n\nIf `baremetal_provision` is `true`, deploy baremetal with \"openstack\novercloud node provision\" and deployed-server, instead of with\nheat/nova.\n\nChange-Id: I57893a64f4c3224ffe8bd12330f12bd9d3cbbf89\nBlueprint: nova-less-deploy\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/07f038b4718fa0c58ec8493b69c68f54c833b7c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/07f038b4718fa0c58ec8493b69c68f54c833b7c9"}]},"branch":"refs/heads/master"},"157e989100ed767388f201ad508ff6b3c611a7b5":{"kind":"REWORK","_number":2,"created":"2019-06-11 04:11:58.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/56/664456/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/56/664456/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/2"}}},"commit":{"parents":[{"commit":"c72f25ea31ffbf8f64b36ab8959b2c276b0ca2b1","subject":"Define var undercloud_enable_nova","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/c72f25ea31ffbf8f64b36ab8959b2c276b0ca2b1"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-06-10 22:38:42.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-06-11 04:07:58.000000000","tz":720},"subject":"Support direct provisioning of bare metal","message":"Support direct provisioning of bare metal\n\nIf `baremetal_provision` is `true`, deploy baremetal with \"openstack\novercloud node provision\" and deployed-server, instead of with\nheat/nova.\n\nChange-Id: I57893a64f4c3224ffe8bd12330f12bd9d3cbbf89\nBlueprint: nova-less-deploy\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/157e989100ed767388f201ad508ff6b3c611a7b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/157e989100ed767388f201ad508ff6b3c611a7b5"}]},"branch":"refs/heads/master"},"704811d6c99b7134fb5f1b6f5ff9d2550fc292e1":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-06-11 04:12:27.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/56/664456/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/56/664456/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/3"}}},"commit":{"parents":[{"commit":"3e9560ef56cbcf94eb15fefe4ffab13491bc8c1e","subject":"Merge \"Pin ARA to \u003c1.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/3e9560ef56cbcf94eb15fefe4ffab13491bc8c1e"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-06-10 22:38:42.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-06-11 04:12:16.000000000","tz":720},"subject":"Support direct provisioning of bare metal","message":"Support direct provisioning of bare metal\n\nIf `baremetal_provision` is `true`, deploy baremetal with \"openstack\novercloud node provision\" and deployed-server, instead of with\nheat/nova.\n\nChange-Id: I57893a64f4c3224ffe8bd12330f12bd9d3cbbf89\nBlueprint: nova-less-deploy\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/704811d6c99b7134fb5f1b6f5ff9d2550fc292e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/704811d6c99b7134fb5f1b6f5ff9d2550fc292e1"}]},"branch":"refs/heads/master"},"457113e135793f7701e9495eeea7a0b27b8a5439":{"kind":"REWORK","_number":4,"created":"2019-06-11 04:21:46.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/56/664456/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/56/664456/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/4"}}},"commit":{"parents":[{"commit":"3e9560ef56cbcf94eb15fefe4ffab13491bc8c1e","subject":"Merge \"Pin ARA to \u003c1.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/3e9560ef56cbcf94eb15fefe4ffab13491bc8c1e"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-06-10 22:38:42.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-06-11 04:21:34.000000000","tz":720},"subject":"Support direct provisioning of bare metal","message":"Support direct provisioning of bare metal\n\nIf `baremetal_provision` is `true`, deploy baremetal with \"openstack\novercloud node provision\" and deployed-server, instead of with\nheat/nova.\n\nChange-Id: I57893a64f4c3224ffe8bd12330f12bd9d3cbbf89\nBlueprint: nova-less-deploy\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/457113e135793f7701e9495eeea7a0b27b8a5439"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/457113e135793f7701e9495eeea7a0b27b8a5439"}]},"branch":"refs/heads/master"},"5d0c3953f18e53d9b6c7f748f8fafc6899c07973":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-06-16 21:32:29.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/56/664456/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/56/664456/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/5"}}},"commit":{"parents":[{"commit":"a916cdf805489319d1c7d45f5398d8b210d743da","subject":"Merge \"validations: Allow to customize the max retries\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/a916cdf805489319d1c7d45f5398d8b210d743da"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-06-10 22:38:42.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-06-16 21:30:44.000000000","tz":720},"subject":"Support direct provisioning of bare metal","message":"Support direct provisioning of bare metal\n\nIf `baremetal_provision` is `true`, deploy baremetal with \"openstack\novercloud node provision\" and deployed-server, instead of with\nheat/nova.\n\nChange-Id: I57893a64f4c3224ffe8bd12330f12bd9d3cbbf89\nBlueprint: nova-less-deploy\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/5d0c3953f18e53d9b6c7f748f8fafc6899c07973"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/5d0c3953f18e53d9b6c7f748f8fafc6899c07973"}]},"branch":"refs/heads/master"},"6b05a0add77e7ade9e10f4cfc4d00f4d851a3d2a":{"kind":"REWORK","_number":6,"created":"2019-08-15 02:33:39.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/56/664456/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/56/664456/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/6"}}},"commit":{"parents":[{"commit":"dfe7238db1e2c2b6b2d33d449097affab9603c7f","subject":"Configure rhel container for tempest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/dfe7238db1e2c2b6b2d33d449097affab9603c7f"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-06-10 22:38:42.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-08-15 02:33:04.000000000","tz":720},"subject":"Support direct provisioning of bare metal","message":"Support direct provisioning of bare metal\n\nIf `baremetal_provision` is `true`, deploy baremetal with \"openstack\novercloud node provision\" and deployed-server, instead of with\nheat/nova.\n\nChange-Id: I57893a64f4c3224ffe8bd12330f12bd9d3cbbf89\nBlueprint: nova-less-deploy\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/6b05a0add77e7ade9e10f4cfc4d00f4d851a3d2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/6b05a0add77e7ade9e10f4cfc4d00f4d851a3d2a"}]},"branch":"refs/heads/master"},"b4e628e2c83a6ce1af1ec407340dd3a5f03b2314":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-08-18 21:11:10.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/56/664456/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/56/664456/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/7"}}},"commit":{"parents":[{"commit":"376eb5f29bda6aadafb0411d173f22f9e5a0ad65","subject":"Define var undercloud_enable_nova","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/376eb5f29bda6aadafb0411d173f22f9e5a0ad65"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-06-10 22:38:42.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-08-18 21:10:38.000000000","tz":720},"subject":"Support direct provisioning of bare metal","message":"Support direct provisioning of bare metal\n\nIf `baremetal_provision` is `true`, deploy baremetal with \"openstack\novercloud node provision\" and deployed-server, instead of with\nheat/nova.\n\nChange-Id: I57893a64f4c3224ffe8bd12330f12bd9d3cbbf89\nBlueprint: nova-less-deploy\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/b4e628e2c83a6ce1af1ec407340dd3a5f03b2314"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/b4e628e2c83a6ce1af1ec407340dd3a5f03b2314"}]},"branch":"refs/heads/master"},"0f687fe25d1af5022a5e2e417f30c90b87c4145a":{"kind":"REWORK","_number":8,"created":"2019-08-21 22:18:40.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/56/664456/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/56/664456/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/8"}}},"commit":{"parents":[{"commit":"6a20363ef06d31099cd6b99831afc20cc33210b7","subject":"Merge \"Define var undercloud_enable_nova\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/6a20363ef06d31099cd6b99831afc20cc33210b7"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-06-10 22:38:42.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-08-21 22:12:02.000000000","tz":720},"subject":"Support direct provisioning of bare metal","message":"Support direct provisioning of bare metal\n\nIf `baremetal_provision` is `true`, deploy baremetal with \"openstack\novercloud node provision\" and deployed-server, instead of with\nheat/nova.\n\nChange-Id: I57893a64f4c3224ffe8bd12330f12bd9d3cbbf89\nBlueprint: nova-less-deploy\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/0f687fe25d1af5022a5e2e417f30c90b87c4145a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/0f687fe25d1af5022a5e2e417f30c90b87c4145a"}]},"branch":"refs/heads/master"},"add7420ebbf9505f939a9ce2d97ceb741e0592cc":{"kind":"REWORK","_number":9,"created":"2019-08-22 22:38:22.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/56/664456/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/56/664456/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/9"}}},"commit":{"parents":[{"commit":"6a20363ef06d31099cd6b99831afc20cc33210b7","subject":"Merge \"Define var undercloud_enable_nova\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/6a20363ef06d31099cd6b99831afc20cc33210b7"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-06-10 22:38:42.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-08-22 22:10:36.000000000","tz":720},"subject":"Support direct provisioning of bare metal","message":"Support direct provisioning of bare metal\n\nIf `baremetal_provision` is `true`, deploy baremetal with \"openstack\novercloud node provision\" and deployed-server, instead of with\nheat/nova.\n\nChange-Id: I57893a64f4c3224ffe8bd12330f12bd9d3cbbf89\nBlueprint: nova-less-deploy\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/add7420ebbf9505f939a9ce2d97ceb741e0592cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/add7420ebbf9505f939a9ce2d97ceb741e0592cc"}]},"branch":"refs/heads/master"},"d58823d52754bca9b96d959b8630ee597d6fbff9":{"kind":"REWORK","_number":10,"created":"2019-08-27 03:53:51.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/56/664456/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/56/664456/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/10"}}},"commit":{"parents":[{"commit":"8f3bb293484c844182f6215fb323302efb06e8c0","subject":"Merge \"Move reproducer README file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/8f3bb293484c844182f6215fb323302efb06e8c0"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-06-10 22:38:42.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-08-27 03:42:09.000000000","tz":720},"subject":"Support direct provisioning of bare metal","message":"Support direct provisioning of bare metal\n\nIf `baremetal_provision` is `true`, deploy baremetal with \"openstack\novercloud node provision\" and deployed-server, instead of with\nheat/nova.\n\nChange-Id: I57893a64f4c3224ffe8bd12330f12bd9d3cbbf89\nBlueprint: nova-less-deploy\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/d58823d52754bca9b96d959b8630ee597d6fbff9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/d58823d52754bca9b96d959b8630ee597d6fbff9"}]},"branch":"refs/heads/master"},"5a33db15f1eecf2273ec87e4b5e434c754efc226":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2019-09-06 01:33:49.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/56/664456/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/56/664456/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/11"}}},"commit":{"parents":[{"commit":"a308ff0992917b9341a3e46bfbf03902bff39dc1","subject":"Merge \"update tox minversion version\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/a308ff0992917b9341a3e46bfbf03902bff39dc1"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-06-10 22:38:42.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-09-06 01:32:58.000000000","tz":720},"subject":"Support direct provisioning of bare metal","message":"Support direct provisioning of bare metal\n\nIf `baremetal_provision` is `true`, deploy baremetal with \"openstack\novercloud node provision\" and deployed-server, instead of with\nheat/nova.\n\nChange-Id: I57893a64f4c3224ffe8bd12330f12bd9d3cbbf89\nBlueprint: nova-less-deploy\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/5a33db15f1eecf2273ec87e4b5e434c754efc226"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/5a33db15f1eecf2273ec87e4b5e434c754efc226"}]},"branch":"refs/heads/master"},"1428c7e387a493d95401a59becd36981cc3a6d3d":{"kind":"REWORK","_number":12,"created":"2019-09-06 04:14:10.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/56/664456/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/56/664456/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/12"}}},"commit":{"parents":[{"commit":"a308ff0992917b9341a3e46bfbf03902bff39dc1","subject":"Merge \"update tox minversion version\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/a308ff0992917b9341a3e46bfbf03902bff39dc1"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-06-10 22:38:42.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-09-06 04:13:59.000000000","tz":720},"subject":"Support direct provisioning of bare metal","message":"Support direct provisioning of bare metal\n\nIf `baremetal_provision` is `true`, deploy baremetal with \"openstack\novercloud node provision\" and deployed-server, instead of with\nheat/nova.\n\nChange-Id: I57893a64f4c3224ffe8bd12330f12bd9d3cbbf89\nBlueprint: nova-less-deploy\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/1428c7e387a493d95401a59becd36981cc3a6d3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/1428c7e387a493d95401a59becd36981cc3a6d3d"}]},"branch":"refs/heads/master"},"386c243e41c7d4b50f00f2337d3b513a732965ce":{"kind":"REWORK","_number":13,"created":"2019-09-12 02:28:03.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/56/664456/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/56/664456/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/13"}}},"commit":{"parents":[{"commit":"0b62606370fce9fc2e76fc96545ff9a548723a72","subject":"Merge \"\u0027is in\u0027 is not a valid Jinja test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/0b62606370fce9fc2e76fc96545ff9a548723a72"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-06-10 22:38:42.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-09-12 02:27:51.000000000","tz":720},"subject":"Support direct provisioning of bare metal","message":"Support direct provisioning of bare metal\n\nIf `baremetal_provision` is `true`, deploy baremetal with \"openstack\novercloud node provision\" and deployed-server, instead of with\nheat/nova.\n\nChange-Id: I57893a64f4c3224ffe8bd12330f12bd9d3cbbf89\nBlueprint: nova-less-deploy\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/386c243e41c7d4b50f00f2337d3b513a732965ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/386c243e41c7d4b50f00f2337d3b513a732965ce"}]},"branch":"refs/heads/master"},"b8f83ee564c3ee64d0f35f9475a675d758f35469":{"kind":"REWORK","_number":14,"created":"2019-09-13 02:07:09.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/56/664456/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/56/664456/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/14"}}},"commit":{"parents":[{"commit":"0b62606370fce9fc2e76fc96545ff9a548723a72","subject":"Merge \"\u0027is in\u0027 is not a valid Jinja test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/0b62606370fce9fc2e76fc96545ff9a548723a72"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-06-10 22:38:42.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-09-13 02:06:38.000000000","tz":720},"subject":"Support direct provisioning of bare metal","message":"Support direct provisioning of bare metal\n\nIf `baremetal_provision` is `true`, deploy baremetal with \"openstack\novercloud node provision\" and deployed-server, instead of with\nheat/nova.\n\nChange-Id: I57893a64f4c3224ffe8bd12330f12bd9d3cbbf89\nBlueprint: nova-less-deploy\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/b8f83ee564c3ee64d0f35f9475a675d758f35469"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/b8f83ee564c3ee64d0f35f9475a675d758f35469"}]},"branch":"refs/heads/master"},"645490d5c3d023c04a873f8d9097db2b47d30557":{"kind":"REWORK","_number":15,"created":"2019-09-13 04:51:16.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/56/664456/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/56/664456/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/15"}}},"commit":{"parents":[{"commit":"0b62606370fce9fc2e76fc96545ff9a548723a72","subject":"Merge \"\u0027is in\u0027 is not a valid Jinja test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/0b62606370fce9fc2e76fc96545ff9a548723a72"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-06-10 22:38:42.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-09-13 04:51:06.000000000","tz":720},"subject":"Support direct provisioning of bare metal","message":"Support direct provisioning of bare metal\n\nIf `baremetal_provision` is `true`, deploy baremetal with \"openstack\novercloud node provision\" and deployed-server, instead of with\nheat/nova.\n\nChange-Id: I57893a64f4c3224ffe8bd12330f12bd9d3cbbf89\nBlueprint: nova-less-deploy\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/645490d5c3d023c04a873f8d9097db2b47d30557"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/645490d5c3d023c04a873f8d9097db2b47d30557"}]},"branch":"refs/heads/master"},"7a9676908fa9e6ec814f8cbe12f86162864ec436":{"kind":"REWORK","_number":16,"created":"2019-09-15 13:58:24.000000000","uploader":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"ref":"refs/changes/56/664456/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/56/664456/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/664456/16"}}},"commit":{"parents":[{"commit":"0b62606370fce9fc2e76fc96545ff9a548723a72","subject":"Merge \"\u0027is in\u0027 is not a valid Jinja test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/0b62606370fce9fc2e76fc96545ff9a548723a72"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-06-10 22:38:42.000000000","tz":720},"committer":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-09-15 13:58:20.000000000","tz":0},"subject":"Support direct provisioning of bare metal","message":"Support direct provisioning of bare metal\n\nIf `baremetal_provision` is `true`, deploy baremetal with \"openstack\novercloud node provision\" and deployed-server, instead of with\nheat/nova.\n\nChange-Id: I57893a64f4c3224ffe8bd12330f12bd9d3cbbf89\nBlueprint: nova-less-deploy\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7a9676908fa9e6ec814f8cbe12f86162864ec436"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7a9676908fa9e6ec814f8cbe12f86162864ec436"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
