)]}'
{"id":"openstack%2Ftripleo-quickstart-extras~669344","triplet_id":"openstack%2Ftripleo-quickstart-extras~master~Ie387f1274059ec80d19c5ef0bf9aed7a3a999e2a","project":"openstack/tripleo-quickstart-extras","branch":"master","topic":"octavia-ci","hashtags":[],"change_id":"Ie387f1274059ec80d19c5ef0bf9aed7a3a999e2a","subject":"Add a tasks to download amphora image for octavia deployments","status":"MERGED","created":"2019-07-05 12:15:29.000000000","updated":"2019-10-21 19:52:25.000000000","submitted":"2019-10-21 19:52:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669344-1571687545096-70f2cce9","meta_rev_id":"60867ee3774a40cee15839e725f6beaed9451bbd","_number":669344,"virtual_id_number":669344,"owner":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"date":"2019-10-10 13:09:52.000000000","_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-21 19:52:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-10-10 16:21:10.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"recommended":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":1,"date":"2019-10-16 11:57:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":2,"date":"2019-09-11 15:31:05.000000000","_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":2,"date":"2019-09-11 15:31:05.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":1,"date":"2019-10-21 14:15:22.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-16 07:15:21.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2019-08-12 12:54:51.000000000","updated_by":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-01 08:41:57.000000000","updated_by":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"reviewer":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"state":"REVIEWER"},{"updated":"2019-10-10 16:21:10.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-16 11:57:47.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-10-21 14:15:22.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-10-21 19:52:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d866680bb443fa5db53fe69a519ab05e064a76e6","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2019-07-05 12:15:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"22fc58304bb8acb5da7dc1d5eef78c6593be5235","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-07-05 12:15:51.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/3605/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"eeead03a4d2a960c5d3731cfec3ae93b556c6f77","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2019-07-05 12:30:44.000000000","message":"Patch Set 1: Workflow-1\n\nPart of a collection of changes in progress","accounts_in_message":[],"_revision_number":1},{"id":"186616adca5aa5d98510e34d1197d9a88701ac91","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-05 15:12:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/44/669344/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/2674eb6/ : SUCCESS in 2h 48m 14s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/44/669344/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/1922bb4/ : FAILURE in 2h 11m 28s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/44/669344/1/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/4fd0aca/ : SUCCESS in 1h 20m 57s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/44/669344/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/7021015/ : SUCCESS in 2h 06m 31s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/44/669344/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/0b4c9df/ : SUCCESS in 2h 09m 44s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/44/669344/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/c3bdb3d/ : SUCCESS in 2h 45m 46s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/44/669344/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/a7249e9/ : SUCCESS in 2h 54m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8ebf25f85cb19740ef63f1b4ad4f9760a593ef3c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 15:17:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/44/669344/1/check/tripleo-ci-centos-7-undercloud-containers/22ff327/ : SUCCESS in 1h 47m 34s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/44/669344/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/a2e813a/ : FAILURE in 2h 18m 16s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/44/669344/1/check/tripleo-ci-centos-7-containers-multinode/16e5eae/ : SUCCESS in 2h 08m 37s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/44/669344/1/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/b9d859d/ : SUCCESS in 2h 21m 50s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container http://logs.openstack.org/44/669344/1/check/tripleo-ci-centos-7-scenario012-multinode-oooq-container/b640418/ : FAILURE in 2h 25m 44s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/44/669344/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/acea206/ : SUCCESS in 2h 11m 36s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/44/669344/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/1f2e663/ : SUCCESS in 2h 36m 05s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/44/669344/1/check/tripleo-ci-centos-7-standalone/380cc96/ : SUCCESS in 1h 47m 57s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/44/669344/1/check/tripleo-ci-fedora-28-standalone/7973097/ : SUCCESS in 1h 38m 16s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/44/669344/1/check/tripleo-ci-centos-7-standalone-upgrade/90dc9e9/ : SUCCESS in 2h 22m 33s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/44/669344/1/check/tripleo-ci-centos-7-standalone-upgrade-stein/940864d/ : SUCCESS in 2h 21m 14s\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/44/669344/1/check/tripleo-ci-centos-7-standalone-os-tempest/7eccb79/ : SUCCESS in 1h 15m 30s\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/44/669344/1/check/tripleo-ci-centos-7-containers-multinode-stein/a71fbd0/ : SUCCESS in 1h 58m 51s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/44/669344/1/check/tripleo-ci-centos-7-containers-multinode-rocky/91d76a2/ : SUCCESS in 2h 08m 52s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/44/669344/1/check/tripleo-ci-centos-7-containers-multinode-queens/a13ca74/ : SUCCESS in 2h 00m 14s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/44/669344/1/check/tripleo-ci-centos-7-containers-multinode-pike/5f65413/ : SUCCESS in 2h 08m 18s\n- openstack-tox-linters http://logs.openstack.org/44/669344/1/check/openstack-tox-linters/5148843/ : FAILURE in 3m 40s","accounts_in_message":[],"_revision_number":1},{"id":"ffea1e35429873a703dae0d1bee96eb1cece5356","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2019-07-05 17:09:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"744f4bf8514ccc4b49586c21cbf8d85de1548fb8","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2019-07-05 17:10:06.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"b18b120c11595a8d9f8dfcc4ee6273b3df1788eb","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-07-05 17:10:06.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/3608/ : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"8a730bcb73b2c57e680f0a2f083e0f541726af5f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-05 20:04:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/44/669344/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/7215555/ : SUCCESS in 2h 28m 42s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/44/669344/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/a4bc3df/ : SUCCESS in 2h 32m 06s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/44/669344/2/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/fe57f69/ : SUCCESS in 1h 17m 02s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/44/669344/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/79611d6/ : FAILURE in 2h 04m 37s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/44/669344/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/45e63be/ : SUCCESS in 2h 06m 49s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/44/669344/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/e20e94e/ : SUCCESS in 2h 49m 01s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/44/669344/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/5e9437a/ : SUCCESS in 2h 52m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5f2b50dceb6f6c47e5000c72a63dd9205ba8c0f7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 20:17:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/44/669344/2/check/tripleo-ci-centos-7-undercloud-containers/8c58461/ : SUCCESS in 1h 30m 41s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/44/669344/2/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/4777e37/ : FAILURE in 1h 54m 34s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/44/669344/2/check/tripleo-ci-centos-7-containers-multinode/33f7bf6/ : SUCCESS in 2h 05m 06s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/44/669344/2/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/3814ab5/ : SUCCESS in 2h 43m 27s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container http://logs.openstack.org/44/669344/2/check/tripleo-ci-centos-7-scenario012-multinode-oooq-container/cbacfdf/ : FAILURE in 2h 15m 13s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/44/669344/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/e94ae76/ : SUCCESS in 2h 18m 22s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/44/669344/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/fc7fc5e/ : SUCCESS in 2h 29m 24s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/44/669344/2/check/tripleo-ci-centos-7-standalone/2decec0/ : SUCCESS in 1h 50m 29s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/44/669344/2/check/tripleo-ci-fedora-28-standalone/cc9f7b0/ : SUCCESS in 1h 43m 05s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/44/669344/2/check/tripleo-ci-centos-7-standalone-upgrade/0ceca90/ : SUCCESS in 2h 13m 46s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/44/669344/2/check/tripleo-ci-centos-7-standalone-upgrade-stein/92efce5/ : FAILURE in 1h 20m 39s\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/44/669344/2/check/tripleo-ci-centos-7-standalone-os-tempest/134a455/ : SUCCESS in 1h 24m 40s\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/44/669344/2/check/tripleo-ci-centos-7-containers-multinode-stein/fde69c9/ : SUCCESS in 2h 40m 34s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/44/669344/2/check/tripleo-ci-centos-7-containers-multinode-rocky/0ed3e31/ : SUCCESS in 2h 40m 06s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/44/669344/2/check/tripleo-ci-centos-7-containers-multinode-queens/27fc12e/ : SUCCESS in 2h 14m 39s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/44/669344/2/check/tripleo-ci-centos-7-containers-multinode-pike/d477fca/ : SUCCESS in 2h 21m 29s\n- openstack-tox-linters http://logs.openstack.org/44/669344/2/check/openstack-tox-linters/ca8f352/ : FAILURE in 3m 51s","accounts_in_message":[],"_revision_number":2},{"id":"0afb25039c7ff46a91079cdc12c283d99f220bcd","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2019-07-08 13:16:28.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"92a9ce6a22d9d1d4173165e767edd2d464bbb9cf","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-07-08 13:16:47.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/3611/ : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"366212e7aa440cf02c134fcfd72a29cd43d3c351","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-08 16:35:40.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/44/669344/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/15a6be2/ : RETRY_LIMIT in 6m 09s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/44/669344/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/e9aad5f/ : RETRY_LIMIT in 2m 41s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch finger://ze02.softwarefactory-project.io:7979/9f3b88390079480cbadbc8966f121396 : RETRY_LIMIT in 7m 05s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch finger://ze03.softwarefactory-project.io:7979/4904a87bd64040e2887da4c41ac11bb6 : RETRY_LIMIT in 10m 02s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/44/669344/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/a31c06d/ : FAILURE in 57m 53s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/44/669344/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/55c1080/ : FAILURE in 55m 31s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"daaf05e527e3bf2ac3f31eeb291dc32a198ccfd4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 20:34:09.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/44/669344/3/check/tripleo-ci-centos-7-undercloud-containers/afe3eff/ : SUCCESS in 1h 54m 29s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/44/669344/3/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/ac01161/ : FAILURE in 1h 51m 39s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/44/669344/3/check/tripleo-ci-centos-7-containers-multinode/b451170/ : FAILURE in 1h 00m 32s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/44/669344/3/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/3ffa472/ : SUCCESS in 2h 20m 54s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container http://logs.openstack.org/44/669344/3/check/tripleo-ci-centos-7-scenario012-multinode-oooq-container/10a1041/ : FAILURE in 2h 37m 30s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/44/669344/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/4464bcf/ : SUCCESS in 2h 32m 26s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/44/669344/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/7c6ce38/ : SUCCESS in 2h 41m 07s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/44/669344/3/check/tripleo-ci-centos-7-standalone/738b272/ : SUCCESS in 2h 00m 52s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/44/669344/3/check/tripleo-ci-fedora-28-standalone/4aa0db1/ : SUCCESS in 1h 41m 35s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/44/669344/3/check/tripleo-ci-centos-7-standalone-upgrade/994c979/ : SUCCESS in 2h 14m 12s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/44/669344/3/check/tripleo-ci-centos-7-standalone-upgrade-stein/08129a9/ : SUCCESS in 2h 26m 35s\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/44/669344/3/check/tripleo-ci-centos-7-standalone-os-tempest/66d7513/ : SUCCESS in 1h 19m 52s\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/44/669344/3/check/tripleo-ci-centos-7-containers-multinode-stein/066da98/ : SUCCESS in 2h 28m 04s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/44/669344/3/check/tripleo-ci-centos-7-containers-multinode-rocky/df9f768/ : SUCCESS in 2h 18m 31s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/44/669344/3/check/tripleo-ci-centos-7-containers-multinode-queens/8dde7a5/ : SUCCESS in 2h 10m 49s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/44/669344/3/check/tripleo-ci-centos-7-containers-multinode-pike/83c8b89/ : SUCCESS in 2h 31m 56s\n- openstack-tox-linters http://logs.openstack.org/44/669344/3/check/openstack-tox-linters/e774dfa/ : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":3},{"id":"1b574a59d6f360872e5ff1c6f069d1d0f3497ec6","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2019-07-15 11:11:45.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4ce1fbb746f34d92ea99be769297d3b0a9f37c1f","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-07-15 11:12:06.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/3635/ : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"eca514d37ed942f7ef351a8b4c94fedd758d4815","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-15 13:57:31.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/44/669344/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/c7a1227/ : FAILURE in 2h 15m 31s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/44/669344/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/39ede9c/ : SUCCESS in 2h 31m 01s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/44/669344/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/5ab9dac/ : FAILURE in 2h 08m 54s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/44/669344/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/de1f5fd/ : SUCCESS in 2h 15m 14s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/44/669344/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/344b859/ : SUCCESS in 2h 28m 26s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/44/669344/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/fa2f2a0/ : SUCCESS in 2h 43m 42s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c471d329bc9c9001db9e8275159526daaeb00f60","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2019-07-15 14:04:11.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"228f2c96507c076f8234c98935cb44d333fd47ff","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-07-15 14:04:30.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/3637/ : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"ec39b12e8fac387e0a941ab32ac22ab52f79c418","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-15 14:08:05.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4a78a35c9fb02e90485b027965370a1ac25a8c22","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-15 17:07:03.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/44/669344/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/7e365a1/ : SUCCESS in 2h 28m 13s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/44/669344/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/03ae2ba/ : SUCCESS in 2h 34m 19s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/44/669344/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/60608c3/ : SUCCESS in 2h 14m 56s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/44/669344/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/df5dba1/ : SUCCESS in 2h 39m 06s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/44/669344/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/17d72c7/ : SUCCESS in 2h 40m 13s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/44/669344/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/4c6b16a/ : SUCCESS in 3h 01m 06s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"794d1461009e94cc0e1eee323e054860f651885a","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2019-07-15 18:14:46.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"eb8fe20bce086209cc78cb1222c05f7ce8be9d10","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-07-15 18:15:02.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/3639/ : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"200a04da61457c4bc5ee24bf76c3bdf90e9d2f31","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 22:50:11.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/44/669344/6/check/tripleo-ci-centos-7-undercloud-containers/add88c7/ : SUCCESS in 1h 43m 21s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/44/669344/6/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/b006b75/ : FAILURE in 2h 19m 16s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/44/669344/6/check/tripleo-ci-centos-7-containers-multinode/ec87a7d/ : SUCCESS in 2h 04m 12s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/44/669344/6/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/3d024ef/ : SUCCESS in 2h 16m 01s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container http://logs.openstack.org/44/669344/6/check/tripleo-ci-centos-7-scenario012-multinode-oooq-container/251a3aa/ : FAILURE in 2h 03m 35s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/44/669344/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/98a50f1/ : SUCCESS in 2h 14m 43s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/44/669344/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/45f6b4e/ : FAILURE in 2h 34m 24s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/44/669344/6/check/tripleo-ci-centos-7-standalone/0bf8723/ : SUCCESS in 1h 39m 02s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/44/669344/6/check/tripleo-ci-fedora-28-standalone/8f3b20f/ : SUCCESS in 1h 37m 04s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/44/669344/6/check/tripleo-ci-centos-7-standalone-upgrade/0a848ce/ : FAILURE in 1h 54m 08s (non-voting)\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/44/669344/6/check/tripleo-ci-centos-7-standalone-os-tempest/4751666/ : SUCCESS in 1h 20m 37s\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/44/669344/6/check/tripleo-ci-centos-7-containers-multinode-stein/c8018b4/ : SUCCESS in 2h 35m 46s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/44/669344/6/check/tripleo-ci-centos-7-containers-multinode-rocky/2840f5f/ : SUCCESS in 2h 08m 14s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/44/669344/6/check/tripleo-ci-centos-7-containers-multinode-queens/566b1d0/ : SUCCESS in 2h 01m 20s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/44/669344/6/check/tripleo-ci-centos-7-containers-multinode-pike/103b6da/ : SUCCESS in 2h 25m 24s\n- openstack-tox-linters http://logs.openstack.org/44/669344/6/check/openstack-tox-linters/cc74cdc/ : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":6},{"id":"75188f98dd83f6a4db685143a327b4007a9e76ff","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-15 23:24:59.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/44/669344/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/6821313/ : SUCCESS in 3h 11m 36s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/44/669344/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/adf5e86/ : RETRY_LIMIT in 5m 53s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/44/669344/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/49f5b64/ : SUCCESS in 2h 08m 10s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/44/669344/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/9627810/ : SUCCESS in 2h 28m 18s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/44/669344/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/928bf2c/ : FAILURE in 1h 35m 59s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 finger://ze02.softwarefactory-project.io:7979/6a29c7d5a0b143cbbe77789a2f95d86f : RETRY_LIMIT in 6s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"596c56ea5e3b8c23163fa0aece18795583983aac","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2019-07-16 07:15:21.000000000","message":"Patch Set 6: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"323c5f0b2f195c03b444558ed8c02f02ab858b60","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2019-07-16 12:08:42.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"81e2aaac9ed7a114fb12ee37f622403dec36debf","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-07-16 12:09:02.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/3641/ : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"6055d9db0155d82953c973d5615ccf782d38246b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 15:04:28.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/44/669344/7/check/tripleo-ci-centos-7-undercloud-containers/81ab951/ : SUCCESS in 1h 38m 13s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/44/669344/7/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/d929bc5/ : FAILURE in 2h 06m 08s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/44/669344/7/check/tripleo-ci-centos-7-containers-multinode/a9d570f/ : SUCCESS in 2h 08m 13s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/44/669344/7/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/de6f821/ : FAILURE in 1h 00m 25s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container http://logs.openstack.org/44/669344/7/check/tripleo-ci-centos-7-scenario012-multinode-oooq-container/6ed9d1c/ : FAILURE in 2h 23m 40s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/44/669344/7/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/676b144/ : SUCCESS in 2h 34m 43s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/44/669344/7/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/991951b/ : FAILURE in 2h 37m 20s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/44/669344/7/check/tripleo-ci-centos-7-standalone/6a1f607/ : FAILURE in 53m 53s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/44/669344/7/check/tripleo-ci-fedora-28-standalone/b02e18a/ : SUCCESS in 1h 53m 38s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/44/669344/7/check/tripleo-ci-centos-7-standalone-upgrade/d4b8567/ : FAILURE in 1h 30m 19s (non-voting)\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/44/669344/7/check/tripleo-ci-centos-7-standalone-os-tempest/2faf340/ : FAILURE in 50m 42s\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/44/669344/7/check/tripleo-ci-centos-7-containers-multinode-stein/9b6fa20/ : SUCCESS in 2h 13m 41s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/44/669344/7/check/tripleo-ci-centos-7-containers-multinode-rocky/7f9eb96/ : SUCCESS in 2h 31m 31s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/44/669344/7/check/tripleo-ci-centos-7-containers-multinode-queens/a628f5f/ : SUCCESS in 2h 13m 18s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/44/669344/7/check/tripleo-ci-centos-7-containers-multinode-pike/794fce0/ : SUCCESS in 2h 43m 45s\n- openstack-tox-linters http://logs.openstack.org/44/669344/7/check/openstack-tox-linters/610ec43/ : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":7},{"id":"08c09b0a71418164878c98085ea20030cf471774","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-16 15:23:01.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/44/669344/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/f8ab50c/ : SUCCESS in 2h 36m 51s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/44/669344/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/2c48eb1/ : SUCCESS in 3h 12m 00s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/44/669344/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/05056fc/ : SUCCESS in 2h 22m 32s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/44/669344/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/a921053/ : SUCCESS in 2h 20m 57s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/44/669344/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/a63a78e/ : SUCCESS in 2h 34m 32s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/44/669344/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/30182bd/ : SUCCESS in 2h 49m 12s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"3c9810531860da0a117549f174a50c9632acfef2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-19 14:31:46.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/44/669344/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/31dca8d/ : SUCCESS in 3h 39m 27s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/44/669344/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/06b82e5/ : RETRY_LIMIT in 6m 07s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/44/669344/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/b28f511/ : RETRY_LIMIT in 3m 30s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/44/669344/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/ef48142/ : RETRY_LIMIT in 5m 08s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/44/669344/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/e8df1f4/ : RETRY_LIMIT in 5m 24s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/44/669344/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/80f8e52/ : SUCCESS in 3h 06m 10s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"bd49c342384bb6d45e18c5d8e9498ee2d60e35f0","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2019-07-24 15:05:40.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"80a73f57b14982193a0462bc1c6ebade11d4cd12","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-07-24 15:05:58.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/3682/ : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"de6474bb51d22675b86074921fdc45b2d30203c4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 19:17:39.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/44/669344/8/check/tripleo-ci-centos-7-undercloud-containers/9f358b7/ : SUCCESS in 1h 31m 15s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/44/669344/8/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/f8fc7cf/ : FAILURE in 2h 08m 05s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/44/669344/8/check/tripleo-ci-centos-7-containers-multinode/46200d4/ : SUCCESS in 2h 30m 42s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/44/669344/8/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/c991625/ : SUCCESS in 3h 02m 14s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container http://logs.openstack.org/44/669344/8/check/tripleo-ci-centos-7-scenario012-multinode-oooq-container/abed7f4/ : FAILURE in 2h 27m 50s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/44/669344/8/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/57f13e3/ : SUCCESS in 2h 30m 03s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/44/669344/8/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/4b7f299/ : SUCCESS in 2h 30m 13s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/44/669344/8/check/tripleo-ci-centos-7-standalone/2a2d7dc/ : FAILURE in 1h 01m 38s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/44/669344/8/check/tripleo-ci-fedora-28-standalone/be37455/ : FAILURE in 1h 00m 20s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/44/669344/8/check/tripleo-ci-centos-7-standalone-upgrade/8cc49d1/ : FAILURE in 1h 45m 53s (non-voting)\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/44/669344/8/check/tripleo-ci-centos-7-standalone-os-tempest/89ebd4f/ : SUCCESS in 1h 21m 38s\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/44/669344/8/check/tripleo-ci-centos-7-containers-multinode-stein/25c3be6/ : SUCCESS in 2h 15m 04s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/44/669344/8/check/tripleo-ci-centos-7-containers-multinode-rocky/6759a35/ : SUCCESS in 2h 28m 51s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/44/669344/8/check/tripleo-ci-centos-7-containers-multinode-queens/ee2b752/ : SUCCESS in 2h 05m 39s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/44/669344/8/check/tripleo-ci-centos-7-containers-multinode-pike/5778a80/ : SUCCESS in 2h 28m 17s\n- openstack-tox-linters http://logs.openstack.org/44/669344/8/check/openstack-tox-linters/0eec249/ : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":8},{"id":"298428142600b7dcda8a11ffe3a44f0c7a67c463","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-24 19:21:01.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 finger://ze03.softwarefactory-project.io:7979/824c8fac65284cc291d4314df2320427 : POST_FAILURE in 3h 32m 46s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 finger://ze07.softwarefactory-project.io:7979/ee777ee250c34ac0b78fc0d1fd19a1fa : POST_FAILURE in 3h 42m 47s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/44/669344/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/86257e2/ : FAILURE in 3h 17m 36s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/44/669344/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/a4eb234/ : FAILURE in 1h 35m 22s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch finger://ze03.softwarefactory-project.io:7979/37530b1b0150428baa7daa373ecaaf42 : POST_FAILURE in 3h 38m 39s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 finger://ze02.softwarefactory-project.io:7979/8e6c0db9227f43b193f60099fd3b39ba : POST_FAILURE in 4h 00m 20s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"21ae83f3b503fb1b81fc89214049bcbc2387fcb1","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2019-08-01 16:58:06.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"94c4f63a608f23b7ea59fe1ebecb3a48af26742f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-01 19:44:52.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/44/669344/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/2c57fe0/ : SUCCESS in 2h 35m 55s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/44/669344/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/02af3c9/ : SUCCESS in 2h 36m 25s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/44/669344/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/87a10a3/ : FAILURE in 1h 10m 40s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/44/669344/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/737b9c2/ : FAILURE in 1h 20m 20s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/44/669344/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/88b4661/ : SUCCESS in 2h 40m 17s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/44/669344/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/3950a2d/ : SUCCESS in 2h 44m 12s (non-voting)\n- tripleo-ci-rhel-8-standalone-rdo http://logs.rdoproject.org/44/669344/8/openstack-check/tripleo-ci-rhel-8-standalone-rdo/b4abe41/ : FAILURE in 14m 21s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"899c875dfdc3fbb238f85218777a6e0d4693f2ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 03:41:12.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/44/669344/8/check/tripleo-ci-centos-7-undercloud-containers/61027e4/ : SUCCESS in 1h 27m 10s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/44/669344/8/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/beca7cf/ : TIMED_OUT in 3h 12m 42s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/44/669344/8/check/tripleo-ci-centos-7-containers-multinode/da15a9e/ : SUCCESS in 2h 18m 53s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://logs.opendev.org/44/669344/8/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/81b9eed/ : SUCCESS in 2h 29m 15s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://logs.opendev.org/44/669344/8/check/tripleo-ci-centos-7-scenario012-multinode-oooq-container/b4f9292/ : FAILURE in 2h 28m 05s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/44/669344/8/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/da11063/ : SUCCESS in 2h 15m 59s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/44/669344/8/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/5063635/ : SUCCESS in 2h 38m 53s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/44/669344/8/check/tripleo-ci-centos-7-standalone/be94527/ : SUCCESS in 1h 52m 48s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/44/669344/8/check/tripleo-ci-fedora-28-standalone/c91eb7a/ : FAILURE in 47m 07s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/44/669344/8/check/tripleo-ci-centos-7-standalone-upgrade/afb0b10/ : FAILURE in 1h 33m 43s (non-voting)\n- tripleo-ci-centos-7-standalone-os-tempest https://logs.opendev.org/44/669344/8/check/tripleo-ci-centos-7-standalone-os-tempest/49ddf9e/ : SUCCESS in 1h 08m 26s\n- tripleo-ci-centos-7-containers-multinode-stein https://logs.opendev.org/44/669344/8/check/tripleo-ci-centos-7-containers-multinode-stein/3f478e3/ : SUCCESS in 2h 05m 38s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky https://logs.opendev.org/44/669344/8/check/tripleo-ci-centos-7-containers-multinode-rocky/ec51a82/ : SUCCESS in 2h 21m 05s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens https://logs.opendev.org/44/669344/8/check/tripleo-ci-centos-7-containers-multinode-queens/6871748/ : SUCCESS in 2h 11m 15s\n- tripleo-ci-centos-7-containers-multinode-pike https://logs.opendev.org/44/669344/8/check/tripleo-ci-centos-7-containers-multinode-pike/30effee/ : SUCCESS in 2h 15m 17s\n- openstack-tox-linters https://logs.opendev.org/44/669344/8/check/openstack-tox-linters/be74df8/ : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":8},{"id":"195d624d0437ba6f95d33e4ebb43c3e27f128c8a","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2019-08-12 12:54:33.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"76dc1e72858c301f7606eaaa9a1ca8097fd55283","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-08-12 12:54:51.000000000","message":"Patch Set 9:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/3768/ : FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"9cec287f06ecbe129fccb0678e623cd2b950a469","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 16:23:52.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/d04d52ced9d343de966deed9545dc64e : SUCCESS in 1h 32m 41s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/58c95eee41c244dc9f6f308100175baf : FAILURE in 1h 57m 25s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/4659fb94df174d93be2bbb3ef6df4573 : SUCCESS in 2h 17m 08s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/f65d2d8bfd964522b7db91e28196239e : SUCCESS in 2h 25m 57s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/5b9e130937294e478c93c3683b49cc3f : FAILURE in 2h 22m 53s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/274d64aa8d0b4482af1d97fcb8d766f0 : SUCCESS in 2h 20m 28s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/f6919f2d2f8847f59121ccc620dd293c : SUCCESS in 2h 15m 18s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/5938df246f964f39a0ea58c6ff3bc4d5 : FAILURE in 55m 10s\n- tripleo-ci-fedora-28-standalone https://zuul.opendev.org/t/openstack/build/b39bfd48c5074572a184ef50be589368 : SUCCESS in 1h 30m 35s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/02663dcb9bc9417e899e5db63cacfb67 : SUCCESS in 2h 03m 15s (non-voting)\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/aa7b745a3de64340aded8eac3f85cabb : SUCCESS in 1h 09m 51s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/1251cf9dfff44219aef4596b8cabc429 : SUCCESS in 2h 24m 30s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/7bb38de4b11d465fa1e728bc37fe4832 : SUCCESS in 2h 23m 21s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/5bd3cf33b80e4e0081e72034b406582e : SUCCESS in 2h 02m 22s\n- tripleo-ci-centos-7-containers-multinode-pike https://zuul.opendev.org/t/openstack/build/bd5f2eb22a984fdeae0b3bcface210f1 : SUCCESS in 2h 27m 16s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7d75a4a23d29464aa968529ede05ffe9 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":9},{"id":"e2f3705b21cc1acff46ae02727d79ed3e50ae29f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-12 16:37:43.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 finger://ze01.softwarefactory-project.io:7979/cc0a8d9076e64566a0b9fd2cc2663c4d : POST_FAILURE in 3h 23m 23s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 finger://ze06.softwarefactory-project.io:7979/d84f140cdd0a44698b81010b0bb5b91c : POST_FAILURE in 3h 29m 42s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch finger://ze07.softwarefactory-project.io:7979/b2c44268e4e848d4bc9e01081982217a : POST_FAILURE in 3h 14m 54s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch finger://ze01.softwarefactory-project.io:7979/3da1b0451225496486217dd9f968538f : POST_FAILURE in 3h 15m 16s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/44/669344/9/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/5d867e7/ : FAILURE in 1h 21m 47s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 finger://ze06.softwarefactory-project.io:7979/0d7a899df7ed40f9be724e981e33826d : POST_FAILURE in 3h 42m 07s (non-voting)\n- tripleo-ci-rhel-8-standalone-rdo http://logs.rdoproject.org/44/669344/9/openstack-check/tripleo-ci-rhel-8-standalone-rdo/5f6f7dd/ : FAILURE in 50m 08s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"405868f7e9f79858d0428776b39fc489be733d87","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2019-08-23 18:54:36.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"33d6e18b2354da64414aab7a8efd3a8b1c0d98d4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 22:04:07.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/aff7add149a042dfbff13977bdd4683a : SUCCESS in 1h 59m 08s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/cc20679b15d749a0b4bc8f17d757e680 : FAILURE in 1h 51m 59s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/6807503fe41948d4b9e681462b9ed7db : SUCCESS in 2h 43m 43s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/031c9e97b338420abb2640ef5398dc4d : SUCCESS in 3h 01m 41s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/f7c3225bf09341d992940e3db8189902 : FAILURE in 3h 03m 41s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/0f446d55330a46799cc4b0e958bd039a : SUCCESS in 2h 54m 56s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/c4a19be09f6d48e78bde005ba7d43ae7 : SUCCESS in 2h 39m 48s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/e728b64bf668404baa266c1d32146d22 : SUCCESS in 2h 10m 08s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/e30c1ab4d02a49b0bb36229d76cb740a : SUCCESS in 2h 04m 47s (non-voting)\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/7c4af51ed2d141bfa37ee97a5ab83432 : SUCCESS in 1h 29m 34s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/7671b1b2eea443818f4f23301be5d0f9 : SUCCESS in 2h 15m 08s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/4c5e4bfcc2894fd0b8161582eb50597b : SUCCESS in 2h 20m 12s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/1af02c5941f04607bcf930131326c559 : SUCCESS in 2h 14m 31s\n- tripleo-ci-centos-7-containers-multinode-pike https://zuul.opendev.org/t/openstack/build/03bca4b1f02749a1b3a265c9db0caef1 : SUCCESS in 2h 45m 39s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/527d5bb1cf754ecbac6ea71c53d9c06a : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":10},{"id":"67313b6dd8624805aa0fab7275b6510f8787226a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-23 22:43:51.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/44/669344/10/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/21a02da/ : SUCCESS in 3h 29m 52s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/44/669344/10/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/d4f2334/ : FAILURE in 3h 47m 36s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/44/669344/10/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/d5e1010/ : TIMED_OUT in 3h 39m 57s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/44/669344/10/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/57d172a/ : SUCCESS in 2h 16m 17s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/44/669344/10/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/2ed9de6/ : SUCCESS in 2h 40m 33s","accounts_in_message":[],"_revision_number":10},{"id":"1de7883089f8fa08c348665a2d93c548a7ae6d74","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-08-25 09:30:41.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"b52af2a259d3de2825191aeea3cc5047a3a6e8fa","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-28 13:46:30.000000000","message":"Patch Set 10: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"217acd3c23c7c7382de6990f9a9df197be0caaaa","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2019-08-28 14:23:21.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"ba98b0657fd859df1d1fc784280909a1a98cea58","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-09-01 08:41:57.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"8accfc63ff60e034f2c90df73b744eeaa9858c1b","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-09-01 08:49:26.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"7d7f6d7fdc1458c37e950128501ee964f4acc44d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-01 09:11:00.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/44/669344/10/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/c6fbefa/ : FAILURE in 17m 36s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/44/669344/10/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/603a199/ : FAILURE in 17m 44s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/44/669344/10/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/8cef379/ : FAILURE in 20m 07s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/44/669344/10/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/6ec5a02/ : FAILURE in 20m 09s","accounts_in_message":[],"_revision_number":10},{"id":"04ba54b5e2482f4d71389fd2d9d59aa677fb68dd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-01 09:21:51.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/7fc3bc0566424fec98aad779a54db017 : FAILURE in 17m 59s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/51621491d09d421f9cac70f349babb53 : FAILURE in 8m 53s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/194cd8682e1e48399fc4834c1d3f4a2a : FAILURE in 12m 40s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/3d7e7c166c2e41e5bd30d934c0b3f66e : FAILURE in 19m 32s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/1797c1c4401641f49cd469d46a4c42e8 : FAILURE in 22m 02s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/fd575bc92826409baa1cf3568b0bfa7d : FAILURE in 11m 59s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/9279f67a2f894b8ea3e4f1a0e533c358 : FAILURE in 23m 10s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/278c34abf2c94ba88ea46b86311d31bc : FAILURE in 16m 59s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/15943a6cd34e4cdbab81434180c188e2 : FAILURE in 9m 28s (non-voting)\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/5aec8e49724e42d09607df9199e6defb : FAILURE in 16m 32s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/0e99f4aaa677441c897d09779788b850 : FAILURE in 16m 50s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/8fdbf76db5824d358aa59c1a2727b605 : FAILURE in 13m 17s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/01f52b5d20a1442993c1b5d944c49b1b : FAILURE in 11m 36s\n- tripleo-ci-centos-7-containers-multinode-pike https://zuul.opendev.org/t/openstack/build/de7a56d582044a399848e344d81cd5a1 : FAILURE in 14m 49s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8b2900455bda4de7a4c7f963185d2c04 : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":10},{"id":"e3b5299f0db89daa4fc281ba34929a254d2dffb2","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2019-09-11 15:21:19.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"f3ea8a928e79df7db925ba2c232f5d19f41803c7","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2019-09-11 15:30:54.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"2ac6951ed6d44c80568454eebd43466c4ad3a9ec","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2019-09-11 15:31:05.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.","accounts_in_message":[],"_revision_number":12},{"id":"595a0f283d8915f64ccdb45cc84b16f550431ae6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 18:47:45.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/bc7ab76f87f94f2d8cf58d5ada699b28 : SUCCESS in 1h 50m 22s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/a67780bae35f43848768182be1da05f0 : FAILURE in 2h 01m 58s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/6e30f8ca816b469683abb98c1880090b : SUCCESS in 3h 04m 39s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/5a653f4d86574445a56ae449f7e1dd7f : SUCCESS in 3h 01m 13s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/6ca3ab0b442a40148a2b6d67b2fca212 : FAILURE in 2h 07m 35s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/f8bf2a71d2e94b5daeeb8fade81a80f6 : SUCCESS in 2h 28m 51s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/455220ebf2704aef89322bed5a25ab45 : SUCCESS in 2h 34m 53s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/0eda92fd131240308ec5ddf0dc13cd3a : FAILURE in 1h 22m 57s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/218b1d94089f41a98092b099e6daf640 : FAILURE in 1h 49m 38s (non-voting)\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/d2cce51c43dd43f4bcf968978bdc5518 : SUCCESS in 1h 28m 07s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/4d8be04a0e064d8390be0d1d9bd54614 : SUCCESS in 2h 18m 47s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/c52274fd8329410d957ab26f13e83733 : SUCCESS in 2h 33m 15s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/a156504463fe4286a284bd09cf419309 : SUCCESS in 2h 28m 09s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d949f33c26904ab5ba816e26ca7b7654 : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":12},{"id":"22002b712ba31bc7cfe130671665f0df584285a3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-11 19:05:06.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/44/669344/12/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/d20425c/ : SUCCESS in 2h 51m 13s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/44/669344/12/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/2a00746/ : SUCCESS in 3h 32m 21s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/44/669344/12/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/1e09db3/ : SUCCESS in 2h 30m 43s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/44/669344/12/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/d7cd230/ : SUCCESS in 2h 49m 13s","accounts_in_message":[],"_revision_number":12},{"id":"2c7f3982aaf960ef86b5eed31b45369208549e02","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2019-09-11 19:19:53.000000000","message":"Patch Set 12:\n\nnova conductor healthcheck failing","accounts_in_message":[],"_revision_number":12},{"id":"8d60a1eed965998917922f6681be4f822a83cbc6","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2019-10-10 13:09:52.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"ecde3dc2cd58eef521352ccdd1680f35bbc7b92e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 16:19:57.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/2d79f1d052fe4d0e8ee2f281c9b91ddb : SUCCESS in 1h 55m 58s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/6527926157324c5893d01c2c4eabec00 : FAILURE in 2h 10m 29s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/2447938511b147d98e6707508a9a0ff7 : SUCCESS in 3h 02m 21s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/49ebabadb5f64b94aa4eb46dc7e18265 : SUCCESS in 3h 03m 29s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/6dce1f0be4654cdbab85c74aa41d6d08 : FAILURE in 2h 10m 48s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/e932709fbc454272bda243c5387b55ab : SUCCESS in 2h 29m 17s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/6566d49fa3bf4795b1d6fdaaacf78e71 : SUCCESS in 2h 28m 19s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/e556523e739f458ba1fbccd3c20b1313 : SUCCESS in 2h 38m 09s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/9c494700121d437c98dc4ab5a6aecd7e : FAILURE in 2h 01m 58s (non-voting)\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/1c792e1662ff49a4a9466cd44861a375 : SUCCESS in 1h 36m 38s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/d4d1c304463747609e569e7a2569b2c4 : SUCCESS in 2h 53m 23s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/0e0c74f0b0ca4aa587cb964fe7d13358 : SUCCESS in 2h 42m 07s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/158323d5204c4dc2bdefea1a571cda5e : SUCCESS in 2h 20m 02s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5b8da5ceb7d94d1ca7168bcc69dd3ed4 : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":12},{"id":"777f50e56c4eb6b017d25e1c3423a88ac24f1c24","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-10 16:21:10.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/44/669344/12/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/97e5e52/ : SUCCESS in 2h 46m 20s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/44/669344/12/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/e8d6441/ : FAILURE in 1h 42m 02s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/44/669344/12/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/47fc89a/ : SUCCESS in 2h 17m 00s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/44/669344/12/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/f79821e/ : SUCCESS in 3h 06m 26s","accounts_in_message":[],"_revision_number":12},{"id":"0077bb672a860658dac60953926e690008bfb88c","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-10-16 11:57:47.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"d025b02c41924cf2709b6fff227adf608a81e2ef","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-10-21 14:15:22.000000000","message":"Patch Set 12: Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"2aeeb165c5ed77fbdabc1bcf7c8f4b8d5af0a368","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 14:15:45.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"35eb33277850a9b87644d5d039877c8f2f09e777","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 19:52:24.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/4931c8db36b0464c8c22d4884206d9ab : SUCCESS in 2h 02m 44s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/8f04a8a7d0424aec87682d1e44427c0b : SUCCESS in 2h 58m 27s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/4a848f64e78d404598799d0eeff80522 : SUCCESS in 1h 53m 14s\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/b65bc21caef64576a9bf6d0732d30bb6 : SUCCESS in 2h 04m 25s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/34f8110817d94f0e890fa2ee5311e763 : SUCCESS in 2h 29m 12s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/b34ebccf3c9d448394945e2c1c9bcea9 : SUCCESS in 2h 32m 45s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/c5a6063c598248a9a6b623a4e77cf912 : SUCCESS in 2h 38m 12s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/e1c58f60c56648ebb22ea396a3a68420 : SUCCESS in 2h 28m 30s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/60c4fb28b8f44d45bd7b1cfc236f4108 : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":12},{"id":"42fe63643e1afaa827ecfac07fffa32da8db8031","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 19:52:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"4612cf32f114801780bd1078e358bfe51179fc86","revisions":{"1e79cf136cf5b95da2f75261f6291605d1935805":{"kind":"REWORK","_number":1,"created":"2019-07-05 12:15:29.000000000","uploader":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"ref":"refs/changes/44/669344/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/44/669344/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/44/669344/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/44/669344/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/44/669344/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/44/669344/1"}}},"commit":{"parents":[{"commit":"1f59d0c54391369a81eeac434cd9ca2f4a69fe46","subject":"Merge \"Replace git.openstack.org URLs with opendev.org URLs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/1f59d0c54391369a81eeac434cd9ca2f4a69fe46"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-06-28 14:53:10.000000000","tz":-150},"committer":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-07-05 12:15:13.000000000","tz":-240},"subject":"Add a tasks to download amphora image for octavia deployments","message":"Add a tasks to download amphora image for octavia deployments\n\nThis patch adds a task to download the Octavia Amphora image to the\nworking directory on the undercloud so that it can be pulled in by the\nOctavia overcloud deployment tasks.\n\nChange-Id: Ie387f1274059ec80d19c5ef0bf9aed7a3a999e2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/1e79cf136cf5b95da2f75261f6291605d1935805"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/1e79cf136cf5b95da2f75261f6291605d1935805"}]},"branch":"refs/heads/master"},"e94410143013cc8ce7c529499d33ced82b35bbad":{"kind":"REWORK","_number":2,"created":"2019-07-05 17:09:47.000000000","uploader":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"ref":"refs/changes/44/669344/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/44/669344/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/44/669344/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/44/669344/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/44/669344/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/44/669344/2"}}},"commit":{"parents":[{"commit":"1f59d0c54391369a81eeac434cd9ca2f4a69fe46","subject":"Merge \"Replace git.openstack.org URLs with opendev.org URLs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/1f59d0c54391369a81eeac434cd9ca2f4a69fe46"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-06-28 14:53:10.000000000","tz":-150},"committer":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-07-05 17:09:38.000000000","tz":-240},"subject":"Add a tasks to download amphora image for octavia deployments","message":"Add a tasks to download amphora image for octavia deployments\n\nThis patch adds a task to download the Octavia Amphora image to the\nworking directory on the undercloud so that it can be pulled in by the\nOctavia overcloud deployment tasks.\n\nChange-Id: Ie387f1274059ec80d19c5ef0bf9aed7a3a999e2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/e94410143013cc8ce7c529499d33ced82b35bbad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/e94410143013cc8ce7c529499d33ced82b35bbad"}]},"branch":"refs/heads/master"},"1f9deae7d187ed678f84efb7a177bb0db8fb0c61":{"kind":"REWORK","_number":3,"created":"2019-07-08 13:16:28.000000000","uploader":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"ref":"refs/changes/44/669344/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/44/669344/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/44/669344/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/44/669344/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/44/669344/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/44/669344/3"}}},"commit":{"parents":[{"commit":"1f59d0c54391369a81eeac434cd9ca2f4a69fe46","subject":"Merge \"Replace git.openstack.org URLs with opendev.org URLs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/1f59d0c54391369a81eeac434cd9ca2f4a69fe46"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-06-28 14:53:10.000000000","tz":-150},"committer":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-07-08 13:16:23.000000000","tz":-240},"subject":"Add a tasks to download amphora image for octavia deployments","message":"Add a tasks to download amphora image for octavia deployments\n\nThis patch adds a task to download the Octavia Amphora image to the\nworking directory on the undercloud so that it can be pulled in by the\nOctavia overcloud deployment tasks.\n\nChange-Id: Ie387f1274059ec80d19c5ef0bf9aed7a3a999e2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/1f9deae7d187ed678f84efb7a177bb0db8fb0c61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/1f9deae7d187ed678f84efb7a177bb0db8fb0c61"}]},"branch":"refs/heads/master"},"c9105fb2ca035ceddae0389d64d1c1385f07d5e2":{"kind":"REWORK","_number":4,"created":"2019-07-15 11:11:45.000000000","uploader":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"ref":"refs/changes/44/669344/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/44/669344/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/44/669344/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/44/669344/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/44/669344/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/44/669344/4"}}},"commit":{"parents":[{"commit":"ffda8daad0950fe281f554e7996740a117158538","subject":"Merge \"Always prefer local ansible install in reproducer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/ffda8daad0950fe281f554e7996740a117158538"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-06-28 14:53:10.000000000","tz":-150},"committer":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-07-15 11:11:33.000000000","tz":-240},"subject":"Add a tasks to download amphora image for octavia deployments","message":"Add a tasks to download amphora image for octavia deployments\n\nThis patch adds a task to download the Octavia Amphora image to the\nworking directory on the undercloud so that it can be pulled in by the\nOctavia overcloud deployment tasks.\n\nChange-Id: Ie387f1274059ec80d19c5ef0bf9aed7a3a999e2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/c9105fb2ca035ceddae0389d64d1c1385f07d5e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/c9105fb2ca035ceddae0389d64d1c1385f07d5e2"}]},"branch":"refs/heads/master"},"5f2c643237930b56744c75bcefe016ebf79ecb4c":{"kind":"REWORK","_number":5,"created":"2019-07-15 14:04:11.000000000","uploader":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"ref":"refs/changes/44/669344/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/44/669344/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/44/669344/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/44/669344/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/44/669344/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/44/669344/5"}}},"commit":{"parents":[{"commit":"ffda8daad0950fe281f554e7996740a117158538","subject":"Merge \"Always prefer local ansible install in reproducer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/ffda8daad0950fe281f554e7996740a117158538"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-06-28 14:53:10.000000000","tz":-150},"committer":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-07-15 14:03:55.000000000","tz":-240},"subject":"Add a tasks to download amphora image for octavia deployments","message":"Add a tasks to download amphora image for octavia deployments\n\nThis patch adds a task to download the Octavia Amphora image to the\nworking directory on the undercloud so that it can be pulled in by the\nOctavia overcloud deployment tasks.\n\nChange-Id: Ie387f1274059ec80d19c5ef0bf9aed7a3a999e2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/5f2c643237930b56744c75bcefe016ebf79ecb4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/5f2c643237930b56744c75bcefe016ebf79ecb4c"}]},"branch":"refs/heads/master"},"85f9221141a7b072f9041cd84504fab4f8d32ff5":{"kind":"REWORK","_number":6,"created":"2019-07-15 18:14:46.000000000","uploader":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"ref":"refs/changes/44/669344/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/44/669344/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/44/669344/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/44/669344/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/44/669344/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/44/669344/6"}}},"commit":{"parents":[{"commit":"ffda8daad0950fe281f554e7996740a117158538","subject":"Merge \"Always prefer local ansible install in reproducer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/ffda8daad0950fe281f554e7996740a117158538"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-06-28 14:53:10.000000000","tz":-150},"committer":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-07-15 18:13:03.000000000","tz":-240},"subject":"Add a tasks to download amphora image for octavia deployments","message":"Add a tasks to download amphora image for octavia deployments\n\nThis patch adds a task to download the Octavia Amphora image to the\nworking directory on the undercloud so that it can be pulled in by the\nOctavia overcloud deployment tasks.\n\nChange-Id: Ie387f1274059ec80d19c5ef0bf9aed7a3a999e2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/85f9221141a7b072f9041cd84504fab4f8d32ff5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/85f9221141a7b072f9041cd84504fab4f8d32ff5"}]},"branch":"refs/heads/master"},"a702d8c4cbf9accfef28c01989ac9626ed3b8212":{"kind":"REWORK","_number":7,"created":"2019-07-16 12:08:42.000000000","uploader":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"ref":"refs/changes/44/669344/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/44/669344/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/44/669344/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/44/669344/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/44/669344/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/44/669344/7"}}},"commit":{"parents":[{"commit":"ffda8daad0950fe281f554e7996740a117158538","subject":"Merge \"Always prefer local ansible install in reproducer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/ffda8daad0950fe281f554e7996740a117158538"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-06-28 14:53:10.000000000","tz":-150},"committer":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-07-16 12:08:24.000000000","tz":-240},"subject":"Add a tasks to download amphora image for octavia deployments","message":"Add a tasks to download amphora image for octavia deployments\n\nThis patch adds a task to download the Octavia Amphora image to the\nworking directory on the undercloud so that it can be pulled in by the\nOctavia overcloud deployment tasks.\n\nChange-Id: Ie387f1274059ec80d19c5ef0bf9aed7a3a999e2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/a702d8c4cbf9accfef28c01989ac9626ed3b8212"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/a702d8c4cbf9accfef28c01989ac9626ed3b8212"}]},"branch":"refs/heads/master"},"758dad5677f7da8ef2ca4e49bf3bb512dd90c893":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2019-07-24 15:05:40.000000000","uploader":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"ref":"refs/changes/44/669344/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/44/669344/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/44/669344/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/44/669344/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/44/669344/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/44/669344/8"}}},"commit":{"parents":[{"commit":"9406975449c6a84c423378c606df69670cc2e5a4","subject":"Merge \"Use the insecure registry w/o port with podman on RedHat 8 and above\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/9406975449c6a84c423378c606df69670cc2e5a4"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-06-28 14:53:10.000000000","tz":-150},"committer":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-07-24 15:05:35.000000000","tz":-240},"subject":"Add a tasks to download amphora image for octavia deployments","message":"Add a tasks to download amphora image for octavia deployments\n\nThis patch adds a task to download the Octavia Amphora image to the\nworking directory on the undercloud so that it can be pulled in by the\nOctavia overcloud deployment tasks.\n\nChange-Id: Ie387f1274059ec80d19c5ef0bf9aed7a3a999e2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/758dad5677f7da8ef2ca4e49bf3bb512dd90c893"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/758dad5677f7da8ef2ca4e49bf3bb512dd90c893"}]},"branch":"refs/heads/master"},"01820bfc278cc1916fb94002c45aed3924926192":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2019-08-12 12:54:33.000000000","uploader":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"ref":"refs/changes/44/669344/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/44/669344/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/44/669344/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/44/669344/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/44/669344/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/44/669344/9"}}},"commit":{"parents":[{"commit":"fb6c4cc1656b5e89a67e49fe9f0b0f19599309e8","subject":"Merge \"Remove VolumesSnapshotTestJSON from stein skip list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/fb6c4cc1656b5e89a67e49fe9f0b0f19599309e8"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-06-28 14:53:10.000000000","tz":-150},"committer":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-08-12 12:54:26.000000000","tz":-240},"subject":"Add a tasks to download amphora image for octavia deployments","message":"Add a tasks to download amphora image for octavia deployments\n\nThis patch adds a task to download the Octavia Amphora image to the\nworking directory on the undercloud so that it can be pulled in by the\nOctavia overcloud deployment tasks.\n\nChange-Id: Ie387f1274059ec80d19c5ef0bf9aed7a3a999e2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/01820bfc278cc1916fb94002c45aed3924926192"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/01820bfc278cc1916fb94002c45aed3924926192"}]},"branch":"refs/heads/master"},"c13c33f04b166be665582b5011cca0351c8af5a0":{"kind":"REWORK","_number":10,"created":"2019-08-23 18:54:36.000000000","uploader":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"ref":"refs/changes/44/669344/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/44/669344/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/44/669344/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/44/669344/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/44/669344/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/44/669344/10"}}},"commit":{"parents":[{"commit":"6a20363ef06d31099cd6b99831afc20cc33210b7","subject":"Merge \"Define var undercloud_enable_nova\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/6a20363ef06d31099cd6b99831afc20cc33210b7"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-06-28 14:53:10.000000000","tz":-150},"committer":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-08-23 18:54:14.000000000","tz":-240},"subject":"Add a tasks to download amphora image for octavia deployments","message":"Add a tasks to download amphora image for octavia deployments\n\nThis patch adds a task to download the Octavia Amphora image to the\nworking directory on the undercloud so that it can be pulled in by the\nOctavia overcloud deployment tasks.\n\nChange-Id: Ie387f1274059ec80d19c5ef0bf9aed7a3a999e2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/c13c33f04b166be665582b5011cca0351c8af5a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/c13c33f04b166be665582b5011cca0351c8af5a0"}]},"branch":"refs/heads/master"},"ab4c608cedb28a2f56d8137a78c42a0b820c99d5":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2019-09-11 15:30:54.000000000","uploader":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"ref":"refs/changes/44/669344/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/44/669344/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/44/669344/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/44/669344/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/44/669344/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/44/669344/11"}}},"commit":{"parents":[{"commit":"2ec8d4cfe81770ee5ba37347cbe2bf158b0cf6fc","subject":"Merge \"Add *_container_cli settings to extras-common\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/2ec8d4cfe81770ee5ba37347cbe2bf158b0cf6fc"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-06-28 14:53:10.000000000","tz":-150},"committer":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-08-29 12:41:01.000000000","tz":-240},"subject":"Add a tasks to download amphora image for octavia deployments","message":"Add a tasks to download amphora image for octavia deployments\n\nThis patch adds a task to download the Octavia Amphora image to the\nworking directory on the undercloud so that it can be pulled in by the\nOctavia overcloud deployment tasks.\n\nChange-Id: Ie387f1274059ec80d19c5ef0bf9aed7a3a999e2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/ab4c608cedb28a2f56d8137a78c42a0b820c99d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/ab4c608cedb28a2f56d8137a78c42a0b820c99d5"}]},"branch":"refs/heads/master"},"4612cf32f114801780bd1078e358bfe51179fc86":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2019-09-11 15:31:05.000000000","uploader":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"ref":"refs/changes/44/669344/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/44/669344/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/44/669344/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/44/669344/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/44/669344/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/44/669344/12"}}},"commit":{"parents":[{"commit":"0b62606370fce9fc2e76fc96545ff9a548723a72","subject":"Merge \"\u0027is in\u0027 is not a valid Jinja test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/0b62606370fce9fc2e76fc96545ff9a548723a72"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-06-28 14:53:10.000000000","tz":-150},"committer":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-09-11 15:30:58.000000000","tz":-240},"subject":"Add a tasks to download amphora image for octavia deployments","message":"Add a tasks to download amphora image for octavia deployments\n\nThis patch adds a task to download the Octavia Amphora image to the\nworking directory on the undercloud so that it can be pulled in by the\nOctavia overcloud deployment tasks.\n\nChange-Id: Ie387f1274059ec80d19c5ef0bf9aed7a3a999e2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/4612cf32f114801780bd1078e358bfe51179fc86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/4612cf32f114801780bd1078e358bfe51179fc86"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
