)]}'
{"id":"openstack%2Ftripleo-quickstart-extras~670135","triplet_id":"openstack%2Ftripleo-quickstart-extras~master~Iff54ff52c9db65e3675af69fc17dd9461eb596a5","project":"openstack/tripleo-quickstart-extras","branch":"master","topic":"octavia-ci","hashtags":[],"change_id":"Iff54ff52c9db65e3675af69fc17dd9461eb596a5","subject":"Preserve Octavia networks when running tempest","status":"MERGED","created":"2019-07-10 17:06:48.000000000","updated":"2019-10-25 23:16:15.000000000","submitted":"2019-10-25 23:16:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670135-1572045375288-0fd08e9d","meta_rev_id":"c59e33ae7190f4326adc46446849589f5c6918f4","_number":670135,"virtual_id_number":670135,"owner":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-25 23:16:15.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-09-11 18:35:01.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"date":"2019-09-17 22:48:43.000000000","_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":2,"date":"2019-09-12 07:17:48.000000000","_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":1,"date":"2019-10-16 11:49:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"value":2,"date":"2019-10-25 17:32:33.000000000","_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"value":1,"date":"2019-10-25 17:32:33.000000000","_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"},{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-11 08:12:26.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2019-07-11 13:09:10.000000000","updated_by":{"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"},"reviewer":{"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"},"state":"REVIEWER"},{"updated":"2019-08-12 12:54:52.000000000","updated_by":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-27 13:35:24.000000000","updated_by":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"reviewer":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"state":"REVIEWER"},{"updated":"2019-08-28 13:24:59.000000000","updated_by":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"reviewer":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"state":"REVIEWER"},{"updated":"2019-09-09 09:20:59.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2019-09-09 12:41:09.000000000","updated_by":{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},"reviewer":{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},"state":"REVIEWER"},{"updated":"2019-09-09 12:41:20.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"REVIEWER"},{"updated":"2019-09-11 18:35:01.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-12 07:17:48.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2019-10-16 11:49:15.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-10-25 17:32:33.000000000","updated_by":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"reviewer":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"state":"REVIEWER"},{"updated":"2019-10-25 23:16:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3aecd823e39cae1ac6cab59be703c4e70a339ed8","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2019-07-10 17:06:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fb2c09acb65ac8a8c7b0261a81b8823fee389902","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-07-10 17:07:05.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/3626/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"68f16eeda830d2b3c0fe1f3a7224e6866167d055","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-10 17:28:03.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3bfa2cf778e26142fbb4308ec81a5a889f9a0c35","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-10 20:24:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/35/670135/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/50dd2cb/ : RETRY_LIMIT in 6m 49s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/35/670135/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/19a31ba/ : FAILURE in 59m 29s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/35/670135/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/d6e4734/ : FAILURE in 1h 02m 05s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/35/670135/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/071d855/ : FAILURE in 32m 37s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/35/670135/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/81ca9c2/ : FAILURE in 1h 17m 31s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/35/670135/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/2b958c0/ : SUCCESS in 3h 15m 45s (non-voting)\n- tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020 http://logs.rdoproject.org/35/670135/1/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020/e189b7a/ : FAILURE in 59m 20s","accounts_in_message":[],"_revision_number":1},{"id":"4b95ef802cc2074e9dbabf3aee93c912bcf0d799","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 22:22:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/35/670135/1/check/tripleo-ci-centos-7-undercloud-containers/16440f7/ : SUCCESS in 1h 35m 27s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/35/670135/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/f816e4c/ : FAILURE in 2h 05m 20s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/35/670135/1/check/tripleo-ci-centos-7-containers-multinode/7b0c607/ : SUCCESS in 2h 11m 15s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/35/670135/1/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/4ececd6/ : SUCCESS in 2h 29m 16s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container http://logs.openstack.org/35/670135/1/check/tripleo-ci-centos-7-scenario012-multinode-oooq-container/e84e37c/ : FAILURE in 2h 25m 04s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/35/670135/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/70a93aa/ : SUCCESS in 2h 07m 46s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/35/670135/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/5ae3e92/ : FAILURE in 2h 18m 01s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/35/670135/1/check/tripleo-ci-centos-7-standalone/6fd019a/ : SUCCESS in 1h 46m 44s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/35/670135/1/check/tripleo-ci-fedora-28-standalone/36ca541/ : SUCCESS in 1h 40m 25s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/35/670135/1/check/tripleo-ci-centos-7-standalone-upgrade/c20069e/ : FAILURE in 1h 58m 24s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/35/670135/1/check/tripleo-ci-centos-7-standalone-upgrade-stein/c6271bc/ : SUCCESS in 2h 31m 39s\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/35/670135/1/check/tripleo-ci-centos-7-scenario001-standalone/06cb9a1/ : SUCCESS in 1h 43m 34s\n- tripleo-ci-centos-7-scenario002-standalone http://logs.openstack.org/35/670135/1/check/tripleo-ci-centos-7-scenario002-standalone/28562a1/ : SUCCESS in 1h 38m 44s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/35/670135/1/check/tripleo-ci-centos-7-scenario003-standalone/7076c41/ : SUCCESS in 1h 10m 21s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/35/670135/1/check/tripleo-ci-centos-7-scenario004-standalone/b290c9e/ : SUCCESS in 1h 41m 02s\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/35/670135/1/check/tripleo-ci-centos-7-standalone-os-tempest/2e2fff6/ : SUCCESS in 1h 12m 06s\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/35/670135/1/check/tripleo-ci-centos-7-containers-multinode-stein/1ca2e8f/ : SUCCESS in 2h 19m 54s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/35/670135/1/check/tripleo-ci-centos-7-containers-multinode-rocky/33d09b8/ : FAILURE in 38m 43s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/35/670135/1/check/tripleo-ci-centos-7-containers-multinode-queens/7205460/ : SUCCESS in 2h 19m 19s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/35/670135/1/check/tripleo-ci-centos-7-containers-multinode-pike/b29f08a/ : SUCCESS in 2h 16m 12s\n- tripleo-ci-centos-7-scenario010-standalone http://logs.openstack.org/35/670135/1/check/tripleo-ci-centos-7-scenario010-standalone/bd4eeda/ : SUCCESS in 1h 13m 53s\n- tripleo-ci-centos-7-scenario012-standalone http://logs.openstack.org/35/670135/1/check/tripleo-ci-centos-7-scenario012-standalone/203fece/ : SUCCESS in 1h 46m 39s (non-voting)\n- openstack-tox-linters http://logs.openstack.org/35/670135/1/check/openstack-tox-linters/4da6a88/ : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":1},{"id":"7a449d192c64fa0340d5ebac0232ccf6d647f468","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2019-07-11 08:11:40.000000000","message":"Patch Set 1: Code-Review-1\n\nso there is no tempest there http://logs.openstack.org/35/670135/1/check/tripleo-ci-centos-7-scenario010-standalone/bd4eeda/logs/undercloud/home/zuul/ \n\nI went digging and looks like it isn\u0027t enabled in the job definition?\n\nhttps://github.com/openstack/tripleo-ci/blob/master/zuul.d/standalone-jobs.yaml#L648\n\nSO either you need to remove that \"run_tempest: false\" or, keep it, and add the \"use_os_tempest: true\" (e.g. see scenarios 1-4) but you\u0027ll need to tweak the params too (e.g. \"tempest_whitelist\" becomes \"tempest_test_whitelist\")\n\n\"run_tempest\" was the old way and \"use_os_tempest\" uses upstream tempest \n\n+1 on skipping but lets get it to run before we merge this?","accounts_in_message":[],"_revision_number":1},{"id":"3f4d964667fd92bc8e78f6a639d1b0b489706f31","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2019-07-11 08:13:21.000000000","message":"Patch Set 1:\n\nadding the tripleo-ci tempest folks\n\nis there any reason to not have \" use_os_tempest\" for scen10 (see my last comment here )","accounts_in_message":[],"_revision_number":1},{"id":"fa495f49c9a4ea461e8c4cc283a60d5e9d776626","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-11 08:15:43.000000000","message":"Patch Set 1:\n\nMarios, please see https://review.opendev.org/#/q/topic:octavia-ci and specifically this one where tempest is running with Octavia tests http://logs.openstack.org/40/669740/5/check/tripleo-ci-centos-7-scenario010-standalone/45bdb83/logs/tempest.html.gz","accounts_in_message":[],"_revision_number":1},{"id":"996e65cef0fe5a2bb56de874caf47bed1f7d3842","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2019-07-11 08:24:20.000000000","message":"Patch Set 1: Code-Review+2\n\n\u003e Marios, please see https://review.opendev.org/#/q/topic:octavia-ci\n \u003e and specifically this one where tempest is running with Octavia\n \u003e tests http://logs.openstack.org/40/669740/5/check/tripleo-ci-centos-7-scenario010-standalone/45bdb83/logs/tempest.html.gz\n\nack thanks revoting then (i see that https://review.opendev.org/#/c/669740/ dependson this)\n\n feels a bit weird that we\u0027ll be unskipping them (since you need this first and its depended-on by the other patch) but they won\u0027t be running","accounts_in_message":[],"_revision_number":1},{"id":"0db0a0594f1c04e975988e1ae4315fca7c35398f","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-11 08:56:45.000000000","message":"Patch Set 1: Code-Review+1\n\n\u003e so there is no tempest there http://logs.openstack.org/35/670135/1/check/tripleo-ci-centos-7-scenario010-standalone/bd4eeda/logs/undercloud/home/zuul/\n \u003e \n \u003e I went digging and looks like it isn\u0027t enabled in the job\n \u003e definition?\n \u003e \n \u003e https://github.com/openstack/tripleo-ci/blob/master/zuul.d/standalone-jobs.yaml#L648\n \u003e \n \u003e SO either you need to remove that \"run_tempest: false\" or, keep it,\n \u003e and add the \"use_os_tempest: true\" (e.g. see scenarios 1-4) but\n \u003e you\u0027ll need to tweak the params too (e.g. \"tempest_whitelist\"\n \u003e becomes \"tempest_test_whitelist\")\n \u003e \n \u003e \"run_tempest\" was the old way and \"use_os_tempest\" uses upstream\n \u003e tempest\n \u003e \n \u003e +1 on skipping but lets get it to run before we merge this?\n\n \u003e adding the tripleo-ci tempest folks\n \u003e \n \u003e is there any reason to not have \" use_os_tempest\" for scen10 (see\n \u003e my last comment here )\n\nFS010 is not yet ported to os_tempest, that\u0027s why use_os_tempest flag is not used.","accounts_in_message":[],"_revision_number":1},{"id":"764c8f4cb9ffdabb1969ef6b7a9eddfde980753d","author":{"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"},"date":"2019-07-11 13:09:10.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c92d40d931cef891d415189df34c4462910cb8ba","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2019-07-15 11:11:45.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"84423ec61105511ce2d9c05f7f7d6e72afbcbb83","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-07-15 11:12:06.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/3636/ : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"72ff8bffaf30a398d848c83931a89c3c13238cf9","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2019-07-15 14:04:11.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"f382ad70383ace18441f037ae5c7f4a729ab9433","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-07-15 14:04:30.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/3638/ : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"6aa14a13dd40e394f88827b825b128fbc46a1cbd","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2019-07-15 18:14:46.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"be6722960f50e879d037d9e00d7d736cc73bef22","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-07-15 18:15:02.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/3640/ : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"a1ad2605e823c4871160a55a304bff35f8c02cbe","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-15 22:22:08.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/35/670135/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/c2fdb0a/ : RETRY_LIMIT in 3m 50s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/35/670135/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/2e30a45/ : RETRY_LIMIT in 6m 19s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/35/670135/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/c2b4b99/ : RETRY_LIMIT in 3m 02s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/35/670135/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/1362d4a/ : SUCCESS in 2h 17m 51s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/35/670135/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/25241d9/ : SUCCESS in 2h 33m 52s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/35/670135/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/0bfdef0/ : FAILURE in 2h 03m 45s (non-voting)\n- tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020 http://logs.rdoproject.org/35/670135/4/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020/49be93d/ : RETRY_LIMIT in 5m 22s","accounts_in_message":[],"_revision_number":4},{"id":"9d740ef135b94b45162ff0ba7706d65869762e7e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 22:59:34.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/35/670135/4/check/tripleo-ci-centos-7-undercloud-containers/a1843c9/ : SUCCESS in 1h 36m 21s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/35/670135/4/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/00f42e8/ : FAILURE in 1h 27m 05s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/35/670135/4/check/tripleo-ci-centos-7-containers-multinode/a213749/ : SUCCESS in 2h 12m 56s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/35/670135/4/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/231af4d/ : SUCCESS in 2h 15m 52s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container http://logs.openstack.org/35/670135/4/check/tripleo-ci-centos-7-scenario012-multinode-oooq-container/056ea65/ : FAILURE in 2h 17m 46s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/35/670135/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/9baf7c7/ : SUCCESS in 2h 21m 53s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/35/670135/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/5baeeef/ : POST_FAILURE in 2h 11m 27s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/35/670135/4/check/tripleo-ci-centos-7-standalone/89ec936/ : SUCCESS in 1h 39m 31s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/35/670135/4/check/tripleo-ci-fedora-28-standalone/1e42ab8/ : SUCCESS in 1h 41m 03s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/35/670135/4/check/tripleo-ci-centos-7-standalone-upgrade/cb738fc/ : FAILURE in 1h 46m 39s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/35/670135/4/check/tripleo-ci-centos-7-scenario001-standalone/216ae36/ : SUCCESS in 1h 31m 57s\n- tripleo-ci-centos-7-scenario002-standalone http://logs.openstack.org/35/670135/4/check/tripleo-ci-centos-7-scenario002-standalone/f2f27a2/ : SUCCESS in 1h 31m 06s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/35/670135/4/check/tripleo-ci-centos-7-scenario003-standalone/af9b294/ : SUCCESS in 1h 17m 16s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/35/670135/4/check/tripleo-ci-centos-7-scenario004-standalone/2507cb4/ : SUCCESS in 1h 55m 24s\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/35/670135/4/check/tripleo-ci-centos-7-standalone-os-tempest/8b7d08b/ : SUCCESS in 1h 21m 53s\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/35/670135/4/check/tripleo-ci-centos-7-containers-multinode-stein/510880e/ : SUCCESS in 2h 18m 26s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/35/670135/4/check/tripleo-ci-centos-7-containers-multinode-rocky/1afeca7/ : SUCCESS in 2h 14m 08s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/35/670135/4/check/tripleo-ci-centos-7-containers-multinode-queens/1775ada/ : SUCCESS in 2h 11m 54s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/35/670135/4/check/tripleo-ci-centos-7-containers-multinode-pike/09c9cb2/ : SUCCESS in 2h 16m 16s\n- tripleo-ci-centos-7-scenario010-standalone http://logs.openstack.org/35/670135/4/check/tripleo-ci-centos-7-scenario010-standalone/785cf58/ : SUCCESS in 1h 11m 20s\n- tripleo-ci-centos-7-scenario012-standalone http://logs.openstack.org/35/670135/4/check/tripleo-ci-centos-7-scenario012-standalone/4cbba1c/ : SUCCESS in 1h 32m 47s (non-voting)\n- openstack-tox-linters http://logs.openstack.org/35/670135/4/check/openstack-tox-linters/416ce38/ : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":4},{"id":"732232de142d9ed7e33857d2601ec701f948f25a","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-07-16 09:28:04.000000000","message":"Patch Set 4:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":4},{"id":"80f1bd8360f33775065bd72d11a9eb3289ea2ce7","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2019-07-16 12:08:42.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"1227394da1e0a46c0c9af8c42e19a17101d036b5","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-07-16 12:09:02.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/3642/ : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"6a43f1a3b737de2b2cbf6910ba7b89694cd73a2b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 14:59:49.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/35/670135/5/check/tripleo-ci-centos-7-undercloud-containers/8284c9a/ : FAILURE in 57m 04s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/35/670135/5/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/b89efa2/ : FAILURE in 2h 09m 40s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/35/670135/5/check/tripleo-ci-centos-7-containers-multinode/5696357/ : FAILURE in 57m 44s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/35/670135/5/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/4dddb4e/ : FAILURE in 1h 02m 21s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container http://logs.openstack.org/35/670135/5/check/tripleo-ci-centos-7-scenario012-multinode-oooq-container/dadb577/ : FAILURE in 58m 04s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/35/670135/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/298e4bb/ : FAILURE in 1h 03m 56s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/35/670135/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/faca5d2/ : FAILURE in 1h 02m 03s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/35/670135/5/check/tripleo-ci-centos-7-standalone/598d1d4/ : FAILURE in 48m 38s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/35/670135/5/check/tripleo-ci-fedora-28-standalone/002c926/ : FAILURE in 40m 58s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/35/670135/5/check/tripleo-ci-centos-7-standalone-upgrade/f51a773/ : FAILURE in 1h 45m 03s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/35/670135/5/check/tripleo-ci-centos-7-scenario001-standalone/c3a6310/ : SUCCESS in 1h 36m 57s\n- tripleo-ci-centos-7-scenario002-standalone http://logs.openstack.org/35/670135/5/check/tripleo-ci-centos-7-scenario002-standalone/610625c/ : FAILURE in 55m 32s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/35/670135/5/check/tripleo-ci-centos-7-scenario003-standalone/bcfb53d/ : SUCCESS in 1h 13m 31s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/35/670135/5/check/tripleo-ci-centos-7-scenario004-standalone/d7ec37f/ : SUCCESS in 1h 40m 49s\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/35/670135/5/check/tripleo-ci-centos-7-standalone-os-tempest/f297ba7/ : FAILURE in 45m 44s\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/35/670135/5/check/tripleo-ci-centos-7-containers-multinode-stein/ec00a78/ : SUCCESS in 2h 39m 45s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/35/670135/5/check/tripleo-ci-centos-7-containers-multinode-rocky/1622ab6/ : SUCCESS in 2h 10m 49s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/35/670135/5/check/tripleo-ci-centos-7-containers-multinode-queens/8dcfd21/ : SUCCESS in 2h 16m 27s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/35/670135/5/check/tripleo-ci-centos-7-containers-multinode-pike/85f72e9/ : SUCCESS in 2h 25m 55s\n- tripleo-ci-centos-7-scenario010-standalone http://logs.openstack.org/35/670135/5/check/tripleo-ci-centos-7-scenario010-standalone/25d5346/ : SUCCESS in 1h 19m 19s\n- tripleo-ci-centos-7-scenario012-standalone http://logs.openstack.org/35/670135/5/check/tripleo-ci-centos-7-scenario012-standalone/7e95b74/ : SUCCESS in 1h 59m 02s (non-voting)\n- openstack-tox-linters http://logs.openstack.org/35/670135/5/check/openstack-tox-linters/32ecfa6/ : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":5},{"id":"65cc14b9c2da41c14c5d344707fecbad6a06b629","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-16 17:23:16.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/35/670135/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/f9246be/ : SUCCESS in 2h 37m 11s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/35/670135/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/379d348/ : SUCCESS in 2h 46m 34s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/35/670135/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/0ce0fe1/ : FAILURE in 2h 17m 12s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/35/670135/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/02db3ca/ : SUCCESS in 2h 21m 16s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/35/670135/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/d69862b/ : SUCCESS in 2h 41m 07s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/35/670135/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/eb94918/ : FAILURE in 1h 42m 31s (non-voting)\n- tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020 http://logs.rdoproject.org/35/670135/5/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020/eb15638/ : FAILURE in 5h 12m 24s","accounts_in_message":[],"_revision_number":5},{"id":"f98c2ae53e1ccac6cf5263def70e7e41e08b2d87","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2019-07-18 15:07:07.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"4cae57dc4a97a605661adc1ee2006eb7850beba6","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-07-18 15:07:23.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/3653/ : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"79a0b939c1419c23626ec043d87b8e8fd13fd51a","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-18 16:37:16.000000000","message":"Patch Set 6:\n\nWe are deprecating validate-tempest role in favor of os_tempest, can we move this stuff to os_tempest role?","accounts_in_message":[],"_revision_number":6},{"id":"be8345f08b2dbca95ff24e858723554a487820b9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 20:02:53.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/35/670135/6/check/tripleo-ci-centos-7-undercloud-containers/672e6cc/ : SUCCESS in 1h 33m 44s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/35/670135/6/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/578e251/ : FAILURE in 2h 06m 13s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/35/670135/6/check/tripleo-ci-centos-7-containers-multinode/6e1e407/ : SUCCESS in 2h 09m 39s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/35/670135/6/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/c388e27/ : SUCCESS in 2h 27m 54s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container http://logs.openstack.org/35/670135/6/check/tripleo-ci-centos-7-scenario012-multinode-oooq-container/c8adecf/ : FAILURE in 52m 55s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/35/670135/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/2dfc5a4/ : SUCCESS in 2h 17m 51s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/35/670135/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/9f5a7ea/ : FAILURE in 2h 27m 11s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/35/670135/6/check/tripleo-ci-centos-7-standalone/867eb3f/ : SUCCESS in 1h 57m 27s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/35/670135/6/check/tripleo-ci-fedora-28-standalone/fdb626b/ : FAILURE in 23m 37s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/35/670135/6/check/tripleo-ci-centos-7-standalone-upgrade/eafce00/ : FAILURE in 17m 37s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/35/670135/6/check/tripleo-ci-centos-7-scenario001-standalone/ac332e4/ : FAILURE in 18m 51s\n- tripleo-ci-centos-7-scenario002-standalone http://logs.openstack.org/35/670135/6/check/tripleo-ci-centos-7-scenario002-standalone/436150a/ : SUCCESS in 1h 28m 26s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/35/670135/6/check/tripleo-ci-centos-7-scenario003-standalone/6e90d67/ : SUCCESS in 1h 12m 32s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/35/670135/6/check/tripleo-ci-centos-7-scenario004-standalone/078b1a2/ : SUCCESS in 1h 37m 45s\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/35/670135/6/check/tripleo-ci-centos-7-standalone-os-tempest/48a6e08/ : SUCCESS in 1h 07m 41s\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/35/670135/6/check/tripleo-ci-centos-7-containers-multinode-stein/1c16c59/ : SUCCESS in 2h 16m 00s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/35/670135/6/check/tripleo-ci-centos-7-containers-multinode-rocky/4776e9d/ : SUCCESS in 2h 27m 06s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/35/670135/6/check/tripleo-ci-centos-7-containers-multinode-queens/e9f1082/ : SUCCESS in 2h 21m 47s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/35/670135/6/check/tripleo-ci-centos-7-containers-multinode-pike/c1522ce/ : SUCCESS in 2h 22m 24s\n- tripleo-ci-centos-7-scenario010-standalone http://logs.openstack.org/35/670135/6/check/tripleo-ci-centos-7-scenario010-standalone/6b01374/ : FAILURE in 18m 38s\n- tripleo-ci-centos-7-scenario012-standalone http://logs.openstack.org/35/670135/6/check/tripleo-ci-centos-7-scenario012-standalone/cf892da/ : SUCCESS in 2h 03m 53s (non-voting)\n- openstack-tox-linters http://logs.openstack.org/35/670135/6/check/openstack-tox-linters/41cbedb/ : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":6},{"id":"a0b32b9466c117abf0e7ff9d8452e96ae787319d","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2019-07-19 07:10:49.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"9fc086513a1b91dd81271e700b883644f0b2b6c5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-19 13:41:03.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/35/670135/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/9813bae/ : RETRY_LIMIT in 5m 18s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/35/670135/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/de8511a/ : SUCCESS in 3h 17m 02s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/35/670135/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/5cfc91a/ : RETRY_LIMIT in 6m 37s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/35/670135/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/6c62e2d/ : RETRY_LIMIT in 6m 46s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/35/670135/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/f245e10/ : RETRY_LIMIT in 4m 21s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/35/670135/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/d2dedc9/ : RETRY_LIMIT in 3m 40s (non-voting)\n- tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020 http://logs.rdoproject.org/35/670135/6/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020/3dca445/ : RETRY_LIMIT in 6m 04s","accounts_in_message":[],"_revision_number":6},{"id":"21b8f01c8b46643597b992fdb38168c531e4ff86","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2019-07-24 15:05:40.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"a7201f36369c75ebc8da77b29e109dd5e9e1893b","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-07-24 15:05:58.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/3683/ : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"0f03be4ac347e3ee11dd7120b011c0f4225d105a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 18:58:23.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/35/670135/7/check/tripleo-ci-centos-7-undercloud-containers/ccf72dc/ : SUCCESS in 1h 36m 46s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/35/670135/7/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/6fc9237/ : FAILURE in 2h 04m 09s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/35/670135/7/check/tripleo-ci-centos-7-containers-multinode/996e5f7/ : SUCCESS in 2h 26m 28s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/35/670135/7/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/76059d8/ : SUCCESS in 2h 32m 27s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container http://logs.openstack.org/35/670135/7/check/tripleo-ci-centos-7-scenario012-multinode-oooq-container/df4814f/ : FAILURE in 2h 27m 09s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/35/670135/7/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/98509f1/ : SUCCESS in 2h 29m 58s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/35/670135/7/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/d23554b/ : SUCCESS in 2h 22m 47s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/35/670135/7/check/tripleo-ci-centos-7-standalone/3c20266/ : FAILURE in 1h 01m 42s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/35/670135/7/check/tripleo-ci-fedora-28-standalone/b5ddde2/ : SUCCESS in 1h 35m 15s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/35/670135/7/check/tripleo-ci-centos-7-standalone-upgrade/6a6e8b6/ : FAILURE in 1h 49m 27s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/35/670135/7/check/tripleo-ci-centos-7-scenario001-standalone/db2b8aa/ : SUCCESS in 1h 52m 38s\n- tripleo-ci-centos-7-scenario002-standalone http://logs.openstack.org/35/670135/7/check/tripleo-ci-centos-7-scenario002-standalone/678ab9c/ : SUCCESS in 1h 38m 16s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/35/670135/7/check/tripleo-ci-centos-7-scenario003-standalone/8e117e1/ : SUCCESS in 1h 14m 18s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/35/670135/7/check/tripleo-ci-centos-7-scenario004-standalone/a8e2595/ : SUCCESS in 1h 48m 28s\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/35/670135/7/check/tripleo-ci-centos-7-standalone-os-tempest/fd230d9/ : FAILURE in 49m 39s\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/35/670135/7/check/tripleo-ci-centos-7-containers-multinode-stein/5adc97c/ : SUCCESS in 2h 32m 03s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/35/670135/7/check/tripleo-ci-centos-7-containers-multinode-rocky/eb7c86e/ : SUCCESS in 2h 31m 00s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/35/670135/7/check/tripleo-ci-centos-7-containers-multinode-queens/fd84496/ : SUCCESS in 2h 27m 10s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/35/670135/7/check/tripleo-ci-centos-7-containers-multinode-pike/8849f66/ : SUCCESS in 2h 18m 21s\n- tripleo-ci-centos-7-scenario010-standalone http://logs.openstack.org/35/670135/7/check/tripleo-ci-centos-7-scenario010-standalone/74aa012/ : SUCCESS in 1h 17m 29s\n- tripleo-ci-centos-7-scenario012-standalone http://logs.openstack.org/35/670135/7/check/tripleo-ci-centos-7-scenario012-standalone/1846455/ : SUCCESS in 2h 26m 29s (non-voting)\n- openstack-tox-linters http://logs.openstack.org/35/670135/7/check/openstack-tox-linters/0737ad7/ : SUCCESS in 7m 30s","accounts_in_message":[],"_revision_number":7},{"id":"906efa407ee1994d961719937381f2e044f3a2e8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-24 18:58:49.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 finger://ze05.softwarefactory-project.io:7979/816abdfc02c54959a7ac14d40f9d355f : POST_FAILURE in 3h 33m 16s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 finger://ze05.softwarefactory-project.io:7979/3be7b1a5a9f64a469d998c98bfc25718 : POST_FAILURE in 3h 49m 04s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch finger://ze06.softwarefactory-project.io:7979/53dd2b3971ab40f4b0d9133faa28258b : POST_FAILURE in 3h 17m 06s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch finger://ze01.softwarefactory-project.io:7979/118f5afc513e419e841da89353b76706 : POST_FAILURE in 3h 18m 18s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch finger://ze01.softwarefactory-project.io:7979/04dc465a5e204ce49e63f90c8a22599c : POST_FAILURE in 3h 35m 05s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 finger://ze04.softwarefactory-project.io:7979/377382ecfae845f5b04ed7cbc61cb80c : POST_FAILURE in 3h 50m 52s (non-voting)\n- tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020 http://logs.rdoproject.org/35/670135/7/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020/87bb7c9/ : FAILURE in 2h 10m 47s","accounts_in_message":[],"_revision_number":7},{"id":"6025fffb3c12f367f69f768c77fc508ca0c91f51","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2019-08-12 12:54:33.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"f30d155b73816627077bf706922b8f7b8e77c3fe","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-08-12 12:54:52.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/3769/ : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"e188008e2a5e92187b6cfcec0345090d6ee7d340","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-12 16:32:22.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 finger://ze05.softwarefactory-project.io:7979/5c6a0b7e23364a2dac833df79faee396 : POST_FAILURE in 3h 26m 57s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 finger://ze07.softwarefactory-project.io:7979/cac2d40596a145f7bda2c84e0b6336f1 : POST_FAILURE in 3h 33m 47s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch finger://ze04.softwarefactory-project.io:7979/e8ef2385c6c5412197040f3c43acef8b : POST_FAILURE in 3h 17m 37s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch finger://ze03.softwarefactory-project.io:7979/55b8f52ac4894986ac9e59c1120c2da2 : POST_FAILURE in 3h 20m 33s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch finger://ze02.softwarefactory-project.io:7979/9c9a4066ce0d4954b04bbab07cc2beaa : POST_FAILURE in 3h 35m 46s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 finger://ze02.softwarefactory-project.io:7979/84fb2c31432f400daa1511a923cf5756 : POST_FAILURE in 3h 34m 40s (non-voting)\n- tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020 http://logs.rdoproject.org/35/670135/8/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020/f7489d1/ : FAILURE in 1h 46m 14s\n- tripleo-ci-rhel-8-standalone-rdo http://logs.rdoproject.org/35/670135/8/openstack-check/tripleo-ci-rhel-8-standalone-rdo/6842d80/ : FAILURE in 46m 27s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"376568e5d9ce1dba879b7cb37040c60aa06ddd45","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 16:56:14.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/e39a34f3d5b54659bf89e48d99e4b113 : SUCCESS in 1h 35m 17s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/9ce2cffacd4342d9aef80f7626014897 : FAILURE in 2h 05m 03s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/00c3c764680e46658df46147a01d53ba : SUCCESS in 2h 38m 39s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/2ca6b841f7e44ea49e5437c5d90d58f8 : SUCCESS in 2h 20m 42s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/77b4c13cce8c4fd1a70fcc8fa83b33f8 : FAILURE in 2h 39m 22s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/6ad2eeac7bd94158aa3f5653875b6702 : SUCCESS in 2h 16m 57s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/e28cf857c8464c1794af3011090f30d0 : SUCCESS in 2h 07m 35s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/8b19c7e50e0e4d8a99176561c2269fa0 : SUCCESS in 1h 41m 28s\n- tripleo-ci-fedora-28-standalone https://zuul.opendev.org/t/openstack/build/ba43747a2bf14f538d3cee58c0e67fe2 : SUCCESS in 1h 50m 10s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/2c96f5444a014738885ab014fa01ed87 : SUCCESS in 2h 00m 18s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/7f6cd2e62cca48aab70fcdd3099fdd8a : SUCCESS in 1h 52m 16s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/c5ab400c30514bc19a1258f90af73e1b : SUCCESS in 1h 46m 57s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/7945cd946bf6475ca03464f8379e5e45 : SUCCESS in 1h 28m 42s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/0519cd3b994e444c9a71473649bb8637 : SUCCESS in 1h 51m 28s\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/70d706f5421442e68905ec7d0f51ca9d : SUCCESS in 1h 07m 03s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/ac243f428cb3454caf65274b6b557832 : SUCCESS in 2h 09m 29s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/6e403829d2de48a29fc116787f035616 : SUCCESS in 2h 22m 04s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/3c246e52aee043c8885d25b759080041 : SUCCESS in 2h 16m 05s\n- tripleo-ci-centos-7-containers-multinode-pike https://zuul.opendev.org/t/openstack/build/6f27259543484b88a161182021e7be29 : SUCCESS in 2h 23m 11s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/cfd7dd679b91400b958ba0655efb2c57 : SUCCESS in 1h 11m 33s (non-voting)\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/5ac86d2d79d747efa51941850923aede : SUCCESS in 1h 49m 21s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7d0638bcef5248df8eba4cebd089b071 : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":8},{"id":"a625dd88fa69a24a06d4d9bb1d370645794d5150","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2019-08-13 07:48:46.000000000","message":"Patch Set 8: Code-Review-1\n\nPlease, notice that we are moving from validate-tempest role, to openstack-ansible-os_tempest role, we are still using (for now) the skip list on this role, but in the future, this skip list will be moved to somewhere else. I sugest you take a look in os_tempest and which jobs are using it.","accounts_in_message":[],"_revision_number":8},{"id":"307805297613ca7bc32d807a311afaa533bfca1e","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2019-08-13 16:14:56.000000000","message":"Patch Set 8:\n\n\u003e Please, notice that we are moving from validate-tempest role, to\n \u003e openstack-ansible-os_tempest role, we are still using (for now) the\n \u003e skip list on this role, but in the future, this skip list will be\n \u003e moved to somewhere else. I sugest you take a look in os_tempest and\n \u003e which jobs are using it.\nI understand but the last I looked os_tempest and octavia wasn\u0027t working properly. Do you have any updates on this?","accounts_in_message":[],"_revision_number":8},{"id":"9d784c0a28a093b4976d5a7f76123733572c3991","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2019-08-23 18:54:36.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"9802fd53356859d299b1db16e6df909f90df7022","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 22:18:33.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/00d1cb7fdf354deb94acedf04133096a : SUCCESS in 1h 51m 08s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/c62e99ea6f3d4bc48516b33748b2cb00 : FAILURE in 2h 15m 50s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/5c28a932773b45c9864044038a5c5c02 : SUCCESS in 2h 45m 01s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/96556b195134441093e2cf597e83f1a9 : SUCCESS in 2h 48m 55s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/690983e42b0945fdaff500e84966883f : FAILURE in 2h 59m 56s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/76c72e927a304e35ab1d350e907e538f : SUCCESS in 2h 39m 21s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/334098b64b6e4cd38bbcd0a3aeeefc2c : SUCCESS in 2h 44m 42s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/581faace4d074fa38cf8fdfe0ed69275 : SUCCESS in 2h 11m 33s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/e644bd4c0d1f4218b5633971941b2e72 : SUCCESS in 2h 36m 22s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/a7a0e23b401343aa83d71d533179f305 : SUCCESS in 2h 19m 56s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/8e68fe518f1d4ada87e9240cfa2f7e3f : SUCCESS in 2h 01m 40s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/2e42ad7f8514432d8e6a170d7bf41463 : SUCCESS in 1h 36m 45s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/abcb47cb276d43338a9b5af3c0384e19 : SUCCESS in 2h 05m 40s\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/4708d0c03f85492c825327b29c28be70 : SUCCESS in 1h 42m 28s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/952a20b01c254aa2a495d35f4c93bb5f : SUCCESS in 2h 39m 48s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/a684561fd00942a7b202397a455c60f7 : SUCCESS in 2h 26m 47s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/d4c390d4648248d18b9fb86ffa175d35 : SUCCESS in 2h 24m 25s\n- tripleo-ci-centos-7-containers-multinode-pike https://zuul.opendev.org/t/openstack/build/01276411d10d45299351aaa8ca11a16e : SUCCESS in 2h 21m 48s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/981817bf7bbb42a6acbd56c226d21d8f : SUCCESS in 1h 28m 14s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/9282933d986c4e38bd2c5cb825b56043 : FAILURE in 1h 36m 05s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6a0fb324940747a197e28878ff9261b6 : SUCCESS in 6m 14s","accounts_in_message":[],"_revision_number":9},{"id":"e2f8a35f8a4cf28e2e7ff9efab6fd382b9ad36c5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-24 00:01:49.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/35/670135/9/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/dbc41f9/ : SUCCESS in 3h 26m 26s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/35/670135/9/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/a317a4b/ : FAILURE in 3h 32m 25s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/35/670135/9/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/5557086/ : FAILURE in 1h 06m 16s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/35/670135/9/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/e7f8f43/ : SUCCESS in 2h 19m 56s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/35/670135/9/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/7b2465f/ : FAILURE in 1h 42m 17s\n- tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020 http://logs.rdoproject.org/35/670135/9/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020/521c7c9/ : FAILURE in 5h 05m 20s","accounts_in_message":[],"_revision_number":9},{"id":"a550a8822daccd6823018a04eb9932acd7ffb1d6","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2019-08-27 12:19:28.000000000","message":"Patch Set 9:\n\n\u003e \u003e Please, notice that we are moving from validate-tempest role, to\n \u003e \u003e openstack-ansible-os_tempest role, we are still using (for now)\n \u003e the\n \u003e \u003e skip list on this role, but in the future, this skip list will be\n \u003e \u003e moved to somewhere else. I sugest you take a look in os_tempest\n \u003e and\n \u003e \u003e which jobs are using it.\n \u003e I understand but the last I looked os_tempest and octavia wasn\u0027t\n \u003e working properly. Do you have any updates on this?\n\nAlso please note that there are lot of moving parts to getting tempest running with octavia and we are eager to land the basics ASAP. We would of course use this as a baseline to  move to os_tempest and work out the issues there.","accounts_in_message":[],"_revision_number":9},{"id":"78438618a3eeab020526a0bd298bfab61c489a6d","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2019-08-27 13:35:24.000000000","message":"Patch Set 9: Code-Review+1\n\nChanging my vote, since you are aware of the deprecation and you need it :)","accounts_in_message":[],"_revision_number":9},{"id":"1b1d2a4c3c735d0379f943b31e9120a902da41da","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-08-28 13:24:59.000000000","message":"Patch Set 9: Code-Review-1\n\nAs it seems in job tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020 octavia tests still don\u0027t run: http://logs.rdoproject.org/35/670135/9/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020/521c7c9/","accounts_in_message":[],"_revision_number":9},{"id":"5401e5c345c7493ca8d2142f9b7d5e4b0152eb07","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-08-28 13:33:26.000000000","message":"Patch Set 9:\n\nWe need add a small change here:\nhttps://opendev.org/openstack/tripleo-quickstart-extras/src/branch/master/roles/validate-tempest/templates/configure-tempest.sh.j2#L44\nopenstack role show load-balancer_member \u003e /dev/null || openstack role create load-balancer_member \nput it in conditional when octavia is enabled.\n\nAnd here is the error: https://logs.rdoproject.org/35/670135/9/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020/521c7c9/logs/undercloud/home/zuul/tempest.log.txt.gz#_2019-08-23_23_07_55","accounts_in_message":[],"_revision_number":9},{"id":"9d853d0e01ed99a311735a3aa75857ae2938820f","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2019-08-29 12:46:17.000000000","message":"Patch Set 9:\n\n\u003e As it seems in job tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020\n \u003e octavia tests still don\u0027t run: http://logs.rdoproject.org/35/670135/9/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020/521c7c9/\n\nActually, octavia isn\u0027t part of the default deployment. So what I need is someway to enable it only for octavia deployments but disable the tests for everything else.","accounts_in_message":[],"_revision_number":9},{"id":"edc957fc79fe42cd961bdeb8d609c964f32497fe","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2019-09-09 08:56:01.000000000","message":"Patch Set 9: Code-Review+1\n\n\u003e \u003e As it seems in job tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020\n \u003e \u003e octavia tests still don\u0027t run: http://logs.rdoproject.org/35/670135/9/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020/521c7c9/\n \u003e \n \u003e Actually, octavia isn\u0027t part of the default deployment. So what I\n \u003e need is someway to enable it only for octavia deployments but\n \u003e disable the tests for everything else.\n\npretty sure you can do that with one of the tempest variable overrides in the job definition, i\u0027ll go ping someone from tempest team to check here too in a sec - I think it is tempest_test_blacklist ?\n (scenario10 looks like looking at the topic branch)\n\nbut then am not sure if the job definition var will trump the skiplist, surely it must","accounts_in_message":[],"_revision_number":9},{"id":"6fb2a9c8e8186c7c06760ab2bef4ee02908ba06e","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-09-09 08:59:41.000000000","message":"Patch Set 9:\n\nplease check the comments on ps#9","accounts_in_message":[],"_revision_number":9},{"id":"e388b9cc5f75d4c5199f1be862c24624437eef02","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-09-09 09:20:59.000000000","message":"Patch Set 9:\n\n\u003e \u003e \u003e As it seems in job tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020\n \u003e \u003e \u003e octavia tests still don\u0027t run: http://logs.rdoproject.org/35/670135/9/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020/521c7c9/\n \u003e \u003e\n \u003e \u003e Actually, octavia isn\u0027t part of the default deployment. So what I\n \u003e \u003e need is someway to enable it only for octavia deployments but\n \u003e \u003e disable the tests for everything else.\n \u003e \n \u003e pretty sure you can do that with one of the tempest variable\n \u003e overrides in the job definition, i\u0027ll go ping someone from tempest\n \u003e team to check here too in a sec - I think it is tempest_test_blacklist\n \u003e ?\n \u003e (scenario10 looks like looking at the topic branch)\n \u003e \n\nIf we donot pass the test_regex explicitly the test will not get triggered.\nIn case of scenrario 10:\nhttps://github.com/openstack/tripleo-ci/blob/master/zuul.d/standalone-jobs.yaml#L552 which is based on validate_tempest, we can just modify the scenario 10 job based on scenario 7https://github.com/openstack/tripleo-ci/blob/master/zuul.d/standalone-jobs.yaml#L532 and pass the appropriate test name there then it will get triggered for scenario 10 only.\n\n \u003e but then am not sure if the job definition var will trump the\n \u003e skiplist, surely it must","accounts_in_message":[],"_revision_number":9},{"id":"3c82d00a6cd32a6cf128067c2cd8226f9b920f4f","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2019-09-09 12:40:58.000000000","message":"Patch Set 9:\n\n\u003e \u003e \u003e As it seems in job tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020\n \u003e \u003e \u003e octavia tests still don\u0027t run: http://logs.rdoproject.org/35/670135/9/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020/521c7c9/\n \u003e \u003e\n \u003e \u003e Actually, octavia isn\u0027t part of the default deployment. So what I\n \u003e \u003e need is someway to enable it only for octavia deployments but\n \u003e \u003e disable the tests for everything else.\n \u003e \n \u003e pretty sure you can do that with one of the tempest variable\n \u003e overrides in the job definition, i\u0027ll go ping someone from tempest\n \u003e team to check here too in a sec - I think it is tempest_test_blacklist\n \u003e ?\n \u003e (scenario10 looks like looking at the topic branch)\n \u003e \n \u003e but then am not sure if the job definition var will trump the\n \u003e skiplist, surely it must\n\nTried and it seems that the job definition does not trump the skiplist (maybe this is a bug? - it\u0027d be nice if it were). With respect to comments about simply using the 52/standalone (?) featureset - there are filters on featureset overrides. I could extends the list of allowable featureset overrides, but IIRC the pattern around the allowable overrides seem to indicate an \"intent\" to allow control over some generic items - not expanding for specific setups. In Octavia\u0027s case, I need to enable some ansible in the quickstart extras for downloading the amphora and that seems inconsistent with the implied scope of the allowed overrides. Perhaps people who\u0027ve modified of the list of allowed_overrides in tripleo-ci/roles/run-test/tasks/main.yaml could comment?","accounts_in_message":[],"_revision_number":9},{"id":"9767d1336aec376b0e0f93d9bb2d69e011ea78bb","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2019-09-10 08:06:20.000000000","message":"Patch Set 9:\n\n\u003e \u003e \u003e \u003e As it seems in job tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020\n \u003e \u003e \u003e \u003e octavia tests still don\u0027t run: http://logs.rdoproject.org/35/670135/9/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020/521c7c9/\n \u003e \u003e \u003e\n \u003e \u003e \u003e Actually, octavia isn\u0027t part of the default deployment. So what\n \u003e I\n \u003e \u003e \u003e need is someway to enable it only for octavia deployments but\n \u003e \u003e \u003e disable the tests for everything else.\n \u003e \u003e\n \u003e \u003e pretty sure you can do that with one of the tempest variable\n \u003e \u003e overrides in the job definition, i\u0027ll go ping someone from\n \u003e tempest\n \u003e \u003e team to check here too in a sec - I think it is tempest_test_blacklist\n \u003e \u003e ?\n \u003e \u003e (scenario10 looks like looking at the topic branch)\n \u003e \u003e\n \u003e \u003e but then am not sure if the job definition var will trump the\n \u003e \u003e skiplist, surely it must\n \u003e \n \u003e Tried and it seems that the job definition does not trump the\n \u003e skiplist (maybe this is a bug? - it\u0027d be nice if it were). With\n \u003e respect to comments about simply using the 52/standalone (?)\n \u003e featureset - there are filters on featureset overrides. I could\n \u003e extends the list of allowable featureset overrides, but IIRC the\n \u003e pattern around the allowable overrides seem to indicate an \"intent\"\n \u003e to allow control over some generic items - not expanding for\n \u003e specific setups. In Octavia\u0027s case, I need to enable some ansible\n \u003e in the quickstart extras for downloading the amphora and that seems\n \u003e inconsistent with the implied scope of the allowed overrides.\n \u003e Perhaps people who\u0027ve modified of the list of allowed_overrides in\n \u003e tripleo-ci/roles/run-test/tasks/main.yaml could comment?\n\n\nLooking at scenario10 i see you have run_tempest false (so won\u0027t run the \u0027old\u0027 validate_tempest) but DO NOT HAVE use_os_tempest true. Also the tempest vars you are passing are for validate tempest (compare to scneario 7 - e.g. tempest_test_whitelist vs just tempest_whitelist). https://github.com/openstack/tripleo-ci/blob/a7043592537ec4227a23cd6478a15b5c3fccd5b3/zuul.d/standalone-jobs.yaml#L552\n\nJust posted that now https://review.opendev.org/681179 to see what happens with os-tempest enabled - wasn\u0027t sure what tests to enable for octavia went looking but couldn\u0027t quickly find a octavia specific scenario. i added the network basic ops scenario for now.\n\n\nWRT the featureset_override -  you\u0027re right it used to be much more restricted. It still is, as we have to add into the allowed_overrides, but we have added more recently as was necessary (e.g. for the standalones to have a shared featureset, we use the override instead). So it *is* possible to add something there ;) i think if you have a use-case just post the review and we can have the discussion there.\n\nBut it might not be necessary lets see if we can get the os-tempest working","accounts_in_message":[],"_revision_number":9},{"id":"f3443095439c7f669561ae10a8aae1421a4f6ea9","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2019-09-11 15:30:54.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"dd8cb2eb4477511f3bf6ea3d43ae99232bba7a10","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2019-09-11 15:31:05.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"cf950c95ddd2e6ef9c7b29e9132380de4ae1a585","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2019-09-11 15:46:23.000000000","message":"Patch Set 11:\n\nremoved modification of skip list here as it affects everything and we only want to enable it in tripleo and only for Octavia CI jobs when we known octavia should be deployed.","accounts_in_message":[],"_revision_number":11},{"id":"b9749c8b26207e4f06fcef2eb99957b640df071b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-11 18:35:01.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/35/670135/11/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/711aa9e/ : FAILURE in 3h 02m 18s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/35/670135/11/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/4126f3e/ : SUCCESS in 2h 10m 21s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/35/670135/11/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/19b1b37/ : SUCCESS in 2h 14m 01s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/35/670135/11/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/0c6d577/ : FAILURE in 2h 59m 16s","accounts_in_message":[],"_revision_number":11},{"id":"389a868e8987208d12f54a2c7810a11b247b5cde","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 20:56:23.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/87d5b271daa4442482c5a69b1ea5e545 : SUCCESS in 1h 53m 07s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/9dc348c93e1540129ffadf95d0acdda7 : FAILURE in 1h 56m 08s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/78d4ef86755e4166abfb3dd15fc068c0 : SUCCESS in 2h 41m 12s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/6bdef59a21aa4ca084997c1708adaf95 : SUCCESS in 3h 04m 45s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/4e8c216966324a0e9ac9ac159aa0fbd0 : FAILURE in 1h 31m 05s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/b8c846536f094357a4a46dfcd818f767 : SUCCESS in 2h 25m 22s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/baff24bb151d4f5196e59566a44c6b77 : FAILURE in 1h 33m 14s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/af00362fae7641cbbf47ecc416b8e143 : SUCCESS in 2h 02m 37s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/aa3a6a4137db412f888bd5e929a25523 : FAILURE in 1h 53m 22s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/b084e6bf1505405e9e699bd40b8266f6 : SUCCESS in 1h 59m 11s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/b4e9d3c2ed0245ef88a424bf88a3f514 : FAILURE in 1h 51m 42s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/09e32dbd31be4444841f3fdb83291dd5 : SUCCESS in 1h 45m 02s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/41ca9dbbf86746c7a7bb06f6cf542548 : SUCCESS in 2h 22m 57s\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/0615de7da40e471ca2b4477df31f9b7b : SUCCESS in 1h 21m 44s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/61ae93da48a047bd8cbd0b9e5100d5b7 : SUCCESS in 2h 29m 50s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/d5b8df9ae5c548b9b1f630c66f6bf033 : SUCCESS in 2h 20m 22s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/0a1bba03ac174cb387cbda4f0b815704 : SUCCESS in 2h 44m 19s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/45d6b008e5d24dd48d9aec4252e4c3f3 : FAILURE in 1h 44m 39s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/593813dfe12f4154af777b762d1c8ca6 : SUCCESS in 1h 58m 43s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ff99599af39641f89d9b340dc1140a78 : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":11},{"id":"ca32731d81f3a6f20fb78ec0ae3a1d0d7a200f21","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2019-09-12 07:17:48.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"89a9c93fe9d099e352c89654582a1baa2e6a363b","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2019-09-17 22:48:43.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"6099c187f8d5598e2b8a11f06f266cb4cdc60ac9","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2019-09-17 22:51:29.000000000","message":"Abandoned\n\nwhoops noticed we are clearing the gate. will rekick post crisis","accounts_in_message":[],"_revision_number":11},{"id":"34034b416e1613facae32fe0cda0013676208eae","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2019-09-19 15:08:56.000000000","message":"Restored","accounts_in_message":[],"_revision_number":11},{"id":"a44e6f0d00cb96bc04563863017c1513d647f9c2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-19 19:49:57.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/35/670135/11/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/3662964/ : FAILURE in 2h 57m 27s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/35/670135/11/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/45f44cb/ : FAILURE in 1h 54m 52s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/35/670135/11/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/f74fac3/ : SUCCESS in 2h 23m 49s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/35/670135/11/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/4cd1a2d/ : SUCCESS in 3h 20m 37s","accounts_in_message":[],"_revision_number":11},{"id":"e5a67302e8605e5b12361c2aa86ef27c028db842","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 05:41:20.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/6dd53635954c4423bb2da69214cef2f9 : SUCCESS in 2h 27m 11s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/c77cfd5311d8477497d47f5fec100162 : FAILURE in 2h 09m 34s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/1ed51ac8f9954ced8898a91ff0e06b16 : SUCCESS in 3h 54m 05s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/12355eb31ae446f6bc892bb10829b825 : SUCCESS in 3h 35m 19s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/877d1ef4dd3343d58d5d29b8cb14c554 : FAILURE in 2h 36m 13s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/2997cc6e2f0d42338420f44205157850 : SUCCESS in 3h 25m 33s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/0c52f462da9347b896bc57a6f5e63cec : SUCCESS in 3h 30m 49s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/e732937a4af94945bf6e610358eed4d9 : SUCCESS in 2h 22m 46s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/0c7beacb1e364db7b85091e3c9388522 : SUCCESS in 2h 38m 51s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/3d9d318cb4fa43ed89fca1ae17a818e1 : SUCCESS in 2h 44m 32s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/44544677d99546cdb380b164516ff1f6 : SUCCESS in 2h 33m 45s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/25f16252e7c34c188d1ac0ef9bdaa126 : FAILURE in 1h 33m 16s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/3f70c7df7a27415191eb3e4b3d43857e : SUCCESS in 2h 26m 37s\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/e6934d726b474d7da71e1da4c57d46a1 : SUCCESS in 1h 58m 54s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/8eeafa56bda54291a0cd2cb851389b16 : SUCCESS in 2h 58m 47s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/b550ae8f01254df89b1fe95a66782658 : SUCCESS in 2h 54m 37s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/992d0043b21f4210aff5d61aa02355ab : SUCCESS in 2h 27m 58s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/9f2a731b79ef4cd78be6e8a1b0896630 : SUCCESS in 1h 46m 25s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/b1e5c1d98d2845419fa36f59d22b95c0 : FAILURE in 2h 03m 22s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/162f2f54be9b41d1aaa794d23757c073 : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":11},{"id":"1440ae642abb2b3fa18ccbeb429f0bfacf908208","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-10-16 11:49:15.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"153d44cef307d7817fd9ebda4e8b17dcf97acc92","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2019-10-25 17:32:33.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"d169b1a832edfc676d53c044e2a638f2721c9e5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 20:29:05.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/e81721a31d1d48fd9c7eecc2d1126c20 : SUCCESS in 1h 47m 44s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/04f64edf00cc498d95f18f6be377b752 : FAILURE in 1h 51m 14s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/24d492c2ff304539a3c42f53e8f106f7 : SUCCESS in 2h 54m 56s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/ca7eaaa49f854643aaa1ff5d72b17aaf : SUCCESS in 2h 24m 31s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/36e9d93f1b7d4bc889238ee6282540c3 : FAILURE in 1h 58m 17s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/84a45bb4dd0345149b63b901146be270 : SUCCESS in 2h 19m 03s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/bae8e603f4754a53857dad60fbd7b55e : SUCCESS in 2h 00m 56s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/6deccc16195140308fdca5a3f92db462 : SUCCESS in 1h 47m 32s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/922a30c65c7348cb99a74f769ccd1b4f : FAILURE in 1h 20m 43s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/ba3dd9f1e4f34ae6b96cd2535ee93bd9 : SUCCESS in 1h 35m 27s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/d270e3b4832e4a268654e19830b1e654 : SUCCESS in 1h 54m 56s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/c408f2713b974d0595ac10a2fa6c8bb2 : SUCCESS in 1h 39m 44s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/50ea32be5786429fa317cc1898f0d2d6 : SUCCESS in 1h 58m 50s\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/a5a491c7a7f64ebea61de95fba8edf6c : SUCCESS in 1h 37m 26s\n- tripleo-ci-centos-7-containers-multinode-train https://zuul.opendev.org/t/openstack/build/fbcf7490e410426b926013d934e46251 : SUCCESS in 2h 39m 15s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/b83118ecf7a848ae8e99e5f26935d8e6 : SUCCESS in 2h 32m 02s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/bd390f7841d84afa8fb1548f7256908a : SUCCESS in 2h 24m 10s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/79e875bc718f48ecbd17a931551a0945 : SUCCESS in 2h 08m 57s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/3c5d986825864b91a02a778678a888a6 : SUCCESS in 1h 32m 08s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/79db778ae04b419a8880ca7c830a5c50 : FAILURE in 1h 13m 50s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/218e146d10864fd580f4459ba2ea0a93 : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":11},{"id":"9a6abefcc2842573a98e749c6f13a69701fc1bb9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 20:29:22.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"03c25e48eaa0c38d4b0c732d9c1aa1c992ea6e99","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 23:16:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":11},{"id":"19b5a4ba19a8a320961aa1cdabc157444ab38fec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 23:16:15.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/10b11caf8afc43d680894e2a994c698a : SUCCESS in 1h 42m 52s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/1afcd296f488444695e30b1ba993cb66 : SUCCESS in 2h 34m 13s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/4673c7b568204c1e8c893d6602fcf5de : SUCCESS in 1h 57m 45s\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/36e032cd7a4e405fa8dfc932e606ce87 : SUCCESS in 1h 25m 56s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/824c05dd6a654909ba7414fc5435ce58 : SUCCESS in 2h 20m 11s\n- tripleo-ci-centos-7-containers-multinode-train https://zuul.opendev.org/t/openstack/build/103ccd534c3243019777501a2ecfe0c8 : SUCCESS in 2h 38m 12s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/15dd9ec09223408f876a36acfd8c95a0 : SUCCESS in 2h 41m 49s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/e4edc6ee22774ee4b861edae47c3d820 : SUCCESS in 2h 36m 34s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/9fa8365214144239a9c23d05586f4f14 : SUCCESS in 2h 28m 10s\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/85fde2fe77a043ff918d0c97ea984549 : SUCCESS in 1h 41m 12s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/4555486d1052440f9816b43a8e75faef : SUCCESS in 1h 46m 24s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/1ccfcfd52b4544fcbcba04790d1b1978 : SUCCESS in 1h 29m 30s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/5738ff8c4e944cc6b7eed5a68862e276 : SUCCESS in 2h 11m 29s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/e60e348c45744ca6be5ac82479703073 : SUCCESS in 1h 13m 45s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/cc5dd67454774fdab5e5133be6827281 : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"5b8665b18d056c7d15b2f5224d8d9cfddb560ec3","revisions":{"ceff7085150a168bf483fafacb6106488efb4ccd":{"kind":"REWORK","_number":1,"created":"2019-07-10 17:06:48.000000000","uploader":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"ref":"refs/changes/35/670135/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/670135/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/670135/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/670135/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/670135/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/670135/1"}}},"commit":{"parents":[{"commit":"1f9deae7d187ed678f84efb7a177bb0db8fb0c61","subject":"Add a tasks to download amphora image for octavia deployments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/1f9deae7d187ed678f84efb7a177bb0db8fb0c61"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-07-10 17:06:03.000000000","tz":-240},"committer":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-07-10 17:06:03.000000000","tz":-240},"subject":"Remove Octavia from tempest skip","message":"Remove Octavia from tempest skip\n\nWe should be able to run Octavia tempest tests with the latest changes.\n\nChange-Id: Iff54ff52c9db65e3675af69fc17dd9461eb596a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/ceff7085150a168bf483fafacb6106488efb4ccd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/ceff7085150a168bf483fafacb6106488efb4ccd"}]},"branch":"refs/heads/master"},"654d1bd5e0ce62995e9afb52c16f5274c9b57499":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-07-15 11:11:45.000000000","uploader":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"ref":"refs/changes/35/670135/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/670135/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/670135/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/670135/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/670135/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/670135/2"}}},"commit":{"parents":[{"commit":"c9105fb2ca035ceddae0389d64d1c1385f07d5e2","subject":"Add a tasks to download amphora image for octavia deployments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/c9105fb2ca035ceddae0389d64d1c1385f07d5e2"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-07-10 17:06:03.000000000","tz":-240},"committer":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-07-15 11:11:33.000000000","tz":-240},"subject":"Remove Octavia from tempest skip","message":"Remove Octavia from tempest skip\n\nWe should be able to run Octavia tempest tests with the latest changes.\n\nChange-Id: Iff54ff52c9db65e3675af69fc17dd9461eb596a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/654d1bd5e0ce62995e9afb52c16f5274c9b57499"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/654d1bd5e0ce62995e9afb52c16f5274c9b57499"}]},"branch":"refs/heads/master"},"f1ecda7fa98989d132662829e7e3717031e3f8ca":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-07-15 14:04:11.000000000","uploader":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"ref":"refs/changes/35/670135/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/670135/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/670135/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/670135/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/670135/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/670135/3"}}},"commit":{"parents":[{"commit":"5f2c643237930b56744c75bcefe016ebf79ecb4c","subject":"Add a tasks to download amphora image for octavia deployments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/5f2c643237930b56744c75bcefe016ebf79ecb4c"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-07-10 17:06:03.000000000","tz":-240},"committer":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-07-15 14:04:02.000000000","tz":-240},"subject":"Remove Octavia from tempest skip","message":"Remove Octavia from tempest skip\n\nWe should be able to run Octavia tempest tests with the latest changes.\n\nChange-Id: Iff54ff52c9db65e3675af69fc17dd9461eb596a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/f1ecda7fa98989d132662829e7e3717031e3f8ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/f1ecda7fa98989d132662829e7e3717031e3f8ca"}]},"branch":"refs/heads/master"},"70c2ab22eac5f881a20c356eb50235bc74cab48f":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-07-15 18:14:46.000000000","uploader":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"ref":"refs/changes/35/670135/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/670135/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/670135/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/670135/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/670135/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/670135/4"}}},"commit":{"parents":[{"commit":"85f9221141a7b072f9041cd84504fab4f8d32ff5","subject":"Add a tasks to download amphora image for octavia deployments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/85f9221141a7b072f9041cd84504fab4f8d32ff5"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-07-10 17:06:03.000000000","tz":-240},"committer":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-07-15 18:14:30.000000000","tz":-240},"subject":"Remove Octavia from tempest skip","message":"Remove Octavia from tempest skip\n\nWe should be able to run Octavia tempest tests with the latest changes.\n\nChange-Id: Iff54ff52c9db65e3675af69fc17dd9461eb596a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/70c2ab22eac5f881a20c356eb50235bc74cab48f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/70c2ab22eac5f881a20c356eb50235bc74cab48f"}]},"branch":"refs/heads/master"},"112679540e53922c01efc681fb3d2f42d8fc74ed":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-07-16 12:08:42.000000000","uploader":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"ref":"refs/changes/35/670135/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/670135/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/670135/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/670135/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/670135/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/670135/5"}}},"commit":{"parents":[{"commit":"a702d8c4cbf9accfef28c01989ac9626ed3b8212","subject":"Add a tasks to download amphora image for octavia deployments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/a702d8c4cbf9accfef28c01989ac9626ed3b8212"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-07-10 17:06:03.000000000","tz":-240},"committer":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-07-16 12:08:29.000000000","tz":-240},"subject":"Remove Octavia from tempest skip","message":"Remove Octavia from tempest skip\n\nWe should be able to run Octavia tempest tests with the latest changes.\n\nChange-Id: Iff54ff52c9db65e3675af69fc17dd9461eb596a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/112679540e53922c01efc681fb3d2f42d8fc74ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/112679540e53922c01efc681fb3d2f42d8fc74ed"}]},"branch":"refs/heads/master"},"4a7a8f70e46e7000959d8473ba923c5771e6905b":{"kind":"REWORK","_number":6,"created":"2019-07-18 15:07:07.000000000","uploader":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"ref":"refs/changes/35/670135/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/670135/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/670135/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/670135/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/670135/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/670135/6"}}},"commit":{"parents":[{"commit":"a702d8c4cbf9accfef28c01989ac9626ed3b8212","subject":"Add a tasks to download amphora image for octavia deployments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/a702d8c4cbf9accfef28c01989ac9626ed3b8212"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-07-10 17:06:03.000000000","tz":-240},"committer":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-07-18 15:06:05.000000000","tz":-240},"subject":"Add support for running Octavia tempest tasks","message":"Add support for running Octavia tempest tasks\n\nTo enable we need to remove from skip list and prevent cleanups of the\noctavia network resources that are created in the overcloud during\ndeployment.\n\nChange-Id: Iff54ff52c9db65e3675af69fc17dd9461eb596a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/4a7a8f70e46e7000959d8473ba923c5771e6905b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/4a7a8f70e46e7000959d8473ba923c5771e6905b"}]},"branch":"refs/heads/master"},"03e4a46d1074e4f3c23d2a8f40d9a062f425d74a":{"kind":"REWORK","_number":7,"created":"2019-07-24 15:05:40.000000000","uploader":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"ref":"refs/changes/35/670135/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/670135/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/670135/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/670135/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/670135/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/670135/7"}}},"commit":{"parents":[{"commit":"758dad5677f7da8ef2ca4e49bf3bb512dd90c893","subject":"Add a tasks to download amphora image for octavia deployments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/758dad5677f7da8ef2ca4e49bf3bb512dd90c893"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-07-10 17:06:03.000000000","tz":-240},"committer":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-07-24 15:05:35.000000000","tz":-240},"subject":"Add support for running Octavia tempest tasks","message":"Add support for running Octavia tempest tasks\n\nTo enable we need to remove from skip list and prevent cleanups of the\noctavia network resources that are created in the overcloud during\ndeployment.\n\nChange-Id: Iff54ff52c9db65e3675af69fc17dd9461eb596a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/03e4a46d1074e4f3c23d2a8f40d9a062f425d74a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/03e4a46d1074e4f3c23d2a8f40d9a062f425d74a"}]},"branch":"refs/heads/master"},"6cdedefefb5f9a60d4408307400bde5ebc4cf919":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2019-08-12 12:54:33.000000000","uploader":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"ref":"refs/changes/35/670135/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/670135/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/670135/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/670135/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/670135/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/670135/8"}}},"commit":{"parents":[{"commit":"01820bfc278cc1916fb94002c45aed3924926192","subject":"Add a tasks to download amphora image for octavia deployments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/01820bfc278cc1916fb94002c45aed3924926192"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-07-10 17:06:03.000000000","tz":-240},"committer":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-08-12 12:54:26.000000000","tz":-240},"subject":"Add support for running Octavia tempest tasks","message":"Add support for running Octavia tempest tasks\n\nTo enable we need to remove from skip list and prevent cleanups of the\noctavia network resources that are created in the overcloud during\ndeployment.\n\nChange-Id: Iff54ff52c9db65e3675af69fc17dd9461eb596a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/6cdedefefb5f9a60d4408307400bde5ebc4cf919"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/6cdedefefb5f9a60d4408307400bde5ebc4cf919"}]},"branch":"refs/heads/master"},"568700fe0ba7219da7ea59bdc1cc80cb41d80ab4":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2019-08-23 18:54:36.000000000","uploader":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"ref":"refs/changes/35/670135/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/670135/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/670135/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/670135/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/670135/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/670135/9"}}},"commit":{"parents":[{"commit":"c13c33f04b166be665582b5011cca0351c8af5a0","subject":"Add a tasks to download amphora image for octavia deployments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/c13c33f04b166be665582b5011cca0351c8af5a0"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-07-10 17:06:03.000000000","tz":-240},"committer":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-08-23 18:54:26.000000000","tz":-240},"subject":"Add support for running Octavia tempest tasks","message":"Add support for running Octavia tempest tasks\n\nTo enable we need to remove from skip list and prevent cleanups of the\noctavia network resources that are created in the overcloud during\ndeployment.\n\nChange-Id: Iff54ff52c9db65e3675af69fc17dd9461eb596a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/568700fe0ba7219da7ea59bdc1cc80cb41d80ab4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/568700fe0ba7219da7ea59bdc1cc80cb41d80ab4"}]},"branch":"refs/heads/master"},"6ea152c344b9f3a9ddb6dadf4de185961bc761c7":{"kind":"REWORK","_number":10,"created":"2019-09-11 15:30:54.000000000","uploader":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"ref":"refs/changes/35/670135/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/670135/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/670135/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/670135/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/670135/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/670135/10"}}},"commit":{"parents":[{"commit":"ab4c608cedb28a2f56d8137a78c42a0b820c99d5","subject":"Add a tasks to download amphora image for octavia deployments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/ab4c608cedb28a2f56d8137a78c42a0b820c99d5"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-07-10 17:06:03.000000000","tz":-240},"committer":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-09-11 15:30:13.000000000","tz":-240},"subject":"Preserve Octavia networks when running tempest","message":"Preserve Octavia networks when running tempest\n\nTo prevent cleanups of the octavia network resources that are created in\nthe overcloud during deployment.\n\nChange-Id: Iff54ff52c9db65e3675af69fc17dd9461eb596a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/6ea152c344b9f3a9ddb6dadf4de185961bc761c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/6ea152c344b9f3a9ddb6dadf4de185961bc761c7"}]},"branch":"refs/heads/master"},"5b8665b18d056c7d15b2f5224d8d9cfddb560ec3":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2019-09-11 15:31:05.000000000","uploader":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"ref":"refs/changes/35/670135/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/670135/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/670135/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/670135/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/670135/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/670135/11"}}},"commit":{"parents":[{"commit":"4612cf32f114801780bd1078e358bfe51179fc86","subject":"Add a tasks to download amphora image for octavia deployments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/4612cf32f114801780bd1078e358bfe51179fc86"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-07-10 17:06:03.000000000","tz":-240},"committer":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-09-11 15:30:58.000000000","tz":-240},"subject":"Preserve Octavia networks when running tempest","message":"Preserve Octavia networks when running tempest\n\nTo prevent cleanups of the octavia network resources that are created in\nthe overcloud during deployment.\n\nChange-Id: Iff54ff52c9db65e3675af69fc17dd9461eb596a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/5b8665b18d056c7d15b2f5224d8d9cfddb560ec3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/5b8665b18d056c7d15b2f5224d8d9cfddb560ec3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
