)]}'
{"id":"openstack%2Ftripleo-quickstart-extras~671262","triplet_id":"openstack%2Ftripleo-quickstart-extras~master~I4ac636572f537b98213745dd0a68ba89b4a1d74b","project":"openstack/tripleo-quickstart-extras","branch":"master","topic":"bug/1836415","hashtags":[],"change_id":"I4ac636572f537b98213745dd0a68ba89b4a1d74b","subject":"Add keycloak-setup role","status":"ABANDONED","created":"2019-07-17 13:15:42.000000000","updated":"2020-05-22 19:56:32.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"55076dcf477b29fa8c066ae6f2bbfdfa01b619e2","_number":671262,"virtual_id_number":671262,"owner":{"_account_id":8116,"name":"John Dennis","email":"jdennis@redhat.com","username":"jdennis"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"date":"2020-05-22 19:54:46.000000000","_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},{"_account_id":8116,"name":"John Dennis","email":"jdennis@redhat.com","username":"jdennis"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":22136,"name":"Harry Rybacki","email":"hrybacki@gmail.com"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":9098,"name":"Nathan Kinder","email":"nkinder@redhat.com","username":"nkinder"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-05-22 19:16:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":-1,"date":"2020-05-22 19:24:24.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8116,"name":"John Dennis","email":"jdennis@redhat.com","username":"jdennis"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22136,"name":"Harry Rybacki","email":"hrybacki@gmail.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9098,"name":"Nathan Kinder","email":"nkinder@redhat.com","username":"nkinder"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":8116,"name":"John Dennis","email":"jdennis@redhat.com","username":"jdennis"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":22136,"name":"Harry Rybacki","email":"hrybacki@gmail.com"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":9098,"name":"Nathan Kinder","email":"nkinder@redhat.com","username":"nkinder"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":8116,"name":"John Dennis","email":"jdennis@redhat.com","username":"jdennis"},{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},{"_account_id":9098,"name":"Nathan Kinder","email":"nkinder@redhat.com","username":"nkinder"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},{"_account_id":22136,"name":"Harry Rybacki","email":"hrybacki@gmail.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-18 01:58:21.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2019-07-18 20:59:39.000000000","updated_by":{"_account_id":22136,"name":"Harry Rybacki","email":"hrybacki@gmail.com"},"reviewer":{"_account_id":22136,"name":"Harry Rybacki","email":"hrybacki@gmail.com"},"state":"REVIEWER"},{"updated":"2019-07-31 16:07:31.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-01 05:00:09.000000000","updated_by":{"_account_id":9098,"name":"Nathan Kinder","email":"nkinder@redhat.com","username":"nkinder"},"reviewer":{"_account_id":9098,"name":"Nathan Kinder","email":"nkinder@redhat.com","username":"nkinder"},"state":"REVIEWER"},{"updated":"2019-08-01 12:34:22.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2019-08-05 18:57:17.000000000","updated_by":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-01-21 12:15:52.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-22 19:16:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-22 19:24:24.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-05-22 19:54:46.000000000","updated_by":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"reviewer":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"state":"REVIEWER"}],"messages":[{"id":"8571323cecdbd8592864e22196493fe5a259d046","author":{"_account_id":8116,"name":"John Dennis","email":"jdennis@redhat.com","username":"jdennis"},"date":"2019-07-17 13:15:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0d9dba2ac96a3ef27457c2fc99b74ceb934a20d1","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-07-17 13:16:00.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/3645/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"69f222c5a4003e10b69afb03d2da29f6c9d29c72","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-17 14:34:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 finger://ze07.softwarefactory-project.io:7979/d7ea4d4a1cc34d6c812c27314855a847 : RETRY_LIMIT in 6s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 finger://ze05.softwarefactory-project.io:7979/ebee4c1dafc947c0a5fc55d1db9588d9 : RETRY_LIMIT in 6s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/62/671262/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/053e3e3/ : FAILURE in 1h 00m 29s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/62/671262/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/41de4c1/ : FAILURE in 1h 14m 56s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/62/671262/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/a5b02ae/ : FAILURE in 1h 05m 28s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/62/671262/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/c5e1457/ : FAILURE in 1h 03m 58s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/62/671262/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/6f5ad4b/ : RETRY_LIMIT in 20m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4dc138e867619d08dab3f5f710fe1d0df7eed7d7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 18:01:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/62/671262/1/check/tripleo-ci-centos-7-undercloud-containers/3bd4e5d/ : SUCCESS in 1h 30m 55s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/62/671262/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/12f680a/ : FAILURE in 2h 12m 47s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/62/671262/1/check/tripleo-ci-centos-7-containers-multinode/0cb6327/ : FAILURE in 2h 14m 18s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/62/671262/1/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/2c71bfb/ : SUCCESS in 2h 33m 21s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container http://logs.openstack.org/62/671262/1/check/tripleo-ci-centos-7-scenario012-multinode-oooq-container/59fd177/ : FAILURE in 2h 45m 51s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/62/671262/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/e403ae7/ : SUCCESS in 2h 17m 38s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/62/671262/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/44041c9/ : FAILURE in 2h 16m 25s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/62/671262/1/check/tripleo-ci-centos-7-standalone/3e1c163/ : FAILURE in 1h 06m 45s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/62/671262/1/check/tripleo-ci-fedora-28-standalone/457621d/ : SUCCESS in 1h 40m 49s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/62/671262/1/check/tripleo-ci-centos-7-standalone-upgrade/fb41c79/ : FAILURE in 1h 38m 18s (non-voting)\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/62/671262/1/check/tripleo-ci-centos-7-standalone-os-tempest/575817f/ : FAILURE in 1h 08m 35s\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/62/671262/1/check/tripleo-ci-centos-7-containers-multinode-stein/3fbc189/ : SUCCESS in 2h 24m 28s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/62/671262/1/check/tripleo-ci-centos-7-containers-multinode-rocky/34b07b1/ : SUCCESS in 2h 27m 23s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/62/671262/1/check/tripleo-ci-centos-7-containers-multinode-queens/80ac7fc/ : SUCCESS in 2h 17m 12s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/62/671262/1/check/tripleo-ci-centos-7-containers-multinode-pike/8a6cc23/ : SUCCESS in 2h 03m 31s\n- openstack-tox-linters http://logs.openstack.org/62/671262/1/check/openstack-tox-linters/b302e37/ : FAILURE in 4m 12s","accounts_in_message":[],"_revision_number":1},{"id":"0996754d923d0512df47a64c3ff31be519929d79","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-07-18 01:58:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"545fa24b07768c70a02e4ed1cd620f8107484bec","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-07-18 01:58:36.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/3647/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"98e41707ab44a47c121b690b545a46f3a37f8200","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 05:46:22.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/62/671262/1/check/tripleo-ci-centos-7-undercloud-containers/3947686/ : SUCCESS in 1h 42m 51s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/62/671262/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/9ed3bb6/ : FAILURE in 2h 09m 49s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/62/671262/1/check/tripleo-ci-centos-7-containers-multinode/1a4c672/ : SUCCESS in 2h 24m 01s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/62/671262/1/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/9d02772/ : SUCCESS in 2h 20m 42s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container http://logs.openstack.org/62/671262/1/check/tripleo-ci-centos-7-scenario012-multinode-oooq-container/68759fa/ : FAILURE in 2h 43m 24s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/62/671262/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/6caf630/ : SUCCESS in 2h 14m 15s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/62/671262/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/34f4a56/ : FAILURE in 2h 37m 52s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/62/671262/1/check/tripleo-ci-centos-7-standalone/fd37698/ : FAILURE in 57m 30s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/62/671262/1/check/tripleo-ci-fedora-28-standalone/99112f9/ : SUCCESS in 1h 18m 30s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/62/671262/1/check/tripleo-ci-centos-7-standalone-upgrade/b86d298/ : FAILURE in 2h 08m 24s (non-voting)\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/62/671262/1/check/tripleo-ci-centos-7-standalone-os-tempest/ca0ea2f/ : FAILURE in 56m 55s\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/62/671262/1/check/tripleo-ci-centos-7-containers-multinode-stein/f300fc1/ : SUCCESS in 2h 34m 56s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/62/671262/1/check/tripleo-ci-centos-7-containers-multinode-rocky/de99722/ : SUCCESS in 2h 12m 24s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/62/671262/1/check/tripleo-ci-centos-7-containers-multinode-queens/7f33f78/ : SUCCESS in 2h 14m 58s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/62/671262/1/check/tripleo-ci-centos-7-containers-multinode-pike/08d7ade/ : SUCCESS in 2h 12m 12s\n- openstack-tox-linters http://logs.openstack.org/62/671262/1/check/openstack-tox-linters/296b658/ : FAILURE in 3m 57s","accounts_in_message":[],"_revision_number":1},{"id":"620059d69c59e7d1adc02f578339b92b50ce4921","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-18 07:25:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 finger://ze01.softwarefactory-project.io:7979/4ffba93449a84de2965109e7e25375d6 : POST_FAILURE in 3h 50m 37s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/62/671262/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/6bd772f/ : RETRY_LIMIT in 6m 13s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/62/671262/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/1a176c1/ : SUCCESS in 2h 16m 03s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch finger://ze07.softwarefactory-project.io:7979/acdf838c7c9e42e585189b517592a99d : POST_FAILURE in 3h 19m 09s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/62/671262/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/49ab3a3/ : RETRY_LIMIT in 4m 44s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/62/671262/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/ab3e2b8/ : RETRY_LIMIT in 4m 22s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/62/671262/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/3bf5b5a/ : RETRY_LIMIT in 4m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ed0a3cdddc85e83256f548c5a1f60c6d9f214bfe","author":{"_account_id":8116,"name":"John Dennis","email":"jdennis@redhat.com","username":"jdennis"},"date":"2019-07-18 21:12:04.000000000","message":"Patch Set 1:\n\nNote: the review for invoking the keycloak-setup role in featureset039 is https://review.opendev.org/#/c/671567","accounts_in_message":[],"_revision_number":1},{"id":"e8291a20ac8c0c3485cf06dbfd2630013975008c","author":{"_account_id":8116,"name":"John Dennis","email":"jdennis@redhat.com","username":"jdennis"},"date":"2019-07-30 12:03:03.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"56f9932802eafcb46795da98bede95cd90e8e079","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-07-30 12:03:22.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/3703/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"f3b1cc8d3380b08218179588cf1e62f562b548fb","author":{"_account_id":8116,"name":"John Dennis","email":"jdennis@redhat.com","username":"jdennis"},"date":"2019-07-30 15:53:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0839fa4bcefaa508eef3ba23916229a385977ece","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-07-30 15:53:19.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/3707/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"470ba5dc6a3b49eee7fadb744a45ddf1eb74091b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-30 16:20:27.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/62/671262/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/60aff9e/ : SUCCESS in 3h 35m 47s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/62/671262/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/dc46ac6/ : SUCCESS in 3h 38m 46s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/62/671262/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/625a8af/ : FAILURE in 2h 34m 00s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/62/671262/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/8b47a15/ : SUCCESS in 3h 46m 20s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch finger://ze02.softwarefactory-project.io:7979/b8dc37a1886c40a798b3658b576f6f23 : POST_FAILURE in 3h 33m 38s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/62/671262/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/164c797/ : TIMED_OUT in 4h 15m 32s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/62/671262/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/b55bbb2/ : SUCCESS in 3h 46m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a4da9d19a7a2a8191a875749e129de420ab4f2e5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 18:37:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/62/671262/1/check/tripleo-ci-centos-7-undercloud-containers/74c9715/ : SUCCESS in 1h 21m 04s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/62/671262/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/b4e48e7/ : FAILURE in 2h 04m 46s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/62/671262/1/check/tripleo-ci-centos-7-containers-multinode/5c58281/ : SUCCESS in 2h 22m 34s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/62/671262/1/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/a91fccb/ : SUCCESS in 2h 24m 19s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container http://logs.openstack.org/62/671262/1/check/tripleo-ci-centos-7-scenario012-multinode-oooq-container/717124f/ : FAILURE in 2h 46m 46s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/62/671262/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/15c4ca0/ : SUCCESS in 2h 54m 43s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/62/671262/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/41a10c7/ : SUCCESS in 2h 14m 46s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/62/671262/1/check/tripleo-ci-centos-7-standalone/c28b5cd/ : SUCCESS in 1h 38m 19s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/62/671262/1/check/tripleo-ci-fedora-28-standalone/b5545e3/ : SUCCESS in 1h 36m 42s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/62/671262/1/check/tripleo-ci-centos-7-standalone-upgrade/99699dd/ : SUCCESS in 2h 11m 17s (non-voting)\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/62/671262/1/check/tripleo-ci-centos-7-standalone-os-tempest/a0844fd/ : SUCCESS in 1h 12m 49s\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/62/671262/1/check/tripleo-ci-centos-7-containers-multinode-stein/7875f8f/ : SUCCESS in 2h 33m 01s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/62/671262/1/check/tripleo-ci-centos-7-containers-multinode-rocky/fb382f6/ : SUCCESS in 1h 58m 09s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/62/671262/1/check/tripleo-ci-centos-7-containers-multinode-queens/4ed3bec/ : SUCCESS in 2h 21m 46s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/62/671262/1/check/tripleo-ci-centos-7-containers-multinode-pike/9b5e958/ : SUCCESS in 2h 19m 36s\n- openstack-tox-linters http://logs.openstack.org/62/671262/1/check/openstack-tox-linters/edd9d67/ : FAILURE in 4m 53s","accounts_in_message":[],"_revision_number":1},{"id":"4618c7bc32fea71fa3f27aa7605937d6cde5f2ea","author":{"_account_id":9098,"name":"Nathan Kinder","email":"nkinder@redhat.com","username":"nkinder"},"date":"2019-07-31 16:08:47.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b6e80aba430f83f88ed4fb3894afbcf8fa0a8e77","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-07-31 16:09:04.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/3713/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"5a6bd25afe3b7125830ec70320f39a357b05570d","author":{"_account_id":8116,"name":"John Dennis","email":"jdennis@redhat.com","username":"jdennis"},"date":"2019-07-31 17:10:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fdfe11cf635cd6550543a6fc7307f828dd18ae39","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-07-31 17:10:33.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/3715/ : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"d930baa5a189d2a4b7827ef2e18461d02012e993","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-31 19:58:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/62/671262/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/91daab1/ : FAILURE in 1h 12m 24s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/62/671262/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/b9b8950/ : FAILURE in 1h 10m 18s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/62/671262/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/0626096/ : FAILURE in 2h 29m 50s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/62/671262/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/f8ee588/ : SUCCESS in 2h 15m 22s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/62/671262/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/00c0b89/ : FAILURE in 1h 05m 16s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/62/671262/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/38425e4/ : SUCCESS in 2h 46m 12s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/62/671262/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/0427af9/ : FAILURE in 24m 34s (non-voting)\n- tripleo-ci-rhel-8-standalone-rdo http://logs.rdoproject.org/62/671262/2/openstack-check/tripleo-ci-rhel-8-standalone-rdo/8391df2/ : FAILURE in 14m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a325416808e65c47bc36452e25db12fc24e3ac5a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 00:06:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/62/671262/2/check/tripleo-ci-centos-7-undercloud-containers/2494ec5/ : SUCCESS in 1h 30m 01s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/62/671262/2/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/1396668/ : FAILURE in 1h 44m 23s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/62/671262/2/check/tripleo-ci-centos-7-containers-multinode/aecd688/ : SUCCESS in 2h 34m 40s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/62/671262/2/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/3db4ccd/ : SUCCESS in 2h 23m 27s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container http://logs.openstack.org/62/671262/2/check/tripleo-ci-centos-7-scenario012-multinode-oooq-container/0df12b1/ : FAILURE in 2h 13m 54s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/62/671262/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/834057f/ : SUCCESS in 2h 01m 02s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/62/671262/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/510c3dc/ : TIMED_OUT in 3h 26m 41s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/62/671262/2/check/tripleo-ci-centos-7-standalone/2ffe2da/ : SUCCESS in 1h 53m 46s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/62/671262/2/check/tripleo-ci-fedora-28-standalone/29adece/ : FAILURE in 1h 07m 41s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/62/671262/2/check/tripleo-ci-centos-7-standalone-upgrade/eff7e8b/ : FAILURE in 1h 35m 53s (non-voting)\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/62/671262/2/check/tripleo-ci-centos-7-standalone-os-tempest/1dbcd15/ : SUCCESS in 1h 20m 00s\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/62/671262/2/check/tripleo-ci-centos-7-containers-multinode-stein/346b293/ : SUCCESS in 2h 12m 33s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/62/671262/2/check/tripleo-ci-centos-7-containers-multinode-rocky/6870bb0/ : SUCCESS in 2h 30m 52s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/62/671262/2/check/tripleo-ci-centos-7-containers-multinode-queens/761924d/ : SUCCESS in 2h 29m 38s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/62/671262/2/check/tripleo-ci-centos-7-containers-multinode-pike/573a978/ : SUCCESS in 2h 25m 23s\n- openstack-tox-linters http://logs.openstack.org/62/671262/2/check/openstack-tox-linters/b2fc27d/ : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":2},{"id":"a78baa30e281a6076b1d51bf6bff63f27ae1b8ec","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2019-08-01 00:50:37.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"5881cbc1295df0199f367fbedafd83e345c98f5c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-01 03:40:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/62/671262/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/b10ac3a/ : SUCCESS in 2h 23m 44s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/62/671262/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/336f42a/ : SUCCESS in 2h 44m 12s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/62/671262/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/28a2eef/ : SUCCESS in 2h 18m 54s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/62/671262/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/3f05121/ : SUCCESS in 2h 32m 10s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/62/671262/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/686b1e6/ : SUCCESS in 2h 47m 53s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/62/671262/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/ccecf66/ : SUCCESS in 2h 31m 06s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/62/671262/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/6fcab73/ : SUCCESS in 2h 38m 27s (non-voting)\n- tripleo-ci-rhel-8-standalone-rdo http://logs.rdoproject.org/62/671262/3/openstack-check/tripleo-ci-rhel-8-standalone-rdo/ed66e21/ : FAILURE in 14m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e93abc506c2eea4ded17bf4eaacee34989032250","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 04:31:19.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/62/671262/3/check/tripleo-ci-centos-7-undercloud-containers/f289ba9/ : SUCCESS in 1h 26m 05s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/62/671262/3/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/ecc86c3/ : FAILURE in 1h 36m 02s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/62/671262/3/check/tripleo-ci-centos-7-containers-multinode/eb6c262/ : SUCCESS in 2h 26m 49s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://logs.opendev.org/62/671262/3/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/54278ad/ : SUCCESS in 2h 17m 31s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://logs.opendev.org/62/671262/3/check/tripleo-ci-centos-7-scenario012-multinode-oooq-container/a8b51d1/ : FAILURE in 2h 19m 51s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/62/671262/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/811363e/ : SUCCESS in 2h 06m 47s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/62/671262/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/36529c2/ : FAILURE in 1h 42m 30s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/62/671262/3/check/tripleo-ci-centos-7-standalone/149c8d7/ : SUCCESS in 1h 40m 57s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/62/671262/3/check/tripleo-ci-fedora-28-standalone/d53292d/ : FAILURE in 45m 30s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/62/671262/3/check/tripleo-ci-centos-7-standalone-upgrade/eaccda9/ : FAILURE in 1h 47m 01s (non-voting)\n- tripleo-ci-centos-7-standalone-os-tempest https://logs.opendev.org/62/671262/3/check/tripleo-ci-centos-7-standalone-os-tempest/4f63695/ : SUCCESS in 1h 17m 30s\n- tripleo-ci-centos-7-containers-multinode-stein https://logs.opendev.org/62/671262/3/check/tripleo-ci-centos-7-containers-multinode-stein/d835dfd/ : SUCCESS in 2h 20m 20s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky https://logs.opendev.org/62/671262/3/check/tripleo-ci-centos-7-containers-multinode-rocky/34a12c7/ : SUCCESS in 3h 08m 49s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens https://logs.opendev.org/62/671262/3/check/tripleo-ci-centos-7-containers-multinode-queens/b70f393/ : SUCCESS in 2h 04m 28s\n- tripleo-ci-centos-7-containers-multinode-pike https://logs.opendev.org/62/671262/3/check/tripleo-ci-centos-7-containers-multinode-pike/2f43ac3/ : FAILURE in 1h 37m 50s\n- openstack-tox-linters https://logs.opendev.org/62/671262/3/check/openstack-tox-linters/21feded/ : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":3},{"id":"e04498413b1d2471fd2e07fc99a7488ae937a93e","author":{"_account_id":9098,"name":"Nathan Kinder","email":"nkinder@redhat.com","username":"nkinder"},"date":"2019-08-01 05:00:09.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"c51d9d2a50a746befc1354fe0bfa09084e00c2b6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-01 07:48:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/62/671262/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/ff3bdf0/ : SUCCESS in 2h 31m 32s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/62/671262/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/a672e49/ : SUCCESS in 2h 37m 38s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/62/671262/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/2065d1e/ : SUCCESS in 2h 11m 26s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/62/671262/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/67300d6/ : SUCCESS in 2h 11m 34s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/62/671262/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/c9c561e/ : SUCCESS in 2h 25m 22s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/62/671262/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/512db17/ : FAILURE in 1h 39m 19s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/62/671262/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/a0b3e9e/ : SUCCESS in 2h 46m 57s (non-voting)\n- tripleo-ci-rhel-8-standalone-rdo http://logs.rdoproject.org/62/671262/3/openstack-check/tripleo-ci-rhel-8-standalone-rdo/1508a18/ : FAILURE in 14m 12s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5d62cd7f99a77afda7c82fade89bbb1531dede77","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 10:38:44.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/62/671262/3/check/tripleo-ci-centos-7-undercloud-containers/f7100c2/ : SUCCESS in 1h 28m 21s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/62/671262/3/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/53c20c7/ : TIMED_OUT in 3h 12m 05s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/62/671262/3/check/tripleo-ci-centos-7-containers-multinode/ed1afa0/ : SUCCESS in 2h 11m 28s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://logs.opendev.org/62/671262/3/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/b85e717/ : SUCCESS in 2h 09m 09s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://logs.opendev.org/62/671262/3/check/tripleo-ci-centos-7-scenario012-multinode-oooq-container/e13c974/ : FAILURE in 2h 23m 01s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/62/671262/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/93c1ca2/ : SUCCESS in 2h 17m 43s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/62/671262/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/feb6204/ : SUCCESS in 2h 11m 36s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/62/671262/3/check/tripleo-ci-centos-7-standalone/aa62a65/ : SUCCESS in 1h 36m 56s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/62/671262/3/check/tripleo-ci-fedora-28-standalone/eae7fa0/ : FAILURE in 45m 20s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/62/671262/3/check/tripleo-ci-centos-7-standalone-upgrade/2ff1b7c/ : FAILURE in 1h 42m 01s (non-voting)\n- tripleo-ci-centos-7-standalone-os-tempest https://logs.opendev.org/62/671262/3/check/tripleo-ci-centos-7-standalone-os-tempest/81ba317/ : SUCCESS in 1h 06m 39s\n- tripleo-ci-centos-7-containers-multinode-stein https://logs.opendev.org/62/671262/3/check/tripleo-ci-centos-7-containers-multinode-stein/f7c29e6/ : SUCCESS in 1h 58m 44s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky https://logs.opendev.org/62/671262/3/check/tripleo-ci-centos-7-containers-multinode-rocky/3a33662/ : SUCCESS in 2h 03m 38s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens https://logs.opendev.org/62/671262/3/check/tripleo-ci-centos-7-containers-multinode-queens/3f51725/ : FAILURE in 1h 12m 49s\n- tripleo-ci-centos-7-containers-multinode-pike https://logs.opendev.org/62/671262/3/check/tripleo-ci-centos-7-containers-multinode-pike/c7bc3ed/ : SUCCESS in 2h 09m 26s\n- openstack-tox-linters https://logs.opendev.org/62/671262/3/check/openstack-tox-linters/d4ba22b/ : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":3},{"id":"2de05511e486b867ccc9489ac9ac983a2cddd945","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2019-08-01 12:34:22.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cef872c57f745ae57dc221fdbc24942e24a943ee","author":{"_account_id":8116,"name":"John Dennis","email":"jdennis@redhat.com","username":"jdennis"},"date":"2019-08-01 13:02:57.000000000","message":"Patch Set 3:\n\nI imagine that\u0027s because you don\u0027t have the additional variables set in featureset039 that is part of this review https://review.opendev.org/#/c/671567. Did you apply both set of changes? It always seemed a bit odd to me that tripleo-quickstart and tripleo-quickstart-extras were in distinct and different repos when there is so much shared between them and causes what should be one unified set of changes to be split between different reviews.","accounts_in_message":[],"_revision_number":3},{"id":"a9a8568960b09eed720c7f49f2afcedaf1a18af6","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2019-08-01 14:52:58.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"6339444614b45f9d558734d7d05b0fd69864ff2a","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2019-08-01 16:51:17.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"de7423d1989f070bf90ab5b6086ed91c14b372d3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-01 20:16:27.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/62/671262/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/408082d/ : SUCCESS in 2h 40m 31s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/62/671262/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/659e80c/ : SUCCESS in 2h 39m 19s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/62/671262/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/73db141/ : SUCCESS in 2h 33m 16s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/62/671262/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/392f153/ : SUCCESS in 2h 20m 11s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/62/671262/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/46b6706/ : SUCCESS in 2h 35m 38s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/62/671262/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/50bb8d1/ : SUCCESS in 3h 23m 14s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/62/671262/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/339e605/ : SUCCESS in 2h 55m 00s (non-voting)\n- tripleo-ci-rhel-8-standalone-rdo http://logs.rdoproject.org/62/671262/5/openstack-check/tripleo-ci-rhel-8-standalone-rdo/7e8ba55/ : FAILURE in 14m 27s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"918f0819c437edc57251d6254b402667d0978f48","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 03:17:44.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/62/671262/5/check/tripleo-ci-centos-7-undercloud-containers/5abc6de/ : SUCCESS in 1h 17m 27s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/62/671262/5/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/9c8b32c/ : TIMED_OUT in 3h 11m 27s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/62/671262/5/check/tripleo-ci-centos-7-containers-multinode/10b9f72/ : SUCCESS in 2h 17m 35s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://logs.opendev.org/62/671262/5/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/e69aa1a/ : SUCCESS in 2h 27m 42s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://logs.opendev.org/62/671262/5/check/tripleo-ci-centos-7-scenario012-multinode-oooq-container/ebd3120/ : FAILURE in 2h 22m 23s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/62/671262/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/d6f263c/ : SUCCESS in 2h 17m 55s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/62/671262/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/e629891/ : SUCCESS in 2h 25m 12s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/62/671262/5/check/tripleo-ci-centos-7-standalone/c6cc21c/ : SUCCESS in 1h 44m 45s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/62/671262/5/check/tripleo-ci-fedora-28-standalone/ef260c3/ : FAILURE in 45m 16s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/62/671262/5/check/tripleo-ci-centos-7-standalone-upgrade/2265318/ : FAILURE in 1h 38m 06s (non-voting)\n- tripleo-ci-centos-7-standalone-os-tempest https://logs.opendev.org/62/671262/5/check/tripleo-ci-centos-7-standalone-os-tempest/fd8542a/ : SUCCESS in 1h 16m 42s\n- tripleo-ci-centos-7-containers-multinode-stein https://logs.opendev.org/62/671262/5/check/tripleo-ci-centos-7-containers-multinode-stein/8e87b1b/ : SUCCESS in 2h 43m 54s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky https://logs.opendev.org/62/671262/5/check/tripleo-ci-centos-7-containers-multinode-rocky/79590dc/ : SUCCESS in 2h 30m 48s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens https://logs.opendev.org/62/671262/5/check/tripleo-ci-centos-7-containers-multinode-queens/656887b/ : SUCCESS in 2h 12m 48s\n- tripleo-ci-centos-7-containers-multinode-pike https://logs.opendev.org/62/671262/5/check/tripleo-ci-centos-7-containers-multinode-pike/efc085b/ : SUCCESS in 2h 10m 37s\n- openstack-tox-linters https://logs.opendev.org/62/671262/5/check/openstack-tox-linters/63ed660/ : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":5},{"id":"7622fcf1c10113816cdc9aa2faa1a6a129da73a5","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2019-08-05 18:56:39.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"915a1938114b919a1e01c87008a31a7c84cf9741","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-08-05 18:56:56.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/3723/ : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"e52b0d4610db0948f93f66e0ef94647c6e597bc0","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2019-08-05 18:57:00.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"0625cb04f6fa43be35eddff5c420813305ecaffd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 18:57:08.000000000","message":"Patch Set 6:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":6},{"id":"0eef910ac6b61d172e4d87a21aca53cc1448fd80","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-05 18:57:08.000000000","message":"Patch Set 6: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":6},{"id":"8f14bfbbde67b8ec68bc3cc946e9521cd2fc54c2","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-08-05 18:57:17.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/3725/ : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"0549b85f1aef00ca59bf45006c65c10693c5c3a1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 21:35:07.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/62/671262/7/check/tripleo-ci-centos-7-undercloud-containers/4e04297/ : SUCCESS in 1h 31m 20s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/62/671262/7/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/fae3529/ : FAILURE in 1h 40m 35s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/62/671262/7/check/tripleo-ci-centos-7-containers-multinode/83f24f7/ : SUCCESS in 2h 22m 51s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://logs.opendev.org/62/671262/7/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/3ad8c02/ : SUCCESS in 2h 10m 24s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://logs.opendev.org/62/671262/7/check/tripleo-ci-centos-7-scenario012-multinode-oooq-container/364f314/ : FAILURE in 2h 19m 22s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/62/671262/7/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/58ebccc/ : SUCCESS in 2h 23m 38s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/62/671262/7/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/9510ae0/ : FAILURE in 1h 00m 02s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/62/671262/7/check/tripleo-ci-centos-7-standalone/be7d665/ : SUCCESS in 1h 47m 54s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/62/671262/7/check/tripleo-ci-fedora-28-standalone/aa543e6/ : FAILURE in 50m 26s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/62/671262/7/check/tripleo-ci-centos-7-standalone-upgrade/a02a61c/ : SUCCESS in 2h 03m 58s (non-voting)\n- tripleo-ci-centos-7-standalone-os-tempest https://logs.opendev.org/62/671262/7/check/tripleo-ci-centos-7-standalone-os-tempest/4873e03/ : FAILURE in 51m 04s\n- tripleo-ci-centos-7-containers-multinode-stein https://logs.opendev.org/62/671262/7/check/tripleo-ci-centos-7-containers-multinode-stein/5526aa8/ : SUCCESS in 2h 03m 09s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky https://logs.opendev.org/62/671262/7/check/tripleo-ci-centos-7-containers-multinode-rocky/f4601cf/ : SUCCESS in 2h 23m 05s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens https://logs.opendev.org/62/671262/7/check/tripleo-ci-centos-7-containers-multinode-queens/86d5093/ : SUCCESS in 2h 19m 27s\n- tripleo-ci-centos-7-containers-multinode-pike https://logs.opendev.org/62/671262/7/check/tripleo-ci-centos-7-containers-multinode-pike/4c3579c/ : SUCCESS in 2h 17m 53s\n- openstack-tox-linters https://logs.opendev.org/62/671262/7/check/openstack-tox-linters/2f6895e/ : FAILURE in 5m 34s","accounts_in_message":[],"_revision_number":7},{"id":"63eff93f70327057ae07ef94264807d1705d10e1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-05 22:40:20.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 finger://ze03.softwarefactory-project.io:7979/29a52ad8cae54f3ab4e3026d869d77d8 : POST_FAILURE in 3h 35m 23s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 finger://ze01.softwarefactory-project.io:7979/6f07143a753b4a539b6cfa881e251090 : POST_FAILURE in 3h 41m 08s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/62/671262/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/d657670/ : FAILURE in 1h 10m 34s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/62/671262/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/b68edf4/ : FAILURE in 1h 22m 44s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch finger://ze02.softwarefactory-project.io:7979/dca8006ace9e46948e5980e23a9d8223 : POST_FAILURE in 3h 30m 04s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 finger://ze03.softwarefactory-project.io:7979/2c5db8968c7a4755a07a71e7e23d0d30 : POST_FAILURE in 3h 36m 43s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/62/671262/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/0702821/ : FAILURE in 39m 58s (non-voting)\n- tripleo-ci-rhel-8-standalone-rdo http://logs.rdoproject.org/62/671262/7/openstack-check/tripleo-ci-rhel-8-standalone-rdo/767480b/ : FAILURE in 13m 19s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"7b2c412faea2f0c897ddb62d3beb47f39137a89e","author":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"date":"2019-09-17 12:34:38.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"ee5120821923f213e493d210f7f9111e44b469a4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-17 15:34:03.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/62/671262/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/e4fcc16/ : FAILURE in 2h 57m 48s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/62/671262/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/067b5a4/ : SUCCESS in 2h 10m 36s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/62/671262/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/8874443/ : SUCCESS in 2h 05m 27s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/62/671262/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/182da41/ : FAILURE in 1h 44m 02s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/62/671262/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/0eca42a/ : FAILURE in 34m 58s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"e4d7ffce8c496bd23e175d2263a91cfbc46ef5e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 17:04:56.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/1e1d1a2f61d240ef972542064e5e7e42 : SUCCESS in 2h 08m 06s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/d22b0b71aee3496ea5974a11c48974b5 : FAILURE in 2h 24m 07s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/8f20dfe2155d47e7b689361d397ae195 : SUCCESS in 3h 08m 47s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/797db6b792f9481e90780d955b2022b3 : SUCCESS in 3h 02m 49s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/646a0f9ea65e4f909f31256e5a6541f1 : FAILURE in 2h 33m 28s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/ee1b0033b1cd43a5b2733545f0c7144c : SUCCESS in 2h 42m 50s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/c835b22772d446dba9840c57cf1e8db4 : SUCCESS in 2h 28m 36s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/56ce499ea2354c81a6ce90ca4971a6a5 : SUCCESS in 2h 14m 27s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/377a7e2ddd6b443c988700077504a5ae : SUCCESS in 2h 22m 14s (non-voting)\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/6b6cc90c50934879829fca3533e10e72 : SUCCESS in 1h 39m 04s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/925f59d386cc43f3a3f3259609c805f4 : SUCCESS in 2h 39m 17s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/544c18000dba45068f5adaf7a0ef46e5 : SUCCESS in 2h 31m 57s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/8b31aff1fc494faba4ea803b384134b8 : SUCCESS in 1h 52m 49s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/df5e64950b8a42749742c3e29d759bf7 : FAILURE in 4m 45s","accounts_in_message":[],"_revision_number":8},{"id":"5957b8126b7f28dd02b274642b3cabd48ab2927e","author":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"date":"2019-09-30 15:44:20.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"758b4f68cb7c61591c3a9b80a83411314fe4e640","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-30 18:42:17.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/62/671262/9/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/c80b8b6/ : SUCCESS in 2h 43m 07s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/62/671262/9/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/262dc0e/ : FAILURE in 1h 53m 01s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/62/671262/9/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/470b7a9/ : FAILURE in 36m 14s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/62/671262/9/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/5721a04/ : SUCCESS in 2h 55m 51s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/62/671262/9/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/ce0ba13/ : FAILURE in 40m 12s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"c1eea366697d5b40ba9aafd6472f32bab5aab196","author":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"date":"2019-09-30 19:07:12.000000000","message":"Patch Set 9: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"46457b66c6965fa0581167f1008c442c27e35703","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 19:59:17.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/395df69c483948b6b197faa7521542cb : SUCCESS in 1h 59m 30s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/9773508bef0e430cbaa374fdea213bd3 : FAILURE in 2h 11m 41s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/dab1217cf4504c13af28c52e6408794b : SUCCESS in 3h 11m 15s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/85f8a1c9ed284047bdd440396d860a06 : SUCCESS in 2h 50m 55s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/bf2ab6b20e7a43cfae96cbc94c809689 : FAILURE in 2h 28m 55s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/5cd65b44cbfe4f40b36648728220b0a5 : SUCCESS in 2h 45m 51s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/84943ece4f724e01b5096bd073a74843 : SUCCESS in 3h 02m 09s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/21d6e47bc7c84257873b5434cdd238e3 : SUCCESS in 2h 31m 01s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/7b45fbb0eed24d0bac2e5e126ef51bf6 : FAILURE in 2h 10m 44s (non-voting)\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/1fa57e5b0c3e436dba21e9ec0cb54a68 : SUCCESS in 1h 45m 15s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/5ac7341b1bde41c1929e148bed4c9ec9 : SUCCESS in 3h 19m 22s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/2c50fcd1ac3040d2a7adede7a09de88b : FAILURE in 46m 38s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/bf6d997a04504816b4b7d7bf28da2770 : SUCCESS in 2h 35m 15s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/669066d5e0ad419cbb90a3c808528423 : FAILURE in 4m 30s","accounts_in_message":[],"_revision_number":9},{"id":"06f1dedbc1cdc25f32bf88ef40e0333bdbc147bc","author":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"date":"2019-10-02 18:31:13.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"231a9f1bed891dfd02b7a077e6352cb3261b3b4f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-02 21:35:25.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/62/671262/10/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/04640c8/ : FAILURE in 1h 48m 40s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/62/671262/10/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/8c08471/ : FAILURE in 2h 05m 35s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/62/671262/10/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/ead0051/ : FAILURE in 2h 03m 10s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/62/671262/10/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/b5178eb/ : SUCCESS in 3h 02m 42s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/62/671262/10/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/1316082/ : FAILURE in 39m 58s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"df6160670cf679b54502b396bcc1ae8e6ddbdc3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 22:24:39.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/49d69c8a454e4932b3321dbd94d67316 : SUCCESS in 1h 33m 45s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/89b8d58d80924bb4a8fd2d30e61f7b46 : FAILURE in 2h 03m 35s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/1a562b4df2b54df883bc402b0775e3cf : SUCCESS in 3h 19m 26s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/1a62bf61a4464d719be15ed1ba77bb13 : SUCCESS in 3h 18m 13s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/5e172825418340dfa5a18bf6486eb7e3 : FAILURE in 2h 13m 32s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/e18b1bf48d524dd49749542a8f89b517 : SUCCESS in 2h 30m 03s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/8eaa2f0a6b5c42c5b231ebd5df856a2b : SUCCESS in 2h 44m 05s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/325f460bc8554933a0961e3b79bcd070 : SUCCESS in 2h 26m 10s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/3c78322f94334b93a26e7cd13dc94694 : FAILURE in 1h 48m 37s (non-voting)\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/84fe8939875747b994095c28c8799cc4 : SUCCESS in 1h 47m 22s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/cfe2457bed6943b3bf7ba57c09c9d619 : SUCCESS in 3h 05m 08s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/37514abb45834b408df46c157b54daca : SUCCESS in 3h 06m 30s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/e78b376e239c42719d7b90b0d9d4ae43 : SUCCESS in 2h 31m 59s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fa4402105f034dd9afbc7adf326b1577 : FAILURE in 5m 35s","accounts_in_message":[],"_revision_number":10},{"id":"1cbde3c895549858dffc3120c483d3ec9df1b229","author":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"date":"2019-10-07 15:57:53.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"87b4781c0d2c3f00c123b4e6f1aeadb02b1cfb90","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-07 19:13:25.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/62/671262/11/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/22c7dbc/ : SUCCESS in 3h 13m 53s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/62/671262/11/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/33d40f8/ : FAILURE in 1h 43m 28s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/62/671262/11/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/4d3c56a/ : SUCCESS in 2h 26m 59s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/62/671262/11/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/af8fee2/ : SUCCESS in 3h 00m 23s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/62/671262/11/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/8c539e9/ : FAILURE in 40m 17s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"bfeadb855f44616fd6604222e926f635096a433f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 19:46:06.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/4e9d703b617442a1b86086547c21cc56 : SUCCESS in 1h 56m 14s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/96da1a7946a246acbf7bb8884c04ddd0 : FAILURE in 1h 58m 54s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/041ab4d1bbdf4f64a7573cb351edc236 : SUCCESS in 2h 40m 45s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/bc5fa7b4013d4813b8df573d7b150a21 : SUCCESS in 3h 08m 33s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/f377607100f74e30a02253b9d8d146d9 : FAILURE in 2h 46m 48s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/34df7174f8ef48ea81b1b15b60639e40 : SUCCESS in 2h 47m 11s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/753043d212624cd780dffa27ce741a91 : SUCCESS in 2h 24m 41s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/c3dea4144efe4958aa65e4c2227b30eb : SUCCESS in 2h 31m 20s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/6f90ec5493f643b2812979bf353b1d4a : FAILURE in 1h 45m 59s (non-voting)\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/fc33e5c9bd394f92b7a79fe72f88f362 : SUCCESS in 1h 42m 43s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/0ceed16c256f46cf8d02c47efbcb7e92 : SUCCESS in 2h 50m 41s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/c28c9068b3504cd7b71202e5f3e5d9dd : SUCCESS in 2h 26m 54s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/378d8ae25a7e4f7d9b0ef89b60fc9dfc : SUCCESS in 2h 18m 11s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/14dbd1c293fa499eb36aa5d6810190bb : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":11},{"id":"244d15cf45b45cbcca45ac3fd07dfaef7af4a43b","author":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"date":"2019-11-05 17:50:56.000000000","message":"Patch Set 11:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":11},{"id":"17d5a47feecaec569ac90e71ddd6c9c960749be2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-05 21:03:40.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/62/671262/11/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/20214cc/ : SUCCESS in 2h 50m 50s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/62/671262/11/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/3e9fcf4/ : SUCCESS in 2h 02m 21s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/62/671262/11/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/306f3a6/ : SUCCESS in 2h 08m 10s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/62/671262/11/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/e18e2db/ : SUCCESS in 3h 10m 53s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/62/671262/11/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/43248a5/ : FAILURE in 38m 19s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"6d5cc3c57008fa1cfe992d9dcef3c1efe27e6d0d","author":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"date":"2020-01-21 11:51:59.000000000","message":"Patch Set 12: Patch Set 11 was rebased","accounts_in_message":[],"_revision_number":12},{"id":"a902d93a1df31e00de649f8846b8e856a0b8a55b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-21 12:15:52.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/4812da2a3fa24067950da8d993cb7e3e : RETRY_LIMIT in 5m 36s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch https://review.rdoproject.org/zuul/build/7fa3b95e20b84c9593640012b62f02b8 : RETRY_LIMIT in 5m 53s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch https://review.rdoproject.org/zuul/build/1fc0f4e386bd4ba1a61a482de72b4aaf : RETRY_LIMIT in 5m 38s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch https://review.rdoproject.org/zuul/build/cb2bec01cffb4e4b9a60c287cf1e8c31 : RETRY_LIMIT in 5m 58s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 https://review.rdoproject.org/zuul/build/913685973d1c409d88d31b640f73c9fe : RETRY_LIMIT in 5m 48s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"d11312c992928c66057c59ee2b8113e8251336ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 12:22:12.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/2adb782265cb4f80a4a6ea1ad3f72374 : FAILURE in 9m 22s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/1d54af86c16445c2ae2b7a3d45dfc6e0 : FAILURE in 10m 14s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/04cbfd53b1d74f42a9d46bd6e840607c : FAILURE in 17m 54s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/a7f8721c33184de6809375c046c6dad8 : FAILURE in 12m 20s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/697a2849ffc444cbb33ae8f2331bc3b2 : FAILURE in 15m 14s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/bec4da8fae074cba8125bd695c5e83af : FAILURE in 15m 52s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/6345186e74634e5ebb79a5d43f2bf05b : FAILURE in 15m 35s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/f9a39a3b0eaf4472a85d8edea2ca77fe : FAILURE in 13m 37s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-train https://zuul.opendev.org/t/openstack/build/5d70be9a204f46a19e6e958ac383bc8f : FAILURE in 18m 10s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/5bc602a6348e4ab28a09f0cf25086d33 : FAILURE in 18m 15s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/67ee2c91a8cc47a5b72a8b1f98822e16 : FAILURE in 11m 16s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/5bed9f41fcc0437d8bddcd48aade8d63 : FAILURE in 14m 05s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/69911eac12d44ae9a3ad1eedd9f44ea0 : SUCCESS in 7m 32s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/fca2399c25414ef7b23b80105d9650ff : FAILURE in 12m 43s","accounts_in_message":[],"_revision_number":12},{"id":"1d1845ac77b55dbeabc5cd78c3a0ade8d271e019","author":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"date":"2020-05-22 16:13:00.000000000","message":"Patch Set 13: Patch Set 12 was rebased","accounts_in_message":[],"_revision_number":13},{"id":"b285f1209f163e89d1329a59250bff40cfe475b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 19:16:42.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/fe14af08a97344c19dbaf16ab65aac66 : SUCCESS in 1h 29m 03s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/be1affb3663946ec8be9a60b74dfeb8d : SUCCESS in 2h 23m 36s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/4a7debbafa5d4c0292a2d97a98c76ca4 : SUCCESS in 2h 08m 02s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/1a950f61a4f9469ea7d5ccfb54488346 : SUCCESS in 2h 16m 49s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/722d7656f8e64df0a20eac8f54381e2b : SUCCESS in 1h 50m 52s\n- tripleo-ci-centos-7-containers-multinode-train https://zuul.opendev.org/t/openstack/build/7a6806ca75194eb886422e3d47ea90a5 : SUCCESS in 2h 33m 27s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/8270557bfd3a4b40ab3cb2f477852ed5 : SUCCESS in 2h 07m 26s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/71d38471b39743d99bdc4a79de471ddb : SUCCESS in 2h 17m 02s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/0902029ce152461a8b20177a9db328f7 : SUCCESS in 2h 03m 08s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0d4aca116a7747acbd70fed608c5a888 : SUCCESS in 9m 42s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/beff94a214db4bbbb71dc37caabfd908 : SUCCESS in 15m 21s","accounts_in_message":[],"_revision_number":13},{"id":"ee4ab329a070f786b90d4f18e949fed4885a7564","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-05-22 19:24:24.000000000","message":"Patch Set 13: Code-Review-1\n\nWhy are we adding that role into tripleo-quickstart-extras?\nCan\u0027t it be re-used in TripleO outside of Quickstart?\n\nIMO the best place for that role could be https://opendev.org/x/tripleo-ipa/src/branch/master/tripleo_ipa/roles maybe?","accounts_in_message":[],"_revision_number":13},{"id":"ee077cfe4a82c37be9e00cb8c414348f07eb94fd","author":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"date":"2020-05-22 19:54:46.000000000","message":"Patch Set 13:\n\nThis can be abandoned for now, we need to rework this following the new approach for dealing with ansible roles.","accounts_in_message":[],"_revision_number":13},{"id":"73fff5558a097c7be59b322571e4287de913d2f5","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-05-22 19:56:29.000000000","message":"Patch Set 13:\n\n\u003e This can be abandoned for now, we need to rework this following the\n \u003e new approach for dealing with ansible roles.\n\nok","accounts_in_message":[],"_revision_number":13},{"id":"6fa21dad7bdb6f0c3bb8b600aaf176fd72934875","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-05-22 19:56:32.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"8b3dfb4a68ce504d7c627f740859b513d27fbc6d","revisions":{"22e1c696350f33f5cc7aa0c5376c2c150154d4bb":{"kind":"REWORK","_number":1,"created":"2019-07-17 13:15:42.000000000","uploader":{"_account_id":8116,"name":"John Dennis","email":"jdennis@redhat.com","username":"jdennis"},"ref":"refs/changes/62/671262/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/62/671262/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/62/671262/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/62/671262/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/62/671262/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/62/671262/1"}}},"commit":{"parents":[{"commit":"235b73aea9343591ca5f5d91fa51bb6a818f3149","subject":"[Core] Improve the try part.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/235b73aea9343591ca5f5d91fa51bb6a818f3149"}]}],"author":{"name":"John Dennis","email":"jdennis@redhat.com","date":"2019-04-08 19:48:12.000000000","tz":-240},"committer":{"name":"John Dennis","email":"jdennis@redhat.com","date":"2019-07-17 12:55:56.000000000","tz":-240},"subject":"Add keycloak-setup role","message":"Add keycloak-setup role\n\nkeycloak-setup is a new role used to install the Keycloak IdP on the\nsupplemental node. It utilizes the nkinder.keycloak Ansible Galaxy\nrole to perform the installation. This role depends on the existing\nfreeipa-setup role because it utilizes FreeIPA to obtain the necessary\nTLS cert and key, this is checked via the enable_tls_everywhere\nboolean. The Keycloak IdP is necessary to support testing Keystone\nfederated authentication which is enabled via the new\nenable_federation boolean.\n\nPartial-Bug: 1836415\n\nChange-Id: I4ac636572f537b98213745dd0a68ba89b4a1d74b\nSigned-off-by: John Dennis \u003cjdennis@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/22e1c696350f33f5cc7aa0c5376c2c150154d4bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/22e1c696350f33f5cc7aa0c5376c2c150154d4bb"}]},"branch":"refs/heads/master"},"4aabdbb2c9c0c2459a9c9aac3297a666a1437963":{"kind":"REWORK","_number":2,"created":"2019-07-31 17:10:15.000000000","uploader":{"_account_id":8116,"name":"John Dennis","email":"jdennis@redhat.com","username":"jdennis"},"ref":"refs/changes/62/671262/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/62/671262/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/62/671262/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/62/671262/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/62/671262/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/62/671262/2"}}},"commit":{"parents":[{"commit":"235b73aea9343591ca5f5d91fa51bb6a818f3149","subject":"[Core] Improve the try part.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/235b73aea9343591ca5f5d91fa51bb6a818f3149"}]}],"author":{"name":"John Dennis","email":"jdennis@redhat.com","date":"2019-04-08 19:48:12.000000000","tz":-240},"committer":{"name":"John Dennis","email":"jdennis@redhat.com","date":"2019-07-31 17:09:28.000000000","tz":-240},"subject":"Add keycloak-setup role","message":"Add keycloak-setup role\n\nkeycloak-setup is a new role used to install the Keycloak IdP on the\nsupplemental node. It utilizes the nkinder.keycloak Ansible Galaxy\nrole to perform the installation. This role depends on the existing\nfreeipa-setup role because it utilizes FreeIPA to obtain the necessary\nTLS cert and key, this is checked via the enable_tls_everywhere\nboolean. The Keycloak IdP is necessary to support testing Keystone\nfederated authentication which is enabled via the new\nenable_federation boolean.\n\nPartial-Bug: 1836415\n\nChange-Id: I4ac636572f537b98213745dd0a68ba89b4a1d74b\nSigned-off-by: John Dennis \u003cjdennis@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/4aabdbb2c9c0c2459a9c9aac3297a666a1437963"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/4aabdbb2c9c0c2459a9c9aac3297a666a1437963"}]},"branch":"refs/heads/master"},"d425498099aab88d583d45dfa37f62ca6d00a48e":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-08-01 00:50:37.000000000","uploader":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"ref":"refs/changes/62/671262/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/62/671262/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/62/671262/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/62/671262/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/62/671262/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/62/671262/3"}}},"commit":{"parents":[{"commit":"08a6913eaef03441b02cc3035b664f415ace6a07","subject":"WIP: allow external ansible role for 3rd party job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/08a6913eaef03441b02cc3035b664f415ace6a07"}]}],"author":{"name":"John Dennis","email":"jdennis@redhat.com","date":"2019-04-08 19:48:12.000000000","tz":-240},"committer":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2019-08-01 00:50:11.000000000","tz":-360},"subject":"Add keycloak-setup role","message":"Add keycloak-setup role\n\nkeycloak-setup is a new role used to install the Keycloak IdP on the\nsupplemental node. It utilizes the nkinder.keycloak Ansible Galaxy\nrole to perform the installation. This role depends on the existing\nfreeipa-setup role because it utilizes FreeIPA to obtain the necessary\nTLS cert and key, this is checked via the enable_tls_everywhere\nboolean. The Keycloak IdP is necessary to support testing Keystone\nfederated authentication which is enabled via the new\nenable_federation boolean.\n\nPartial-Bug: 1836415\n\nChange-Id: I4ac636572f537b98213745dd0a68ba89b4a1d74b\nSigned-off-by: John Dennis \u003cjdennis@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/d425498099aab88d583d45dfa37f62ca6d00a48e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/d425498099aab88d583d45dfa37f62ca6d00a48e"}]},"branch":"refs/heads/master"},"589468f92f993fb0d98adf17b6897e8daa317cb5":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-08-01 14:52:58.000000000","uploader":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"ref":"refs/changes/62/671262/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/62/671262/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/62/671262/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/62/671262/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/62/671262/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/62/671262/4"}}},"commit":{"parents":[{"commit":"f2a2a6aed768cafa1b425c59771308b5f8309b0e","subject":"WIP: allow external ansible role for 3rd party job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/f2a2a6aed768cafa1b425c59771308b5f8309b0e"}]}],"author":{"name":"John Dennis","email":"jdennis@redhat.com","date":"2019-04-08 19:48:12.000000000","tz":-240},"committer":{"name":"wes hayutin","email":"whayutin@redhat.com","date":"2019-08-01 14:52:58.000000000","tz":0},"subject":"Add keycloak-setup role","message":"Add keycloak-setup role\n\nkeycloak-setup is a new role used to install the Keycloak IdP on the\nsupplemental node. It utilizes the nkinder.keycloak Ansible Galaxy\nrole to perform the installation. This role depends on the existing\nfreeipa-setup role because it utilizes FreeIPA to obtain the necessary\nTLS cert and key, this is checked via the enable_tls_everywhere\nboolean. The Keycloak IdP is necessary to support testing Keystone\nfederated authentication which is enabled via the new\nenable_federation boolean.\n\nPartial-Bug: 1836415\n\nChange-Id: I4ac636572f537b98213745dd0a68ba89b4a1d74b\nSigned-off-by: John Dennis \u003cjdennis@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/589468f92f993fb0d98adf17b6897e8daa317cb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/589468f92f993fb0d98adf17b6897e8daa317cb5"}]},"branch":"refs/heads/master"},"b8bd910f9f9cb2776cd15a1a353e8c27556b8f8f":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-08-01 16:51:17.000000000","uploader":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"ref":"refs/changes/62/671262/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/62/671262/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/62/671262/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/62/671262/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/62/671262/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/62/671262/5"}}},"commit":{"parents":[{"commit":"5cd61be2795bf670c4f4721a221a9d7702c317cf","subject":"WIP: allow external ansible role for 3rd party job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/5cd61be2795bf670c4f4721a221a9d7702c317cf"}]}],"author":{"name":"John Dennis","email":"jdennis@redhat.com","date":"2019-04-08 19:48:12.000000000","tz":-240},"committer":{"name":"wes hayutin","email":"whayutin@redhat.com","date":"2019-08-01 16:51:17.000000000","tz":0},"subject":"Add keycloak-setup role","message":"Add keycloak-setup role\n\nkeycloak-setup is a new role used to install the Keycloak IdP on the\nsupplemental node. It utilizes the nkinder.keycloak Ansible Galaxy\nrole to perform the installation. This role depends on the existing\nfreeipa-setup role because it utilizes FreeIPA to obtain the necessary\nTLS cert and key, this is checked via the enable_tls_everywhere\nboolean. The Keycloak IdP is necessary to support testing Keystone\nfederated authentication which is enabled via the new\nenable_federation boolean.\n\nPartial-Bug: 1836415\n\nChange-Id: I4ac636572f537b98213745dd0a68ba89b4a1d74b\nSigned-off-by: John Dennis \u003cjdennis@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/b8bd910f9f9cb2776cd15a1a353e8c27556b8f8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/b8bd910f9f9cb2776cd15a1a353e8c27556b8f8f"}]},"branch":"refs/heads/master"},"9722db87c8c6ee7eece542775b26d1ef18a18689":{"kind":"REWORK","_number":6,"created":"2019-08-05 18:56:39.000000000","uploader":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"ref":"refs/changes/62/671262/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/62/671262/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/62/671262/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/62/671262/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/62/671262/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/62/671262/6"}}},"commit":{"parents":[{"commit":"5cd61be2795bf670c4f4721a221a9d7702c317cf","subject":"WIP: allow external ansible role for 3rd party job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/5cd61be2795bf670c4f4721a221a9d7702c317cf"}]}],"author":{"name":"John Dennis","email":"jdennis@redhat.com","date":"2019-04-08 19:48:12.000000000","tz":-240},"committer":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2019-08-05 18:56:32.000000000","tz":-360},"subject":"Add keycloak-setup role","message":"Add keycloak-setup role\n\nkeycloak-setup is a new role used to install the Keycloak IdP on the\nsupplemental node. It utilizes the nkinder.keycloak Ansible Galaxy\nrole to perform the installation. This role depends on the existing\nfreeipa-setup role because it utilizes FreeIPA to obtain the necessary\nTLS cert and key, this is checked via the enable_tls_everywhere\nboolean. The Keycloak IdP is necessary to support testing Keystone\nfederated authentication which is enabled via the new\nenable_federation boolean.\n\nPartial-Bug: 1836415\n\nChange-Id: I4ac636572f537b98213745dd0a68ba89b4a1d74b\nSigned-off-by: John Dennis \u003cjdennis@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/9722db87c8c6ee7eece542775b26d1ef18a18689"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/9722db87c8c6ee7eece542775b26d1ef18a18689"}]},"branch":"refs/heads/master"},"ad287d4422b73f563cf0acfb62ce75bdca6ca574":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-08-05 18:57:00.000000000","uploader":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"ref":"refs/changes/62/671262/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/62/671262/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/62/671262/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/62/671262/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/62/671262/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/62/671262/7"}}},"commit":{"parents":[{"commit":"f951c7705db07b4c3d6d65f1a61bac488ff0377d","subject":"WIP: allow external ansible role for 3rd party job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/f951c7705db07b4c3d6d65f1a61bac488ff0377d"}]}],"author":{"name":"John Dennis","email":"jdennis@redhat.com","date":"2019-04-08 19:48:12.000000000","tz":-240},"committer":{"name":"wes hayutin","email":"whayutin@redhat.com","date":"2019-08-05 18:57:00.000000000","tz":0},"subject":"Add keycloak-setup role","message":"Add keycloak-setup role\n\nkeycloak-setup is a new role used to install the Keycloak IdP on the\nsupplemental node. It utilizes the nkinder.keycloak Ansible Galaxy\nrole to perform the installation. This role depends on the existing\nfreeipa-setup role because it utilizes FreeIPA to obtain the necessary\nTLS cert and key, this is checked via the enable_tls_everywhere\nboolean. The Keycloak IdP is necessary to support testing Keystone\nfederated authentication which is enabled via the new\nenable_federation boolean.\n\nPartial-Bug: 1836415\n\nChange-Id: I4ac636572f537b98213745dd0a68ba89b4a1d74b\nSigned-off-by: John Dennis \u003cjdennis@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/ad287d4422b73f563cf0acfb62ce75bdca6ca574"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/ad287d4422b73f563cf0acfb62ce75bdca6ca574"}]},"branch":"refs/heads/master"},"829c327bc9a9b0e22fbd7749b31f3422bb087321":{"kind":"REWORK","_number":8,"created":"2019-09-17 12:34:38.000000000","uploader":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"ref":"refs/changes/62/671262/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/62/671262/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/62/671262/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/62/671262/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/62/671262/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/62/671262/8"}}},"commit":{"parents":[{"commit":"258a3772b91fc7c98aeaf6bb904179ff49506585","subject":"WIP: allow external ansible role for 3rd party job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/258a3772b91fc7c98aeaf6bb904179ff49506585"}]}],"author":{"name":"John Dennis","email":"jdennis@redhat.com","date":"2019-04-08 19:48:12.000000000","tz":-240},"committer":{"name":"Raildo Mascena","email":"rmascena@redhat.com","date":"2019-09-17 12:34:12.000000000","tz":-180},"subject":"Add keycloak-setup role","message":"Add keycloak-setup role\n\nkeycloak-setup is a new role used to install the Keycloak IdP on the\nsupplemental node. It utilizes the nkinder.keycloak Ansible Galaxy\nrole to perform the installation. This role depends on the existing\nfreeipa-setup role because it utilizes FreeIPA to obtain the necessary\nTLS cert and key, this is checked via the enable_tls_everywhere\nboolean. The Keycloak IdP is necessary to support testing Keystone\nfederated authentication which is enabled via the new\nenable_federation boolean.\n\nPartial-Bug: 1836415\n\nChange-Id: I4ac636572f537b98213745dd0a68ba89b4a1d74b\nSigned-off-by: John Dennis \u003cjdennis@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/829c327bc9a9b0e22fbd7749b31f3422bb087321"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/829c327bc9a9b0e22fbd7749b31f3422bb087321"}]},"branch":"refs/heads/master"},"a68962a6a86f40e386ad523797e4c6f91593534b":{"kind":"REWORK","_number":9,"created":"2019-09-30 15:44:20.000000000","uploader":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"ref":"refs/changes/62/671262/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/62/671262/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/62/671262/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/62/671262/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/62/671262/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/62/671262/9"}}},"commit":{"parents":[{"commit":"258a3772b91fc7c98aeaf6bb904179ff49506585","subject":"WIP: allow external ansible role for 3rd party job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/258a3772b91fc7c98aeaf6bb904179ff49506585"}]}],"author":{"name":"John Dennis","email":"jdennis@redhat.com","date":"2019-04-08 19:48:12.000000000","tz":-240},"committer":{"name":"Raildo Mascena","email":"rmascena@redhat.com","date":"2019-09-30 15:44:01.000000000","tz":-180},"subject":"Add keycloak-setup role","message":"Add keycloak-setup role\n\nkeycloak-setup is a new role used to install the Keycloak IdP on the\nsupplemental node. It utilizes the nkinder.keycloak Ansible Galaxy\nrole to perform the installation. This role depends on the existing\nfreeipa-setup role because it utilizes FreeIPA to obtain the necessary\nTLS cert and key, this is checked via the enable_tls_everywhere\nboolean. The Keycloak IdP is necessary to support testing Keystone\nfederated authentication which is enabled via the new\nenable_federation boolean.\n\nPartial-Bug: 1836415\n\nChange-Id: I4ac636572f537b98213745dd0a68ba89b4a1d74b\nSigned-off-by: John Dennis \u003cjdennis@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/a68962a6a86f40e386ad523797e4c6f91593534b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/a68962a6a86f40e386ad523797e4c6f91593534b"}]},"branch":"refs/heads/master"},"ecee6b7be8c479227917b4d0c1819ca9c55bf519":{"kind":"REWORK","_number":10,"created":"2019-10-02 18:31:13.000000000","uploader":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"ref":"refs/changes/62/671262/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/62/671262/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/62/671262/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/62/671262/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/62/671262/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/62/671262/10"}}},"commit":{"parents":[{"commit":"258a3772b91fc7c98aeaf6bb904179ff49506585","subject":"WIP: allow external ansible role for 3rd party job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/258a3772b91fc7c98aeaf6bb904179ff49506585"}]}],"author":{"name":"John Dennis","email":"jdennis@redhat.com","date":"2019-04-08 19:48:12.000000000","tz":-240},"committer":{"name":"Raildo Mascena","email":"rmascena@redhat.com","date":"2019-10-02 18:30:59.000000000","tz":-180},"subject":"Add keycloak-setup role","message":"Add keycloak-setup role\n\nkeycloak-setup is a new role used to install the Keycloak IdP on the\nsupplemental node. It utilizes the nkinder.keycloak Ansible Galaxy\nrole to perform the installation. This role depends on the existing\nfreeipa-setup role because it utilizes FreeIPA to obtain the necessary\nTLS cert and key, this is checked via the enable_tls_everywhere\nboolean. The Keycloak IdP is necessary to support testing Keystone\nfederated authentication which is enabled via the new\nenable_federation boolean.\n\nPartial-Bug: 1836415\n\nChange-Id: I4ac636572f537b98213745dd0a68ba89b4a1d74b\nSigned-off-by: John Dennis \u003cjdennis@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/ecee6b7be8c479227917b4d0c1819ca9c55bf519"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/ecee6b7be8c479227917b4d0c1819ca9c55bf519"}]},"branch":"refs/heads/master"},"b39013fe589e3bade9c01651033cdeede3cc1ab7":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2019-10-07 15:57:53.000000000","uploader":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"ref":"refs/changes/62/671262/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/62/671262/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/62/671262/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/62/671262/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/62/671262/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/62/671262/11"}}},"commit":{"parents":[{"commit":"6e1fedc10dd7bd4d8470f87efae4cfdc3cd98f5e","subject":"allow external ansible role for 3rd party job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/6e1fedc10dd7bd4d8470f87efae4cfdc3cd98f5e"}]}],"author":{"name":"John Dennis","email":"jdennis@redhat.com","date":"2019-04-08 19:48:12.000000000","tz":-240},"committer":{"name":"Raildo Mascena","email":"rmascena@redhat.com","date":"2019-10-07 15:57:53.000000000","tz":0},"subject":"Add keycloak-setup role","message":"Add keycloak-setup role\n\nkeycloak-setup is a new role used to install the Keycloak IdP on the\nsupplemental node. It utilizes the nkinder.keycloak Ansible Galaxy\nrole to perform the installation. This role depends on the existing\nfreeipa-setup role because it utilizes FreeIPA to obtain the necessary\nTLS cert and key, this is checked via the enable_tls_everywhere\nboolean. The Keycloak IdP is necessary to support testing Keystone\nfederated authentication which is enabled via the new\nenable_federation boolean.\n\nPartial-Bug: 1836415\n\nChange-Id: I4ac636572f537b98213745dd0a68ba89b4a1d74b\nSigned-off-by: John Dennis \u003cjdennis@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/b39013fe589e3bade9c01651033cdeede3cc1ab7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/b39013fe589e3bade9c01651033cdeede3cc1ab7"}]},"branch":"refs/heads/master"},"dda25dc46cf9d34a8d27001e7fa30651b9c316f6":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2020-01-21 11:51:59.000000000","uploader":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"ref":"refs/changes/62/671262/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/62/671262/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/62/671262/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/62/671262/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/62/671262/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/62/671262/12"}}},"commit":{"parents":[{"commit":"d23136331b9a5a51ab76224ac391dd8f4dfa2499","subject":"Merge \"Fix a typo for error message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/d23136331b9a5a51ab76224ac391dd8f4dfa2499"}]}],"author":{"name":"John Dennis","email":"jdennis@redhat.com","date":"2019-04-08 19:48:12.000000000","tz":-240},"committer":{"name":"Raildo Mascena","email":"rmascena@redhat.com","date":"2020-01-21 11:51:59.000000000","tz":0},"subject":"Add keycloak-setup role","message":"Add keycloak-setup role\n\nkeycloak-setup is a new role used to install the Keycloak IdP on the\nsupplemental node. It utilizes the nkinder.keycloak Ansible Galaxy\nrole to perform the installation. This role depends on the existing\nfreeipa-setup role because it utilizes FreeIPA to obtain the necessary\nTLS cert and key, this is checked via the enable_tls_everywhere\nboolean. The Keycloak IdP is necessary to support testing Keystone\nfederated authentication which is enabled via the new\nenable_federation boolean.\n\nPartial-Bug: 1836415\n\nChange-Id: I4ac636572f537b98213745dd0a68ba89b4a1d74b\nSigned-off-by: John Dennis \u003cjdennis@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/dda25dc46cf9d34a8d27001e7fa30651b9c316f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/dda25dc46cf9d34a8d27001e7fa30651b9c316f6"}]},"branch":"refs/heads/master"},"8b3dfb4a68ce504d7c627f740859b513d27fbc6d":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2020-05-22 16:13:00.000000000","uploader":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"ref":"refs/changes/62/671262/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/62/671262/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/62/671262/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/62/671262/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/62/671262/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/62/671262/13"}}},"commit":{"parents":[{"commit":"5a7f69b1a3f2501c74c6530c4f4d2bb5a60e3d43","subject":"[skiplist] NetworkSecGroupTest timeout (train)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/5a7f69b1a3f2501c74c6530c4f4d2bb5a60e3d43"}]}],"author":{"name":"John Dennis","email":"jdennis@redhat.com","date":"2019-04-08 19:48:12.000000000","tz":-240},"committer":{"name":"Raildo Mascena","email":"rmascena@redhat.com","date":"2020-05-22 16:13:00.000000000","tz":0},"subject":"Add keycloak-setup role","message":"Add keycloak-setup role\n\nkeycloak-setup is a new role used to install the Keycloak IdP on the\nsupplemental node. It utilizes the nkinder.keycloak Ansible Galaxy\nrole to perform the installation. This role depends on the existing\nfreeipa-setup role because it utilizes FreeIPA to obtain the necessary\nTLS cert and key, this is checked via the enable_tls_everywhere\nboolean. The Keycloak IdP is necessary to support testing Keystone\nfederated authentication which is enabled via the new\nenable_federation boolean.\n\nPartial-Bug: 1836415\n\nChange-Id: I4ac636572f537b98213745dd0a68ba89b4a1d74b\nSigned-off-by: John Dennis \u003cjdennis@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/8b3dfb4a68ce504d7c627f740859b513d27fbc6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/8b3dfb4a68ce504d7c627f740859b513d27fbc6d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
