)]}'
{"id":"openstack%2Ftripleo-quickstart-extras~675201","triplet_id":"openstack%2Ftripleo-quickstart-extras~master~I41859f93980dd8cd723ea4ad0f4e6592a78d0082","project":"openstack/tripleo-quickstart-extras","branch":"master","topic":"bp/undercloud-minion","hashtags":[],"change_id":"I41859f93980dd8cd723ea4ad0f4e6592a78d0082","subject":"Add basic minion validation","status":"MERGED","created":"2019-08-07 21:22:19.000000000","updated":"2019-08-20 05:17:09.000000000","submitted":"2019-08-20 05:17:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"675201-1566278229709-9a06ee28","meta_rev_id":"022571e07e7d762c037a88669c376630a5ab6b86","_number":675201,"virtual_id_number":675201,"owner":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"date":"2019-08-15 14:02:25.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-08-20 05:17:09.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-15 16:30:52.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-15 14:02:49.000000000","_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-15 15:28:38.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-08-20 00:55:23.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-08 17:58:26.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2019-08-12 14:57:51.000000000","updated_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"reviewer":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"state":"REVIEWER"},{"updated":"2019-08-15 14:02:49.000000000","updated_by":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-15 15:28:38.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-08-15 16:30:52.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-20 00:55:23.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-08-20 05:17:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0b6d09957a349f6cac2904d66fe6afee72beba13","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-07 21:22:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d073c88beeff9228b4a12a1306daa923b024a8ce","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-08-07 21:22:36.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/3739/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"81008ac1f92e4a1a4645551aa2f6caea94435421","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-07 21:31:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"009d9d775ac4a0364ae4e097531b92eed6aee85f","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-08-07 21:31:41.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/3740/ : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"601c188e7253336358a71f0446d667ba999df2c5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-08 00:43:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/01/675201/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/5833fd1/ : SUCCESS in 2h 37m 16s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/01/675201/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/162ba5b/ : SUCCESS in 2h 31m 42s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/01/675201/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/56c94c3/ : FAILURE in 3h 09m 12s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/01/675201/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/147530c/ : SUCCESS in 2h 13m 25s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/01/675201/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/aef7f0a/ : SUCCESS in 2h 38m 07s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/01/675201/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/b3193f5/ : FAILURE in 2h 40m 26s (non-voting)\n- tripleo-ci-rhel-8-standalone-rdo http://logs.rdoproject.org/01/675201/2/openstack-check/tripleo-ci-rhel-8-standalone-rdo/f0d3fec/ : FAILURE in 13m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7cf57c0e0895683387af5ca16677552d7672c6e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 00:57:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/01/675201/2/check/tripleo-ci-centos-7-undercloud-containers/41a0e26/ : SUCCESS in 1h 29m 38s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/01/675201/2/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/29cd87d/ : FAILURE in 1h 48m 29s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/01/675201/2/check/tripleo-ci-centos-7-containers-multinode/090cbd7/ : SUCCESS in 2h 38m 47s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://logs.opendev.org/01/675201/2/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/5ba6b10/ : SUCCESS in 2h 31m 59s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://logs.opendev.org/01/675201/2/check/tripleo-ci-centos-7-scenario012-multinode-oooq-container/93080a6/ : TIMED_OUT in 3h 23m 41s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/01/675201/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/211574c/ : SUCCESS in 2h 11m 46s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/01/675201/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/3902eba/ : SUCCESS in 2h 40m 27s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/01/675201/2/check/tripleo-ci-centos-7-standalone/15a7cec/ : SUCCESS in 1h 47m 03s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/01/675201/2/check/tripleo-ci-fedora-28-standalone/a1b35f5/ : SUCCESS in 1h 33m 45s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/01/675201/2/check/tripleo-ci-centos-7-standalone-upgrade/3971020/ : SUCCESS in 3h 02m 30s (non-voting)\n- tripleo-ci-centos-7-standalone-os-tempest https://logs.opendev.org/01/675201/2/check/tripleo-ci-centos-7-standalone-os-tempest/93c0ca2/ : SUCCESS in 1h 10m 35s\n- tripleo-ci-centos-7-containers-multinode-stein https://logs.opendev.org/01/675201/2/check/tripleo-ci-centos-7-containers-multinode-stein/f87ed96/ : SUCCESS in 2h 16m 24s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky https://logs.opendev.org/01/675201/2/check/tripleo-ci-centos-7-containers-multinode-rocky/1bf87b6/ : SUCCESS in 2h 37m 05s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens https://logs.opendev.org/01/675201/2/check/tripleo-ci-centos-7-containers-multinode-queens/c3aad91/ : SUCCESS in 2h 18m 01s\n- tripleo-ci-centos-7-containers-multinode-pike https://logs.opendev.org/01/675201/2/check/tripleo-ci-centos-7-containers-multinode-pike/21740c6/ : SUCCESS in 2h 18m 41s\n- openstack-tox-linters https://logs.opendev.org/01/675201/2/check/openstack-tox-linters/a61b3de/ : FAILURE in 4m 30s","accounts_in_message":[],"_revision_number":2},{"id":"c9013f6023c73050f31a1b2ff2c654ecf00b156e","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-08 14:08:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ea006d8b9f0c876791a7588b2f1261e870d767bf","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-08 14:08:27.000000000","message":"Patch Set 4: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"917c22c9656ce05aa7405c3869c6a3e9ff2ad0c2","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-08-08 14:08:28.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/3751/ : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"f518e97714d8180844ad698c17aee44cb7504ccd","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-08-08 14:08:45.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/3752/ : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"355cc115dbd121ae36ca340af71e6b1b3f976e9c","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2019-08-08 17:58:26.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"e23cfd9f1dca1d44df3187a06a53de2c7b7c590b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-08 18:43:34.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 : NODE_FAILURE in 0s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/01/675201/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/070005f/ : TIMED_OUT in 4h 30m 23s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/01/675201/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/0ea0c6e/ : TIMED_OUT in 3h 43m 43s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/01/675201/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/88f10b7/ : FAILURE in 1h 15m 16s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch finger://ze05.softwarefactory-project.io:7979/9ecf60940cb346c3a8d7e04d2472ff7e : RETRY_LIMIT in 7s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/01/675201/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/1c21816/ : SUCCESS in 2h 49m 09s (non-voting)\n- tripleo-ci-rhel-8-standalone-rdo http://logs.rdoproject.org/01/675201/4/openstack-check/tripleo-ci-rhel-8-standalone-rdo/ecccbe0/ : FAILURE in 16m 13s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"528dda8360aaf1ee231cdb69afa5eab267faa425","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 21:49:00.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/01/675201/4/check/tripleo-ci-centos-7-undercloud-containers/0693f9e/ : SUCCESS in 1h 37m 32s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/01/675201/4/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/0a8e545/ : FAILURE in 1h 56m 53s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/01/675201/4/check/tripleo-ci-centos-7-containers-multinode/93c76a8/ : SUCCESS in 2h 44m 57s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://logs.opendev.org/01/675201/4/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/8968d66/ : SUCCESS in 2h 34m 40s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://logs.opendev.org/01/675201/4/check/tripleo-ci-centos-7-scenario012-multinode-oooq-container/e922bd5/ : FAILURE in 2h 39m 49s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/01/675201/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/b5fbe43/ : SUCCESS in 2h 17m 52s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/01/675201/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/d9eaf7c/ : SUCCESS in 2h 40m 42s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/01/675201/4/check/tripleo-ci-centos-7-standalone/ebe4007/ : SUCCESS in 1h 48m 34s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/01/675201/4/check/tripleo-ci-fedora-28-standalone/d5b1753/ : SUCCESS in 1h 41m 32s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/01/675201/4/check/tripleo-ci-centos-7-standalone-upgrade/6fc846d/ : SUCCESS in 2h 09m 57s (non-voting)\n- tripleo-ci-centos-7-standalone-os-tempest https://logs.opendev.org/01/675201/4/check/tripleo-ci-centos-7-standalone-os-tempest/365841a/ : SUCCESS in 1h 10m 35s\n- tripleo-ci-centos-7-containers-multinode-stein https://logs.opendev.org/01/675201/4/check/tripleo-ci-centos-7-containers-multinode-stein/87f0844/ : SUCCESS in 2h 02m 53s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky https://logs.opendev.org/01/675201/4/check/tripleo-ci-centos-7-containers-multinode-rocky/33bfbc6/ : SUCCESS in 2h 14m 49s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens https://logs.opendev.org/01/675201/4/check/tripleo-ci-centos-7-containers-multinode-queens/53369ff/ : SUCCESS in 2h 12m 58s\n- tripleo-ci-centos-7-containers-multinode-pike https://logs.opendev.org/01/675201/4/check/tripleo-ci-centos-7-containers-multinode-pike/6e7bcd0/ : SUCCESS in 2h 19m 17s\n- openstack-tox-linters https://logs.opendev.org/01/675201/4/check/openstack-tox-linters/458503a/ : FAILURE in 5m 11s\n- tripleo-ci-centos-7-containers-undercloud-minion https://logs.opendev.org/01/675201/4/check/tripleo-ci-centos-7-containers-undercloud-minion/49a80fb/ : SUCCESS in 1h 31m 50s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4ff56044bc02aa20400cc84f0408507e03d6d589","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-12 14:52:50.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"33e024a0ad8e9bd1b6e28cf361ba5cf33e6a05ec","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-08-12 14:53:08.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/3772/ : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"86938d826ea729c067f5d85289545c7bba6fdfb5","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-12 14:57:51.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"b3e0af01318b50dedebe74775ac2cf21492d6f7e","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-12 15:16:04.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"a89c664a87ead1049ae59634cabeaee2a0d0be6a","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-08-12 15:16:19.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/3773/ : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"813c94da2fadcbf3121e0ac72281a2f717919e4b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-12 19:32:28.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 finger://ze07.softwarefactory-project.io:7979/c1cb15260bba451898e277fc873b3ddf : POST_FAILURE in 3h 26m 45s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/01/675201/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/9fa5102/ : FAILURE in 1h 37m 53s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/01/675201/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/a2badef/ : FAILURE in 3h 21m 10s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/01/675201/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/d9c3ad0/ : FAILURE in 1h 47m 32s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/01/675201/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/9d300a6/ : SUCCESS in 3h 29m 11s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/01/675201/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/fa72c5c/ : FAILURE in 2h 09m 52s (non-voting)\n- tripleo-ci-rhel-8-standalone-rdo http://logs.rdoproject.org/01/675201/6/openstack-check/tripleo-ci-rhel-8-standalone-rdo/54bda39/ : FAILURE in 20m 09s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"754d9e96c1835ba11f995b48fedf26ce8b59a5cd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 22:07:03.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/02345cd4c57847f58234485f1a94e4a2 : SUCCESS in 1h 27m 53s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/1beaf89d3fa54f079290031381654b1f : FAILURE in 2h 35m 08s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/7da9b02ad17348c699cb7ef7a3a5aedb : SUCCESS in 2h 17m 35s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/6b9359a2be094c59af742f352be36ac3 : SUCCESS in 2h 16m 04s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/08e762aa6b744c01a7e5bb56c4bd8b5d : FAILURE in 2h 28m 15s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/9f4bebab5ca044408809f768b6b7d459 : SUCCESS in 2h 01m 43s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/5c4e66eb5ea04f7887222a2e4eea0564 : SUCCESS in 2h 21m 47s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/23625c08b1d54664809f7f8ed7d2f138 : SUCCESS in 1h 44m 52s\n- tripleo-ci-fedora-28-standalone https://zuul.opendev.org/t/openstack/build/08e6d6808617498bb92307b4132cbfd1 : SUCCESS in 1h 34m 46s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/60a1144ad29b4fd7a300617da823615b : SUCCESS in 1h 57m 58s (non-voting)\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/c1accccd29624503941b33f55d19a36e : SUCCESS in 1h 06m 34s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/33683c45ff2344c3baf3fab1e8dee273 : SUCCESS in 1h 57m 46s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/b8c2192e3e2d4b999f170b8d27021d1b : SUCCESS in 2h 13m 02s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/3425926cba334d1e8cb7d5ac23ca09f8 : SUCCESS in 1h 55m 09s\n- tripleo-ci-centos-7-containers-multinode-pike https://zuul.opendev.org/t/openstack/build/fcb33989be334b53baed709205f3ce2a : SUCCESS in 2h 09m 49s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3183f96cc0d5443db314984823f9c00c : SUCCESS in 3m 57s\n- tripleo-ci-centos-7-containers-undercloud-minion https://zuul.opendev.org/t/openstack/build/b3d1d7aa53c44796ae30c111c8790f3b : SUCCESS in 1h 37m 05s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ad9bf795db02fa4c955c1155e16c928e418a0c75","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-12 22:41:26.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"89e74ae5aae23ba43758ee16e9f3fbdb17cb1239","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-08-12 22:41:40.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/3775/ : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"84fefba80bcfdf072e21ce1859563723602de708","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-13 01:20:51.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/01/675201/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/874ee01/ : FAILURE in 1h 33m 25s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/01/675201/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/3763afc/ : FAILURE in 2h 32m 47s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/01/675201/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/5471fd7/ : SUCCESS in 2h 14m 37s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/01/675201/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/fee0531/ : SUCCESS in 2h 17m 39s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/01/675201/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/d5d0c47/ : FAILURE in 1h 40m 57s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/01/675201/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/308636a/ : SUCCESS in 2h 37m 07s (non-voting)\n- tripleo-ci-rhel-8-standalone-rdo http://logs.rdoproject.org/01/675201/7/openstack-check/tripleo-ci-rhel-8-standalone-rdo/1ba41bc/ : FAILURE in 15m 48s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"64faa49b4e4ad97296342971f40a080b18c98794","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 01:40:25.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/04fbc1e0c6e14991927b63122d9c2968 : SUCCESS in 2h 01m 21s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/4489e52865bc4cba9ff50cd6a4c9561b : FAILURE in 1h 59m 11s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/20e3d173d44f4e8a989c0ee728af2436 : SUCCESS in 2h 16m 01s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/1157196ffef54e2ab4ad52c2b024976a : SUCCESS in 2h 57m 27s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/64cc4eb05df74550853612a29734460a : FAILURE in 13m 20s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/a067af8bc71a4b18b322375cb39db9a8 : SUCCESS in 2h 10m 43s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/383cc33d58f140b68c5b31511d1bfbe0 : SUCCESS in 2h 21m 57s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/c7ac3b1f7f7f475d8140976f996c582d : SUCCESS in 1h 45m 50s\n- tripleo-ci-fedora-28-standalone https://zuul.opendev.org/t/openstack/build/6d03e8bcaedc4a2c8c0bacb1c1359f1d : FAILURE in 1h 37m 07s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/98931b27f9a64f5ab347640b42dc0703 : SUCCESS in 2h 05m 28s (non-voting)\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/747dc7836a364476a368a4a43d2c0d05 : SUCCESS in 1h 07m 10s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/136b9f201fa5430d9fe02b9d65e89049 : SUCCESS in 2h 00m 29s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/4a2407e19fe848fbbfc9d922f7db7055 : SUCCESS in 2h 17m 24s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/f40ef1fce7824dcf97cf56aa00283d67 : SUCCESS in 2h 06m 37s\n- tripleo-ci-centos-7-containers-multinode-pike https://zuul.opendev.org/t/openstack/build/a2a28ba1bda24e99abeedad320ef4dd0 : SUCCESS in 2h 24m 35s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/986563a01e9348c9b0981847b7744899 : SUCCESS in 3m 50s\n- tripleo-ci-centos-7-containers-undercloud-minion https://zuul.opendev.org/t/openstack/build/57d9a0ea9ce745908e9826368a3fc2ce : FAILURE in 1h 18m 51s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"308832866ebbbcc9d667c7ac5cfb928d885c3e69","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-13 19:43:39.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"c2ab9397b405aa6ebe672a404aad1d0e4300c52d","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-08-13 19:43:56.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/3783/ : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"4435621866587925bd2a424cae4a4ab0cf75773e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-13 22:48:55.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/01/675201/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/5939a78/ : FAILURE in 2h 49m 55s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/01/675201/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/0dc5497/ : SUCCESS in 3h 03m 31s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/01/675201/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/f80a5ef/ : SUCCESS in 2h 18m 24s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/01/675201/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/a197101/ : SUCCESS in 2h 16m 23s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/01/675201/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/cf6b4a2/ : SUCCESS in 2h 38m 58s","accounts_in_message":[],"_revision_number":8},{"id":"3fd60fc3a86b103f543660e0310abb9338026fae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 22:59:15.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/8085b1517d6a4155ad420ed73e2bbb0f : SUCCESS in 1h 28m 44s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/8d16a6677b584961b88bd04a9259175c : FAILURE in 2h 06m 16s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/d3c0f31c58b6479c9f6d0a00b79f9b88 : SUCCESS in 2h 29m 36s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/03af8e7822e74a8e81923952c94c6352 : SUCCESS in 2h 29m 19s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/36f9e861a0ea4e61b5bc7c2ddcab4117 : FAILURE in 2h 26m 17s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/c2a48b13b4a14be8b851909b54a89f2f : SUCCESS in 2h 24m 37s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/3339e46e6ca94212968a7693a4c93e58 : SUCCESS in 2h 47m 20s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/fc1c65078660437fad57a25beb02a72b : SUCCESS in 2h 49m 11s\n- tripleo-ci-fedora-28-standalone https://zuul.opendev.org/t/openstack/build/b8b6bdbfa23941dab5fc87c458017b90 : SUCCESS in 1h 49m 46s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/c2bb981f1ced48c6a06873cc5aefd12f : SUCCESS in 1h 56m 50s (non-voting)\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/b6c214b88fd9443dbef80886a7555b3a : SUCCESS in 1h 10m 00s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/22b9ab9f59e24e8f91f9389b7a9fcb1f : SUCCESS in 2h 10m 15s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/356d9bf2ab484e05adfe6cc543b41c69 : SUCCESS in 2h 20m 52s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/9a7666959b584c33b6e9956af356ceda : SUCCESS in 2h 03m 23s\n- tripleo-ci-centos-7-containers-multinode-pike https://zuul.opendev.org/t/openstack/build/38bf44d53df040cb8cac0c98842d957e : FAILURE in 14m 34s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6827226103f04aaa8970d47f5a899ffc : SUCCESS in 3m 58s\n- tripleo-ci-centos-7-containers-undercloud-minion https://zuul.opendev.org/t/openstack/build/658955b26f4743f7acdf9f6059c48ee6 : SUCCESS in 1h 34m 11s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"53582d5e88145879e80faa0f79c086a85e2b23af","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-15 14:02:25.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"f50644b4a3c4137b2bb507d7ef2e1010d709ad42","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-08-15 14:02:49.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/3793/ : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"9253362a1152174018989216b8786f55befafa73","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-08-15 15:28:38.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"914f74ef3eba6595edc6e764b23b2231f6a5ae02","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-15 16:30:52.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/01/675201/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/d44bd62/ : SUCCESS in 2h 23m 45s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/01/675201/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/842015b/ : SUCCESS in 2h 24m 17s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/01/675201/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/9c25a5d/ : SUCCESS in 2h 19m 08s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/01/675201/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/f060013/ : SUCCESS in 2h 11m 44s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/01/675201/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/7688c19/ : SUCCESS in 2h 26m 52s","accounts_in_message":[],"_revision_number":8},{"id":"4fad5920a3729b733cea2ebea4dd7039283fbcce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 16:50:58.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/249533d2a5b249f997fbe3209c85f6c8 : SUCCESS in 1h 34m 28s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/a49b7122cd2244b89b82f9969f30cccd : FAILURE in 1h 59m 48s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/7d5acc7f77dd4ed38115b49e2b7f02bf : SUCCESS in 2h 22m 36s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/88009d01fa5f454d9cec0d38ab4ba069 : SUCCESS in 2h 12m 08s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/3ba4fdb5c3d94ac39bea620c2429c345 : FAILURE in 2h 16m 31s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/349ee3269ea24d6f88c52b8f457149b3 : SUCCESS in 2h 02m 11s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/b1c41b94405b4ed684017dbe5e963018 : SUCCESS in 2h 22m 49s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/ae3d3523299444c3a61a7e8ec4f2219b : SUCCESS in 1h 53m 57s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/d9158e6209a34dec942227d91b01ee0d : SUCCESS in 1h 59m 42s (non-voting)\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/857fcfb843dc440690a7d05fbbb39b15 : SUCCESS in 1h 27m 51s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/dfd2e8ed51994480968c627d351e78a3 : SUCCESS in 2h 01m 59s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/56ec4b4db21d468c9352eb78a5bef009 : SUCCESS in 2h 15m 28s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/b1579ca380ba4dd0a45a50e7b819afee : SUCCESS in 2h 10m 20s\n- tripleo-ci-centos-7-containers-multinode-pike https://zuul.opendev.org/t/openstack/build/648f7d3ea10c488c83e193e53e5991f5 : SUCCESS in 2h 16m 44s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e4eb8b160341412e87ae2da7fadbbe08 : SUCCESS in 4m 06s\n- tripleo-ci-centos-7-containers-undercloud-minion https://zuul.opendev.org/t/openstack/build/e8f347e302be4f55a69269a318c65f34 : SUCCESS in 1h 29m 55s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"edb815b9b4245ed8fa7b6d5e26e2cb0c37849e40","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-20 00:55:23.000000000","message":"Patch Set 8: Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"8af601b2022dbe2506b3caf755138142d761afee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 00:55:39.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"be166adfb6f629122b27b12d4596c295df11c6a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 05:17:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"8b20479df2150a555515b11c73ac09c03b4cf6bb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 05:17:09.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/e217b40fb510486dbc80469404a13908 : SUCCESS in 1h 33m 19s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/a88c3d76c3aa4807b2603fa5f4be6078 : SUCCESS in 2h 28m 42s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/7b79ab53c5d34d22be296d34234de72d : SUCCESS in 1h 52m 39s\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/8a29644cdb1d4acba16c5b84d70e485d : SUCCESS in 1h 12m 27s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/7cc3ca63269843b5a48183fbd67257e8 : SUCCESS in 2h 21m 38s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/0c822fbe55b84df8b4adf95be2dea922 : SUCCESS in 2h 09m 30s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/e58fb3db272f4557a84028e78ac9939e : SUCCESS in 2h 25m 31s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/1e231b1ad4ed46fda6e1f9fd17efc4b8 : SUCCESS in 2h 03m 30s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7cdd3c188bda435fb0e41b521df73476 : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"766e75e2a84781b405df6f0288332316c13265f2","revisions":{"2d31779f7601d94f21f40840d44bdd77b1875b91":{"kind":"REWORK","_number":1,"created":"2019-08-07 21:22:19.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/01/675201/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/01/675201/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/01/675201/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/01/675201/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/01/675201/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/01/675201/1"}}},"commit":{"parents":[{"commit":"8bb2dc094c17dbdb4c3dfbd6ce05bfb658d84d3c","subject":"Allow building packages for RHEL8 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/8bb2dc094c17dbdb4c3dfbd6ce05bfb658d84d3c"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-08-07 21:21:04.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-08-07 21:21:04.000000000","tz":-360},"subject":"Add basic minion validation","message":"Add basic minion validation\n\nThe minion doesn\u0027t actually have any api services so we validate it by\nchecking the service status on the undercloud. This change adds a role\nto validate the heat engine and ironic conductor status on the\nundercloud includes additional hosts (undercloud + minion) and there are\nno failed services.\n\nChange-Id: I41859f93980dd8cd723ea4ad0f4e6592a78d0082\nRelated-Blueprint: undercloud-minion\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/2d31779f7601d94f21f40840d44bdd77b1875b91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/2d31779f7601d94f21f40840d44bdd77b1875b91"}]},"branch":"refs/heads/master"},"891ba110e0ebd1927db6ed4a415ab44d840e101c":{"kind":"REWORK","_number":2,"created":"2019-08-07 21:31:26.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/01/675201/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/01/675201/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/01/675201/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/01/675201/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/01/675201/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/01/675201/2"}}},"commit":{"parents":[{"commit":"8bb2dc094c17dbdb4c3dfbd6ce05bfb658d84d3c","subject":"Allow building packages for RHEL8 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/8bb2dc094c17dbdb4c3dfbd6ce05bfb658d84d3c"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-08-07 21:21:04.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-08-07 21:31:07.000000000","tz":-360},"subject":"Add basic minion validation","message":"Add basic minion validation\n\nThe minion doesn\u0027t actually have any api services so we validate it by\nchecking the service status on the undercloud. This change adds a role\nto validate the heat engine and ironic conductor status on the\nundercloud includes additional hosts (undercloud + minion) and there are\nno failed services.\n\nChange-Id: I41859f93980dd8cd723ea4ad0f4e6592a78d0082\nRelated-Blueprint: undercloud-minion\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/891ba110e0ebd1927db6ed4a415ab44d840e101c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/891ba110e0ebd1927db6ed4a415ab44d840e101c"}]},"branch":"refs/heads/master"},"7196802e5470c4152493143be95fe4317d910394":{"kind":"REWORK","_number":3,"created":"2019-08-08 14:08:14.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/01/675201/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/01/675201/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/01/675201/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/01/675201/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/01/675201/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/01/675201/3"}}},"commit":{"parents":[{"commit":"8bb2dc094c17dbdb4c3dfbd6ce05bfb658d84d3c","subject":"Allow building packages for RHEL8 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/8bb2dc094c17dbdb4c3dfbd6ce05bfb658d84d3c"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-08-07 21:21:04.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-08-08 14:07:51.000000000","tz":-360},"subject":"Add basic minion validation","message":"Add basic minion validation\n\nThe minion doesn\u0027t actually have any api services so we validate it by\nchecking the service status on the undercloud. This change adds a role\nto validate the heat engine and ironic conductor status on the\nundercloud includes additional hosts (undercloud + minion) and there are\nno failed services.\n\nChange-Id: I41859f93980dd8cd723ea4ad0f4e6592a78d0082\nRelated-Blueprint: undercloud-minion\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7196802e5470c4152493143be95fe4317d910394"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7196802e5470c4152493143be95fe4317d910394"}]},"branch":"refs/heads/master"},"adc85197a889ece4db5cb6c66c45c95a1bcf0a3f":{"kind":"REWORK","_number":4,"created":"2019-08-08 14:08:27.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/01/675201/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/01/675201/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/01/675201/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/01/675201/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/01/675201/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/01/675201/4"}}},"commit":{"parents":[{"commit":"e0a51fca9357222a77b6c60330c61e90a90f0d15","subject":"Add undercloud minion job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/e0a51fca9357222a77b6c60330c61e90a90f0d15"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-08-07 21:21:04.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-08-08 14:08:27.000000000","tz":0},"subject":"Add basic minion validation","message":"Add basic minion validation\n\nThe minion doesn\u0027t actually have any api services so we validate it by\nchecking the service status on the undercloud. This change adds a role\nto validate the heat engine and ironic conductor status on the\nundercloud includes additional hosts (undercloud + minion) and there are\nno failed services.\n\nChange-Id: I41859f93980dd8cd723ea4ad0f4e6592a78d0082\nRelated-Blueprint: undercloud-minion\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/adc85197a889ece4db5cb6c66c45c95a1bcf0a3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/adc85197a889ece4db5cb6c66c45c95a1bcf0a3f"}]},"branch":"refs/heads/master"},"b5be0fe1630a9f1c8c2c9a339728d700c5664bab":{"kind":"REWORK","_number":5,"created":"2019-08-12 14:52:50.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/01/675201/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/01/675201/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/01/675201/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/01/675201/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/01/675201/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/01/675201/5"}}},"commit":{"parents":[{"commit":"e0a51fca9357222a77b6c60330c61e90a90f0d15","subject":"Add undercloud minion job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/e0a51fca9357222a77b6c60330c61e90a90f0d15"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-08-07 21:21:04.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-08-12 14:52:30.000000000","tz":-360},"subject":"Add basic minion validation","message":"Add basic minion validation\n\nThe minion doesn\u0027t actually have any api services so we validate it by\nchecking the service status on the undercloud. This change adds a role\nto validate the heat engine and ironic conductor status on the\nundercloud includes additional hosts (undercloud + minion) and there are\nno failed services.\n\nChange-Id: I41859f93980dd8cd723ea4ad0f4e6592a78d0082\nRelated-Blueprint: undercloud-minion\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/b5be0fe1630a9f1c8c2c9a339728d700c5664bab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/b5be0fe1630a9f1c8c2c9a339728d700c5664bab"}]},"branch":"refs/heads/master"},"36fb93d6956c0821dbf1d1fb3342baf1e4d72a50":{"kind":"REWORK","_number":6,"created":"2019-08-12 15:16:04.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/01/675201/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/01/675201/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/01/675201/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/01/675201/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/01/675201/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/01/675201/6"}}},"commit":{"parents":[{"commit":"e0a51fca9357222a77b6c60330c61e90a90f0d15","subject":"Add undercloud minion job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/e0a51fca9357222a77b6c60330c61e90a90f0d15"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-08-07 21:21:04.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-08-12 15:02:22.000000000","tz":-360},"subject":"Add basic minion validation","message":"Add basic minion validation\n\nThe minion doesn\u0027t actually have any api services so we validate it by\nchecking the service status on the undercloud. This change adds a role\nto validate the heat engine and ironic conductor status on the\nundercloud includes additional hosts (undercloud + minion) and there are\nno failed services.\n\nChange-Id: I41859f93980dd8cd723ea4ad0f4e6592a78d0082\nRelated-Blueprint: undercloud-minion\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/36fb93d6956c0821dbf1d1fb3342baf1e4d72a50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/36fb93d6956c0821dbf1d1fb3342baf1e4d72a50"}]},"branch":"refs/heads/master"},"ed657deb7c0eaa9e46d75ba16e3941bbc3a37e55":{"kind":"REWORK","_number":7,"created":"2019-08-12 22:41:26.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/01/675201/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/01/675201/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/01/675201/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/01/675201/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/01/675201/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/01/675201/7"}}},"commit":{"parents":[{"commit":"e0a51fca9357222a77b6c60330c61e90a90f0d15","subject":"Add undercloud minion job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/e0a51fca9357222a77b6c60330c61e90a90f0d15"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-08-07 21:21:04.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-08-12 22:41:12.000000000","tz":-360},"subject":"Add basic minion validation","message":"Add basic minion validation\n\nThe minion doesn\u0027t actually have any api services so we validate it by\nchecking the service status on the undercloud. This change adds a role\nto validate the heat engine and ironic conductor status on the\nundercloud includes additional hosts (undercloud + minion) and there are\nno failed services.\n\nChange-Id: I41859f93980dd8cd723ea4ad0f4e6592a78d0082\nRelated-Blueprint: undercloud-minion\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/ed657deb7c0eaa9e46d75ba16e3941bbc3a37e55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/ed657deb7c0eaa9e46d75ba16e3941bbc3a37e55"}]},"branch":"refs/heads/master"},"766e75e2a84781b405df6f0288332316c13265f2":{"kind":"REWORK","_number":8,"created":"2019-08-13 19:43:39.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/01/675201/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/01/675201/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/01/675201/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/01/675201/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/01/675201/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/01/675201/8"}}},"commit":{"parents":[{"commit":"e0a51fca9357222a77b6c60330c61e90a90f0d15","subject":"Add undercloud minion job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/e0a51fca9357222a77b6c60330c61e90a90f0d15"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-08-07 21:21:04.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-08-13 19:43:09.000000000","tz":-360},"subject":"Add basic minion validation","message":"Add basic minion validation\n\nThe minion doesn\u0027t actually have any api services so we validate it by\nchecking the service status on the undercloud. This change adds a role\nto validate the heat engine and ironic conductor status on the\nundercloud includes additional hosts (undercloud + minion) and there are\nno failed services.\n\nChange-Id: I41859f93980dd8cd723ea4ad0f4e6592a78d0082\nRelated-Blueprint: undercloud-minion\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/766e75e2a84781b405df6f0288332316c13265f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/766e75e2a84781b405df6f0288332316c13265f2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
