)]}'
{"id":"openstack%2Ftripleo-quickstart-extras~676196","triplet_id":"openstack%2Ftripleo-quickstart-extras~master~If0e190344fe8f1375d8b2c1f4aa8ccd54811f16d","project":"openstack/tripleo-quickstart-extras","branch":"master","topic":"restore_ignore_errors_tempest_results","hashtags":[],"change_id":"If0e190344fe8f1375d8b2c1f4aa8ccd54811f16d","subject":"Restore ignore_errors for tempest-results tasks","status":"ABANDONED","created":"2019-08-13 13:54:40.000000000","updated":"2021-11-09 13:23:45.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0f7570c9bb05572885c783bc68f5989f614fd11b","_number":676196,"virtual_id_number":676196,"owner":{"_account_id":1918,"name":"François Charlier","email":"francois.charlier@redhat.com","username":"francois-charlier"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"all":[{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":1,"date":"2019-09-15 11:01:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-09-15 10:40:21.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":1918,"name":"François Charlier","email":"francois.charlier@redhat.com","username":"francois-charlier"},{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},{"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"},{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"recommended":{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},"all":[{"value":1,"date":"2019-09-15 07:41:00.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"value":2,"date":"2019-09-15 07:42:57.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1918,"name":"François Charlier","email":"francois.charlier@redhat.com","username":"francois-charlier"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-15 07:41:00.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"},{"value":1,"date":"2019-09-15 07:41:00.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":1918,"name":"François Charlier","email":"francois.charlier@redhat.com","username":"francois-charlier"},{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},{"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"},{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1918,"name":"François Charlier","email":"francois.charlier@redhat.com","username":"francois-charlier"},{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"},{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-13 13:54:57.000000000","updated_by":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-25 09:45:26.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2019-09-04 14:09:51.000000000","updated_by":{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},"reviewer":{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},"state":"REVIEWER"},{"updated":"2019-09-09 09:56:54.000000000","updated_by":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"reviewer":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"state":"REVIEWER"},{"updated":"2019-09-13 06:00:41.000000000","updated_by":{"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"},"reviewer":{"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"},"state":"REVIEWER"},{"updated":"2019-09-15 07:42:57.000000000","updated_by":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"reviewer":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"state":"REVIEWER"},{"updated":"2019-09-15 10:40:21.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-15 11:01:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"651b10bb54927cd8c996126477d18cf50e031e4a","author":{"_account_id":1918,"name":"François Charlier","email":"francois.charlier@redhat.com","username":"francois-charlier"},"date":"2019-08-13 13:54:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cd68fe04e6032cdc989c9ec6c78acbbb44da63c6","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-08-13 13:54:57.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/3780/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"51473fc0a55e73a2b6ee2aa04435de25f21f4487","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-13 17:15:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/96/676196/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/c4a5690/ : SUCCESS in 3h 03m 29s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/96/676196/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/d739168/ : FAILURE in 2h 30m 54s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/96/676196/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/b5044b6/ : SUCCESS in 2h 40m 21s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/96/676196/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/59e44ef/ : SUCCESS in 2h 28m 40s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/96/676196/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/f263c27/ : SUCCESS in 2h 50m 18s","accounts_in_message":[],"_revision_number":1},{"id":"884053939b91eff8bbaf4022624753f7dc554996","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 17:57:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/bde3a35939d8498d9b1cf9e802ea6e02 : SUCCESS in 1h 34m 54s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/e6df8cc5b2074c8baa64b75ac6c553ef : FAILURE in 1h 43m 43s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/60b1002d3bf34b2684414cae74d93e36 : SUCCESS in 2h 26m 51s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/2d7f8f2c360646c19094df0344c6d1a6 : SUCCESS in 2h 38m 30s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/b00476cd11ca4a45a2fc2105f07e55ba : FAILURE in 2h 31m 31s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/3bd4a3956fd54f4a941c9f9f513d1e65 : SUCCESS in 2h 26m 31s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/fa0b09f7d03a47b4996d54a2dfbd1ec8 : SUCCESS in 2h 30m 25s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/9951f778019841218d3bddefae10a531 : SUCCESS in 2h 02m 04s\n- tripleo-ci-fedora-28-standalone https://zuul.opendev.org/t/openstack/build/925dbdbbc7e64f54baf1edc402a5a0e3 : FAILURE in 49m 06s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/93d83c5d31444de1aba042ed02f8c235 : SUCCESS in 2h 04m 57s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/728956a0fb784380863360ca072d19bf : SUCCESS in 1h 59m 24s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/8b216e689bcb4864a8b17a0d20909923 : SUCCESS in 1h 38m 22s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/45d57132e3304d469c8c7792f256645f : SUCCESS in 1h 15m 21s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/f844ee4768254753a19e471a65f4ce8d : SUCCESS in 1h 55m 25s\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/d281f5307cbc4e408f0fef06fc814b8b : SUCCESS in 1h 06m 08s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/6bcef287fc1d4cde8422b9c2ebdeba68 : SUCCESS in 2h 07m 52s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/622a571f96e1491fbd726e83f472608c : SUCCESS in 2h 24m 26s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/baff148ad7d24c3885c8980f53303e75 : SUCCESS in 2h 14m 29s\n- tripleo-ci-centos-7-containers-multinode-pike https://zuul.opendev.org/t/openstack/build/d46def018dfe45b391f7467778358073 : SUCCESS in 2h 27m 27s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/69eb85c3f5864b81bc92ee07daca437f : SUCCESS in 1h 10m 01s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/3941c0ef5278447c8738eb2cb4e959d8 : SUCCESS in 1h 56m 42s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3958bfeb63f4471196cab93681901930 : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":1},{"id":"ffa2420671fb384261e59ab509b620d2751c098b","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-08-25 09:45:09.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"eab26a870e65d0a76cb0851b95e50d3a369c9871","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2019-08-26 09:15:08.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"bd143db59954e557173446010ea133596c216ab4","author":{"_account_id":1918,"name":"François Charlier","email":"francois.charlier@redhat.com","username":"francois-charlier"},"date":"2019-09-02 09:14:07.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"5d96f8527270d1cf35b757d8ece18bc3c2bc4749","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-02 09:36:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/96/676196/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/a52762e/ : FAILURE in 17m 16s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/96/676196/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/e501943/ : FAILURE in 18m 02s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/96/676196/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/51e0a1b/ : FAILURE in 20m 18s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/96/676196/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/e609d3c/ : FAILURE in 20m 51s","accounts_in_message":[],"_revision_number":2},{"id":"5d336d31ce07309ae30b0ede9567c0284dc10bf9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-02 10:17:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/f6b3c90fdd3047138272dd9e24406362 : FAILURE in 20m 15s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/58d65707f96649b4843c2c8fa38ba0df : FAILURE in 21m 27s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/9c4051d5939f48aab10806b4091a6445 : FAILURE in 15m 22s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/864227dfb3d54960a6824fa6fbaa84e8 : FAILURE in 23m 52s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/b81e7c5c869b4c1da4fe22fb9ca787da : FAILURE in 21m 17s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/d5b9d1306128441ebdc2f737e39656ee : FAILURE in 13m 58s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/2654f992ff3f42f58ca4959262173815 : FAILURE in 18m 16s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/9a444ddb11dc4a6d988e0b5d5cf628e9 : FAILURE in 14m 59s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/56704450621b46678670027cb80377c7 : FAILURE in 9m 46s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/99d02265be6e4fa3acad4755a74fd5d6 : FAILURE in 18m 16s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/da782fdf9a584c4882e90725881873a8 : FAILURE in 17m 47s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/bf67d06a0a1642aea0fc3fb649c99cfc : FAILURE in 16m 18s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/82eeaa0cc2464b0fa8cbb49ff128bee4 : FAILURE in 18m 38s\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/4e7622c93530483bba1808226b69af7a : FAILURE in 16m 28s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/3578295f49ad48bfa6d32d1c9688929c : FAILURE in 11m 47s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/24fa6d3479d0466795a2304c9b470ab6 : FAILURE in 13m 05s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/cda3de8cc70b48969322fe3397614a4c : FAILURE in 17m 26s\n- tripleo-ci-centos-7-containers-multinode-pike https://zuul.opendev.org/t/openstack/build/290d9c57c958439599222ddc793d7e01 : FAILURE in 12m 03s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/cd8d37ce611a476ebd9706c98c5a2457 : FAILURE in 14m 54s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/5358564d19f14a178e88470ca8c211eb : FAILURE in 14m 48s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7739e0df7bae4fea91580393f358b995 : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":2},{"id":"91b3d02a09de6a4a1bb1dfedb26925545c2a93c8","author":{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},"date":"2019-09-04 14:09:51.000000000","message":"Patch Set 2: Code-Review+1\n\nWhich job is affected by an error on validate_tempest ? Canit be moved to os_tempest instead ?","accounts_in_message":[],"_revision_number":2},{"id":"cbc8563c49f7f176868afeb85219e41446fc3775","author":{"_account_id":1918,"name":"François Charlier","email":"francois.charlier@redhat.com","username":"francois-charlier"},"date":"2019-09-04 15:11:20.000000000","message":"Patch Set 2:\n\nGabriele: we\u0027re affected by this issue in DCI.\n\nI don\u0027t know about os_tempest, where can I find mode info about this ?","accounts_in_message":[],"_revision_number":2},{"id":"9c939fd38b5b96adab8afbc803df3a509d3663ec","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2019-09-09 09:56:54.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"c6459f5e5b3b48d049ff3a204bef8805292fa379","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2019-09-09 09:58:28.000000000","message":"Patch Set 2:\n\nFrançois Charlier wrote:\n\n\u003e Gabriele: we\u0027re affected by this issue in DCI.\n\u003e\n\u003e I don\u0027t know about os_tempest, where can I find mode info about this\n\u003e ?\nYou can check os_tempest at https://github.com/openstack/openstack-ansible-os_tempest/ we are deprecating this. This doesn\u0027t affect our jobs because we are using os_tempest and I believe we don\u0027t need this anymore (for upstream jobs)","accounts_in_message":[],"_revision_number":2},{"id":"2f6eb5505613ddf3da126659ea54d337277fd0fe","author":{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},"date":"2019-09-09 12:06:27.000000000","message":"Patch Set 2:\n\nrecheck\n\nos_tempest is not supported in rocky, so we need to keep validate tempest maintained\n+2 after this passes tests","accounts_in_message":[],"_revision_number":2},{"id":"1ba6203308611914f9ef8789e37a1838de1b4374","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-09 15:09:16.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/96/676196/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/7671cc8/ : FAILURE in 1h 32m 00s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/96/676196/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/b3c08c3/ : FAILURE in 2h 33m 35s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/96/676196/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/4ad1ef0/ : FAILURE in 2h 51m 33s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/96/676196/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/f346243/ : SUCCESS in 2h 58m 22s","accounts_in_message":[],"_revision_number":2},{"id":"a01245cfd1aacd7abe0e734145c8fc19733b9618","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 15:15:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/e8e0d7e481d3472b80d3e112ff601316 : SUCCESS in 1h 55m 57s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/b44f0dbcc10a4c3ebacd5be4031823ae : FAILURE in 2h 10m 47s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/ca8f1dea190a461895bb059465cb2213 : SUCCESS in 2h 53m 40s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/62f802c280b64ecfb5d48606bcd3b57b : FAILURE in 1h 59m 27s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/bbfcfe74313343c2b49e6697a35104ec : FAILURE in 2h 22m 05s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/31950d739c184d1a99f43878a29fdb93 : SUCCESS in 2h 15m 04s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/a6a2f12c6cdd46bfb742c5eca8c9b29b : SUCCESS in 2h 23m 40s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/da4f29d399e54d9989e29f76280a8978 : SUCCESS in 1h 55m 06s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/a4f7b27e0d864b2ab91544045fa43c52 : SUCCESS in 2h 31m 40s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/b863d10432e042c495344e33022d8d34 : SUCCESS in 2h 18m 24s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/0dd0ca4c858940d1a08c04d4a848c135 : SUCCESS in 1h 48m 31s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/672d99e593684986a441a8994c6c2ba3 : SUCCESS in 1h 32m 54s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/5226955601f34dc58a0c81b53d1970d0 : SUCCESS in 1h 54m 25s\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/84c1690b7bb84438862b4b67be00586d : SUCCESS in 1h 21m 21s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/669a8ef56055498e8373bdb29c248f14 : SUCCESS in 2h 42m 25s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/4ed462d3b1cc486da99d58826394f038 : SUCCESS in 2h 24m 33s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/da127dd9e86d47239586f0467c20eadf : SUCCESS in 2h 15m 24s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/dadfedd274da41c081c66176a0e85027 : SUCCESS in 1h 27m 36s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/a4a3cf87b7104e1c9ec38e3dfd3d37d2 : FAILURE in 1h 40m 54s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/023958e74ff7405cb0e235f808db9305 : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":2},{"id":"d208a0a307aa74ef362d32e422b5b057ea235da6","author":{"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"},"date":"2019-09-13 06:00:41.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"18d4de9e3af8d3d8985b375766d3170b5f0293e4","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-09-15 07:41:00.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"d6b5b44622fb6888ad06ff0db261f4e0387712ac","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-09-15 07:42:45.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"e7d8820234e7f9fe6452c375305793a1f2a51664","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-09-15 07:42:57.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"bc76630d09c8842af0809dd4d75e2a75dcb398ab","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-15 10:40:21.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/96/676196/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/4b3bc58/ : FAILURE in 1h 45m 53s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/96/676196/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/a5bb2b1/ : FAILURE in 2h 11m 17s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/96/676196/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/9a676e7/ : SUCCESS in 2h 16m 30s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/96/676196/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/fc8524b/ : SUCCESS in 2h 58m 04s","accounts_in_message":[],"_revision_number":3},{"id":"365b56386f08e64d00afbaf8beba2703ff7a84be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 11:01:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/902eafc78b3e40718065f30d28cb3252 : SUCCESS in 1h 48m 15s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/4060621058fc41258f4476d443599243 : FAILURE in 2h 03m 54s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/0833ee72e1f04d2fa8fecbf641accddc : SUCCESS in 3h 10m 04s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/38674cd70d7a44e0b8d2e6418736af02 : SUCCESS in 2h 48m 53s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/b1659615004a4493addffbbfe26353d3 : FAILURE in 1h 58m 45s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/c3254b9e7c7646f8b5e1f883bc8b4339 : SUCCESS in 2h 33m 42s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/5261c4859e2a4db2bc56e56d0cfc0ed0 : SUCCESS in 3h 05m 41s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/b57b06dcc5c54c9faa89ee3850c0c769 : SUCCESS in 2h 00m 27s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/251cfcf08ab04b17a8668e7387758dcd : FAILURE in 2h 05m 17s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/8c984cf6e5ca48f5a932d294e6b6718f : SUCCESS in 2h 13m 08s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/b1466d98ab904de5b06fc18406c9896c : SUCCESS in 1h 53m 08s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/e3551af537bc48ceafda0503996d2a47 : SUCCESS in 1h 27m 59s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/fb7ef636422a488f9ed7b119fa5ea6db : SUCCESS in 2h 05m 51s\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/84829ef56a8349cda3476478a18cafd9 : SUCCESS in 1h 33m 09s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/f93c83cc7cc346e5818c4d86725f6cf3 : SUCCESS in 2h 33m 29s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/8e420be57e4944ac9696735fb5a53096 : SUCCESS in 2h 25m 44s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/1adcd522f8a1450baa656894fdca26b1 : SUCCESS in 2h 40m 08s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/27273fd4844a407aa1181ed618ac9ff4 : SUCCESS in 1h 37m 06s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/c62336e1322540e28871623445a6568f : SUCCESS in 2h 12m 27s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/db2d9888b0f74ba197d759097dc3472a : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":3},{"id":"0f7570c9bb05572885c783bc68f5989f614fd11b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2021-11-09 13:23:45.000000000","message":"Abandoned\n\nWhile positive reviews, 2 years old and unmerged. Abandoning for now, can be restored if needed.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"46a1501afcea12462d4971b4e14a3558645e8f34","revisions":{"5fcb0a7c76397f9f0f7c8c2992688617bc02205d":{"kind":"REWORK","_number":1,"created":"2019-08-13 13:54:40.000000000","uploader":{"_account_id":1918,"name":"François Charlier","email":"francois.charlier@redhat.com","username":"francois-charlier"},"ref":"refs/changes/96/676196/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/96/676196/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/96/676196/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/96/676196/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/96/676196/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/96/676196/1"}}},"commit":{"parents":[{"commit":"fb6c4cc1656b5e89a67e49fe9f0b0f19599309e8","subject":"Merge \"Remove VolumesSnapshotTestJSON from stein skip list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/fb6c4cc1656b5e89a67e49fe9f0b0f19599309e8"}]}],"author":{"name":"François Charlier","email":"francois.charlier@redhat.com","date":"2019-08-13 13:47:52.000000000","tz":120},"committer":{"name":"François Charlier","email":"francois.charlier@redhat.com","date":"2019-08-13 13:54:29.000000000","tz":120},"subject":"Restore ignore_errors for tempest-results tasks","message":"Restore ignore_errors for tempest-results tasks\n\nCommint cd7bb9bc8 removed (by mistake apparently) the \u0027ignore_errors:\ntrue\u0027 statement fro the tasks block generating the tempest results.\n\nChange-Id: If0e190344fe8f1375d8b2c1f4aa8ccd54811f16d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/5fcb0a7c76397f9f0f7c8c2992688617bc02205d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/5fcb0a7c76397f9f0f7c8c2992688617bc02205d"}]},"branch":"refs/heads/master"},"de368c6da9524a260c7c22d0b8b26026dd36fee3":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-09-02 09:14:07.000000000","uploader":{"_account_id":1918,"name":"François Charlier","email":"francois.charlier@redhat.com","username":"francois-charlier"},"ref":"refs/changes/96/676196/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/96/676196/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/96/676196/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/96/676196/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/96/676196/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/96/676196/2"}}},"commit":{"parents":[{"commit":"fb6c4cc1656b5e89a67e49fe9f0b0f19599309e8","subject":"Merge \"Remove VolumesSnapshotTestJSON from stein skip list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/fb6c4cc1656b5e89a67e49fe9f0b0f19599309e8"}]}],"author":{"name":"François Charlier","email":"francois.charlier@redhat.com","date":"2019-08-13 13:47:52.000000000","tz":120},"committer":{"name":"François Charlier","email":"francois.charlier@redhat.com","date":"2019-09-02 09:13:09.000000000","tz":120},"subject":"Restore ignore_errors for tempest-results tasks","message":"Restore ignore_errors for tempest-results tasks\n\nChange I9c7f97f3aa97c0ea061ba6aeb125501bf3a1adbf removed (by mistake\napparently) the \u0027ignore_errors: true\u0027 statement for the tasks block\ngenerating the tempest results.\n\nChange-Id: If0e190344fe8f1375d8b2c1f4aa8ccd54811f16d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/de368c6da9524a260c7c22d0b8b26026dd36fee3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/de368c6da9524a260c7c22d0b8b26026dd36fee3"}]},"branch":"refs/heads/master"},"46a1501afcea12462d4971b4e14a3558645e8f34":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-09-15 07:41:00.000000000","uploader":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"ref":"refs/changes/96/676196/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/96/676196/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/96/676196/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/96/676196/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/96/676196/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/96/676196/3"}}},"commit":{"parents":[{"commit":"b623e952a7e825e46d672a4a80b7318598367ebd","subject":"Merge \"Fix waiting OC VMs after restored from snapshot\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/b623e952a7e825e46d672a4a80b7318598367ebd"}]}],"author":{"name":"François Charlier","email":"francois.charlier@redhat.com","date":"2019-08-13 13:47:52.000000000","tz":120},"committer":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-09-15 07:41:00.000000000","tz":0},"subject":"Restore ignore_errors for tempest-results tasks","message":"Restore ignore_errors for tempest-results tasks\n\nChange I9c7f97f3aa97c0ea061ba6aeb125501bf3a1adbf removed (by mistake\napparently) the \u0027ignore_errors: true\u0027 statement for the tasks block\ngenerating the tempest results.\n\nChange-Id: If0e190344fe8f1375d8b2c1f4aa8ccd54811f16d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/46a1501afcea12462d4971b4e14a3558645e8f34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/46a1501afcea12462d4971b4e14a3558645e8f34"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
