)]}'
{"id":"openstack%2Ftripleo-quickstart-extras~676203","triplet_id":"openstack%2Ftripleo-quickstart-extras~master~I02670fe00f9d9d5ef12a2f3af5bede35442f1397","project":"openstack/tripleo-quickstart-extras","branch":"master","topic":"is_in_is_not_a_valid_test_according_to_ansible","hashtags":[],"change_id":"I02670fe00f9d9d5ef12a2f3af5bede35442f1397","subject":"\u0027is in\u0027 is not a valid Jinja test","status":"MERGED","created":"2019-08-13 14:14:01.000000000","updated":"2019-09-10 22:55:50.000000000","submitted":"2019-09-10 22:55:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"676203-1568156150300-53d8a3d3","meta_rev_id":"fbb0ebd534440d8470eb70a7fc88a86e93130a63","_number":676203,"virtual_id_number":676203,"owner":{"_account_id":1918,"name":"François Charlier","email":"francois.charlier@redhat.com","username":"francois-charlier"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":2,"date":"2019-09-10 22:55:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-04 17:38:53.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},"recommended":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"all":[{"value":1,"date":"2019-09-01 09:34:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":2,"date":"2019-09-04 14:11:38.000000000","_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"value":2,"date":"2019-09-01 09:05:21.000000000","_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":2,"date":"2019-09-10 19:40:23.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2019-09-01 09:05:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":1,"date":"2019-09-10 19:40:23.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-13 14:14:20.000000000","updated_by":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-25 09:43:01.000000000","updated_by":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"reviewer":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"state":"REVIEWER"},{"updated":"2019-08-30 11:23:27.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"state":"REVIEWER"},{"updated":"2019-09-01 09:34:28.000000000","updated_by":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"reviewer":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"state":"REVIEWER"},{"updated":"2019-09-04 14:11:38.000000000","updated_by":{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},"reviewer":{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},"state":"REVIEWER"},{"updated":"2019-09-04 17:38:53.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-10 19:40:23.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-09-10 22:55:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ce16244f9c1473f180a0f6ea764cb766ef4a20af","author":{"_account_id":1918,"name":"François Charlier","email":"francois.charlier@redhat.com","username":"francois-charlier"},"date":"2019-08-13 14:14:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e54611413f0a4570c96c28677b55d9a2b902507c","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-08-13 14:14:20.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/3781/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"3fa0929e516f87a82dc46edc85beb35842951fe2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-13 17:27:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/03/676203/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/8c608a5/ : SUCCESS in 2h 52m 30s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/03/676203/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/fc54299/ : SUCCESS in 2h 59m 18s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/03/676203/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/641ae88/ : SUCCESS in 2h 14m 54s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/03/676203/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/920f11f/ : SUCCESS in 2h 28m 25s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/03/676203/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/5b341ed/ : SUCCESS in 2h 48m 00s","accounts_in_message":[],"_revision_number":1},{"id":"8b093c08c83c8a6fd4ce13348738fa940958f2be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 19:16:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/51d5abe546424f2886c5f494f1ee3ae2 : SUCCESS in 1h 32m 29s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/d8ccf5685cf644d69da09959b6e44dd7 : FAILURE in 1h 53m 49s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/fdc2dae35a7f46048b3d20fbe82bade4 : SUCCESS in 2h 21m 16s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/07bc69ed8b864b9ba7dc1000395036c1 : SUCCESS in 2h 15m 28s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/5461a694ed664a569954e93db44dc4fc : FAILURE in 2h 24m 25s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/5fe5709870a24ec89178c12d6e779686 : SUCCESS in 2h 29m 11s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/181004d20a5c491a8e1dda9de48632d5 : SUCCESS in 2h 27m 30s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/d37bf24c171c43f09215930826d09a77 : SUCCESS in 1h 56m 01s\n- tripleo-ci-fedora-28-standalone https://zuul.opendev.org/t/openstack/build/cb1a11adbbb44e47a02ca005f937e684 : SUCCESS in 1h 40m 38s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/f02d1e4292aa4703a161dd8f0b33a5a6 : SUCCESS in 2h 17m 13s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/6edf053f2bbe43169db883b81272b454 : SUCCESS in 2h 11m 45s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/785283fb50774b9ca55b9038da5873f1 : SUCCESS in 1h 37m 08s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/a343be88f77243ce9adbb50295726071 : SUCCESS in 1h 16m 32s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/12bfe26a74094933a9b7554be48f5afa : SUCCESS in 1h 42m 39s\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/ac2d6b4d771d465581a3a18b8116e294 : SUCCESS in 1h 23m 45s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/765eb7b2daf348c6b4e30e519a76bbf2 : SUCCESS in 2h 10m 19s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/db8c067f2da047038746e31ec8eb080d : SUCCESS in 2h 28m 37s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/9798a141d6c143cc9ea03d480eb9467c : SUCCESS in 2h 15m 24s\n- tripleo-ci-centos-7-containers-multinode-pike https://zuul.opendev.org/t/openstack/build/577692a6335541cbaa484dfdac6c0626 : SUCCESS in 2h 26m 52s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/4b650956b91943a8abf432d968ea7758 : SUCCESS in 1h 09m 20s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/842816d0e18d4cf1a404aa16b19328a8 : SUCCESS in 1h 54m 08s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/228bcfb092b84e2095a152b101eb2fa1 : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":1},{"id":"ca70b5cfa029ddc7d95d0916811b66add4deba4d","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-08-25 09:43:01.000000000","message":"Patch Set 1: Code-Review+2\n\nGood catch!","accounts_in_message":[],"_revision_number":1},{"id":"f6ef864318a1fb496e8a8e3a07e0a9041249992c","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2019-08-30 11:23:27.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9d20d78ac3e0b4c2f32f03bab9f9d14f443decad","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-09-01 09:05:21.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"e2e1598e20ca8b6cff1596effa3b53c98da69f74","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-01 09:28:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/03/676203/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/77d4745/ : FAILURE in 17m 53s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/03/676203/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/73b2c57/ : FAILURE in 17m 05s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/03/676203/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/0f9c6fa/ : FAILURE in 19m 35s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/03/676203/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/c31863d/ : FAILURE in 19m 36s","accounts_in_message":[],"_revision_number":2},{"id":"ff87094d3fa5c864da94aaba6ee353febb27dd3d","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2019-09-01 09:34:28.000000000","message":"Patch Set 2: Code-Review+1\n\nship it!","accounts_in_message":[],"_revision_number":2},{"id":"a111b8c5533696ce9f62ab214daf089a24a366cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-01 09:39:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/faa8f5ef8d9d4d638f96ce74464405b0 : FAILURE in 16m 54s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/65cd781dfe3e4ed7b03136260be209de : FAILURE in 10m 13s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/a96b362a64454d1fb0207764ae9c6139 : FAILURE in 14m 13s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/9fc60637753d46f6a19f2324d1fcddb2 : FAILURE in 20m 32s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/055446cdad814e91b785e34e304b2ab2 : FAILURE in 22m 36s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/bfa2386e6b274ae9b6c47166d40e1637 : FAILURE in 11m 35s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/cd9254fd8ca64d04810dbe4640c476f1 : FAILURE in 21m 40s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/50ecb3a056674840aa8048ad0b5d97ce : FAILURE in 16m 01s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/d3b156362b0c4e6f8f835673d7fa6584 : FAILURE in 9m 30s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/9f62d578b7e240cd8c2bb199218010bb : FAILURE in 20m 15s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/4308735062154216b1b575aa0ef57b79 : FAILURE in 15m 44s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/73e73e66428e460bbf94cd02308abd4c : FAILURE in 16m 23s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/d5751b3237f242fb80afa6df3354c216 : FAILURE in 15m 52s\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/5ff93e2067cf4154bdf70d12b1ca2633 : FAILURE in 18m 26s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/c3a80747d8214f5ba5943223f2b59455 : FAILURE in 11m 54s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/6d02ee2f73f64f439244d7b203283014 : FAILURE in 13m 51s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/4db550509bfa483d99ce16167f1a09b5 : FAILURE in 16m 12s\n- tripleo-ci-centos-7-containers-multinode-pike https://zuul.opendev.org/t/openstack/build/b416be13d4de4ec6b639170f1c27fed0 : FAILURE in 12m 02s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/aac62ebb363c428aa002bc10463cb09a : FAILURE in 16m 48s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/2550fd55d7e04e46a391e111f4db0374 : FAILURE in 16m 05s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/90fe1771b38a41f6b61a71a4f04e100c : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":2},{"id":"11ad71c4e5b00d8fe304233f9e202bf1134dc663","author":{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},"date":"2019-09-04 14:11:38.000000000","message":"Patch Set 2: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d9fb477b2ede00638ab159233a4e8a0ba3167bf8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-04 17:38:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/03/676203/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/a0bed39/ : SUCCESS in 2h 57m 46s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/03/676203/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/ddc51df/ : SUCCESS in 2h 14m 26s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/03/676203/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/4238dea/ : SUCCESS in 2h 32m 22s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/03/676203/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/1c1f61d/ : SUCCESS in 2h 48m 14s","accounts_in_message":[],"_revision_number":2},{"id":"b539782d82fb9bc62aaec04e8eee33510ea9b00d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 21:22:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/45457b767e3d49f28f9e13caf8a7e1c3 : SUCCESS in 1h 45m 47s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/adcdef18aa6e4a02bc877c607930c370 : FAILURE in 2h 02m 56s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/e6162ba0850a477eb85756aa01082be8 : SUCCESS in 2h 51m 03s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/0c37e285c246431697fd4de2f6266338 : SUCCESS in 2h 53m 45s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/ef0b358126fe45f094911a247f4814e9 : FAILURE in 2h 56m 15s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/94a9a50405ee44a0a3b4ea02d70f6f00 : SUCCESS in 2h 49m 48s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/a132bda2032d41a5a26932023a653808 : SUCCESS in 3h 28m 16s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/cf9724fadf7c421aa0ffeffe9d60202d : SUCCESS in 1h 52m 23s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/6199ab5ed4d14e30942f26e0df5ac0c2 : SUCCESS in 1h 57m 06s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/1f0aca4f71404b1e8e25d4350bef4981 : SUCCESS in 1h 47m 50s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/bb831900e53c491d9641bf4949f2dc1c : SUCCESS in 1h 58m 26s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/0618e6cf8bd245d1b653b11723c308e3 : SUCCESS in 1h 42m 25s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/ad7dc2e63fe847c181f3630166ca612f : SUCCESS in 2h 06m 01s\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/234b08337369485e8d9e2f8786bb29e0 : SUCCESS in 1h 29m 42s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/949e74dfd2cb4e9bbd570354e7c151c2 : SUCCESS in 2h 33m 33s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/0f1778abd92b4324b9e90b1c3d0b3149 : SUCCESS in 2h 27m 21s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/078cf43ac13348369a37ee9979cbb8e9 : SUCCESS in 2h 34m 57s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/dfb36b839cf048f9a794830c4be78515 : SUCCESS in 1h 38m 21s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/ee00617b96b54557bf180386c3fb727a : FAILURE in 1h 37m 32s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/eb48a50dd1524136bcaf0377e3cefc2a : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":2},{"id":"14260b95569b5ca29cce0b80059ab0e4aca651dc","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-09-10 19:40:23.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2c8073974a102cbafe11d8287432439ae67617bc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 19:40:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3ef2aea4158ae78687df6493bacd383417765377","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 22:55:49.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/14918a7c3c9b4fceb9a499210b8ad732 : SUCCESS in 2h 00m 02s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/efff1ea8ff6440c999afd3942b660003 : SUCCESS in 2h 58m 06s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/4b3caa668fb14313928992adc4db2809 : SUCCESS in 2h 29m 10s\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/ef7c3229e1034a3daacd1cb5b57bcf8e : SUCCESS in 1h 45m 46s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/987a5420d21f4e0c992063fbabffe739 : SUCCESS in 2h 24m 16s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/539956101a574331ab59b953a0976afe : SUCCESS in 2h 25m 45s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/1e9c72d765ac4d34b9c67394f90a0615 : SUCCESS in 2h 12m 34s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/448edc5c684c48789c47b082b58fe479 : SUCCESS in 2h 10m 18s\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/54f7867202e14cb99c4af04fba2d868b : SUCCESS in 1h 58m 13s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/cc2593e4d0fd46c383994f94b2cd8c64 : SUCCESS in 2h 01m 59s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/653dd0f46a3c41348788a24db4f8bebc : SUCCESS in 1h 32m 25s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/8f2cf20ebf264052b9692e23cc5101d5 : SUCCESS in 2h 10m 55s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/140818a007d64276b072cf3919cb7968 : SUCCESS in 1h 37m 55s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3c85492c94264d878c93f6002f435f03 : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":2},{"id":"a46558c62ea3fbfea07e826900dc9f1053f90bb1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 22:55:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ae40f0ff93106f15bc1be9d214037b2d6bd7d4f6","revisions":{"9b91bd13f21f5b24d2d57639b2f73482fdd0aabc":{"kind":"REWORK","_number":1,"created":"2019-08-13 14:14:01.000000000","uploader":{"_account_id":1918,"name":"François Charlier","email":"francois.charlier@redhat.com","username":"francois-charlier"},"ref":"refs/changes/03/676203/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/03/676203/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/03/676203/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/03/676203/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/03/676203/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/03/676203/1"}}},"commit":{"parents":[{"commit":"fb6c4cc1656b5e89a67e49fe9f0b0f19599309e8","subject":"Merge \"Remove VolumesSnapshotTestJSON from stein skip list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/fb6c4cc1656b5e89a67e49fe9f0b0f19599309e8"}]}],"author":{"name":"François Charlier","email":"francois.charlier@redhat.com","date":"2019-08-13 14:10:44.000000000","tz":120},"committer":{"name":"François Charlier","email":"francois.charlier@redhat.com","date":"2019-08-13 14:10:44.000000000","tz":120},"subject":"\u0027is in\u0027 is not a valid Jinja test","message":"\u0027is in\u0027 is not a valid Jinja test\n\nAnsible complains that \"is in\" is not a valid test, use \"in\" instead.\n(introduced in f20c29689)\n\nChange-Id: I02670fe00f9d9d5ef12a2f3af5bede35442f1397\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/9b91bd13f21f5b24d2d57639b2f73482fdd0aabc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/9b91bd13f21f5b24d2d57639b2f73482fdd0aabc"}]},"branch":"refs/heads/master"},"ae40f0ff93106f15bc1be9d214037b2d6bd7d4f6":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-09-01 09:05:21.000000000","uploader":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"ref":"refs/changes/03/676203/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/03/676203/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/03/676203/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/03/676203/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/03/676203/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/03/676203/2"}}},"commit":{"parents":[{"commit":"ad955589c830420ee4ff75a7ef6655662b911eb9","subject":"Merge \"Increase size of cinder LVM backend\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/ad955589c830420ee4ff75a7ef6655662b911eb9"}]}],"author":{"name":"François Charlier","email":"francois.charlier@redhat.com","date":"2019-08-13 14:10:44.000000000","tz":120},"committer":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-09-01 09:05:21.000000000","tz":0},"subject":"\u0027is in\u0027 is not a valid Jinja test","message":"\u0027is in\u0027 is not a valid Jinja test\n\nAnsible complains that \"is in\" is not a valid test, use \"in\" instead.\n(introduced in f20c29689)\n\nChange-Id: I02670fe00f9d9d5ef12a2f3af5bede35442f1397\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/ae40f0ff93106f15bc1be9d214037b2d6bd7d4f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/ae40f0ff93106f15bc1be9d214037b2d6bd7d4f6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
