)]}'
{"id":"openstack%2Ftripleo-quickstart-extras~677182","triplet_id":"openstack%2Ftripleo-quickstart-extras~master~If8e8f9bac6f4cdedb213eb22ca650e23e2a099ce","project":"openstack/tripleo-quickstart-extras","branch":"master","topic":"dnm_tempest","hashtags":[],"change_id":"If8e8f9bac6f4cdedb213eb22ca650e23e2a099ce","subject":"[DNM]gate testing","status":"ABANDONED","created":"2019-08-19 11:33:59.000000000","updated":"2019-08-26 14:47:38.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6af97f0452082a6bb5b54d783b8558f911946ab9","_number":677182,"virtual_id_number":677182,"owner":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2019-08-19 14:42:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-08-19 14:32:59.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"date":"2019-08-19 11:34:17.000000000","_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-19 11:34:17.000000000","updated_by":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-19 14:32:59.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-19 14:42:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ff9a9f69e60f7bccbb54e8ff46ec7ac9d91dac5b","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-08-19 11:33:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f5eef7c79e3fb409ab8767e280109db4c5dddd4d","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-08-19 11:34:17.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/3805/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"461194d45f317cfc4bf68c8076d31c14835244af","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-19 14:32:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/82/677182/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/340de68/ : FAILURE in 1h 57m 09s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/82/677182/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/7edaaae/ : FAILURE in 1h 57m 51s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/82/677182/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/be7f790/ : FAILURE in 2h 53m 55s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/82/677182/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/3cc08a9/ : FAILURE in 2h 57m 18s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/82/677182/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/e89c610/ : FAILURE in 2h 38m 23s","accounts_in_message":[],"_revision_number":1},{"id":"392d489397bb53e8602e6f80abaa4148fe8df8f4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 14:42:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/43798b18b9f84666b9c53732bbdf0d67 : FAILURE in 2h 27m 07s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/1e8439f23f4e47a0ba085ec24fbd4105 : FAILURE in 2h 23m 31s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/284d7eaf16514414b207592c687c2bcf : FAILURE in 2h 25m 07s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/e852eecfbe3d44da8af4fb034f063bfd : FAILURE in 2h 04m 49s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/78a9cf2c736e4ee78fa9f78142fd4daf : FAILURE in 2h 08m 47s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/93aba329d61143aaade8d1c6c6c0602b : FAILURE in 2h 28m 38s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/2252d30bf8b940a497dda58288494b13 : FAILURE in 1h 57m 14s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/1bd0aa4ff03d40e2a7c1622b0bbe94dc : FAILURE in 2h 39m 59s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/0391906d46b94b129c06f852e26e2301 : SUCCESS in 1h 55m 18s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/faff28e2937240db930a37360e603f82 : FAILURE in 1h 19m 40s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/0e178abff4e54a6e9dd74106f1c46b08 : FAILURE in 2h 24m 44s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/5cb1e4f44fc8452ca016702ca2c1e429 : TIMED_OUT in 3h 07m 40s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/0cf7fb482a6c4bdea58d143b7b7faf24 : FAILURE in 2h 27m 33s\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/91ecb6981fe14969bcbe9dea5bdebd3f : FAILURE in 23m 17s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/0b4a29b662ab4a458182aaf56f1183d9 : FAILURE in 14m 01s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/3e63798f81d546f08a40d2a7d63ef759 : FAILURE in 2h 24m 56s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/776f9430b3e6492e8cb389f0325d4b50 : FAILURE in 1h 59m 47s\n- tripleo-ci-centos-7-containers-multinode-pike https://zuul.opendev.org/t/openstack/build/b14192c8ac744b4eaafd3b55e2067407 : FAILURE in 2h 09m 44s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/8584b05fd43d4be0b77e7de0563f328c : FAILURE in 2h 16m 52s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/bce7b68ea7a54adbbe752719169def3c : FAILURE in 2h 08m 55s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/703909fa2b7549dca700807f1b6fe88c : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":1},{"id":"aa3d0d0143f115522148a4076ff62e5e670ea8ac","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-08-26 14:47:38.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d0b76ae3119f8d914515b4a783b3297a4285a73b","revisions":{"d0b76ae3119f8d914515b4a783b3297a4285a73b":{"kind":"REWORK","_number":1,"created":"2019-08-19 11:33:59.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/82/677182/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/82/677182/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/82/677182/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/82/677182/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/82/677182/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/82/677182/1"}}},"commit":{"parents":[{"commit":"dfe7238db1e2c2b6b2d33d449097affab9603c7f","subject":"Configure rhel container for tempest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/dfe7238db1e2c2b6b2d33d449097affab9603c7f"}]}],"author":{"name":"Chandan kumar","email":"chkumar@redhat.com","date":"2019-08-19 11:33:15.000000000","tz":330},"committer":{"name":"Chandan kumar","email":"chkumar@redhat.com","date":"2019-08-19 11:33:15.000000000","tz":330},"subject":"[DNM]gate testing","message":"[DNM]gate testing\n\nDepends-On: https://review.opendev.org/#/c/674919/\nDepends-On: https://review.opendev.org/#/c/677063/\n\nChange-Id: If8e8f9bac6f4cdedb213eb22ca650e23e2a099ce\nSigned-off-by: Chandan kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/d0b76ae3119f8d914515b4a783b3297a4285a73b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/d0b76ae3119f8d914515b4a783b3297a4285a73b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
