)]}'
{"id":"openstack%2Ftripleo-quickstart-extras~678141","triplet_id":"openstack%2Ftripleo-quickstart-extras~master~I0e73bc0e4b1abb761c2f5f470297b2f16928021c","project":"openstack/tripleo-quickstart-extras","branch":"master","topic":"add_master_skip","hashtags":[],"change_id":"I0e73bc0e4b1abb761c2f5f470297b2f16928021c","subject":"Move SnapshotsActionsTest to master skip list","status":"MERGED","created":"2019-08-23 04:24:52.000000000","updated":"2019-08-23 20:39:29.000000000","submitted":"2019-08-23 20:39:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"678141-1566592769266-8367d9e9","meta_rev_id":"a29d814e9f22f71318cc713959a7e62303a53775","_number":678141,"virtual_id_number":678141,"owner":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":2,"date":"2019-08-23 20:39:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-08-23 18:23:29.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"all":[{"value":0,"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"value":2,"date":"2019-08-23 15:03:31.000000000","_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"all":[{"value":0,"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"value":1,"date":"2019-08-23 17:05:42.000000000","_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-23 04:25:27.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2019-08-23 04:25:32.000000000","updated_by":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"reviewer":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"state":"REVIEWER"},{"updated":"2019-08-23 04:25:43.000000000","updated_by":{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},"reviewer":{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},"state":"REVIEWER"},{"updated":"2019-08-23 10:07:44.000000000","updated_by":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"reviewer":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"state":"REVIEWER"},{"updated":"2019-08-23 12:57:44.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2019-08-23 17:05:42.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"REVIEWER"},{"updated":"2019-08-23 18:23:29.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-23 20:39:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"51d89f165b4fd9846e5a2a901f646ef5cc9bd879","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-08-23 04:24:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2b0b671438dd128850cd5c536549ae3f501e4db3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 07:57:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/858bb2ea227b49cc9fcacec5d4a7d781 : SUCCESS in 1h 40m 50s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/28c49968d31e4552ab66bb93b8f714cd : FAILURE in 16m 48s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/cff1ce6046a24d0d9706ee6ef86edb84 : SUCCESS in 2h 38m 12s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/2b04b2cef2474ef5b60aa7588bc09ba9 : SUCCESS in 3h 04m 25s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/39da29b973854264bcbb196c51ef5227 : FAILURE in 2h 49m 06s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/5c91a854136e4b82aa3b2db6316db283 : SUCCESS in 2h 49m 17s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/576e55ee5f914a74af205972e742664f : SUCCESS in 3h 30m 29s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/edc84ebcc21b41ab93750601df9448a6 : SUCCESS in 1h 59m 40s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/486bfa41a2554d258eabc930162a4807 : SUCCESS in 2h 02m 08s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/ca8e1f55314941a4941d616cb1521a1b : FAILURE in 1h 05m 28s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/1e52fc9c19d841b68432d30ae5278f53 : SUCCESS in 2h 07m 22s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/2595628fc83d4eb4b451d694c5a60fd7 : SUCCESS in 1h 21m 52s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/49c6b1c59b934c27a9e4b1b7364598aa : SUCCESS in 1h 50m 58s\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/6a262f626ca94b669d77f37709b11f6d : SUCCESS in 1h 28m 17s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/40d231a5f0de4ffb9a49d69338e5173f : SUCCESS in 2h 16m 18s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/1194c80832ab43c08a714d6dbd6ac5ad : SUCCESS in 2h 19m 03s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/b25cb265eadf4c50a3ff412cccf2ca8a : SUCCESS in 2h 04m 57s\n- tripleo-ci-centos-7-containers-multinode-pike https://zuul.opendev.org/t/openstack/build/40b6a1f78da2425a8b1140b9734e49a8 : SUCCESS in 2h 49m 49s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/de1b9a4939394982a4f1e3a95af37924 : SUCCESS in 1h 24m 56s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/f4644e3950c04dc1a55bc4b60f5d43f6 : FAILURE in 1h 42m 52s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/878ef607d0514cfb9850e3e2a93adbac : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":1},{"id":"05148935c84b473c72b64d988a36659b1fbc80a0","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-08-23 08:23:10.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fde2690993600f7244b53f4ece29f02bb00061ba","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-23 08:57:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/41/678141/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/6eebef3/ : SUCCESS in 2h 56m 38s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/41/678141/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/db96fa9/ : TIMED_OUT in 4h 32m 10s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/41/678141/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/0e034be/ : SUCCESS in 2h 32m 24s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/41/678141/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/3a0c2fd/ : SUCCESS in 2h 18m 21s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/41/678141/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/c07fab5/ : SUCCESS in 3h 04m 42s\n- tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020 http://logs.rdoproject.org/41/678141/1/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020/ed8db2d/ : FAILURE in 1h 29m 24s","accounts_in_message":[],"_revision_number":1},{"id":"34578d6e1839108ca276cc855c9ff6a9059274c1","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-08-23 09:25:36.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"e87240c1b3fcc7baff30936a4bbb34954fe70601","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2019-08-23 10:07:44.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"20af3c42d79b9136c570263f686ff7f603abe7e7","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2019-08-23 12:30:08.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4612b3be53d3b086a9b604c596d99978a65ea27a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 12:39:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/a6f0cd9536f1494b8dfab2b5b4a51cf6 : SUCCESS in 2h 05m 34s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/bf758291e3c244239191969dc15daeef : FAILURE in 1h 55m 41s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/1fb90e36154d4affba7f72c658d7d54f : SUCCESS in 3h 02m 47s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/68228661c489463d8ae6689c931dc452 : SUCCESS in 2h 54m 47s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/410a78b61ce1459daeb990354f4ecd0c : FAILURE in 3h 02m 17s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/b7fda28a09a94782bd0bfdec0aa19c11 : SUCCESS in 2h 40m 37s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/ba88e9c2eeb84244a6f3d23bc79b6b85 : SUCCESS in 2h 59m 29s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/873560a1e7ce4b7cb5edc723f559bf0d : SUCCESS in 2h 15m 54s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/5e4c4adbae4547c09ded058aa7cb39f7 : SUCCESS in 2h 07m 38s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/63a0b3fd92cd492eba1c221188d348f5 : SUCCESS in 2h 11m 27s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/8d11bc8a77c441f5baf8eb18acab04fe : SUCCESS in 1h 57m 02s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/650379a66e4948f6a4fc74088cbb785a : SUCCESS in 1h 45m 10s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/976f4975994e4de8b1670a6535f9d1cf : SUCCESS in 2h 12m 23s\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/814c1413b098471db0b5dbdb34f4a1b6 : SUCCESS in 1h 46m 49s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/8e0b9ecd3d8a424aa82c3214726ef77a : SUCCESS in 2h 38m 02s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/729e5defde7f4354a79c2e87b6dc5dc2 : SUCCESS in 2h 37m 01s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/840141ace208499d9c5e233b7e320ee5 : SUCCESS in 2h 09m 31s\n- tripleo-ci-centos-7-containers-multinode-pike https://zuul.opendev.org/t/openstack/build/c4af938a3e094e4eb1fe3a29652f750e : SUCCESS in 2h 35m 35s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/ed94104420fb4df0b8801793e4d0541c : SUCCESS in 1h 33m 10s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/90a01ced8a404f2a9f9dc2417919dade : FAILURE in 1h 49m 34s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ca09c70af0244e828baa05e114e900a6 : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":2},{"id":"61071bfbd08fa7ff11fa8817b21a689dbdbf2373","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2019-08-23 12:57:44.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nDepends-on needs to be updated, but I\u0027m in favor of this patch as it will buy time to track down why LVM backend runs out of space (which might be due to a problem with cinder quotas).","accounts_in_message":[],"_revision_number":2},{"id":"7b8d5b200b971838a8a0187f8fb03aa0e2677534","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2019-08-23 13:14:22.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"facb97e91a0f7c28fa512e8f9be523565fd36210","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2019-08-23 13:14:48.000000000","message":"Patch Set 3:\n\nUpdated Depends-On and Related-Bug","accounts_in_message":[],"_revision_number":3},{"id":"bc9b53900bf502d355ac4e508d752e6a05bc4698","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2019-08-23 15:03:31.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"c0163c8ee1819119db36f178a0bdd165e1c09f73","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 16:41:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/b2dc025ce449443285825e6a84c69fbe : SUCCESS in 1h 57m 27s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/67aec84b34f84b95b705406c0345570e : FAILURE in 1h 59m 36s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/587641261eff4f75aa076948e1ce4e04 : SUCCESS in 3h 10m 56s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/695fdb2d8d834a0a9dde3652551f2baf : TIMED_OUT in 3h 24m 17s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/909a51c834bf4cd99f86bb43ad117fca : FAILURE in 3h 17m 15s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/bd6cb21addca4a78a2de58053dd631f5 : SUCCESS in 2h 59m 38s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/e84da3c85c474ce7a1095151a137dbe4 : SUCCESS in 3h 01m 06s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/a7e5d6acb5c444ddbfe936c66f370c2d : SUCCESS in 2h 16m 57s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/fd6a3685963c46f9a88e28362727efca : SUCCESS in 2h 14m 37s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/5029293f3c794507a8feddd38ce3ca94 : SUCCESS in 2h 09m 30s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/0b39ad7062704796ace814dfad652ce5 : SUCCESS in 1h 58m 07s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/bbd6bec6ed684ee8b6d212be7a90efdf : SUCCESS in 1h 38m 54s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/2d3cbe6478484eed91c93ccf5e79a85d : SUCCESS in 2h 00m 25s\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/373c523064854066990049fa574269e8 : SUCCESS in 1h 23m 37s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/8616a382b6bb406bbf0dcb13bf5c96c0 : SUCCESS in 2h 21m 56s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/c5a97dea3d564ef7ad70f5f2bca0370b : SUCCESS in 2h 32m 00s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/2406210361d9431baa5e26c04cecfa53 : SUCCESS in 2h 15m 48s\n- tripleo-ci-centos-7-containers-multinode-pike https://zuul.opendev.org/t/openstack/build/94e2713b2cfe4c48aa37d30a8c65fbe6 : SUCCESS in 2h 25m 36s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/db789753a8a842f0be315e0ae0a55b22 : SUCCESS in 1h 37m 53s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/bd7a27dbed1441178ff4ef1011311d91 : FAILURE in 1h 24m 22s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8d7ab72d9fd94a17bba572600565c8c9 : SUCCESS in 7m 00s","accounts_in_message":[],"_revision_number":3},{"id":"072c9a4c8b6787588e088a0fae45bd562c1ed786","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2019-08-23 17:05:42.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b31836cde3b2613b0c8d59d5d034bd53685b4036","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 17:06:00.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e186ce28f59a587542a96dadc09bdd1a747c9c7b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-23 18:23:29.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/41/678141/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/574aea2/ : FAILURE in 3h 10m 27s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/41/678141/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/c558c7b/ : FAILURE in 3h 01m 40s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/41/678141/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/ec06450/ : SUCCESS in 2h 14m 26s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/41/678141/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/1f0c2f5/ : SUCCESS in 2h 15m 04s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/41/678141/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/e7275b0/ : SUCCESS in 2h 32m 11s\n- tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020 http://logs.rdoproject.org/41/678141/3/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020/08f99a1/ : SUCCESS in 4h 53m 21s","accounts_in_message":[],"_revision_number":3},{"id":"51bc6e700db9f0bf51cb8103d0d6626f202413f6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 20:39:28.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/e91320278fa34bb5b7b37368efc9519d : SUCCESS in 1h 42m 56s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/f756225255de4c68ad6a52da2fe1da89 : SUCCESS in 2h 52m 15s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/7944cf1973fe457fa0245551aab2532c : SUCCESS in 2h 17m 41s\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/df946633339745f492b4a1ef6bb814f2 : SUCCESS in 1h 38m 55s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/96284bf9189749a8908469250fda6d86 : SUCCESS in 2h 44m 12s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/a783c7704df64c859a07547b96b700b1 : SUCCESS in 2h 08m 52s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/66f39b2b4b274d25a4ef056683311768 : SUCCESS in 2h 27m 45s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/091808f62c0f4f96879544f3e1bc47b9 : SUCCESS in 2h 02m 13s\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/54802fed015b4bd5839fe23d15b0b2d5 : SUCCESS in 2h 08m 37s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/823cdf8792de49c28bcd1b5167a1765d : SUCCESS in 2h 03m 24s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/65199b692e7b4b108bebd67be3023acf : SUCCESS in 1h 41m 37s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/d7579d1c3aa84b328f615dc97cd76679 : SUCCESS in 2h 08m 26s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/578767a5b026407a8bc9b4c0e1f85ded : SUCCESS in 1h 33m 07s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7c16fae5db4540969d3759ac81586672 : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":3},{"id":"495b9c1656a173ed3a3a33d3ecddd31090d76c2f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 20:39:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c1801751efe1861915c02d2faa246e74d8aeabfb","revisions":{"eb14d5ead07d302acacf17c13d1695ba58bac1a3":{"kind":"REWORK","_number":1,"created":"2019-08-23 04:24:52.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/41/678141/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/41/678141/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/41/678141/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/41/678141/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/41/678141/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/41/678141/1"}}},"commit":{"parents":[{"commit":"6a20363ef06d31099cd6b99831afc20cc33210b7","subject":"Merge \"Define var undercloud_enable_nova\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/6a20363ef06d31099cd6b99831afc20cc33210b7"}]}],"author":{"name":"Chandan kumar","email":"chkumar@redhat.com","date":"2019-08-23 04:22:16.000000000","tz":330},"committer":{"name":"Chandan kumar","email":"chkumar@redhat.com","date":"2019-08-23 04:22:16.000000000","tz":330},"subject":"Move SnapshotsActionsTest to master skip list","message":"Move SnapshotsActionsTest to master skip list\n\nThese tests are failing randomly due insufficient free virtual\nspace (-12.0GB) to accommodate thin provisioned 1GB volume on\nhost hostgroup in Cinder. Increasing LVM space does not fix the\nactual issue.\n\nInvestigation is in-progress. Adding it to skip list for master\npromotion.\n\nRelated-Bug: #1840451\n\nChange-Id: I0e73bc0e4b1abb761c2f5f470297b2f16928021c\nSigned-off-by: Chandan kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/eb14d5ead07d302acacf17c13d1695ba58bac1a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/eb14d5ead07d302acacf17c13d1695ba58bac1a3"}]},"branch":"refs/heads/master"},"1d78bfe6524f1aafc638fb96ed6ec27a32ea7428":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-08-23 09:25:36.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/41/678141/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/41/678141/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/41/678141/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/41/678141/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/41/678141/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/41/678141/2"}}},"commit":{"parents":[{"commit":"6a20363ef06d31099cd6b99831afc20cc33210b7","subject":"Merge \"Define var undercloud_enable_nova\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/6a20363ef06d31099cd6b99831afc20cc33210b7"}]}],"author":{"name":"Chandan kumar","email":"chkumar@redhat.com","date":"2019-08-23 04:22:16.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-08-23 09:25:32.000000000","tz":0},"subject":"Move SnapshotsActionsTest to master skip list","message":"Move SnapshotsActionsTest to master skip list\n\nThese tests are failing randomly due insufficient free virtual\nspace (-12.0GB) to accommodate thin provisioned 1GB volume on\nhost hostgroup in Cinder. Increasing LVM space does not fix the\nactual issue.\n\nInvestigation is in-progress. Adding it to skip list for master\npromotion.\n\nDepends-On: https://review.opendev.org/678186\n\nRelated-Bug: #1840451\n\nChange-Id: I0e73bc0e4b1abb761c2f5f470297b2f16928021c\nSigned-off-by: Chandan kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/1d78bfe6524f1aafc638fb96ed6ec27a32ea7428"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/1d78bfe6524f1aafc638fb96ed6ec27a32ea7428"}]},"branch":"refs/heads/master"},"c1801751efe1861915c02d2faa246e74d8aeabfb":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-08-23 13:14:22.000000000","uploader":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"ref":"refs/changes/41/678141/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/41/678141/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/41/678141/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/41/678141/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/41/678141/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/41/678141/3"}}},"commit":{"parents":[{"commit":"6a20363ef06d31099cd6b99831afc20cc33210b7","subject":"Merge \"Define var undercloud_enable_nova\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/6a20363ef06d31099cd6b99831afc20cc33210b7"}]}],"author":{"name":"Chandan kumar","email":"chkumar@redhat.com","date":"2019-08-23 04:22:16.000000000","tz":330},"committer":{"name":"Ronelle Landy","email":"rlandy@redhat.com","date":"2019-08-23 13:14:17.000000000","tz":0},"subject":"Move SnapshotsActionsTest to master skip list","message":"Move SnapshotsActionsTest to master skip list\n\nThese tests are failing randomly due insufficient free virtual\nspace (-12.0GB) to accommodate thin provisioned 1GB volume on\nhost hostgroup in Cinder. Increasing LVM space does not fix the\nactual issue.\n\nInvestigation is in-progress. Adding it to skip list for master\npromotion.\n\nDepends-On: https://review.opendev.org/#/c/674556/\n\nRelated-Bug: #1841071\n\nChange-Id: I0e73bc0e4b1abb761c2f5f470297b2f16928021c\nSigned-off-by: Chandan kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/c1801751efe1861915c02d2faa246e74d8aeabfb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/c1801751efe1861915c02d2faa246e74d8aeabfb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
