)]}'
{"id":"openstack%2Ftripleo-quickstart-extras~678894","triplet_id":"openstack%2Ftripleo-quickstart-extras~master~Ie02211f3a6706f6fa1102fb8dc5ee0e9e54252cc","project":"openstack/tripleo-quickstart-extras","branch":"master","topic":"bug/1841071","hashtags":[],"change_id":"Ie02211f3a6706f6fa1102fb8dc5ee0e9e54252cc","subject":"Increase size of cinder LVM backend","status":"MERGED","created":"2019-08-27 16:06:41.000000000","updated":"2019-08-29 14:20:29.000000000","submitted":"2019-08-29 14:20:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"678894-1567088429868-22bc38ef","meta_rev_id":"7d4b7104803b9b2725c0c5985bcef5b4f680fcc0","_number":678894,"virtual_id_number":678894,"owner":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":2,"date":"2019-08-29 14:20:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-08-27 20:57:36.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":2,"date":"2019-08-29 11:11:33.000000000","_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":1,"date":"2019-08-28 13:01:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":2,"date":"2019-08-28 02:03:39.000000000","_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":1,"date":"2019-08-28 05:12:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":1,"date":"2019-08-29 11:11:33.000000000","_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"date":"2019-08-28 12:30:37.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-27 20:57:36.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-28 02:03:39.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"REVIEWER"},{"updated":"2019-08-28 05:12:46.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2019-08-28 13:01:54.000000000","updated_by":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"reviewer":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"state":"REVIEWER"},{"updated":"2019-08-29 11:11:33.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2019-08-29 14:20:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"52785939cd6088446e896002cbbc96cffc72832e","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2019-08-27 16:06:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a33989eb523b05762111216fc27cf035a1c13465","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2019-08-27 16:08:15.000000000","message":"Patch Set 1: Workflow-1\n\nLet\u0027s see if fs020 succeeds a couple of times.","accounts_in_message":[],"_revision_number":1},{"id":"75d74faffa00e481f4596dd6ad2b606f304ce82c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 20:42:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/9942a8cf63a64d718091be666d1e6b19 : SUCCESS in 1h 53m 08s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/29509d96e8744d34a8b68ce17b58dbb9 : FAILURE in 2h 04m 25s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/8d445102e827447f8ac5fa6e4d2f19b4 : SUCCESS in 2h 47m 57s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/caf6fb3afb074b3492bf77d91b7ca84f : SUCCESS in 2h 58m 19s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/d2607095a06e4563bc44db4c61c03b6c : FAILURE in 3h 01m 46s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/a09fc9f20b2b4fd99dc1aef00b0e097d : SUCCESS in 2h 47m 53s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/7341cc64c7df48b583efe907507a84e7 : SUCCESS in 2h 35m 07s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/9dd8e1bbce72434781d30de5112ecc8a : SUCCESS in 2h 02m 24s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/f94ec3f7bb8e4ae5abedecd4921f6a9e : SUCCESS in 2h 20m 55s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/7e61753228004a588f837abe9dd093c5 : SUCCESS in 2h 05m 13s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/ed2161cda83a49e9a01fd87ed9c12dba : SUCCESS in 2h 01m 46s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/036479373d574aa4a8b36db19980d67c : SUCCESS in 1h 30m 11s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/eff9d3f8fa174e7297c75f5cd6682eb5 : SUCCESS in 2h 15m 57s\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/374a97af582b466e87813b145f1777c2 : SUCCESS in 1h 28m 19s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/77f4ef138b534c4693856e95b7121823 : SUCCESS in 2h 42m 49s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/cd5c158ebde14a9492d78ccad5045531 : SUCCESS in 2h 21m 53s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/a878781b17f74724bfc6c4a4c55d9ac4 : SUCCESS in 2h 38m 19s\n- tripleo-ci-centos-7-containers-multinode-pike https://zuul.opendev.org/t/openstack/build/fe989818d2414b01b1e1973bd4598b27 : SUCCESS in 2h 16m 16s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/f86625aee05a418cb56b2f6146e5a5f5 : SUCCESS in 1h 26m 21s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/5b7d1b75cafa42c383aa5506bc564beb : FAILURE in 1h 34m 05s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e575a07073744fcea91871d51541463e : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":1},{"id":"fb271573118109bed5afc8cadefde691e59e43b4","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2019-08-27 20:48:12.000000000","message":"Patch Set 1:\n\nlet\u0027s test this with fs020 ...\nhttps://review.rdoproject.org/r/21683","accounts_in_message":[],"_revision_number":1},{"id":"2d0be5ac979c67ff067bd921a19fc9cc13ae23cd","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-27 20:57:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/94/678894/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/bc99f7d/ : FAILURE in 3h 41m 23s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 finger://ze03.softwarefactory-project.io:7979/51e5fdad8e66450d8bd8b263b3e4be31 : POST_FAILURE in 3h 51m 04s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/94/678894/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/bcbe7c5/ : TIMED_OUT in 3h 51m 58s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/94/678894/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/8ade37b/ : FAILURE in 2h 42m 41s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch finger://ze04.softwarefactory-project.io:7979/b91d0384f3714e8e8a40d5352e91a37a : POST_FAILURE in 3h 51m 57s\n- tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020 http://logs.rdoproject.org/94/678894/1/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020/beba263/ : FAILURE in 2h 03m 24s","accounts_in_message":[],"_revision_number":1},{"id":"9bd81c5b1430f9af86384caccc4b695992707156","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2019-08-28 02:03:39.000000000","message":"Patch Set 1: Code-Review+2\n\ntest results look good ...\n2019-08-28 00:21:30 | {2} tempest.api.volume.admin.test_snapshots_actions.SnapshotsActionsTest.test_snapshot_force_delete_when_snapshot_is_creating [7.206595s] ... ok\n2019-08-28 00:21:33 | {1} tempest.api.volume.test_volume_transfers.VolumesTransfersTest.test_create_get_list_accept_volume_transfer [8.335725s] ... ok\n2019-08-28 00:21:34 | {2} tempest.api.volume.admin.test_snapshots_actions.SnapshotsActionsTest.test_snapshot_force_delete_when_snapshot_is_deleting [3.856088s] ... ok\n2019-08-28 00:21:38 | {1} tempest.api.volume.test_volume_transfers.VolumesTransfersTest.test_create_list_delete_volume_transfer [4.435717s] ... ok\n2019-08-28 00:21:39 | {2} tempest.api.volume.admin.test_snapshots_actions.SnapshotsActionsTest.test_snapshot_force_delete_when_snapshot_is_error [4.849884s] ... ok\n2019-08-28 00:21:41 | {2} tempest.api.volume.admin.test_snapshots_actions.SnapshotsActionsTest.test_snapshot_force_delete_when_snapshot_is_error_deleting [2.657468s] ... ok\n\nhttp://logs.rdoproject.org/83/21683/21/check/periodic-tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020-master/38bff31/logs/undercloud/home/zuul/tempest.log.txt.gz","accounts_in_message":[],"_revision_number":1},{"id":"2e8a4bdf4d9d544d047508ff3dc91c6ecd3494b6","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2019-08-28 02:16:20.000000000","message":"Patch Set 1:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":1},{"id":"1c674d52246ed6bf237a61359e20a0bb6b51398e","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-08-28 05:12:46.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"11b818b539b9eaf7022443b5940260c136211fea","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-28 06:57:27.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/94/678894/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/9b3b87d/ : FAILURE in 2h 46m 27s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/94/678894/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/5885cff/ : SUCCESS in 3h 08m 01s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/94/678894/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/a478cf7/ : FAILURE in 1h 33m 43s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/94/678894/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/5908193/ : FAILURE in 3h 06m 32s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/94/678894/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/b07bf96/ : SUCCESS in 3h 07m 42s\n- tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020 http://logs.rdoproject.org/94/678894/1/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020/dc60c59/ : SUCCESS in 4h 39m 58s","accounts_in_message":[],"_revision_number":1},{"id":"ee5d671549446ffd5d65ac3b244d3a58799545d2","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2019-08-28 12:30:37.000000000","message":"Patch Set 1: -Workflow\n\nLooks good, has passed twice.","accounts_in_message":[],"_revision_number":1},{"id":"921dd1bfab1bd368eab81a7ffeaa873e69466230","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2019-08-28 13:01:54.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"20f6274daf6785bd8493aa440da573434d25c429","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-08-29 10:30:28.000000000","message":"Patch Set 1:\n\nTested here https://review.rdoproject.org/r/21683","accounts_in_message":[],"_revision_number":1},{"id":"e260543d038b9402d711a7a780f646db4aeb8d5e","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2019-08-29 11:11:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"954b9753629703863fabd893f7782d4bc08854f1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 11:11:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cd9e60488ccf94e21c2967c81b65c259b4338580","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 14:20:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/9eb0adcf35d2408cad102035d5c62fa2 : SUCCESS in 2h 01m 29s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/272c05361dff4602b4a7ba24df0f8016 : SUCCESS in 2h 49m 20s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/1c2a087f391e4b7da6f42128ef4f362e : SUCCESS in 2h 25m 29s\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/95693797061b4717a42d71125ff8bf6e : SUCCESS in 1h 41m 54s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/c25d10b57c024c71affb442f95fbb3df : SUCCESS in 2h 35m 12s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/beae6e7b5e0d40599b5c5163d3072b82 : SUCCESS in 2h 52m 30s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/8cbb9688b6df43b6bffd532c4ebbeca9 : SUCCESS in 2h 37m 22s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/5ea46d311e8245b29d0f808b1ec09b22 : SUCCESS in 2h 21m 47s\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/bf6babb56f404399bd7c6a4fec4bf7ae : SUCCESS in 2h 17m 01s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/f34bc671a0304378b889033983ab6409 : SUCCESS in 1h 40m 51s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/5838691d29bd4891b6ac41d96065c692 : SUCCESS in 1h 37m 02s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/5ba5b31840274ce9879b9eb041e7d997 : SUCCESS in 2h 24m 21s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/45ed8ebb48654c1c8200fd5d2353cc7f : SUCCESS in 1h 42m 59s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f15ce9a3ecc841f0956181bbcb13d341 : SUCCESS in 8m 06s","accounts_in_message":[],"_revision_number":1},{"id":"6a20148d2b3321e395e5e48cfb47e6070d418004","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 14:20:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c5a6d012a18f8f8f571eba9218a9f7e527b67f87","revisions":{"c5a6d012a18f8f8f571eba9218a9f7e527b67f87":{"kind":"REWORK","_number":1,"created":"2019-08-27 16:06:41.000000000","uploader":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"ref":"refs/changes/94/678894/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/94/678894/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/94/678894/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/94/678894/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/94/678894/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/94/678894/1"}}},"commit":{"parents":[{"commit":"8f3bb293484c844182f6215fb323302efb06e8c0","subject":"Merge \"Move reproducer README file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/8f3bb293484c844182f6215fb323302efb06e8c0"}]}],"author":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2019-08-27 15:45:43.000000000","tz":-420},"committer":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2019-08-27 16:05:51.000000000","tz":-420},"subject":"Increase size of cinder LVM backend","message":"Increase size of cinder LVM backend\n\nIncrease the size of the cinder LVM backend to 20GB, and re-enable\ntempest tests that were previously skipped due to space limitation.\n\nTempest\u0027s volume tests don\u0027t actually need 20GB of storage, but rapidly\ncreating and deleting volumes in a short time span may cause cinder\u0027s\nscheduler to report a reduced amount of free space. The actual amount\nof space in use is updated by the cinder-volume service, but this is\ndone periodically. Increasing the size of the LVM backend allows the\nscheduler to sustain a burst of volume creation requests during the\ntime interval between updates from the volume service on the actual\namount of storage in use.\n\nCloses-Bug: #1841071\nChange-Id: Ie02211f3a6706f6fa1102fb8dc5ee0e9e54252cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/c5a6d012a18f8f8f571eba9218a9f7e527b67f87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/c5a6d012a18f8f8f571eba9218a9f7e527b67f87"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
