)]}'
{"id":"openstack%2Ftripleo-quickstart-extras~683350","triplet_id":"openstack%2Ftripleo-quickstart-extras~master~I3aaa38ff0180826a6e749d0686d1a0e37e97eb84","project":"openstack/tripleo-quickstart-extras","branch":"master","topic":"lp/1843259","hashtags":[],"change_id":"I3aaa38ff0180826a6e749d0686d1a0e37e97eb84","subject":"Adding tests to skip list in Rocky","status":"MERGED","created":"2019-09-20 10:08:03.000000000","updated":"2019-10-05 23:09:11.000000000","submitted":"2019-10-05 23:09:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"683350-1570316951023-bad3269c","meta_rev_id":"cae6dd607aa02e3200c144846c5fb01bce90d2d5","_number":683350,"virtual_id_number":683350,"owner":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-05 23:09:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-10-01 10:20:22.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},{"value":0,"date":"2019-10-05 16:28:39.000000000","_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},"recommended":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"all":[{"value":1,"date":"2019-10-01 07:29:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2019-10-01 07:28:47.000000000","_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"value":2,"date":"2019-10-02 12:57:48.000000000","_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":1,"date":"2019-10-01 07:28:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":2,"date":"2019-10-01 07:28:47.000000000","_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"all":[{"value":0,"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":1,"date":"2019-10-02 14:23:22.000000000","_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-20 11:03:16.000000000","updated_by":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"reviewer":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"state":"REVIEWER"},{"updated":"2019-09-20 11:05:01.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2019-09-24 08:50:37.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2019-09-26 17:34:17.000000000","updated_by":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"reviewer":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"state":"REVIEWER"},{"updated":"2019-09-26 17:54:38.000000000","updated_by":{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},"reviewer":{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},"state":"REVIEWER"},{"updated":"2019-10-01 07:29:08.000000000","updated_by":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"reviewer":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"state":"REVIEWER"},{"updated":"2019-10-01 10:20:22.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-02 12:57:48.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"REVIEWER"},{"updated":"2019-10-02 14:23:22.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2019-10-05 23:09:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8fe911c323d16469e1ac4459b4677ee04fd58a77","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2019-09-20 10:08:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"72f83e2e4d9d84e71fe5ffe01d8c7bfd0c97d844","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2019-09-20 10:09:22.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"b505fe86c1c0c8fa4fbab0545daefdffb6687f2a","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-09-20 11:05:01.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"524de08076b82db707849ca769264e555a2a76ad","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-20 13:38:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/50/683350/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/78f8f2a/ : SUCCESS in 2h 53m 38s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/50/683350/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/c2dd82a/ : FAILURE in 1h 50m 49s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/50/683350/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/0c281cb/ : FAILURE in 1h 50m 21s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/50/683350/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/40e09f6/ : SUCCESS in 2h 57m 39s\n- tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020-rocky http://logs.rdoproject.org/50/683350/2/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020-rocky/53d64ac/ : FAILURE in 3h 27m 04s","accounts_in_message":[],"_revision_number":2},{"id":"4827a31e19c2a1e8a80bd0991092150fa876e677","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 23:24:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/54f0160e885641ac8a793bc4fa4b312e : SUCCESS in 2h 03m 04s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/536a985f149f441582ec8350f4aa6ae6 : FAILURE in 2h 11m 40s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/7dbcac301dbb4c359c7103d256f337ed : TIMED_OUT in 3h 57m 54s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/ade9260b541248d78d4099159a082b9b : SUCCESS in 3h 30m 30s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/706ef4051c284632919daf5ff252b9b3 : FAILURE in 2h 11m 16s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/adea4c1d71764beeba97bb8c4d0e94d2 : SUCCESS in 2h 53m 33s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/1bb6ca4ff6ab435c97f2c83fbdb2a840 : SUCCESS in 3h 04m 43s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/b63a35c1adf7468b9156ebea775707bb : SUCCESS in 3h 13m 44s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/8394d81b371d4ad3b0f29e55505784d5 : TIMED_OUT in 3h 52m 50s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/56a3f62f68c24e9bb64ac24663cd32e5 : SUCCESS in 3h 24m 50s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/ae882b3c871e48c6af8f0d494c09ed9e : SUCCESS in 2h 24m 31s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/0da15ee82e63405f86da077091db0cad : SUCCESS in 1h 43m 59s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/3026a0f90e8948f6bb2e4f8326a6e32a : SUCCESS in 2h 06m 26s\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/5a5ff98f296a464b866a727cfe869e8a : SUCCESS in 1h 35m 58s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/0a53dfa92f2944e48bd109aab5e1460f : SUCCESS in 2h 56m 33s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/4466b2aba2e14e63a96e6ef605a67fc1 : SUCCESS in 3h 14m 40s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/29cad999fab841629f5fffa38f144005 : SUCCESS in 2h 54m 26s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/af390b4a49124d8f9af7e440a1d3a3a1 : SUCCESS in 2h 05m 30s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/fb5f6e5446bb41d0ae28ca0184b7284e : FAILURE in 2h 14m 38s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e1ec1edb72dc41f38656741a59d2ed09 : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":2},{"id":"f39416fbfccddd7dca4f9ef68a02b53c566e2423","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-09-24 08:50:37.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"b3f345547cb1dfd38d4d37108e605aad7975cee8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-24 13:27:41.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/50/683350/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/a8aeda0/ : SUCCESS in 2h 55m 39s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/50/683350/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/b6a7794/ : FAILURE in 1h 54m 04s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/50/683350/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/d9e5dd4/ : SUCCESS in 2h 15m 05s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/50/683350/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/9e71b6e/ : FAILURE in 3h 27m 11s\n- tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020-rocky http://logs.rdoproject.org/50/683350/2/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020-rocky/ea20688/ : SUCCESS in 4h 35m 42s","accounts_in_message":[],"_revision_number":2},{"id":"49aef6d2e04af76e79cd0746cd572b60c395cee2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 15:50:04.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/d270dabb8564468dbc118ace303e6155 : SUCCESS in 2h 07m 14s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/55936fb6251644799cd93793a1321fec : FAILURE in 2h 45m 35s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/8ee490b4e36c4e1dbb7c71e6c3e25725 : TIMED_OUT in 3h 38m 47s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/2ca1315ef54744d7bea069bf8b4c8588 : SUCCESS in 3h 56m 42s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/3e3473b162f64128a96b29571fde7fa5 : FAILURE in 2h 55m 12s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/1698c7ed05994065a26db702d9e608b3 : SUCCESS in 3h 03m 30s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/88f6c416d6164d20b3b76ea6597582b3 : SUCCESS in 3h 42m 22s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/b55ddf40b5154fdb987df58594da7066 : SUCCESS in 2h 52m 00s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/c9fe05bd578842699343010a07bc6a64 : FAILURE in 2h 35m 09s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/551a9db4a32246648eba435cf56ce8b6 : SUCCESS in 3h 02m 09s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/4e9c8852263344dca56dd88bc2ede975 : SUCCESS in 2h 45m 57s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/893a6cf7e13046258c5c07b756ab42b7 : SUCCESS in 1h 50m 20s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/e34d4f1baf9948e5b05388bf1df7b1bc : SUCCESS in 2h 14m 59s\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/a11cb224fa4942ee845e648941cf6309 : SUCCESS in 1h 46m 47s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/834ca6911c9a4fc08ec9fc268dab689e : SUCCESS in 2h 59m 49s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/dc0039d775c544dda2e4d124e5b0ce11 : SUCCESS in 3h 06m 58s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/8343a9fe2eeb4bb688869e9a8c514d63 : SUCCESS in 2h 25m 23s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/92f23c9ee87c4d0f901a34a7cafda38f : SUCCESS in 1h 55m 47s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/116544c05a2d440c948b71cf13c3e3ee : FAILURE in 2h 19m 22s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e45025e75041420ab58556d498217675 : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":2},{"id":"765a39ad881c183e4fe60542f95536dca3af4473","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2019-09-26 09:05:47.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"bf322477b367144308f2bcdad13f6670947de283","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-26 13:36:05.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/50/683350/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/f5dc54c/ : FAILURE in 3h 06m 08s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/50/683350/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/2cc8784/ : FAILURE in 1h 56m 17s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/50/683350/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/4997ac5/ : SUCCESS in 2h 33m 54s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/50/683350/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/de33fb1/ : SUCCESS in 2h 56m 40s\n- tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020-rocky http://logs.rdoproject.org/50/683350/2/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020-rocky/6b6a0f0/ : SUCCESS in 4h 28m 31s","accounts_in_message":[],"_revision_number":2},{"id":"4ae90e73d2e0bf9ef04974027e5a184390c7d83c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 17:06:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/f45e85de049e44a0b64237d95be3ee96 : SUCCESS in 2h 58m 36s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/7ba5b4a468bc4cfb8a47df1a03bb0b1b : FAILURE in 1h 18m 30s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/f240f61f7baa4fbfb3f3c788ceadad34 : SUCCESS in 3h 38m 42s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/239339021a264104a7fd12c114fee56e : TIMED_OUT in 3h 48m 35s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/ce352bd4d03b4933be9b6b4a2597a98a : FAILURE in 2h 29m 58s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/417f5af68b31455797b80488209e8a8c : SUCCESS in 3h 42m 12s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/60f5131e51d54640bc2ce083e0e3d9be : SUCCESS in 3h 43m 43s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/df4af6b768c44ab1a06d207c7372bbc7 : SUCCESS in 2h 57m 16s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/7ef6c22ac774424eabf72ff141d3faff : FAILURE in 2h 29m 27s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/3b398d6687fe431e920aa5e8fa1ba2ee : SUCCESS in 2h 30m 08s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/408c95df31d8437d96c385a6537fb798 : SUCCESS in 2h 31m 33s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/1056c28d35c148c7bc3a807ff9ba9600 : SUCCESS in 2h 04m 48s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/a26d0c90e7ec4f228822eb1eb4742d7b : SUCCESS in 2h 38m 22s\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/9c88dded06db4634b6cd2c1d291b8c10 : SUCCESS in 1h 55m 00s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/20d30e508e7349fab56da0f6c9beaa00 : SUCCESS in 3h 17m 50s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/29db59a1c2674ac0aa0960b5075fe9df : SUCCESS in 2h 39m 57s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/72dddd6afd45442baca0cefb0e4eac7f : SUCCESS in 2h 33m 52s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/73392b86dd9342e9adf360297f1857da : SUCCESS in 2h 00m 57s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/2738225039234e28886c28c716e8c720 : FAILURE in 2h 46m 27s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f1b61273b14e4fc384f9cc1d75ce28fe : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":2},{"id":"693cbddb9465b962a5c59a524c9e9f77b1c47852","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-09-26 17:34:17.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"795bbb2a81f060d68ad4c0f98d822e508116d867","author":{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},"date":"2019-09-26 17:54:38.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d1ee29be3d98f44b9bd48f5f83d723ddaf3955c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 17:56:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3effc7f73db3b793b6f1cb6399dffc58137024e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 04:26:12.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/20206069c8834de3910e25c6f5450772 : SUCCESS in 2h 24m 33s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/7bd694c8f9d245bc999dd49db7bb7039 : TIMED_OUT in 3h 35m 44s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/a2805847c71840fe9a46d8f0ee8894b8 : SUCCESS in 2h 37m 51s\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/69b1ec95717b48f39240d444442ec428 : SUCCESS in 2h 54m 05s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/8c46d567116a4854bd25a0d140014836 : SUCCESS in 3h 25m 12s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/205e05a4018e433f8d5c499f57e896e2 : SUCCESS in 2h 55m 25s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/cf82400b746c4b41a87c2053e2141fd9 : TIMED_OUT in 3h 49m 10s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/fcffb16d405740879c58eca875297539 : SUCCESS in 3h 06m 11s\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/8bf8fb3d787649c597f70887c4ab2821 : SUCCESS in 3h 35m 25s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/fa4bd0c0953d45368facd170213eb5ef : SUCCESS in 3h 22m 38s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/713d79d864da451d874767423c73614d : SUCCESS in 2h 10m 07s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/ce1eb3231c564842b945ab58c95d2184 : SUCCESS in 3h 02m 41s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/7796619e70fc423ea12b07a63250770a : SUCCESS in 2h 08m 42s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b791e432c63d4644a710af9f1944cf9a : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":2},{"id":"bf811d798b2425f36ef02551d8e4f3e438c599a7","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2019-09-27 13:25:36.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c46177d62dd02c2036f22156894c9a8680b8a28b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-27 16:29:24.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/50/683350/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/c46e29c/ : FAILURE in 3h 00m 33s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/50/683350/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/bd67a78/ : FAILURE in 1h 52m 03s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/50/683350/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/131fbd2/ : FAILURE in 37m 39s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/50/683350/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/68bb8b7/ : SUCCESS in 3h 02m 30s\n- tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020-rocky http://logs.rdoproject.org/50/683350/2/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020-rocky/19542f2/ : FAILURE in 37m 29s","accounts_in_message":[],"_revision_number":2},{"id":"87038d38909e118bb28622d5f78ee1d82b8b3208","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 20:53:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/89e6d280cc7e487d967a436275f4cd13 : SUCCESS in 3h 06m 24s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/5035569ac8e442eab06b71b4b8fe8492 : FAILURE in 2h 34m 28s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/b44cd00206f446c5b5b9c3dbebf320f2 : TIMED_OUT in 3h 59m 51s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/3fa6572f8ff941dc80d47036c258525e : SUCCESS in 3h 30m 12s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/8393712a5fbd44baa2ac43678fb81f95 : FAILURE in 3h 06m 38s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/01d8f5516e544bb29e39d318edd0da3e : SUCCESS in 3h 21m 55s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/16e6230b9bda43da8c55b626fcae32e0 : SUCCESS in 3h 25m 35s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/dd9b368c01134166a416a3b508295301 : SUCCESS in 2h 46m 20s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/72b7623f0ccd4923a28f735826a28392 : FAILURE in 2h 17m 39s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/a607a0e7fed54391b800488d65345c40 : SUCCESS in 3h 31m 53s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/f367d8058f184e02adfee25ec113bbfd : SUCCESS in 3h 09m 12s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/17a0112a5a80440d8c19529136b8e3ff : SUCCESS in 2h 35m 30s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/2848d4e96ec24efd91433d2c54e2d22c : SUCCESS in 2h 59m 38s\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/bad410203e0d4abd818e95c9da6f5e6d : SUCCESS in 2h 18m 24s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/29ef73519c604bf0bbccbeac9cc1b0c8 : SUCCESS in 3h 12m 05s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/729c65e814df404782bc1643531354d5 : FAILURE in 55m 47s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/21599f5f5186453990c3e9253861649a : SUCCESS in 2h 35m 54s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/e94cdbe9efa24f219d91321114081673 : SUCCESS in 2h 19m 16s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/102a5032ac7845a596dba03240abd0be : FAILURE in 2h 24m 34s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fa207a85fd63484bba2254fb0fc1014d : SUCCESS in 5m 28s","accounts_in_message":[],"_revision_number":2},{"id":"2afc40e2cdf82d773ed364de53881d70d3e656d8","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2019-09-28 10:31:47.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"b1c59376171b9657bd33ceb59e452e080e07af9c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-28 13:41:28.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/50/683350/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/8333187/ : SUCCESS in 3h 08m 57s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/50/683350/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/7f210e0/ : FAILURE in 1h 46m 05s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/50/683350/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/053f8b6/ : FAILURE in 35m 10s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/50/683350/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/8bc3fee/ : SUCCESS in 2h 50m 41s\n- tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020-rocky http://logs.rdoproject.org/50/683350/2/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020-rocky/4258dbd/ : FAILURE in 34m 13s","accounts_in_message":[],"_revision_number":2},{"id":"73dd35b5ea4af11b1b10d930d0e9505fb11f5f42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-28 14:38:55.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/f0b27bfab2114a829ad475bba35d04d4 : SUCCESS in 3h 36m 17s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/4d80245e636a4505bc96473d5dbc3a9a : FAILURE in 2h 02m 57s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/39f266aa78b34bdab4e1ec3878a31dd7 : SUCCESS in 3h 14m 06s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/7a27bdbfa67b49d39f0d1f259d690728 : SUCCESS in 3h 11m 06s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/02965ed650154300a6096225ce0b997f : FAILURE in 2h 24m 30s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/045a55db48454b20a2b43f04c7ae6a50 : TIMED_OUT in 4h 03m 10s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/0f2d03f7ca154faeb9a39984ab49f428 : SUCCESS in 2h 55m 26s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/465cfa30d81c4b7a950c8bf291ff09c4 : SUCCESS in 2h 39m 58s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/7e6079958b3848e9b414fd80757d3c4b : FAILURE in 2h 01m 38s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/937110beb9124f8cae87bb26d9f74c4d : SUCCESS in 3h 38m 04s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/1658b89a711b404bb5107c78707b3e29 : SUCCESS in 2h 13m 34s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/63a3482743f347adb91b6e3515444cf2 : SUCCESS in 2h 49m 28s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/cff7b8d740974819a0f80a35f30554a0 : SUCCESS in 3h 41m 56s\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/b55a166febec436394d709baa65a769e : SUCCESS in 1h 42m 26s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/a093d677b4514bdc9a9284944e2b8146 : SUCCESS in 3h 40m 40s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/56d96650e2304cad99371fd3d8ed63c8 : FAILURE in 51m 48s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/d4db6d4fbb35402191a66b96b3a76e81 : SUCCESS in 2h 40m 22s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/a31151a945fa4e5e8ba17d04281e3ff1 : SUCCESS in 1h 43m 55s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/de4de94f336a464b8cec9e6d8c47a110 : FAILURE in 3h 27m 26s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9aa6090ff02f48aaa49ca79b07ff00fa : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":2},{"id":"262536279002109bcccd2b3544eb68ba3edb3705","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2019-09-28 15:03:14.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f1c74db446f0c0382bb2c4baef1b2b54343a807e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-28 17:59:10.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/50/683350/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/787d45d/ : SUCCESS in 2h 54m 19s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/50/683350/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/57cf286/ : FAILURE in 1h 47m 14s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/50/683350/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/c48ef1f/ : FAILURE in 34m 15s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/50/683350/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/458245c/ : SUCCESS in 2h 53m 06s\n- tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020-rocky http://logs.rdoproject.org/50/683350/2/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020-rocky/d487ac1/ : FAILURE in 36m 51s","accounts_in_message":[],"_revision_number":2},{"id":"0232acc52024ffbdc3f1a3a2193021471c914ec1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-28 19:15:37.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/f6c73cad914a40239fdbea28ed708e9e : SUCCESS in 2h 24m 09s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/1660623a402d4a5d97c8ec47a9b93473 : FAILURE in 1h 52m 48s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/b979bc563bd04c76a94f751b08ca321e : SUCCESS in 3h 32m 25s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/c9eebf71805a49869c007582ed4482a7 : SUCCESS in 3h 27m 36s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/7598d4f28db5431c9029cc355792d881 : FAILURE in 2h 30m 08s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/d03dd704f4214d639caab8d3c603b379 : SUCCESS in 3h 57m 11s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/c3e5a67b9b1c43bda56dc09168d4cb3e : SUCCESS in 2h 42m 07s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/3cee3669d15b413ab3925e9a5c4ff22b : SUCCESS in 2h 20m 19s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/509c740084aa48a39c778d7898d5018f : FAILURE in 1h 54m 04s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/8f5d3fce339148fe85c4014dc8704eb5 : SUCCESS in 2h 46m 19s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/c85d8feb4f694d85b4e09a79bf340a5b : SUCCESS in 2h 34m 11s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/b21621965bd6472a9392c8a90a80ad45 : SUCCESS in 2h 06m 00s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/fc8f10f6a16041c9a32ff0fef14a9ca3 : SUCCESS in 2h 34m 41s\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/183624f53f0d4cc68f41583223c4efa7 : SUCCESS in 1h 56m 28s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/81ab6f10792d4091ae4c1a8a6bc9cea8 : SUCCESS in 3h 26m 00s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/364e69058f254de798be4c25905ddd9c : FAILURE in 54m 15s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/82e13686097d4a949b57ad8e99320b25 : SUCCESS in 2h 26m 28s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/5f9212af01fd43d78167e0d0c484c852 : SUCCESS in 2h 04m 01s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/fc303d8c5f03488abe970e72062b094b : FAILURE in 2h 16m 56s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/edb868c9ec1b408783fd0cebbd91b91e : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":2},{"id":"03ed3196a0a31a3a8fd17945e154251422333202","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2019-10-01 07:28:47.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"f049e8653a5c464d97dbc0f9d80bc96b9243fba0","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2019-10-01 07:29:08.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"963fb546ef46d4c3bd74324ebbf5c265ffde9c26","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-01 10:20:22.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/50/683350/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/356f153/ : SUCCESS in 2h 49m 52s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/50/683350/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/9637360/ : FAILURE in 1h 55m 38s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/50/683350/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/70ce7e9/ : FAILURE in 36m 27s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/50/683350/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/fb5dd5f/ : SUCCESS in 2h 46m 39s\n- tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020-rocky http://logs.rdoproject.org/50/683350/3/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020-rocky/8b815c7/ : FAILURE in 35m 48s","accounts_in_message":[],"_revision_number":3},{"id":"abda9e77bea3948117047831f5b250ac69357c73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 13:48:56.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/b860c136110f4fcda299bce27d9185c2 : SUCCESS in 1h 46m 43s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/959c8c5954d040ba838690b63b3d8847 : FAILURE in 2h 04m 09s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/12ba1ff4686e4adea02e62133cef8fd0 : SUCCESS in 3h 36m 00s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/4570d3809d4f4c06962453ca6c4e9f58 : TIMED_OUT in 3h 45m 22s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/83fa138e8cce4edcb99a70687db7a75c : FAILURE in 2h 10m 29s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/376e8723aac44c5880f7dc69fcb63026 : SUCCESS in 2h 42m 17s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/24e30e20a6094ab3b6ee5c9d148b350e : SUCCESS in 2h 41m 32s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/a4ebabff08f94ac7a74f4c5df166a722 : SUCCESS in 2h 36m 34s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/b39d3c1088e14622b0229f403e088cba : FAILURE in 1h 58m 37s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/e120c1b7e8dd4a489ce43f3da93520bc : SUCCESS in 2h 30m 04s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/75c3bb63f33441b39e3aac3ce2805743 : SUCCESS in 2h 10m 59s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/992e66d1b117420a9e0766ffe98ba0db : SUCCESS in 1h 40m 36s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/5f76824d34014371bcc34243f7a2ba1f : SUCCESS in 2h 21m 13s\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/7f3f39e6365f46f7b675d626cd9ea776 : SUCCESS in 1h 38m 52s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/e78b43a27b8c4dbdb84e7d407705d009 : SUCCESS in 3h 18m 38s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/5dd4ac275de84135bc412468ff67daee : FAILURE in 52m 27s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/4a9d1c70ad47405793f6fff674675586 : SUCCESS in 2h 28m 56s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/4955cc8d21fc4cf0819d712ba2767265 : SUCCESS in 1h 46m 48s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/6a86324e630a47e6b68c0a61f6361e7f : FAILURE in 1h 19m 40s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/25c2dc05e59546fa9b1a0ecc578b4a9d : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":3},{"id":"4ba126ad26563a6820b9454b7be8608d60121903","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2019-10-02 08:13:53.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"721abf0858dd6614c950b03779d07ba70a04637c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 12:03:10.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/fcf900bf426343a7a1daea725ec49738 : SUCCESS in 2h 22m 29s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/11bdc95c8e7a45d081e9d5d8fc211f43 : FAILURE in 1h 57m 42s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/4f841c0e9da94c59a572694640d15854 : SUCCESS in 3h 30m 25s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/b62fbdd2bf3f475caf023783a017e5ef : SUCCESS in 3h 14m 21s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/90d7eb24f8e34fab8df5699ebafb9ff0 : FAILURE in 2h 35m 25s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/550f8f05040a440c8630c314d81dc536 : SUCCESS in 2h 48m 54s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/efd4bf4ac3a7442b9b4a497937ecfc44 : SUCCESS in 2h 52m 07s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/c0f88f71569e4c409bcd3ba0de4db960 : TIMED_OUT in 3h 24m 57s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/59334fdb5e8443a3976dd700561e3c11 : FAILURE in 1h 44m 26s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/6e7bf7fd9d7e4f32ae2b8b28c519801c : FAILURE in 2h 46m 44s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/bb3b27444f224a2d8c0aa8e96acfd5bf : SUCCESS in 2h 13m 31s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/5f3b259ba8b146c1a6ed1c92e410d383 : SUCCESS in 1h 42m 04s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/c28a21f068f64238a0494ddee1170669 : SUCCESS in 2h 33m 05s\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/2a4ff5fbe263459e8dbb1072d9c84ab4 : TIMED_OUT in 3h 29m 58s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/3ed13208f47c4beaad0e5ab422586333 : SUCCESS in 3h 08m 35s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/0bca925b86a049d8b7b4084032f1d4cf : SUCCESS in 3h 15m 02s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/7ff9ba647c844f41911943d7555276e1 : SUCCESS in 2h 22m 30s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/aa01f4e281c84b1d8d4e3f6ceb74085c : FAILURE in 1h 52m 05s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/e4eee2dd42df42c288a631bb992a25a9 : FAILURE in 1h 33m 40s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b130e11b0d8c49bb97142e3702f08f9a : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":3},{"id":"3f9e4236acd752bda0024164c013e342b056a475","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-02 12:27:45.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/50/683350/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/c22ed35/ : SUCCESS in 2h 32m 38s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/50/683350/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/6e11469/ : FAILURE in 1h 58m 48s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/50/683350/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/123525f/ : SUCCESS in 2h 18m 06s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/50/683350/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/4f5f6bd/ : SUCCESS in 2h 59m 09s\n- tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020-rocky http://logs.rdoproject.org/50/683350/3/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020-rocky/43f2737/ : SUCCESS in 4h 12m 27s","accounts_in_message":[],"_revision_number":3},{"id":"5c39c70b9cc3efcf554c658e0acf5ae387e1ba0b","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2019-10-02 12:57:48.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"8ddd920dc1fee8bcb2dd5a00d9255bd00883264c","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2019-10-02 14:23:22.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"fa5eeac329f1ae49f6968ca88f5706be34db7970","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 19:00:04.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/4fe3178a624645dfbb8e6bfa61779363 : SUCCESS in 1h 57m 53s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/b8755677dbde4d54806f936aec6a7095 : FAILURE in 1h 55m 16s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/91c87e63f8ae48eda15eb723e84fb48c : SUCCESS in 2h 58m 12s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/f65f00af3b244ea7aaf7c86f52602f28 : SUCCESS in 3h 00m 47s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/78fcc0fcf9444613b7e7f7efc318cb46 : FAILURE in 2h 21m 36s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/449487002db541abad1301546aee02d9 : SUCCESS in 2h 34m 35s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/fa89ab0bc7de4786bd5f205ef105d90a : SUCCESS in 2h 40m 13s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/f6e17ba2dbc548de8722d06c7113bdb8 : SUCCESS in 2h 23m 46s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/e6b69e9dda5f41e28fe7e58aafab165a : FAILURE in 2h 04m 34s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/e4d84c762fe54c3b864ea6eabf774d93 : SUCCESS in 3h 01m 08s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/5a10c395b6ed4ddcac929ffa8ecb20c4 : SUCCESS in 2h 01m 22s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/efd0bc7663de4ab79414c6dd08b9aaa3 : SUCCESS in 1h 43m 10s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/781f89b95fc64c8e86f45aaa3578dde7 : SUCCESS in 2h 13m 06s\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/663842c3cb3b49bfa0948c67c2a64080 : SUCCESS in 1h 39m 28s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/b32d761691d74147b94736fc8ec0c02f : SUCCESS in 3h 14m 16s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/7efed0cb51c64329bb34c84de16d4ed2 : SUCCESS in 2h 59m 45s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/68671f86603142d4a8918c90b6532cf3 : SUCCESS in 2h 37m 15s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/5592c6b5de814ee1a4083686bd5b8fcc : FAILURE in 1h 27m 47s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/e1e53e0df73d45c8800c87f6169660d0 : FAILURE in 1h 11m 06s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/485f35d8b2ee482cb4137788ca7d7773 : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":3},{"id":"1ba099744b6d630162bda719994e19933ee48155","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2019-10-03 11:34:33.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"982f0f71d5bffa333c7e6a6020c9739bb7576f19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 15:35:35.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/3cdf893524ef41eabc2c10319a82f2f2 : SUCCESS in 2h 07m 30s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/17b3657b1abe40ef86ef88e30956bb27 : FAILURE in 2h 06m 21s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/c32ac118d68947c5815cfa9d5eb3a37b : SUCCESS in 3h 00m 25s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/3f9d709ef93940cf9c4c93da7489933d : SUCCESS in 2h 45m 47s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/1b3a57891b24465292a5baaaf1b7a60e : FAILURE in 2h 24m 10s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/d381c5551272424a977b8fc18984917c : SUCCESS in 2h 38m 16s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/4e9f5257e46c47d78da18138e5e55889 : SUCCESS in 2h 50m 36s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/2ac0d94f3f594039a179c606aa8ac7e2 : SUCCESS in 2h 14m 40s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/4063c18527bb49cdbbd9945ba15d2be1 : FAILURE in 1h 50m 06s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/629b5e679d0b4ec0b663f20239a5ad14 : TIMED_OUT in 3h 56m 20s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/740b95f989574c1c8627f45fd3ddecc1 : SUCCESS in 3h 51m 19s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/4791f3ff25b4459db77f74b6868e9cfd : SUCCESS in 2h 56m 37s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/9aceadfe023d43598a399e48b1245ad6 : TIMED_OUT in 3h 44m 18s\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/da45a5d86b88448fbabcc78a5920c379 : SUCCESS in 1h 36m 07s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/29058f14a1884d16b338c11374e8a98d : SUCCESS in 2h 50m 08s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/3e1a5dce709a408cb57dbb929f32b53a : SUCCESS in 2h 47m 49s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/35334c7bf63845a4a5b4366ee0ce414d : SUCCESS in 2h 37m 08s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/dc366d4abca04ae3a0f0f8c12a9ce4e5 : FAILURE in 1h 36m 18s (non-voting)\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/e9b159a097ff413a8247a1842d722415 : FAILURE in 1h 13m 18s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/18672d45d104409b9c262332fae919d9 : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":3},{"id":"59e24a18dbf25ab2f12a7e3c7c29574ef9a0c196","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2019-10-03 16:21:02.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"5d92ca20a375f228b46a0b71f8b7e884c6864c67","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-03 16:28:37.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/50/683350/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/5b654ae/ : TIMED_OUT in 4h 37m 15s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/50/683350/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/1f5195d/ : FAILURE in 1h 51m 39s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/50/683350/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/d1ca8e3/ : SUCCESS in 2h 18m 07s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/50/683350/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/05fceac/ : SUCCESS in 2h 59m 41s\n- tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020-rocky http://logs.rdoproject.org/50/683350/3/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020-rocky/cf9e016/ : SUCCESS in 4h 52m 38s","accounts_in_message":[],"_revision_number":3},{"id":"872e37bcc78d7c7311ea1849af0af5e8cac59f05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 20:57:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/0bd00fddba1d442388e3a95a443ecbc0 : SUCCESS in 1h 53m 53s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/0d5bc63c1ceb4b50acca4943fccf097c : FAILURE in 2h 15m 43s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/87d75489fe804b9cb732b2e0b3e17e2e : SUCCESS in 3h 07m 33s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/f7fe9b45d30444438bff5b15e6dc46f3 : SUCCESS in 3h 28m 44s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/b47da7a3e80d431baa0a5d6bcb563b4f : FAILURE in 2h 27m 10s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/823865a42f834e9eac7c6c96f651f9fe : SUCCESS in 2h 29m 54s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/c22af7ef5f16488190f09443dc16938e : SUCCESS in 2h 58m 20s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/0b82b4231c3747408182dd6117e2a272 : SUCCESS in 2h 16m 48s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/7b91ca496234484e98fbe9d3c2a17353 : FAILURE in 2h 01m 03s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/b4b2a39207f2479aaecb9e8ce1b10913 : SUCCESS in 2h 41m 27s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/b6d233fe1b8a48eaa334dc09b4ffc4ad : SUCCESS in 1h 40m 39s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/23e98a0183914768a5e59fc3babd448a : SUCCESS in 1h 55m 06s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/32e0f0cf527d4b31bd1ae0964d4aa6ae : SUCCESS in 2h 28m 41s\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/bb95686e2fd94bc796053b267fe381a9 : SUCCESS in 2h 00m 10s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/427fa13488eb4852a2490d5d3b93b2b7 : SUCCESS in 2h 54m 17s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/213e1210b42b4a39a102a76bf118cc1d : SUCCESS in 3h 11m 21s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/a3d2ef98770241beb8ec8c65bc391f07 : SUCCESS in 2h 23m 59s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/30e8d6240a4645f3b6f94ba77736776d : FAILURE in 1h 51m 56s (non-voting)\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/c5b0db7b13b445a986fda4ac3eb18cbe : FAILURE in 1h 32m 07s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7f1957d1b8e745f8b1e222d3aa451512 : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":3},{"id":"f7c3379075f9e418aade8405f24faf227643c989","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 20:57:31.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"420dc05dee69a46e2df04bb4de0629455c534e80","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 00:33:03.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/1ff8d86be5fd4af1b663ca525ed4073d : SUCCESS in 2h 02m 53s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/306ee64472fd45ea8e9ac45dd43d89a3 : FAILURE in 3h 31m 11s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/61e8735f92c64c8cb0978659288eca18 : SUCCESS in 2h 15m 50s\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/a2d7aa34b69f45a4a683d97ba2b120f3 : SUCCESS in 1h 24m 21s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/1b438ff1c8484bef85c91eb435a3f892 : SUCCESS in 2h 29m 22s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/f9daee000b8b467ea93ef31d79161000 : SUCCESS in 3h 06m 07s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/bf07b532f3d54b2da4bb19e9e85444bb : SUCCESS in 3h 19m 13s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/b8289312f9df4a98aeb7e0b0106f50eb : SUCCESS in 2h 33m 02s\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/cf1ee05e877b42cda5fb3f5f519c78cd : SUCCESS in 2h 18m 48s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/279401d394d54e5abfb0c77a62bae9ee : SUCCESS in 2h 17m 08s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/f503ff54316d4e4992e058d73d06e5ff : SUCCESS in 2h 01m 45s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/67a87bde7a68418699b92391ed29a1c5 : SUCCESS in 2h 41m 41s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f426a205faef4dde923ffeaf8df1177f : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":3},{"id":"2b8d603645250ce5f1ec4615a07b4a0df6d23885","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2019-10-04 08:30:37.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"51b5e953896b5827790a8155e4b5b904878420d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 11:44:10.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/f23ab7b7073b435890afe3b0e6871051 : SUCCESS in 2h 01m 42s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/392092874c554ae199c9905e5e291c70 : FAILURE in 2h 10m 06s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/716321b631ca43c9aeb6a4babfe413dc : SUCCESS in 3h 11m 03s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/b7602499c8544497a733b8142f474279 : SUCCESS in 2h 59m 03s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/7756d925212a4bacb2a487352b704597 : FAILURE in 2h 06m 16s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/fbe96dfbf611412397d1ba16c7d7f46c : FAILURE in 40m 16s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/6600e037d3674595bafa85a4ced88c91 : SUCCESS in 2h 42m 52s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/82114eceb5ff4a2d8b60a8c0e069d644 : SUCCESS in 2h 17m 05s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/8dac60d13c324e50a057b4eeec934ddb : FAILURE in 1h 48m 22s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/cca44d012a144626b7eadca6c39a97b2 : SUCCESS in 2h 43m 45s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/803b1d9ec72842adb2a865dbd9d81e4e : SUCCESS in 2h 13m 05s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/c88b7388c8ee4586ab66eb9b5773905b : SUCCESS in 1h 29m 34s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/d3d172e492d84c62a3e84379612dd3ed : SUCCESS in 2h 26m 04s\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/340ea1078f9b4ab6978e9481acdf94aa : SUCCESS in 1h 53m 02s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/d4c874e755ee4912ac1fd788e98daca0 : SUCCESS in 2h 40m 20s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/54935a74d4264d0e91a6f65bafd6167e : SUCCESS in 2h 46m 33s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/baef19ae2a9d48e688c57871d5764c83 : SUCCESS in 2h 35m 09s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/6ed2cffa9e8b464aae445560dee88d96 : FAILURE in 1h 36m 13s (non-voting)\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/cfb80a85009a4aaba8201261831e34bd : FAILURE in 1h 21m 37s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/65f8feb0b2844a57b4732fc5bd943291 : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":3},{"id":"f70b30a23c7e2f4e339130dd8f17c2c0369e6278","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-04 12:45:08.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/50/683350/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/b03c778/ : FAILURE in 2h 50m 41s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/50/683350/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/8007d4d/ : FAILURE in 1h 38m 09s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/50/683350/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/e8beb78/ : SUCCESS in 2h 19m 21s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/50/683350/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/bb412a2/ : SUCCESS in 2h 50m 58s\n- tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020-rocky http://logs.rdoproject.org/50/683350/3/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020-rocky/df29090/ : SUCCESS in 4h 13m 11s","accounts_in_message":[],"_revision_number":3},{"id":"96e87591056d66d0d4980f53c86b55bc0ee96b6f","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2019-10-04 18:24:31.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"597b42a3639b699815796359026bca49056dcd57","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-04 22:53:50.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/50/683350/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/54d2c40/ : SUCCESS in 2h 42m 21s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/50/683350/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/7ad331a/ : FAILURE in 1h 47m 39s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/50/683350/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/104003d/ : SUCCESS in 2h 22m 38s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/50/683350/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/1f7877b/ : SUCCESS in 2h 52m 59s\n- tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020-rocky http://logs.rdoproject.org/50/683350/3/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020-rocky/523e759/ : SUCCESS in 4h 06m 45s","accounts_in_message":[],"_revision_number":3},{"id":"146d865ac00af56f75498d883b47ab6551a172d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-05 03:47:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/cfd6c6d3d61944bbac838040760f50f7 : SUCCESS in 1h 53m 10s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/eecfc2245caf491bb8c64382fc6b945a : FAILURE in 1h 39m 12s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/56b5c8e13a90407198a760b5c4dca87b : SUCCESS in 3h 08m 09s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/3407b2e02ca64a8e8644cd6d20ca6141 : SUCCESS in 2h 40m 55s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/2b62aece040e48698e5ac66f713c63cd : FAILURE in 2h 14m 55s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/2116677c97474ab3ac3212c57798440c : SUCCESS in 2h 36m 38s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/756fbc83ae624bcdadc8707a1e674d6e : SUCCESS in 2h 51m 41s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/d8d7f9ac2c4c472bbf8e769c199772a8 : SUCCESS in 2h 09m 50s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/5cb4286107a44a7db70183dca6740070 : FAILURE in 1h 33m 49s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/69be116d66174cd188a918783676fa92 : SUCCESS in 2h 26m 18s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/c748813c7037402e94b037d90ecf15ea : SUCCESS in 2h 05m 28s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/cdc2edbd32064785bf528b8ce63e6078 : SUCCESS in 1h 43m 04s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/0e6501734028417898760fa6bfea815d : SUCCESS in 2h 16m 43s\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/de9ddd55184e4f15a27d8fe10780ad36 : SUCCESS in 1h 28m 49s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/8738e5f63a5b4171bff3d41d6fe4078f : SUCCESS in 2h 45m 13s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/6675839409b3499291686165a5b35e87 : SUCCESS in 2h 53m 34s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/6aef120d28a542e69d7316be91f91909 : SUCCESS in 2h 28m 42s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/e9a7ead448d242f798d4c885c49007c0 : FAILURE in 1h 41m 57s (non-voting)\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/805a735c9fb2427b9c3a2225614f1033 : FAILURE in 1h 16m 31s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b23adbaddc094c16964b43d9dba2b498 : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":3},{"id":"84dd6994b8cc58f8bde07c5396a5b3b043e4de72","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-05 03:47:39.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3d4519eddddd494792c76ebd037102bde0b000dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-05 06:53:59.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/c251d05398b141bb87de46266c7edfb8 : SUCCESS in 1h 56m 44s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/f8e067f9504349819697a9beec1038a9 : SUCCESS in 2h 49m 23s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/801ce804a22d47559a13bbbf4d3ac231 : SUCCESS in 2h 20m 37s\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/74140ed2fc6c4f8a8d8c45a9e9ff3d37 : SUCCESS in 1h 35m 40s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/50c73f82ca3a4f9994517ebe8168ce44 : SUCCESS in 2h 28m 50s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/6920c8db742c4d4b85c8eaec164578a4 : SUCCESS in 2h 49m 06s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/10d953e7763446d49231373d5ccf8f49 : SUCCESS in 2h 47m 46s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/9a1bc02dd3794333a2add592edb86b6b : SUCCESS in 2h 21m 51s\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/07690dc567b749068a9a8ff8506026f3 : SUCCESS in 2h 36m 03s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/d45d9840097c4d18b76be95fe6a90b9f : SUCCESS in 1h 50m 06s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/f6a34fe5cd4d4f0988fc203402ae484a : SUCCESS in 1h 47m 59s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/c8e095dd8a694afabf569264ee1433db : FAILURE in 2h 31m 45s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b0b678d3e8a444fe9e6eaa7664a0142a : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":3},{"id":"71975ad8fa4f01b1ca75d4f0b966369111fb5763","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2019-10-05 16:28:39.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"4510d17f9e832419e3e2414f219924c4626c44c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-05 19:57:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/56eca1551d254dcc8e8b9200a46b26ab : SUCCESS in 2h 10m 54s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/ee44caa1a076473da2506f320d81d054 : FAILURE in 1h 44m 29s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/8d55bf663ed34faa939f8e8f105aa032 : SUCCESS in 2h 50m 48s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/5ce009d9bd244fdfadd47eb3f4602d15 : SUCCESS in 2h 39m 08s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/abc8ff8656484719a85df0be16bc5e47 : FAILURE in 1h 58m 28s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/bb02b5aafa9f4d9d94ec8fee683a1ae0 : SUCCESS in 2h 35m 31s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/9657f4bec9284fc7966899ea302824ed : POST_FAILURE in 3h 16m 51s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/96ff772f90e74f54b9cb53d2f8c6e246 : SUCCESS in 2h 07m 11s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/099080915b854dfb90b9421ade8d9ebb : FAILURE in 1h 53m 00s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/5f8ca57ab9d0434399f8dfecc7e67d71 : SUCCESS in 2h 40m 57s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/31c2baece8d441c6b64a624d4f7dbccd : SUCCESS in 2h 11m 10s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/fb105ee2130444329fff04cdfee19efa : SUCCESS in 1h 21m 08s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/418d9546d07744a7ae3e115007cded62 : SUCCESS in 2h 24m 05s\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/6a4eca2a60a64e978372359207a7c470 : SUCCESS in 1h 29m 06s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/b5199f6e84db426d963ccb8d3ec722e4 : SUCCESS in 3h 02m 51s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/3c93576fed84485b8fe5ef267b9a4fbd : SUCCESS in 2h 22m 05s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/aed181e5b3854772997e069e6d231c45 : SUCCESS in 2h 16m 34s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/72f25f88434d490ea687a2f0165169f6 : FAILURE in 1h 28m 44s (non-voting)\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/f3aa1b9b0d884087a2e3775b7d757526 : FAILURE in 1h 22m 15s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d516f2871a6a4af39ffaadbf8322e49d : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":3},{"id":"472da9c8c3e91bf257ac564b6a57a4c67ff53b44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-05 19:57:54.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"608dafc0d6228f6c701abc4d4d2cb884e0a43fad","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-05 20:28:45.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/50/683350/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/05e6c02/ : SUCCESS in 2h 51m 02s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/50/683350/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/29e742a/ : FAILURE in 1h 38m 03s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/50/683350/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/a5d1f2c/ : SUCCESS in 2h 36m 06s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/50/683350/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/4ac48b8/ : SUCCESS in 2h 58m 33s\n- tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020-rocky http://logs.rdoproject.org/50/683350/3/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020-rocky/eb682fc/ : FAILURE in 3h 58m 42s","accounts_in_message":[],"_revision_number":3},{"id":"7de166a366f3f93de7a91a0cd1162b3c5d0069a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-05 23:09:10.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/2e9d2b8cda5440bea08ef05588010fe7 : SUCCESS in 1h 49m 31s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/3ddec76d68c245c4b92d27b1fec1b64f : SUCCESS in 2h 53m 14s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/588e168726134184a41e7d7768a3f195 : SUCCESS in 2h 09m 31s\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/15df54db8ebe447197def16ba50a66b1 : SUCCESS in 1h 35m 06s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/75bcac38f2ab4549b57268ec5affd3d6 : SUCCESS in 2h 35m 38s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/6ab1c4b39c0a4d81bcc5484ac0d9eccc : SUCCESS in 2h 51m 02s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/f66543dbe3fb45498c059606587a7f69 : SUCCESS in 2h 22m 51s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/9f6c17d7426343b9873ae42d8bf67b0d : SUCCESS in 2h 11m 23s\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/51f6d931bfaa42f1bdbf6bc43890f43b : SUCCESS in 2h 38m 35s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/153883fc9a2d4434a3ee6a9e09e02daf : SUCCESS in 1h 41m 30s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/0bbcf54e7251415bb788867210304aaa : SUCCESS in 1h 29m 08s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/da76113dc06d4c899d95fd3fb1310a37 : SUCCESS in 1h 58m 46s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/657862bb032247069cd84b7ad5e05945 : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":3},{"id":"fe20afb0acb2cc4ff0f290fff6d5ef9a96f4920f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-05 23:09:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"62babbfe379dad4dbc9a87c216f236e3a70d711b","revisions":{"a3ed0ea252c4a7f445ee9905102c57ac96368c9a":{"kind":"REWORK","_number":1,"created":"2019-09-20 10:08:03.000000000","uploader":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"ref":"refs/changes/50/683350/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/50/683350/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/50/683350/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/50/683350/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/50/683350/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/50/683350/1"}}},"commit":{"parents":[{"commit":"6b1dbe6729d899dfcc88e71cdcb3b5298aa7cb05","subject":"Merge \"Support direct provisioning of bare metal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/6b1dbe6729d899dfcc88e71cdcb3b5298aa7cb05"}]}],"author":{"name":"Arx Cruz","email":"arxcruz@gmail.com","date":"2019-09-20 10:07:49.000000000","tz":120},"committer":{"name":"Arx Cruz","email":"arxcruz@gmail.com","date":"2019-09-20 10:07:49.000000000","tz":120},"subject":"Adding tests to skip list in Rocky","message":"Adding tests to skip list in Rocky\n\nThe two tests added to skip list are failing on featureset020. A\nlaunchpad bug is open and the responsible team are investigating. Once\nget fixed, we will remove from skip list.\n\nChange-Id: I3aaa38ff0180826a6e749d0686d1a0e37e97eb84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/a3ed0ea252c4a7f445ee9905102c57ac96368c9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/a3ed0ea252c4a7f445ee9905102c57ac96368c9a"}]},"branch":"refs/heads/master"},"f65689983216197eb6f2f4fac82cd58e3ff2d403":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-09-20 10:09:22.000000000","uploader":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"ref":"refs/changes/50/683350/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/50/683350/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/50/683350/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/50/683350/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/50/683350/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/50/683350/2"}}},"commit":{"parents":[{"commit":"6b1dbe6729d899dfcc88e71cdcb3b5298aa7cb05","subject":"Merge \"Support direct provisioning of bare metal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/6b1dbe6729d899dfcc88e71cdcb3b5298aa7cb05"}]}],"author":{"name":"Arx Cruz","email":"arxcruz@gmail.com","date":"2019-09-20 10:07:49.000000000","tz":120},"committer":{"name":"Arx Cruz","email":"arxcruz@gmail.com","date":"2019-09-20 10:08:36.000000000","tz":120},"subject":"Adding tests to skip list in Rocky","message":"Adding tests to skip list in Rocky\n\nThe two tests added to skip list are failing on featureset020. A\nlaunchpad bug is open and the responsible team are investigating. Once\nget fixed, we will remove from skip list.\n\nRelated-Bug: 1843259\nChange-Id: I3aaa38ff0180826a6e749d0686d1a0e37e97eb84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/f65689983216197eb6f2f4fac82cd58e3ff2d403"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/f65689983216197eb6f2f4fac82cd58e3ff2d403"}]},"branch":"refs/heads/master"},"62babbfe379dad4dbc9a87c216f236e3a70d711b":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-10-01 07:28:47.000000000","uploader":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"ref":"refs/changes/50/683350/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/50/683350/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/50/683350/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/50/683350/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/50/683350/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/50/683350/3"}}},"commit":{"parents":[{"commit":"dc0ba8b9afedd280ea43212ce8cd8f7f5eb660e9","subject":"Merge \"Statically import CI reproducer roles\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/dc0ba8b9afedd280ea43212ce8cd8f7f5eb660e9"}]}],"author":{"name":"Arx Cruz","email":"arxcruz@gmail.com","date":"2019-09-20 10:07:49.000000000","tz":120},"committer":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2019-10-01 07:28:47.000000000","tz":0},"subject":"Adding tests to skip list in Rocky","message":"Adding tests to skip list in Rocky\n\nThe two tests added to skip list are failing on featureset020. A\nlaunchpad bug is open and the responsible team are investigating. Once\nget fixed, we will remove from skip list.\n\nRelated-Bug: 1843259\nChange-Id: I3aaa38ff0180826a6e749d0686d1a0e37e97eb84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/62babbfe379dad4dbc9a87c216f236e3a70d711b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/62babbfe379dad4dbc9a87c216f236e3a70d711b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
