)]}'
{"id":"openstack%2Ftripleo-quickstart-extras~685184","triplet_id":"openstack%2Ftripleo-quickstart-extras~master~I3cd1404e48e3b76fd19f4301a00d8beaad274dcd","project":"openstack/tripleo-quickstart-extras","branch":"master","hashtags":[],"change_id":"I3cd1404e48e3b76fd19f4301a00d8beaad274dcd","subject":"ensure periodic job requirements are visible","status":"MERGED","created":"2019-09-26 22:25:31.000000000","updated":"2019-10-02 20:09:05.000000000","submitted":"2019-10-02 20:09:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"685184-1570046945889-e03b07bf","meta_rev_id":"e08f68cb35b395a22cd1f3897637b923fa4aaa3e","_number":685184,"virtual_id_number":685184,"owner":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-02 20:09:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-10-01 20:32:51.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"all":[{"value":2,"date":"2019-10-01 18:04:53.000000000","_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"all":[{"value":1,"date":"2019-10-01 18:04:53.000000000","_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-01 15:07:28.000000000","updated_by":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"reviewer":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"state":"REVIEWER"},{"updated":"2019-10-01 20:32:51.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-02 20:09:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"36abd53129b115d8a4237fe2d3dcf4da88ee0c05","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2019-09-26 22:25:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dbcd2cbe1eb7031f7700705683d8e66e78afbd88","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2019-09-26 23:47:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0078769dd6558a5d1293f7772fbd69d4a4e526f7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-27 01:48:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/84/685184/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/2114eb5/ : SUCCESS in 2h 58m 47s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/84/685184/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/d80eed8/ : FAILURE in 1h 48m 13s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/84/685184/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/0daf34f/ : SUCCESS in 2h 13m 18s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/84/685184/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/da8bf07/ : FAILURE in 3h 21m 59s","accounts_in_message":[],"_revision_number":1},{"id":"5d87b367228e5dadf162e8159a8c5a68d44242ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 03:51:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/268071367e6047b48ffc147e38b4088e : SUCCESS in 2h 42m 02s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/e3b2644d26f649c0b8edb7363f9d7779 : FAILURE in 2h 35m 09s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/e9c7c68d96934d5dbf855a8496ae0817 : SUCCESS in 3h 41m 15s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/5f91ea9cfed14fbd91bd97d321352b82 : SUCCESS in 3h 20m 03s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/7f2a5b2fe1134f51bd583a78c3ea5d92 : FAILURE in 2h 48m 15s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/af856b8318e14f4bac759b78fd8689f9 : SUCCESS in 3h 16m 59s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/5fc0e4ae547d4b8098d4bc635e225490 : SUCCESS in 3h 18m 53s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/466603f48f5842e3a8a5d95cee2e1055 : SUCCESS in 2h 44m 31s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/9963b39ff19f478ab8aa54413413a5f0 : FAILURE in 1h 09m 33s (non-voting)\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/feb2628b6542416994f87c37f8c64bc9 : SUCCESS in 1h 41m 09s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/95a27401349b4a65a93f1640c9ed082a : SUCCESS in 2h 43m 03s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/a1b877d3e7c64c1ba089b764617c972a : SUCCESS in 3h 03m 34s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/ecca78b863bf42339f70a8a4be22237c : SUCCESS in 2h 36m 56s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/25cd0cc29a1b4d3fbfd3005a116bf72f : SUCCESS in 6m 17s","accounts_in_message":[],"_revision_number":1},{"id":"12eb11b5111ce6f7e51f63b85e0494ccffe1d50c","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2019-09-28 03:23:40.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"18106657b35dcbc1f400a1b0786cd340f6026cda","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-28 06:36:32.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/84/685184/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/bdc8ddd/ : FAILURE in 3h 11m 17s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/84/685184/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/3c95776/ : FAILURE in 2h 02m 51s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/84/685184/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/8d5a271/ : FAILURE in 38m 40s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/84/685184/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/f8b2ad9/ : SUCCESS in 3h 01m 14s","accounts_in_message":[],"_revision_number":1},{"id":"9996cc1b89bba3ea9a5f5467d8102ac6b872ff25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-28 07:03:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/f2be9cd5f57b4cac8c2712e8056dcb32 : SUCCESS in 2h 14m 36s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/23a689c3a4e442ebbd34d2917edc5dd3 : FAILURE in 2h 18m 48s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/feec1557a5a045dea60c3d1b08b5f126 : SUCCESS in 3h 34m 11s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/44feb4c73bee47c79e63783c2de2cad9 : SUCCESS in 3h 19m 00s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/dd9c7c7f8eb240438ee65560d40095ff : FAILURE in 2h 48m 11s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/732fc0ada5c9412fa6ee9fb306c80d80 : SUCCESS in 2h 57m 10s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/a9770854fb5741788aef1305c66e5d91 : SUCCESS in 2h 46m 51s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/29108a37fb924075834bef4bbc2990aa : SUCCESS in 2h 21m 40s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/1c3ca6ccbacf489bb43c00b39da71054 : FAILURE in 1h 48m 02s (non-voting)\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/ff6e07964556416bb07b0b9fea02a4ec : SUCCESS in 1h 36m 04s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/18526fca2714461aa46b32c940c1ce86 : SUCCESS in 2h 57m 40s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/cd273e2b386c4a6c8d1af19a6cb04aac : FAILURE in 52m 06s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/002b731e66b340f7a47b7f9d9858f2c5 : SUCCESS in 2h 19m 33s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/76522c4357df428987c2b0270d9ae4ed : SUCCESS in 3m 28s","accounts_in_message":[],"_revision_number":1},{"id":"10869749c42f3881e58c7a392d55d356229b7c9b","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2019-09-28 14:35:53.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f5bed1a67d704605a41fede88303ebaba99c4440","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-28 17:34:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/84/685184/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/7c17c3e/ : SUCCESS in 2h 57m 44s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/84/685184/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/53f2616/ : FAILURE in 1h 45m 58s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/84/685184/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/9d74536/ : FAILURE in 35m 59s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/84/685184/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/74a803d/ : SUCCESS in 2h 51m 02s","accounts_in_message":[],"_revision_number":1},{"id":"280272c6b5921769c41b155f5a7a6413f70b7544","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-28 18:20:43.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/cb428255a7cc4cde890a812b5741673a : SUCCESS in 2h 22m 03s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/99127aaf68e548b2aaf561f015e34d60 : FAILURE in 2h 06m 04s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/2f8a80764c1349809a4549e7b3d5f3e6 : SUCCESS in 3h 20m 13s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/99d3c4508e4c4c3e9363752874b2765b : SUCCESS in 3h 34m 48s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/918675df4c48409886bdd5a9e81dd456 : FAILURE in 2h 27m 17s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/1ca4449f0e874005b842eb85d8a48d42 : SUCCESS in 3h 06m 17s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/b5da2f2f2f6b44b7b66eb3a932337686 : SUCCESS in 2h 58m 01s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/f3dbc7a129684de08ca7a14e1b5433d6 : SUCCESS in 2h 21m 13s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/fd71e7c8ad3143598634697c724d8505 : FAILURE in 1h 53m 06s (non-voting)\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/8e579a6bd4384840ae51ce837ea9deb2 : SUCCESS in 1h 51m 49s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/5dcb5fdb91d344ba8037b965a598bbd5 : SUCCESS in 2h 50m 00s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/7205f9ad61ea479687079698d9d29609 : FAILURE in 46m 11s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/ec138811cb4e4109979c18f2cd83ab48 : SUCCESS in 2h 43m 23s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2859c322e6eb4d67ba92804923795b77 : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":1},{"id":"73e9dbb2ae96a494a187208c7a417013d56bca67","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2019-10-01 00:49:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9060ab8c837a6759d0f439610ae280182224bf04","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-01 03:47:47.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/84/685184/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/1112bbf/ : SUCCESS in 2h 57m 15s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/84/685184/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/2bf140c/ : FAILURE in 1h 50m 06s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/84/685184/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/477a303/ : FAILURE in 35m 30s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/84/685184/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/bcfe648/ : SUCCESS in 2h 46m 33s","accounts_in_message":[],"_revision_number":1},{"id":"e0f81c79e601b07c18faaa1e56e4702be4d12fd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 04:14:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/a8a6ca541b59416c92efcb24378f81de : SUCCESS in 1h 57m 37s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/5d5047de0ae94f89b8ffc4f3f9d0e25a : FAILURE in 1h 52m 32s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/2d26a20aabb341aea8d8d7567fd70cd4 : SUCCESS in 2h 47m 09s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/22b2fce6de064e40a329fcd174a9fbe0 : SUCCESS in 3h 05m 25s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/7eeba6c94e4f42559404bb17dfc5fba4 : FAILURE in 2h 06m 59s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/b34acf8f1c0b4b14ad2dc5708a41860a : SUCCESS in 2h 50m 08s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/412c7c69c2c44b18be3664797401e1a2 : SUCCESS in 2h 38m 54s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/f5f28782c1ba4f0f9938e06d52f6136c : SUCCESS in 2h 22m 12s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/413defaa1542403992705a5f662866cd : FAILURE in 1h 53m 52s (non-voting)\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/212e775bf718479c83ceacb8d521d12b : SUCCESS in 1h 53m 21s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/73c95caf28d84ee48a40e308919ec34f : SUCCESS in 2h 59m 40s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/007153cbb8864c119127a66e69c02245 : FAILURE in 45m 21s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/e56a8e6e63a24e9bb68fb6701157123f : SUCCESS in 2h 22m 56s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c37a8a6953d643bfbc79e6cb6519c162 : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":1},{"id":"7da189e248966c6e309a129f8c428756674430cf","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2019-10-01 15:07:28.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"924237ca0c742c148869601fbff269517c3442bb","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2019-10-01 16:48:05.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"c32b73937c2b1ca8359dc018a1339072d60d679f","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2019-10-01 16:48:44.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"1f1926b94d1fb3470e12c77a03f9fae0d847e36c","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2019-10-01 18:04:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c05a9d3e77fd5da23558a76d4e9a99be1498dfb6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-01 20:32:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/84/685184/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/cf7d657/ : TIMED_OUT in 3h 42m 19s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/84/685184/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/7f1facb/ : FAILURE in 1h 56m 08s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/84/685184/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/53c89f8/ : FAILURE in 35m 53s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/84/685184/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/9b171aa/ : SUCCESS in 2h 51m 26s","accounts_in_message":[],"_revision_number":2},{"id":"69b44c837e590f61c287432f82200aed3fc13632","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 00:02:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/0335c3d8f34e445688b6f90f29675626 : SUCCESS in 2h 16m 16s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/c52ebf7e7d8d4495bab4055dd52befc2 : FAILURE in 2h 03m 54s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/f4d6fa1dcafc4ea4bf08163654e601dc : SUCCESS in 2h 43m 17s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/e17654ce606743c1827a7c3e8e455366 : SUCCESS in 3h 02m 09s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/659fa145d92b44818da2b9c8fa7fdafd : FAILURE in 2h 19m 02s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/b2f6b5a2723a4754bfc8c06639b5a706 : SUCCESS in 2h 25m 42s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/68090d7b0a6a461eaef6fa9b8ae14f58 : SUCCESS in 2h 34m 59s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/ba39a5e955cd4699aa19137e2d1ad848 : SUCCESS in 2h 27m 18s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/76ae5aec82854f37854c53e1758d9e66 : FAILURE in 2h 09m 18s (non-voting)\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/252eb93761e04d09bb155b2d539bde0e : SUCCESS in 1h 21m 29s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/663de0caf4d14e76b8d7d5199c8f677a : SUCCESS in 2h 51m 14s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/b0f29b5e785d496fb60516bff2fe7ac3 : SUCCESS in 2h 55m 02s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/034f672311294e42ab2c67ccba193c3d : SUCCESS in 2h 24m 15s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d0dc7f0abb274928927b5355becc9f5a : SUCCESS in 6m 40s","accounts_in_message":[],"_revision_number":2},{"id":"849dd4e085f0a48507bc927eea2d6f554cc58f61","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2019-10-02 12:45:18.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"27c7ede56be6bbbe90c58173c5755d2611d4e3c1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-02 15:51:40.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/84/685184/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/c81a331/ : SUCCESS in 2h 45m 23s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/84/685184/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/8e3bed7/ : FAILURE in 1h 57m 45s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/84/685184/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/a465804/ : SUCCESS in 2h 18m 35s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/84/685184/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/5c5657b/ : SUCCESS in 3h 02m 05s","accounts_in_message":[],"_revision_number":2},{"id":"de443739c51c875af89ce60a28ea537731d4c289","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 16:24:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"36ea388003e93798b91f16d310a05b28cca44c2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 17:29:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/e4c56fe86e7543e8a6cfc591f805644e : SUCCESS in 1h 59m 25s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/dde5624f536941baaf7002b1ff9aa488 : FAILURE in 2h 00m 24s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/6b67a0d2fcbe4142ba21859978d0405c : SUCCESS in 3h 12m 37s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/01db837600d04a10b6c8a5e53a65fc0f : SUCCESS in 2h 56m 42s (non-voting)\n- tripleo-ci-centos-7-scenario012-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/0526153c3cbc4f1a8421d035b3ee9d8a : FAILURE in 1h 56m 43s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/3d119ec34ff44c66a388cf28eb082737 : SUCCESS in 2h 37m 33s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/5fff5f2af1244b4e80c87d5e68a4df7f : SUCCESS in 2h 50m 40s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/000eefcb3c5a4f97ab023537c05d954c : SUCCESS in 2h 23m 44s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/84f15fd95aa04d81b41a4b343a63b531 : FAILURE in 2h 04m 22s (non-voting)\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/c8954440e1014565a02bbe988f972c5a : SUCCESS in 1h 59m 48s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/d4399bc309a24d3880acfd84410722c4 : SUCCESS in 2h 53m 49s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/35cf46c22df64543984c6802ea605baf : SUCCESS in 3h 45m 20s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/47fd35348dfa427bb6a62ff4e25ea82f : SUCCESS in 2h 24m 37s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1c5a319c78354fb2a3ea02f006353c15 : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":2},{"id":"8545a6f9603db7e24ffac09b60fc9d414c18449c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 20:09:05.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/55b848bb85bb4b9188a7d102721d886c : SUCCESS in 2h 19m 00s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/ea38e56fa6c645178c3003d4665175d1 : SUCCESS in 3h 40m 47s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/5f961aceaab140c1a39a0c78f02fe3b5 : SUCCESS in 2h 37m 08s\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/c76ba03a97574499a586edbc1f950f16 : SUCCESS in 1h 56m 43s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/23559f3d794d42f6887a263197dd5328 : SUCCESS in 3h 02m 49s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/df1efc0cd62448cca3e8cc7b0105d11c : SUCCESS in 3h 01m 45s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/7d87b76f31af44928d749d1dbaaecf3c : SUCCESS in 3h 07m 53s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/e305e0a80a7145d8a32e8549abfb140d : SUCCESS in 2h 05m 08s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/23fc241252854243ab02da76b66c1d37 : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":2},{"id":"e08f68cb35b395a22cd1f3897637b923fa4aaa3e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 20:09:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8fb370f5be8bcc991f2d7beef556ba22abc1d646","revisions":{"b078ce548d9575d99c602b62772c69ad96545c18":{"kind":"REWORK","_number":1,"created":"2019-09-26 22:25:31.000000000","uploader":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"ref":"refs/changes/84/685184/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/84/685184/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/84/685184/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/84/685184/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/84/685184/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/84/685184/1"}}},"commit":{"parents":[{"commit":"f29d6c275c9d5bac5fcebe40d1e5978417905460","subject":"update centos-release openstack to rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/f29d6c275c9d5bac5fcebe40d1e5978417905460"}]}],"author":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2019-09-26 22:25:06.000000000","tz":-360},"committer":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2019-09-26 22:25:09.000000000","tz":-360},"subject":"ensure periodic job requirements are visible","message":"ensure periodic job requirements are visible\n\nperiodic requires to_build\u003dtrue and\nto use the rdo registry\n\nChange-Id: I3cd1404e48e3b76fd19f4301a00d8beaad274dcd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/b078ce548d9575d99c602b62772c69ad96545c18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/b078ce548d9575d99c602b62772c69ad96545c18"}]},"branch":"refs/heads/master"},"8fb370f5be8bcc991f2d7beef556ba22abc1d646":{"kind":"REWORK","_number":2,"created":"2019-10-01 16:48:44.000000000","uploader":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"ref":"refs/changes/84/685184/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/84/685184/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/84/685184/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/84/685184/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/84/685184/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/84/685184/2"}}},"commit":{"parents":[{"commit":"f29d6c275c9d5bac5fcebe40d1e5978417905460","subject":"update centos-release openstack to rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/f29d6c275c9d5bac5fcebe40d1e5978417905460"}]}],"author":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2019-09-26 22:25:06.000000000","tz":-360},"committer":{"name":"wes hayutin","email":"whayutin@redhat.com","date":"2019-10-01 16:48:41.000000000","tz":0},"subject":"ensure periodic job requirements are visible","message":"ensure periodic job requirements are visible\n\nperiodic requires to_build\u003dtrue and\nto use the rdo registry\n\nChange-Id: I3cd1404e48e3b76fd19f4301a00d8beaad274dcd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/8fb370f5be8bcc991f2d7beef556ba22abc1d646"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/8fb370f5be8bcc991f2d7beef556ba22abc1d646"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
