)]}'
{"id":"openstack%2Ftripleo-quickstart-extras~685639","triplet_id":"openstack%2Ftripleo-quickstart-extras~master~I2a95bed29f39e49b35d7e0cb90a95be8d9b1ab34","project":"openstack/tripleo-quickstart-extras","branch":"master","topic":"685639","hashtags":[],"change_id":"I2a95bed29f39e49b35d7e0cb90a95be8d9b1ab34","subject":"[DNM] run master package with non master branch","status":"ABANDONED","created":"2019-09-30 03:46:52.000000000","updated":"2020-01-27 05:08:59.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1d422df9ddbe360c31bc1f6d009fef704a3fb77a","_number":685639,"virtual_id_number":685639,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"all":[{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-10-01 14:07:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-10-01 13:10:14.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"all":[{"value":-2,"date":"2019-10-01 09:26:02.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-30 16:44:09.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2019-10-01 13:10:14.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-01 14:07:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"de558967044ffdb17cb22f7b746b3c9069645002","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-09-30 03:46:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c57efebe77ce4502112d60e38f347dd8682d13b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 06:39:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/9c97347ad8e44e02838e0600b2a3015f : SUCCESS in 2h 22m 02s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/5652f5e7b41541b591c021497722719f : FAILURE in 2h 08m 20s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/53c803827e4144ed9601b1da435ba346 : SUCCESS in 2h 17m 52s","accounts_in_message":[],"_revision_number":1},{"id":"2cca78ab455355e3811d27c131381f7b94fda7e5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-30 06:43:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/39/685639/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/e018121/ : SUCCESS in 2h 53m 20s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/39/685639/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/7e2ae67/ : FAILURE in 1h 49m 41s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/39/685639/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/8b91682/ : FAILURE in 36m 41s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/39/685639/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/4b41ec2/ : SUCCESS in 2h 55m 29s","accounts_in_message":[],"_revision_number":1},{"id":"1bebd0084fea0d1048675d9680cc0ab518a0685d","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2019-09-30 16:44:09.000000000","message":"Patch Set 1: Code-Review-2\n\nlet\u0027s not do this in defaults.","accounts_in_message":[],"_revision_number":1},{"id":"820df4b5e76e17a3b5c35e5611c712c92783092c","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2019-09-30 16:49:32.000000000","message":"Patch Set 1:\n\nah.. this is a DNM patch","accounts_in_message":[],"_revision_number":1},{"id":"5df5867fa44b90e5c73a65a3cb05cd126cc54b65","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-10-01 09:26:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b62221d34187d622d2b72f1ef96f3e2499c7cc8b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-01 13:10:14.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/39/685639/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/4751642/ : SUCCESS in 2h 58m 22s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/39/685639/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/c8270d1/ : FAILURE in 2h 07m 39s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/39/685639/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/3a6c4e2/ : SUCCESS in 2h 24m 19s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/39/685639/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/7e3444b/ : TIMED_OUT in 3h 42m 03s","accounts_in_message":[],"_revision_number":2},{"id":"9fdb23da4645d85f473216db6a9ced176653ccb5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 14:07:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/2957c81cf87941cdbfaadfc102322e5a : SUCCESS in 2h 05m 57s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/74457e32fef7463f8dcb761c5cf392c2 : FAILURE in 1h 35m 53s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/0817f667a0ac47b39ee9d305cdb7f355 : SUCCESS in 2h 12m 18s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/c0dbf92a53f343ab898318442249dc98 : SUCCESS in 2h 47m 01s","accounts_in_message":[],"_revision_number":2},{"id":"56570e6b773b10093ebf64b2f3e7d7370496b0dd","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-01-27 05:08:59.000000000","message":"Abandoned\n\nWas just for testing","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2fd1aa7e40992311959816dd5151447ddede898b","revisions":{"e5f5a9f746dde050573d7c94f73df2247063bf55":{"kind":"REWORK","_number":1,"created":"2019-09-30 03:46:52.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/39/685639/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/39/685639/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/39/685639/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/39/685639/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/39/685639/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/39/685639/1"}}},"commit":{"parents":[{"commit":"dc0ba8b9afedd280ea43212ce8cd8f7f5eb660e9","subject":"Merge \"Statically import CI reproducer roles\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/dc0ba8b9afedd280ea43212ce8cd8f7f5eb660e9"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2019-09-30 03:45:57.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2019-09-30 03:46:30.000000000","tz":330},"subject":"Try setting DnsServers for undercloud","message":"Try setting DnsServers for undercloud\n\nChange-Id: I2a95bed29f39e49b35d7e0cb90a95be8d9b1ab34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/e5f5a9f746dde050573d7c94f73df2247063bf55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/e5f5a9f746dde050573d7c94f73df2247063bf55"}]},"branch":"refs/heads/master"},"2fd1aa7e40992311959816dd5151447ddede898b":{"kind":"REWORK","_number":2,"created":"2019-10-01 09:26:02.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/39/685639/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/39/685639/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/39/685639/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/39/685639/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/39/685639/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/39/685639/2"}}},"commit":{"parents":[{"commit":"dc0ba8b9afedd280ea43212ce8cd8f7f5eb660e9","subject":"Merge \"Statically import CI reproducer roles\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/dc0ba8b9afedd280ea43212ce8cd8f7f5eb660e9"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2019-09-30 03:45:57.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2019-10-01 09:24:56.000000000","tz":330},"subject":"[DNM] run master package with non master branch","message":"[DNM] run master package with non master branch\n\nDepends-On: https://review.opendev.org/#/c/685877/\nChange-Id: I2a95bed29f39e49b35d7e0cb90a95be8d9b1ab34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/2fd1aa7e40992311959816dd5151447ddede898b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/2fd1aa7e40992311959816dd5151447ddede898b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
