)]}'
{"id":"openstack%2Ftripleo-quickstart-extras~713542","triplet_id":"openstack%2Ftripleo-quickstart-extras~master~I728183f608eb7ab9571d66cc8c515fe52f4dafc0","project":"openstack/tripleo-quickstart-extras","branch":"master","topic":"bug/1867035","hashtags":[],"change_id":"I728183f608eb7ab9571d66cc8c515fe52f4dafc0","subject":"Adjust molecule requirements","status":"MERGED","created":"2020-03-17 21:15:57.000000000","updated":"2020-03-19 16:12:26.000000000","submitted":"2020-03-19 16:09:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"713542-1584634156573-51d1296f","meta_rev_id":"2a6a29b91f970ffc4fd60d14ac790043fb6bbd6d","_number":713542,"virtual_id_number":713542,"owner":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-19 16:09:16.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2020-03-18 00:11:44.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-03-17 21:38:07.000000000","_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-03-19 12:04:16.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-17 21:38:07.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2020-03-18 00:11:44.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-19 12:04:16.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-03-19 16:09:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"910b8be601215435fb2d3f696c09fe45c525f00b","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-03-17 21:15:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3d71ccfa59be8ce9d3999faecbe4f988a10d3725","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-03-17 21:38:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"84edf9371e79e65e1fdcd0ee81372759eebe705b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-03-18 00:11:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/1ac6a73b78d347f0b8d2b6d576009002 : FAILURE in 2h 32m 05s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch https://review.rdoproject.org/zuul/build/603bad1e883a485ea9380358102c839c : SUCCESS in 2h 09m 43s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch https://review.rdoproject.org/zuul/build/42c6392ea9c24b1aa788b4e83e526886 : SUCCESS in 2h 26m 21s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch https://review.rdoproject.org/zuul/build/a2ab4263f7714f18a33c675c9a326948 : SUCCESS in 2h 41m 45s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-master-vexxhost https://review.rdoproject.org/zuul/build/1323bf9b1aef4f9aacc418c0bf77c8a2 : SUCCESS in 2h 54m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"faf6a48c4a3fd0338eec07deac765268d43c872b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 00:42:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/a0d9fa2cebc549c7a4c63217148b723c : SUCCESS in 1h 33m 27s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/2ca8f01775fd4a7ab29d60fc1592f487 : TIMED_OUT in 3h 24m 37s (non-voting)\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/4a31291a569041e2b4f0904dfe731ad2 : SUCCESS in 2h 39m 32s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/a26a72a6a77f41d5a1dfd572ebbaa705 : FAILURE in 1h 30m 57s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/dcac2889271746b5a2e44af4cff93720 : SUCCESS in 2h 15m 36s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/d7d2e124ae3e48fda103192713e32b2c : SUCCESS in 2h 05m 13s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/3485ab6d77634494b9763413c11ea92e : SUCCESS in 2h 19m 11s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/8dd2dd8c789a4d1f9b775c0b301f6a20 : SUCCESS in 1h 32m 30s\n- tripleo-ci-centos-7-containers-multinode-train https://zuul.opendev.org/t/openstack/build/189729b635774238b577e9efda10ef3c : SUCCESS in 2h 21m 57s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/bb4f686fe7e34e39b7f2076f533dd7ab : SUCCESS in 1h 57m 48s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/8c776c1ae56b48b0b47261297f7f1a6e : SUCCESS in 2h 18m 14s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/b74f53e0551b42d2b3ff834181287690 : SUCCESS in 1h 58m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/88fdd09d33884892906dad1cd6a597f4 : SUCCESS in 4m 01s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c7cdcab5844f4d7d8c66394e85e8ca41 : SUCCESS in 5m 58s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/663d7e55777440a99abed64511818c76 : SUCCESS in 11m 25s","accounts_in_message":[],"_revision_number":1},{"id":"95edad033b3de621ab9bc6e0ec7a6c2e28c71c3b","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-03-19 12:04:16.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"17ebbfba571247d6773f1cd5a8679e3a811af51f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 12:04:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a80362817f80827b18a322eec6bd45e340b8a1bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 16:09:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/d88d714c68cb4866bce1d5688519f635 : SUCCESS in 1h 40m 27s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/45dfed1d675f4acb8b5dd9fc5db1e0ce : SUCCESS in 2h 46m 53s\n- tripleo-ci-centos-8-containers-undercloud-minion https://zuul.opendev.org/t/openstack/build/6e85847bd7b64078b6ab2f28f75049e2 : SUCCESS in 1h 35m 09s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/a87cec5922994ff4b8c1c23c5e4ba455 : SUCCESS in 1h 31m 26s\n- tripleo-ci-centos-7-containers-multinode-train https://zuul.opendev.org/t/openstack/build/9c604f45b5374c54b5dc5570adbdb041 : SUCCESS in 2h 27m 36s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/e1cd8ad9cbd148daa5770d0658466b3d : SUCCESS in 2h 14m 23s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/a4dae5cb87f64359947d9aa7734e6fa9 : SUCCESS in 2h 18m 06s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/3a2fb70c5c1047aabc249dffeba8cac9 : SUCCESS in 2h 08m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/deef445b59344214830156c03443d166 : SUCCESS in 5m 31s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/eefabbad91714152a8e0977278ad2daf : SUCCESS in 7m 34s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/71d502dbc8814e878e764a0ce739b7e3 : SUCCESS in 13m 20s","accounts_in_message":[],"_revision_number":1},{"id":"b0b664ad35f21d40aec3bcb0fad59db5c5b2d5dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 16:09:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f38b4ca50f7ef0dd79834ed88c01b4c4c685763b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 16:12:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a999cd25de6d423c9d6a6d3dff19f96b : SUCCESS in 2m 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1b5274793b92383255df18b41ee49045458b7522","revisions":{"1b5274793b92383255df18b41ee49045458b7522":{"kind":"REWORK","_number":1,"created":"2020-03-17 21:15:57.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/42/713542/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/42/713542/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/42/713542/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/42/713542/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/42/713542/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/42/713542/1"}}},"commit":{"parents":[{"commit":"ffd8f24e0a29b7621196be4d23b3409806989fc4","subject":"Merge \"Make reproducer script centos-8 friendly\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/ffd8f24e0a29b7621196be4d23b3409806989fc4"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-03-17 21:15:16.000000000","tz":780},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-03-17 21:15:16.000000000","tz":780},"subject":"Adjust molecule requirements","message":"Adjust molecule requirements\n\nThis change uses the same pytest/molecule requirements spec as\ntripleo-quickstart and breaks out the requirements into their own\nfile.\n\n[1] https://review.opendev.org/#/c/713532/\nCloses-Bug: #1867035\n\nChange-Id: I728183f608eb7ab9571d66cc8c515fe52f4dafc0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/1b5274793b92383255df18b41ee49045458b7522"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/1b5274793b92383255df18b41ee49045458b7522"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
