)]}'
{"id":"openstack%2Ftripleo-quickstart-extras~734570","triplet_id":"openstack%2Ftripleo-quickstart-extras~master~Ia06fa93973b153a265ddd0cb2538f5bdabf764aa","project":"openstack/tripleo-quickstart-extras","branch":"master","topic":"drop_mock","hashtags":[],"change_id":"Ia06fa93973b153a265ddd0cb2538f5bdabf764aa","subject":"Use unittest.mock instead of mock","status":"MERGED","created":"2020-06-09 10:18:14.000000000","updated":"2020-12-28 11:08:16.000000000","submitted":"2020-12-28 11:08:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"734570-drop_mock","meta_rev_id":"1a87fcec672f01fdf9ed24b388fc92235c70a9cb","_number":734570,"virtual_id_number":734570,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-28 11:08:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-09 15:35:21.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-06-10 14:25:10.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-11-27 06:59:53.000000000","_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2020-11-24 19:00:55.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2020-11-24 19:00:55.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-09 15:35:21.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-09 16:17:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-10 14:25:10.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-11-24 19:00:55.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-11-27 06:59:53.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"}],"messages":[{"id":"b9b558d7eac497a5a9ec70cae7b373ab033bfa67","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-09 10:18:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3b4a75e5b32a0173facdaa5d44e878deb1ea9490","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-09 15:35:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/984076e8c9ee4f4abc568ee6445f5d71 : SUCCESS in 2h 39m 10s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch https://review.rdoproject.org/zuul/build/6bd0cafc7b714a2298933eaee55a1b1e : SUCCESS in 2h 05m 49s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch https://review.rdoproject.org/zuul/build/f51c2b283b754e3c82e89b3dec16a2dc : SUCCESS in 2h 33m 21s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch https://review.rdoproject.org/zuul/build/4c8e84849e58405b850442e22d66f343 : SUCCESS in 2h 46m 37s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-train-branch https://review.rdoproject.org/zuul/build/3305512ee57a4b529f336c1856e9303a : SUCCESS in 2h 45m 07s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-ussuri-branch https://review.rdoproject.org/zuul/build/b69fb1362ea74a408af5612f0e94eaa8 : SUCCESS in 2h 52m 47s (non-voting)\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/b4598b1dd3ed4b4cb3c814fe2f6a4f6c : SUCCESS in 3h 04m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e8bc43f71a05cd02ff0f806b7a2d56adca4c5d41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 16:17:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/6ee087c031a44944b890488ade3c1fef : SUCCESS in 1h 31m 05s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/119926aa3cdd4c8185462b975d9823bf : SUCCESS in 2h 23m 38s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/e2b08f254fd7495791b9bb31a1550728 : SUCCESS in 2h 10m 08s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/6abe5610d0fd4552b8253ecfb1c3a332 : SUCCESS in 2h 10m 59s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/6618dbf8f09a4e13848f8ad6c0c43515 : SUCCESS in 1h 35m 04s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/b420fdeffd564c3cb97a8cac5641f17b : SUCCESS in 2h 38m 14s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/34636f42f50843d394c01e3cce0b2d71 : SUCCESS in 1h 25m 03s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/9f7468ca54e84e8b9e3963962e2d292b : SUCCESS in 1h 28m 42s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/d45230376bf9430dad51299232345d69 : SUCCESS in 1h 44m 25s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/0ea9d78c4d3b4fa48b7805eb98f3141c : SUCCESS in 2h 21m 59s\n- tripleo-ci-centos-7-containers-multinode-train https://zuul.opendev.org/t/openstack/build/08bee23a098542eda249665dd0168e83 : SUCCESS in 2h 57m 50s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/fbdea261f20243599f3fb592a298a549 : SUCCESS in 2h 18m 04s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/bab91fb2cb294922b302ad7e78f08659 : SUCCESS in 2h 29m 26s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/e03061dde8414fa8b6b7652941176ba5 : SUCCESS in 2h 24m 56s\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/5247ca0e04214a698f8854b1487da3c2 : SUCCESS in 1h 30m 44s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/dbb83fbccbf946079022a8ffe93676dd : SUCCESS in 9m 28s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/7e630ba5ceac48509e5fba0acb5342c7 : SUCCESS in 16m 14s","accounts_in_message":[],"_revision_number":1},{"id":"a339e4b0316f7533609530f1c95593090e92c1f7","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-06-10 14:25:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"38ffd68f6069f2d67850a60bbef99af9597a8675","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-11-24 19:00:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d1c36cc27395f62864331ab488aefe931dfdfeb2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 00:11:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ac0546d4b66114ddb3fc96c4d7b594581118adf1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 19:25:20.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/93691f52f28b4d9bbb215a2a85183fe8 : FAILURE in 52m 46s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/9d12b0390e09432e92d71130cb38fa77 : FAILURE in 1h 17m 18s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/5fc873a28d5e4884bda83eb07f3dcb32 : FAILURE in 1h 09m 48s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/7ea7283339b8451a9f58d395100036c9 : FAILURE in 30m 05s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-undercloud-minion https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1766b07988084178aa569fc965c4a40a : SUCCESS in 5m 04s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/328bd36474424093b3153551a53c50d2 : SUCCESS in 14m 58s","accounts_in_message":[],"_revision_number":1},{"id":"feb37799cee94cfab580bb87301b7473fd208324","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-11-27 06:59:53.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c529cea77205b7211e02675610b2ece7e02a19bd","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-11-27 07:00:22.000000000","message":"Patch Set 1:\n\nvalidate-tempest is no longer used though, likely the code be cleaned up.","accounts_in_message":[],"_revision_number":1},{"id":"c9c9d26d65e4cf79d284df22831757ccfa0cf9e6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-11-27 12:44:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/13835d5ab50b4f9eab6d9621f7186c68 : SUCCESS in 2h 49m 40s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-ussuri-branch https://review.rdoproject.org/zuul/build/53665fd22992486d8b328eea6844ea5e : SUCCESS in 2h 35m 24s (non-voting)\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/1bb315b77ab848108345cd7f37cd57d1 : TIMED_OUT in 5h 13m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"03425e4fc071f4674be7d52bb98305fd024ca516","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 13:01:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/8a464f448dcf40e5ab6525e18e680320 : SUCCESS in 4h 46m 10s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/f539d63d30af41d9ac724dcb5499e86c : SUCCESS in 4h 46m 21s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/f7682fd0f120452fb39f9bc770123e2f : SUCCESS in 3h 32m 49s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/4fb80a4241004080bed9b3758be9b5f3 : SUCCESS in 2h 49m 15s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/fc2bf21de67a4ecb949ca4e19d165687 : SUCCESS in 1h 11m 14s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/5e28a80fae9949e09a0d3cee872bf730 : TIMED_OUT in 3h 21m 56s (non-voting)\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/d55ab150669f438bba7b5cdfcd1b6ce6 : SUCCESS in 1h 55m 04s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/bf947864f1aa4fc4a3412e0b2fa4e6a3 : SUCCESS in 1h 51m 50s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/65fa6eefb66f4167ab006cdaa0843bc7 : SUCCESS in 1h 39m 42s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/852eb798577d440787b02d1505ff6b43 : SUCCESS in 1h 54m 28s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/79bdbb99409644d68388c56a3d5fc88b : SUCCESS in 1h 27m 05s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/07964d9348c64d1c96ebada2b51978c8 : SUCCESS in 1h 41m 58s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/4478ab7b20bf495da9562593c25dcb46 : SUCCESS in 1h 25m 13s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/e3244571944d4540a1bb0b077ad598d6 : SUCCESS in 1h 08m 00s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/8b8384571b36452ca85ec097b548f8f4 : SUCCESS in 1h 24m 38s\n- tripleo-ci-centos-7-content-provider-train https://zuul.opendev.org/t/openstack/build/46f9e1e4220448ae8e68788480780885 : SUCCESS in 3h 37m 52s\n- tripleo-ci-centos-7-content-provider-queens https://zuul.opendev.org/t/openstack/build/d1f33ce0689940a4b1e63344f5907d1a : SUCCESS in 3h 18m 10s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/77e1abf4c1204021887832806743320d : SUCCESS in 2h 19m 36s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/cb23b00dab8947cc98178b162e8d1c2a : SUCCESS in 1h 55m 56s\n- tripleo-ci-centos-7-containers-multinode-train https://zuul.opendev.org/t/openstack/build/6a4bf846290e4074b880a798858dfcba : SUCCESS in 2h 11m 04s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/dfca9ac3a23549d898c2c7c8f8ddc9cd : FAILURE in 1h 58m 17s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/f1acb532f304442c93f6cdb3fd05c297 : SUCCESS in 1h 14m 08s\n- tripleo-ci-centos-8-standalone-upgrade https://zuul.opendev.org/t/openstack/build/7b5601c6cfa5408799574aa2109de5bc : FAILURE in 1h 44m 56s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4b56936c308149858dae792849e4145d : SUCCESS in 5m 23s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/24637cd0d69a40948cf8690dfd8419f0 : SUCCESS in 12m 19s","accounts_in_message":[],"_revision_number":1},{"id":"63655758eb6c36fc8abe21fd4e3106bbe6da3733","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 13:02:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8c55171a86e29281dccfcfd84e694e8d1c3c4835","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 21:58:32.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/a703ca9dfe4c4b7b97babe88f7285c93 : SUCCESS in 3h 18m 01s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/4cda96ee15da4f06884b945ef9d7ee54 : SUCCESS in 55m 56s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/a59c59b88d0d47709bc05ea7944c5d5d : SUCCESS in 3h 14m 05s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/2d48c45f9509480aa07df8cdc70afa0d : SUCCESS in 3h 24m 29s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/40cdbf8147c743b59d634ea60914549a : SUCCESS in 1h 33m 56s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/8137e14ea6ba43c3ba06fc027f00df9c : SUCCESS in 2h 02m 15s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/c3e989b95c17471a8e91bb8598912efa : SUCCESS in 1h 57m 58s\n- tripleo-ci-centos-8-containers-undercloud-minion https://zuul.opendev.org/t/openstack/build/2e3559f1710f4769b77326b6c1800d9c : SUCCESS in 1h 11m 10s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/536bf3879a844c3888bc1edcfbee864c : SUCCESS in 1h 29m 52s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/467c8ee3b0654421b9e56a9546142931 : SUCCESS in 2h 24m 00s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/ff327fb249bb40669efd877866c886d9 : SUCCESS in 2h 37m 45s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/e03977c68f4143d280d293cdc551b326 : SUCCESS in 1h 43m 02s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/e3a91bef6dec4fc7b3d4641915a2235b : FAILURE in 1h 03m 33s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/a7ceadbf2f9f45e1b96421fa22e7ba94 : SUCCESS in 1h 08m 21s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/50961d66c6be4bae96528503fa949bec : SUCCESS in 1h 38m 22s\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/5e1af3bcf3344388bac01a6a46874c51 : SUCCESS in 1h 20m 21s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/065108033c794c409399af0a942fd321 : SUCCESS in 4m 55s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/44f6400e3a994c229a9c38e07bb7b1c8 : SUCCESS in 12m 08s","accounts_in_message":[],"_revision_number":1},{"id":"abb05b2043faec2a5060648d08fb7c764babe56d","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-12-28 04:26:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9a7aae76221b04b18c58cfad6ed4d73994f14fb6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-12-28 07:16:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/2664057b21b64d318970649236b0b445 : SUCCESS in 2h 38m 56s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-ussuri-branch https://review.rdoproject.org/zuul/build/12bce6a1b9fc49f5b0c4fdcae4b399a1 : SUCCESS in 2h 44m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c3f39b815ddcd3af00a80586b1c017066520974a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-28 08:14:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/01059112310748be822e92c490270112 : SUCCESS in 3h 40m 54s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/8dd665c77b2a4f49b599e92c8be7d577 : SUCCESS in 3h 39m 45s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/9eaf67a905204c47ba0232b48dc821aa : SUCCESS in 3h 27m 22s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/17266229b6c7480f920e311eb69298cb : SUCCESS in 2h 55m 04s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/bd4e523688784786b12cbfea461afa23 : SUCCESS in 1h 08m 56s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/b6354f57c79f4e0c95c553e34b64773d : SUCCESS in 1h 36m 27s (non-voting)\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/40216d36f7464352b677c165e52911c6 : SUCCESS in 1h 46m 25s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/244afadbcd34407c908c34eaa3cb6e54 : SUCCESS in 1h 39m 35s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/dcb4ba2f2c1646ebb5d82feab5bbe941 : SUCCESS in 1h 37m 31s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/43a6c80510f54bc9b937e1e697d93c2b : SUCCESS in 1h 45m 07s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/5752dc96c64641c6bd73caf5d3abd3bb : SUCCESS in 1h 19m 08s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/b905f8a46b724c5d9c01fc3ac6079e7e : SUCCESS in 1h 35m 56s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/92f9454ae1584f3db48e06cbe08dc172 : SUCCESS in 1h 22m 46s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/b15e21bda3f94f85b92f935ffa412bbc : SUCCESS in 1h 03m 55s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/a941747acc464a0790efdc7e27a8e18f : SUCCESS in 1h 31m 41s\n- tripleo-ci-centos-7-content-provider-train https://zuul.opendev.org/t/openstack/build/88ac326364aa4152a5c821419286e8ab : SUCCESS in 3h 44m 40s\n- tripleo-ci-centos-7-content-provider-queens https://zuul.opendev.org/t/openstack/build/92cf4965bb6949ea937f05709f166170 : SUCCESS in 2h 51m 29s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/de5d1c48c16f40a081c9955d6a2f4519 : SUCCESS in 1h 47m 07s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/ab3a7f29d9444141b0a907922423e20e : SUCCESS in 2h 03m 16s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/ef5afbb2c9ef4736996f3ed7ec2e09f5 : SUCCESS in 1h 43m 08s\n- tripleo-ci-centos-7-containers-multinode-train https://zuul.opendev.org/t/openstack/build/906341a5498840a38f289adb260152dd : SUCCESS in 2h 04m 04s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/96685b56c7864acfb469db30ca695f9a : SUCCESS in 1h 54m 21s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/f89ecef6d3a146329ddd8d6e97dd88ee : SUCCESS in 1h 08m 12s\n- tripleo-ci-centos-8-standalone-upgrade https://zuul.opendev.org/t/openstack/build/c83f1b6a220f4db7a7657b6074b74d60 : SUCCESS in 2h 14m 02s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/bd38049512a9462c942a19574362177c : SUCCESS in 4m 21s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/b4e3d837cbc14c63a5940b1c0e032236 : SUCCESS in 15m 15s","accounts_in_message":[],"_revision_number":1},{"id":"ec6bb8dc37eaccd3bac0e3c185b1885761253e73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-28 08:14:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7d252b9ba4143f94f2dccb3d79964d5ad775d055","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-28 11:08:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/00a3d7cce6ad4577860d223821721a89 : SUCCESS in 2h 52m 03s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/8fb622ac8a1a4e239effc59f9edcccdb : SUCCESS in 2h 40m 09s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/6896876de82c4df1808756508ab0bc93 : SUCCESS in 2h 47m 34s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/b24e8136050545309ed58bf8cd5c2252 : SUCCESS in 2h 35m 10s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/70e582b35c434dd9b90c60883d784a10 : SUCCESS in 1h 11m 30s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/882c03df6537492493b267a6a67dc35a : SUCCESS in 1h 48m 36s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/cb45cb61281448878e964b5db68c7e86 : SUCCESS in 1h 38m 14s\n- tripleo-ci-centos-8-containers-undercloud-minion https://zuul.opendev.org/t/openstack/build/005a7d12aae540b2b812740c9ad4f575 : SUCCESS in 1h 07m 37s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/36f0ba5b938f41df84b5404f0c113728 : SUCCESS in 1h 16m 30s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/59aaddb2167144b9b6989d3d23d12a89 : SUCCESS in 1h 42m 57s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/e5308b4b7e8248bf9a0db7374a0dcad4 : SUCCESS in 1h 45m 12s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/5cf298dab1874515bce10434230fbfe4 : SUCCESS in 1h 44m 09s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/b51b9e5753bd4c1cb9544185b65ea1b4 : SUCCESS in 1h 28m 55s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/fe63918b834d4c55b84dcf1bf43bedde : SUCCESS in 1h 24m 52s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/3c1c2994eb9c47cc8d61a40e536af9c9 : SUCCESS in 1h 08m 14s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/a84b91e1a04f4dc3a04a7aa646319d0a : SUCCESS in 1h 25m 10s\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/f8209bc6219e47f6a9a24995368a162e : SUCCESS in 1h 08m 30s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/aca915e879d14f539916c9b8495232bc : SUCCESS in 4m 00s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/2d55a535fe4d49d5a547ea009bff3536 : SUCCESS in 14m 00s","accounts_in_message":[],"_revision_number":1},{"id":"1a87fcec672f01fdf9ed24b388fc92235c70a9cb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-28 11:08:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"508446450d3c273498b04ec0316edd226f588e9b","revisions":{"508446450d3c273498b04ec0316edd226f588e9b":{"kind":"REWORK","_number":1,"created":"2020-06-09 10:18:14.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/70/734570/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/70/734570/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/734570/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/734570/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/734570/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/70/734570/1"}}},"commit":{"parents":[{"commit":"026f0f80a17f1f72975a5e37d16e08a0caae2f77","subject":"Merge \"add ipv6 hotplug tests to skip list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/026f0f80a17f1f72975a5e37d16e08a0caae2f77"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-09 10:18:07.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-09 10:18:07.000000000","tz":120},"subject":"Use unittest.mock instead of mock","message":"Use unittest.mock instead of mock\n\nThe mock third party library was needed for mock support in py2\nruntimes. Since we now only support py36 and later, we can use the\nstandard lib unittest.mock module instead.\n\nChange-Id: Ia06fa93973b153a265ddd0cb2538f5bdabf764aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/508446450d3c273498b04ec0316edd226f588e9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/508446450d3c273498b04ec0316edd226f588e9b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
