)]}'
{"id":"openstack%2Ftripleo-quickstart-extras~762674","triplet_id":"openstack%2Ftripleo-quickstart-extras~master~I58b817336c77808b676783e75eee14255896ee84","project":"openstack/tripleo-quickstart-extras","branch":"master","topic":"bugs/1903993","hashtags":[],"change_id":"I58b817336c77808b676783e75eee14255896ee84","subject":"Skip running tempest after deployment for standalone upgrade jobs","status":"MERGED","created":"2020-11-13 15:39:46.000000000","updated":"2020-11-17 14:09:48.000000000","submitted":"2020-11-17 14:09:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"762674-1605622188486-0535b842","meta_rev_id":"5f0b3c5a4208837a0483796145861bb88e59b5f6","_number":762674,"virtual_id_number":762674,"owner":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},{"value":0,"_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-17 14:09:48.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-11-16 18:35:04.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"date":"2020-11-17 03:10:37.000000000","_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"},"recommended":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"all":[{"value":1,"date":"2020-11-16 16:42:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},{"value":2,"date":"2020-11-16 20:55:00.000000000","_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"},{"value":2,"date":"2020-11-16 21:33:28.000000000","_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":2,"date":"2020-11-16 19:47:50.000000000","_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":1,"date":"2020-11-17 10:52:42.000000000","_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},{"value":0,"_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-16 16:16:52.000000000","updated_by":{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},"reviewer":{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},"state":"REVIEWER"},{"updated":"2020-11-16 16:42:50.000000000","updated_by":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"reviewer":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"state":"REVIEWER"},{"updated":"2020-11-16 18:35:04.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-16 19:47:50.000000000","updated_by":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"reviewer":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"state":"REVIEWER"},{"updated":"2020-11-16 20:55:00.000000000","updated_by":{"_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"},"reviewer":{"_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"},"state":"REVIEWER"},{"updated":"2020-11-16 21:33:28.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"REVIEWER"},{"updated":"2020-11-17 03:10:37.000000000","updated_by":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-17 14:09:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d01fdcc90d80621216769f04366f855f2f8c375c","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2020-11-13 15:39:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4df874fb2e9d4de78753467a123c91e47a9021ba","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2020-11-13 15:44:03.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"cf26a24783d3066a870ad47c6db7ff4eec91473f","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2020-11-13 16:27:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5a635015bedd0ed6ad20501b76f6bc23b98279a9","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2020-11-13 16:40:34.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4b32bcb99567c4d18797f3030113780871194b63","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2020-11-13 16:40:45.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"faa63a6edc2df187e01fa5b3dea1d17a9b56bb81","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2020-11-13 16:42:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fb42e6302801241a1c97f3225f400ca48e7bc905","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2020-11-13 16:43:30.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/4606/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"6c3f97c8758d7a272d51c60a023905aff3c1bcf2","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2020-11-13 19:24:05.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/4607/ : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"a10dc37cedcc5fae20265387e9bb40a132a69794","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2020-11-13 19:42:56.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"49305ede4e0f7bcde9ecd98b5a3932002d678785","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-11-13 19:50:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/5260748192f4461b9674922092166a99 : SUCCESS in 2h 06m 04s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-ussuri-branch https://review.rdoproject.org/zuul/build/a48a544bd69b4514af8282054220905b : SUCCESS in 2h 04m 39s (non-voting)\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/c4cae7fd33d14dc782bfc5c72ae804a5 : SUCCESS in 2h 34m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1d9cedda95c6bf956797b79f67cc9da1ae201f10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-13 22:52:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/67d27369dcb042be8d2365778f94c38e : SUCCESS in 3h 25m 08s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/77994372ea63425693007d458e1e2009 : SUCCESS in 3h 21m 24s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/6342ed8e7a7e4345a4e8ea75d88ebb90 : SUCCESS in 2h 58m 39s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/78ccdc7780da4db0b54272fdaea32332 : SUCCESS in 1h 22m 28s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/106851ef225a463892916e745ba4dcc8 : FAILURE in 27m 49s (non-voting)\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/0a55178d70424d0296feb9d823795486 : SUCCESS in 2h 07m 57s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/315ea53b2f7547828004680012e94b25 : SUCCESS in 1h 56m 01s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/71148f2eb9d84e789a2d24ed7a2c1de6 : SUCCESS in 2h 01m 19s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/d77b86c967ed47fb8d4984a168122623 : SUCCESS in 2h 03m 17s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/1e63551b0ef045bfb8b087f378d820a5 : FAILURE in 21m 40s\n- tripleo-ci-centos-7-content-provider-train https://zuul.opendev.org/t/openstack/build/8ef90de5069847fcb8650c5c0651e0b6 : SUCCESS in 4h 07m 47s\n- tripleo-ci-centos-7-content-provider-queens https://zuul.opendev.org/t/openstack/build/4e53ed61e376410d9cf6e7fb3a5bef3b : SUCCESS in 3h 21m 33s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/f2ca925406614077b57be97ef69e3360 : SUCCESS in 2h 06m 16s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/8a7717e9ea214a9c8df646c142c90803 : SUCCESS in 2h 06m 32s\n- tripleo-ci-centos-7-containers-multinode-train https://zuul.opendev.org/t/openstack/build/1588584cae574c918dcd89e0ad6c1f53 : SUCCESS in 2h 28m 36s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/fecc9cbae8ab46228f57d2269779578f : SUCCESS in 2h 04m 13s\n- tripleo-ci-centos-8-standalone-upgrade https://zuul.opendev.org/t/openstack/build/66c271884bd24a5fb675c94f778aeeed : FAILURE in 19m 21s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ea74a66e369a466fb60029b847bc1257 : SUCCESS in 5m 27s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/ca059ed2894f43c3b04c3d3cfda69987 : SUCCESS in 13m 29s","accounts_in_message":[],"_revision_number":2},{"id":"d6be4f2c5aee163c3a665e6bd72a9ef97a888e02","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2020-11-16 16:14:53.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"5fbe8073eb13320bdcf1e9b12befd8dccb94cff1","author":{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},"date":"2020-11-16 16:16:52.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"b4ce84cae6d4c7b906a53bfc3081d2e0b4edf467","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2020-11-16 16:19:35.000000000","message":"Patch Set 3: Code-Review-1\n\ni think there is a problem here https://28afbbbeb395c38b77eb-80e67ec01f194d5b25d665df26ee3378.ssl.cf1.rackcdn.com/762674/2/check/tripleo-ci-centos-8-standalone-upgrade/66c2718/job-output.txt","accounts_in_message":[],"_revision_number":3},{"id":"d3763ca3bf8af328bda94e1e36f568cbb305fe25","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2020-11-16 16:20:05.000000000","message":"Patch Set 3:\n\n\u003e i think there is a problem here https://28afbbbeb395c38b77eb-80e67ec01f194d5b25d665df26ee3378.ssl.cf1.rackcdn.com/762674/2/check/tripleo-ci-centos-8-standalone-upgrade/66c2718/job-output.txt\n\n2020-11-13 20:17:08.883557 | primary | The error appears to be in \u0027/home/zuul/workspace/.quickstart/playbooks/multinode-standalone.yml\u0027: line 59, column 3, but may\n2020-11-13 20:17:08.883582 | primary | be elsewhere in the file depending on the exact syntax problem.\n2020-11-13 20:17:08.883606 | primary |\n2020-11-13 20:17:08.883630 | primary | The offending line appears to be:\n2020-11-13 20:17:08.883654 | primary |\n2020-11-13 20:17:08.883678 | primary | # standalone-upgrade runtime. If this is an upgrade job fs56 skip this run.\n2020-11-13 20:17:08.883701 | primary | - name: Validate the deployment","accounts_in_message":[],"_revision_number":3},{"id":"a3c2e7b3e2fd00f7b76df695e351cd27cd1e2c1e","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2020-11-16 16:27:56.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"bd362f89fa9cedac74ff2f13baa8920d5179bb1e","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2020-11-16 16:30:39.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/4614/ : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"39b727d57dea8112f8986e06220caa6caf2b1d46","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2020-11-16 16:42:50.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"6378ba6db08eb63caa5b8fcedcabf4c7a7e2b269","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2020-11-16 18:06:16.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/4615/ : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"2691f1f85b766b898dde5244da8deeb16cb628b5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-11-16 18:35:04.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/8613ab1d7427403091ffba6b7f3fe6db : SUCCESS in 2h 05m 06s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-ussuri-branch https://review.rdoproject.org/zuul/build/8992e5b00ddd402ebd08b4679fa34386 : SUCCESS in 2h 00m 34s (non-voting)\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/19e839433e2e4eaea092ee1b18006f44 : FAILURE in 1h 27m 04s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"340ee287d4942bdc432a811ae36a8467631028f3","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2020-11-16 19:47:50.000000000","message":"Patch Set 4: Code-Review+2\n\nThat\u0027s a heck of a topic branch :)","accounts_in_message":[],"_revision_number":4},{"id":"919f8c509e7b9523a8bf2b6b07f4983bb6905432","author":{"_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"},"date":"2020-11-16 20:55:00.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"00507446586cd18c4c8b649d836b950c995f32d2","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2020-11-16 21:33:28.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"7b973a019ac70b6324f55696d4a43d66083939f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 21:46:23.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/c3f4754f706b4b4db52216dab6734381 : SUCCESS in 4h 20m 35s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/654952c8b9d843b5b8ac385990494298 : SUCCESS in 4h 18m 54s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/4586587bbbc44d51ba37ad85bb28f093 : SUCCESS in 4h 36m 07s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/e0186af2063847868495d4922a46cf44 : SUCCESS in 3h 04m 44s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/75ad295317f54a44b5db2c194f7b981c : SUCCESS in 1h 32m 34s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/6bc0c36a602a4d378968f5c325f90d0d : SUCCESS in 1h 48m 32s (non-voting)\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/c68ca9144c794edc8acb43a61bc74a22 : SUCCESS in 2h 02m 36s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/adf634ab30f74a86900bfbaa554971c8 : SUCCESS in 2h 02m 44s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/772d0231cafe40b9a26c9e8ef8d4402f : SUCCESS in 2h 06m 38s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/b25013b403b54ebe88abf2c662f67d27 : SUCCESS in 2h 08m 27s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/e6d5c40423a34c53a5c81a24d614bf80 : SUCCESS in 1h 38m 22s\n- tripleo-ci-centos-7-content-provider-train https://zuul.opendev.org/t/openstack/build/eb3aaf7e1e444f4dbcd67f1c5b9c7c83 : SUCCESS in 2h 27m 55s\n- tripleo-ci-centos-7-content-provider-queens https://zuul.opendev.org/t/openstack/build/4edba850bd814bca9c53039f1d58313d : SUCCESS in 2h 19m 14s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/17ef17d1a5f147ff851cce41527e4866 : SUCCESS in 2h 38m 50s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/314b3b69645e4a3a942028505a5b1727 : SUCCESS in 2h 06m 46s\n- tripleo-ci-centos-7-containers-multinode-train https://zuul.opendev.org/t/openstack/build/69f0decb706a47f6b4d2c81b4036f200 : FAILURE in 1h 20m 45s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/7d3c3d93cd034f7e84de8562af4be93e : FAILURE in 1h 09m 08s\n- tripleo-ci-centos-8-standalone-upgrade https://zuul.opendev.org/t/openstack/build/ca2d2a3bb79f4b98a3baa0e3a98c6972 : SUCCESS in 2h 54m 48s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1bf14902d4254a5d88dea9b633452bbe : SUCCESS in 4m 50s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/94bf671ec1e84270a56feb0c062ee67c : SUCCESS in 13m 05s","accounts_in_message":[],"_revision_number":4},{"id":"94db4b3cbf928980151427996124ad64da638b9a","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2020-11-17 00:29:50.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"406b7f71cedfbf37d17f4305e367a8b53e0d4da4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-11-17 02:39:33.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/31aa4c2026af4e1096ef19e67dea787b : SUCCESS in 2h 08m 27s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-ussuri-branch https://review.rdoproject.org/zuul/build/e43734c71b5e4871bb4fb7162e72264d : SUCCESS in 2h 03m 15s (non-voting)\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/cdfe9c1b4fd1461eb3e28c4aa6d36e9e : FAILURE in 1h 21m 37s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3f06af39afb55e30ec25c6902853c61527c0a042","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2020-11-17 03:10:37.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* tripleo-quickstart-extras-gate-master-delorean-full-featureset052 https://ci.centos.org/job/tripleo-quickstart-extras-gate-master-delorean-full-featureset052/4616/ : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"0c8c17a2494a61425bd4508d6b2c7bc79f7c89c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 05:02:24.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/b25a39152af343288b20e8de32bdea11 : SUCCESS in 4h 31m 41s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/8546156c10f7408d984162c10927e8c5 : SUCCESS in 4h 27m 59s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/87cc174be2f44f43847cd5b383bfc62e : SUCCESS in 3h 15m 52s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/0824e745694446838d072770ca32e238 : SUCCESS in 3h 15m 09s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/501739abe44f428a890f5be558e5d79f : SUCCESS in 1h 11m 33s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/c88709da1203453faf918d97e7d00237 : SUCCESS in 1h 36m 41s (non-voting)\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/7460b39dcd174673b7620bc7799ff1bb : SUCCESS in 2h 00m 35s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/c9630dba98e24ab19a5d433f461ead08 : SUCCESS in 1h 41m 50s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/a4f12a9ee0f74584841b91758b93b648 : SUCCESS in 1h 59m 25s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/295311e478f24845bc7cbfa0a81fd71c : SUCCESS in 2h 28m 44s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/3034e61dacc1465ab279775e4b1950ff : SUCCESS in 1h 56m 47s\n- tripleo-ci-centos-7-content-provider-train https://zuul.opendev.org/t/openstack/build/daefbd982be04e9c978b25cc144d2cbc : SUCCESS in 3h 07m 52s\n- tripleo-ci-centos-7-content-provider-queens https://zuul.opendev.org/t/openstack/build/ec73d77d9a614fec89bb1d8808ff9d47 : SUCCESS in 3h 02m 50s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/754faa56c29f488fb2bacded4e5ae7a5 : SUCCESS in 1h 47m 14s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/1afdf07a8051493ba1ab8b559fe355c3 : SUCCESS in 2h 17m 20s\n- tripleo-ci-centos-7-containers-multinode-train https://zuul.opendev.org/t/openstack/build/2e6c82b164114501a92ed351ce1476fd : SUCCESS in 2h 02m 59s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/ada078ab5dd64eda800efbfc63a1147c : SUCCESS in 2h 02m 36s\n- tripleo-ci-centos-8-standalone-upgrade https://zuul.opendev.org/t/openstack/build/0acee73eb8d84e1ea13626d4578c0f10 : SUCCESS in 3h 07m 09s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/95b8a39164064737bbffbbdf6449c890 : SUCCESS in 4m 08s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/d50ec00f6e904c4ba6365bdf1413958e : SUCCESS in 14m 14s","accounts_in_message":[],"_revision_number":4},{"id":"ae55513718dcd3c682ab801d37f1cd5b905b6626","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2020-11-17 06:36:46.000000000","message":"Patch Set 4:\n\n\u003e That\u0027s a heck of a topic branch :)\n\n:) no idea how that happened","accounts_in_message":[],"_revision_number":4},{"id":"213dc364502dbe30e53ed566346f4268933af949","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2020-11-17 10:52:42.000000000","message":"Patch Set 4: Workflow+1\n\nk going to merge lgtm from the logs here \n\n\n        * https://99f521f9472da75892a8-e65fbda5c4a8fc14eb81d398bd7b0a80.ssl.cf1.rackcdn.com/762674/4/check/tripleo-ci-centos-8-standalone-upgrade/0acee73/job-output.txt\n\n\"skipped after deployment\"\n\n        * 2020-11-17 03:12:38.117981 | primary | PLAY [Validate the undercloud] *************************************************\n        * 2020-11-17 03:12:38.134664 | primary | TASK [validate-services : include_tasks] ***************************************\n\t2020-11-17 03:12:38.134698 | primary | Tuesday 17 November 2020  03:12:38 +0000 (0:00:00.050)       1:06:48.256 ******\n\t2020-11-17 03:12:38.158175 | primary | skipping: [undercloud]\n\t2020-11-17 03:12:38.172856 | primary |\n\t2020-11-17 03:12:38.172893 | primary | TASK [validate-services : include_tasks] ***************************************\n\t2020-11-17 03:12:38.172907 | primary | Tuesday 17 November 2020  03:12:38 +0000 (0:00:00.038)       1:06:48.294 ******\n\t2020-11-17 03:12:38.197549 | primary | skipping: [undercloud]\n\t2020-11-17 03:12:38.202243 | primary |\n\t2020-11-17 03:12:38.202323 | primary | PLAY [Validate the deployment] *************************************************\n\t2020-11-17 03:12:38.217858 | primary |\n\t2020-11-17 03:12:38.217891 | primary | TASK [include_tasks] ***********************************************************\n\t2020-11-17 03:12:38.217926 | primary | Tuesday 17 November 2020  03:12:38 +0000 (0:00:00.045)       1:06:48.339 ******\n\t2020-11-17 03:12:38.244772 | primary | skipping: [undercloud]\n\n\"runs after upgrade\"\n\n      * 2020-11-17 04:14:32.751111 | primary | PLAY [Validate the Standalone upgrade] *****************************************\n\t2020-11-17 04:14:32.769126 | primary |\n\t2020-11-17 04:14:32.769208 | primary | TASK [include_tasks] ***********************************************************\n\t2020-11-17 04:14:32.769222 | primary | Tuesday 17 November 2020  04:14:32 +0000 (0:59:55.956)       1:01:34.271 ******\n\t2020-11-17 04:14:33.159119 | primary | included: /home/zuul/workspace/.quickstart/playbooks/tasks/tempest.yml for undercloud","accounts_in_message":[],"_revision_number":4},{"id":"23ac02a5baef2b2caff7e9175c92d572119591b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 10:52:55.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"4ecc38b071accfeedd1550fac148b17cdac2bc34","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2020-11-17 10:53:09.000000000","message":"Topic changed from I58b817336c77808b676783e75eee14255896ee84 to https://launchpad.net/bugs/1903993","accounts_in_message":[],"_revision_number":4},{"id":"b13a9eb0f396af18be069dcba9b0dd5dc1c11c78","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2020-11-17 10:53:29.000000000","message":"Topic changed from https://launchpad.net/bugs/1903993 to bugs/1903993","accounts_in_message":[],"_revision_number":4},{"id":"1177f3ef6a18aaab2246cb00e7c4f40163251171","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 14:09:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"5f0b3c5a4208837a0483796145861bb88e59b5f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 14:09:48.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/36b81e313d88417b890f98b141210e86 : SUCCESS in 2h 56m 46s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/715b6feb19074fc6bdba7c72fb581122 : SUCCESS in 1h 14m 09s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/8ce9aecf21f64c6a9689665e1de7a449 : SUCCESS in 3h 14m 22s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/2755642df1134f9eaddec8c08da789ee : SUCCESS in 2h 43m 02s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/e28b96ca73b948fdb2af38865d9ca918 : SUCCESS in 1h 20m 16s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/736b19dbf2b7446281f34af03e22e104 : SUCCESS in 1h 55m 01s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/9cf936d1f40b4a3bbba3547d2730c578 : SUCCESS in 1h 58m 26s\n- tripleo-ci-centos-8-containers-undercloud-minion https://zuul.opendev.org/t/openstack/build/fb578200496f410a8291fb0fa2633f25 : SUCCESS in 1h 13m 18s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/3513eaabbf5541d284db8fcf32a6052d : SUCCESS in 1h 33m 09s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/3fd615bb2f4a4e8c9229a10004827de0 : SUCCESS in 2h 21m 41s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/1db6a34081b54a9181b04831d2b9f887 : SUCCESS in 1h 57m 24s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a3f82029e6994a90a6e00d5355c1f4e5 : SUCCESS in 5m 32s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/b4e0abfb99254b09867c516b6b84ddbb : SUCCESS in 12m 46s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"f9ff7116d675d8c9fa042c3d64b123784ab47961","revisions":{"4d9fe3055896c0871717dde6adcdbdb6a98a0547":{"kind":"REWORK","_number":1,"created":"2020-11-13 15:39:46.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/74/762674/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/74/762674/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/74/762674/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/74/762674/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/74/762674/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/74/762674/1"}}},"commit":{"parents":[{"commit":"f90376d7594bd3a2f0ed19c764035bf4b40a02cb","subject":"Merge \"Bump linters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/f90376d7594bd3a2f0ed19c764035bf4b40a02cb"}]}],"author":{"name":"MSA","email":"marios@redhat.com","date":"2020-11-13 15:34:48.000000000","tz":120},"committer":{"name":"MSA","email":"marios@redhat.com","date":"2020-11-13 15:34:48.000000000","tz":120},"subject":"Skip running tempest after deployment for standalone upgrade jobs","message":"Skip running tempest after deployment for standalone upgrade jobs\n\nPer the related bug, the standalone upgrade is running too close\nto 3 hours and times out. Let\u0027s try to reduce that by removing\nthe tempest after deployment with a conditional. It will still\nrun tempest after the upgrade.\n\nChange-Id: I58b817336c77808b676783e75eee14255896ee84\nRelated-Bug: 1903993\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/4d9fe3055896c0871717dde6adcdbdb6a98a0547"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/4d9fe3055896c0871717dde6adcdbdb6a98a0547"}]},"branch":"refs/heads/master"},"a6c6340efea1ec3228b669c44c84d9083aa92ba3":{"kind":"REWORK","_number":2,"created":"2020-11-13 16:40:45.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/74/762674/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/74/762674/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/74/762674/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/74/762674/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/74/762674/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/74/762674/2"}}},"commit":{"parents":[{"commit":"f90376d7594bd3a2f0ed19c764035bf4b40a02cb","subject":"Merge \"Bump linters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/f90376d7594bd3a2f0ed19c764035bf4b40a02cb"}]}],"author":{"name":"MSA","email":"marios@redhat.com","date":"2020-11-13 15:34:48.000000000","tz":120},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2020-11-13 16:40:43.000000000","tz":0},"subject":"Skip running tempest after deployment for standalone upgrade jobs","message":"Skip running tempest after deployment for standalone upgrade jobs\n\nPer the related bug, the standalone upgrade is running too close\nto 3 hours and times out. Let\u0027s try to reduce that by removing\nthe tempest after deployment with a conditional. It will still\nrun tempest after the upgrade.\n\nChange-Id: I58b817336c77808b676783e75eee14255896ee84\nRelated-Bug: 1903993\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/a6c6340efea1ec3228b669c44c84d9083aa92ba3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/a6c6340efea1ec3228b669c44c84d9083aa92ba3"}]},"branch":"refs/heads/master"},"e7c31767b45fae3ac864b2c6707a8d930e591254":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-11-16 16:14:53.000000000","uploader":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"ref":"refs/changes/74/762674/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/74/762674/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/74/762674/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/74/762674/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/74/762674/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/74/762674/3"}}},"commit":{"parents":[{"commit":"b339732aedff7e503a131b63958470c118ca78da","subject":"Merge \"Add dns_nameservers options to undercloud.conf.j2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/b339732aedff7e503a131b63958470c118ca78da"}]}],"author":{"name":"MSA","email":"marios@redhat.com","date":"2020-11-13 15:34:48.000000000","tz":120},"committer":{"name":"Ronelle Landy","email":"rlandy@redhat.com","date":"2020-11-16 16:14:53.000000000","tz":0},"subject":"Skip running tempest after deployment for standalone upgrade jobs","message":"Skip running tempest after deployment for standalone upgrade jobs\n\nPer the related bug, the standalone upgrade is running too close\nto 3 hours and times out. Let\u0027s try to reduce that by removing\nthe tempest after deployment with a conditional. It will still\nrun tempest after the upgrade.\n\nChange-Id: I58b817336c77808b676783e75eee14255896ee84\nRelated-Bug: 1903993\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/e7c31767b45fae3ac864b2c6707a8d930e591254"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/e7c31767b45fae3ac864b2c6707a8d930e591254"}]},"branch":"refs/heads/master"},"f9ff7116d675d8c9fa042c3d64b123784ab47961":{"kind":"REWORK","_number":4,"created":"2020-11-16 16:27:56.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/74/762674/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/74/762674/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/74/762674/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/74/762674/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/74/762674/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/74/762674/4"}}},"commit":{"parents":[{"commit":"b339732aedff7e503a131b63958470c118ca78da","subject":"Merge \"Add dns_nameservers options to undercloud.conf.j2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/b339732aedff7e503a131b63958470c118ca78da"}]}],"author":{"name":"MSA","email":"marios@redhat.com","date":"2020-11-13 15:34:48.000000000","tz":120},"committer":{"name":"MSA","email":"marios@redhat.com","date":"2020-11-16 16:27:11.000000000","tz":120},"subject":"Skip running tempest after deployment for standalone upgrade jobs","message":"Skip running tempest after deployment for standalone upgrade jobs\n\nPer the related bug, the standalone upgrade is running too close\nto 3 hours and times out. Let\u0027s try to reduce that by removing\nthe tempest after deployment with a conditional. It will still\nrun tempest after the upgrade.\n\nChange-Id: I58b817336c77808b676783e75eee14255896ee84\nRelated-Bug: 1903993\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/f9ff7116d675d8c9fa042c3d64b123784ab47961"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/f9ff7116d675d8c9fa042c3d64b123784ab47961"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
