)]}'
{"id":"openstack%2Ftripleo-quickstart-extras~782020","triplet_id":"openstack%2Ftripleo-quickstart-extras~master~Ida4349a1f0fdee5dce0ec2bbc96d0567c6d075d0","project":"openstack/tripleo-quickstart-extras","branch":"master","topic":"fix_octavia_kvm","hashtags":[],"change_id":"Ida4349a1f0fdee5dce0ec2bbc96d0567c6d075d0","subject":"Enable nested virt for scenario 10 octavia kvm job","status":"MERGED","created":"2021-03-22 07:13:28.000000000","updated":"2021-04-06 12:53:01.000000000","submitted":"2021-04-06 12:53:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"782020-fix_octavia_kvm","meta_rev_id":"54f592d6dbec68c2b3076e8f11cec352d1ca70bd","_number":782020,"virtual_id_number":782020,"owner":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-06 12:53:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-04-05 15:57:07.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":2,"date":"2021-04-06 06:48:29.000000000","_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":2,"date":"2021-04-06 08:59:52.000000000","_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":1,"date":"2021-04-06 09:00:05.000000000","_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"}],"reviewers":{"REVIEWER":[{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-22 08:31:57.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-22 08:40:18.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2021-03-22 12:16:04.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2021-03-22 12:24:46.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"REVIEWER"},{"updated":"2021-03-22 18:30:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-23 16:29:23.000000000","updated_by":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"reviewer":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"state":"CC"},{"updated":"2021-04-01 10:40:54.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"},{"updated":"2021-04-01 10:42:42.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"state":"REVIEWER"},{"updated":"2021-04-06 08:59:52.000000000","updated_by":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"reviewer":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"state":"REVIEWER"}],"messages":[{"id":"a1aec47f24065a8fb8b06926bef208233965ad65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-03-22 07:13:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"64cafb171dbcd0fbaf07c8785b1bc757de6343a1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-03-22 08:31:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/14837133b24a4912a86c4b6341e2552f : FAILURE in 1h 16m 02s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/435643ca1fd5408295d429047fedcb71 : FAILURE in 1h 07m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6c44433bb77bc6be48c6d76cad3514c52e5a0596","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-03-22 08:40:18.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ec9e76078db272edb07ef5da5bbf4233dbd233a9","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-03-22 10:48:16.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nHere it got tested : https://logserver.rdoproject.org/73/32573/5/check/periodic-tripleo-ci-centos-8-scenario010-kvm-standalone-master/02280d9/logs/undercloud/var/log/extra/nested_virt_amd.txt.gz","accounts_in_message":[],"_revision_number":1},{"id":"ce80c8ab961baf01cc468e34ec60ad1273c2714f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-03-22 10:48:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1c2d1d485f7fd18823fcf9372fc01782d0f1af8c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-03-22 11:55:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/865c46fc88774113b4847b570c78c3c9 : FAILURE in 1h 04m 46s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/af5ca91a54244428be0ae80c2343c4b7 : FAILURE in 1h 03m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4e7b8edb9a7cd754ead058ebe9b1383c15551a81","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-03-22 12:16:04.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f88ba639e8e92ec488b8d4b3e640342327af02a7","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2021-03-22 12:24:46.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"28c23d7d240c4993acb4ada14d9ed5b3b54deeed","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-03-22 12:51:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7b218dd2c0164e17b9c26b3f2940057bae5b99e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 18:30:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/c8a42753fece4c7cb1409cb40e7b5452 : SUCCESS in 3h 03m 36s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/92ea0d6cc24445a594a3e07f8410e2fd : SUCCESS in 1h 13m 04s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/90d515dc1311471482722fb2ec28308d : SUCCESS in 1h 47m 12s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/e08d2d663ed14f18a66e3b8c1a302585 : SUCCESS in 1h 40m 21s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a803058b7791491898ef76771748caec : SUCCESS in 5m 21s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/6f5b250912d54886baec44d494ae7d59 : SUCCESS in 3h 09m 11s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/c1dbd92a5fa340c9a36af1c28f677b87 : SUCCESS in 3h 16m 18s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/49610fa5a77d4c2a88c2a6927b02d2ff : SUCCESS in 2h 53m 16s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/67eaae4e99b1491fa3652f878e276ada : SUCCESS in 1h 51m 12s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/3cee2a9ada864825be5ef1ba7c7b939f : SUCCESS in 1h 42m 01s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/778db01ff4f74c12a35bd7b47321079e : SUCCESS in 1h 58m 05s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/89314518852b4f67a46b2a03db39b621 : SUCCESS in 1h 26m 20s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/915ccbb372a0473f87b9de5544dfa510 : SUCCESS in 1h 50m 07s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/1ab43c8cc8a44cd2a4faea0d6a194670 : SUCCESS in 12m 42s","accounts_in_message":[],"_revision_number":2},{"id":"5784b84c4e6113c4f50b280c92c4b37919d312d6","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2021-03-23 16:29:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"54629110699a8494a84e176a0c120824b2c60cdf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-04-01 09:30:30.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8103f79a8c48b8974fc13db931fd418e5d42ac7d","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-04-01 09:35:37.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"bd3d3bea2839ef707bf414004567cbbacdacf9eb","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2021-04-01 10:40:54.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"5b6f66c08fc2c8e32004019b93e0dc76e540ada7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-04-01 12:32:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/6b1f48c8281d4b139739780b710ab41f : SUCCESS in 2h 45m 42s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/73e52fd485c040259e34724f9c3b4d5b : SUCCESS in 2h 58m 18s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"870e45ce7e5efbdf8c11759141f8ec2a93165d19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 19:23:56.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/410d2998191f43e8bee13516baf7ef78 : SUCCESS in 2h 47m 43s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/a909f56f4b4c4a28ae0b479075a4fdc0 : SUCCESS in 1h 05m 41s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/ddbf00cd75fe4304883ada792f3421e3 : SUCCESS in 1h 44m 02s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/8230d4f0285948c78c82777b1545400d : FAILURE in 1h 34m 34s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/47de3765be284a23957c0748810a2006 : SUCCESS in 6m 46s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/266b22ce2d0a48b6bcbc20775937e708 : SUCCESS in 2h 54m 48s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/c8881358af0f48e390c38adf15643f39 : SUCCESS in 2h 55m 18s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/40358ebcd4544d35b943b38717fa8986 : SUCCESS in 2h 51m 21s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/68bd8ac5bd2542c683ace80d434acc88 : SUCCESS in 1h 51m 07s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/bb76523491864226a9bdbc4d782809f5 : SUCCESS in 1h 50m 14s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/6884e14ae82f4222a257dcaedb0cae3d : SUCCESS in 1h 54m 43s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/73d521cbe8634a5abb8c4bd52099ca84 : FAILURE in 13m 07s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/cecbb86eb8554456be87e2d8ba72cd7b : SUCCESS in 1h 44m 08s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/e194ff7a150d4adabf348adef60759a0 : SUCCESS in 14m 53s","accounts_in_message":[],"_revision_number":3},{"id":"ebdb5fdf972c3310d8873d320f17230ee884a4e6","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-04-02 06:50:19.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)\n\n-1 for inline sanity check please on the conditional","accounts_in_message":[],"_revision_number":3},{"id":"5ff4aac9c5ba53e4a4fcb74e46db81fbb710fba2","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2021-04-02 14:15:35.000000000","message":"Patch Set 3: -Code-Review","accounts_in_message":[],"_revision_number":3},{"id":"a58ffc372c0508f4862dc52fd1eb75275bb4a020","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-04-05 12:51:05.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0c898f11086b9f198d8fbfd1a1ca1e1e0368fc82","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-04-05 12:51:10.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"9ab2218204ceaa7074ae78cd03ad0aff1354255b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-04-05 15:57:07.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/aa8ab829d8614008a64856e2018d17b6 : SUCCESS in 2h 30m 55s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/414fcfe59ee843459128ede32da3c366 : SUCCESS in 2h 42m 43s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"464b936f60c3aadbc1ae0f943e805b0900b74582","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-05 16:14:03.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/92d641b230e641a79cd8ffe510ea2369 : SUCCESS in 2h 42m 12s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/45ddebca45274b448e8c20342bc9f562 : SUCCESS in 1h 12m 00s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/882134504e6e400393d1cef81f3f933d : SUCCESS in 1h 46m 12s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/90887e25bdaa4bd5a4c0a7d7585a5361 : SUCCESS in 1h 32m 14s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a79df54f0b8e40788c7eb8a6e08076bd : SUCCESS in 5m 18s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/f5c5fd4b1b784ed9bc9daf58d817b042 : SUCCESS in 3h 09m 12s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/8d4631b495d84d38bb2ded42ae4a64ff : SUCCESS in 2h 49m 27s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/2fdbc1bcff57426ab882af3720f67803 : SUCCESS in 2h 19m 13s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/bdc66f6f082c4d43bb32c399bd026cf3 : SUCCESS in 1h 56m 18s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/08a34a6ae9e34f2faaa92f1273c2605e : SUCCESS in 1h 49m 35s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/7952e9761b9a41deba5a03ecbdc608a2 : SUCCESS in 1h 32m 02s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/6a8ed46bc125460689baab6a11a388b5 : SUCCESS in 1h 20m 03s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/909b699ea7ca46e7b2f12166cd975b98 : SUCCESS in 1h 38m 02s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/a8a3fa1373ac49b690637cf65973189f : SUCCESS in 13m 23s","accounts_in_message":[],"_revision_number":4},{"id":"dbf93dd015f714896a86b7050dd22fb85266a562","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-04-06 06:48:29.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"39dab0ef31a087cc93285b8a7dc0a0dcf2aa79a0","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2021-04-06 08:59:52.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"a7c7222897637bded56b6dee5c0de0aa1f8578e6","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2021-04-06 09:00:05.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"5e71e55da647afb7c1757b0b3f3a1d068f2ab6d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 09:00:30.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"7e3c98f8404d4239f22b8a2b2026acdabd31708c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 12:53:00.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/074184d271e942d79bd4007dd6de5a43 : SUCCESS in 3h 06m 46s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/514488db367e438fa3dd10ccf85bcd29 : SUCCESS in 1h 09m 33s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/0de3c238fad741e59470404a21053397 : SUCCESS in 1h 48m 22s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/384fe6c689824c09887248cbc277e0df : SUCCESS in 1h 39m 18s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/a310883644c8470cbf48863efa03f69d : SUCCESS in 3h 04m 34s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/d5fe6fdcc2af48788c2d9a673d533987 : SUCCESS in 2h 59m 11s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/67996494c9494baaae0805e3e5e9743e : SUCCESS in 3h 13m 24s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/37b9759ba2d24d71a366648189f96e90 : SUCCESS in 1h 49m 25s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/fbcaa2eb43604331b576b37e3909f18e : SUCCESS in 1h 51m 54s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/0e0d88ad12064c2b84c278037609a6ce : SUCCESS in 1h 59m 20s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/a1687632c1e6464084f53549d3c28aaa : SUCCESS in 1h 27m 06s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/7daf37283a324026b7e57116af5bf655 : SUCCESS in 1h 49m 38s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/acc59281df0f4eca8cde63ad33e0c9d8 : SUCCESS in 5m 18s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/3749cce37c3b41d4ad1b2a42f8d5afa2 : SUCCESS in 13m 39s","accounts_in_message":[],"_revision_number":4},{"id":"54f592d6dbec68c2b3076e8f11cec352d1ca70bd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 12:53:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"63fea2971dc0d742715455ee6d664cb317453e8c","revisions":{"8bffb4dc159ef3da917dc3dca39a9fad31cf2323":{"kind":"REWORK","_number":1,"created":"2021-03-22 07:13:28.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/20/782020/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/20/782020/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/20/782020/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/20/782020/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/20/782020/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/20/782020/1"}}},"commit":{"parents":[{"commit":"37109c55aa3d089fa09c5f4fa33c424eccccdd16","subject":"Revert \"Upgrade linters (ansible-lint v5)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/37109c55aa3d089fa09c5f4fa33c424eccccdd16"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-03-22 07:12:42.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-03-22 07:13:25.000000000","tz":330},"subject":"Enable nested virt for standalone","message":"Enable nested virt for standalone\n\nFor sc10 kvm jobs, we need to enable nested virt on the host.\nIt adds the execution point for the same.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: Ida4349a1f0fdee5dce0ec2bbc96d0567c6d075d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/8bffb4dc159ef3da917dc3dca39a9fad31cf2323"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/8bffb4dc159ef3da917dc3dca39a9fad31cf2323"}]},"branch":"refs/heads/master"},"4c76a3329d622c062d13bf3f68901e240eb30cbc":{"kind":"REWORK","_number":2,"created":"2021-03-22 10:48:49.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/20/782020/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/20/782020/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/20/782020/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/20/782020/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/20/782020/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/20/782020/2"}}},"commit":{"parents":[{"commit":"37109c55aa3d089fa09c5f4fa33c424eccccdd16","subject":"Revert \"Upgrade linters (ansible-lint v5)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/37109c55aa3d089fa09c5f4fa33c424eccccdd16"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-03-22 07:12:42.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-03-22 10:48:36.000000000","tz":330},"subject":"Enable nested virt for standalone","message":"Enable nested virt for standalone\n\nFor sc10 kvm jobs, we need to enable nested virt on the host.\nIt adds the execution point for the same.\n\nEarlier sc10 kvm jobs were not added to the pipeline.\nWe are working on enabling it here:\nhttps://review.rdoproject.org/r/q/topic:%22fix_octavia_kvm%22+(status:open%20OR%20status:merged)\n\nand here is the test project change:\nhttps://review.rdoproject.org/r/c/testproject/+/32573\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: Ida4349a1f0fdee5dce0ec2bbc96d0567c6d075d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/4c76a3329d622c062d13bf3f68901e240eb30cbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/4c76a3329d622c062d13bf3f68901e240eb30cbc"}]},"branch":"refs/heads/master"},"d1926099cba31397de599bc543b2347770da36f3":{"kind":"REWORK","_number":3,"created":"2021-04-01 09:30:30.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/20/782020/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/20/782020/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/20/782020/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/20/782020/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/20/782020/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/20/782020/3"}}},"commit":{"parents":[{"commit":"37109c55aa3d089fa09c5f4fa33c424eccccdd16","subject":"Revert \"Upgrade linters (ansible-lint v5)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/37109c55aa3d089fa09c5f4fa33c424eccccdd16"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-03-22 07:12:42.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-04-01 09:30:07.000000000","tz":330},"subject":"Enable nested virt for standalone","message":"Enable nested virt for standalone\n\nFor sc10 kvm jobs, we need to enable nested virt on the host.\nIt adds the execution point for the same.\n\nEarlier sc10 kvm jobs were not added to the pipeline.\nWe are working on enabling it here:\nhttps://review.rdoproject.org/r/q/topic:%22fix_octavia_kvm%22+(status:open%20OR%20status:merged)\n\nand here is the test project change:\nhttps://review.rdoproject.org/r/c/testproject/+/32573\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: Ida4349a1f0fdee5dce0ec2bbc96d0567c6d075d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/d1926099cba31397de599bc543b2347770da36f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/d1926099cba31397de599bc543b2347770da36f3"}]},"branch":"refs/heads/master"},"63fea2971dc0d742715455ee6d664cb317453e8c":{"kind":"REWORK","_number":4,"created":"2021-04-05 12:51:05.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/20/782020/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/20/782020/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/20/782020/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/20/782020/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/20/782020/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/20/782020/4"}}},"commit":{"parents":[{"commit":"37109c55aa3d089fa09c5f4fa33c424eccccdd16","subject":"Revert \"Upgrade linters (ansible-lint v5)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/37109c55aa3d089fa09c5f4fa33c424eccccdd16"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-03-22 07:12:42.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-04-05 12:49:33.000000000","tz":330},"subject":"Enable nested virt for scenario 10 octavia kvm job","message":"Enable nested virt for scenario 10 octavia kvm job\n\nFor sc10 kvm jobs, we need to enable nested virt on the host.\nIt adds the execution point for the same.\n\nIt cannot be enabled by default for sc10 kvm jobs as\nnot all upstream cloud support nested virtualization.\n\nEarlier sc10 kvm jobs were not added to the pipeline.\nWe are working on enabling it here:\nhttps://review.rdoproject.org/r/q/topic:%22fix_octavia_kvm%22+(status:open%20OR%20status:merged)\n\nand here is the test project change:\nhttps://review.rdoproject.org/r/c/testproject/+/32573\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: Ida4349a1f0fdee5dce0ec2bbc96d0567c6d075d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/63fea2971dc0d742715455ee6d664cb317453e8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/63fea2971dc0d742715455ee6d664cb317453e8c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
