)]}'
{"id":"openstack%2Ftripleo-quickstart-extras~785040","triplet_id":"openstack%2Ftripleo-quickstart-extras~master~I420d04c4ae83f8471a2cdd8db1a1d1fa5db63e30","project":"openstack/tripleo-quickstart-extras","branch":"master","hashtags":[],"change_id":"I420d04c4ae83f8471a2cdd8db1a1d1fa5db63e30","subject":"add feature to execute repoclosure","status":"MERGED","created":"2021-04-06 20:07:08.000000000","updated":"2021-04-13 18:17:08.000000000","submitted":"2021-04-13 18:17:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"785040","meta_rev_id":"5fb561f5b8153c62290958abe7ca07efadbd0f54","_number":785040,"virtual_id_number":785040,"owner":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8161,"name":"Lon Hohberger","email":"lhh@redhat.com","username":"lon"},{"value":0,"_account_id":22722,"name":"Jason Joyce","email":"fuzzball81@gmail.com","username":"jjoyce"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-13 18:17:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-04-10 00:03:35.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"recommended":{"_account_id":22722,"name":"Jason Joyce","email":"fuzzball81@gmail.com","username":"jjoyce"},"all":[{"value":1,"date":"2021-04-12 12:00:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8161,"name":"Lon Hohberger","email":"lhh@redhat.com","username":"lon"},{"value":1,"date":"2021-04-12 18:16:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22722,"name":"Jason Joyce","email":"fuzzball81@gmail.com","username":"jjoyce"},{"value":2,"date":"2021-04-09 21:06:09.000000000","_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":2,"date":"2021-04-09 21:10:11.000000000","_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"all":[{"value":0,"_account_id":8161,"name":"Lon Hohberger","email":"lhh@redhat.com","username":"lon"},{"value":0,"_account_id":22722,"name":"Jason Joyce","email":"fuzzball81@gmail.com","username":"jjoyce"},{"value":1,"date":"2021-04-09 21:55:40.000000000","_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8161,"name":"Lon Hohberger","email":"lhh@redhat.com","username":"lon"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22722,"name":"Jason Joyce","email":"fuzzball81@gmail.com","username":"jjoyce"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-07 00:46:31.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-07 00:53:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-09 21:10:11.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"REVIEWER"},{"updated":"2021-04-12 12:00:33.000000000","updated_by":{"_account_id":8161,"name":"Lon Hohberger","email":"lhh@redhat.com","username":"lon"},"reviewer":{"_account_id":8161,"name":"Lon Hohberger","email":"lhh@redhat.com","username":"lon"},"state":"REVIEWER"},{"updated":"2021-04-12 18:16:29.000000000","updated_by":{"_account_id":22722,"name":"Jason Joyce","email":"fuzzball81@gmail.com","username":"jjoyce"},"reviewer":{"_account_id":22722,"name":"Jason Joyce","email":"fuzzball81@gmail.com","username":"jjoyce"},"state":"REVIEWER"}],"messages":[{"id":"60c071492091823dd33a765ca1b7b87373360311","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-04-06 20:07:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6585a1977b2c2e1cf6f8acc809b91bd4458d932d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-04-06 21:51:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0a45d2e34b2ef0fc89036a533fdbc86ab60a43df","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-04-07 00:46:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/9b946f3e1f704d9cb062ef6d77638d12 : SUCCESS in 2h 53m 39s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/a9f2cbca4e7e41af9f21963577760539 : SUCCESS in 2h 47m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bcd27266568cb8f40348896516f2642ed9979686","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 00:53:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/16946d156cf24e8db4cc18d8f734c978 : SUCCESS in 2h 46m 21s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/13ef72131c5f4810b1b6833920f232bc : SUCCESS in 1h 09m 16s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/500e025232bc4ef69ebf55f6886b72ad : SUCCESS in 1h 44m 52s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/482f7dc59e774b10b1509dc373319593 : SUCCESS in 1h 33m 02s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1f4086de7e8d4b0cb879b9137fc33c7d : SUCCESS in 6m 12s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/46190930d2c8452c952a51d1bcb17db6 : SUCCESS in 2h 52m 58s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/e528017c2c9640f9adff542f3210fdae : SUCCESS in 2h 50m 28s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/e7e2a2284b44421ba3e5fa25e5e04b26 : SUCCESS in 2h 16m 35s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/d529a2c7c5f44714adcb19f0f15f1b2c : SUCCESS in 1h 50m 02s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/d05e2e13493c456596a0043ae99cc2cb : SUCCESS in 1h 51m 52s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/08a99459ae6546b28e8207cd643a0d75 : SUCCESS in 1h 31m 51s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/38bee7e6eb8f4e56aefa14a370ef832c : SUCCESS in 21m 50s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/99f6723e30d9412880128409df4b7503 : SUCCESS in 1h 43m 01s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/0990a4ee59d94d01a008b7538138b33d : SUCCESS in 12m 40s","accounts_in_message":[],"_revision_number":2},{"id":"38092f7e701f3035cc3b3b29920d5e39054ab966","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-04-07 17:23:29.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8a48ff21a43a1ce4854d743c605f18c77a262e40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 17:52:48.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/465c53665425491ca4cd68fa177ff56b : SUCCESS in 4m 41s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/30d55332cf87462b934bfa480b8eebc6 : SUCCESS in 10m 10s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/1899835345bd41e2b262421dbe9f3978 : FAILURE in 21m 14s","accounts_in_message":[],"_revision_number":3},{"id":"603cf3178366ce66541d3ef283c1627538927b40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-04-07 19:13:11.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6cc4d305e01f1fa1196349321574b8e53660069d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 19:55:12.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4caa53e412dd4191b1ee94b85d293eb0 : SUCCESS in 4m 38s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/ec3da9b5ba84478996a7c8f9550c3e46 : SUCCESS in 11m 32s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/909ce663a3f14d848bb8db020532e6a6 : SUCCESS in 29m 28s","accounts_in_message":[],"_revision_number":4},{"id":"2c09820a29485259bc1d117cb6becdf844493de5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-04-07 20:15:19.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"1077762c6ae0ec0122c6cc7d95d914db814fc22d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 20:58:41.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0452c78d5b0647d880a03568eed8b108 : SUCCESS in 6m 42s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/6c0337567cfe438b9797e59bdf14aacf : SUCCESS in 13m 29s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/18ef3b02ce054e2799a0047d5f71e791 : SUCCESS in 28m 20s","accounts_in_message":[],"_revision_number":5},{"id":"c3f09e6ba0ada40a1de056439f6d8463465958ef","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-04-07 23:40:21.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/5cf8d4ab6fda4941a285240f7c483554 : SUCCESS in 2h 33m 02s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/cfb2f9a0f9bc411796e1edc3a864f36c : SUCCESS in 2h 43m 31s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"10a559403adee817cd044f789f1843d99512f80f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-04-08 00:12:53.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"dbe5632fab3af75b5b235f4a73d9ebf99dfbc312","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 01:18:35.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/93aeca1413a94564880357a3f7864ac5 : SUCCESS in 5m 14s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/6c930863e0bc495e9c93db465d926ac7 : SUCCESS in 12m 17s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/7913a760719046888c9bd18dd1b4db1e : SUCCESS in 50m 41s","accounts_in_message":[],"_revision_number":6},{"id":"2c958243f12530c92d565d21d71333b937777e2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-04-08 02:12:22.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"4fae97540ed7eb4a1c8ec03b7421a1fd33d45acf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 02:41:05.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/56faaaf25d3b4301bffe9ea7efa2ed5c : SUCCESS in 5m 23s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/140fd02e8bf44237a02ef6001b22dcdc : SUCCESS in 12m 57s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/2f0d49d8daa6469d9c3461f836cfc761 : SUCCESS in 17m 32s","accounts_in_message":[],"_revision_number":7},{"id":"fa3d9b59b15435ca2d960dbd51fe2c0ed8a25261","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-04-08 05:01:15.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/bd853cb6bff943618453a2a3c637916b : SUCCESS in 2h 41m 20s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/c48f844f8bf849678a282bc329a7afbe : SUCCESS in 2h 10m 51s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"ca9d6af1563f24b711e19cdc0ab3ee4e14795e7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-04-08 16:33:18.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"3271aad88e64aeb6c7136ff4bef440d401108230","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 17:11:35.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fd1a7dd131fe46d3b57cfde60c09aadf : SUCCESS in 5m 41s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/42b455d53bfd46b0b55fce3910d98b77 : SUCCESS in 12m 50s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/eaad0ab2061d4b83916a84b42fd0fe51 : FAILURE in 22m 28s","accounts_in_message":[],"_revision_number":8},{"id":"1946af02baf67f3d18862e5f8575bbe23da3f726","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-04-08 17:12:52.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"d8f187e0f79710534c129af92d873ee1ef16e5b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-04-08 17:44:36.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"3bda4105eaaaf0c133613ea4cb688641d32e02dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-04-08 18:33:03.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"fbbf9e1080633a7f41e527fc3753a6b7461bdda5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 20:10:03.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/42835a0e77794982b5661a4c89aab8c9 : SUCCESS in 6m 00s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/27a6f8dce04f4313bbf2d977b580e181 : SUCCESS in 13m 58s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/9f5288ed37b64ecfa0a6a501dd2a29a4 : SUCCESS in 1h 26m 40s","accounts_in_message":[],"_revision_number":11},{"id":"7fdbe16fb933e7dc69bb7e9a194033168e504d27","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-04-08 21:09:36.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"721067abf80a2865c4c08e02bdda92f1faaf666a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 21:28:31.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b88de122ecfd466a889c55b1d42f6ab3 : FAILURE in 5m 32s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/04b4327ac7554ad78fbd92ce12eeea45 : SUCCESS in 12m 48s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED","accounts_in_message":[],"_revision_number":12},{"id":"92ad1b1fc55a0c4cafb980b9dc9001f3c05e0d44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-04-08 21:44:02.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"b3cfb86bdf1b18cbb516c5c5d702b86ea40f5a7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 22:06:21.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d6f8a796e9d1464fafb89256bdf3e187 : SUCCESS in 4m 38s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/4fe29b20d6474d8bbdc940d7d64a309d : SUCCESS in 10m 02s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/69681b68f4eb4f75a9a4a5970bcfc9c9 : FAILURE in 13m 35s","accounts_in_message":[],"_revision_number":13},{"id":"a8d1e43242430e9dc72d47a55221b289ac0e9c6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-04-08 22:57:48.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"9958a2fd5c4c7a94c9f0999c751b2f0d2d78901d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 00:39:00.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8c70bc3dfe3c4af48bb5a596e64217d7 : SUCCESS in 5m 47s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/cbe5755f080d474cb86d0130c0903e76 : SUCCESS in 10m 56s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/a3c6c5bb16ed4de28da4c1bba9e24f1b : SUCCESS in 1h 28m 35s","accounts_in_message":[],"_revision_number":14},{"id":"2572dff90c8039b682e269fe9e6dc7761f872832","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-04-09 02:00:52.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/e9acf9a109eb457c94cf31c12f1f9eae : SUCCESS in 2h 41m 05s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/05212688fa894909866630205360b808 : FAILURE in 2h 10m 19s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"b766b215517a399c04bc3135d3f67e7e1f85f964","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-04-09 18:25:58.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"553e5d523e8cd91c42eb17ccad1f7e5759fbe6bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 18:47:09.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c05eec51a1f14bf0b8984b3ed97390c5 : FAILURE in 4m 51s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/9d393e7b716e4194a409053218eb1adf : SUCCESS in 11m 50s\n- tripleo-ci-centos-8-repoclosure https://zuul.opendev.org/t/openstack/build/fa13d85898e84889a90aac5ce7437ec0 : FAILURE in 19m 32s","accounts_in_message":[],"_revision_number":15},{"id":"fda4e76c6e089662813126c6855ef8ec847d3207","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-04-09 18:47:17.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"8814fd1aaf46006f0e12990a1e67113132fba393","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 19:13:59.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/01bce54e06604debba314751fbb141e0 : FAILURE in 6m 06s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/b2664003c4fd4e63b8dd830544482160 : SUCCESS in 17m 53s\n- tripleo-ci-centos-8-repoclosure https://zuul.opendev.org/t/openstack/build/934c5c92f4104f98bd634231035678a2 : FAILURE in 23m 01s","accounts_in_message":[],"_revision_number":16},{"id":"b6b56ba20779530a63ee9772783f214756a88414","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-04-09 19:28:56.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"18e9942bb018372b751edba490fca4fbc91f7ce1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 19:52:54.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b2673899a53e4f56b5c7fd462694305c : SUCCESS in 5m 24s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/3b4bb12f75a548d38ab3b732cfac6de7 : SUCCESS in 15m 36s\n- tripleo-ci-centos-8-repoclosure https://zuul.opendev.org/t/openstack/build/61194283ee3d485cba4b1c6fec4b33cf : FAILURE in 22m 29s","accounts_in_message":[],"_revision_number":17},{"id":"1fa67e6bf3f58b94e6bdc0264af24363bee2ac98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-04-09 20:05:27.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"d652b76f421c2270cc80da5e0375ab2da0e99b86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-04-09 20:38:10.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"f68c95f0bc13e66ea8974a9bb6f96d31c9f2e78f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 21:00:20.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/284e5c7983654e9a9063e2cbc4a96d30 : SUCCESS in 5m 53s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/1d97e0f0c473435ba885aa2a2cf1d058 : SUCCESS in 11m 19s\n- tripleo-ci-centos-8-repoclosure https://zuul.opendev.org/t/openstack/build/c329f17f833f46fd9739464207d8b408 : FAILURE in 20m 14s","accounts_in_message":[],"_revision_number":19},{"id":"bf76d708d2cabaa6a9ad4fc6d3cd3f2ab9cca889","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-04-09 21:01:04.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"01cb4323bb18c09abab270b82d613cd9abd51997","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-04-09 21:06:09.000000000","message":"Patch Set 20: Code-Review+2","accounts_in_message":[],"_revision_number":20},{"id":"6301848adfa5eb10180ec50c5990f9ceea46a3e5","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2021-04-09 21:10:11.000000000","message":"Patch Set 20: Code-Review+2\n\n(1 comment)\n\nonly runs on roles/repoclosure - let\u0027s merge this and go from there","accounts_in_message":[],"_revision_number":20},{"id":"fca6727040e26d9217b5fb92f6be4288dbb5f9fa","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-04-09 21:55:40.000000000","message":"Patch Set 20: Workflow+1","accounts_in_message":[],"_revision_number":20},{"id":"a32712bb52db682bf902298118b8cda7586b7944","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-04-10 00:03:35.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/77eee09750db4fd9a4480cf3853eb622 : SUCCESS in 2h 35m 28s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/310639a6118e4924835bc3ee531f9002 : SUCCESS in 2h 39m 50s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"7c7fe7af508651f08805ba3d6ea48cda08066ffb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-10 00:44:53.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/1be3e9abcd5c4839be5997afb73fca55 : SUCCESS in 3h 35m 16s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/7460db142cfa43e683220625f0b1806a : SUCCESS in 1h 15m 44s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/ed7882c7e1a141c4953488dec730bf2f : SUCCESS in 1h 33m 06s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/be31d13d6cd740deb71d9cfa9c733e31 : SUCCESS in 1h 54m 06s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b206a1e8ad3c4661899954e837d7012f : SUCCESS in 5m 03s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/f53aa69414674b7f87e32c8ceadc19e6 : SUCCESS in 2h 51m 17s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/40dcdcfe32c740b2b4ac47eee0e9039b : SUCCESS in 3h 00m 30s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/faaf4441337a48b19d5e2eae491350b4 : SUCCESS in 3h 23m 31s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/ae1f18e53ee94e179d9ae3d8773807c8 : SUCCESS in 2h 52m 14s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/9e06f28b12e0455ca89d1ff0a3035350 : SUCCESS in 1h 46m 53s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/d3f935cfaf9a47b199c8b6acec1f18dd : SUCCESS in 1h 53m 07s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/db88644921dc4681bc0c7433d120fa8a : SUCCESS in 2h 04m 23s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/f84eacbb82e84306b8ee92f1dc9a48fe : SUCCESS in 1h 53m 26s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/dd57dc591bfc4345bd269805ea0bffa7 : SUCCESS in 1h 32m 10s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/0beea58aadeb4342b3b24eae06ca4d1a : FAILURE in 53m 48s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/b6b741a0c16d4fd89293f3b39d62a588 : SUCCESS in 13m 03s\n- tripleo-ci-centos-8-repoclosure https://zuul.opendev.org/t/openstack/build/a9afdc3e1a884ec288fc0694441577fa : SUCCESS in 23m 15s","accounts_in_message":[],"_revision_number":20},{"id":"e2faf809bf72a569b04c2ea67542c30fdeffdc47","author":{"_account_id":8161,"name":"Lon Hohberger","email":"lhh@redhat.com","username":"lon"},"date":"2021-04-12 12:00:33.000000000","message":"Patch Set 20: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"e777492aae2c57f3e4fdc0df95e1c7e668d2d550","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-04-12 16:56:02.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"f60967abf4abb0fbad37b126d4100edcb406f23c","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-04-12 16:56:17.000000000","message":"Patch Set 20:\n\nrecheck","accounts_in_message":[],"_revision_number":20},{"id":"0b381e2068708ce07fd90b71876ddf180b5d7e56","author":{"_account_id":22722,"name":"Jason Joyce","email":"fuzzball81@gmail.com","username":"jjoyce"},"date":"2021-04-12 18:16:29.000000000","message":"Patch Set 20: Code-Review+1","accounts_in_message":[],"_revision_number":20},{"id":"92e6fa3e01db102f039ffcb30af81ab81605056c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-04-12 19:55:58.000000000","message":"Patch Set 20:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/e8fbc53992f54172b671c522b94fd7c4 : SUCCESS in 2h 46m 38s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/af62cb7efb2c4388a6b2d755329f0380 : SUCCESS in 2h 38m 02s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"f502950c1d8db2786fc0d572aa58443eb4bc80e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 20:23:37.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/7d39a5b173c3411d9c2e7a8b40f276d2 : SUCCESS in 3h 04m 25s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/849e68bbdd6b4d689790bc495fb11e4e : SUCCESS in 1h 13m 27s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/d0ca388db0c94a7e95bd04aed76f6ebe : SUCCESS in 1h 48m 42s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/6e1d7eb7e2d14865944dd87b4b83e972 : SUCCESS in 1h 38m 21s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/bae5c108648746899d9be8af0e37e654 : SUCCESS in 6m 49s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/91f3d7be661248d9ac57a91fd2ed7d3a : SUCCESS in 2h 44m 45s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/5d9d0b4e58e64d51aec551d494d7391d : SUCCESS in 3h 00m 06s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/bc85aca06db944108c745b5cb95dc51e : SUCCESS in 2h 39m 39s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/58af1017a23244dc9caa2ae99cb6fb04 : SUCCESS in 3h 14m 04s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/f77793fc8ba04da898b48860ba249dce : SUCCESS in 1h 46m 54s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/264c765dc8494f1abf8d0ef02bc66c34 : SUCCESS in 1h 52m 59s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/c4ffbc90200745cabdb07cde1cd2ff96 : SUCCESS in 1h 38m 40s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/7ba26a0782e0489dae93eaab4395ed8d : SUCCESS in 2h 05m 59s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/8435799d4caf4ba2bc4cbc44a7bef49e : SUCCESS in 1h 21m 13s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/c4922e12531e4eb48ffa865abf95232b : SUCCESS in 1h 50m 16s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/fa3ebe2f26984454bece4546364a85c2 : SUCCESS in 14m 49s\n- tripleo-ci-centos-8-repoclosure https://zuul.opendev.org/t/openstack/build/e03d6398bba54f3cae0fb045d4b3ca14 : SUCCESS in 24m 49s","accounts_in_message":[],"_revision_number":20},{"id":"38450d30be6b579100c186aeb4987017ef9e60e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 08:44:33.000000000","message":"Patch Set 20: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":20},{"id":"d7160d8cbc6a2578985f19e47171480e170a9d65","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 18:17:07.000000000","message":"Patch Set 20: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/25fa5133ad5840f4aca145944ff9d608 : SUCCESS in 3h 11m 52s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/cd6348d825154803b7977167fd42abe5 : SUCCESS in 1h 15m 00s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/86675bc99bdb463f876ab0cc22b1145e : SUCCESS in 1h 50m 45s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/b7d1c7ae14c44ff7b12890df5c001811 : SUCCESS in 1h 36m 02s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/8458fb0b1d2d4e31a7df8feb0f941489 : SUCCESS in 2h 50m 03s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/700b319f16b34ba28d24d9057b68736c : SUCCESS in 3h 27m 01s (non-voting)\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/35a76a0b82ac49b1a61d1789721768f2 : SUCCESS in 3h 14m 08s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/eba35e4c76e848db922c9d2adfbd5b06 : SUCCESS in 3h 02m 46s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/77399eb83efc493394524a721abbc9c4 : SUCCESS in 1h 52m 14s (non-voting)\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/a3abfb38ed3e46b8aa3690b76c43c898 : SUCCESS in 2h 04m 08s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/44ebc4f732a04c51a8e36bf8399d07ec : SUCCESS in 1h 51m 43s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/9aec7962183c4495ac800074fc3da667 : SUCCESS in 2h 07m 09s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/f179cdced3e74b5699483451b98309d0 : SUCCESS in 1h 20m 42s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/5c5b4edad4b14354aefff33ea0c25814 : SUCCESS in 1h 52m 21s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f01f45fb848c4faa9acda77802c3f274 : SUCCESS in 5m 22s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/84ed6c7c1f8042628469289bfef2589b : SUCCESS in 11m 27s","accounts_in_message":[],"_revision_number":20},{"id":"5fb561f5b8153c62290958abe7ca07efadbd0f54","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 18:17:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":20}],"current_revision_number":20,"current_revision":"b0d8b11297765dee249660b3b7a2b4f38ee7a1e3","revisions":{"746d6c7eaafa2a09e3c21dfcb1b7a01e244b06d7":{"kind":"REWORK","_number":1,"created":"2021-04-06 20:07:08.000000000","uploader":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"ref":"refs/changes/40/785040/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/40/785040/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/1"}}},"commit":{"parents":[{"commit":"d6b7ec625afef9b96f3a60637b89a496c9010d1d","subject":"Merge \"Set cirros image to local one\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/d6b7ec625afef9b96f3a60637b89a496c9010d1d"}]}],"author":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2021-04-06 20:06:36.000000000","tz":-360},"committer":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2021-04-06 20:06:40.000000000","tz":-360},"subject":"WIP: add feature to execute repoclosure","message":"WIP: add feature to execute repoclosure\n\n* execute repoclosure directly after the repos are setup\n* allow the ansible playbook run to stop directly after the\n  repoclosure command.\n* since multiple playbooks are used a follow up patch to\n  a playbook in tqe is required.  For example w/ a standalone\n  job quickstart.yml and multinode-standalone.yml are use.\n* if zuul.repoclosure is defined, stop multinode-standalone.yml\n  from fully executing.\n\nChange-Id: I420d04c4ae83f8471a2cdd8db1a1d1fa5db63e30\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/746d6c7eaafa2a09e3c21dfcb1b7a01e244b06d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/746d6c7eaafa2a09e3c21dfcb1b7a01e244b06d7"}]},"branch":"refs/heads/master"},"bf2ef41ac478e3c192f39898cbb0b3351770370a":{"kind":"REWORK","_number":2,"created":"2021-04-06 21:51:57.000000000","uploader":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"ref":"refs/changes/40/785040/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/40/785040/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/2"}}},"commit":{"parents":[{"commit":"d6b7ec625afef9b96f3a60637b89a496c9010d1d","subject":"Merge \"Set cirros image to local one\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/d6b7ec625afef9b96f3a60637b89a496c9010d1d"}]}],"author":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2021-04-06 20:06:36.000000000","tz":-360},"committer":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2021-04-06 21:51:53.000000000","tz":-360},"subject":"WIP: add feature to execute repoclosure","message":"WIP: add feature to execute repoclosure\n\n* execute repoclosure directly after the repos are setup\n* allow the ansible playbook run to stop directly after the\n  repoclosure command.\n* since multiple playbooks are used a follow up patch to\n  a playbook in tqe is required.  For example w/ a standalone\n  job quickstart.yml and multinode-standalone.yml are use.\n* if zuul.repoclosure is defined, stop multinode-standalone.yml\n  from fully executing.\n\nChange-Id: I420d04c4ae83f8471a2cdd8db1a1d1fa5db63e30\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/bf2ef41ac478e3c192f39898cbb0b3351770370a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/bf2ef41ac478e3c192f39898cbb0b3351770370a"}]},"branch":"refs/heads/master"},"3105f44568a41e3c8676416d67080b64ed5a6492":{"kind":"REWORK","_number":3,"created":"2021-04-07 17:23:29.000000000","uploader":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"ref":"refs/changes/40/785040/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/40/785040/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/3"}}},"commit":{"parents":[{"commit":"d6b7ec625afef9b96f3a60637b89a496c9010d1d","subject":"Merge \"Set cirros image to local one\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/d6b7ec625afef9b96f3a60637b89a496c9010d1d"}]}],"author":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2021-04-06 20:06:36.000000000","tz":-360},"committer":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2021-04-07 17:23:22.000000000","tz":-360},"subject":"WIP: add feature to execute repoclosure","message":"WIP: add feature to execute repoclosure\n\n* execute repoclosure directly after the repos are setup\n* allow the ansible playbook run to stop directly after the\n  repoclosure command.\n* since multiple playbooks are used a follow up patch to\n  a playbook in tqe is required.  For example w/ a standalone\n  job quickstart.yml and multinode-standalone.yml are use.\n* if zuul.repoclosure is defined, stop multinode-standalone.yml\n  from fully executing.\n\nChange-Id: I420d04c4ae83f8471a2cdd8db1a1d1fa5db63e30\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/3105f44568a41e3c8676416d67080b64ed5a6492"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/3105f44568a41e3c8676416d67080b64ed5a6492"}]},"branch":"refs/heads/master"},"ef05a168535038b5f1a0f1d9b2b442c063017780":{"kind":"REWORK","_number":4,"created":"2021-04-07 19:13:11.000000000","uploader":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"ref":"refs/changes/40/785040/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/40/785040/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/4"}}},"commit":{"parents":[{"commit":"d6b7ec625afef9b96f3a60637b89a496c9010d1d","subject":"Merge \"Set cirros image to local one\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/d6b7ec625afef9b96f3a60637b89a496c9010d1d"}]}],"author":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2021-04-06 20:06:36.000000000","tz":-360},"committer":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2021-04-07 19:13:06.000000000","tz":-360},"subject":"WIP: add feature to execute repoclosure","message":"WIP: add feature to execute repoclosure\n\n* execute repoclosure directly after the repos are setup\n* allow the ansible playbook run to stop directly after the\n  repoclosure command.\n* since multiple playbooks are used a follow up patch to\n  a playbook in tqe is required.  For example w/ a standalone\n  job quickstart.yml and multinode-standalone.yml are use.\n* if zuul.repoclosure is defined, stop multinode-standalone.yml\n  from fully executing.\n\nChange-Id: I420d04c4ae83f8471a2cdd8db1a1d1fa5db63e30\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/ef05a168535038b5f1a0f1d9b2b442c063017780"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/ef05a168535038b5f1a0f1d9b2b442c063017780"}]},"branch":"refs/heads/master"},"5bb0b6d50f223c6171be4c42c7fe02da7b336487":{"kind":"REWORK","_number":5,"created":"2021-04-07 20:15:19.000000000","uploader":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"ref":"refs/changes/40/785040/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/40/785040/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/5"}}},"commit":{"parents":[{"commit":"d6b7ec625afef9b96f3a60637b89a496c9010d1d","subject":"Merge \"Set cirros image to local one\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/d6b7ec625afef9b96f3a60637b89a496c9010d1d"}]}],"author":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2021-04-06 20:06:36.000000000","tz":-360},"committer":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2021-04-07 20:15:11.000000000","tz":-360},"subject":"WIP: add feature to execute repoclosure","message":"WIP: add feature to execute repoclosure\n\n* execute repoclosure directly after the repos are setup\n* allow the ansible playbook run to stop directly after the\n  repoclosure command.\n* since multiple playbooks are used a follow up patch to\n  a playbook in tqe is required.  For example w/ a standalone\n  job quickstart.yml and multinode-standalone.yml are use.\n* if zuul.repoclosure is defined, stop multinode-standalone.yml\n  from fully executing.\n\nChange-Id: I420d04c4ae83f8471a2cdd8db1a1d1fa5db63e30\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/5bb0b6d50f223c6171be4c42c7fe02da7b336487"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/5bb0b6d50f223c6171be4c42c7fe02da7b336487"}]},"branch":"refs/heads/master"},"feaf526669c1248c20eebdedbbaa8a72716652b8":{"kind":"REWORK","_number":6,"created":"2021-04-08 00:12:53.000000000","uploader":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"ref":"refs/changes/40/785040/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/40/785040/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/6"}}},"commit":{"parents":[{"commit":"d6b7ec625afef9b96f3a60637b89a496c9010d1d","subject":"Merge \"Set cirros image to local one\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/d6b7ec625afef9b96f3a60637b89a496c9010d1d"}]}],"author":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2021-04-06 20:06:36.000000000","tz":-360},"committer":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2021-04-08 00:12:47.000000000","tz":-360},"subject":"WIP: add feature to execute repoclosure","message":"WIP: add feature to execute repoclosure\n\n* execute repoclosure directly after the repos are setup\n* allow the ansible playbook run to stop directly after the\n  repoclosure command.\n* since multiple playbooks are used a follow up patch to\n  a playbook in tqe is required.  For example w/ a standalone\n  job quickstart.yml and multinode-standalone.yml are use.\n* if zuul.repoclosure is defined, stop multinode-standalone.yml\n  from fully executing.\n\nChange-Id: I420d04c4ae83f8471a2cdd8db1a1d1fa5db63e30\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/feaf526669c1248c20eebdedbbaa8a72716652b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/feaf526669c1248c20eebdedbbaa8a72716652b8"}]},"branch":"refs/heads/master"},"82f17c0679a42d3864cad96f6e4b93d0321fc22a":{"kind":"REWORK","_number":7,"created":"2021-04-08 02:12:22.000000000","uploader":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"ref":"refs/changes/40/785040/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/40/785040/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/7"}}},"commit":{"parents":[{"commit":"d6b7ec625afef9b96f3a60637b89a496c9010d1d","subject":"Merge \"Set cirros image to local one\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/d6b7ec625afef9b96f3a60637b89a496c9010d1d"}]}],"author":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2021-04-06 20:06:36.000000000","tz":-360},"committer":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2021-04-08 02:12:16.000000000","tz":-360},"subject":"WIP: add feature to execute repoclosure","message":"WIP: add feature to execute repoclosure\n\n* execute repoclosure directly after the repos are setup\n* allow the ansible playbook run to stop directly after the\n  repoclosure command.\n* since multiple playbooks are used a follow up patch to\n  a playbook in tqe is required.  For example w/ a standalone\n  job quickstart.yml and multinode-standalone.yml are use.\n* if zuul.repoclosure is defined, stop multinode-standalone.yml\n  from fully executing.\n\nChange-Id: I420d04c4ae83f8471a2cdd8db1a1d1fa5db63e30\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/82f17c0679a42d3864cad96f6e4b93d0321fc22a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/82f17c0679a42d3864cad96f6e4b93d0321fc22a"}]},"branch":"refs/heads/master"},"ca17ec95c95e3d87df4259631df3da1202e4180a":{"kind":"REWORK","_number":8,"created":"2021-04-08 16:33:18.000000000","uploader":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"ref":"refs/changes/40/785040/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/40/785040/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/8"}}},"commit":{"parents":[{"commit":"d6b7ec625afef9b96f3a60637b89a496c9010d1d","subject":"Merge \"Set cirros image to local one\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/d6b7ec625afef9b96f3a60637b89a496c9010d1d"}]}],"author":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2021-04-06 20:06:36.000000000","tz":-360},"committer":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2021-04-08 16:33:11.000000000","tz":-360},"subject":"WIP: add feature to execute repoclosure","message":"WIP: add feature to execute repoclosure\n\n* execute repoclosure directly after the repos are setup\n* allow the ansible playbook run to stop directly after the\n  repoclosure command.\n* since multiple playbooks are used a follow up patch to\n  a playbook in tqe is required.  For example w/ a standalone\n  job quickstart.yml and multinode-standalone.yml are use.\n* if zuul.repoclosure is defined, stop multinode-standalone.yml\n  from fully executing.\n\nChange-Id: I420d04c4ae83f8471a2cdd8db1a1d1fa5db63e30\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/ca17ec95c95e3d87df4259631df3da1202e4180a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/ca17ec95c95e3d87df4259631df3da1202e4180a"}]},"branch":"refs/heads/master"},"1d18ddf257b9978aaabc563e353b77266ecdb2b1":{"kind":"REWORK","_number":9,"created":"2021-04-08 17:12:52.000000000","uploader":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"ref":"refs/changes/40/785040/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/40/785040/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/9"}}},"commit":{"parents":[{"commit":"d6b7ec625afef9b96f3a60637b89a496c9010d1d","subject":"Merge \"Set cirros image to local one\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/d6b7ec625afef9b96f3a60637b89a496c9010d1d"}]}],"author":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2021-04-06 20:06:36.000000000","tz":-360},"committer":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2021-04-08 17:12:47.000000000","tz":-360},"subject":"WIP: add feature to execute repoclosure","message":"WIP: add feature to execute repoclosure\n\n* execute repoclosure directly after the repos are setup\n* allow the ansible playbook run to stop directly after the\n  repoclosure command.\n* since multiple playbooks are used a follow up patch to\n  a playbook in tqe is required.  For example w/ a standalone\n  job quickstart.yml and multinode-standalone.yml are use.\n* if zuul.repoclosure is defined, stop multinode-standalone.yml\n  from fully executing.\n\nChange-Id: I420d04c4ae83f8471a2cdd8db1a1d1fa5db63e30\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/1d18ddf257b9978aaabc563e353b77266ecdb2b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/1d18ddf257b9978aaabc563e353b77266ecdb2b1"}]},"branch":"refs/heads/master"},"173ed607b1f03535780ab63ddc688db4e19523cf":{"kind":"REWORK","_number":10,"created":"2021-04-08 17:44:36.000000000","uploader":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"ref":"refs/changes/40/785040/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/40/785040/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/10"}}},"commit":{"parents":[{"commit":"d6b7ec625afef9b96f3a60637b89a496c9010d1d","subject":"Merge \"Set cirros image to local one\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/d6b7ec625afef9b96f3a60637b89a496c9010d1d"}]}],"author":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2021-04-06 20:06:36.000000000","tz":-360},"committer":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2021-04-08 17:44:28.000000000","tz":-360},"subject":"WIP: add feature to execute repoclosure","message":"WIP: add feature to execute repoclosure\n\n* execute repoclosure directly after the repos are setup\n* allow the ansible playbook run to stop directly after the\n  repoclosure command.\n* since multiple playbooks are used a follow up patch to\n  a playbook in tqe is required.  For example w/ a standalone\n  job quickstart.yml and multinode-standalone.yml are use.\n* if zuul.repoclosure is defined, stop multinode-standalone.yml\n  from fully executing.\n\nChange-Id: I420d04c4ae83f8471a2cdd8db1a1d1fa5db63e30\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/173ed607b1f03535780ab63ddc688db4e19523cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/173ed607b1f03535780ab63ddc688db4e19523cf"}]},"branch":"refs/heads/master"},"353632a2f3f8fef784977d735eba746f548029c0":{"kind":"REWORK","_number":11,"created":"2021-04-08 18:33:03.000000000","uploader":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"ref":"refs/changes/40/785040/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/40/785040/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/11"}}},"commit":{"parents":[{"commit":"d6b7ec625afef9b96f3a60637b89a496c9010d1d","subject":"Merge \"Set cirros image to local one\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/d6b7ec625afef9b96f3a60637b89a496c9010d1d"}]}],"author":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2021-04-06 20:06:36.000000000","tz":-360},"committer":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2021-04-08 18:32:58.000000000","tz":-360},"subject":"WIP: add feature to execute repoclosure","message":"WIP: add feature to execute repoclosure\n\n* execute repoclosure directly after the repos are setup\n* allow the ansible playbook run to stop directly after the\n  repoclosure command.\n* since multiple playbooks are used a follow up patch to\n  a playbook in tqe is required.  For example w/ a standalone\n  job quickstart.yml and multinode-standalone.yml are use.\n* if zuul.repoclosure is defined, stop multinode-standalone.yml\n  from fully executing.\n\nChange-Id: I420d04c4ae83f8471a2cdd8db1a1d1fa5db63e30\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/353632a2f3f8fef784977d735eba746f548029c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/353632a2f3f8fef784977d735eba746f548029c0"}]},"branch":"refs/heads/master"},"d3361a9a10c3c2140c2725084dec31d02be0c424":{"kind":"REWORK","_number":12,"created":"2021-04-08 21:09:36.000000000","uploader":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"ref":"refs/changes/40/785040/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/40/785040/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/12"}}},"commit":{"parents":[{"commit":"d6b7ec625afef9b96f3a60637b89a496c9010d1d","subject":"Merge \"Set cirros image to local one\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/d6b7ec625afef9b96f3a60637b89a496c9010d1d"}]}],"author":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2021-04-06 20:06:36.000000000","tz":-360},"committer":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2021-04-08 21:09:31.000000000","tz":-360},"subject":"WIP: add feature to execute repoclosure","message":"WIP: add feature to execute repoclosure\n\n* execute repoclosure directly after the repos are setup\n* allow the ansible playbook run to stop directly after the\n  repoclosure command.\n* since multiple playbooks are used a follow up patch to\n  a playbook in tqe is required.  For example w/ a standalone\n  job quickstart.yml and multinode-standalone.yml are use.\n* if zuul.repoclosure is defined, stop multinode-standalone.yml\n  from fully executing.\n\nChange-Id: I420d04c4ae83f8471a2cdd8db1a1d1fa5db63e30\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/d3361a9a10c3c2140c2725084dec31d02be0c424"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/d3361a9a10c3c2140c2725084dec31d02be0c424"}]},"branch":"refs/heads/master"},"00f25c511af527d8d89463d3c8f1ebe93f340580":{"kind":"REWORK","_number":13,"created":"2021-04-08 21:44:02.000000000","uploader":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"ref":"refs/changes/40/785040/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/40/785040/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/13"}}},"commit":{"parents":[{"commit":"d6b7ec625afef9b96f3a60637b89a496c9010d1d","subject":"Merge \"Set cirros image to local one\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/d6b7ec625afef9b96f3a60637b89a496c9010d1d"}]}],"author":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2021-04-06 20:06:36.000000000","tz":-360},"committer":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2021-04-08 21:43:57.000000000","tz":-360},"subject":"WIP: add feature to execute repoclosure","message":"WIP: add feature to execute repoclosure\n\n* execute repoclosure directly after the repos are setup\n* allow the ansible playbook run to stop directly after the\n  repoclosure command.\n* since multiple playbooks are used a follow up patch to\n  a playbook in tqe is required.  For example w/ a standalone\n  job quickstart.yml and multinode-standalone.yml are use.\n* if zuul.repoclosure is defined, stop multinode-standalone.yml\n  from fully executing.\n\nChange-Id: I420d04c4ae83f8471a2cdd8db1a1d1fa5db63e30\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/00f25c511af527d8d89463d3c8f1ebe93f340580"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/00f25c511af527d8d89463d3c8f1ebe93f340580"}]},"branch":"refs/heads/master"},"d0d597793b0fcf94c34898aac24534f3cd94d8ea":{"kind":"REWORK","_number":14,"created":"2021-04-08 22:57:48.000000000","uploader":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"ref":"refs/changes/40/785040/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/40/785040/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/14"}}},"commit":{"parents":[{"commit":"d6b7ec625afef9b96f3a60637b89a496c9010d1d","subject":"Merge \"Set cirros image to local one\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/d6b7ec625afef9b96f3a60637b89a496c9010d1d"}]}],"author":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2021-04-06 20:06:36.000000000","tz":-360},"committer":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2021-04-08 22:57:43.000000000","tz":-360},"subject":"WIP: add feature to execute repoclosure","message":"WIP: add feature to execute repoclosure\n\n* execute repoclosure directly after the repos are setup\n* allow the ansible playbook run to stop directly after the\n  repoclosure command.\n* since multiple playbooks are used a follow up patch to\n  a playbook in tqe is required.  For example w/ a standalone\n  job quickstart.yml and multinode-standalone.yml are use.\n* if zuul.repoclosure is defined, stop multinode-standalone.yml\n  from fully executing.\n\nChange-Id: I420d04c4ae83f8471a2cdd8db1a1d1fa5db63e30\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/d0d597793b0fcf94c34898aac24534f3cd94d8ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/d0d597793b0fcf94c34898aac24534f3cd94d8ea"}]},"branch":"refs/heads/master"},"f2d6d5e7582fc122975620ebda6b71f89edae01a":{"kind":"REWORK","_number":15,"created":"2021-04-09 18:25:58.000000000","uploader":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"ref":"refs/changes/40/785040/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/40/785040/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/15"}}},"commit":{"parents":[{"commit":"d6b7ec625afef9b96f3a60637b89a496c9010d1d","subject":"Merge \"Set cirros image to local one\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/d6b7ec625afef9b96f3a60637b89a496c9010d1d"}]}],"author":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2021-04-06 20:06:36.000000000","tz":-360},"committer":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2021-04-09 18:25:30.000000000","tz":-360},"subject":"WIP: add feature to execute repoclosure","message":"WIP: add feature to execute repoclosure\n\n* execute repoclosure directly after the repos are setup\n* allow the ansible playbook run to stop directly after the\n  repoclosure command.\n* since multiple playbooks are used a follow up patch to\n  a playbook in tqe is required.  For example w/ a standalone\n  job quickstart.yml and multinode-standalone.yml are use.\n* if zuul.repoclosure is defined, stop multinode-standalone.yml\n  from fully executing.\n\nDepends-On: https://review.opendev.org/c/openstack/tripleo-ci/+/785690\nChange-Id: I420d04c4ae83f8471a2cdd8db1a1d1fa5db63e30\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/f2d6d5e7582fc122975620ebda6b71f89edae01a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/f2d6d5e7582fc122975620ebda6b71f89edae01a"}]},"branch":"refs/heads/master"},"801b3b530eba7a6c2221d6bfa21095e1c722c1a9":{"kind":"REWORK","_number":16,"created":"2021-04-09 18:47:17.000000000","uploader":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"ref":"refs/changes/40/785040/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/40/785040/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/16"}}},"commit":{"parents":[{"commit":"d6b7ec625afef9b96f3a60637b89a496c9010d1d","subject":"Merge \"Set cirros image to local one\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/d6b7ec625afef9b96f3a60637b89a496c9010d1d"}]}],"author":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2021-04-06 20:06:36.000000000","tz":-360},"committer":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2021-04-09 18:46:22.000000000","tz":-360},"subject":"add feature to execute repoclosure","message":"add feature to execute repoclosure\n\n* execute repoclosure directly after the repos are setup\n* if zuul.repoclosure is defined, stop multinode-standalone.yml\n  from fully executing.\n* new playbook add in 785690 that exits after repoclosure\n\nDepends-On: https://review.opendev.org/c/openstack/tripleo-ci/+/785690\nChange-Id: I420d04c4ae83f8471a2cdd8db1a1d1fa5db63e30\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/801b3b530eba7a6c2221d6bfa21095e1c722c1a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/801b3b530eba7a6c2221d6bfa21095e1c722c1a9"}]},"branch":"refs/heads/master"},"3c54d47b810855fb36d0d6c8706c521af8106541":{"kind":"REWORK","_number":17,"created":"2021-04-09 19:28:56.000000000","uploader":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"ref":"refs/changes/40/785040/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/40/785040/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/17"}}},"commit":{"parents":[{"commit":"d6b7ec625afef9b96f3a60637b89a496c9010d1d","subject":"Merge \"Set cirros image to local one\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/d6b7ec625afef9b96f3a60637b89a496c9010d1d"}]}],"author":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2021-04-06 20:06:36.000000000","tz":-360},"committer":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2021-04-09 19:28:34.000000000","tz":-360},"subject":"add feature to execute repoclosure","message":"add feature to execute repoclosure\n\n* execute repoclosure directly after the repos are setup\n* if zuul.repoclosure is defined, stop multinode-standalone.yml\n  from fully executing.\n* new playbook add in 785690 that exits after repoclosure\n\nDepends-On: https://review.opendev.org/c/openstack/tripleo-ci/+/785690\nChange-Id: I420d04c4ae83f8471a2cdd8db1a1d1fa5db63e30\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/3c54d47b810855fb36d0d6c8706c521af8106541"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/3c54d47b810855fb36d0d6c8706c521af8106541"}]},"branch":"refs/heads/master"},"220a0aef92b712a138b341d42763ef75c75380af":{"kind":"REWORK","_number":18,"created":"2021-04-09 20:05:27.000000000","uploader":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"ref":"refs/changes/40/785040/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/40/785040/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/18"}}},"commit":{"parents":[{"commit":"d6b7ec625afef9b96f3a60637b89a496c9010d1d","subject":"Merge \"Set cirros image to local one\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/d6b7ec625afef9b96f3a60637b89a496c9010d1d"}]}],"author":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2021-04-06 20:06:36.000000000","tz":-360},"committer":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2021-04-09 20:05:22.000000000","tz":-360},"subject":"add feature to execute repoclosure","message":"add feature to execute repoclosure\n\n* execute repoclosure directly after the repos are setup\n* if zuul.repoclosure is defined, stop multinode-standalone.yml\n  from fully executing.\n* new playbook add in 785690 that exits after repoclosure\n\nDepends-On: https://review.opendev.org/c/openstack/tripleo-ci/+/785690\nChange-Id: I420d04c4ae83f8471a2cdd8db1a1d1fa5db63e30\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/220a0aef92b712a138b341d42763ef75c75380af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/220a0aef92b712a138b341d42763ef75c75380af"}]},"branch":"refs/heads/master"},"e076f742190c85587cf49a5ed684080a3d3d5f4a":{"kind":"REWORK","_number":19,"created":"2021-04-09 20:38:10.000000000","uploader":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"ref":"refs/changes/40/785040/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/40/785040/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/19"}}},"commit":{"parents":[{"commit":"d6b7ec625afef9b96f3a60637b89a496c9010d1d","subject":"Merge \"Set cirros image to local one\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/d6b7ec625afef9b96f3a60637b89a496c9010d1d"}]}],"author":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2021-04-06 20:06:36.000000000","tz":-360},"committer":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2021-04-09 20:38:05.000000000","tz":-360},"subject":"add feature to execute repoclosure","message":"add feature to execute repoclosure\n\n* execute repoclosure directly after the repos are setup\n* if zuul.repoclosure is defined, stop multinode-standalone.yml\n  from fully executing.\n* new playbook add in 785690 that exits after repoclosure\n\nDepends-On: https://review.opendev.org/c/openstack/tripleo-ci/+/785690\nChange-Id: I420d04c4ae83f8471a2cdd8db1a1d1fa5db63e30\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/e076f742190c85587cf49a5ed684080a3d3d5f4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/e076f742190c85587cf49a5ed684080a3d3d5f4a"}]},"branch":"refs/heads/master"},"b0d8b11297765dee249660b3b7a2b4f38ee7a1e3":{"kind":"REWORK","_number":20,"created":"2021-04-09 21:01:04.000000000","uploader":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"ref":"refs/changes/40/785040/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/40/785040/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/40/785040/20"}}},"commit":{"parents":[{"commit":"d6b7ec625afef9b96f3a60637b89a496c9010d1d","subject":"Merge \"Set cirros image to local one\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/d6b7ec625afef9b96f3a60637b89a496c9010d1d"}]}],"author":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2021-04-06 20:06:36.000000000","tz":-360},"committer":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2021-04-09 21:00:58.000000000","tz":-360},"subject":"add feature to execute repoclosure","message":"add feature to execute repoclosure\n\n* execute repoclosure directly after the repos are setup\n* if zuul.repoclosure is defined, stop multinode-standalone.yml\n  from fully executing.\n* new playbook add in 785690 that exits after repoclosure\n\nDepends-On: https://review.opendev.org/c/openstack/tripleo-ci/+/785690\nChange-Id: I420d04c4ae83f8471a2cdd8db1a1d1fa5db63e30\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/b0d8b11297765dee249660b3b7a2b4f38ee7a1e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/b0d8b11297765dee249660b3b7a2b4f38ee7a1e3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
