)]}'
{"id":"openstack%2Ftripleo-quickstart-extras~790926","triplet_id":"openstack%2Ftripleo-quickstart-extras~master~I4f9f3f13346761d3dec35844234cb6ae03ad4107","project":"openstack/tripleo-quickstart-extras","branch":"master","topic":"remove_uneeded_tq_requirements","hashtags":[],"change_id":"I4f9f3f13346761d3dec35844234cb6ae03ad4107","subject":"Move build-test-packages as a part of undercloud setup","status":"MERGED","created":"2021-05-12 09:24:24.000000000","updated":"2021-08-10 16:33:50.000000000","submitted":"2021-08-10 16:33:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":16,"unresolved_comment_count":10,"has_review_started":true,"submission_id":"790926-remove_uneeded_tq_requirements","meta_rev_id":"c4cd1b2325d4a10b1a03998dc6bf3fddf7aa3214","_number":790926,"virtual_id_number":790926,"owner":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-10 16:33:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-08-10 06:21:10.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"recommended":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":1,"date":"2021-08-10 06:49:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":2,"date":"2021-08-10 14:34:02.000000000","_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2021-08-10 12:36:41.000000000","_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":2,"date":"2021-08-10 11:24:43.000000000","_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2021-08-10 12:36:41.000000000","_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"CC":[{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-12 09:25:15.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"state":"REVIEWER"},{"updated":"2021-05-12 09:25:15.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"REVIEWER"},{"updated":"2021-05-12 09:25:15.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2021-05-12 09:40:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-12 13:24:04.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-15 12:43:37.000000000","updated_by":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"reviewer":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"state":"REVIEWER"},{"updated":"2021-06-21 03:08:35.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2021-08-02 07:38:32.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2021-08-05 12:04:52.000000000","updated_by":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"reviewer":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"state":"CC"}],"messages":[{"id":"e9fdc0dc6920a87fcac36beaf39344a7d452cc86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-05-12 09:24:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a8f315842bf5cd2369dfa59375b4afe620643b79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 09:40:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9c348ef27c5644ba8c11b9519f3ab858 : FAILURE in 4m 41s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/fde0001ad83243c88a0d5f0292bf6248 : SUCCESS in 15m 00s","accounts_in_message":[],"_revision_number":1},{"id":"7356098423bf01cf5f1eba18a3e613f0c8e77149","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-05-12 09:41:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"97d80a774f772b8282fa29b57891522a1ef70118","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 10:04:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b35367db1d274cbaa9338f439b755d4d : FAILURE in 4m 56s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/6c39bd64e1c14da2a6116ed5533caf44 : SUCCESS in 11m 22s","accounts_in_message":[],"_revision_number":2},{"id":"61396c8a56290fd6253d24cba2a495aea4d1b9e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-05-12 10:16:11.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"abe3ce65777053358fbf7e788b02a71a9c661675","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 13:18:47.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/fa07e4698cbf4f5a86936f117a0f501d : SUCCESS in 2h 40m 22s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/0a6a688b27a045cf89a9e783f656cef6 : SUCCESS in 59m 33s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/388dbdc80ca8404383dec61ce56d17ee : POST_FAILURE in 1h 34m 25s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/d7fb0f70d4744c63b06dad4199efdcb5 : SUCCESS in 1h 29m 02s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/544121131b1d4ce790117528cb90dbaf : SUCCESS in 4m 53s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/9475137f2b8e44ac99b75646bac49caa : SUCCESS in 2h 34m 44s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/a80151947aca4167aa3ab2beca5412d7 : SUCCESS in 2h 54m 54s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/d0df6d00d15a4d9996061a94d85fcd0a : POST_FAILURE in 2h 54m 35s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/aa3ee423654a43bcafb95760b875c1de : FAILURE in 19m 31s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/c4e7fb9a796540ac84c99a70d7aa0689 : SUCCESS in 1h 29m 41s (non-voting)\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/ffa4dc303874497094455e8f7abf6054 : SUCCESS in 1h 44m 23s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/77a4f40b6a954ee28f45a1cd202e4d65 : SUCCESS in 1h 44m 37s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/c4c9634cbb444677b36edb735d8f91b2 : POST_FAILURE in 1h 12m 48s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/b3350786cc8342609dbb85f7ecbf5b04 : SUCCESS in 1h 36m 55s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/197b7a12188b435c9f3fd64fd3bff68f : SUCCESS in 9m 44s","accounts_in_message":[],"_revision_number":3},{"id":"8838c0886fdfd43322e35e3b7cc95fa6af8e1033","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-05-12 13:24:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/5765c3b728694ae7b864af32cc5c09a8 : SUCCESS in 2h 52m 52s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/cc282b35d4c041aa8a22612959e1725e : SUCCESS in 2h 36m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7842266d5b3375a3eb5f46e0ea562600a38ed574","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-05-13 07:24:04.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"69ea947b4f591974b6dbd96155237e3c2ed60b7d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-05-13 09:55:46.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/a473a4dfb3bf4e759c9906e6afd60592 : FAILURE in 36m 07s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/43b6d95d397b4ba7aed8d90d90b85a66 : SUCCESS in 2h 29m 01s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"252b8a246e3c4e370ec2133963454fc0a71697fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 10:38:41.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/3180bba2117142329b52f8cc476fb932 : SUCCESS in 2h 49m 44s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/a7a2d9aa79494e809f8d6bb797688f43 : SUCCESS in 53m 05s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/39e80117971c43f08ceddfbddf203f72 : SUCCESS in 1h 43m 10s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/1f48a0f5eeea4b3a88100151a08f5004 : SUCCESS in 1h 41m 10s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d59cb32d51de4dd197cf25ce3b7871d5 : SUCCESS in 5m 26s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/c23daab75fb248b4b57a93f0bb7411a9 : SUCCESS in 2h 42m 41s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/11844c752781406e9b0f2c564401cc91 : SUCCESS in 3h 04m 04s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/305512628729455e8c566c74e42eefac : SUCCESS in 2h 34m 00s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/0079f39bc24244eba7cd0ab73d6662fe : FAILURE in 19m 11s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/8826453f6b794819875c3f34498c1eb0 : SUCCESS in 1h 49m 52s (non-voting)\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/81bbabebf7394161873d13198f18126d : SUCCESS in 1h 59m 10s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/8aff35624c214d32907a7ddf6de26532 : SUCCESS in 1h 46m 57s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/ed997c754f1e4bf8921b8e18115d5c38 : SUCCESS in 1h 09m 49s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/ebbb32db521a4bab857976fd37a488f6 : SUCCESS in 1h 31m 20s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/bec0c354c2124ea781bb7a25306e7f20 : SUCCESS in 12m 14s","accounts_in_message":[],"_revision_number":4},{"id":"203e5d2db8d5b392d84fd81ca279620d4ac458fb","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-05-13 10:41:22.000000000","message":"Patch Set 4:\n\nSo It is working fine:\nhttps://storage.bhs.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_318/790926/4/check/tripleo-ci-centos-8-content-provider/3180bba/logs/undercloud/home/zuul/install_packages.sh.log\n```\n2021-05-13 07:47:27 | Installing:\n2021-05-13 07:47:27 |  ceph-ansible                           noarch  6.0.2-1.el8                               quickstart-centos-ceph-pacific  219 k\n2021-05-13 07:47:27 |  python3-tripleoclient                  noarch  16.0.1-0.20210513005212.b5ca0f4.el8       delorean-current                534 k\n2021-05-13 07:47:27 |  tripleo-operator-ansible               noarch  0.5.1-0.20210513074459.e80abe7.el8        gating-repo                     266 k\n```","accounts_in_message":[],"_revision_number":4},{"id":"0018a08dcbabb21b0fa095d00eb2bead49bbc504","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2021-05-13 11:04:30.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d0e6f204399d1d82157836e6b316cdb007892dc6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-05-13 12:20:05.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b0cc298f3f8ebe2841079c00dad9216633e7eae7","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-05-13 12:24:40.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a668d4076ff5a3b8dd08fcb340d7cbe0d6e27945","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-05-13 14:57:40.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/9c3e7e84937f4ba4954738323e0c3a12 : FAILURE in 38m 09s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/9cd64f525c394d01b084a1aa03a86064 : FAILURE in 1h 55m 41s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5ec8722d76fae884780e8c10d279d191aa04287a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 16:19:29.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/be19647d1e0c4d4a8fcbeb61c7341b0b : SUCCESS in 3h 43m 54s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/183b7217db6d488eab54429978da2743 : SUCCESS in 1h 23m 46s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/23524c34953d47589ff964664aca12b1 : SUCCESS in 2h 07m 37s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/728f0b0d6ad14a549c23b1becac1bdbe : SUCCESS in 1h 52m 24s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/bcaa6c5bd38c452396e6728ba4f24b64 : SUCCESS in 4m 52s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/8008f53b53224f6e96831c7e3df036be : SUCCESS in 3h 20m 49s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/33af0b33988d430b90c8e9aa96a11fc1 : SUCCESS in 3h 41m 07s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/ebd91529c8d24ea588e9d2dd3de8e2a1 : SUCCESS in 3h 25m 04s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/490dee8397a942588891dee289b9360d : FAILURE in 27m 18s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/02108269237c49d5947834ead39ed88a : SUCCESS in 1h 55m 08s (non-voting)\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/4651f5f38acd46c69750eeb025c8472b : SUCCESS in 2h 10m 26s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/70bc642d070f49e68531cd7185080486 : SUCCESS in 2h 05m 00s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/aab49ee5f26842d2abaaea852e31d484 : SUCCESS in 1h 53m 16s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/491a5eda7adf4812ae1c9f1df75d5245 : SUCCESS in 2h 14m 08s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/97592f8448b94e9fb670c0e303418edf : SUCCESS in 12m 30s","accounts_in_message":[],"_revision_number":5},{"id":"92ef863314392539adb2dccc8a018929033d5609","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-05-13 20:40:07.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"f49dc6316d78fa1a9e340f759180b9eff27ea664","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-05-13 23:31:28.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/5a3030a3f25747c784fcbcba4d95be90 : FAILURE in 36m 08s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/f585b03e12824e77b1e538ee11778415 : SUCCESS in 2h 48m 31s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"517a9a0fa0a5413f92a2cae59ac1aa64c32eb2d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 03:53:32.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/7352cf0171cc4230ac141042e210c145 : SUCCESS in 2h 57m 35s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/625f61354a0548c19bc94138d5fd277c : SUCCESS in 1h 07m 07s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/b8f90962db8d41c59df46257c4aa2a1f : SUCCESS in 1h 48m 47s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/494383706ac14646b4e0baa1b41823e7 : SUCCESS in 1h 39m 31s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6d6b2ed9c94349c8a1813a0b18b2f086 : SUCCESS in 8m 02s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/ac13d9f878cf4f16b76bd8326181894a : SUCCESS in 2h 50m 25s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/70cb6c1f362a4ea9937916e69beb61dc : SUCCESS in 3h 46m 26s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/050db882645848cd92963a18ed107310 : SUCCESS in 2h 57m 43s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/74d8cee3ee2b44da88dd6f17842c6121 : FAILURE in 28m 10s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/449077c5294a442a95477985f24dbd3f : SUCCESS in 1h 45m 54s (non-voting)\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/2dac4efa1dec435f9e71d791b00fcb6f : SUCCESS in 1h 54m 45s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/6136dfced81a4f37b330abc68bff9313 : SUCCESS in 1h 55m 22s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/354d2fdc1d8e43efb46a7c3bc1f49477 : SUCCESS in 1h 24m 26s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/bda5030c8ef947fabfd1f2437ff295d5 : SUCCESS in 1h 48m 49s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/4034e8973f0a4a0fa6901185b569b463 : SUCCESS in 14m 16s","accounts_in_message":[],"_revision_number":5},{"id":"69dffae52463a19857b45f4719d5d42db122f638","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2021-05-25 08:10:14.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"f5208f6c8ffe405d4e0f198ec47cd590246c0968","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-05-25 10:05:27.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/b65ec4ebadd642b6ba617d5fe77ec008 : RETRY_LIMIT in 27m 07s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/845c7877e5e44e2bb59bc3e1fbec6cf6 : RETRY_LIMIT in 21m 38s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"010ac7ab270114937327946e34010185034cb8be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 16:17:20.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/8bee0b3ccd6d4a31acc8e7bb00c93f86 : SUCCESS in 31h 56m 40s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/12b87e5ab9b345cd98b5566b2ec4563e : SUCCESS in 1h 03m 13s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/7b65fe1f780a459e9ad852f0262e9673 : SUCCESS in 1h 37m 16s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/4b7bb15a0dfd4cdeb26ef57b6181ccd2 : SUCCESS in 1h 29m 58s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3bed4da12c874af58bc50aac2370eee7 : SUCCESS in 6m 14s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/68750146189748f89d8d0b6b4c34e3a2 : SUCCESS in 5h 18m 04s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/61249108559b48be881ac62b7fe0c6bc : FAILURE in 31m 40s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/b9fc61047a494ae4b1a64355abb61742 : FAILURE in 31m 03s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/705bb4f4799842b1b2469fe66b0d219c : FAILURE in 18m 34s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/e6bfd14dc56f4956814e83305866e30d : SUCCESS in 1h 40m 11s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/c9f5a511120a454dbbe17fbee7b2a4ed : SUCCESS in 1h 41m 54s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/da143cd6bdae4d769e1bf0c664c59bdd : SUCCESS in 1h 40m 55s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/e3c192aadc1b499082a8b138a289029d : SUCCESS in 12m 41s","accounts_in_message":[],"_revision_number":5},{"id":"bbacafe8301dda7ca832a1d4d9e8b764a9ac804e","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2021-06-15 12:43:37.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"fa4a6b3847964b5d1e893480470fd91fc9e6e394","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-06-18 13:12:20.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"c814d8a96695cfbcdbbbb69a30b576fb54a45037","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-06-18 15:40:11.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/5f7f35a1501047cb8a3e0efb3fb7a1ec : FAILURE in 36m 09s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/671e74f6d5c643b583cb1ec400dbc4f0 : FAILURE in 2h 25m 37s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"af11f69fc316f8348d07766d98446ed8fd7b1e85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 17:29:44.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/2bc75a576e1548518eb0fdd26a19ecf3 : SUCCESS in 3h 51m 16s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/6ac3f759cd304276bd4b9a9654e0dde9 : SUCCESS in 1h 22m 51s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/80336acca3514ce5a8e3d63322826024 : SUCCESS in 1h 55m 38s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/124bf88c0c834c1da01d36dbf85057c7 : SUCCESS in 1h 57m 25s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6646e44f0b944532804048907b04426c : SUCCESS in 9m 03s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/25f1971f47674ca8b1baf3d9c08e5f3f : SUCCESS in 3h 53m 54s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/605e2f9802f14d89b8bf9af77d92b2bb : SUCCESS in 3h 19m 45s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/2292c79f505e432abf14fde6066151b7 : SUCCESS in 3h 51m 09s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/7b6544601595433a8d42866f1f95baa9 : SUCCESS in 3h 31m 54s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/6c99715985fe4edeaa2e0ee47a1bc0e9 : SUCCESS in 2h 02m 37s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/37b6e0c150e742e9b45b89f779e66ad7 : SUCCESS in 2h 05m 49s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/6c8b3762fd7c4eabb5c3c148d2b33e70 : SUCCESS in 2h 29m 09s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/e0e6cf14b68c407eb15c8c6b7c8a5a40 : SUCCESS in 2h 18m 42s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/55b898febf1d4b54b1fdebfd575fc3ef : SUCCESS in 1h 43m 18s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/8dea2ab59142477c9be6dd3ac22f90a6 : SUCCESS in 1h 58m 45s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/580e1d9c72554c9090c9309ac0655644 : SUCCESS in 13m 21s","accounts_in_message":[],"_revision_number":6},{"id":"8758762bf554581c5ac6832f0a4e3f26a0f6d330","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-07-02 06:26:34.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"d581d393da0c1b521cb1d94f71ae4078bb89d272","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-07-02 10:58:26.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/f7913befdf6f49b9a652fb44770dfdc7 : FAILURE in 41m 43s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/4b37e0bbfb51432b8ee2deb0ef96faab : SUCCESS in 3h 25m 14s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"b60862f0b49b4494e05c810249cfdfd56db1e243","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 12:16:50.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/f422445e05564044b9176cf434caf706 : SUCCESS in 3h 50m 57s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/2b79ac855c984fa39174d52a697b2404 : SUCCESS in 1h 32m 03s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/0de7f1389766484687af8f239b5e84c4 : SUCCESS in 2h 12m 44s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/d9240e92184644c7bc8a5cfe1eba068c : SUCCESS in 2h 11m 34s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a90474114f4c4dc0b4978d60fa4a5aa4 : SUCCESS in 6m 29s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/0dce7625e96d4dd2a48adffe254206e5 : SUCCESS in 3h 38m 24s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/5a5b3ba4bdcd470dbc5e9e66c2a2cfdb : SUCCESS in 4h 22m 18s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/e3468b443d224427b339af9240ad81ea : SUCCESS in 3h 27m 11s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/82532a084271405ebe4128ba0f8c948c : SUCCESS in 3h 12m 23s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/d752f84d73984b738a2ea2f9eeff1d90 : SUCCESS in 2h 05m 44s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/6f260639a68b41e28115b5fee0d2987e : SUCCESS in 2h 30m 47s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/fdc980b868c34fa0a9fb5053fb888695 : SUCCESS in 2h 10m 01s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/c2435f364c6946b3ab58a237a820caed : SUCCESS in 2h 03m 34s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/4b0a05cd7ab44521ba9e755b41f9669b : SUCCESS in 1h 44m 16s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/885a2cba8bed46e08f1c7ed4c3ad16a4 : SUCCESS in 2h 08m 51s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/61c3d99949274f74bbf1630a480fa394 : SUCCESS in 12m 51s","accounts_in_message":[],"_revision_number":7},{"id":"733534c39e97e359077a5473b29d1891848d65eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2021-07-05 11:38:29.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"88883b2e18c2a17e1ec4509a9fa7021726bbd70e","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2021-07-05 11:38:57.000000000","message":"Patch Set 8: Code-Review+1\n\nwould be nice to see tests in testproject with building something","accounts_in_message":[],"_revision_number":8},{"id":"dbf19522da323f31b881c5102ce96ce15b72668b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-07-05 15:43:01.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/85c8c9251b0d411caf2a034c086224fb : RETRY_LIMIT in 23m 36s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/027e3a012c7b4d1cbec52c90884db2bc : SUCCESS in 3h 29m 45s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"c2f3647752a66b28cc21b2975e036d7da60d1662","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-05 21:26:18.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/e927f848789541898241ba85b3bdac06 : SUCCESS in 9h 39m 15s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/2374703910ca4b1a811778fcc2a167d7 : SUCCESS in 1h 11m 26s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/2351db9675564c798542dafda0c78f9c : SUCCESS in 1h 45m 11s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/ea10cfa81add4a938c0fce103d0e13da : SUCCESS in 1h 45m 43s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/23e756b05f6b4b81907337e290fe836f : SUCCESS in 5m 40s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/81c0f3d59d1d4d37938be307089dd990 : SUCCESS in 2h 44m 52s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/4b3ed5f9180a49e68926867d849edb03 : SUCCESS in 3h 27m 30s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/b3a606048035426582f60b37120d551d : SUCCESS in 3h 13m 38s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/e77ae57246144be08dc24228df5e1fdf : SUCCESS in 3h 14m 01s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/0bc8f634f3594053b009944d432caa12 : SUCCESS in 1h 39m 49s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/cd7866dac01a414b8a91b844e418b05b : SUCCESS in 2h 03m 59s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/6c8717ba94a3407fa1ec6ffa6307a545 : SUCCESS in 2h 02m 12s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/41d834b3d14044f1a4841d652368f8b1 : SUCCESS in 2h 09m 41s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/4ad5272ee24f41139571f974f5f6c3bd : SUCCESS in 1h 29m 36s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/d6a85a773cb3406f84092e295a83bcc7 : SUCCESS in 1h 38m 25s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/336de30573c0450d8735927a4b1bf854 : SUCCESS in 13m 20s","accounts_in_message":[],"_revision_number":8},{"id":"34ab757e9572201acfd9e178c780940d15fb13b5","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2021-07-06 16:12:43.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"ad5bc37d476d9e086dc48aecf3bb790f14fef710","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-07-07 20:41:47.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"60ba2b3677d5dad52e3a004b0e4978229d47f3f5","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-07-08 07:21:05.000000000","message":"Patch Set 8: Code-Review-1\n\n(2 comments)\n\nlooks OK but -1 mainly for  inline comment on the depends-on \n\nWould be better to see the results of testing before we merge this","accounts_in_message":[],"_revision_number":8},{"id":"89dcda217eac3ea499da58abf36f993a08cbf253","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-07-26 10:10:05.000000000","message":"Patch Set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"20247eea76eaedffffd928566a6c6f47a1518b26","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-07-26 10:22:48.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"661b2556395971c080f1861ee18dd416ea3e9251","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-07-26 13:11:22.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/01103d81962740f58b566aa0dcdd85a0 : SUCCESS in 2h 34m 19s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/9a6444006a0048d0ad16307fdbb2fdfb : SUCCESS in 2h 58m 30s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"64779d42cc39e5bca00a47da6afb3de50a11c085","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 14:03:36.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/03728b00ec6b4c4d9b2af646922cacb6 : SUCCESS in 3h 19m 43s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/2d0357bc98ac4e1a9635581a10fe0ec9 : SUCCESS in 52m 36s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/fc016011cf3b4a148247cbb401036769 : SUCCESS in 1h 53m 21s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/2594a7aad1a9489da4e1abc92fd644db : SUCCESS in 1h 42m 47s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1ab37ee895a0437c886b5c18674b175d : SUCCESS in 5m 43s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/3ee2feac6f424989815c90a3b0e8da46 : SUCCESS in 3h 37m 04s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/a579f539f1c241e1a5f2867fb87d88be : SUCCESS in 3h 42m 16s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/a8228821d83e4d4fa9b86bed5e148123 : SUCCESS in 3h 36m 54s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/7f5048531e54463582c5947a62989786 : SUCCESS in 3h 36m 55s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/62c924aea61c41818100edef96b95fbe : SUCCESS in 1h 56m 24s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/eff047c774ff484688c00b9e50f71c8d : SUCCESS in 2h 17m 47s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/18e0030a2f474b66894ddc18240fc5f6 : SUCCESS in 2h 12m 23s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/95b1de787c354e7891f88596c5863cdd : SUCCESS in 2h 19m 19s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/ac8435238e8749f9aedb320d84fdf4b4 : SUCCESS in 1h 23m 10s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/8c66582f876d49eb9cad2b97d6493a3b : SUCCESS in 1h 32m 06s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/83b39df442064b7bab52131f1d2e318f : SUCCESS in 1h 23m 51s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/14f739512c8e421394af13c29fe86b10 : SUCCESS in 12m 34s","accounts_in_message":[],"_revision_number":9},{"id":"2aceb091a82ce6930a00ebca1ad874d050e7e5fc","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-07-27 09:00:34.000000000","message":"Patch Set 9: Code-Review+1\n\n(2 comments)\n\nk but would still be nice to see some testing result before being happy to merge that","accounts_in_message":[],"_revision_number":9},{"id":"5a2168114978329e14faa564e3ba5e78b95e195c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-07-27 10:48:58.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"d08574dc7716786ff356d6eb3efccfaa93cfeea2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 11:30:20.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/f68c865d46c34673bc8469f9c533f19c : FAILURE in 21m 22s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/325e2985561b43cfb31d3bfcd53b5381 : SUCCESS in 5m 30s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/346e9806f94343dc848728503aedd4c9 : FAILURE in 20m 37s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/1a6e6a6da6674b46b08aac06b7e0e163 : FAILURE in 20m 37s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/90933da99e7042d7b138a1e8b290a7ed : FAILURE in 18m 43s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/0670e28b515c4419823d93d6a3695b45 : FAILURE in 19m 36s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/79741a4355de4b4085e1b25d7e1f3f95 : SUCCESS in 12m 24s","accounts_in_message":[],"_revision_number":10},{"id":"bf1cd01bfe6ae099e809d296b4ff20784f5b9789","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-07-27 11:58:05.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"bf922021dadab56fee89379539f85ce560ac2fee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-07-27 12:19:04.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"f036deba7ab9127222859af117bb12432535d301","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-07-27 15:13:54.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/b68efb4743c84d499cbdc0e8b159d429 : SUCCESS in 2h 15m 23s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/5857ee27dc5d4ae6ac5c18286ab31813 : SUCCESS in 2h 52m 52s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"7fbd7f5d66af8909dbc93fb71bcdcc8358c10e68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 15:50:13.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/491e96a6ad7a470e9f4f409c7bc1e740 : SUCCESS in 3h 04m 27s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/88ae32f2ff8042859f2ec8777ae54fe1 : SUCCESS in 51m 16s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/99f54a78d2e24193828fe499f8c83780 : SUCCESS in 1h 46m 00s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/8d073182633442acba4ca43c8dacef3f : SUCCESS in 1h 40m 46s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9d14b2d2ece442af83bae69585822e49 : SUCCESS in 5m 31s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/9c24cf3fc6504468979a8932d35cac72 : SUCCESS in 3h 02m 49s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/45c1b76e46e84c29a8cca386b02af25f : SUCCESS in 3h 23m 37s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/4a8a9e0d0d0546cca069dd2d9dbc37fa : SUCCESS in 3h 16m 19s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/9d102f9786db45509951306039e218e4 : SUCCESS in 3h 17m 48s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/526f0f643d6e4119895d2cb0e23f11c0 : SUCCESS in 1h 46m 30s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/2d496a3a40e64824a99f4aa4c9876088 : SUCCESS in 2h 04m 31s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/864f3c066d24432885ea164c6862e626 : SUCCESS in 2h 04m 53s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/3e81c4bf9901443193cd019b7d0b522e : SUCCESS in 2h 11m 32s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/6acfe4e8f6584d12be992fa43a9a6485 : SUCCESS in 1h 18m 33s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/92b6b483ef674310a8f1f79dd0971c6a : SUCCESS in 1h 32m 57s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/e52efd9bfb294eb39a0f69cd538af9dd : SUCCESS in 1h 15m 12s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/d57a257253394f90842486f5c5a442e8 : SUCCESS in 11m 46s","accounts_in_message":[],"_revision_number":12},{"id":"7e367d309f6d59ec19a0986f6bee725c5acd243b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-07-28 07:59:43.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"c9bc35916601220c299550a07b2e7a330000ae8e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-07-28 10:20:32.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/cb6cf6eeedd2493980ee0e3781572453 : SUCCESS in 2h 10m 37s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/3732a0cdc2bb4fb6816874c119fa05ae : FAILURE in 1h 41m 28s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"1cbfaa88e2424c66e320518f5e3e4a4f0a898e41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 11:50:22.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/8c0d9cad624f442f9f75a4c2158b6c0d : SUCCESS in 3h 02m 08s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/8865b92d78f641f78230c803ac8b6ae4 : SUCCESS in 1h 04m 46s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/263f50e7b7fc431faa601e7c8d8fc13b : SUCCESS in 1h 50m 34s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/a47b34894b7c4d42aab25fdf15153cbb : FAILURE in 1h 23m 23s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6bdb91616fde41048ae2b4a69fb95809 : SUCCESS in 5m 28s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/bb098a70748c47d899d47fa554bc0417 : SUCCESS in 2h 45m 20s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/a3d4c58952b54cc98e8208a2f7d93718 : SUCCESS in 3h 37m 54s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/f986eedc84ed46b28723935faf6c2842 : SUCCESS in 3h 09m 44s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/f3a15fe6f3da41a1a607fd3aff476fce : SUCCESS in 2h 53m 38s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/99f3114ab02241f0a639b5e4e0f31a2b : SUCCESS in 1h 39m 19s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/accd3929960b47aeaaa802bf05d5412f : SUCCESS in 1h 50m 41s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/ee69e0d052ad49afad6a6a846290e580 : SUCCESS in 2h 06m 13s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/9a4b335a04c04cff9a082cfe33eca617 : SUCCESS in 1h 54m 46s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/81e3ef7a6f2945eeb1cc753350bbdb45 : SUCCESS in 1h 11m 55s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/f229b5acc38d4292818c88e0243421cf : SUCCESS in 1h 28m 38s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/c36c269d4f504a5a8774b18f72f17ea1 : SUCCESS in 1h 16m 36s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/78ce946b94b5442ead3c878f6a6042ae : SUCCESS in 12m 15s","accounts_in_message":[],"_revision_number":13},{"id":"ac7c5f22f60b1ffac1969c0982c904ad522ed367","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-07-28 14:02:31.000000000","message":"Patch Set 13:\n\nTest Patch: https://review.opendev.org/c/openstack/tripleo-heat-templates/+/801296 [[DNM] tripleo-operator-ansible rpm testing]\n\nWhich depends-on: https://review.opendev.org/c/openstack/tripleo-quickstart/+/790701 [[DNM] check rpm building for tripleo-operator]\n\nand which depends-on: https://review.opendev.org/c/openstack/tripleo-quickstart-extras/+/790926 [Move build-test-packages as a part of undercloud setup]\n\nand which depends-on: https://review.opendev.org/c/openstack/tripleo-quickstart/+/790701 [Remove tripleo-operator-ansible from TQE requires]\n\nGoal of testing\n* Make sure tripleo-operator-ansible gets installed from gating repo in various jobs\n\nhttps://fa66a4d347b291f594e0-a902ad727e7b582f9b0c349db2905b1c.ssl.cf2.rackcdn.com/801296/3/check/tripleo-ci-centos-8-content-provider/fd990e8/logs/undercloud/home/zuul/install_packages.sh.log\n\n```\n2021-07-28 08:43:56 | Installing:\n2021-07-28 08:43:56 |  python3-tripleoclient                  noarch  17.0.1-0.20210727194911.78e148f.el8       delorean-current                548 k\n2021-07-28 08:43:56 |  tripleo-operator-ansible               noarch  0.5.2-0.20210728083459.51a294b.el8        gating-repo                     273 k\n```\n\nhttps://23747c8ea869a495f1ca-a232ce3bdc50fca913ceba9a1c600c62.ssl.cf2.rackcdn.com/801296/3/check/tripleo-ci-centos-8-standalone/1d06920/logs/undercloud/home/zuul/install_packages.sh.log\n```\n2021-07-28 10:59:11 |  python3-tripleoclient                  noarch  17.0.1-0.20210727194911.78e148f.el8       delorean-current                548 k\n2021-07-28 10:59:11 |  tripleo-operator-ansible               noarch  0.5.2-0.20210728105330.51a294b.el8        gating-repo                     273 k\n```\n\nhttps://aaeb0a9024d6da4fc3f8-f9c86595ee722ed5765b04c30efd2d71.ssl.cf5.rackcdn.com/801296/3/check/tripleo-ci-centos-8-containers-multinode/c1a1b2c/logs/undercloud/home/zuul/install_packages.sh.log\n\n```\n2021-07-28 11:05:35 | Installing:\n2021-07-28 11:05:35 |  python3-tripleoclient                  noarch  17.0.1-0.20210727194911.78e148f.el8       delorean-current                548 k\n2021-07-28 11:05:35 |  tripleo-operator-ansible               noarch  0.5.2-0.20210728105916.51a294b.el8        gating-repo                     273 k\n2021-07-28 11:05:35 | Installing dependencies:\n```\n\nhttps://9adf80126907a4287188-5c4643ce01a9b304087712e3e08013b4.ssl.cf5.rackcdn.com/801296/3/check/tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates/8c5c95d/logs/undercloud/home/zuul/install_packages.sh.log\n\n```\n2021-07-28 11:36:01 | Installing:\n2021-07-28 11:36:01 |  python3-tripleoclient                  noarch  17.0.1-0.20210727194911.78e148f.el8       delorean-current                548 k\n2021-07-28 11:36:01 |  tripleo-operator-ansible               noarch  0.5.2-0.20210728112953.51a294b.el8        gating-repo                     273 k\n```\n\nOvb testing: https://review.rdoproject.org/r/c/testproject/+/34676\n\nhttps://logserver.rdoproject.org/76/34676/3/check/periodic-tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-master/072ce4b/logs/undercloud/home/zuul/install_packages.sh.log.txt.gz\n```\n2021-07-28 08:30:49 |  python3-tripleoclient                  noarch  17.0.1-0.20210727194911.78e148f.el8     delorean-current                548 k\n2021-07-28 08:30:49 |  tripleo-operator-ansible               noarch  0.5.2-0.20210728082832.51a294b.el8      gating-repo                     273 k\n```\n\nhttps://logserver.rdoproject.org/76/34676/3/check/periodic-tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-master/072ce4b/logs/undercloud/etc/yum.repos.d/gating.repo.txt.gz\n\n```\n[gating-repo]\nname\u003dGating repository\nbaseurl\u003dfile:///opt/gating_repo\nenabled\u003d1\ngpgcheck\u003d0\npriority\u003d1\n```\nand the same repo needs to exist on overcloud\n\nhttps://logserver.rdoproject.org/76/34676/3/check/periodic-tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-master/072ce4b/logs/overcloud-controller-0/etc/yum.repos.d/gating.repo.txt.gz","accounts_in_message":[],"_revision_number":13},{"id":"e7ba966b5f663660496d8652cee0a2b493430ea5","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2021-07-28 14:07:18.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"a557af2c854f1fdd25e19102344de48e0ee5c89a","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-07-28 14:08:26.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"bf0f38b2c223e697e4bb9b9253d0c863626eb13d","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2021-07-28 14:11:38.000000000","message":"Patch Set 13: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"6fd1d08176491ebfc677912afe72bc717e68bd53","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-07-28 14:13:04.000000000","message":"Patch Set 13:\n\nNeed to do one more round of testing for fs02.","accounts_in_message":[],"_revision_number":13},{"id":"b19d5317e1ca3e645ea1c780ddf1f7fcae4c1130","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-07-28 17:20:18.000000000","message":"Patch Set 13:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/c6ae11c7a447461e8d68638618297ecf : SUCCESS in 2h 22m 53s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/7e7bf400c84c40c393fbeae0bd614df2 : SUCCESS in 2h 53m 55s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"eeab804057a475e64691043b9e98da5e3f8e84e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 17:31:18.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/714288a95edd4dbe9ea6f8b466d581f9 : SUCCESS in 2h 50m 08s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/fb1579bc844841b6be095bb31aab6eeb : SUCCESS in 45m 31s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/311505279b594fd3aa589a5e89ef228d : SUCCESS in 1h 33m 16s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/6a5fe13e0b3d42f2b459b7411b868015 : SUCCESS in 1h 33m 04s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/09200cfbe2e74483be17ed73f5026c08 : SUCCESS in 5m 31s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/de3fdc271efd4f7d9b9214bc9cdf49c6 : SUCCESS in 2h 58m 27s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/6b939a7c384b4b68b8c6cccafb4fe87d : SUCCESS in 3h 11m 14s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/479dac6fbe4f41b9987ec28b34fe2e2e : SUCCESS in 3h 06m 58s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/831bdcb7c5a1480c9514bb1e364832fc : SUCCESS in 3h 09m 27s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/6a117475da6c4dfa97c9f789ded8cd37 : SUCCESS in 1h 38m 33s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/e2c8c6313ebb44ce9746c572131bcdda : SUCCESS in 1h 59m 09s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/c7697f6ac0f345a8ab0a985fb9fd05c3 : SUCCESS in 1h 56m 05s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/b359e4572b3c4e33b987d6cd3a46602d : SUCCESS in 2h 04m 55s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/aecaeafde646446fa045fdb74786a206 : SUCCESS in 1h 17m 03s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/95483c422c8e41e4a276da043c92758b : SUCCESS in 1h 25m 42s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/669b89cf762e43ae9eaaa7f29a5a4e4f : SUCCESS in 1h 13m 16s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/207b4cb7d8b2496d97b73f5f5f26dd7b : SUCCESS in 13m 58s","accounts_in_message":[],"_revision_number":13},{"id":"0aa27d769520eba058b5d2b21f77fe3ea6e84fab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-07-29 05:52:41.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"150400d37642d1f1cb52d6b004975cb2188557a1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-07-29 06:17:16.000000000","message":"Patch Set 14: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":14},{"id":"e8dcd6c7147ceb2e77f79eca2b4e2a9dbc7877ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 06:17:18.000000000","message":"Patch Set 14: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":14},{"id":"5d5c72bc9342444a94da9e825deb7e23012d5e15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-07-29 06:18:03.000000000","message":"Patch Set 15: Patch Set 14 was rebased","accounts_in_message":[],"_revision_number":15},{"id":"0519fdd24676f5ebe5dcb3a9d2b86f50e3cc9db7","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-07-29 07:22:41.000000000","message":"Patch Set 15: Code-Review+2\n\nthanks for adding links tests look good","accounts_in_message":[],"_revision_number":15},{"id":"0ffa179496ddfa189d3c98d695800aeb656621b0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-07-29 09:47:59.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/182a413efe2c40dfb0a6239341f3dafb : SUCCESS in 2h 37m 51s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/fdb73caa6b4b4c768bc792a05339fb2d : SUCCESS in 3h 28m 20s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"d7e089f690347dc68cd171f711e50477165c91ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 10:37:03.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/3b1ecf7433094003a3fdf6040edbc087 : SUCCESS in 3h 34m 11s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/589084cb9b1040a29ae8037eccac2cac : SUCCESS in 53m 26s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/086ab5216a1147f595ca152efb9a18be : SUCCESS in 1h 42m 48s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/6a9e66699d2b4e8c8de43c771f80cfbd : SUCCESS in 1h 48m 48s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d400ad742bd7472eb23bf840caf5a4f1 : SUCCESS in 6m 11s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/3781d707c9424e429df87a51ceaddc4f : SUCCESS in 3h 26m 54s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/abd856b45c074a9e9cc2e40894b3960d : SUCCESS in 3h 37m 10s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/07329028becc465ea1038cc178129975 : SUCCESS in 3h 20m 49s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/51cb1714e72e4137aed42b5cae86bd91 : SUCCESS in 3h 18m 22s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/6a1c4df9cb884e589db956e47e58f78e : SUCCESS in 1h 59m 05s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/a24f5cfe133d469b8d3ae38430cca0f4 : SUCCESS in 2h 05m 37s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/eeff636eb9ad46ab8bd378009d8b9f09 : SUCCESS in 2h 07m 09s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/f4f712995d7e4423a9d21a07a513f188 : SUCCESS in 2h 05m 49s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/4cf839bf5ca84d06bf063205800963d1 : SUCCESS in 1h 17m 54s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/00d17c297dc34905ae25342d6c79f3ed : SUCCESS in 1h 43m 26s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/fc27d9ef569940b9ac1cf6e8ed51af32 : SUCCESS in 1h 21m 22s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/1629acea82564cf88eacc488b1d5bfd8 : SUCCESS in 13m 33s","accounts_in_message":[],"_revision_number":15},{"id":"15a2e414f69ca8ae1720868e1c83507586240516","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-08-02 17:27:34.000000000","message":"Patch Set 15: Code-Review+2 Workflow+1\n\ntht and tripleo-operator built successfully on both multinode and standalone\nwell done\n\nhttps://aaeb0a9024d6da4fc3f8-f9c86595ee722ed5765b04c30efd2d71.ssl.cf5.rackcdn.com/801296/3/check/tripleo-ci-centos-8-containers-multinode/c1a1b2c/logs/delorean_logs/component/tripleo/d6/eb/d6eb282a9393008549ad9edcd100d78c3dfc7863_dev/index.html\n\nhttps://23747c8ea869a495f1ca-a232ce3bdc50fca913ceba9a1c600c62.ssl.cf2.rackcdn.com/801296/3/check/tripleo-ci-centos-8-standalone/1d06920/logs/delorean_logs/component/tripleo/d6/eb/d6eb282a9393008549ad9edcd100d78c3dfc7863_dev/index.html","accounts_in_message":[],"_revision_number":15},{"id":"d123eb5c6dd9365ce468a0d2fd97517ac426e10e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 17:28:14.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"4008af1b3ba0e727841c51af42facce10d13be69","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-08-02 18:25:32.000000000","message":"Patch Set 15:\n\nStderr: \u0027level\u003ddebug msg\u003d\"Pull Policy for pull [PullIfNewer]\"\\nerror creating build container: Error reading signatures: Get \"https://access.redhat.com/webassets/docker/content/sigstore/ubi8@sha256\u003daf76fe5eff247fc2bd97331a28333a696bb539d856613daf7e3fe8b52cbc94fc/signature-2\": net/http: TLS handshake timeout\\nFailed to write to log, write /home/zuul/container-builds/746f3dc2-c702-49b5-8ca6-d0fb6bf6012f/base/base-build.log: file already closed\\nFailed to write to log, write /home/zuul/container-builds/746f3dc2-c702-49b5-8ca6-d0fb6bf6012f/base/base-build.log: file already closed\\n\u0027\n2021-08-02 18:02:48.154 52566 ERROR openstack Traceback (most recent call last):\n2021-08-02 18:02:48.154 52566 ERROR openstack   File \"/usr/lib/python3.6/site-packages/cliff/app.py\", line 402, in run_subcommand\n2021-08-02 18:02:48.154 52566 ERROR openstack     result \u003d cmd.run(parsed_args)\n2021-08-02 18:02:48.154 52566 ERROR openstack   File \"/usr/lib/python3.6/site-packages/tripleoclient/command.py\", line 34, in run\n2021-08-02 18:02:48.154 52566 ERROR openstack     super(Command, self).run(parsed_args)\n2021-08-02 18:02:48.154 52566 ERROR openstack   File \"/usr/lib/python3.6/site-packages/osc_lib/command/command.py\", line 39, in run\n2021-08-02 18:02:48.154 52566 ERROR openstack     return super(Command, self).run(parsed_args)\n2021-08-02 18:02:48.154 52566 ERROR openstack   File \"/usr/lib/python3.6/site-packages/cliff/command.py\", line 185, in run\n2021-08-02 18:02:48.154 52566 ERROR openstack     return_code \u003d self.take_action(parsed_args) or 0\n2021-08-02 18:02:48.154 52566 ERROR openstack   File \"/usr/lib/python3.6/site-packages/tripleoclient/v2/tripleo_container_image.py\", line 688, in take_action\n2021-08-02 18:02:48.154 52566 ERROR openstack     bb.build_all()\n2021-08-02 18:02:48.154 52566 ERROR openstack   File \"/usr/lib/python3.6/site-packages/tripleo_common/image/builder/buildah.py\", line 248, in build_all\n2021-08-02 18:02:48.154 52566 ERROR openstack     self._multi_build(containers\u003dcontainers)\n2021-08-02 18:02:48.154 52566 ERROR openstack   File \"/usr/lib/python3.6/site-packages/tripleo_common/image/builder/buildah.py\", line 362, in _multi_build\n2021-08-02 18:02:48.154 52566 ERROR openstack     exceptions\u003d\u0027\\n\u0027.join(exceptions)\n2021-08-02 18:02:48.154 52566 ERROR openstack RuntimeError: \n2021-08-02 18:02:48.154 52566 ERROR openstack The following errors were detected during container build(s):\n2021-08-02 18:02:48.154 52566 ERROR openstack \n\n\nlevel\u003ddebug msg\u003d\"unable to pull and read image \\\"registry.access.redhat.com/ubi8:latest\\\": Error reading signatures: Get \\\"https://access.redhat.com/webassets/docker/content/sigstore/ubi8@sha256\u003daf76fe5eff247fc2bd97331a28333a696bb539d856613daf7e3fe8b52cbc94fc/signature-2\\\": net/http: TLS handshake timeout\"\n\nhttps://storage.bhs.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_0fd/790926/15/gate/tripleo-ci-centos-8-content-provider/0fdb7e1/logs/undercloud/home/zuul/container_image_build.log\n\n\nhttps://storage.bhs.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_0fd/790926/15/gate/tripleo-ci-centos-8-content-provider/0fdb7e1/logs/container-builds/746f3dc2-c702-49b5-8ca6-d0fb6bf6012f/base/base-build.log","accounts_in_message":[],"_revision_number":15},{"id":"5e25bbc38f7f6c199fba2762ab4c1975683158a7","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-08-02 18:34:12.000000000","message":"Patch Set 15:\n\nhttps://bugs.launchpad.net/tripleo/+bug/1938718\nhttps://review.opendev.org/c/openstack/tripleo-ci-health-queries/+/803246","accounts_in_message":[],"_revision_number":15},{"id":"cfc5126cc178a24c0dbc28e397e38a124c1675b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 21:04:27.000000000","message":"Patch Set 15: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/0fdb7e1a1f9d4254b06b990a547c731b : FAILURE in 42m 00s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/5866f87395694ed39d2315370794de94 : SUCCESS in 3h 13m 23s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/8a9dc8c86fd0419f9c87cd9ebc5bcfee : SUCCESS in 3h 21m 35s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/7e5cd8a728af4311a6bca80354d78654 : SUCCESS in 3h 28m 42s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/700a7c22f0ef49919e84ea73a5d22d91 : SUCCESS in 2h 45m 54s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/9dc2d7e1eca041efa10122685d35598b : SUCCESS in 1h 48m 20s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/f345474cc66146e0879bb15f6051169e : SUCCESS in 2h 02m 55s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/df048e0bd9c24c9e8f27f678ae4abf88 : SUCCESS in 2h 10m 07s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/da8acec181d14b71a86418123e64c842 : SUCCESS in 1h 50m 01s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ce30b9b4b310450489602459f601aaf2 : SUCCESS in 5m 41s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/7ee918e886cb41458b004f0643f18ebe : SUCCESS in 12m 51s","accounts_in_message":[],"_revision_number":15},{"id":"1f0f9879302e08cdc85d954867ee637b36d6a93b","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2021-08-02 21:24:07.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"5becdb35a904d786074a660be581ed2dae33bb2d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-08-03 00:06:49.000000000","message":"Patch Set 15:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/525aaa7c5a874e0d8cc9310814843d7c : SUCCESS in 2h 21m 54s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/960225f357f4443ca78bf4af6d7ff47e : SUCCESS in 2h 40m 56s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"ea06a7fc2aa40cc8b0241270a68884744324e19d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 01:10:56.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/9e63b8f91c2b4e06ab6d8e6abbb851d8 : SUCCESS in 3h 11m 44s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/d60901829944436eafda8e8e982f8642 : SUCCESS in 59m 06s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/7aa6b9347eec4027999aefe935fd1616 : SUCCESS in 1h 45m 41s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/38972c7048f24c90b65906e786e33662 : SUCCESS in 1h 46m 59s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f45274c65c2b4d2eb211697e5776f611 : SUCCESS in 4m 56s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/9b1b5030b42d470683f2ed2dc21cbd70 : SUCCESS in 3h 20m 39s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/b8a4a2ea0f46403c91ff1220d4c2f100 : SUCCESS in 3h 32m 02s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/bd708d81db714149b1e135782e8c179f : SUCCESS in 3h 26m 42s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/dbf8584a13e24b09996686c3934b4069 : SUCCESS in 3h 35m 30s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/b97e0ba6740e4c7786a174da8ec97fc0 : SUCCESS in 1h 54m 55s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/9303b5414ea74819a9b96b0ed7d4a496 : SUCCESS in 2h 12m 28s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/68690c489b9c4be68f7850a53562eda7 : SUCCESS in 2h 03m 25s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/dbe858578b89491c848306fd3042c75a : SUCCESS in 2h 18m 44s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/e68916642ba244bc842a14096486eebb : SUCCESS in 1h 26m 36s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/cc78199f12e04e258712f366a0a29b3f : SUCCESS in 1h 31m 39s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/72bbf621775c4847b697f58f35a81da6 : SUCCESS in 1h 18m 29s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/02b6a7feb85d4746ba95dfd270a0fa10 : SUCCESS in 12m 32s","accounts_in_message":[],"_revision_number":15},{"id":"610b6e8c0a6be82a1f455b5fb7cf7a6f2464e0fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 01:31:20.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"786bf7e23206d45e8306499dc3ce004c453a91f7","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-08-03 08:21:40.000000000","message":"Patch Set 15: Code-Review-1\n\nWith this i noticed packages are built in both provider and consumer jobs which is not desired, also will likely impact non zuul jobs by running builds twice once in virthost and then on undercloud.\nhttps://storage.gra.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_239/801296/2/check/tripleo-ci-centos-8-content-provider/239c006/logs/delorean_logs/index.html\nhttps://storage.gra.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_8eb/801296/2/check/tripleo-ci-centos-8-standalone/8eb1155/logs/delorean_logs/index.html\nhttps://storage.gra.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_8eb/801296/2/check/tripleo-ci-centos-8-standalone/8eb1155/logs/undercloud/etc/yum.repos.d/gating.repo","accounts_in_message":[],"_revision_number":15},{"id":"fba58243c2496b5a3fef2ad6860d26c505500e98","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-08-03 08:37:08.000000000","message":"Removed Workflow+1 by wes hayutin \u003cwhayutin@redhat.com\u003e\n","accounts_in_message":[],"_revision_number":15},{"id":"68451441aa53a9749e7df8de90d9726f800beccb","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-08-03 08:37:21.000000000","message":"Patch Set 15:\n\nNeed to fix some last minute changes","accounts_in_message":[],"_revision_number":15},{"id":"7e54f9ea304aea81c5f73809c81507147528301b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-08-03 08:59:29.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"d4baad2c4fbae7f3e025ff3514d8c315dce8a33e","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-08-03 09:00:34.000000000","message":"Patch Set 16:\n\n\u003e Patch Set 15: Code-Review-1\n\u003e \n\u003e With this i noticed packages are built in both provider and consumer jobs which is not desired, also will likely impact non zuul jobs by running builds twice once in virthost and then on undercloud.\n\u003e https://storage.gra.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_239/801296/2/check/tripleo-ci-centos-8-content-provider/239c006/logs/delorean_logs/index.html\n\u003e https://storage.gra.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_8eb/801296/2/check/tripleo-ci-centos-8-standalone/8eb1155/logs/delorean_logs/index.html\n\u003e https://storage.gra.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_8eb/801296/2/check/tripleo-ci-centos-8-standalone/8eb1155/logs/undercloud/etc/yum.repos.d/gating.repo\n\nNow fixed it, it will not get triggered for consumers jobs.","accounts_in_message":[],"_revision_number":16},{"id":"9ae5efd0fc0cc1c381f29acd9014b0b1df3ab2c2","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-08-03 10:17:01.000000000","message":"Patch Set 16: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"74e974ecf90d3343936efa0c1931866e11669d4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-08-03 11:35:40.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"6bd979bebfe0215c7d2382f7e86c3cddbc9d56a7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-08-03 14:11:54.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/e78855fa535d4d239da24f9d130a5edb : SUCCESS in 2h 32m 47s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/93a0e84082bb40c08448717769be27f2 : SUCCESS in 2h 35m 03s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"c3dc7b0082a89985d139a57b0df9900308321662","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-08-03 14:38:48.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"c833640aaad49f57d48ca02d9a9b9511de006287","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-08-03 16:59:40.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/19726c9b311b4eadbf962ecfffeddddd : SUCCESS in 2h 19m 27s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/3aad01fc2b7249058d0c9533e728fb57 : FAILURE in 2h 13m 13s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"4da863335596af45dee14aa668a99e3108d7774d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 18:14:31.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/d61a4da522e44622aa7c315f0e58dc1b : SUCCESS in 2h 42m 21s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/f3b8a45fa33f45f5b9eb0ee1990f40fc : SUCCESS in 44m 10s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/5339bd86cd6d4911b0efec8717f41f7b : SUCCESS in 1h 29m 16s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/e604bcef1c5f4583a15cfdf97b0bf8f5 : SUCCESS in 1h 24m 48s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/28c9cb2b8fd048abb3962b111fb311e9 : SUCCESS in 5m 39s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/32731c19cf49406d96571702ec69b766 : SUCCESS in 2h 53m 06s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/2225336472d545c3bc3746d86383a698 : SUCCESS in 3h 17m 25s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/58eb39776cb245b0a02cff55563aa5e7 : SUCCESS in 3h 27m 20s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/bd95de2d58f042c592dc0b278bfc01e7 : SUCCESS in 3h 04m 22s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/31a13242df3a4d75a4b5c25bda866e9b : SUCCESS in 1h 37m 52s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/0bbfcb3dcf45408585f1405eb12e635f : SUCCESS in 2h 02m 03s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/804f8391dd624e2d81ea3fa127933ac4 : SUCCESS in 2h 10m 14s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/4de54d69c48b4cedadd8b2770a9bb899 : SUCCESS in 2h 02m 24s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/18de06d69f214955a84600e497f8144f : SUCCESS in 1h 37m 03s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/e04925aa529c4d4b9a68fa3050c20702 : SUCCESS in 1h 23m 24s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/613f52648f6b47aeaf2a3cc0cb192509 : SUCCESS in 1h 05m 34s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/ddd196833b3e47ad97acdb537d56b7e3 : SUCCESS in 13m 45s","accounts_in_message":[],"_revision_number":18},{"id":"c603d04efadac01960f34391c2d11ff3589e2e87","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-08-04 07:16:08.000000000","message":"Patch Set 18: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"2940078dba08b1282606d5116898f84d480203fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-08-04 09:02:56.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"95e09b02973586a406481dec15fbcb9d556ffd9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-08-04 09:31:40.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"b4de6588ea3bfa490d629a240d5a8b16d787f035","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-04 09:44:32.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4ca691c3b14747fa8b1d5e718e8bef14 : FAILURE in 4m 39s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/624351e4da814098b32749f293334187 : SUCCESS in 11m 08s","accounts_in_message":[],"_revision_number":20},{"id":"5bec7ca049c86f6b68aa60415d2d2040a3b3deca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-08-04 09:59:23.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"a88455e63a01e6352e9dbab36b01bd48270a9ff6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-08-04 11:55:30.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"d289c87112800552b6bf496aaaad94ccd9ec348d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2021-08-04 13:49:55.000000000","message":"Patch Set 23: Published edit on patch set 22.","accounts_in_message":[],"_revision_number":23},{"id":"f4199879c3385ea0578af69acb2aa8dc7a6e279c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-08-04 16:28:31.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/c58b909e36ec40eca83ef99a0e83d3f3 : FAILURE in 2h 12m 48s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/ec021e6823834ad99a35a9161838591d : SUCCESS in 2h 37m 14s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"27bd731ca4afe13b4d7ba999913f3ab164ddfbe9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-04 17:17:27.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/6d5409c195104a09a16f09f82a433b94 : SUCCESS in 3h 19m 27s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/84ec7b7c40fb429b8c4f938330c5677b : SUCCESS in 45m 28s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/a27a1b2581e4405caece3e2b3c0f8e32 : SUCCESS in 1h 33m 23s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/770f9170fdc1440ab103d7df7a0b89cc : SUCCESS in 2h 07m 47s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/178b13cc48184245bc39e797ff17096d : SUCCESS in 5m 41s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/ce65ccd27e3b41a592d8b96a23c0073b : SUCCESS in 2h 54m 09s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/cf251281206740d2809e6243df2591a8 : SUCCESS in 2h 59m 24s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/ad22e2e5327e49bfb8fc340f33f12e8a : SUCCESS in 3h 05m 49s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/b2eec583e7f54c5ea672083801b95464 : SUCCESS in 2h 42m 39s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/0aca1aae69234f17aadee3b283e5857f : SUCCESS in 1h 43m 25s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/1719832843784ce4bc72a0e8a209327f : SUCCESS in 1h 47m 05s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/47bca674706d4d4ba086da1696f2dc01 : SUCCESS in 1h 58m 44s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/00e46642344a4287b132a1018a489e30 : SUCCESS in 1h 54m 14s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/b237b4b6e9454480bd0d816aaaed3d94 : SUCCESS in 1h 15m 27s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/290aafdf0155446ea0b3aecbf992d8b1 : SUCCESS in 2h 01m 46s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/1efc5aac74104e1c8a2051c4f9b4810b : SUCCESS in 1h 09m 49s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/7a31913b63b34f819493f68a365f7df2 : SUCCESS in 13m 15s","accounts_in_message":[],"_revision_number":23},{"id":"1774cd67452df34d62b090569799cb5e5def3be3","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2021-08-05 12:04:52.000000000","message":"Patch Set 23:\n\n(2 comments)","accounts_in_message":[],"_revision_number":23},{"id":"c4fe02e63bbc5373afb13749b2ec9097db79b7b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-08-05 12:15:15.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"49de23561cde3e163a8156dc3cce019626010654","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-08-05 14:51:30.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/dcc8d01233a94626876b0450bc43cadf : SUCCESS in 2h 13m 51s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/27c4834fc92a48cca2af61bdba33f6af : SUCCESS in 2h 34m 50s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"d2308e09ebaf836f316523ae7205f85cc6f8afa2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-05 15:44:44.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/0cb1a4b99f344a64a641038dc0cea0e6 : SUCCESS in 2h 42m 56s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/13c51e0bb9e646e8b1fef22c5f7b7cdb : SUCCESS in 47m 05s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/9e9df30c25e1421f984e6ddd972a53eb : SUCCESS in 1h 34m 08s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/33fbeb1988074d0e9d3add61d2d8b0bd : SUCCESS in 1h 31m 25s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a73c5875d145412f935f6648c245e16d : SUCCESS in 5m 45s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/79ee90e9b3bb4bd38317c00fcd5e0391 : SUCCESS in 2h 59m 33s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/034c4c06d94f4026a8fbcf18cbe400e0 : SUCCESS in 3h 04m 41s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/8f417aa655994713ba58ccf111e2f4a3 : SUCCESS in 3h 06m 51s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/0dd0ea94efce4b9da2f37035f91aefe2 : SUCCESS in 3h 14m 36s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/d284d3f10081469fb098976bd03c0a01 : SUCCESS in 1h 41m 01s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/cc1e31022c65492a8914b05cad735e45 : SUCCESS in 1h 56m 49s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/69e6136b43c24555ab3c2620220db35d : SUCCESS in 1h 55m 48s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/dbd03bf0194f462cbfa74567215cbc86 : SUCCESS in 2h 02m 37s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/7c5a70473e324c0baa5015812ea33b14 : SUCCESS in 1h 13m 42s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/150822b550af48ae90de3fc27b8ab63c : SUCCESS in 1h 25m 13s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/59152a4617c24e2b9e7d98129558a4a5 : SUCCESS in 1h 10m 44s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/990f5dd5c18c4f059a54209cae2fa7fe : SUCCESS in 13m 09s","accounts_in_message":[],"_revision_number":24},{"id":"d30006733f53c356ac7d43bb321409b50324cd2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-08-06 05:08:12.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"4379f5f58718eb7913b335b0aaf7b9b83201ca09","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-08-06 07:55:53.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/3fe3048a4e094519ba0149f4bdd4f70f : SUCCESS in 2h 11m 41s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/2a73fea1e5254f42869b8ef0f83d0c34 : SUCCESS in 2h 45m 57s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"919be8f4f62d0232391a1f8925bf25eab80b02fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-06 08:35:37.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/f24146f8b76445f3b163525824cc2553 : SUCCESS in 3h 13m 57s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/faee3477763644a79bcc1fcef8f6cf68 : SUCCESS in 47m 10s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/b375edaaed0d423fbf3bb24ffb926c33 : SUCCESS in 2h 03m 07s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/718044b897154abfa30b1faf91e0fd1e : SUCCESS in 1h 25m 43s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1ca237969e7043a4b19daf6566880697 : SUCCESS in 6m 11s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/5463a873e78e4a02b5089bfa245da9c1 : SUCCESS in 2h 47m 40s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/60b5f9aecfae4e57bf5acff5d2e941bb : SUCCESS in 3h 16m 50s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/362a319301e14739ac0d294e105a470b : SUCCESS in 3h 06m 16s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/0426d3e328fd48d2af73b40de78e6679 : SUCCESS in 2h 59m 38s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/6d4738a9394046c1b61d95e5b58a7a83 : SUCCESS in 1h 37m 45s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/91619c7264c7462ca81551d6b9175635 : SUCCESS in 2h 00m 34s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/b6a71958800c4ece9ceb29405c979113 : SUCCESS in 1h 58m 04s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/7277da61d453427ba02c99ece9764bb0 : SUCCESS in 2h 09m 04s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/791c371f65424ec6b75a04affc232c89 : SUCCESS in 1h 06m 19s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/c3790cf0a8b540e2aafbea15f5bce966 : FAILURE in 38m 03s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/c93e2ada2e5045fabe1fba3e1484fc97 : FAILURE in 57m 24s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/4fcf7d05822647baa4294f74d98ee620 : SUCCESS in 13m 02s","accounts_in_message":[],"_revision_number":25},{"id":"f9f1d7bc21a609968ab2b5c228d24ddd72526a15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-08-09 13:08:06.000000000","message":"Uploaded patch set 26: Commit message was updated.","accounts_in_message":[],"_revision_number":26},{"id":"353de0e3dd98ea44b63562781f9343a131f8e790","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-08-09 13:36:08.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/2fec046ed1634d039b40dac0eb234deb : RETRY_LIMIT in 5m 54s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/a81e433117f0408fb04577a68d8e3427 : RETRY_LIMIT in 6m 03s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"d948273543cacb265cba326a9f624dacf90d3883","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-09 14:14:07.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/d57c33e1c8d34ef29c0f8e30b9a19b19 : RETRY_LIMIT in 10m 11s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/075d82bc50464219bd0d88fa54d3ec1a : SUCCESS in 6m 19s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/f11e61fc29684c54b3b176f3d66d2933 : RETRY_LIMIT in 12m 00s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/9fa5f6c1c8db418a9f52941bf0919fcc : RETRY_LIMIT in 12m 53s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/9f6e79a958c8460c83e3997887c6ccc2 : RETRY_LIMIT in 17m 03s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/2755f8cb672147d6bc43760716c20b51 : RETRY_LIMIT in 16m 37s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/1fdce575bf574d068d1268927e0558ad : SUCCESS in 13m 19s","accounts_in_message":[],"_revision_number":26},{"id":"1fc4f4f8489ee9658a467aab3958c10803155bc5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-08-09 14:36:32.000000000","message":"Patch Set 27: Patch Set 26 was rebased","accounts_in_message":[],"_revision_number":27},{"id":"b3d8e2066f798fd0495c054066c206a667301d95","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-08-09 14:59:04.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/445603c713a2411aa5342814c54f144b : FAILURE in 20m 56s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/b0d4858498dc431a9a27e16165de2396 : FAILURE in 21m 22s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"d7c064ce895a1140912080714c2c1705e374b11d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-09 15:12:27.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/6d6fd6f7e9744097bc3ac529c427ddb1 : FAILURE in 20m 35s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6a7379df0de243328808a1182151e5dc : SUCCESS in 7m 56s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/caf1c7c8629a43868bc059c477b27de8 : FAILURE in 14m 14s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/51209e1569354553b35a3c0f032d2004 : FAILURE in 18m 28s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/237916134d8343c899bad14adcc5054a : FAILURE in 15m 19s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/b4ac408a633141dfbc3e02a6337e1b5f : FAILURE in 17m 41s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/fbf19eb4154f4ea58d1cafb2f5ec3730 : SUCCESS in 12m 53s","accounts_in_message":[],"_revision_number":27},{"id":"df8741506d8ccb2b2fdf8d1339a1548647edf60f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-08-10 03:43:20.000000000","message":"Patch Set 28: Commit message was updated.","accounts_in_message":[],"_revision_number":28},{"id":"32c41ae19a0fcb0c5a62e88801a3ae067cf0ac18","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-08-10 06:21:10.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/f1a609abfe58442ab67d22ecd62aa461 : SUCCESS in 1h 58m 07s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/5936d5c54a7a487fba04c9150c5566c4 : SUCCESS in 2h 36m 38s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"300b63ad6703133157e9183ba4f2fd9c71acc828","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-08-10 06:49:25.000000000","message":"Patch Set 28: Code-Review+1\n\n(1 comment)\n\nlooks OK but the tests runs are ~10 days old now @ https://review.rdoproject.org/r/c/testproject/+/34676 https://review.opendev.org/c/openstack/tripleo-heat-templates/+/801296 do you want to rerun that before merge","accounts_in_message":[],"_revision_number":28},{"id":"e7a9f11303f4a596df9b7556b2de7aec70585aa6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 07:53:31.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/357a11a35d7549ed84c96cd48c14d8dc : SUCCESS in 3h 14m 11s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/f75cc7564852409ca06cdd02d9f659fb : SUCCESS in 49m 05s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/e4ba56a66ec045de87eeba0d33591106 : SUCCESS in 1h 40m 13s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/47e4147426e349b29c61e996b26874dd : SUCCESS in 1h 33m 42s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/97a6478792e6491791f4cbca60c8ad28 : SUCCESS in 4m 51s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/e09a0089b0c04aa5a304eed478b3f1d0 : SUCCESS in 3h 41m 44s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/75017e4552ae41aa8d6b82281d0fba80 : SUCCESS in 3h 45m 59s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/52c8732bf7474c56a8aa46702ea6b5da : SUCCESS in 3h 40m 43s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/2a9efe3f31ec4632989f6a3781a4ad96 : SUCCESS in 4h 00m 29s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/462b7f2d47b04fff9e6e971dd2127b8d : SUCCESS in 1h 43m 24s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/e345bbd6640747298fbdc950b5300bb6 : SUCCESS in 2h 02m 07s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/c8fb8846a6604334bc0669633cb8e87c : SUCCESS in 2h 00m 20s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/aff094ff0ace44c4b993215cb7f4e01c : SUCCESS in 2h 06m 22s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/01e96dfeac7041e5b93e12f7c3eda10c : SUCCESS in 1h 17m 08s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/32cc5f9cc8e64e89ae8de0b9cdc0f200 : SUCCESS in 1h 29m 00s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/3119194a0db4470781d82bb83af0afbf : SUCCESS in 1h 13m 43s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/674e0e55d6ec415e8c13705cad9e80b9 : SUCCESS in 11m 14s","accounts_in_message":[],"_revision_number":28},{"id":"d9765a438f77c51ac3f31d4d46f8afd02b890e27","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-08-10 08:14:15.000000000","message":"Patch Set 28:\n\nNow the issue is fixed, currently build-test-packages is only running on content provider job\nhttps://storage.bhs.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_357/790926/28/check/tripleo-ci-centos-8-content-provider/357a11a/job-output.txt\n```\n TASK [Run DLRN gate role] ******************************************************\n2021-08-10 04:11:07.950032 | primary | Tuesday 10 August 2021  04:11:07 +0000 (0:00:00.091)       0:01:42.771 ********\n2021-08-10 04:11:08.211805 | primary |\n2021-08-10 04:11:08.211846 | primary | TASK [build-test-packages : gather facts used by role] *************************\n2021-08-10 04:11:08.212030 | primary | Tuesday 10 August 2021  04:11:08 +0000 (0:00:00.261)       0:01:43.033 ********\n2021-08-10 04:11:09.027580 | primary | ok: [undercloud]\n2021-08-10 04:11:09.041181 | primary |\n2021-08-10 04:11:09.041215 | primary | TASK [build-test-packages : Ensure DLRN package dependencies] ******************\n2021-08-10 04:11:09.041397 | primary | Tuesday 10 August 2021  04:11:09 +0000 (0:00:00.829)       0:01:43.862 ********\n2021-08-10 04:11:18.115039 | primary | changed: [undercloud]\n\n```\nand non-consumer jobs\nhttps://logserver.rdoproject.org/26/790926/28/openstack-check/tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001/f1a609a/job-output.txt\n```\nTASK [Run DLRN gate role] ******************************************************\n2021-08-10 03:59:39.253273 | primary | Tuesday 10 August 2021  03:59:39 +0000 (0:00:00.062)       0:00:53.620 ********\n2021-08-10 03:59:39.497036 | primary |\n2021-08-10 03:59:39.497106 | primary | TASK [build-test-packages : gather facts used by role] *************************\n2021-08-10 03:59:39.497126 | primary | Tuesday 10 August 2021  03:59:39 +0000 (0:00:00.243)       0:00:53.863 ********\n2021-08-10 03:59:40.168638 | primary | ok: [undercloud]\n2021-08-10 03:59:40.179138 | primary |\n2021-08-10 03:59:40.179160 | primary | TASK [build-test-packages : Ensure DLRN package dependencies] ******************\n2021-08-10 03:59:40.179238 | primary | Tuesday 10 August 2021  03:59:40 +0000 (0:00:00.682)       0:00:54.546 ********\n2021-08-10 03:59:47.032143 | primary | changed: [undercloud]\n```\n\nand it is getting skipped on consumer jobs\nhttps://56678b26a100ad6e31f2-7e86d39a7e52fd605616e5700386a2b9.ssl.cf2.rackcdn.com/790926/28/check/tripleo-ci-centos-8-undercloud-containers/f75cc75/job-output.txt\n```\nexport \u0027DEFAULT_ARGS\u003d--extra-vars local_working_dir\u003d/home/zuul/workspace/.quickstart     --extra-vars virthost\u003d127.0.0.2     --inventory /home/zuul/workspace/.quickstart/hosts     --extra-vars tripleo_root\u003d/home/zuul/src/opendev.org/openstack     --extra-vars working_dir\u003d/home/zuul     --extra-vars tripleo_generate_scripts\u003dtrue     --skip-tags tripleo-validations,teardown-all,build \u0027\n2021-08-10 05:34:31.208198 | primary | +(./toci_quickstart.sh:22): main(): DEFAULT_ARGS\u003d\u0027--extra-vars local_working_dir\u003d/home/zuul/workspace/.quickstart     --extra-vars virthost\u003d127.0.0.2     --inventory /home/zuul/workspace/.quickstart/hosts     --extra-vars tripleo_root\u003d/home/zuul/src/opendev.org/openstack     --extra-vars working_dir\u003d/home/zuul     --extra-vars tripleo_generate_scripts\u003dtrue     --skip-tags tripleo-validations,teardown-all,build \u0027\n2021-08-10 05:34:31.208241 | primary | +(./toci_quickstart.sh:29): main(): QUICKSTART_PREPARE_CMD\u003d\u0027\n```","accounts_in_message":[],"_revision_number":28},{"id":"89f3f81c1d808ad4beb83c45eb0fcaabf7a7a112","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-08-10 08:15:31.000000000","message":"Patch Set 28:\n\n\u003e Patch Set 28: Code-Review+1\n\u003e \n\u003e (1 comment)\n\u003e \n\u003e looks OK but the tests runs are ~10 days old now @ https://review.rdoproject.org/r/c/testproject/+/34676 https://review.opendev.org/c/openstack/tripleo-heat-templates/+/801296 do you want to rerun that before merge\n\nAdded the latest results here: https://review.opendev.org/c/openstack/tripleo-quickstart-extras/+/790926/28#message-d9765a438f77c51ac3f31d4d46f8afd02b890e27\n\nThe only issue we were facing earlier as build-test-packages were running on consumer job and is got fixed by https://review.opendev.org/c/openstack/tripleo-ci/+/803919","accounts_in_message":[],"_revision_number":28},{"id":"9bcda0009c42d75e6a91a2d52f0c56c5f58bf9b5","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2021-08-10 11:24:43.000000000","message":"Patch Set 28: Code-Review+2\n\nThanks!","accounts_in_message":[],"_revision_number":28},{"id":"e3af99242b9ca3fa3289acb59673c3563150886d","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2021-08-10 12:36:41.000000000","message":"Patch Set 28: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":28},{"id":"7feae0275ee3001eeaa304064f5911b115e27534","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 12:37:24.000000000","message":"Patch Set 28: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":28},{"id":"efd18ca520577adeb51c59a55b62776e3e6f5b9e","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-08-10 14:34:02.000000000","message":"Patch Set 28: Code-Review+2\n\n(1 comment)\n\nThanks, much better now, an inline comment, if fails in gate can do follow up.","accounts_in_message":[],"_revision_number":28},{"id":"f9552d889edf7f76748a8311fe4c25cbe4b97c29","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 16:33:49.000000000","message":"Patch Set 28: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/7b3d9a98579147ff9ae3fb43d9d9871d : SUCCESS in 3h 13m 35s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/af6102e7da624b1c86d24f997ed751c2 : SUCCESS in 50m 12s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/49f8658b72b94c7fb99ade9bd0765f31 : SUCCESS in 1h 39m 28s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/5702dca135364f138301fa9a4cd6e235 : SUCCESS in 1h 31m 55s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/efbe6e72128046de9bbbeee44b28c087 : SUCCESS in 3h 51m 37s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/99cdbc3d9d4c46a28c0f9f8eec6efdeb : SUCCESS in 3h 21m 50s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/de8dba63ae6744f991f495dbd0b22c05 : SUCCESS in 3h 30m 00s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/7e727fd3ce9f4bf9adf701ced2dddc67 : SUCCESS in 3h 27m 40s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/400e6524d8b648e0b1ce1366af865c33 : SUCCESS in 2h 12m 53s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/555f92b2352d4b408e581cdaa2ea352f : SUCCESS in 1h 58m 27s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/cb23c492073d494fbc95dfa8b6b00056 : SUCCESS in 2h 08m 10s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/6a917887056c494ca46acc644e636312 : SUCCESS in 2h 07m 36s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/43dc04d227db494fae4e0ea98df4a1e7 : SUCCESS in 1h 32m 35s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/4b161cb0f9f84ab782914fcc0b3fea6f : SUCCESS in 1h 33m 40s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/ac1e1e851ac8454bbc28362a3f01967f : SUCCESS in 1h 14m 47s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/cad120335eee4ee4a73c159b007e4284 : SUCCESS in 6m 03s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/0ca195a7bb08417ab1e10845bad9e3bf : SUCCESS in 13m 56s","accounts_in_message":[],"_revision_number":28},{"id":"c4cd1b2325d4a10b1a03998dc6bf3fddf7aa3214","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 16:33:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":28}],"current_revision_number":28,"current_revision":"4c225d64654bffc057daf89edaecb866b8890d3b","revisions":{"05d917b65ccfa19474ab249b4b9fab3194c4ee9b":{"kind":"REWORK","_number":1,"created":"2021-05-12 09:24:24.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/26/790926/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/26/790926/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/1"}}},"commit":{"parents":[{"commit":"7178744e793b0dc707eed5c72d8abd16028da582","subject":"Merge \"Set flavor_min_mem as a tempestconf overrides\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7178744e793b0dc707eed5c72d8abd16028da582"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-05-12 09:23:38.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-05-12 09:23:38.000000000","tz":330},"subject":"[WIP] move build-test-packages as a part of undercloud setup","message":"[WIP] move build-test-packages as a part of undercloud setup\n\nDepends-On: https://review.opendev.org/c/openstack/tripleo-quickstart/+/790701\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I4f9f3f13346761d3dec35844234cb6ae03ad4107\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/05d917b65ccfa19474ab249b4b9fab3194c4ee9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/05d917b65ccfa19474ab249b4b9fab3194c4ee9b"}]},"branch":"refs/heads/master"},"3b1cd25f8164df9c78fb486712425069ac1ab921":{"kind":"REWORK","_number":2,"created":"2021-05-12 09:41:39.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/26/790926/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/26/790926/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/2"}}},"commit":{"parents":[{"commit":"7178744e793b0dc707eed5c72d8abd16028da582","subject":"Merge \"Set flavor_min_mem as a tempestconf overrides\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7178744e793b0dc707eed5c72d8abd16028da582"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-05-12 09:23:38.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-05-12 09:41:26.000000000","tz":330},"subject":"[WIP] move build-test-packages as a part of undercloud setup","message":"[WIP] move build-test-packages as a part of undercloud setup\n\nDepends-On: https://review.opendev.org/c/openstack/tripleo-quickstart/+/790701\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I4f9f3f13346761d3dec35844234cb6ae03ad4107\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/3b1cd25f8164df9c78fb486712425069ac1ab921"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/3b1cd25f8164df9c78fb486712425069ac1ab921"}]},"branch":"refs/heads/master"},"0af3919ae85bb40d224e99e78688f754269892ac":{"kind":"REWORK","_number":3,"created":"2021-05-12 10:16:11.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/26/790926/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/26/790926/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/3"}}},"commit":{"parents":[{"commit":"7178744e793b0dc707eed5c72d8abd16028da582","subject":"Merge \"Set flavor_min_mem as a tempestconf overrides\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7178744e793b0dc707eed5c72d8abd16028da582"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-05-12 09:23:38.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-05-12 10:15:58.000000000","tz":330},"subject":"[WIP] move build-test-packages as a part of undercloud setup","message":"[WIP] move build-test-packages as a part of undercloud setup\n\nDepends-On: https://review.opendev.org/c/openstack/tripleo-quickstart/+/790701\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I4f9f3f13346761d3dec35844234cb6ae03ad4107\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/0af3919ae85bb40d224e99e78688f754269892ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/0af3919ae85bb40d224e99e78688f754269892ac"}]},"branch":"refs/heads/master"},"bc48720583d61b01465350cd2f2db8b77a7bd874":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2021-05-13 07:24:04.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/26/790926/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/26/790926/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/4"}}},"commit":{"parents":[{"commit":"7178744e793b0dc707eed5c72d8abd16028da582","subject":"Merge \"Set flavor_min_mem as a tempestconf overrides\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7178744e793b0dc707eed5c72d8abd16028da582"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-05-12 09:23:38.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-05-13 07:22:54.000000000","tz":330},"subject":"[WIP] move build-test-packages as a part of undercloud setup","message":"[WIP] move build-test-packages as a part of undercloud setup\n\nDepends-On: https://review.opendev.org/c/openstack/tripleo-quickstart/+/790701\nDepends-On: https://review.opendev.org/c/openstack/tripleo-operator-ansible/+/791118\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I4f9f3f13346761d3dec35844234cb6ae03ad4107\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/bc48720583d61b01465350cd2f2db8b77a7bd874"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/bc48720583d61b01465350cd2f2db8b77a7bd874"}]},"branch":"refs/heads/master"},"8efa4bf1ffa95c44f913bd5d7d4ea2b1502c9a61":{"kind":"REWORK","_number":5,"created":"2021-05-13 12:20:05.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/26/790926/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/26/790926/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/5"}}},"commit":{"parents":[{"commit":"7178744e793b0dc707eed5c72d8abd16028da582","subject":"Merge \"Set flavor_min_mem as a tempestconf overrides\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7178744e793b0dc707eed5c72d8abd16028da582"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-05-12 09:23:38.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-05-13 12:19:26.000000000","tz":330},"subject":"Move build-test-packages as a part of undercloud setup","message":"Move build-test-packages as a part of undercloud setup\n\nhttps://review.opendev.org/c/openstack/tripleo-quickstart/+/790701\nremoves tripleo-operator-ansible from TQE requires and will be get\ninstalled from rpm as a part for undercloud deps.\n\nIn current setup, dlrn build the rpm packages just after undercloud\nrpm packages installation but tripleo-operator-ansible is used before\nthat so that dlrn needs to build the changes early before undercloud\ndeps installation so that we can consume the changes in the deployment\nvia updated rpm installed from gating repo.\n\nIn order to do that, it moves build-test-packages roles task after repo\nsetup to install latest packages.\n\nIt also removes tripleo-operator-ansible from cloning on nodepool.\n\nDepends-On: https://review.opendev.org/c/openstack/tripleo-quickstart/+/790701\nDepends-On: https://review.opendev.org/c/openstack/tripleo-operator-ansible/+/791118\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I4f9f3f13346761d3dec35844234cb6ae03ad4107\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/8efa4bf1ffa95c44f913bd5d7d4ea2b1502c9a61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/8efa4bf1ffa95c44f913bd5d7d4ea2b1502c9a61"}]},"branch":"refs/heads/master"},"54565c1db8efaae8285b19d615fab8e4611e7a08":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2021-06-18 13:12:20.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/26/790926/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/26/790926/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/6"}}},"commit":{"parents":[{"commit":"bd757f01eb9c1081c1898c43f06266b61e868890","subject":"Merge \"Include component repo in image builds\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/bd757f01eb9c1081c1898c43f06266b61e868890"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-05-12 09:23:38.000000000","tz":330},"committer":{"name":"chandan kumar","email":"chkumar@redhat.com","date":"2021-06-18 13:12:20.000000000","tz":0},"subject":"Move build-test-packages as a part of undercloud setup","message":"Move build-test-packages as a part of undercloud setup\n\nhttps://review.opendev.org/c/openstack/tripleo-quickstart/+/790701\nremoves tripleo-operator-ansible from TQE requires and will be get\ninstalled from rpm as a part for undercloud deps.\n\nIn current setup, dlrn build the rpm packages just after undercloud\nrpm packages installation but tripleo-operator-ansible is used before\nthat so that dlrn needs to build the changes early before undercloud\ndeps installation so that we can consume the changes in the deployment\nvia updated rpm installed from gating repo.\n\nIn order to do that, it moves build-test-packages roles task after repo\nsetup to install latest packages.\n\nIt also removes tripleo-operator-ansible from cloning on nodepool.\n\nDepends-On: https://review.opendev.org/c/openstack/tripleo-quickstart/+/790701\nDepends-On: https://review.opendev.org/c/openstack/tripleo-operator-ansible/+/791118\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I4f9f3f13346761d3dec35844234cb6ae03ad4107\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/54565c1db8efaae8285b19d615fab8e4611e7a08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/54565c1db8efaae8285b19d615fab8e4611e7a08"}]},"branch":"refs/heads/master","description":"Rebase"},"85a70e0928eed8e41023f96e185e1863d04fc0ad":{"kind":"REWORK","_number":7,"created":"2021-07-02 06:26:34.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/26/790926/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/26/790926/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/7"}}},"commit":{"parents":[{"commit":"c907b825fa9750ab795221b80cf8be97c2ae4091","subject":"Merge \"Move atop installation after repo setup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/c907b825fa9750ab795221b80cf8be97c2ae4091"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-05-12 09:23:38.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-07-02 06:26:21.000000000","tz":330},"subject":"Move build-test-packages as a part of undercloud setup","message":"Move build-test-packages as a part of undercloud setup\n\nhttps://review.opendev.org/c/openstack/tripleo-quickstart/+/790701\nremoves tripleo-operator-ansible from TQE requires and will be get\ninstalled from rpm as a part for undercloud deps.\n\nIn current setup, dlrn build the rpm packages just after undercloud\nrpm packages installation but tripleo-operator-ansible is used before\nthat so that dlrn needs to build the changes early before undercloud\ndeps installation so that we can consume the changes in the deployment\nvia updated rpm installed from gating repo.\n\nIn order to do that, it moves build-test-packages roles task after repo\nsetup to install latest packages.\n\nIt also removes tripleo-operator-ansible from cloning on nodepool.\n\nDepends-On: https://review.opendev.org/c/openstack/tripleo-quickstart/+/790701\nDepends-On: https://review.opendev.org/c/openstack/tripleo-operator-ansible/+/791118\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I4f9f3f13346761d3dec35844234cb6ae03ad4107\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/85a70e0928eed8e41023f96e185e1863d04fc0ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/85a70e0928eed8e41023f96e185e1863d04fc0ad"}]},"branch":"refs/heads/master"},"7bd816e6e8ea7a8820370df615f3c355e449ebd7":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2021-07-05 11:38:29.000000000","uploader":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"ref":"refs/changes/26/790926/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/26/790926/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/8"}}},"commit":{"parents":[{"commit":"7e896d63ed7330ac19035414793a6777fe5180e3","subject":"Merge \"CI use \u0027--skip-nodes-and-networks\u0027 in oc deploy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7e896d63ed7330ac19035414793a6777fe5180e3"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-05-12 09:23:38.000000000","tz":330},"committer":{"name":"Shnaidman Sagi (Sergey)","email":"sshnaidm@redhat.com","date":"2021-07-05 11:38:29.000000000","tz":0},"subject":"Move build-test-packages as a part of undercloud setup","message":"Move build-test-packages as a part of undercloud setup\n\nhttps://review.opendev.org/c/openstack/tripleo-quickstart/+/790701\nremoves tripleo-operator-ansible from TQE requires and will be get\ninstalled from rpm as a part for undercloud deps.\n\nIn current setup, dlrn build the rpm packages just after undercloud\nrpm packages installation but tripleo-operator-ansible is used before\nthat so that dlrn needs to build the changes early before undercloud\ndeps installation so that we can consume the changes in the deployment\nvia updated rpm installed from gating repo.\n\nIn order to do that, it moves build-test-packages roles task after repo\nsetup to install latest packages.\n\nIt also removes tripleo-operator-ansible from cloning on nodepool.\n\nDepends-On: https://review.opendev.org/c/openstack/tripleo-quickstart/+/790701\nDepends-On: https://review.opendev.org/c/openstack/tripleo-operator-ansible/+/791118\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I4f9f3f13346761d3dec35844234cb6ae03ad4107\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7bd816e6e8ea7a8820370df615f3c355e449ebd7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7bd816e6e8ea7a8820370df615f3c355e449ebd7"}]},"branch":"refs/heads/master","description":"Rebase"},"f2f418a8fde0eeb483cd93d963a84409652832d7":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2021-07-26 10:10:05.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/26/790926/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/26/790926/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/9"}}},"commit":{"parents":[{"commit":"7e896d63ed7330ac19035414793a6777fe5180e3","subject":"Merge \"CI use \u0027--skip-nodes-and-networks\u0027 in oc deploy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7e896d63ed7330ac19035414793a6777fe5180e3"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-05-12 09:23:38.000000000","tz":330},"committer":{"name":"chandan kumar","email":"chkumar@redhat.com","date":"2021-07-26 10:09:49.000000000","tz":0},"subject":"Move build-test-packages as a part of undercloud setup","message":"Move build-test-packages as a part of undercloud setup\n\nhttps://review.opendev.org/c/openstack/tripleo-quickstart/+/790701\nremoves tripleo-operator-ansible from TQE requires and will be get\ninstalled from rpm as a part for undercloud deps.\n\nIn current setup, dlrn build the rpm packages just after undercloud\nrpm packages installation but tripleo-operator-ansible is used before\nthat so that dlrn needs to build the changes early before undercloud\ndeps installation so that we can consume the changes in the deployment\nvia updated rpm installed from gating repo.\n\nIn order to do that, it moves build-test-packages roles task after repo\nsetup to install latest packages.\n\nIt also removes tripleo-operator-ansible from cloning on nodepool.\n\nDepends-On: https://review.opendev.org/c/openstack/tripleo-quickstart/+/790701\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I4f9f3f13346761d3dec35844234cb6ae03ad4107\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/f2f418a8fde0eeb483cd93d963a84409652832d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/f2f418a8fde0eeb483cd93d963a84409652832d7"}]},"branch":"refs/heads/master","description":"Edit commit message"},"06d6c845adbc9fe4e4575463dfea78827a969b21":{"kind":"REWORK","_number":10,"created":"2021-07-27 10:48:58.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/26/790926/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/26/790926/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/10"}}},"commit":{"parents":[{"commit":"7e896d63ed7330ac19035414793a6777fe5180e3","subject":"Merge \"CI use \u0027--skip-nodes-and-networks\u0027 in oc deploy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7e896d63ed7330ac19035414793a6777fe5180e3"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-05-12 09:23:38.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-07-27 10:48:55.000000000","tz":330},"subject":"Move build-test-packages as a part of undercloud setup","message":"Move build-test-packages as a part of undercloud setup\n\nhttps://review.opendev.org/c/openstack/tripleo-quickstart/+/790701\nremoves tripleo-operator-ansible from TQE requires and will be get\ninstalled from rpm as a part for undercloud deps.\n\nIn current setup, dlrn build the rpm packages just after undercloud\nrpm packages installation but tripleo-operator-ansible is used before\nthat so that dlrn needs to build the changes early before undercloud\ndeps installation so that we can consume the changes in the deployment\nvia updated rpm installed from gating repo.\n\nIn order to do that, it moves build-test-packages roles task after repo\nsetup to install latest packages.\n\nIt also removes tripleo-operator-ansible from cloning on nodepool.\n\nDepends-On: https://review.opendev.org/c/openstack/tripleo-quickstart/+/790701\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I4f9f3f13346761d3dec35844234cb6ae03ad4107\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/06d6c845adbc9fe4e4575463dfea78827a969b21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/06d6c845adbc9fe4e4575463dfea78827a969b21"}]},"branch":"refs/heads/master"},"0be8fc05727f50b14f8c06119c7443a60459eca1":{"kind":"REWORK","_number":11,"created":"2021-07-27 11:58:05.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/26/790926/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/26/790926/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/11"}}},"commit":{"parents":[{"commit":"7e896d63ed7330ac19035414793a6777fe5180e3","subject":"Merge \"CI use \u0027--skip-nodes-and-networks\u0027 in oc deploy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7e896d63ed7330ac19035414793a6777fe5180e3"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-05-12 09:23:38.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-07-27 11:57:55.000000000","tz":330},"subject":"Move build-test-packages as a part of undercloud setup","message":"Move build-test-packages as a part of undercloud setup\n\nhttps://review.opendev.org/c/openstack/tripleo-quickstart/+/790701\nremoves tripleo-operator-ansible from TQE requires and will be get\ninstalled from rpm as a part for undercloud deps.\n\nIn current setup, dlrn build the rpm packages just after undercloud\nrpm packages installation but tripleo-operator-ansible is used before\nthat so that dlrn needs to build the changes early before undercloud\ndeps installation so that we can consume the changes in the deployment\nvia updated rpm installed from gating repo.\n\nIn order to do that, it moves build-test-packages roles task after repo\nsetup to install latest packages.\n\nIt also removes tripleo-operator-ansible from cloning on nodepool.\n\nDepends-On: https://review.opendev.org/c/openstack/tripleo-quickstart/+/790701\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I4f9f3f13346761d3dec35844234cb6ae03ad4107\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/0be8fc05727f50b14f8c06119c7443a60459eca1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/0be8fc05727f50b14f8c06119c7443a60459eca1"}]},"branch":"refs/heads/master"},"6813b147344b959c59182c92f364c4210fea2d3b":{"kind":"REWORK","_number":12,"created":"2021-07-27 12:19:04.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/26/790926/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/26/790926/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/12"}}},"commit":{"parents":[{"commit":"7e896d63ed7330ac19035414793a6777fe5180e3","subject":"Merge \"CI use \u0027--skip-nodes-and-networks\u0027 in oc deploy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7e896d63ed7330ac19035414793a6777fe5180e3"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-05-12 09:23:38.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-07-27 12:18:53.000000000","tz":330},"subject":"Move build-test-packages as a part of undercloud setup","message":"Move build-test-packages as a part of undercloud setup\n\nhttps://review.opendev.org/c/openstack/tripleo-quickstart/+/790701\nremoves tripleo-operator-ansible from TQE requires and will be get\ninstalled from rpm as a part for undercloud deps.\n\nIn current setup, dlrn build the rpm packages just after undercloud\nrpm packages installation but tripleo-operator-ansible is used before\nthat so that dlrn needs to build the changes early before undercloud\ndeps installation so that we can consume the changes in the deployment\nvia updated rpm installed from gating repo.\n\nIn order to do that, it moves build-test-packages roles task after repo\nsetup to install latest packages.\n\nIt also removes tripleo-operator-ansible from cloning on nodepool.\n\nDepends-On: https://review.opendev.org/c/openstack/tripleo-quickstart/+/790701\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I4f9f3f13346761d3dec35844234cb6ae03ad4107\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/6813b147344b959c59182c92f364c4210fea2d3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/6813b147344b959c59182c92f364c4210fea2d3b"}]},"branch":"refs/heads/master"},"21e055555252dffc22d18b63f4e28d784c51154c":{"kind":"REWORK","_number":13,"created":"2021-07-28 07:59:43.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/26/790926/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/26/790926/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/13"}}},"commit":{"parents":[{"commit":"7e896d63ed7330ac19035414793a6777fe5180e3","subject":"Merge \"CI use \u0027--skip-nodes-and-networks\u0027 in oc deploy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7e896d63ed7330ac19035414793a6777fe5180e3"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-05-12 09:23:38.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-07-28 07:58:28.000000000","tz":330},"subject":"Move build-test-packages as a part of undercloud setup","message":"Move build-test-packages as a part of undercloud setup\n\nhttps://review.opendev.org/c/openstack/tripleo-quickstart/+/790701\nremoves tripleo-operator-ansible from TQE requires and will be get\ninstalled from rpm as a part for undercloud deps.\n\nIn current setup, dlrn build the rpm packages just after undercloud\nrpm packages installation but tripleo-operator-ansible is used before\nthat so that dlrn needs to build the changes early before undercloud\ndeps installation so that we can consume the changes in the deployment\nvia updated rpm installed from gating repo.\n\nIn order to do that, it moves build-test-packages roles task after repo\nsetup to install latest packages.\n\nIt also removes tripleo-operator-ansible from cloning on nodepool.\n\nDepends-On: https://review.opendev.org/c/openstack/tripleo-quickstart/+/790701\n\nTesting different scenarios:\n1. https://review.opendev.org/c/openstack/tripleo-heat-templates/+/801296\n2. https://review.rdoproject.org/r/c/testproject/+/34676\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I4f9f3f13346761d3dec35844234cb6ae03ad4107\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/21e055555252dffc22d18b63f4e28d784c51154c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/21e055555252dffc22d18b63f4e28d784c51154c"}]},"branch":"refs/heads/master"},"9b8667a1f786ba638c59fe4efaddff0ac807b999":{"kind":"REWORK","_number":14,"created":"2021-07-29 05:52:41.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/26/790926/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/26/790926/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/14"}}},"commit":{"parents":[{"commit":"7e896d63ed7330ac19035414793a6777fe5180e3","subject":"Merge \"CI use \u0027--skip-nodes-and-networks\u0027 in oc deploy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7e896d63ed7330ac19035414793a6777fe5180e3"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-05-12 09:23:38.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-07-29 05:52:28.000000000","tz":330},"subject":"Move build-test-packages as a part of undercloud setup","message":"Move build-test-packages as a part of undercloud setup\n\nhttps://review.opendev.org/c/openstack/tripleo-quickstart/+/790701\nremoves tripleo-operator-ansible from TQE requires and will be get\ninstalled from rpm as a part for undercloud deps.\n\nIn current setup, dlrn build the rpm packages just after undercloud\nrpm packages installation but tripleo-operator-ansible is used before\nthat so that dlrn needs to build the changes early before undercloud\ndeps installation so that we can consume the changes in the deployment\nvia updated rpm installed from gating repo.\n\nIn order to do that, it moves build-test-packages roles task after repo\nsetup to install latest packages.\n\nIt also removes tripleo-operator-ansible from cloning on nodepool.\n\nDepends-On: https://review.opendev.org/c/openstack/tripleo-quickstart/+/790701\n\nTesting different scenarios:\n1. https://review.opendev.org/c/openstack/tripleo-heat-templates/+/801296\n2. https://review.rdoproject.org/r/c/testproject/+/34676\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I4f9f3f13346761d3dec35844234cb6ae03ad4107\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/9b8667a1f786ba638c59fe4efaddff0ac807b999"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/9b8667a1f786ba638c59fe4efaddff0ac807b999"}]},"branch":"refs/heads/master"},"42c15cb62a1df14a527374432c6f8cf5d2666481":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2021-07-29 06:18:03.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/26/790926/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/26/790926/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/15"}}},"commit":{"parents":[{"commit":"5fe107403798dcf45a03cebd762db21fc8a2622f","subject":"Merge \"Add tripleo_overcloud_image_upload_local back\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/5fe107403798dcf45a03cebd762db21fc8a2622f"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-05-12 09:23:38.000000000","tz":330},"committer":{"name":"chandan kumar","email":"chkumar@redhat.com","date":"2021-07-29 06:18:03.000000000","tz":0},"subject":"Move build-test-packages as a part of undercloud setup","message":"Move build-test-packages as a part of undercloud setup\n\nhttps://review.opendev.org/c/openstack/tripleo-quickstart/+/790701\nremoves tripleo-operator-ansible from TQE requires and will be get\ninstalled from rpm as a part for undercloud deps.\n\nIn current setup, dlrn build the rpm packages just after undercloud\nrpm packages installation but tripleo-operator-ansible is used before\nthat so that dlrn needs to build the changes early before undercloud\ndeps installation so that we can consume the changes in the deployment\nvia updated rpm installed from gating repo.\n\nIn order to do that, it moves build-test-packages roles task after repo\nsetup to install latest packages.\n\nIt also removes tripleo-operator-ansible from cloning on nodepool.\n\nDepends-On: https://review.opendev.org/c/openstack/tripleo-quickstart/+/790701\n\nTesting different scenarios:\n1. https://review.opendev.org/c/openstack/tripleo-heat-templates/+/801296\n2. https://review.rdoproject.org/r/c/testproject/+/34676\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I4f9f3f13346761d3dec35844234cb6ae03ad4107\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/42c15cb62a1df14a527374432c6f8cf5d2666481"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/42c15cb62a1df14a527374432c6f8cf5d2666481"}]},"branch":"refs/heads/master","description":"Rebase"},"742b1b13a83a4d1febf8a4301b8a88ed366109d8":{"kind":"REWORK","_number":16,"created":"2021-08-03 08:59:29.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/26/790926/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/26/790926/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/16"}}},"commit":{"parents":[{"commit":"5fe107403798dcf45a03cebd762db21fc8a2622f","subject":"Merge \"Add tripleo_overcloud_image_upload_local back\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/5fe107403798dcf45a03cebd762db21fc8a2622f"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-05-12 09:23:38.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-08-03 08:58:50.000000000","tz":330},"subject":"Move build-test-packages as a part of undercloud setup","message":"Move build-test-packages as a part of undercloud setup\n\nhttps://review.opendev.org/c/openstack/tripleo-quickstart/+/790701\nremoves tripleo-operator-ansible from TQE requires and will be get\ninstalled from rpm as a part for undercloud deps.\n\nIn current setup, dlrn build the rpm packages just after undercloud\nrpm packages installation but tripleo-operator-ansible is used before\nthat so that dlrn needs to build the changes early before undercloud\ndeps installation so that we can consume the changes in the deployment\nvia updated rpm installed from gating repo.\n\nIn order to do that, it moves build-test-packages roles task after repo\nsetup to install latest packages.\n\nIt also removes tripleo-operator-ansible from cloning on nodepool.\nIt also make sure build-test-packages is not triggered in consumer\njobs.\n\nTesting different scenarios:\n1. https://review.opendev.org/c/openstack/tripleo-heat-templates/+/801296\n2. https://review.rdoproject.org/r/c/testproject/+/34676\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I4f9f3f13346761d3dec35844234cb6ae03ad4107\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/742b1b13a83a4d1febf8a4301b8a88ed366109d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/742b1b13a83a4d1febf8a4301b8a88ed366109d8"}]},"branch":"refs/heads/master"},"2eb9918035af29103c6f318388d7c34bab3cdfe5":{"kind":"REWORK","_number":17,"created":"2021-08-03 11:35:40.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/26/790926/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/26/790926/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/17"}}},"commit":{"parents":[{"commit":"5fe107403798dcf45a03cebd762db21fc8a2622f","subject":"Merge \"Add tripleo_overcloud_image_upload_local back\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/5fe107403798dcf45a03cebd762db21fc8a2622f"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-05-12 09:23:38.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-08-03 11:35:18.000000000","tz":330},"subject":"Move build-test-packages as a part of undercloud setup","message":"Move build-test-packages as a part of undercloud setup\n\nhttps://review.opendev.org/c/openstack/tripleo-quickstart/+/790701\nremoves tripleo-operator-ansible from TQE requires and will be get\ninstalled from rpm as a part for undercloud deps.\n\nIn current setup, dlrn build the rpm packages just after undercloud\nrpm packages installation but tripleo-operator-ansible is used before\nthat so that dlrn needs to build the changes early before undercloud\ndeps installation so that we can consume the changes in the deployment\nvia updated rpm installed from gating repo.\n\nIn order to do that, it moves build-test-packages roles task after repo\nsetup to install latest packages.\n\nIt also removes tripleo-operator-ansible from cloning on nodepool.\n\nTesting different scenarios:\n1. https://review.opendev.org/c/openstack/tripleo-heat-templates/+/801296\n2. https://review.rdoproject.org/r/c/testproject/+/34676\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I4f9f3f13346761d3dec35844234cb6ae03ad4107\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/2eb9918035af29103c6f318388d7c34bab3cdfe5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/2eb9918035af29103c6f318388d7c34bab3cdfe5"}]},"branch":"refs/heads/master"},"fc07f22d5d424032d028ae8914b5e75ceebd63a0":{"kind":"REWORK","_number":18,"created":"2021-08-03 14:38:48.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/26/790926/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/26/790926/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/18"}}},"commit":{"parents":[{"commit":"5fe107403798dcf45a03cebd762db21fc8a2622f","subject":"Merge \"Add tripleo_overcloud_image_upload_local back\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/5fe107403798dcf45a03cebd762db21fc8a2622f"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-05-12 09:23:38.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-08-03 14:37:58.000000000","tz":330},"subject":"Move build-test-packages as a part of undercloud setup","message":"Move build-test-packages as a part of undercloud setup\n\nhttps://review.opendev.org/c/openstack/tripleo-quickstart/+/790701\nremoves tripleo-operator-ansible from TQE requires and will be get\ninstalled from rpm as a part for undercloud deps.\n\nIn current setup, dlrn build the rpm packages just after undercloud\nrpm packages installation but tripleo-operator-ansible is used before\nthat so that dlrn needs to build the changes early before undercloud\ndeps installation so that we can consume the changes in the deployment\nvia updated rpm installed from gating repo.\n\nIn order to do that, it moves build-test-packages roles task after repo\nsetup to install latest packages.\n\nIt also removes tripleo-operator-ansible from cloning on nodepool.\n\nTesting different scenarios:\n1. https://review.opendev.org/c/openstack/tripleo-heat-templates/+/801296\n2. https://review.rdoproject.org/r/c/testproject/+/34676\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I4f9f3f13346761d3dec35844234cb6ae03ad4107\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/fc07f22d5d424032d028ae8914b5e75ceebd63a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/fc07f22d5d424032d028ae8914b5e75ceebd63a0"}]},"branch":"refs/heads/master"},"0393344b46f84a91c6585c59d8c38bfb9d08a22f":{"kind":"REWORK","_number":19,"created":"2021-08-04 09:02:56.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/26/790926/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/26/790926/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/19"}}},"commit":{"parents":[{"commit":"5fe107403798dcf45a03cebd762db21fc8a2622f","subject":"Merge \"Add tripleo_overcloud_image_upload_local back\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/5fe107403798dcf45a03cebd762db21fc8a2622f"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-05-12 09:23:38.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-08-04 09:02:44.000000000","tz":330},"subject":"Move build-test-packages as a part of undercloud setup","message":"Move build-test-packages as a part of undercloud setup\n\nhttps://review.opendev.org/c/openstack/tripleo-quickstart/+/790701\nremoves tripleo-operator-ansible from TQE requires and will be get\ninstalled from rpm as a part for undercloud deps.\n\nIn current setup, dlrn build the rpm packages just after undercloud\nrpm packages installation but tripleo-operator-ansible is used before\nthat so that dlrn needs to build the changes early before undercloud\ndeps installation so that we can consume the changes in the deployment\nvia updated rpm installed from gating repo.\n\nIn order to do that, it moves build-test-packages roles task after repo\nsetup to install latest packages.\n\nIt also removes tripleo-operator-ansible from cloning on nodepool.\n\nTesting different scenarios:\n1. https://review.opendev.org/c/openstack/tripleo-heat-templates/+/801296\n2. https://review.rdoproject.org/r/c/testproject/+/34676\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I4f9f3f13346761d3dec35844234cb6ae03ad4107\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/0393344b46f84a91c6585c59d8c38bfb9d08a22f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/0393344b46f84a91c6585c59d8c38bfb9d08a22f"}]},"branch":"refs/heads/master"},"3db68d37eb8a106399fb5d6dbf25c5ff3ab3c564":{"kind":"REWORK","_number":20,"created":"2021-08-04 09:31:40.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/26/790926/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/26/790926/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/20"}}},"commit":{"parents":[{"commit":"5fe107403798dcf45a03cebd762db21fc8a2622f","subject":"Merge \"Add tripleo_overcloud_image_upload_local back\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/5fe107403798dcf45a03cebd762db21fc8a2622f"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-05-12 09:23:38.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-08-04 09:31:29.000000000","tz":330},"subject":"Move build-test-packages as a part of undercloud setup","message":"Move build-test-packages as a part of undercloud setup\n\nhttps://review.opendev.org/c/openstack/tripleo-quickstart/+/790701\nremoves tripleo-operator-ansible from TQE requires and will be get\ninstalled from rpm as a part for undercloud deps.\n\nIn current setup, dlrn build the rpm packages just after undercloud\nrpm packages installation but tripleo-operator-ansible is used before\nthat so that dlrn needs to build the changes early before undercloud\ndeps installation so that we can consume the changes in the deployment\nvia updated rpm installed from gating repo.\n\nIn order to do that, it moves build-test-packages roles task after repo\nsetup to install latest packages.\n\nIt also removes tripleo-operator-ansible from cloning on nodepool.\n\nTesting different scenarios:\n1. https://review.opendev.org/c/openstack/tripleo-heat-templates/+/801296\n2. https://review.rdoproject.org/r/c/testproject/+/34676\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I4f9f3f13346761d3dec35844234cb6ae03ad4107\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/3db68d37eb8a106399fb5d6dbf25c5ff3ab3c564"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/3db68d37eb8a106399fb5d6dbf25c5ff3ab3c564"}]},"branch":"refs/heads/master"},"21cd262d7e6aa74437c58166fac319e5539ad47b":{"kind":"REWORK","_number":21,"created":"2021-08-04 09:59:23.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/26/790926/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/26/790926/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/21"}}},"commit":{"parents":[{"commit":"5fe107403798dcf45a03cebd762db21fc8a2622f","subject":"Merge \"Add tripleo_overcloud_image_upload_local back\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/5fe107403798dcf45a03cebd762db21fc8a2622f"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-05-12 09:23:38.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-08-04 09:58:15.000000000","tz":330},"subject":"Move build-test-packages as a part of undercloud setup","message":"Move build-test-packages as a part of undercloud setup\n\nhttps://review.opendev.org/c/openstack/tripleo-quickstart/+/790701\nremoves tripleo-operator-ansible from TQE requires and will be get\ninstalled from rpm as a part for undercloud deps.\n\nIn current setup, dlrn build the rpm packages just after undercloud\nrpm packages installation but tripleo-operator-ansible is used before\nthat so that dlrn needs to build the changes early before undercloud\ndeps installation so that we can consume the changes in the deployment\nvia updated rpm installed from gating repo.\n\nIn order to do that, it moves build-test-packages roles task after repo\nsetup to install latest packages.\n\nIt also removes tripleo-operator-ansible from cloning on nodepool.\n\nTesting different scenarios:\n1. https://review.opendev.org/c/openstack/tripleo-heat-templates/+/801296\n2. https://review.rdoproject.org/r/c/testproject/+/34676\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I4f9f3f13346761d3dec35844234cb6ae03ad4107\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/21cd262d7e6aa74437c58166fac319e5539ad47b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/21cd262d7e6aa74437c58166fac319e5539ad47b"}]},"branch":"refs/heads/master"},"7f1163e82e3b64f6b9962ca32d47c97f62d5013d":{"kind":"REWORK","_number":22,"created":"2021-08-04 11:55:30.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/26/790926/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/26/790926/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/22"}}},"commit":{"parents":[{"commit":"5fe107403798dcf45a03cebd762db21fc8a2622f","subject":"Merge \"Add tripleo_overcloud_image_upload_local back\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/5fe107403798dcf45a03cebd762db21fc8a2622f"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-05-12 09:23:38.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-08-04 11:55:20.000000000","tz":330},"subject":"Move build-test-packages as a part of undercloud setup","message":"Move build-test-packages as a part of undercloud setup\n\nhttps://review.opendev.org/c/openstack/tripleo-quickstart/+/790701\nremoves tripleo-operator-ansible from TQE requires and will be get\ninstalled from rpm as a part for undercloud deps.\n\nIn current setup, dlrn build the rpm packages just after undercloud\nrpm packages installation but tripleo-operator-ansible is used before\nthat so that dlrn needs to build the changes early before undercloud\ndeps installation so that we can consume the changes in the deployment\nvia updated rpm installed from gating repo.\n\nIn order to do that, it moves build-test-packages roles task after repo\nsetup to install latest packages.\n\nIt also removes tripleo-operator-ansible from cloning on nodepool.\n\nTesting different scenarios:\n1. https://review.opendev.org/c/openstack/tripleo-heat-templates/+/801296\n2. https://review.rdoproject.org/r/c/testproject/+/34676\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I4f9f3f13346761d3dec35844234cb6ae03ad4107\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7f1163e82e3b64f6b9962ca32d47c97f62d5013d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7f1163e82e3b64f6b9962ca32d47c97f62d5013d"}]},"branch":"refs/heads/master"},"6beefc96384900154a4a49b966c5a8b9a92af1f9":{"kind":"REWORK","_number":23,"created":"2021-08-04 13:49:55.000000000","uploader":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"ref":"refs/changes/26/790926/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/26/790926/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/23"}}},"commit":{"parents":[{"commit":"5fe107403798dcf45a03cebd762db21fc8a2622f","subject":"Merge \"Add tripleo_overcloud_image_upload_local back\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/5fe107403798dcf45a03cebd762db21fc8a2622f"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-05-12 09:23:38.000000000","tz":330},"committer":{"name":"Shnaidman Sagi (Sergey)","email":"sshnaidm@redhat.com","date":"2021-08-04 13:49:49.000000000","tz":0},"subject":"Move build-test-packages as a part of undercloud setup","message":"Move build-test-packages as a part of undercloud setup\n\nhttps://review.opendev.org/c/openstack/tripleo-quickstart/+/790701\nremoves tripleo-operator-ansible from TQE requires and will be get\ninstalled from rpm as a part for undercloud deps.\n\nIn current setup, dlrn build the rpm packages just after undercloud\nrpm packages installation but tripleo-operator-ansible is used before\nthat so that dlrn needs to build the changes early before undercloud\ndeps installation so that we can consume the changes in the deployment\nvia updated rpm installed from gating repo.\n\nIn order to do that, it moves build-test-packages roles task after repo\nsetup to install latest packages.\n\nIt also removes tripleo-operator-ansible from cloning on nodepool.\n\nTesting different scenarios:\n1. https://review.opendev.org/c/openstack/tripleo-heat-templates/+/801296\n2. https://review.rdoproject.org/r/c/testproject/+/34676\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I4f9f3f13346761d3dec35844234cb6ae03ad4107\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/6beefc96384900154a4a49b966c5a8b9a92af1f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/6beefc96384900154a4a49b966c5a8b9a92af1f9"}]},"branch":"refs/heads/master"},"c62faf74b7426a9c5dc67989de4e6afaa14ebd8d":{"kind":"REWORK","_number":24,"created":"2021-08-05 12:15:15.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/26/790926/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/26/790926/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/24"}}},"commit":{"parents":[{"commit":"5fe107403798dcf45a03cebd762db21fc8a2622f","subject":"Merge \"Add tripleo_overcloud_image_upload_local back\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/5fe107403798dcf45a03cebd762db21fc8a2622f"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-05-12 09:23:38.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-08-05 12:15:04.000000000","tz":330},"subject":"Move build-test-packages as a part of undercloud setup","message":"Move build-test-packages as a part of undercloud setup\n\nhttps://review.opendev.org/c/openstack/tripleo-quickstart/+/790701\nremoves tripleo-operator-ansible from TQE requires and will be get\ninstalled from rpm as a part for undercloud deps.\n\nIn current setup, dlrn build the rpm packages just after undercloud\nrpm packages installation but tripleo-operator-ansible is used before\nthat so that dlrn needs to build the changes early before undercloud\ndeps installation so that we can consume the changes in the deployment\nvia updated rpm installed from gating repo.\n\nIn order to do that, it moves build-test-packages roles task after repo\nsetup to install latest packages.\n\nIt also removes tripleo-operator-ansible from cloning on nodepool.\n\nTesting different scenarios:\n1. https://review.opendev.org/c/openstack/tripleo-heat-templates/+/801296\n2. https://review.rdoproject.org/r/c/testproject/+/34676\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I4f9f3f13346761d3dec35844234cb6ae03ad4107\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/c62faf74b7426a9c5dc67989de4e6afaa14ebd8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/c62faf74b7426a9c5dc67989de4e6afaa14ebd8d"}]},"branch":"refs/heads/master"},"5866316356cd9409da67bd04bce5ed025e709628":{"kind":"REWORK","_number":25,"created":"2021-08-06 05:08:12.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/26/790926/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/26/790926/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/25"}}},"commit":{"parents":[{"commit":"5fe107403798dcf45a03cebd762db21fc8a2622f","subject":"Merge \"Add tripleo_overcloud_image_upload_local back\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/5fe107403798dcf45a03cebd762db21fc8a2622f"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-05-12 09:23:38.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-08-06 05:08:02.000000000","tz":330},"subject":"Move build-test-packages as a part of undercloud setup","message":"Move build-test-packages as a part of undercloud setup\n\nhttps://review.opendev.org/c/openstack/tripleo-quickstart/+/790701\nremoves tripleo-operator-ansible from TQE requires and will be get\ninstalled from rpm as a part for undercloud deps.\n\nIn current setup, dlrn build the rpm packages just after undercloud\nrpm packages installation but tripleo-operator-ansible is used before\nthat so that dlrn needs to build the changes early before undercloud\ndeps installation so that we can consume the changes in the deployment\nvia updated rpm installed from gating repo.\n\nIn order to do that, it moves build-test-packages roles task after repo\nsetup to install latest packages.\n\nIt also removes tripleo-operator-ansible from cloning on nodepool.\n\nTesting different scenarios:\n1. https://review.opendev.org/c/openstack/tripleo-heat-templates/+/801296\n2. https://review.rdoproject.org/r/c/testproject/+/34676\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I4f9f3f13346761d3dec35844234cb6ae03ad4107\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/5866316356cd9409da67bd04bce5ed025e709628"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/5866316356cd9409da67bd04bce5ed025e709628"}]},"branch":"refs/heads/master"},"8a7d22eda4da75f899cc97927a1746eaaafc6c7f":{"kind":"NO_CODE_CHANGE","_number":26,"created":"2021-08-09 13:08:06.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/26/790926/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/26/790926/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/26"}}},"commit":{"parents":[{"commit":"5fe107403798dcf45a03cebd762db21fc8a2622f","subject":"Merge \"Add tripleo_overcloud_image_upload_local back\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/5fe107403798dcf45a03cebd762db21fc8a2622f"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-05-12 09:23:38.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-08-09 13:08:03.000000000","tz":330},"subject":"Move build-test-packages as a part of undercloud setup","message":"Move build-test-packages as a part of undercloud setup\n\nhttps://review.opendev.org/c/openstack/tripleo-quickstart/+/790701\nremoves tripleo-operator-ansible from TQE requires and will be get\ninstalled from rpm as a part for undercloud deps.\n\nIn current setup, dlrn build the rpm packages just after undercloud\nrpm packages installation but tripleo-operator-ansible is used before\nthat so that dlrn needs to build the changes early before undercloud\ndeps installation so that we can consume the changes in the deployment\nvia updated rpm installed from gating repo.\n\nIn order to do that, it moves build-test-packages roles task after repo\nsetup to install latest packages.\n\nIt also removes tripleo-operator-ansible from cloning on nodepool.\n\nTesting different scenarios:\n1. https://review.opendev.org/c/openstack/tripleo-heat-templates/+/801296\n2. https://review.rdoproject.org/r/c/testproject/+/34676\n\nDepends-On: https://review.opendev.org/c/openstack/tripleo-ci/+/803919\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I4f9f3f13346761d3dec35844234cb6ae03ad4107\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/8a7d22eda4da75f899cc97927a1746eaaafc6c7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/8a7d22eda4da75f899cc97927a1746eaaafc6c7f"}]},"branch":"refs/heads/master"},"e6fe17caf282889ee19da3103dd9610f28303de4":{"kind":"TRIVIAL_REBASE","_number":27,"created":"2021-08-09 14:36:32.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/26/790926/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/26/790926/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/27"}}},"commit":{"parents":[{"commit":"19e8bdff5f566b978cb448493c9eae6f57a409a8","subject":"Merge \"Update list_allowed variables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/19e8bdff5f566b978cb448493c9eae6f57a409a8"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-05-12 09:23:38.000000000","tz":330},"committer":{"name":"chandan kumar","email":"chkumar@redhat.com","date":"2021-08-09 14:36:32.000000000","tz":0},"subject":"Move build-test-packages as a part of undercloud setup","message":"Move build-test-packages as a part of undercloud setup\n\nhttps://review.opendev.org/c/openstack/tripleo-quickstart/+/790701\nremoves tripleo-operator-ansible from TQE requires and will be get\ninstalled from rpm as a part for undercloud deps.\n\nIn current setup, dlrn build the rpm packages just after undercloud\nrpm packages installation but tripleo-operator-ansible is used before\nthat so that dlrn needs to build the changes early before undercloud\ndeps installation so that we can consume the changes in the deployment\nvia updated rpm installed from gating repo.\n\nIn order to do that, it moves build-test-packages roles task after repo\nsetup to install latest packages.\n\nIt also removes tripleo-operator-ansible from cloning on nodepool.\n\nTesting different scenarios:\n1. https://review.opendev.org/c/openstack/tripleo-heat-templates/+/801296\n2. https://review.rdoproject.org/r/c/testproject/+/34676\n\nDepends-On: https://review.opendev.org/c/openstack/tripleo-ci/+/803919\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I4f9f3f13346761d3dec35844234cb6ae03ad4107\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/e6fe17caf282889ee19da3103dd9610f28303de4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/e6fe17caf282889ee19da3103dd9610f28303de4"}]},"branch":"refs/heads/master","description":"Rebase"},"4c225d64654bffc057daf89edaecb866b8890d3b":{"kind":"NO_CODE_CHANGE","_number":28,"created":"2021-08-10 03:43:20.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/26/790926/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/26/790926/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/26/790926/28"}}},"commit":{"parents":[{"commit":"19e8bdff5f566b978cb448493c9eae6f57a409a8","subject":"Merge \"Update list_allowed variables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/19e8bdff5f566b978cb448493c9eae6f57a409a8"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-05-12 09:23:38.000000000","tz":330},"committer":{"name":"chandan kumar","email":"chkumar@redhat.com","date":"2021-08-10 03:43:13.000000000","tz":0},"subject":"Move build-test-packages as a part of undercloud setup","message":"Move build-test-packages as a part of undercloud setup\n\nhttps://review.opendev.org/c/openstack/tripleo-quickstart/+/790701\nremoves tripleo-operator-ansible from TQE requires and will be get\ninstalled from rpm as a part for undercloud deps.\n\nIn current setup, dlrn build the rpm packages just after undercloud\nrpm packages installation but tripleo-operator-ansible is used before\nthat so that dlrn needs to build the changes early before undercloud\ndeps installation so that we can consume the changes in the deployment\nvia updated rpm installed from gating repo.\n\nIn order to do that, it moves build-test-packages roles task after repo\nsetup to install latest packages.\n\nIt also removes tripleo-operator-ansible from cloning on nodepool.\n\nTesting different scenarios:\n1. https://review.opendev.org/c/openstack/tripleo-heat-templates/+/801296\n2. https://review.rdoproject.org/r/c/testproject/+/34676\n\nPatch for moving remove_tags to skip_tag_list:\n\nDepends-On: https://review.opendev.org/c/openstack/tripleo-ci/+/803919\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I4f9f3f13346761d3dec35844234cb6ae03ad4107\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/4c225d64654bffc057daf89edaecb866b8890d3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/4c225d64654bffc057daf89edaecb866b8890d3b"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
