)]}'
{"id":"openstack%2Ftripleo-quickstart-extras~793331","triplet_id":"openstack%2Ftripleo-quickstart-extras~master~Ib843f59f16ad30c07296a4328b3547396b94f71a","project":"openstack/tripleo-quickstart-extras","branch":"master","topic":"tripleo-get-hash","hashtags":[],"change_id":"Ib843f59f16ad30c07296a4328b3547396b94f71a","subject":"WIP: Install tripleo.repos in tqe plays undercloud/standalone","status":"ABANDONED","created":"2021-05-27 09:32:31.000000000","updated":"2021-08-12 12:46:01.000000000","total_comment_count":23,"unresolved_comment_count":14,"has_review_started":true,"meta_rev_id":"4b76168680970f6e2d328ce0a11da3bcad267249","_number":793331,"virtual_id_number":793331,"owner":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-08-09 18:09:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-08-09 17:33:06.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-27 11:43:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-27 12:02:45.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-15 11:34:35.000000000","updated_by":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"reviewer":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"state":"REVIEWER"},{"updated":"2021-07-05 13:31:25.000000000","updated_by":{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},"reviewer":{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},"state":"REVIEWER"},{"updated":"2021-07-13 12:33:27.000000000","updated_by":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"reviewer":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"state":"REVIEWER"},{"updated":"2021-07-13 14:41:16.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2021-07-15 14:06:56.000000000","updated_by":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"reviewer":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"state":"CC"},{"updated":"2021-07-15 21:37:47.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"CC"},{"updated":"2021-07-15 21:44:29.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"REVIEWER"},{"updated":"2021-07-19 15:16:16.000000000","updated_by":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"reviewer":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"state":"REVIEWER"}],"messages":[{"id":"844e808d4bbc79b4e9d950eb048d83b17d3d3e1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-05-27 09:32:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"185c9acbd46a4d3ae8a3f8d112bcaa5ede3fe576","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 11:43:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/a176da80794c434a808e2ef2195bdb3e : SUCCESS in 2h 00m 03s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/eceb712381e14018af5cf1e4d2a208cc : FAILURE in 14m 39s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/a7f370678020473d9831d981e1718fb8 : FAILURE in 18m 03s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/328dccfbf97f45c798d81dd128e31c0e : FAILURE in 16m 42s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ce4185312fa9475db2d279d043972caa : SUCCESS in 6m 41s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/73d602f74fe24979bb57aec8b98c7122 : SUCCESS in 1h 32m 50s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/83e947ad050242488cd791669271ffc1 : SUCCESS in 1h 51m 52s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/9a09140f96664fde8307a5b1f0a3e544 : SUCCESS in 1h 28m 12s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/a0b68837bd284fb78663d3a355083401 : SUCCESS in 1h 29m 43s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/6195a7975b874b53a8f594a758c008a8 : FAILURE in 19m 42s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/10c8e07479ac415f8f79d55b52f96b4d : FAILURE in 24m 05s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/9756f83b215a4cdaa4bc785530124a8a : FAILURE in 19m 53s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/5067221c54bc45b98b7f74563cf1c275 : FAILURE in 18m 45s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/c761cf419cfa44329fe77c5a3a987320 : SUCCESS in 1h 02m 50s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/33cddcaf48f44733b4a677372617a575 : FAILURE in 30m 16s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/22a032067239445b81a88453a548767f : SUCCESS in 13m 18s","accounts_in_message":[],"_revision_number":1},{"id":"dcc697350e437cc8d479281cedaa00818def61ed","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-05-27 12:02:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/a0b8429bec094877998d954213e2c131 : FAILURE in 2h 09m 16s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/5fc25aa8f1fa46cb9fd06d4f2fc35bb5 : FAILURE in 2h 10m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a9e002167a685ddc8396ac9a1be6aa198ceaf730","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-05-27 14:11:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"80aee4e052e50399a8fbb21bcf410e03ef089a05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 16:37:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/fdaf61c6f5074bed8f3f09ac8da5b641 : SUCCESS in 2h 16m 06s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/7dfffd64d9d149e1b6fba457c6b520dd : SUCCESS in 58m 10s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/08acdaaf96854c92827233829e8640f8 : FAILURE in 19m 01s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/31a85e391feb4943ac483e289eeb4f32 : FAILURE in 20m 07s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d2922c64c47041c287813650943b5b4a : SUCCESS in 6m 00s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/1f217a25f8a94ff599d4369ca0b79c79 : SUCCESS in 1h 24m 00s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/ec55d7aeee934329b02377fcdfbe277b : SUCCESS in 1h 27m 47s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/cce1960dad9a43ab977a69a0b9952800 : SUCCESS in 1h 31m 47s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/35d3a48cdae8453c96bf88f1a6061524 : SUCCESS in 1h 10m 49s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/b6257a97408e4e49a40e4f972628e7b4 : FAILURE in 19m 27s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/b74d4e8ff9d64916ac4b927f027ae44d : FAILURE in 21m 06s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/eff90bff7f0b49299c2fc099fd9abddf : FAILURE in 20m 26s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/29e63d0781974962ab0a2c35a091b80e : FAILURE in 21m 29s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/22355b6eb53848bf8e6ee4d0f6ef8b2f : FAILURE in 1h 13m 00s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/98865cc0bd3445c2b2a6dae5b777100a : FAILURE in 19m 35s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/8195603378ed43718c58a4a65098e7de : SUCCESS in 14m 27s","accounts_in_message":[],"_revision_number":2},{"id":"7b076439188a6c9c03b7f8b4812dc3855f075211","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-05-27 17:47:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/57b6cca8a85e4df9b2aa290471844ad1 : SUCCESS in 2h 43m 02s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/c0f4d71725564e1ab170247ecc250be7 : SUCCESS in 3h 10m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ad6b699bc5cdfaef8605dd8e8eaab7ba0be6bccd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-05-28 10:06:31.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"24defa168084f8e0181f929fab0d5d5125ef7d6d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-05-28 10:16:04.000000000","message":"Topic Ib843f59f16ad30c07296a4328b3547396b94f71a removed","accounts_in_message":[],"_revision_number":3},{"id":"20a2e3720b0bac47773fc8840bc5bd59e6e7f466","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-05-28 10:16:07.000000000","message":"Topic set to tripleo-get-hash","accounts_in_message":[],"_revision_number":3},{"id":"afedad8046609fa399877ed0341234db4bef95db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 12:33:15.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/20dc4c7471a54997a9dffca765e204ce : SUCCESS in 1h 51m 10s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/154e25207fdf46b2aa468a732ecaa8ac : SUCCESS in 56m 05s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/fb69d17bdaf64bf9b3daee279b157e56 : FAILURE in 35m 13s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/17a1b42e929e4b15a0532b971c8f715c : FAILURE in 22m 35s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/52b88fdabfcd494ea5e97c8f7c13f903 : SUCCESS in 5m 44s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/ec00719b8bd6484db5b3f412da23d9f7 : SUCCESS in 1h 33m 38s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/6b2228f7519547c98c1794c1b87972a2 : SUCCESS in 1h 46m 44s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/8e79bac1c9a04a7f9a8180bbf66443cd : SUCCESS in 1h 37m 32s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/176e6fa9928b4d6187a803e7218c9cae : SUCCESS in 1h 04m 46s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/572a7c1dda50418bade3110eee9964fe : FAILURE in 17m 58s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/70a5918e796c4f1e9bccd83135438ff2 : FAILURE in 24m 27s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/72f001ca308845b290834e1cdb930673 : FAILURE in 23m 40s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/af87dad77d184d01b6f0afe90311aaf6 : FAILURE in 18m 45s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/683c04023fd64d1a95abc93bf068efc9 : FAILURE in 16m 54s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/b8ba1b61cccc4da2a19b6c8c0294f002 : FAILURE in 23m 07s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/71ba22af340f4941afc57406c8af996f : SUCCESS in 12m 47s","accounts_in_message":[],"_revision_number":3},{"id":"58133d94a34ccb1152a765e311b6423d810954e4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-05-28 13:30:10.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/eb23a83514a744bc872f684f101a814b : FAILURE in 2h 31m 23s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/8d186f31716840158020732b192b74e4 : FAILURE in 2h 35m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0d7e1d67c014af5b1a9c0d483f283675fba0d276","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-05-31 11:34:55.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4b62329e98ca83e34d99e997f07fc5063a0fffb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 12:09:30.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/326122306eed49fc80a012a5a9dc7916 : FAILURE in 15m 23s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/508a20dc91894eefb78b79c4edd11c70 : SUCCESS in 5m 47s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/089a8b75475f427492075e1b3031ef14 : FAILURE in 16m 53s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/7c088eb4d8cf49c1a3355ae1145b9f0d : FAILURE in 24m 24s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/4d10b22685dd4354a24b57dc309bf56f : FAILURE in 26m 40s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/8b8f8a1e9e1349b2b3614e9202d1bc13 : FAILURE in 24m 55s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone-on-multinode-ipa https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/85636611d4e047bc95ca5a14942ffaea : SUCCESS in 12m 50s","accounts_in_message":[],"_revision_number":4},{"id":"9780c52b869b9a0679b40555bdae0248fc7c83c1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-05-31 14:53:31.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/f1881649af7d4ddca94f4bf7411d8434 : FAILURE in 2h 36m 35s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/a21037313c9c4adda7423fa7bf925682 : FAILURE in 2h 30m 42s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039-train-branch https://review.rdoproject.org/zuul/build/265b80afac36414ea86651cf3509b8ef : RETRY_LIMIT in 15m 39s (non-voting)\n- tripleo-ci-centos-8-ovb-3ctlr_1comp_1supp-featureset039-ussuri-branch https://review.rdoproject.org/zuul/build/e3289971a7c540f8a6c765b5c36d1c6f : RETRY_LIMIT in 23m 54s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp_1supp-featureset039 https://review.rdoproject.org/zuul/build/8014efb407a943b5ad5bc1faf7cf179f : RETRY_LIMIT in 21m 59s","accounts_in_message":[],"_revision_number":4},{"id":"928a75db574ca56909681f51bf8ff3473524c46b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-06-01 10:15:28.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"804783e0e2a03c87a370bd772fd693300fae0292","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 10:45:28.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/e4543cac496e4b39b29ce112ba5b2c9b : FAILURE in 18m 45s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/753d830e17b2484ab046af026794b633 : SUCCESS in 5m 51s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/f4f2595dbd5541cea1b0fd1bef4bd251 : FAILURE in 16m 59s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/05ab750e97124e89a3d4be3250270ab1 : FAILURE in 18m 07s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/ad1cfc54afe540dab75a09a8535d72b2 : FAILURE in 17m 31s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/21cabe4ac3a044e39135687bf13f28dd : FAILURE in 17m 10s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone-on-multinode-ipa https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/04d20861ea104133b44f4dfd5b5f08e8 : SUCCESS in 12m 13s","accounts_in_message":[],"_revision_number":5},{"id":"09a53abcc6cf1cfe4110170117035f0e995fbf22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-06-01 10:59:19.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"1b9c4e016cb4afd6e237f2293c8e0e0c392816f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 11:54:57.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/12695ceb41ea4cc3be019d8e0091d676 : FAILURE in 17m 39s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c08dba184ca84db1923c9d4a87c5e692 : SUCCESS in 5m 10s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/f8517fc9dd1e4b5d9ad5f37485e7a142 : FAILURE in 22m 36s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/67a66fa87e7b4143ba3feb26addb818e : FAILURE in 13m 59s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/98745171977342b99bbde5812f59520f : FAILURE in 16m 04s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/390f8a030adf478daa7f74eadc3a2990 : FAILURE in 17m 23s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone-on-multinode-ipa https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/6e5a4a0a3e5643d1a32adf4f77efac7f : SUCCESS in 12m 50s","accounts_in_message":[],"_revision_number":6},{"id":"9471762ebcd212bb80810bfa7c5dde59f9e7a932","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-06-01 13:51:40.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/cb0538d9c001452cbba2ba3567bf3a6d : FAILURE in 2h 29m 34s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/ac63df0d3fa344468953f618a08aece0 : SUCCESS in 2h 45m 04s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039-train-branch https://review.rdoproject.org/zuul/build/a86010fe4f1940909f1c470a2e7c583b : FAILURE in 33m 58s (non-voting)\n- tripleo-ci-centos-8-ovb-3ctlr_1comp_1supp-featureset039-ussuri-branch https://review.rdoproject.org/zuul/build/ea86966a72e540209297d6c7b604450e : FAILURE in 41m 37s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp_1supp-featureset039 https://review.rdoproject.org/zuul/build/e134dd4f52c64d46aba3f958f8b5eef2 : FAILURE in 43m 31s","accounts_in_message":[],"_revision_number":6},{"id":"9824616628b777b914f962211f2e1f4e709ff68e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-06-02 11:01:43.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"1f59db3339e4c6a1839b018ccf35f25b1344b385","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 11:48:27.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/930ac5931f634689bd874b1343392f97 : FAILURE in 23m 47s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/99ae0bb713bd42f39e74666a7ce80afe : SUCCESS in 4m 41s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/bbe2b89f271d4c7d9fa9e37f23f456e6 : FAILURE in 24m 27s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/a10141c0c49747d282ad695a4d68439f : FAILURE in 24m 08s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/d7ceca8887a74d0692ed307d3ea64473 : FAILURE in 24m 12s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/ee07686dd8684135adbfcb039858a367 : FAILURE in 24m 30s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone-on-multinode-ipa https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/686617078965487489b8602651bb7854 : SUCCESS in 13m 08s","accounts_in_message":[],"_revision_number":7},{"id":"7b8f07b7804993714666cda63c0c2418d3729e7a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-06-02 15:27:09.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/958baee0dac647fd91ce0a53c8894b21 : FAILURE in 2h 18m 20s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/c0ea061c935f4efba4d0ddfc7f4bff01 : FAILURE in 2h 47m 02s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039-train-branch https://review.rdoproject.org/zuul/build/44fe505dd7154d50b51b736b56ed169c : FAILURE in 37m 49s (non-voting)\n- tripleo-ci-centos-8-ovb-3ctlr_1comp_1supp-featureset039-ussuri-branch https://review.rdoproject.org/zuul/build/93326c3999be4a1caf5e96168c059d51 : FAILURE in 44m 18s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp_1supp-featureset039 https://review.rdoproject.org/zuul/build/1782d6f646d9435ca42deab29574e8c6 : FAILURE in 45m 54s","accounts_in_message":[],"_revision_number":7},{"id":"6a0f99d81b999b605e76f0ebd6a224bf71b92e27","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-06-03 10:29:54.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"7c15c7a19fb3781a4081d2cbc59754a11265c3d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 11:13:59.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/7869c49766ac447d9cd1e76ea112a5cc : FAILURE in 12m 55s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5e1c7af0366641ac9b8e6509345a5da9 : SUCCESS in 6m 25s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/f955d186997e4629956971afe6bb75c5 : FAILURE in 21m 33s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/ea440a0730834f9c9ea7fd251e94f1a8 : FAILURE in 22m 18s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/4cbc44ca6ed0409b855f03d81a6e3ae7 : FAILURE in 24m 22s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/c2c05d1ca67b40a399bac01767dfff45 : FAILURE in 24m 41s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone-on-multinode-ipa https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/4d26a03b54be43c4930489ea3788810e : SUCCESS in 11m 16s","accounts_in_message":[],"_revision_number":8},{"id":"ec5010d802cb84226e2e6a6b265f9d9045e4b3ed","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-06-03 11:40:13.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"fb0e1d107b43d1e0618869aa2efeb1c9639dee86","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-06-03 14:48:41.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/b58356cb45754e98b4680be226489234 : RETRY_LIMIT in 11m 32s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/1829cb643c474d2bbe028e3f5f861d2c : FAILURE in 2h 21m 40s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039-train-branch https://review.rdoproject.org/zuul/build/4e496c8477ba4f5fae4f6364f5413dad : FAILURE in 37m 42s (non-voting)\n- tripleo-ci-centos-8-ovb-3ctlr_1comp_1supp-featureset039-ussuri-branch https://review.rdoproject.org/zuul/build/849e732a2d0741e6a2101656e744809d : FAILURE in 51m 16s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp_1supp-featureset039 https://review.rdoproject.org/zuul/build/74b7a01a994b4f0fa4c42cc0d8240c8c : FAILURE in 45m 42s","accounts_in_message":[],"_revision_number":8},{"id":"6aef674e1a6edd3689ba4a7e7b8d2b1cf2dbd55a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-06-04 13:52:24.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"dfd08b788d31869d1b94f3a49367bdf4d894f3a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 15:28:03.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/d2eca4a20d39443aa1bb13a130be5f92 : FAILURE in 15m 54s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9060d398ccb34d2c9c38154b84c08c8d : SUCCESS in 5m 52s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/3d826ce5d2714d9ba5b91388e6db9d14 : FAILURE in 15m 24s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/6a0d668ccaa2485c94fdc20ef1a7362a : FAILURE in 26m 51s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/f7f6be471f3244449d398372b3ced009 : FAILURE in 19m 49s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/61e7d56f04ba4a1fafb094c12c6c771f : FAILURE in 19m 35s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone-on-multinode-ipa https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/6c424c118d72489097bf44c4cf77f3bc : SUCCESS in 12m 19s","accounts_in_message":[],"_revision_number":9},{"id":"73b954c2f7ea9f0dccd9c32f85e48ce2b788fc2f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-06-04 17:21:21.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/c23adc1bcc8444908f9e5a3910d9ebde : SUCCESS in 2h 54m 36s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/f61382ef9a8c4134a331f34eed70ecca : FAILURE in 2h 27m 35s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039-train-branch https://review.rdoproject.org/zuul/build/37a477ce6ce1492d87c6524f8e3d461c : FAILURE in 40m 13s (non-voting)\n- tripleo-ci-centos-8-ovb-3ctlr_1comp_1supp-featureset039-ussuri-branch https://review.rdoproject.org/zuul/build/4a97dcb2e69443458584635e2786c06c : FAILURE in 52m 47s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp_1supp-featureset039 https://review.rdoproject.org/zuul/build/344ada0fafed4067877ddffa2d3d7d32 : FAILURE in 43m 48s","accounts_in_message":[],"_revision_number":9},{"id":"eeb4848afbcc9a2672845c78b9e10b81e35f8a8b","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-06-07 11:30:49.000000000","message":"Patch Set 9: Code-Review-1\n\n(5 comments)\n\nand removing wip","accounts_in_message":[],"_revision_number":9},{"id":"73d388171c3bd01acfc64c524f55778108b03ba7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-06-07 12:45:41.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"a00e4099ffee0ac6b8dff771c59c24c222b8022c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-06-07 15:15:54.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/f5c1b442ba5441b393473697e88a04fa : FAILURE in 2h 24m 57s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/150bcacdbb88460b8d80936a168d9fdf : SUCCESS in 2h 25m 14s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039-train-branch https://review.rdoproject.org/zuul/build/175dbb9781d441c1b9bf7dc5079910e6 : FAILURE in 41m 38s (non-voting)\n- tripleo-ci-centos-8-ovb-3ctlr_1comp_1supp-featureset039-ussuri-branch https://review.rdoproject.org/zuul/build/0e9921df25084f3bb19311f242a5c77c : FAILURE in 55m 31s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp_1supp-featureset039 https://review.rdoproject.org/zuul/build/ce42564c77cd4b18af9ee2bf6dddeab1 : FAILURE in 47m 03s","accounts_in_message":[],"_revision_number":10},{"id":"1c756605048673a14b191463058e2a976635a3bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 17:19:00.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/eb8ece88e1404c38961465df700547a0 : SUCCESS in 3h 33m 50s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/82fe008899af4452acbd9aa54c3cbb18 : SUCCESS in 1h 09m 18s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/7fc2fae06b6444e0b9d031bb742aa21d : SUCCESS in 1h 40m 16s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/16e89d33c2f549638db452b4a22b519a : SUCCESS in 1h 39m 26s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/dfb4a6da0398475e87830186d909595c : SUCCESS in 6m 02s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/1ec53832d36548a5bfe5513f1f59529e : SUCCESS in 3h 37m 28s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/3d8f3dd8d5004e66bb8d71d2d2d42c2c : SUCCESS in 2h 57m 08s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/59dc7a9388cb45ed9693b7e80263e354 : SUCCESS in 3h 28m 33s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/a1714faf1995494b8aacd2eab8c4228f : SUCCESS in 3h 04m 04s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/03f7e819acd840c8a870e6f9a28365f3 : SUCCESS in 1h 53m 04s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/977ee3ef73a44419bff8f824914c46b0 : SUCCESS in 1h 51m 33s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/c114d03e13244d62ae037aefbc5fcb45 : SUCCESS in 2h 01m 50s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/a12169cddc534a1493f8e8b8be9d25b3 : SUCCESS in 1h 57m 12s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/2a93a31535f34b9ba3efdee2a8e97796 : SUCCESS in 1h 31m 30s\n- tripleo-ci-centos-8-standalone-on-multinode-ipa https://zuul.opendev.org/t/openstack/build/b03948d37a244d619d5b052c0ee05a38 : SUCCESS in 1h 51m 20s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/ceab6da578f94ec989c00fcc7dd29e7a : SUCCESS in 1h 35m 41s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/9bbe202e51514890b2a266000fda2081 : SUCCESS in 12m 53s","accounts_in_message":[],"_revision_number":10},{"id":"8102ff2d614554db33d9539d7cf9231569da5e58","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2021-06-15 11:34:35.000000000","message":"Patch Set 10: Code-Review+1\n\nI don\u0027t think this is the best approach, this should be done by tripleo-quickstart when we add tripleo-repo in the requirements, but I will not block it.\nI would create a task to install the python3-requests using ansible module instead of shell script, just to keep it more \"ansible\"","accounts_in_message":[],"_revision_number":10},{"id":"508d4993806ca6c4a6513d48aff111e99beeb032","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-07-01 08:46:53.000000000","message":"Patch Set 10:\n\n\u003e Patch Set 10: Code-Review+1\n\u003e \n\u003e I don\u0027t think this is the best approach, this should be done by tripleo-quickstart when we add tripleo-repo in the requirements, but I will not block it.\n\ndo you mean like there https://review.opendev.org/c/openstack/tripleo-quickstart/+/791486/15/quickstart-extras-requirements.txt#13 ? If so I tried that but had issues with finding the ansible module after installation in that way.\n\n\u003e I would create a task to install the python3-requests using ansible module instead of shell script, just to keep it more \"ansible\"\n\ncould do that but I would still need the shell task for the python setup.py install etc\n\nI know this is not ideal and I tried to get this working both with the git+ install as pointed to above, as well as installation via pip directly but in both cases I couldn\u0027t get the ansible module to be found.\n\nI would like to merge this so we can start using the get hash for now with the plan to replace with a \u0027better\u0027 way. This could be to use a lookup instead of a plugin for the ansible module (apparently this makes it easier for discovery on all the nodes you want to run the ansible) and/or making the installation part of the tripleo-repos installation/setup.cfg.","accounts_in_message":[],"_revision_number":10},{"id":"3c11eded829224586c8b4fac4a9f8d79e89cfeb3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-07-01 11:54:30.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"59556469ecda27293165793c4ef4f27432d5b187","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-07-01 15:54:09.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/7076a3ace31a4e8c8a1a2ac52b0dcd9c : RETRY_LIMIT in 2m 29s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/6c0b2a0f2c0d4529b07eabcaa177d510 : SUCCESS in 3h 36m 24s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039-train-branch https://review.rdoproject.org/zuul/build/ee0e65116ac24d8688b279ee49becdae : FAILURE in 43m 31s (non-voting)\n- tripleo-ci-centos-8-ovb-3ctlr_1comp_1supp-featureset039-ussuri-branch https://review.rdoproject.org/zuul/build/34dc2d9ac0714815a9f5c50e4aa9b949 : FAILURE in 39m 33s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp_1supp-featureset039 https://review.rdoproject.org/zuul/build/e97390cbf5cf45faadb51eadfd119bdd : FAILURE in 38m 23s","accounts_in_message":[],"_revision_number":11},{"id":"4109276fc2d5647cf8994f511d66e5c39eab5c83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 16:03:08.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/d4d287ddf0e34b9eb560ab0d9bef0235 : SUCCESS in 3h 12m 46s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/947d7b2e49774e969d02358df8170b63 : SUCCESS in 1h 07m 54s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/e5e0b65343d549dd901110e7f1faddb3 : SUCCESS in 1h 40m 36s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/18ddf0d68fd048f8b886ff7f8d8bc7fa : SUCCESS in 1h 37m 48s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/312f6a0c8388470c8df4722b77971306 : SUCCESS in 5m 23s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/f5e29240cecc45c5a49168bc96b6a40e : SUCCESS in 3h 27m 56s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/d3f00bfa19c24c738dc089529c631dee : SUCCESS in 3h 48m 26s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/45907d7faa2f4c1995e54a43f8499d91 : SUCCESS in 3h 46m 28s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/12da54b06b794bd4b315023732ece237 : SUCCESS in 3h 40m 18s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/c10981a99d044dbaa76fcc97359945ba : SUCCESS in 1h 55m 52s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/0532148f0c20442483f366d294cd6d9e : SUCCESS in 2h 22m 58s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/ac4b56010cc24521b4c1649824ca1af1 : SUCCESS in 2h 14m 33s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/c022b03ca69947b6b329786e7f4b283f : SUCCESS in 2h 23m 19s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/2913488e472b4bb79fef013f04ef9e50 : SUCCESS in 1h 16m 30s\n- tripleo-ci-centos-8-standalone-on-multinode-ipa https://zuul.opendev.org/t/openstack/build/f9bc18431b37469783a83fe644c0d1db : FAILURE in 36m 54s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/1a104c9940204b40bf4185579039159c : SUCCESS in 1h 38m 46s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/0f1ae19b70de4eb4b4966b3c8cadd9b2 : SUCCESS in 11m 33s","accounts_in_message":[],"_revision_number":11},{"id":"3b6a7b90277e101390bd33cbe7b3d433d254c71e","author":{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},"date":"2021-07-05 13:31:25.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"5ce37035d9185e877a5b6a531f0d7173fc62e61e","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-07-07 10:05:25.000000000","message":"Patch Set 11: Code-Review-1\n\nalso needs ovb playbook","accounts_in_message":[],"_revision_number":11},{"id":"1ac9f841fb26c4cdad8a5c964ecb6cf121104894","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-07-07 10:26:24.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"793cc3435e3dd616cb637eb7f8b626d66a2cc75c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 13:57:10.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/396f0e0b04c34385af2645c36a70d886 : SUCCESS in 2h 40m 24s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/1aa7859741844134b8f473dd70a5546e : SUCCESS in 1h 01m 22s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/c937c319eb624ba58dd25599d5f90274 : FAILURE in 18m 42s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/cb85ff6146734062b7251cb72d940a4c : FAILURE in 18m 38s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/117fed3faac24c1299f8218130935a74 : SUCCESS in 4m 55s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/7150fe1ff29a424394d4f391de8debda : SUCCESS in 1h 38m 25s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/3a06ae8f3df24ca9955153acc40ead9e : SUCCESS in 3h 23m 47s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/30560610875c4a18a0b9bd68ea008a8e : SUCCESS in 3h 12m 51s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/8364ab16b4f04009b813916c21f4b92a : SUCCESS in 3h 06m 00s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/eaa850a776f74f48a54cad6b5345d0f6 : FAILURE in 28m 43s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/ef20a6cc5d3a49f0a6bfb5d318f1c752 : SUCCESS in 2h 01m 02s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/d6abb84d1165442f9a5a167a52cc1135 : SUCCESS in 2h 03m 25s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/ee98ac8fa5524065a9daf2afe2f7686b : SUCCESS in 2h 05m 03s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/000afefbc236449c9510b1dde385d3c1 : SUCCESS in 1h 14m 10s\n- tripleo-ci-centos-8-standalone-on-multinode-ipa https://zuul.opendev.org/t/openstack/build/ca36e95e7f5a4cdeb728b911447cbfdb : FAILURE in 30m 05s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/0358cde8101c41d68da0944e4257f1c2 : SUCCESS in 1h 28m 52s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/b88a3f2faa5c4db684bbfc2e180ad33e : SUCCESS in 11m 29s","accounts_in_message":[],"_revision_number":12},{"id":"889a0c56f06f9b84ee62a9c6d4481c4043305cdb","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-07-07 14:50:28.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/472ea29c48404a17989819b2ed8c52c4 : SUCCESS in 2h 42m 59s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/39e3e1d073bb40418cbfc2d21f4a49ea : TIMED_OUT in 4h 22m 27s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039-train-branch https://review.rdoproject.org/zuul/build/a0d797b1711a428b894d1a43c5759c48 : FAILURE in 45m 24s (non-voting)\n- tripleo-ci-centos-8-ovb-3ctlr_1comp_1supp-featureset039-ussuri-branch https://review.rdoproject.org/zuul/build/be5a68e097b341cda4c807c79c8dc2e6 : FAILURE in 42m 36s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp_1supp-featureset039 https://review.rdoproject.org/zuul/build/6e1cb706258c4a0c845fa18627d389ba : FAILURE in 41m 56s","accounts_in_message":[],"_revision_number":12},{"id":"8182ee51a847a79835ca5e8c24fbb371ce6830bd","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-07-07 16:12:19.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"a20715fce9d714f4d4e6535a124f09f9cb98880b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-07-07 20:28:14.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/d44329f22f0a4982a18a5021b90fb611 : FAILURE in 2h 15m 20s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/994206c0b8a14f2988272b83f8b815a6 : SUCCESS in 3h 40m 32s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039-train-branch https://review.rdoproject.org/zuul/build/bc6e731684a24f72951029495a2d38d3 : FAILURE in 42m 43s (non-voting)\n- tripleo-ci-centos-8-ovb-3ctlr_1comp_1supp-featureset039-ussuri-branch https://review.rdoproject.org/zuul/build/386a48c37b824a1d9f55e87062f725f8 : FAILURE in 39m 28s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp_1supp-featureset039 https://review.rdoproject.org/zuul/build/35802540caeb4514bd01e2bb2c0f1194 : FAILURE in 40m 11s","accounts_in_message":[],"_revision_number":12},{"id":"e6381122d216e5d5e7663df78c39a5c1e1f63e0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 20:52:24.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/9372787eb4f74189988bd5745da2fc10 : SUCCESS in 4h 25m 31s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/e61fda989a744b3b9fe4cbaabe78aab4 : SUCCESS in 1h 04m 59s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/c2d8374474174f83bbe17ffde87e15ee : SUCCESS in 1h 30m 58s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/f6f16e64d26f4512964ab2a73c5fc354 : SUCCESS in 1h 38m 07s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ef9500440f574ba5a29ca22d703de726 : SUCCESS in 6m 09s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/4795aadfdc764fc1a33e66523305baec : SUCCESS in 3h 01m 44s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/cc579123495a4c07b0e0c30ae47a33f6 : SUCCESS in 3h 21m 41s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/4d7701cc39f2407988ddb57113428391 : SUCCESS in 3h 54m 25s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/d23bb4b0ee744b6e97dd2f0e09e90262 : SUCCESS in 3h 19m 17s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/99bc55332f2647ad80dc2849760e33e1 : POST_FAILURE in 1h 45m 18s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/863c381ee4f24ab8adf6f70a779b90bd : SUCCESS in 2h 00m 50s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/83233bfdc70a4617a17b24574c262d92 : SUCCESS in 2h 22m 02s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/27625a7d1e0d493a8822e284bb344e4b : SUCCESS in 2h 09m 49s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/fa01ede65e1d4ce5a3b0006df0d11fbe : SUCCESS in 1h 24m 48s\n- tripleo-ci-centos-8-standalone-on-multinode-ipa https://zuul.opendev.org/t/openstack/build/e81348a322fe41838675c2eb68fa1388 : FAILURE in 42m 14s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/6561f802527542568b292ac785c20ca4 : SUCCESS in 1h 35m 23s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/38289e14207b4287b24d489ce62e62d9 : SUCCESS in 12m 06s","accounts_in_message":[],"_revision_number":12},{"id":"9604fb76b10226d91612837f6ce4e7fcdd44dc3f","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-07-08 05:38:22.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"c50b6e3595e4ba70b54e3fb1c5adc4e288a9ac4b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-07-08 08:59:09.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/2e2abd4997a24e3384476272d0e04b4c : SUCCESS in 2h 57m 16s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/f0766cd29d794b1fb64dac3112173b7e : SUCCESS in 3h 18m 53s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039-train-branch https://review.rdoproject.org/zuul/build/f1e726338b01456b82b5cf0ba221039c : FAILURE in 40m 36s (non-voting)\n- tripleo-ci-centos-8-ovb-3ctlr_1comp_1supp-featureset039-ussuri-branch https://review.rdoproject.org/zuul/build/33cd75bcf27a42139c2dd3bb807cc91f : FAILURE in 35m 49s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp_1supp-featureset039 https://review.rdoproject.org/zuul/build/6048c403e66d4c719922760a18ce3c3b : FAILURE in 40m 45s","accounts_in_message":[],"_revision_number":12},{"id":"862509afc874e72087cd5d66eae3e43e27143b14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-08 09:07:14.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/bb855ad3d6ae49bc9e82984ecc57e97f : SUCCESS in 2h 47m 54s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/44893eea52aa4a8b9501f3854a35195e : SUCCESS in 1h 00m 27s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/d5017b78f29d4c8890e20af92bc323fe : SUCCESS in 1h 34m 27s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/830a0f5f569f4d788654bc6de75466d3 : SUCCESS in 1h 32m 31s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f9d4f850790b422da7fae9a5373e90c8 : SUCCESS in 5m 05s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/63ee24115ad745df8e1f937c88fa762e : SUCCESS in 2h 54m 09s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/fa01f04300df4d86897c9fe003a4ad87 : SUCCESS in 3h 16m 21s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/6e1fdd4d1234413e8653a9978b5392ed : SUCCESS in 3h 19m 02s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/8565cccde99749b5b82eb45ca191b8f6 : SUCCESS in 3h 07m 14s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/69eaa9964379449bac6f39949bec6df1 : SUCCESS in 1h 37m 22s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/36b7e9c21a594467b151eaeb914df619 : SUCCESS in 2h 02m 10s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/201f7865f2ec455b9d883c3c1ac47b07 : SUCCESS in 2h 07m 29s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/528ede10e1814b7ab4fcefd8b5046526 : SUCCESS in 2h 02m 57s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/e80f785104ab473fad25ea65e40790a4 : SUCCESS in 1h 11m 08s\n- tripleo-ci-centos-8-standalone-on-multinode-ipa https://zuul.opendev.org/t/openstack/build/85ffedb0b1b94d78802fd802c7ce82e3 : FAILURE in 32m 50s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/e2f686f701a34d178235eed09b2c1692 : SUCCESS in 1h 30m 40s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/ff89e0b21dd84439a989d96145a204bd : SUCCESS in 11m 47s","accounts_in_message":[],"_revision_number":12},{"id":"e2d8672cf071ad9e736906c0e5e116b32c5a627c","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2021-07-13 12:33:27.000000000","message":"Patch Set 12: Code-Review-1\n\n(2 comments)\n\nUsing \"sudo\" should be changed to \"become\"","accounts_in_message":[],"_revision_number":12},{"id":"3f8b29d44176fa1cd49066ccfef4d3e891534cfe","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-07-13 12:40:43.000000000","message":"Patch Set 12: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"89d188d73367dc3a5c166b77c03aa86c37d92ad0","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-07-13 12:47:53.000000000","message":"Patch Set 12:\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"0b56bca70c29aa405123c3ef764654514690e618","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-07-13 12:56:35.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"f81de6fba80b0dac6ff0eb3fd5143580831c4125","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2021-07-13 14:37:45.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"efa3bb4bff680b76936be35241b48d539d1ec834","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-07-13 15:40:10.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/45db2548860c41208352d329106959b4 : SUCCESS in 2h 42m 19s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/2e3dd792bfc54de39ea315ca928035e6 : SUCCESS in 2h 31m 56s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039-train-branch https://review.rdoproject.org/zuul/build/c9d10bc300a249049a603c457776cedf : FAILURE in 40m 21s (non-voting)\n- tripleo-ci-centos-8-ovb-3ctlr_1comp_1supp-featureset039-ussuri-branch https://review.rdoproject.org/zuul/build/2aea9d7db991419f9f42da101efe3405 : FAILURE in 44m 26s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp_1supp-featureset039 https://review.rdoproject.org/zuul/build/a923cb88eb884d7aa9f10c687e177994 : FAILURE in 43m 41s","accounts_in_message":[],"_revision_number":13},{"id":"fa71bf0168254f7ef715fb327370db2643f75881","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-13 17:01:57.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/bb75944208c74d97a6f8799d441c2764 : SUCCESS in 3h 26m 48s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/318549bd322b4aab99ccff47fc9dbba2 : SUCCESS in 51m 36s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/fd6e7f986be447f6b46e218e64eff269 : SUCCESS in 1h 55m 06s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/2085baf7c932409291dc26b2154d554b : SUCCESS in 1h 35m 34s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/204b89701f6344f5a68c6700776daec0 : SUCCESS in 5m 16s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/4feb1ed5145a4ff1a9ad181c7e7d4248 : SUCCESS in 3h 16m 54s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/a788b4dcac064f16a692a80e5ddef569 : SUCCESS in 3h 51m 03s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/95d5f6c20c574c589add9384653b31cb : SUCCESS in 3h 52m 37s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/5dffabb47efc41a1941c260e867d8919 : SUCCESS in 3h 46m 08s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/9fcab0dfdbbe4cc694e22c5a5714a3bc : SUCCESS in 1h 43m 40s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/592370a8c6144fbab3bdd555f792a7e3 : SUCCESS in 2h 12m 45s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/640130b7a9cd404a9d598aa9baf25930 : POST_FAILURE in 2h 24m 38s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/76f5d98e90854cae8dfb9d7c6b65d586 : SUCCESS in 2h 19m 25s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/be9440bfe7994f3799299b7f5bffb8ca : SUCCESS in 1h 15m 28s\n- tripleo-ci-centos-8-standalone-on-multinode-ipa https://zuul.opendev.org/t/openstack/build/b8122877a2734ad0a86a002c136da7b1 : FAILURE in 35m 55s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/7b23f60c6fc940488991428ee3876a94 : SUCCESS in 1h 29m 57s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/e9f873d4830c4de0bdc45e7d10e90316 : SUCCESS in 11m 58s","accounts_in_message":[],"_revision_number":13},{"id":"2fbc0252d0ad114cefe263a8a8b1a5d8f076686c","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2021-07-13 18:06:22.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"ac9adf8a8386553fc1654058954876ab9a50c07c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-07-13 20:57:45.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/f3be71535db24a89a3a5ed338ef934ab : SUCCESS in 2h 24m 14s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/d75c693854744870985be5cddaedb9a1 : SUCCESS in 2h 45m 31s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039-train-branch https://review.rdoproject.org/zuul/build/b77ffff397964809824437e46352186f : FAILURE in 44m 41s (non-voting)\n- tripleo-ci-centos-8-ovb-3ctlr_1comp_1supp-featureset039-ussuri-branch https://review.rdoproject.org/zuul/build/d6ef7c8d15a147d3b91d6ba69d8504ab : FAILURE in 45m 24s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp_1supp-featureset039 https://review.rdoproject.org/zuul/build/5296ae96eadf419dbc5eb5478eb1e3ea : FAILURE in 43m 43s","accounts_in_message":[],"_revision_number":13},{"id":"1455d22de16e94b6a5c20d809089162bf37134b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-14 00:22:23.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/3d9019cdec4e44b79dc6d9a78b0b4896 : SUCCESS in 3h 23m 37s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/412260ea9af84593ba295a04e8776def : SUCCESS in 57m 28s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/dce73e1de1264753adb51ab31a141051 : SUCCESS in 1h 41m 33s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/a4da98f7762647e684a6d6c3692e5da2 : SUCCESS in 1h 32m 03s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/402d4531b2a943be8528b7b403f69763 : SUCCESS in 6m 23s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/11f344c52310480385c09bf145f32616 : SUCCESS in 3h 08m 52s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/0d98d69ef06b4aceb0b462255808ab93 : SUCCESS in 4h 04m 26s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/e49c478a99da44bf89cc272908ee46b5 : SUCCESS in 3h 54m 04s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/e4e88e5bc09149b795efa8f3bc05474a : SUCCESS in 3h 35m 05s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/0bf9eab338a145d7bbfb98bddda75446 : SUCCESS in 1h 42m 01s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/9f5ac19ca33b4755a76d1cca09f5c89d : SUCCESS in 2h 26m 09s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/e1a1b74d215a46c595e4dcd3da0ba874 : SUCCESS in 2h 18m 00s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/f2b4687d8f984c44bee33c6855122d8f : SUCCESS in 2h 17m 20s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/ec64626925034fe7910079aa0a3e0004 : SUCCESS in 1h 19m 39s\n- tripleo-ci-centos-8-standalone-on-multinode-ipa https://zuul.opendev.org/t/openstack/build/459f13bd3d5949b3a69d5b11315a3802 : FAILURE in 37m 16s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/008e5e08d83949c99ff41620e83b16a4 : SUCCESS in 1h 38m 22s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/4d5a031ba82d48c499ff9bb1baa2cbc8 : SUCCESS in 13m 29s","accounts_in_message":[],"_revision_number":13},{"id":"81b5251748548484c041a2705f9ce93ee11c710d","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-07-14 11:14:48.000000000","message":"Patch Set 13: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"b53142bbf84691da06b44786bfbfa8c5e668c1c5","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-07-14 11:23:12.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"d2fb240193eec85aedbd1ea83992177990aa935e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-07-14 11:35:33.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"f50ff942eb641db41a122b8cb9425c09dfd133ff","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-07-14 14:11:57.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/f94626719fdc402d9f37158a82e22ff0 : FAILURE in 2h 07m 51s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/e6b86f4becc14951ad58aad1c145eb87 : SUCCESS in 2h 34m 23s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039-train-branch https://review.rdoproject.org/zuul/build/3e9bd0ae706548c0a1eb384ed8069108 : FAILURE in 40m 20s (non-voting)\n- tripleo-ci-centos-8-ovb-3ctlr_1comp_1supp-featureset039-ussuri-branch https://review.rdoproject.org/zuul/build/3245cf8e43fc43bb8fb4cac6f534c2f0 : FAILURE in 43m 02s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp_1supp-featureset039 https://review.rdoproject.org/zuul/build/5ae533bd65cd4deb91da58ddd5255641 : FAILURE in 44m 09s","accounts_in_message":[],"_revision_number":14},{"id":"48dc2bbebde4957a2ce7ffb5169d250b4adda8ca","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2021-07-14 14:14:23.000000000","message":"Patch Set 14:\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"bb15ca779641055f0cad422cbbb6d6945542517f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-14 15:57:43.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/edba0a72334d4aad84a5cd9e19350b81 : SUCCESS in 4h 09m 47s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/18749c095a3943a7a2be5d50f63264b3 : SUCCESS in 44m 34s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/c95c3ef089494ecabe0dfd51fa31a3e2 : SUCCESS in 1h 34m 53s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/ea30d687a8964e8981daff966b0b765a : SUCCESS in 1h 29m 51s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/98f3acb54a0248a7b6c046040d7a4621 : SUCCESS in 5m 28s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/67ed19b047d744939375f3f4d2f4cb55 : SUCCESS in 3h 14m 18s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/fc115bed4c5341d4a9984dcfa69706bd : SUCCESS in 3h 56m 25s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/fffe56632c374255bb4d6f1d3bc5e4e1 : SUCCESS in 3h 20m 37s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/7bd6ff4a3e1345929a2bf7641d09e28c : SUCCESS in 3h 25m 59s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/74828e4355a1480b88daa0240a591235 : SUCCESS in 1h 45m 07s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/8449fa858bee493a8cc70e66c7cd2aea : SUCCESS in 2h 24m 29s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/89485d9999d14be2a935f40545e5b880 : SUCCESS in 2h 08m 18s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/10b20cf3291b4ce98f36182a3a3ae5bf : SUCCESS in 2h 15m 56s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/8fd6cc0468a64998af1dc54e7e29fb41 : SUCCESS in 1h 02m 03s\n- tripleo-ci-centos-8-standalone-on-multinode-ipa https://zuul.opendev.org/t/openstack/build/ee0e374781b64e608a37dbfd59f70bb5 : FAILURE in 32m 57s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/ec6fe82ba0e24aa3ab483d69f332e250 : SUCCESS in 1h 21m 28s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/cb5571a292b943c596ccf9f067a546d6 : SUCCESS in 1h 19m 07s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/498b562b113948f6bba3046e1f646471 : SUCCESS in 12m 13s","accounts_in_message":[],"_revision_number":14},{"id":"8e221153fe33586675eb5e36f6f0b54fc639908f","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-07-15 11:46:31.000000000","message":"Patch Set 14:\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"ffd75705dab0a8d768c8014951138c601a656da1","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2021-07-15 14:06:56.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"470b5af5ee060f0a6b0aa4b8fa3d20372b1248b0","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2021-07-15 21:37:47.000000000","message":"Patch Set 14:\n\nare we missing tasks to install for container build and image jobs?","accounts_in_message":[],"_revision_number":14},{"id":"90f81ca6686726a00a1faa66cf26627e2ea4c915","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2021-07-15 21:44:29.000000000","message":"Patch Set 14: Code-Review+1\n\nInstall seems ok from the jobs logs.\nComment left above","accounts_in_message":[],"_revision_number":14},{"id":"d900d179e29b78a2855b3cdedf15907770686911","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-07-16 08:51:07.000000000","message":"Patch Set 14:\n\n\u003e Patch Set 14:\n\u003e \n\u003e are we missing tasks to install for container build and image jobs?\n\nI think we\u0027re OK\n\nFor container build, we aren\u0027t really running those jobs any more far as i can see eg [1][2] didn\u0027t run for a while. I think besides content providers we only do explicit container builds with the tripleo-build-containers-ubi-8 job.\n\nNeither the containers-ubi-8 job or the build image jobs use the repo-setup role (which is where we\u0027re wiring up tripleo-get-hash in https://review.opendev.org/c/openstack/tripleo-quickstart/+/791486\n\nInstead they are directly using tripleo-repos [3][4] - so they should be unaffected by this change.\n\n\n\n\n\n[1] https://zuul.openstack.org/builds?job_name\u003dtripleo-build-containers-centos-8 \n[2] https://zuul.openstack.org/builds?job_name\u003dtripleo-build-containers-centos-8-ussuri\n[3] https://opendev.org/openstack/tripleo-ci/src/commit/ba710b8056922ad8f00e4f8440039e63e60d180f/playbooks/tripleo-buildcontainers/pre.yaml#L22\n[4] https://opendev.org/openstack/tripleo-ci/src/commit/ba710b8056922ad8f00e4f8440039e63e60d180f/playbooks/tripleo-buildimages/pre.yaml#L22","accounts_in_message":[],"_revision_number":14},{"id":"7173d6edcaa01800e7d64ca04d3b0029d9238f99","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-07-16 10:09:23.000000000","message":"Patch Set 14:\n\n\u003e Patch Set 14:\n\u003e \n\u003e \u003e Patch Set 14:\n\u003e \u003e \n\u003e \u003e are we missing tasks to install for container build and image jobs?\n\u003e \n\u003e I think we\u0027re OK\n\u003e \n\u003e For container build, we aren\u0027t really running those jobs any more far as i can see eg [1][2] didn\u0027t run for a while. I think besides content providers we only do explicit container builds with the tripleo-build-containers-ubi-8 job.\n\u003e \n\u003e Neither the containers-ubi-8 job or the build image jobs use the repo-setup role (which is where we\u0027re wiring up tripleo-get-hash in https://review.opendev.org/c/openstack/tripleo-quickstart/+/791486\n\u003e \n\u003e Instead they are directly using tripleo-repos [3][4] - so they should be unaffected by this change.\n\u003e \n\u003e \n\u003e \n\u003e \n\u003e \n\u003e [1] https://zuul.openstack.org/builds?job_name\u003dtripleo-build-containers-centos-8 \n\u003e [2] https://zuul.openstack.org/builds?job_name\u003dtripleo-build-containers-centos-8-ussuri\n\u003e [3] https://opendev.org/openstack/tripleo-ci/src/commit/ba710b8056922ad8f00e4f8440039e63e60d180f/playbooks/tripleo-buildcontainers/pre.yaml#L22\n\u003e [4] https://opendev.org/openstack/tripleo-ci/src/commit/ba710b8056922ad8f00e4f8440039e63e60d180f/playbooks/tripleo-buildimages/pre.yaml#L22\n\n\nposted a testproject as an extra sanity check @ \nhttps://review.rdoproject.org/r/c/testproject/+/34567 Sanity check image container build not affected tripleo_get_hash","accounts_in_message":[],"_revision_number":14},{"id":"d18fe446377f0cbf45f6b113c05b0233f33f2e62","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2021-07-16 15:45:01.000000000","message":"Patch Set 14:\n\n\u003e Patch Set 14:\n\u003e \n\u003e \u003e Patch Set 14:\n\u003e \u003e \n\u003e \u003e \u003e Patch Set 14:\n\u003e \u003e \u003e \n\u003e \u003e \u003e are we missing tasks to install for container build and image jobs?\n\u003e \u003e \n\u003e \u003e I think we\u0027re OK\n\u003e \u003e \n\u003e \u003e For container build, we aren\u0027t really running those jobs any more far as i can see eg [1][2] didn\u0027t run for a while. I think besides content providers we only do explicit container builds with the tripleo-build-containers-ubi-8 job.\n\u003e \u003e \n\u003e \u003e Neither the containers-ubi-8 job or the build image jobs use the repo-setup role (which is where we\u0027re wiring up tripleo-get-hash in https://review.opendev.org/c/openstack/tripleo-quickstart/+/791486\n\u003e \u003e \n\u003e \u003e Instead they are directly using tripleo-repos [3][4] - so they should be unaffected by this change.\n\u003e \u003e \n\u003e \u003e \n\u003e \u003e \n\u003e \u003e \n\u003e \u003e \n\u003e \u003e [1] https://zuul.openstack.org/builds?job_name\u003dtripleo-build-containers-centos-8 \n\u003e \u003e [2] https://zuul.openstack.org/builds?job_name\u003dtripleo-build-containers-centos-8-ussuri\n\u003e \u003e [3] https://opendev.org/openstack/tripleo-ci/src/commit/ba710b8056922ad8f00e4f8440039e63e60d180f/playbooks/tripleo-buildcontainers/pre.yaml#L22\n\u003e \u003e [4] https://opendev.org/openstack/tripleo-ci/src/commit/ba710b8056922ad8f00e4f8440039e63e60d180f/playbooks/tripleo-buildimages/pre.yaml#L22\n\u003e \n\u003e \n\u003e posted a testproject as an extra sanity check @ \n\u003e https://review.rdoproject.org/r/c/testproject/+/34567 Sanity check image container build not affected tripleo_get_hash\n\nck - should be fine - image builds are bot using repo-setup either","accounts_in_message":[],"_revision_number":14},{"id":"d2c789beadbc198df1bc998dce89a17739837dbc","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2021-07-19 13:22:00.000000000","message":"Patch Set 14: Code-Review+2","accounts_in_message":[],"_revision_number":14},{"id":"7a0154258358edfa1ef7219db33e1e865c1d342a","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2021-07-19 15:16:09.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"a8fc5d971da114b51347563bb351b73f0c6815d1","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2021-07-19 15:16:16.000000000","message":"Patch Set 14: Code-Review-1","accounts_in_message":[],"_revision_number":14},{"id":"ff9f003fd11d7104463fca25888cecf7fedad28c","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2021-07-21 07:50:48.000000000","message":"Patch Set 14: Code-Review+2","accounts_in_message":[],"_revision_number":14},{"id":"3987268f36fc46db0cd6da13975ce6b81d18d3cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2021-07-22 10:48:11.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"014ace2f4dbf44c3ae9dc48d60bc632548a7fd1b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-07-22 11:25:26.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/018f43aa7f04407195ba0fc7ad77ff75 : FAILURE in 30m 16s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/03ce33c086544b818111b7014a47c289 : FAILURE in 26m 25s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039-train-branch https://review.rdoproject.org/zuul/build/57f22d6305394789aaac29ef45151996 : FAILURE in 27m 42s (non-voting)\n- tripleo-ci-centos-8-ovb-3ctlr_1comp_1supp-featureset039-ussuri-branch https://review.rdoproject.org/zuul/build/098c56e0e85e41ba82582f6efc4308c9 : FAILURE in 34m 13s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp_1supp-featureset039 https://review.rdoproject.org/zuul/build/1ee1faaa775d4d0bab4a410a6b3269a6 : FAILURE in 31m 33s","accounts_in_message":[],"_revision_number":15},{"id":"0335b620f68fb31f6500b3327a2cf07229cde9cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 11:31:36.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/b902eed5210b4ee8b41b7ccb991169a1 : FAILURE in 24m 16s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/876e15e5a87d4de6a2704a76238229a9 : SUCCESS in 8m 46s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/1c7a0947358646bea8021db37c66640a : FAILURE in 24m 12s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/54de989979d044588b75bce7071c670d : FAILURE in 19m 08s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/2f69de353e0e4898aff20cf9ecee941b : FAILURE in 24m 43s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/551ac2c0a4224f49b2e5333ccb3d70af : FAILURE in 26m 28s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone-on-multinode-ipa https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/99197af153c641db9581c5590227bfef : SUCCESS in 13m 55s","accounts_in_message":[],"_revision_number":15},{"id":"09ccddb218517ec18f713d6b6d40f76ff7316df3","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-08-04 09:24:55.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"84bbe52680a976b58fbf44daf01b7d1045fe90cf","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-08-04 09:27:01.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"c027e7e46498274747d565f6ddba73cf992e3a22","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-08-06 13:50:55.000000000","message":"Patch Set 15:\n\nstarting to revisit this (still ruck for a few more days though) now that the move to collections is done.\n\nrebase incoming to see how well the new install works in https://review.opendev.org/c/openstack/tripleo-quickstart-extras/+/793331/15/playbooks/tasks/install-tripleo-get-hash.yml","accounts_in_message":[],"_revision_number":15},{"id":"aaca04417e5fbf9d602bfa827b859310aa83063f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-08-06 13:56:56.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.","accounts_in_message":[],"_revision_number":16},{"id":"f979cb5234042238f871962fdb4b0d24a04cb91a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-06 14:32:15.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/f62de4ae7dfc478a9350342930a1b6bc : FAILURE in 17m 06s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/26ce5d221e4746d4bc035daaed6848db : SUCCESS in 5m 32s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/34c22cdf872046d0a15ee6de571d882a : FAILURE in 20m 02s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/758b94f87dae4f95b7049952d0ea127c : FAILURE in 18m 33s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/aa93af2e74c64a41b50595352a3afb87 : FAILURE in 18m 20s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/54826ef57ce8411183a36700dd527ac6 : FAILURE in 17m 24s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone-on-multinode-ipa https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/9792960084754148b379a38263dd7a56 : SUCCESS in 12m 41s","accounts_in_message":[],"_revision_number":16},{"id":"6a56d3c0d0add57dcb3daf76a0117c6de26417c6","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-08-06 14:49:40.000000000","message":"Patch Set 16:\n\n* https://storage.gra.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_f62/793331/16/check/tripleo-ci-centos-8-content-provider/f62de4a/job-output.txt\n        * 2021-08-06 14:12:42.743760 | primary | TASK [Install tripleo-repos package and collection from git checkout] **********\n        * 2021-08-06 14:12:46.003275 | primary | + ansible --version\n2021-08-06 14:12:46.003287 | primary | /bin/sh: line 4: ansible: command not found","accounts_in_message":[],"_revision_number":16},{"id":"9df65b0ebe05c2c5a77428f3cb13a51ffe78de89","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-08-06 14:56:42.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/dfa90f403b4a435698fb055b1f98c9f1 : FAILURE in 27m 42s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/39dca632d6ac414296433317f54b72a3 : FAILURE in 26m 06s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039-train-branch https://review.rdoproject.org/zuul/build/e2dc18e2d8b441e09e7d960c11468f2b : FAILURE in 27m 32s (non-voting)\n- tripleo-ci-centos-8-ovb-3ctlr_1comp_1supp-featureset039-ussuri-branch https://review.rdoproject.org/zuul/build/3b8f135e6a5b44ae9d32aaf70e28e15c : FAILURE in 30m 27s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp_1supp-featureset039 https://review.rdoproject.org/zuul/build/a8cf032ac65947959e9735e4c59a5485 : FAILURE in 30m 59s","accounts_in_message":[],"_revision_number":16},{"id":"6fcebf2c682eed7b6f4c13b17c4582976206cdd7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2021-08-09 14:19:07.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"58d80110ac9c0ab97874e477f8b80a58c7a902bf","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-08-09 17:33:06.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/f3bb870ddc1e44a0a659e04b96c8421a : SUCCESS in 2h 02m 36s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/6e5546cf1d3643e58317fd67c1a1ccde : SUCCESS in 2h 27m 57s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039-train-branch https://review.rdoproject.org/zuul/build/204b2dcbd1874c6aad99642ff0943df8 : FAILURE in 51m 52s (non-voting)\n- tripleo-ci-centos-8-ovb-3ctlr_1comp_1supp-featureset039-ussuri-branch https://review.rdoproject.org/zuul/build/8decfc7dde874f7ea716a5dfc932fb32 : SUCCESS in 3h 12m 12s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp_1supp-featureset039 https://review.rdoproject.org/zuul/build/f44428b59a65438a9e3b052c9bf42723 : SUCCESS in 3h 00m 08s","accounts_in_message":[],"_revision_number":17},{"id":"e4c559c8a607801543bd0f573c5f85a18be92575","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-09 18:09:13.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/15bfadd013b54466a1b588f0e35afe6d : SUCCESS in 2h 58m 28s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/c53574db1cb64c26a69f3d765ca1cde7 : SUCCESS in 49m 07s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/84f411ad053447c5978f4288844dd917 : SUCCESS in 1h 37m 04s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/47d85cea4eae4f1cbc40907eff02644f : SUCCESS in 1h 29m 32s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/74a4609f430e4f30b21b99023afa6080 : SUCCESS in 5m 31s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/64fde317fe704bdd865e5ce175cbe0fd : SUCCESS in 3h 34m 57s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/550fed4e91a247aab3faaf6e0620efc3 : SUCCESS in 3h 41m 25s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/a420bf61192145c5991695c2bbb8b619 : SUCCESS in 2h 58m 44s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/2c879fd130d549b0addc1899d423a298 : SUCCESS in 3h 05m 15s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/4ec659dec5b843b089a9ce406d215171 : SUCCESS in 1h 37m 20s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/8accc4da0e6748ee83465e0209f57560 : SUCCESS in 1h 52m 49s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/5864f33509c045d3b95cf095904774de : SUCCESS in 1h 52m 52s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/356175d3d04847d2bdfb276d54071fde : SUCCESS in 1h 59m 35s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/f09a77f2d6ff46028d881fc47f4a32ac : SUCCESS in 1h 16m 00s\n- tripleo-ci-centos-8-standalone-on-multinode-ipa https://zuul.opendev.org/t/openstack/build/562271ed82a14de7a9d0db4d8aeaf33b : SUCCESS in 1h 43m 05s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/13b1153255a24d8aac696af73f677777 : FAILURE in 40m 22s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/dd67247e58c04140a2409ea646f381da : FAILURE in 59m 50s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/f7143e512c234bc591dcd6b8854c8e0a : SUCCESS in 13m 17s","accounts_in_message":[],"_revision_number":17},{"id":"4b76168680970f6e2d328ce0a11da3bcad267249","tag":"autogenerated:gerrit:abandon","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2021-08-12 12:46:01.000000000","message":"Abandoned\n\nObsoleted by https://review.opendev.org/c/openstack/tripleo-quickstart/+/801776","accounts_in_message":[],"_revision_number":17}],"current_revision_number":17,"current_revision":"f5a4f177fc89b894e4acd56389f52f4f5f50740b","revisions":{"2da09c7c8a5fc85c634ebc588d82a9cff2d2163e":{"kind":"REWORK","_number":1,"created":"2021-05-27 09:32:31.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/31/793331/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/31/793331/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/1"}}},"commit":{"parents":[{"commit":"83ae5fdef331fb7a4d0a5b5c7fba17fd51aa169b","subject":"Re-enable container-updates for downstream","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/83ae5fdef331fb7a4d0a5b5c7fba17fd51aa169b"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-05-27 09:26:47.000000000","tz":180},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-05-27 09:32:19.000000000","tz":180},"subject":"DNM test install tripleo-get-hash in tqe play for undercloud","message":"DNM test install tripleo-get-hash in tqe play for undercloud\n\nUsed by https://review.opendev.org/c/openstack/tripleo-quickstart/+/791486\n\nChange-Id: Ib843f59f16ad30c07296a4328b3547396b94f71a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/2da09c7c8a5fc85c634ebc588d82a9cff2d2163e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/2da09c7c8a5fc85c634ebc588d82a9cff2d2163e"}]},"branch":"refs/heads/master"},"c692ead9a3c72603202a991a868773d81fc09ef6":{"kind":"REWORK","_number":2,"created":"2021-05-27 14:11:51.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/31/793331/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/31/793331/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/2"}}},"commit":{"parents":[{"commit":"83ae5fdef331fb7a4d0a5b5c7fba17fd51aa169b","subject":"Re-enable container-updates for downstream","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/83ae5fdef331fb7a4d0a5b5c7fba17fd51aa169b"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-05-27 09:26:47.000000000","tz":180},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-05-27 14:11:42.000000000","tz":180},"subject":"DNM test install tripleo-get-hash in tqe play for undercloud","message":"DNM test install tripleo-get-hash in tqe play for undercloud\n\nUsed by https://review.opendev.org/c/openstack/tripleo-quickstart/+/791486\n\nChange-Id: Ib843f59f16ad30c07296a4328b3547396b94f71a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/c692ead9a3c72603202a991a868773d81fc09ef6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/c692ead9a3c72603202a991a868773d81fc09ef6"}]},"branch":"refs/heads/master"},"fd3958dacc7cc36d662fca50c08abbbd1a13c66a":{"kind":"REWORK","_number":3,"created":"2021-05-28 10:06:31.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/31/793331/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/31/793331/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/3"}}},"commit":{"parents":[{"commit":"7470f4389ed620c85dea7202724d3cbbd5e33675","subject":"Merge \"Collect the old and new path for outputs/passwords first\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7470f4389ed620c85dea7202724d3cbbd5e33675"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-05-27 09:26:47.000000000","tz":180},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-05-28 09:49:06.000000000","tz":180},"subject":"Install tripleo-get-hash in tqe plays undercloud/standalone","message":"Install tripleo-get-hash in tqe plays undercloud/standalone\n\nUsed by https://review.opendev.org/c/openstack/tripleo-quickstart/+/791486\nThe intent is to use this for now and work towards getting into\ntripleo-repos rpm in the long run (and then removing this install).\n\nChange-Id: Ib843f59f16ad30c07296a4328b3547396b94f71a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/fd3958dacc7cc36d662fca50c08abbbd1a13c66a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/fd3958dacc7cc36d662fca50c08abbbd1a13c66a"}]},"branch":"refs/heads/master"},"a1c5cb6dbbe7d262c88027d6bff7ac4ec9e64aad":{"kind":"REWORK","_number":4,"created":"2021-05-31 11:34:55.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/31/793331/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/31/793331/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/4"}}},"commit":{"parents":[{"commit":"8f5db177cd6fcf3265c90dc58998485b9243ee9d","subject":"Separate forwarder condition - with additional debug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/8f5db177cd6fcf3265c90dc58998485b9243ee9d"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-05-27 09:26:47.000000000","tz":180},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-05-31 11:34:38.000000000","tz":180},"subject":"Install tripleo-get-hash in tqe plays undercloud/standalone","message":"Install tripleo-get-hash in tqe plays undercloud/standalone\n\nUsed by https://review.opendev.org/c/openstack/tripleo-quickstart/+/791486\nThe intent is to use this for now and work towards getting into\ntripleo-repos rpm in the long run (and then removing this install).\n\nChange-Id: Ib843f59f16ad30c07296a4328b3547396b94f71a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/a1c5cb6dbbe7d262c88027d6bff7ac4ec9e64aad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/a1c5cb6dbbe7d262c88027d6bff7ac4ec9e64aad"}]},"branch":"refs/heads/master"},"6f9934a4c170bb55adec598ee40b9a1213d64269":{"kind":"REWORK","_number":5,"created":"2021-06-01 10:15:28.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/31/793331/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/31/793331/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/5"}}},"commit":{"parents":[{"commit":"8f5db177cd6fcf3265c90dc58998485b9243ee9d","subject":"Separate forwarder condition - with additional debug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/8f5db177cd6fcf3265c90dc58998485b9243ee9d"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-05-27 09:26:47.000000000","tz":180},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-06-01 10:15:08.000000000","tz":180},"subject":"WIP Install tripleo-get-hash in tqe plays undercloud/standalone","message":"WIP Install tripleo-get-hash in tqe plays undercloud/standalone\n\nUsed by https://review.opendev.org/c/openstack/tripleo-quickstart/+/791486\nThe intent is to use this for now and work towards getting into\ntripleo-repos rpm in the long run (and then removing this install).\n\nChange-Id: Ib843f59f16ad30c07296a4328b3547396b94f71a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/6f9934a4c170bb55adec598ee40b9a1213d64269"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/6f9934a4c170bb55adec598ee40b9a1213d64269"}]},"branch":"refs/heads/master"},"7afe4a804ec1eb77720323208b43e3a8c93bfddf":{"kind":"REWORK","_number":6,"created":"2021-06-01 10:59:19.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/31/793331/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/31/793331/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/6"}}},"commit":{"parents":[{"commit":"8f5db177cd6fcf3265c90dc58998485b9243ee9d","subject":"Separate forwarder condition - with additional debug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/8f5db177cd6fcf3265c90dc58998485b9243ee9d"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-05-27 09:26:47.000000000","tz":180},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-06-01 10:59:08.000000000","tz":180},"subject":"WIP Install tripleo-get-hash in tqe plays undercloud/standalone","message":"WIP Install tripleo-get-hash in tqe plays undercloud/standalone\n\nUsed by https://review.opendev.org/c/openstack/tripleo-quickstart/+/791486\nThe intent is to use this for now and work towards getting into\ntripleo-repos rpm in the long run (and then removing this install).\n\nChange-Id: Ib843f59f16ad30c07296a4328b3547396b94f71a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7afe4a804ec1eb77720323208b43e3a8c93bfddf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7afe4a804ec1eb77720323208b43e3a8c93bfddf"}]},"branch":"refs/heads/master"},"36e623a707d113373230c50e36a2674ca30160d9":{"kind":"REWORK","_number":7,"created":"2021-06-02 11:01:43.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/31/793331/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/31/793331/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/7"}}},"commit":{"parents":[{"commit":"da5514a356d7ed980836fa0dcc1da37566ae994f","subject":"Only add a comma before multiple update repos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/da5514a356d7ed980836fa0dcc1da37566ae994f"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-05-27 09:26:47.000000000","tz":180},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-06-02 11:01:32.000000000","tz":180},"subject":"WIP Install tripleo-get-hash in tqe plays undercloud/standalone","message":"WIP Install tripleo-get-hash in tqe plays undercloud/standalone\n\nUsed by https://review.opendev.org/c/openstack/tripleo-quickstart/+/791486\nThe intent is to use this for now and work towards getting into\ntripleo-repos rpm in the long run (and then removing this install).\n\nChange-Id: Ib843f59f16ad30c07296a4328b3547396b94f71a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/36e623a707d113373230c50e36a2674ca30160d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/36e623a707d113373230c50e36a2674ca30160d9"}]},"branch":"refs/heads/master"},"96d376a00f10a47f71ab73a45200be8bbefb40c4":{"kind":"REWORK","_number":8,"created":"2021-06-03 10:29:54.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/31/793331/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/31/793331/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/8"}}},"commit":{"parents":[{"commit":"da5514a356d7ed980836fa0dcc1da37566ae994f","subject":"Only add a comma before multiple update repos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/da5514a356d7ed980836fa0dcc1da37566ae994f"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-05-27 09:26:47.000000000","tz":180},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-06-03 10:27:06.000000000","tz":180},"subject":"WIP Install tripleo-get-hash in tqe plays undercloud/standalone","message":"WIP Install tripleo-get-hash in tqe plays undercloud/standalone\n\nUsed by https://review.opendev.org/c/openstack/tripleo-quickstart/+/791486\nThe intent is to use this for now and work towards getting into\ntripleo-repos rpm in the long run (and then removing this install).\n\nChange-Id: Ib843f59f16ad30c07296a4328b3547396b94f71a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/96d376a00f10a47f71ab73a45200be8bbefb40c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/96d376a00f10a47f71ab73a45200be8bbefb40c4"}]},"branch":"refs/heads/master"},"852e39c24e0697616960eac60f4ce70e822279ad":{"kind":"REWORK","_number":9,"created":"2021-06-04 13:52:24.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/31/793331/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/31/793331/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/9"}}},"commit":{"parents":[{"commit":"da5514a356d7ed980836fa0dcc1da37566ae994f","subject":"Only add a comma before multiple update repos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/da5514a356d7ed980836fa0dcc1da37566ae994f"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-05-27 09:26:47.000000000","tz":180},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-06-04 13:46:14.000000000","tz":180},"subject":"WIP Install tripleo-get-hash in tqe plays undercloud/standalone","message":"WIP Install tripleo-get-hash in tqe plays undercloud/standalone\n\nUsed by https://review.opendev.org/c/openstack/tripleo-quickstart/+/791486\nThe intent is to use this for now and work towards getting into\ntripleo-repos rpm in the long run (and then removing this install).\n\nChange-Id: Ib843f59f16ad30c07296a4328b3547396b94f71a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/852e39c24e0697616960eac60f4ce70e822279ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/852e39c24e0697616960eac60f4ce70e822279ad"}]},"branch":"refs/heads/master"},"2204bc592cff569b793ece05844c6f38f7032dd5":{"kind":"REWORK","_number":10,"created":"2021-06-07 12:45:41.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/31/793331/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/31/793331/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/10"}}},"commit":{"parents":[{"commit":"582a38e241f77a60eaa1276def64e0a856edda9d","subject":"Merge \"Don\u0027t run ephemeral heat for non enabled scenarios\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/582a38e241f77a60eaa1276def64e0a856edda9d"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-05-27 09:26:47.000000000","tz":180},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-06-07 11:40:21.000000000","tz":180},"subject":"Install tripleo-get-hash in tqe plays undercloud/standalone","message":"Install tripleo-get-hash in tqe plays undercloud/standalone\n\nUsed by https://review.opendev.org/c/openstack/tripleo-quickstart/+/791486\nThe intent is to use this for now and work towards getting into\ntripleo-repos rpm in the long run (and then removing this install).\nThe Depends-On is needed so that zuul closed tripleo-repos and we\ninstall tripleo-get-hash from there.\n\nDepends-On: https://review.opendev.org/c/openstack/tripleo-ci/+/791908\nChange-Id: Ib843f59f16ad30c07296a4328b3547396b94f71a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/2204bc592cff569b793ece05844c6f38f7032dd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/2204bc592cff569b793ece05844c6f38f7032dd5"}]},"branch":"refs/heads/master"},"711bfb6fc6cd160fcc2a806e60611de75df485a2":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2021-07-01 11:54:30.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/31/793331/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/31/793331/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/11"}}},"commit":{"parents":[{"commit":"c907b825fa9750ab795221b80cf8be97c2ae4091","subject":"Merge \"Move atop installation after repo setup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/c907b825fa9750ab795221b80cf8be97c2ae4091"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-05-27 09:26:47.000000000","tz":180},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-07-01 11:54:30.000000000","tz":0},"subject":"Install tripleo-get-hash in tqe plays undercloud/standalone","message":"Install tripleo-get-hash in tqe plays undercloud/standalone\n\nUsed by https://review.opendev.org/c/openstack/tripleo-quickstart/+/791486\nThe intent is to use this for now and work towards getting into\ntripleo-repos rpm in the long run (and then removing this install).\nThe Depends-On is needed so that zuul closed tripleo-repos and we\ninstall tripleo-get-hash from there.\n\nDepends-On: https://review.opendev.org/c/openstack/tripleo-ci/+/791908\nChange-Id: Ib843f59f16ad30c07296a4328b3547396b94f71a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/711bfb6fc6cd160fcc2a806e60611de75df485a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/711bfb6fc6cd160fcc2a806e60611de75df485a2"}]},"branch":"refs/heads/master","description":"Rebase"},"66b9fd8cd86f2a3f028d4a47885b4f2cfa7f0658":{"kind":"REWORK","_number":12,"created":"2021-07-07 10:26:24.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/31/793331/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/31/793331/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/12"}}},"commit":{"parents":[{"commit":"bb4a58b675611b8695b2a9d0f72c410bb73cb2b6","subject":"Merge \"Added missing |bool parameter against use_os_tempest\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/bb4a58b675611b8695b2a9d0f72c410bb73cb2b6"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-05-27 09:26:47.000000000","tz":180},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-07-07 10:15:18.000000000","tz":180},"subject":"Install tripleo-get-hash in tqe plays undercloud/standalone","message":"Install tripleo-get-hash in tqe plays undercloud/standalone\n\nUsed by https://review.opendev.org/c/openstack/tripleo-quickstart/+/791486\nThe intent is to use this for now and work towards getting into\ntripleo-repos rpm in the long run (and then removing this install).\nThe Depends-On is needed so that zuul closed tripleo-repos and we\ninstall tripleo-get-hash from there.\n\nDepends-On: https://review.opendev.org/c/openstack/tripleo-ci/+/791908\nChange-Id: Ib843f59f16ad30c07296a4328b3547396b94f71a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/66b9fd8cd86f2a3f028d4a47885b4f2cfa7f0658"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/66b9fd8cd86f2a3f028d4a47885b4f2cfa7f0658"}]},"branch":"refs/heads/master"},"7644749896301299f825d3e67d3be20a86631177":{"kind":"REWORK","_number":13,"created":"2021-07-13 12:56:35.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/31/793331/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/31/793331/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/13"}}},"commit":{"parents":[{"commit":"bb4a58b675611b8695b2a9d0f72c410bb73cb2b6","subject":"Merge \"Added missing |bool parameter against use_os_tempest\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/bb4a58b675611b8695b2a9d0f72c410bb73cb2b6"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-05-27 09:26:47.000000000","tz":180},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-07-13 12:51:25.000000000","tz":180},"subject":"Install tripleo-get-hash in tqe plays undercloud/standalone","message":"Install tripleo-get-hash in tqe plays undercloud/standalone\n\nUsed by https://review.opendev.org/c/openstack/tripleo-quickstart/+/791486\nThe intent is to use this for now and work towards getting into\ntripleo-repos rpm in the long run (and then removing this install).\nThe patch at [1] (merged) is a dependency of this patch as it\nadds tripleo-repos to the required projects. tripleo-get-hash is\ninstalled from that git checkout.\n\n[1] https://review.opendev.org/c/openstack/tripleo-ci/+/791908\nChange-Id: Ib843f59f16ad30c07296a4328b3547396b94f71a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7644749896301299f825d3e67d3be20a86631177"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7644749896301299f825d3e67d3be20a86631177"}]},"branch":"refs/heads/master"},"284b904ead21a8bcf2de1c55c13a1d4ed33f231b":{"kind":"REWORK","_number":14,"created":"2021-07-14 11:35:33.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/31/793331/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/31/793331/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/14"}}},"commit":{"parents":[{"commit":"8c0f19ae7bfddca7e7e9f272e6156d4102273c1e","subject":"Merge \"Remove `--old-deploy-image` var for image upload\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/8c0f19ae7bfddca7e7e9f272e6156d4102273c1e"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-05-27 09:26:47.000000000","tz":180},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-07-14 11:27:19.000000000","tz":180},"subject":"Install tripleo-get-hash in tqe plays undercloud/standalone","message":"Install tripleo-get-hash in tqe plays undercloud/standalone\n\nUsed by https://review.opendev.org/c/openstack/tripleo-quickstart/+/791486\nThe intent is to use this for now and work towards getting into\ntripleo-repos rpm in the long run (and then removing this install).\nThe patch at [1] (merged) is a dependency of this patch as it\nadds tripleo-repos to the required projects. tripleo-get-hash is\ninstalled from that git checkout.\n\n[1] https://review.opendev.org/c/openstack/tripleo-ci/+/791908\nChange-Id: Ib843f59f16ad30c07296a4328b3547396b94f71a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/284b904ead21a8bcf2de1c55c13a1d4ed33f231b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/284b904ead21a8bcf2de1c55c13a1d4ed33f231b"}]},"branch":"refs/heads/master"},"d8e4a66c71a1a9bee2c4ad23cf7edc16ce415fb1":{"kind":"REWORK","_number":15,"created":"2021-07-22 10:48:11.000000000","uploader":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"ref":"refs/changes/31/793331/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/31/793331/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/15"}}},"commit":{"parents":[{"commit":"062810d721dcbecaec8d605d669720aff2e91b80","subject":"Merge \"Defining tripleo inventory path\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/062810d721dcbecaec8d605d669720aff2e91b80"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-05-27 09:26:47.000000000","tz":180},"committer":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2021-07-22 10:47:37.000000000","tz":60},"subject":"Install tripleo-get-hash in tqe plays undercloud/standalone","message":"Install tripleo-get-hash in tqe plays undercloud/standalone\n\nUsed by https://review.opendev.org/c/openstack/tripleo-quickstart/+/791486\nThe intent is to use this for now and work towards getting into\ntripleo-repos rpm in the long run (and then removing this install).\nThe patch at [1] (merged) is a dependency of this patch as it\nadds tripleo-repos to the required projects. tripleo-get-hash is\ninstalled from that git checkout.\n\n[1] https://review.opendev.org/c/openstack/tripleo-ci/+/791908\nChange-Id: Ib843f59f16ad30c07296a4328b3547396b94f71a\nDepends-On: https://review.opendev.org/c/openstack/tripleo-repos/+/801621\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/d8e4a66c71a1a9bee2c4ad23cf7edc16ce415fb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/d8e4a66c71a1a9bee2c4ad23cf7edc16ce415fb1"}]},"branch":"refs/heads/master"},"0451f0565d15d44599b181b029071c493fab92e1":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2021-08-06 13:56:56.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/31/793331/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/31/793331/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/16"}}},"commit":{"parents":[{"commit":"0735e3916e76adcb01779fe933e2082142d04c52","subject":"Bump nautilus containers to v4.0.19","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/0735e3916e76adcb01779fe933e2082142d04c52"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-05-27 09:26:47.000000000","tz":180},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-08-06 13:47:57.000000000","tz":180},"subject":"Install tripleo-get-hash in tqe plays undercloud/standalone","message":"Install tripleo-get-hash in tqe plays undercloud/standalone\n\nUsed by https://review.opendev.org/c/openstack/tripleo-quickstart/+/791486\nThe intent is to use this for now and work towards getting into\ntripleo-repos rpm in the long run (and then removing this install).\nThe patch at [1] (merged) is a dependency of this patch as it\nadds tripleo-repos to the required projects. tripleo-get-hash is\ninstalled from that git checkout.\n\n[1] https://review.opendev.org/c/openstack/tripleo-ci/+/791908\nChange-Id: Ib843f59f16ad30c07296a4328b3547396b94f71a\nDepends-On: https://review.opendev.org/c/openstack/tripleo-repos/+/801621\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/0451f0565d15d44599b181b029071c493fab92e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/0451f0565d15d44599b181b029071c493fab92e1"}]},"branch":"refs/heads/master"},"f5a4f177fc89b894e4acd56389f52f4f5f50740b":{"kind":"REWORK","_number":17,"created":"2021-08-09 14:19:07.000000000","uploader":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"ref":"refs/changes/31/793331/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/31/793331/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/793331/17"}}},"commit":{"parents":[{"commit":"19e8bdff5f566b978cb448493c9eae6f57a409a8","subject":"Merge \"Update list_allowed variables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/19e8bdff5f566b978cb448493c9eae6f57a409a8"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-05-27 09:26:47.000000000","tz":180},"committer":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2021-08-09 14:18:19.000000000","tz":60},"subject":"WIP: Install tripleo.repos in tqe plays undercloud/standalone","message":"WIP: Install tripleo.repos in tqe plays undercloud/standalone\n\nUsed by https://review.opendev.org/c/openstack/tripleo-quickstart/+/791486\n\nThe intent is to use this for now and work towards getting into\ntripleo-repos rpm in the long run (and then removing this install).\nThe patch at [1] (merged) is a dependency of this patch as it\nadds tripleo-repos to the required projects.\n\ntripleo.repos.get_hash is installed from galaxy or git checkout.\n\n[1] https://review.opendev.org/c/openstack/tripleo-ci/+/791908\nChange-Id: Ib843f59f16ad30c07296a4328b3547396b94f71a\nDepends-On: https://review.opendev.org/c/openstack/tripleo-repos/+/801621\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/f5a4f177fc89b894e4acd56389f52f4f5f50740b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/f5a4f177fc89b894e4acd56389f52f4f5f50740b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
