)]}'
{"id":"openstack%2Ftripleo-quickstart-extras~799517","triplet_id":"openstack%2Ftripleo-quickstart-extras~master~Ie20ccbcbeed0769202a8408e7bae0465daaa1803","project":"openstack/tripleo-quickstart-extras","branch":"master","hashtags":[],"change_id":"Ie20ccbcbeed0769202a8408e7bae0465daaa1803","subject":"Add release_numbering \u0026 release_minor support","status":"MERGED","created":"2021-07-05 16:12:05.000000000","updated":"2021-07-08 02:24:40.000000000","submitted":"2021-07-08 02:24:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":4,"has_review_started":true,"submission_id":"799517","meta_rev_id":"8f72ce0ad2fb1fd7924b9677c4751f8fd9b47318","_number":799517,"virtual_id_number":799517,"owner":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-08 02:24:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-07-07 11:03:35.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"value":0,"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":2,"date":"2021-07-07 07:58:24.000000000","_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2021-07-07 15:28:20.000000000","_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":2,"date":"2021-07-07 21:55:41.000000000","_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2021-07-07 07:54:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"value":1,"date":"2021-07-07 07:50:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"value":0,"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2021-07-07 16:25:00.000000000","_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"value":0,"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"}],"reviewers":{"REVIEWER":[{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-05 19:38:38.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-06 11:17:11.000000000","updated_by":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2021-07-06 11:17:11.000000000","updated_by":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2021-07-06 11:17:11.000000000","updated_by":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"REVIEWER"},{"updated":"2021-07-06 11:17:11.000000000","updated_by":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"reviewer":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"state":"REVIEWER"},{"updated":"2021-07-06 11:17:11.000000000","updated_by":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2021-07-06 11:17:11.000000000","updated_by":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2021-07-06 11:51:34.000000000","updated_by":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"reviewer":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"state":"REVIEWER"},{"updated":"2021-07-06 14:07:58.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2021-07-06 15:50:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cc8572a01974d3689e78aba97d0e70e6712f08ff","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2021-07-05 16:12:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9c920aa65cd965a18ee536608a3fdddaf16ca517","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-07-05 19:38:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/4c391da1f1e04a4dac2f38af44e8b8cd : SUCCESS in 3h 04m 15s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/870ea57d21fd47e381c53186718b0e77 : SUCCESS in 3h 22m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ffd47b3e4a73bc652702f3f1438678f727f97e6d","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2021-07-06 10:30:56.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b81f25649f6b7b49e3d87f55651e453d4b84480a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2021-07-06 10:45:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"49e94902be7aa48e55fe4141c6b5fb6d8ddfa974","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2021-07-06 11:16:38.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"1166148e2c6402edaa8f8999a9574821ea214c44","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2021-07-06 11:51:34.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"4fb964f595a689400c7be9fd2e57668787df0035","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-07-06 12:54:34.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"17c8f90dc27cf7e39f1fc6ce67f86c5f2923b975","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-07-06 13:54:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/1e4ef683c50d4e4181092bf1a014709a : SUCCESS in 3h 07m 02s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/d8a6000eca694df4835997861f7e4ab6 : FAILURE in 2h 30m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2fa3e9d20623814aadfbf7947fc8bf5f7ae81e31","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-07-06 14:07:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e5f30e305a0e236bba30ca42086851a6cef3d25d","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-07-06 14:08:55.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)\n\nLooks good from commit but having doc regarding the same would be awesome,","accounts_in_message":[],"_revision_number":2},{"id":"ec7003961b9ec685f584e6d865ef3f5c2d7a39f9","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2021-07-06 14:45:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a1b05855f79ad7a8c530b2af71a795e6051de77b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 15:50:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/412ded4b1e214e5b83ab8ef64a26b9cf : SUCCESS in 3h 18m 27s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/506a13a918b64b06a0b43091b7d0bb8a : SUCCESS in 1h 07m 36s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/11557c18cb4b4eee97f00b5c5471646c : SUCCESS in 1h 46m 31s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/ef4bfe0f726246708d154fb09f6546da : SUCCESS in 1h 50m 22s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2bacd1595d144e13b50d98cf5fe5ae1f : SUCCESS in 5m 14s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/8cedcabd2dba4c36bfd67486842097c2 : SUCCESS in 3h 04m 11s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/8c39ff70fc9c4f1a8a55ecd64673c605 : SUCCESS in 4h 31m 06s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/8bb8ed25d62744678c5a8fa4c65e792e : SUCCESS in 3h 20m 57s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/e5912675bdf541aaab306470e8bf9852 : SUCCESS in 3h 18m 55s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/f4737e1d04d841898d416fade1e433cd : SUCCESS in 1h 42m 03s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/8e12f1e0ab1f4c98ab11feae29940f84 : SUCCESS in 2h 18m 16s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/b90bcd625c234de395d33a97ecfee933 : SUCCESS in 2h 18m 27s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/ccccd784028346e4a9ec42365299281a : SUCCESS in 2h 25m 25s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/60823344c06844df8527c8f06435cffa : SUCCESS in 1h 27m 26s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/16f4d216054b4c309f0aa4d03fae0dac : SUCCESS in 1h 29m 53s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/1014343f68684829809dc5077cbdd15f : SUCCESS in 12m 59s","accounts_in_message":[],"_revision_number":2},{"id":"3a0842787288e6ce47be1b51e05ea3ad4fde00da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2021-07-07 05:34:20.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"052df53f9b76905e3a632b57ce9852ed93999a21","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2021-07-07 05:41:12.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"2029413644b6f1272324050e936cde7bdc604516","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-07-07 07:08:19.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)\n\nlooks ok and thanks for explanation but see inline spacing nit thanks","accounts_in_message":[],"_revision_number":3},{"id":"feecce00f618772dcd046033320c002bafbc8a7e","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2021-07-07 07:19:57.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5e607a50c2368099eafec03c6f613a27d9f97c32","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2021-07-07 07:20:28.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ad87f506d6eaff73da1f8c70bcb6d55716c48ec1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2021-07-07 07:22:11.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"79d2dfaebd0dc4af5783011fe9c255cc52d67d33","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2021-07-07 07:50:24.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"0bedf2ad3a84141d7a17b3ef80765db601bfbae2","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2021-07-07 07:54:33.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d0f4bd5e6016f9c74dd989d5e83da2914ea3fba5","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-07-07 07:58:24.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"c68bb865748f96c94d50e27912cb80f09e370987","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-07-07 08:00:50.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0338bab5b5ee9a0ee3fb296c8753f3e3165669e3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-07-07 11:03:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/13b5a84b17d949f186f785c82c894187 : SUCCESS in 3h 13m 06s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/40d5088ef43c4231b4bbf941826a9a7b : SUCCESS in 3h 39m 42s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4decbc8b08562b5b6cf11e613fa6a8ed8b8a6dc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 11:18:01.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/32b4609d6b5b4a44af9fc5a45fae046d : SUCCESS in 3h 23m 33s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/b0e21dda73854ebc919e933bebb71d15 : SUCCESS in 1h 11m 11s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/112907eb54a246c0900d3a694481f89d : SUCCESS in 1h 45m 05s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/98afe56856804d2ca40e71a9210dd25f : SUCCESS in 1h 48m 44s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/160a82f6a8a04125be0f594431d8ff49 : SUCCESS in 5m 18s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/9ae54ef97d6442c388b96d858aa8cc2c : SUCCESS in 2h 50m 36s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/ef503ce6c4774411b93c23340c6cde0d : SUCCESS in 3h 45m 22s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/e03266ca758e496b9d435f1fef8eeaa7 : SUCCESS in 3h 50m 03s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/91dcccdd39fb43e5b124e7c82fe3768c : SUCCESS in 3h 07m 34s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/651fd2d07eca4ca3a23eb8569ba1c65d : SUCCESS in 1h 38m 19s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/8eecc775c0cf490686823d4262a99db5 : SUCCESS in 2h 20m 42s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/585f5e31c2664c8aae18ac926dd96418 : SUCCESS in 2h 12m 53s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/b863ba05bb1b456d82f0e6a1f414bc0a : SUCCESS in 2h 02m 53s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/5461284f7d0e479d9205129776ec6802 : SUCCESS in 1h 19m 24s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/b7600c14a3484b6da75939844f13eb87 : SUCCESS in 1h 51m 16s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/cca95c3869a247ffbdae10a10bec1480 : SUCCESS in 11m 15s","accounts_in_message":[],"_revision_number":4},{"id":"803285e13f8b0e6e5dc29d1d0981a979236de405","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2021-07-07 15:28:20.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"ac8546dd424a591ea589dfacc3131ac848b26d1d","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2021-07-07 16:25:00.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"4c294ad5055435841994ee4ab44884cd0ecf59d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 16:52:49.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"339f7dc50df0e995aab86752231e8b99c02c077a","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-07-07 21:55:41.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"2cb8a1e1a6ec375113b22bb1e44e5d5b99b710a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-08 02:24:37.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/e4bb1064a84d47b2a540758abb2717e3 : SUCCESS in 3h 12m 02s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/26f7ee79480c400f8603c60a8a4181dc : SUCCESS in 1h 07m 26s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/c96cb3e231ac484aa6c63103f99a4866 : SUCCESS in 1h 48m 52s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/a506f32b0ae44588a28437308e42bcce : SUCCESS in 1h 30m 53s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/1bcdca86a69f45c4818236ef1fcb6333 : SUCCESS in 3h 04m 08s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/9e5968495f62422993303b2fc70f9391 : SUCCESS in 3h 46m 38s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/affb28bd61f3463d964da40bcfd20e7f : SUCCESS in 3h 32m 45s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/52fa5e8d9b4d452a930bd10d9a6d8701 : SUCCESS in 3h 23m 21s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/3529e05c92ec4185afd25064aed5d33a : SUCCESS in 1h 47m 40s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/8503686f16c3495896befc4da1aa5ef9 : SUCCESS in 2h 16m 10s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/c6424f05733f4fb08308a9531be22a0e : SUCCESS in 2h 04m 50s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/80d0551185a54b7ebc757e8109fc67f3 : SUCCESS in 2h 12m 25s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/4949f687dd4d4598afd055968cefc024 : SUCCESS in 1h 13m 53s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/5b53e4c3a5204518a705d79cea46858e : SUCCESS in 1h 28m 36s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6131054018bd4dd78c1e64e6c387bfa8 : SUCCESS in 7m 21s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/921e103f6d844709b561796c73a68479 : SUCCESS in 16m 48s","accounts_in_message":[],"_revision_number":4},{"id":"8f72ce0ad2fb1fd7924b9677c4751f8fd9b47318","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-08 02:24:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"4907eec94012574e98d104a93ecf88550b41fb07","revisions":{"f40814cb04e276bc0ca38acc0e876262c37059d8":{"kind":"REWORK","_number":1,"created":"2021-07-05 16:12:05.000000000","uploader":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"ref":"refs/changes/17/799517/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/17/799517/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/17/799517/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/17/799517/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/17/799517/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/17/799517/1"}}},"commit":{"parents":[{"commit":"4b4a52632ad1c76d094263b9db74506bc7361af1","subject":"Merge \"Don\u0027t hard-code overcloud-full\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/4b4a52632ad1c76d094263b9db74506bc7361af1"}]}],"author":{"name":"Sandeep Yadav","email":"sandyada@redhat.com","date":"2021-07-05 16:10:19.000000000","tz":330},"committer":{"name":"Sandeep Yadav","email":"sandyada@redhat.com","date":"2021-07-05 16:12:01.000000000","tz":330},"subject":"Add release_numbering \u0026 release_minor support","message":"Add release_numbering \u0026 release_minor support\n\nChange-Id: Ie20ccbcbeed0769202a8408e7bae0465daaa1803\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/f40814cb04e276bc0ca38acc0e876262c37059d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/f40814cb04e276bc0ca38acc0e876262c37059d8"}]},"branch":"refs/heads/master"},"0e3cf2d69ddec659967bc2b06f67c5d16fe93ea1":{"kind":"REWORK","_number":2,"created":"2021-07-06 10:45:53.000000000","uploader":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"ref":"refs/changes/17/799517/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/17/799517/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/17/799517/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/17/799517/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/17/799517/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/17/799517/2"}}},"commit":{"parents":[{"commit":"4b4a52632ad1c76d094263b9db74506bc7361af1","subject":"Merge \"Don\u0027t hard-code overcloud-full\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/4b4a52632ad1c76d094263b9db74506bc7361af1"}]}],"author":{"name":"Sandeep Yadav","email":"sandyada@redhat.com","date":"2021-07-05 16:10:19.000000000","tz":330},"committer":{"name":"Sandeep Yadav","email":"sandyada@redhat.com","date":"2021-07-06 10:45:49.000000000","tz":330},"subject":"Add release_numbering \u0026 release_minor support","message":"Add release_numbering \u0026 release_minor support\n\nWith this patch we are adding support for release_numbering\n\u0026 release_minor for downstream.\n\nWe will pass these parameters through downstream release config\nfiles, should be transparent for upstream.\n\n• `release_numbering` defines the algorithm used by DLRN to assign\n   release numbers to packages.\n\nThe release number is created from the current date and the source\nrepository git hash, and can use different algorithms:\n\n– 0.date.hash if the old method is used: 0.\u003cdate\u003e.\u003chash\u003e\n– 0.1.date.hash if the new method is used: 0.1.\u003cdate\u003e.\u003chash\u003e.\nThis new method provides better compatibility with the Fedora\npackaging guidelines.\n– minor.date.hash allows you to specify the minor version to be\nused, which can be different from 0.\nIf this release numbering schema is used, the value of minor\nwill be determined by release_minor.\n\n• `release_minor` only takes place when release_numbering\nis set to minor.date.hash. For example, if this value is set to 3,\nthe release number for all packages will be computed as 3.date.hash.\n\nChange-Id: Ie20ccbcbeed0769202a8408e7bae0465daaa1803\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/0e3cf2d69ddec659967bc2b06f67c5d16fe93ea1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/0e3cf2d69ddec659967bc2b06f67c5d16fe93ea1"}]},"branch":"refs/heads/master"},"b0ee2586da55567809ba2e479895a152e3a1ad6e":{"kind":"REWORK","_number":3,"created":"2021-07-07 05:34:20.000000000","uploader":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"ref":"refs/changes/17/799517/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/17/799517/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/17/799517/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/17/799517/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/17/799517/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/17/799517/3"}}},"commit":{"parents":[{"commit":"4b4a52632ad1c76d094263b9db74506bc7361af1","subject":"Merge \"Don\u0027t hard-code overcloud-full\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/4b4a52632ad1c76d094263b9db74506bc7361af1"}]}],"author":{"name":"Sandeep Yadav","email":"sandyada@redhat.com","date":"2021-07-05 16:10:19.000000000","tz":330},"committer":{"name":"Sandeep Yadav","email":"sandyada@redhat.com","date":"2021-07-07 05:34:16.000000000","tz":330},"subject":"Add release_numbering \u0026 release_minor support","message":"Add release_numbering \u0026 release_minor support\n\nWith this patch we are adding support for release_numbering\n\u0026 release_minor for downstream.\n\nWe will pass these parameters through downstream release config\nfiles, should be transparent for upstream.\n\n• `release_numbering` defines the algorithm used by DLRN to assign\n   release numbers to packages.\n\nThe release number is created from the current date and the source\nrepository git hash, and can use different algorithms:\n\n– 0.date.hash if the old method is used: 0.\u003cdate\u003e.\u003chash\u003e\n– 0.1.date.hash if the new method is used: 0.1.\u003cdate\u003e.\u003chash\u003e.\nThis new method provides better compatibility with the Fedora\npackaging guidelines.\n– minor.date.hash allows you to specify the minor version to be\nused, which can be different from 0.\nIf this release numbering schema is used, the value of minor\nwill be determined by release_minor.\n\n• `release_minor` only takes place when release_numbering\nis set to minor.date.hash. For example, if this value is set to 3,\nthe release number for all packages will be computed as 3.date.hash.\n\nChange-Id: Ie20ccbcbeed0769202a8408e7bae0465daaa1803\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/b0ee2586da55567809ba2e479895a152e3a1ad6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/b0ee2586da55567809ba2e479895a152e3a1ad6e"}]},"branch":"refs/heads/master"},"4907eec94012574e98d104a93ecf88550b41fb07":{"kind":"REWORK","_number":4,"created":"2021-07-07 07:22:11.000000000","uploader":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"ref":"refs/changes/17/799517/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/17/799517/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/17/799517/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/17/799517/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/17/799517/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/17/799517/4"}}},"commit":{"parents":[{"commit":"4b4a52632ad1c76d094263b9db74506bc7361af1","subject":"Merge \"Don\u0027t hard-code overcloud-full\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/4b4a52632ad1c76d094263b9db74506bc7361af1"}]}],"author":{"name":"Sandeep Yadav","email":"sandyada@redhat.com","date":"2021-07-05 16:10:19.000000000","tz":330},"committer":{"name":"Sandeep Yadav","email":"sandyada@redhat.com","date":"2021-07-07 07:21:55.000000000","tz":330},"subject":"Add release_numbering \u0026 release_minor support","message":"Add release_numbering \u0026 release_minor support\n\nWith this patch we are adding support for release_numbering\n\u0026 release_minor for downstream.\n\nWe will pass these parameters through downstream release config\nfiles, should be transparent for upstream.\n\n• `release_numbering` defines the algorithm used by DLRN to assign\n   release numbers to packages.\n\nThe release number is created from the current date and the source\nrepository git hash, and can use different algorithms:\n\n– 0.date.hash if the old method is used: 0.\u003cdate\u003e.\u003chash\u003e\n– 0.1.date.hash if the new method is used: 0.1.\u003cdate\u003e.\u003chash\u003e.\nThis new method provides better compatibility with the Fedora\npackaging guidelines.\n– minor.date.hash allows you to specify the minor version to be\nused, which can be different from 0.\nIf this release numbering schema is used, the value of minor\nwill be determined by release_minor.\n\n• `release_minor` only takes place when release_numbering\nis set to minor.date.hash. For example, if this value is set to 3,\nthe release number for all packages will be computed as 3.date.hash.\n\nChange-Id: Ie20ccbcbeed0769202a8408e7bae0465daaa1803\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/4907eec94012574e98d104a93ecf88550b41fb07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/4907eec94012574e98d104a93ecf88550b41fb07"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
