)]}'
{"id":"openstack%2Ftripleo-quickstart-extras~814699","triplet_id":"openstack%2Ftripleo-quickstart-extras~master~If6f0c21581039a2318a5b019175852f5c5263f62","project":"openstack/tripleo-quickstart-extras","branch":"master","attention_set":{},"removed_from_attention_set":{"8449":{"account":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"last_update":"2021-10-26 15:14:07.000000000","reason":"Change was submitted"},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2021-10-26 15:14:07.000000000","reason":"Change was submitted"},"9976":{"account":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"last_update":"2021-10-26 15:14:07.000000000","reason":"Change was submitted"},"12393":{"account":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"last_update":"2021-10-26 08:40:30.000000000","reason":"Chandan Kumar replied on the change"},"29775":{"account":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"last_update":"2021-10-25 05:27:32.000000000","reason":"Sandeep Yadav replied on the change"}},"hashtags":[],"change_id":"If6f0c21581039a2318a5b019175852f5c5263f62","subject":"Allow to build containers before undercloud-install","status":"MERGED","created":"2021-10-20 04:39:01.000000000","updated":"2021-11-01 13:24:23.000000000","submitted":"2021-10-26 15:14:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"814699","meta_rev_id":"d14e197946320cbf25b04e883168ef7dd6c5f11e","_number":814699,"virtual_id_number":814699,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-26 15:14:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-10-21 12:22:31.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"recommended":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":1,"date":"2021-10-26 07:08:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":1,"date":"2021-10-25 13:08:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":2,"date":"2021-10-26 08:40:30.000000000","_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":2,"date":"2021-10-25 05:27:32.000000000","_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":1,"date":"2021-10-26 08:40:30.000000000","_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"}],"reviewers":{"REVIEWER":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"CC":[{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-20 04:47:13.000000000","updated_by":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-20 05:30:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-20 21:20:25.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-21 14:06:29.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2021-10-21 14:06:29.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"REVIEWER"},{"updated":"2021-10-21 14:06:29.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"state":"REVIEWER"},{"updated":"2021-10-21 14:06:29.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2021-10-25 13:08:33.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"}],"messages":[{"id":"d4b4c06a9ddb6285d37cf2bf11288b7cc71ef846","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-10-20 04:39:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d39edae6b524b1fcf6d42dc4a7f763b4f5a5b122","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-10-20 04:47:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fd88b77ecc55fc4e68ee7ea0de465981c745c7c7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2021-10-20 04:47:13.000000000","message":"Patch Set 1:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":1},{"id":"71d578c2ca8bb0ea0a96fdaf5ce3206627d812ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 05:30:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/922b5783d26643a78730eaf454f8082d : FAILURE in 18m 54s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5c601f0103c342129a667a6d1dc5c1f7 : SUCCESS in 5m 42s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/6d74ab3a6806461e91759dc5d6154498 : FAILURE in 30m 48s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/630ac6f088f1480fb06e4d7b1a843e9e : FAILURE in 24m 26s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/62eb3c668d1f4b15a218ef7eecc1bb92 : FAILURE in 25m 57s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/54d24fc2b62840ff9ff3f434e2203883 : FAILURE in 24m 52s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/de27ed68627942739c1b0d4046c438c8 : SUCCESS in 10m 40s","accounts_in_message":[],"_revision_number":2},{"id":"015bd381d9f00249d1a1d76619e46648616442a0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2021-10-20 05:32:34.000000000","message":"Patch Set 2:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/267/","accounts_in_message":[],"_revision_number":2},{"id":"7b668a54239f276587624e000fe95580b42c1a71","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2021-10-20 05:37:43.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/267/ : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"0bb5ab723b5226d62e6c52438a203e785ddea10d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-10-20 05:40:48.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"310e807dcd8c58928455db03ecebc9047b6402a4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2021-10-20 05:58:31.000000000","message":"Patch Set 3:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/269/","accounts_in_message":[],"_revision_number":3},{"id":"7e147f3d19fc8390f41b2cc91c82125a127cb105","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2021-10-20 06:03:46.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/269/ : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"31dc873eb0a45198456de11a6c95c298e76f6e39","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-10-20 06:19:27.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a52d674308ef0294e9ec73faefcc556d0deb800f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2021-10-20 07:09:22.000000000","message":"Patch Set 4:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/270/","accounts_in_message":[],"_revision_number":4},{"id":"206c078c17357079f3064ea62a1deb31b2830c7f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2021-10-20 07:13:54.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/270/ : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"14a819b829744019c925a44c7fb58eee2df259a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-10-20 07:40:35.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c4cec2cc38261bbcf2acf031d49ed363604c032a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2021-10-20 08:31:22.000000000","message":"Patch Set 5:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/271/","accounts_in_message":[],"_revision_number":5},{"id":"7fb6dec563853606995082fe4a059a2404954e9c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2021-10-20 08:34:31.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/271/ : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"34a317616429c487f7de409e6167bc77934a1700","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-10-20 09:25:41.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"6cf6341aa79e639f0039c8427812923afb22d0ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-10-20 09:27:01.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"bd99ddbab5e5a69e8d9db294a63c4c7ef33b1954","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2021-10-20 09:27:02.000000000","message":"Patch Set 6:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":6},{"id":"7cb87c68b13a6782c49e0457bcbf0a32c87d70b2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2021-10-20 10:07:52.000000000","message":"Patch Set 7:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/272/","accounts_in_message":[],"_revision_number":7},{"id":"d91fbd5401d2f2b7f93901a6fe1c5d9939be49f5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2021-10-20 10:11:06.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/272/ : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"ecb9f6da98622aba58884373c2d947e26ac6650f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-10-20 11:27:01.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"eebe5f76939dbf7ca97163b21b8ecb81854b8619","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2021-10-20 11:27:20.000000000","message":"Patch Set 8:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/274/","accounts_in_message":[],"_revision_number":8},{"id":"61fad75ba6c22670e2714e19dd5439a6bbef3ce3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2021-10-20 11:32:28.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/274/ : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"c5850523e39fb6f8db6cffbe4921c4079f6421eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-10-20 12:58:22.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"7427e8ce561a7ef20a3c13b31f7f816cdf32adc4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2021-10-20 12:58:43.000000000","message":"Patch Set 9:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/277/","accounts_in_message":[],"_revision_number":9},{"id":"1d76d6dc58f774c6ed1eb4e3bde5dd17f3e0a84b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2021-10-20 13:03:49.000000000","message":"Patch Set 9:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/277/ : FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"89ec4177f0712e8a91e92f05fedd59df1fb172f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-10-20 15:41:53.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"15ac7fdef37c2c2156128d1d77371222fd1bd1d1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2021-10-20 15:57:51.000000000","message":"Patch Set 10:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/278/","accounts_in_message":[],"_revision_number":10},{"id":"a4596b77198b2f2dd28bc5263722eb942172b40c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2021-10-20 18:53:21.000000000","message":"Patch Set 10:\n\nBuild Successful \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/278/ : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"d6d36e2b05da8ecf890ac40000ce6d40c4eb0237","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 20:05:47.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/3588911493384dc1a43f0b519903f65c : SUCCESS in 2h 57m 16s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/e18f879d4e9b45de8b27a9100573d635 : SUCCESS in 52m 00s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/2fb61a99132c4704aa58592d04bff56c : SUCCESS in 1h 37m 59s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/0c46c9c2a05949c0b0308af0d5b3dd61 : SUCCESS in 1h 33m 50s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5182e60f391647589980ac279055e25a : SUCCESS in 5m 43s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/455d1c5c3ebd4b1e93384cf8ffb50f02 : SUCCESS in 2h 48m 21s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/79bf7afb6192445693affd6a662af8e2 : SUCCESS in 3h 04m 52s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/e133e084a3a0461cb1d98a913bc62b2f : SUCCESS in 4h 07m 13s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/2d2fb34e50524421bb5abce362a9cba1 : SUCCESS in 2h 56m 44s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/10bc248eebbf46c58355862c03b1b94a : SUCCESS in 1h 35m 39s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/87694ba4ab614b1ca1cf19a9217d39bb : SUCCESS in 1h 56m 35s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/2b4c106a5af9439d9cc2cca41cfd6f59 : SUCCESS in 2h 23m 26s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/288a5ceeabdb4f1386399a3d3b65b259 : SUCCESS in 1h 58m 01s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/0c56f05018e54a2dad543690e8254b76 : SUCCESS in 1h 20m 57s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/a5bea5bf2a634753a371d941f777803a : FAILURE in 46m 51s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/7d68a4a45a5742b1ab57e20b2a5f056f : SUCCESS in 1h 06m 42s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/72b9b7d37bb04787a28207254b42036d : SUCCESS in 12m 52s","accounts_in_message":[],"_revision_number":10},{"id":"b0e8d07d7b403283388c85932350d7436b7dbaff","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-10-20 21:20:25.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/460af2a1f97d4aa89ff38b2b8e6595b2 : SUCCESS in 2h 13m 53s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/e3837cd3c77241a0bbcbcbcf41469437 : SUCCESS in 2h 33m 45s (non-voting)\n- tripleo-ci-centos-8-ovb-3ctlr_1comp_1supp-featureset039-ussuri-branch https://review.rdoproject.org/zuul/build/f3a810613def44d88ff2557c99cb8846 : POST_FAILURE in 5h 08m 03s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp_1supp-featureset039 https://review.rdoproject.org/zuul/build/06ee309bf6f6483b81eab7f0d8545159 : FAILURE in 2h 29m 35s","accounts_in_message":[],"_revision_number":10},{"id":"e4d4bca8f3dccac440a9cfc0e9a7529a537bc9f9","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-10-21 04:10:36.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"008cd2ea386f2e0a13edd0bb15ef6314b2837f78","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2021-10-21 05:01:11.000000000","message":"Patch Set 10:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/281/","accounts_in_message":[],"_revision_number":10},{"id":"e92ea6e96b6567589976ec35d389996a9967181f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-10-21 06:56:06.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/9cfa500238c342fca1e319a7875c5b29 : SUCCESS in 2h 13m 42s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/afae893d23b64601ac0a5b43c4053541 : SUCCESS in 2h 25m 08s (non-voting)\n- tripleo-ci-centos-8-ovb-3ctlr_1comp_1supp-featureset039-ussuri-branch https://review.rdoproject.org/zuul/build/903b837c07a34b30bb4645f19ba97e07 : FAILURE in 1h 10m 32s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp_1supp-featureset039 https://review.rdoproject.org/zuul/build/d677c150faff45a2a8f906d71e34d07f : SUCCESS in 2h 44m 27s","accounts_in_message":[],"_revision_number":10},{"id":"b9d839d89a9daa6483ebea363a4de5eb15e6393f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 08:16:32.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/397730a59d8547b487bd5a4c4adb3182 : SUCCESS in 2h 50m 16s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/4e3d553d6f584bdb9ad16d89326081c9 : SUCCESS in 42m 13s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/b31f82dd63884c7094bb5b64c658bd74 : SUCCESS in 1h 38m 07s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/a8fb1506945b49a393773f07c5ca0c2c : SUCCESS in 1h 22m 52s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1c73fd4d221b43658c95913053f47385 : SUCCESS in 5m 28s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/bbb84ff0355d4d20839dad34f6fb428d : SUCCESS in 3h 22m 10s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/73f7bb1967544c6a808ee58b82a72e6b : SUCCESS in 3h 31m 03s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/a841c2e488df481ea8cbdd2718beef3d : SUCCESS in 3h 48m 25s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/1c93206500d04fdda2df574a29f51fd0 : SUCCESS in 3h 30m 16s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/9123041c3e874353981da1eb889bd547 : SUCCESS in 1h 56m 27s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/88dcd58dfaff43a8854a5f2bd9eb5782 : SUCCESS in 2h 03m 26s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/cdc40a435c6d4fce97fc57bbf6dad219 : SUCCESS in 2h 06m 52s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/6a2fd4cca37a4a28985bdc5c820d588e : SUCCESS in 2h 08m 38s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/587c3494dd654c1aaeac7131b0791541 : SUCCESS in 1h 16m 49s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/efd5fad4373a4f2ca9d62015ed245354 : FAILURE in 33m 38s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/4736d971746943dfb39b0fd26e058ce8 : FAILURE in 1h 06m 28s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/e3ddf0ee695c4b33832b4699f6283285 : SUCCESS in 12m 42s","accounts_in_message":[],"_revision_number":10},{"id":"7f51b3dc24a81a522fcd5c6c1e76c01c40921d3f","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-10-21 09:08:29.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"9e6248250c0bcf6e409a9604b0aa4c49aeed9e20","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2021-10-21 09:24:42.000000000","message":"Patch Set 10:\n\nBuild Successful \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/281/ : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"f53b70ae27aac2e82807c509afa92c7d128b0058","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-10-21 12:22:31.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/3d83688318e741488618a7836e3c8370 : SUCCESS in 2h 15m 47s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/24784eedca4c48a382b5ec445e9bdfab : SUCCESS in 2h 56m 29s (non-voting)\n- tripleo-ci-centos-8-ovb-3ctlr_1comp_1supp-featureset039-ussuri-branch https://review.rdoproject.org/zuul/build/9e029f0d946e4bbeab0d9891f2d919b6 : SUCCESS in 3h 12m 09s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp_1supp-featureset039 https://review.rdoproject.org/zuul/build/4f7ea6903e3a435e9f42fd9565f6b7fc : SUCCESS in 3h 01m 29s","accounts_in_message":[],"_revision_number":10},{"id":"331fba656dc6d922ee32101e60a3815ec079e635","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 13:06:54.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/66b8f33842644dd4b328185ee0153f25 : SUCCESS in 3h 39m 48s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/0fa38fdde3004fadbb2bc171291e7a4a : SUCCESS in 49m 04s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/b0569bd295d4418baddc930be29cfbc9 : SUCCESS in 1h 40m 29s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/8a1eaa16c77840d4988efc86435d1a1c : SUCCESS in 1h 45m 51s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/013489235aae4c938584f19d3615bb67 : SUCCESS in 7m 17s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/3a90a1fa9c184277a9fa48b75e9d4a8b : SUCCESS in 3h 25m 13s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/0497951458db42d580ac7b126ee26d80 : SUCCESS in 3h 42m 09s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/5fedcc9ddebb404b9ae2aec429cf185e : SUCCESS in 3h 34m 28s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/f69a98909f6d415e8fd32e3152918eaf : SUCCESS in 3h 43m 45s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/bfa3a295bad44b688be0287dab2f6b45 : SUCCESS in 1h 59m 21s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/581bdd42145744e0ad19e037be262ad7 : SUCCESS in 2h 18m 54s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/c323784fbb9740d3a260d314c1d74585 : SUCCESS in 2h 07m 49s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/efc66b77ee9c46918e6fcbb45339e0a4 : SUCCESS in 2h 19m 11s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/854e7cd565994a008ba5d2525b49da3a : SUCCESS in 1h 20m 12s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/e438ba7f78bd4cdcaa04463e4adf1228 : SUCCESS in 1h 34m 37s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/6968ec6a0bb64a15b1790de50f54aa59 : SUCCESS in 1h 11m 53s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/ae62f3a80322408db500d90febc3873c : SUCCESS in 15m 50s","accounts_in_message":[],"_revision_number":10},{"id":"45e4079dda26e634c896ea07d8868db1f029b7b3","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-10-21 14:06:29.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"7173a807b28c12fc109f813f4bb0a2a9403c5003","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2021-10-25 05:27:32.000000000","message":"Patch Set 10: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"a7b872164ef736a620f5cf6eb8665a1f2b17f680","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-10-25 05:35:37.000000000","message":"Patch Set 10: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"09b990c18c2cd1ae1798583d0817669ff6015830","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-10-25 05:40:28.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"7203be654a8bb25e212814776e8f8e35fa2ea745","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2021-10-25 13:08:33.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"63b2b5ea68a96a300341b2a6c29d2a44f97ca740","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-10-26 06:31:31.000000000","message":"Patch Set 10: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"8f3eb7472a4b2567d560fecbbf7eb0574a64b09a","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-10-26 06:50:35.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"f5614c957dcfa0dcae88b10c133bde631aa984ad","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-10-26 07:08:26.000000000","message":"Patch Set 10: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"2f7841f60548611b912cd2c91c7991eda174eddc","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-10-26 08:40:30.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"ba9a07266be1cd0f04d155f5f0a792ea58d3f4df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 08:41:47.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"ab75dc83473b87319120cdd6462fec4353610239","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 15:14:06.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/4fcee1d30c0048148b9d6c29e3ea8ccd : SUCCESS in 3h 04m 50s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/ea9749609e2c454cb5d81c4097984c4a : SUCCESS in 40m 58s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/5c0bf83095e4453cb9a29e6062eed7ff : SUCCESS in 1h 30m 24s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/970e621c0f8f499cbad09ed2eb4272b9 : SUCCESS in 1h 25m 43s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/3ba1273206a84b519c733b4d42811fdc : SUCCESS in 3h 07m 05s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/7c0bc01a7a304711a8595723e22f4ae3 : SUCCESS in 3h 27m 19s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/25d41910ef2f433291bcbcdc752c36ec : SUCCESS in 4h 13m 56s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/c04df97ab80248bfb61f3e7f0cd8dc8c : SUCCESS in 3h 32m 29s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/cdb6559c2d6844d9b75713b2025ebf8a : SUCCESS in 1h 42m 34s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/60768c3ad3a04ae1a3c3a19a0fcf6838 : SUCCESS in 2h 04m 03s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/72a1852aeefe458fa8237898d4582681 : SUCCESS in 2h 05m 03s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/b20a336daa3949ca831b4d3229c9d429 : SUCCESS in 2h 14m 37s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/301eebf29ac042dcbcefa46ad83219c9 : SUCCESS in 1h 23m 25s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/4879b3fcc47140ceb43246b2bcb7257c : SUCCESS in 1h 27m 31s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/1a1ea0ea74e547ab83d0b15acb111d3b : SUCCESS in 1h 14m 41s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/736501aea35a4616ae1e5bc95dd25d18 : SUCCESS in 6m 28s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/6ace64e02c1a4dc686f45b97f1fea265 : SUCCESS in 14m 55s","accounts_in_message":[],"_revision_number":10},{"id":"bb64c257456c3eb415361f0f096df46de88e77ed","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 15:14:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"d14e197946320cbf25b04e883168ef7dd6c5f11e","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-11-01 13:24:23.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"68af35edf1c8885ebd9a0d836e61509276cad5ee","revisions":{"35709bf2a3c6bed6e048751afd9b23043a6c30ae":{"kind":"REWORK","_number":1,"created":"2021-10-20 04:39:01.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/99/814699/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/99/814699/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/99/814699/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/99/814699/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/99/814699/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/99/814699/1"}}},"commit":{"parents":[{"commit":"38e75338f68ffa5dd125913b1e0579cea4036d25","subject":"Fix multinode-ipa job from building-test-packages twice","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/38e75338f68ffa5dd125913b1e0579cea4036d25"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-10-20 04:38:22.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-10-20 04:38:28.000000000","tz":330},"subject":"Allow to build containers in container-prep role","message":"Allow to build containers in container-prep role\n\nIf build_container_images is true build containers\nelse do container updates.\n\nChange-Id: If6f0c21581039a2318a5b019175852f5c5263f62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/35709bf2a3c6bed6e048751afd9b23043a6c30ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/35709bf2a3c6bed6e048751afd9b23043a6c30ae"}]},"branch":"refs/heads/master"},"e84cf21ca1cc239dc5aca093ddebd0e2b09193c0":{"kind":"REWORK","_number":2,"created":"2021-10-20 04:47:12.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/99/814699/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/99/814699/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/99/814699/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/99/814699/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/99/814699/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/99/814699/2"}}},"commit":{"parents":[{"commit":"38e75338f68ffa5dd125913b1e0579cea4036d25","subject":"Fix multinode-ipa job from building-test-packages twice","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/38e75338f68ffa5dd125913b1e0579cea4036d25"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-10-20 04:38:22.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-10-20 04:46:32.000000000","tz":330},"subject":"Allow to build containers in undercloud-setup role","message":"Allow to build containers in undercloud-setup role\n\nIf build_container_images is true build containers\npost repo-setup else do container updates.\n\nChange-Id: If6f0c21581039a2318a5b019175852f5c5263f62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/e84cf21ca1cc239dc5aca093ddebd0e2b09193c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/e84cf21ca1cc239dc5aca093ddebd0e2b09193c0"}]},"branch":"refs/heads/master"},"9ed33217c9473af463c588c17e0a2ee17514444b":{"kind":"REWORK","_number":3,"created":"2021-10-20 05:40:48.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/99/814699/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/99/814699/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/99/814699/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/99/814699/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/99/814699/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/99/814699/3"}}},"commit":{"parents":[{"commit":"38e75338f68ffa5dd125913b1e0579cea4036d25","subject":"Fix multinode-ipa job from building-test-packages twice","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/38e75338f68ffa5dd125913b1e0579cea4036d25"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-10-20 04:38:22.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-10-20 05:39:44.000000000","tz":330},"subject":"Allow to build containers in undercloud-setup role","message":"Allow to build containers in undercloud-setup role\n\nIf build_container_images is true build containers\npost repo-setup else do container updates.\n\nChange-Id: If6f0c21581039a2318a5b019175852f5c5263f62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/9ed33217c9473af463c588c17e0a2ee17514444b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/9ed33217c9473af463c588c17e0a2ee17514444b"}]},"branch":"refs/heads/master"},"75bc1b487efa5b3865da011641ec09ea9eeef5ad":{"kind":"REWORK","_number":4,"created":"2021-10-20 06:19:27.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/99/814699/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/99/814699/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/99/814699/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/99/814699/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/99/814699/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/99/814699/4"}}},"commit":{"parents":[{"commit":"38e75338f68ffa5dd125913b1e0579cea4036d25","subject":"Fix multinode-ipa job from building-test-packages twice","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/38e75338f68ffa5dd125913b1e0579cea4036d25"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-10-20 04:38:22.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-10-20 06:19:14.000000000","tz":330},"subject":"Allow to build containers in undercloud-setup role","message":"Allow to build containers in undercloud-setup role\n\nIf build_container_images is true build containers\npost repo-setup else do container updates.\n\nChange-Id: If6f0c21581039a2318a5b019175852f5c5263f62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/75bc1b487efa5b3865da011641ec09ea9eeef5ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/75bc1b487efa5b3865da011641ec09ea9eeef5ad"}]},"branch":"refs/heads/master"},"81a55299bdf881e2c68afc5c7ee2e984c7b70a2f":{"kind":"REWORK","_number":5,"created":"2021-10-20 07:40:35.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/99/814699/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/99/814699/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/99/814699/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/99/814699/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/99/814699/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/99/814699/5"}}},"commit":{"parents":[{"commit":"38e75338f68ffa5dd125913b1e0579cea4036d25","subject":"Fix multinode-ipa job from building-test-packages twice","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/38e75338f68ffa5dd125913b1e0579cea4036d25"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-10-20 04:38:22.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-10-20 07:40:07.000000000","tz":330},"subject":"Allow to build containers in undercloud-setup role","message":"Allow to build containers in undercloud-setup role\n\nIf build_container_images is true build containers\npost repo-setup else do container updates.\n\nChange-Id: If6f0c21581039a2318a5b019175852f5c5263f62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/81a55299bdf881e2c68afc5c7ee2e984c7b70a2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/81a55299bdf881e2c68afc5c7ee2e984c7b70a2f"}]},"branch":"refs/heads/master"},"eda0da4f43ece879814baf0580638e20691cbaf3":{"kind":"REWORK","_number":6,"created":"2021-10-20 09:25:41.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/99/814699/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/99/814699/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/99/814699/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/99/814699/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/99/814699/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/99/814699/6"}}},"commit":{"parents":[{"commit":"38e75338f68ffa5dd125913b1e0579cea4036d25","subject":"Fix multinode-ipa job from building-test-packages twice","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/38e75338f68ffa5dd125913b1e0579cea4036d25"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-10-20 04:38:22.000000000","tz":330},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2021-10-20 09:25:10.000000000","tz":0},"subject":"Allow to build containers in undercloud-setup role","message":"Allow to build containers in undercloud-setup role\n\nIf build_container_images is true build containers\npost repo-setup else do container updates.\n\nChange-Id: If6f0c21581039a2318a5b019175852f5c5263f62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/eda0da4f43ece879814baf0580638e20691cbaf3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/eda0da4f43ece879814baf0580638e20691cbaf3"}]},"branch":"refs/heads/master"},"f3643c3a843fedf2e3945086f4bfab4321e6beff":{"kind":"REWORK","_number":7,"created":"2021-10-20 09:27:01.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/99/814699/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/99/814699/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/99/814699/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/99/814699/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/99/814699/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/99/814699/7"}}},"commit":{"parents":[{"commit":"38e75338f68ffa5dd125913b1e0579cea4036d25","subject":"Fix multinode-ipa job from building-test-packages twice","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/38e75338f68ffa5dd125913b1e0579cea4036d25"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-10-20 04:38:22.000000000","tz":330},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2021-10-20 09:27:01.000000000","tz":0},"subject":"Allow to build containers in undercloud-setup role","message":"Allow to build containers in undercloud-setup role\n\nIf build_container_images is true build containers\npost repo-setup else do container updates.\n\nChange-Id: If6f0c21581039a2318a5b019175852f5c5263f62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/f3643c3a843fedf2e3945086f4bfab4321e6beff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/f3643c3a843fedf2e3945086f4bfab4321e6beff"}]},"branch":"refs/heads/master"},"d7934eb5334825d50b2f805dbe48af53037e6cbf":{"kind":"REWORK","_number":8,"created":"2021-10-20 11:27:01.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/99/814699/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/99/814699/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/99/814699/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/99/814699/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/99/814699/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/99/814699/8"}}},"commit":{"parents":[{"commit":"38e75338f68ffa5dd125913b1e0579cea4036d25","subject":"Fix multinode-ipa job from building-test-packages twice","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/38e75338f68ffa5dd125913b1e0579cea4036d25"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-10-20 04:38:22.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-10-20 11:26:51.000000000","tz":330},"subject":"Allow to build containers in undercloud-setup role","message":"Allow to build containers in undercloud-setup role\n\nIf build_container_images is true build containers\npost repo-setup else do container updates.\n\nChange-Id: If6f0c21581039a2318a5b019175852f5c5263f62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/d7934eb5334825d50b2f805dbe48af53037e6cbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/d7934eb5334825d50b2f805dbe48af53037e6cbf"}]},"branch":"refs/heads/master"},"f26779fca12e12c4e9d05d5f5937e4ed8fdf5a2a":{"kind":"REWORK","_number":9,"created":"2021-10-20 12:58:22.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/99/814699/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/99/814699/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/99/814699/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/99/814699/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/99/814699/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/99/814699/9"}}},"commit":{"parents":[{"commit":"38e75338f68ffa5dd125913b1e0579cea4036d25","subject":"Fix multinode-ipa job from building-test-packages twice","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/38e75338f68ffa5dd125913b1e0579cea4036d25"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-10-20 04:38:22.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-10-20 12:58:12.000000000","tz":330},"subject":"Allow to build containers in undercloud-setup role","message":"Allow to build containers in undercloud-setup role\n\nIf build_container_images is true build containers\npost repo-setup else do container updates.\n\nChange-Id: If6f0c21581039a2318a5b019175852f5c5263f62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/f26779fca12e12c4e9d05d5f5937e4ed8fdf5a2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/f26779fca12e12c4e9d05d5f5937e4ed8fdf5a2a"}]},"branch":"refs/heads/master"},"68af35edf1c8885ebd9a0d836e61509276cad5ee":{"kind":"REWORK","_number":10,"created":"2021-10-20 15:41:53.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/99/814699/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/99/814699/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/99/814699/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/99/814699/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/99/814699/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/99/814699/10"}}},"commit":{"parents":[{"commit":"ebf8e6a2578c84952cd923a7d89c74ab9065e4a0","subject":"Drop legacy undercloud install","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/ebf8e6a2578c84952cd923a7d89c74ab9065e4a0"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-10-20 04:38:22.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-10-20 15:38:21.000000000","tz":330},"subject":"Allow to build containers before undercloud-install","message":"Allow to build containers before undercloud-install\n\nIf build_container_images is true build containers\nand use these build containers instead of downloading\ncontainers and running container updates.\n\nChange-Id: If6f0c21581039a2318a5b019175852f5c5263f62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/68af35edf1c8885ebd9a0d836e61509276cad5ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/68af35edf1c8885ebd9a0d836e61509276cad5ee"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
