)]}'
{"id":"openstack%2Ftripleo-quickstart-extras~835203","triplet_id":"openstack%2Ftripleo-quickstart-extras~master~I23788ec24c581c037e2535201aeaf61e1bbbbbdc","project":"openstack/tripleo-quickstart-extras","branch":"master","topic":"lp/1966382","attention_set":{},"removed_from_attention_set":{"29775":{"account":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"last_update":"2022-03-25 17:59:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I23788ec24c581c037e2535201aeaf61e1bbbbbdc","subject":"Disable auto discovery as a workaround","status":"MERGED","created":"2022-03-25 10:54:43.000000000","updated":"2022-03-25 19:17:59.000000000","submitted":"2022-03-25 17:59:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"835203-lp/1966382","meta_rev_id":"2c22f74824bb8cdf0a23940f069b3a5c4910b3b8","_number":835203,"virtual_id_number":835203,"owner":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-25 17:59:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"recommended":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"all":[{"value":1,"date":"2022-03-25 14:40:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},{"value":2,"date":"2022-03-25 14:39:10.000000000","_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":2,"date":"2022-03-25 14:42:06.000000000","_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"all":[{"value":0,"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":1,"date":"2022-03-25 14:42:06.000000000","_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"}],"CC":[{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-25 10:55:07.000000000","updated_by":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-25 10:55:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-25 14:39:10.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"REVIEWER"},{"updated":"2022-03-25 14:40:17.000000000","updated_by":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"reviewer":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"state":"REVIEWER"},{"updated":"2022-03-25 14:42:06.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"}],"messages":[{"id":"9411c33f90a3fc103bf18a7fbf44e499133d38c8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2022-03-25 10:54:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8f8a55f053cf9b30316d177b0467cbfc9953cc18","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-03-25 10:55:07.000000000","message":"Patch Set 1:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/796/","accounts_in_message":[],"_revision_number":1},{"id":"a4b00f46954cb021b7215f9fd5fcb630b32409be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-25 10:55:14.000000000","message":"Patch Set 1: Verified-1\n\nUnable to freeze job graph: Job tripleo-ci-centos-9-undercloud-containers depends on tripleo-ci-centos-9-content-provider which was not run.","accounts_in_message":[],"_revision_number":1},{"id":"e7e0b74f442dcd8bbf5bd59a61723339343a03e1","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2022-03-25 10:56:00.000000000","message":"Topic set to lp/1966382","accounts_in_message":[],"_revision_number":1},{"id":"ec731885b4219990e156da898b1d4e4598b366ef","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2022-03-25 13:21:02.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"0f2266d9330133a784f5f6988567a1eead0fd8ad","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-03-25 13:21:19.000000000","message":"Patch Set 2:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/798/","accounts_in_message":[],"_revision_number":2},{"id":"2d1e2eea14f3283ac497d62329844ffc84ffe0f8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-03-25 13:30:19.000000000","message":"Patch Set 1:\n\nBuild Aborted \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/796/ : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"844d269af5725ab40b92dd1ccba10e7ca2a62b7f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2022-03-25 13:33:24.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f3950d004e0cb237b2c9a8a8d4ad83b4af25481e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-03-25 13:33:50.000000000","message":"Patch Set 3:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/800/","accounts_in_message":[],"_revision_number":3},{"id":"ba0e90cd443d92c6bdf1f30a39079d8935fe67ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-25 13:38:22.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"c6ad6d30e3af478350d6a5883479f08d240da7b6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-03-25 13:39:45.000000000","message":"Patch Set 2:\n\nBuild Aborted \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/798/ : ABORTED","accounts_in_message":[],"_revision_number":2},{"id":"a6f8d35a293cfe08d34a544102afea5ac7691d11","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2022-03-25 14:14:14.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":3},{"id":"44f1779c7922b0378bea0567b0e1efc8defbb082","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2022-03-25 14:14:52.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a396cae9a02c2106e71df6c0848417264be7ec02","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-03-25 14:15:09.000000000","message":"Patch Set 3:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/803/","accounts_in_message":[],"_revision_number":3},{"id":"d04621b7ee962af40b7d1a90b903cee741d84574","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2022-03-25 14:39:10.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"abd6c9da326ed0500ab627248af61eb240882cb1","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2022-03-25 14:40:17.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0ce16d2bda51872b1c51ca1b17ed9e7000eae3f8","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-03-25 14:42:06.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"573013112a4fc2891fb0e1b5d88717db86542822","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-25 15:32:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d612bebc937f422f956ca3c2b41599cb : SUCCESS in 4m 15s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/67d34ae249f34468a70fcd34d628441e : SUCCESS in 6m 14s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/bb9560543e504e7aa4f1d0b7c388a74d : SUCCESS in 59m 45s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/3994499f3cf64c429b6f58fafd3ef92c : SUCCESS in 15m 31s","accounts_in_message":[],"_revision_number":3},{"id":"16410da5b177a883fac284a851e82618611360bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-25 15:33:10.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a12abe8b5a5a7ef901f98826d6d7ecb6fdf81c84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-25 17:59:41.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/de059eeaaff44bb4b07d70e66af93ef1 : SUCCESS in 3m 53s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/2625326eeb7f432fb4e8958371c500ef : SUCCESS in 48m 06s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c0787b027ec944c39c4dc59b36002b63 : SUCCESS in 6m 10s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/33f2d692ecee45ad8b379a2585e504e8 : SUCCESS in 14m 58s","accounts_in_message":[],"_revision_number":3},{"id":"17b9925d14350661fa638ccc80c0536322d32c68","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-25 17:59:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"f9ac4e7d913c2856893c7426f93dc7fa59d5b4d9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-03-25 18:31:46.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/800/ : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"2c22f74824bb8cdf0a23940f069b3a5c4910b3b8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-03-25 19:17:59.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/803/ : SUCCESS","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"49d52ddaa392abbfcfd8243f7a1da1c7f6824417","revisions":{"35c711812f9f05894eea40e57e94b766bc3bd6b7":{"kind":"REWORK","_number":1,"created":"2022-03-25 10:54:43.000000000","uploader":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"ref":"refs/changes/03/835203/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/03/835203/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/03/835203/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/03/835203/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/03/835203/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/03/835203/1"}}},"commit":{"parents":[{"commit":"0dc4f9ac62ee38d45333c38cae0179b160fa0c12","subject":"Use from_yaml filter if not mapping","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/0dc4f9ac62ee38d45333c38cae0179b160fa0c12"}]}],"author":{"name":"Sandeep Yadav","email":"sandyada@redhat.com","date":"2022-03-25 10:53:31.000000000","tz":330},"committer":{"name":"Sandeep Yadav","email":"sandyada@redhat.com","date":"2022-03-25 10:54:39.000000000","tz":330},"subject":"Disable auto discovery as a workaround","message":"Disable auto discovery as a workaround\n\nTripleo-ci jobs are broken after latest release of setuptools 61.0,\ndetails in Releated-Bug.\n\nTrying the workaround as mentioned in [1]\n\n[1] https://github.com/pypa/setuptools/issues/3197\n\nReleated-Bug: #1966382\nChange-Id: I23788ec24c581c037e2535201aeaf61e1bbbbbdc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/35c711812f9f05894eea40e57e94b766bc3bd6b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/35c711812f9f05894eea40e57e94b766bc3bd6b7"}]},"branch":"refs/heads/master"},"e34d0ddf9f40a77daf8cb034c27543f8833a4686":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-03-25 13:21:02.000000000","uploader":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"ref":"refs/changes/03/835203/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/03/835203/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/03/835203/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/03/835203/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/03/835203/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/03/835203/2"}}},"commit":{"parents":[{"commit":"0dc4f9ac62ee38d45333c38cae0179b160fa0c12","subject":"Use from_yaml filter if not mapping","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/0dc4f9ac62ee38d45333c38cae0179b160fa0c12"}]}],"author":{"name":"Sandeep Yadav","email":"sandyada@redhat.com","date":"2022-03-25 10:53:31.000000000","tz":330},"committer":{"name":"Sandeep Yadav","email":"sandyada@redhat.com","date":"2022-03-25 13:21:02.000000000","tz":0},"subject":"Disable auto discovery as a workaround","message":"Disable auto discovery as a workaround\n\nTripleo-ci jobs are broken after latest release of setuptools 61.0,\ndetails in Releated-Bug.\n\nTrying the workaround as mentioned in [1]\n\n[1] https://github.com/pypa/setuptools/issues/3197\n\nDepends-On: https://review.opendev.org/c/openstack/tripleo-ci/+/835193\n\nRelated-Bug: #1966382\nChange-Id: I23788ec24c581c037e2535201aeaf61e1bbbbbdc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/e34d0ddf9f40a77daf8cb034c27543f8833a4686"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/e34d0ddf9f40a77daf8cb034c27543f8833a4686"}]},"branch":"refs/heads/master","description":"Edit commit message"},"49d52ddaa392abbfcfd8243f7a1da1c7f6824417":{"kind":"REWORK","_number":3,"created":"2022-03-25 13:33:24.000000000","uploader":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"ref":"refs/changes/03/835203/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/03/835203/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/03/835203/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/03/835203/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/03/835203/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/03/835203/3"}}},"commit":{"parents":[{"commit":"0dc4f9ac62ee38d45333c38cae0179b160fa0c12","subject":"Use from_yaml filter if not mapping","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/0dc4f9ac62ee38d45333c38cae0179b160fa0c12"}]}],"author":{"name":"Sandeep Yadav","email":"sandyada@redhat.com","date":"2022-03-25 10:53:31.000000000","tz":330},"committer":{"name":"Sandeep Yadav","email":"sandyada@redhat.com","date":"2022-03-25 13:33:09.000000000","tz":330},"subject":"Disable auto discovery as a workaround","message":"Disable auto discovery as a workaround\n\nTripleo-ci jobs are broken after latest release of setuptools 61.0,\ndetails in Releated-Bug.\n\nTrying the workaround as mentioned in [1]\n\n[1] https://github.com/pypa/setuptools/issues/3197\n\nDepends-On: https://review.opendev.org/c/openstack/tripleo-ci/+/835193\n\nRelated-Bug: #1966382\nChange-Id: I23788ec24c581c037e2535201aeaf61e1bbbbbdc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/49d52ddaa392abbfcfd8243f7a1da1c7f6824417"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/49d52ddaa392abbfcfd8243f7a1da1c7f6824417"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
