)]}'
{"id":"openstack%2Ftripleo-quickstart-extras~843835","triplet_id":"openstack%2Ftripleo-quickstart-extras~master~I476c7299dec209c89db87e2182a35ec6a4924f92","project":"openstack/tripleo-quickstart-extras","branch":"master","topic":"standalone-roles-testing","attention_set":{},"removed_from_attention_set":{"8449":{"account":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"last_update":"2022-10-18 06:22:25.000000000","reason":"\u003cGERRIT_ACCOUNT_8449\u003e replied on the change","reason_account":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"}},"7144":{"account":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"last_update":"2022-10-20 15:10:18.000000000","reason":"Change was submitted"},"9976":{"account":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"last_update":"2022-10-20 15:10:18.000000000","reason":"Change was submitted"},"33080":{"account":{"_account_id":33080,"name":"Mikolaj Ciecierski","display_name":"Mikolaj Ciecierski","email":"mciecier@redhat.com","username":"mciecierski"},"last_update":"2022-10-20 08:03:56.000000000","reason":"\u003cGERRIT_ACCOUNT_33080\u003e replied on the change","reason_account":{"_account_id":33080,"name":"Mikolaj Ciecierski","display_name":"Mikolaj Ciecierski","email":"mciecier@redhat.com","username":"mciecierski"}},"12393":{"account":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"last_update":"2022-10-20 15:10:18.000000000","reason":"Change was submitted"},"30073":{"account":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"last_update":"2022-10-20 15:10:18.000000000","reason":"Change was submitted"},"6926":{"account":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"last_update":"2022-10-18 11:45:12.000000000","reason":"\u003cGERRIT_ACCOUNT_6926\u003e replied on the change","reason_account":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}},"29775":{"account":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"last_update":"2022-10-20 11:15:52.000000000","reason":"\u003cGERRIT_ACCOUNT_29775\u003e replied on the change","reason_account":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}}},"hashtags":[],"change_id":"I476c7299dec209c89db87e2182a35ec6a4924f92","subject":"Add EDPM role and playbook for standalone-compute deployment","status":"MERGED","created":"2022-05-30 10:33:53.000000000","updated":"2022-10-20 15:10:18.000000000","submitted":"2022-10-20 15:10:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":57,"unresolved_comment_count":8,"has_review_started":true,"submission_id":"843835-standalone-roles-testing","meta_rev_id":"854c987974d876b8424967ef3a5a4ff8d926aad9","_number":843835,"virtual_id_number":843835,"owner":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":33080,"name":"Mikolaj Ciecierski","display_name":"Mikolaj Ciecierski","email":"mciecier@redhat.com","username":"mciecierski"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-20 15:10:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-10-20 08:20:16.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":0,"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"recommended":{"_account_id":33080,"name":"Mikolaj Ciecierski","display_name":"Mikolaj Ciecierski","email":"mciecier@redhat.com","username":"mciecierski"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":1,"date":"2022-10-20 08:03:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33080,"name":"Mikolaj Ciecierski","display_name":"Mikolaj Ciecierski","email":"mciecier@redhat.com","username":"mciecierski"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":2,"date":"2022-10-20 11:47:25.000000000","_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":2,"date":"2022-10-20 11:15:52.000000000","_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":33080,"name":"Mikolaj Ciecierski","display_name":"Mikolaj Ciecierski","email":"mciecier@redhat.com","username":"mciecierski"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2022-10-20 11:47:25.000000000","_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":0,"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"}],"reviewers":{"REVIEWER":[{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},{"_account_id":33080,"name":"Mikolaj Ciecierski","display_name":"Mikolaj Ciecierski","email":"mciecier@redhat.com","username":"mciecierski"}],"CC":[{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-30 10:34:15.000000000","updated_by":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-30 10:51:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-30 14:27:17.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"state":"REVIEWER"},{"updated":"2022-05-30 14:27:17.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"},{"updated":"2022-05-30 16:57:54.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-02 03:30:19.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REMOVED"},{"updated":"2022-06-02 03:30:21.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"state":"REMOVED"},{"updated":"2022-07-13 07:18:18.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2022-07-13 11:25:53.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"REVIEWER"},{"updated":"2022-09-05 14:39:47.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"},{"updated":"2022-10-04 07:13:22.000000000","updated_by":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"reviewer":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"state":"CC"},{"updated":"2022-10-10 08:04:28.000000000","updated_by":{"_account_id":33080,"name":"Mikolaj Ciecierski","display_name":"Mikolaj Ciecierski","email":"mciecier@redhat.com","username":"mciecierski"},"reviewer":{"_account_id":33080,"name":"Mikolaj Ciecierski","display_name":"Mikolaj Ciecierski","email":"mciecier@redhat.com","username":"mciecierski"},"state":"CC"},{"updated":"2022-10-11 07:11:41.000000000","updated_by":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"reviewer":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"state":"CC"},{"updated":"2022-10-18 11:48:10.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"state":"REVIEWER"},{"updated":"2022-10-20 08:03:56.000000000","updated_by":{"_account_id":33080,"name":"Mikolaj Ciecierski","display_name":"Mikolaj Ciecierski","email":"mciecier@redhat.com","username":"mciecierski"},"reviewer":{"_account_id":33080,"name":"Mikolaj Ciecierski","display_name":"Mikolaj Ciecierski","email":"mciecier@redhat.com","username":"mciecierski"},"state":"REVIEWER"},{"updated":"2022-10-20 11:15:52.000000000","updated_by":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"reviewer":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"state":"REVIEWER"}],"messages":[{"id":"ca4fe250ca1b93487347fc934bec29776bdfff0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-05-30 10:33:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"45ac6548927ba30575b7f0152210896d8c11a98b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-05-30 10:34:15.000000000","message":"Patch Set 1:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1000/","accounts_in_message":[],"_revision_number":1},{"id":"c0539fd1428ce816464169944d7042ef6b30df92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 10:51:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/362aeefd6a0f40bab2e8ac2e45f44911 : FAILURE in 10m 03s","accounts_in_message":[],"_revision_number":1},{"id":"3c533e3283ca464b4564f6b5abfe8955f9a9c255","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-05-30 11:44:08.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1000/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"027d2eb5806ee72036f4844071b425a04a4780b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-05-30 12:16:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"10d360469daf13dd3ac04c806afe24805d3c605b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-05-30 12:16:37.000000000","message":"Patch Set 2:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1002/","accounts_in_message":[],"_revision_number":2},{"id":"d637a2404cb34879c27a4f71435e4e31a8334014","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 12:24:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8303140e2777499693e60ffa38eb40de : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":2},{"id":"7b34099d446874cec348c8a136ae677e7a591145","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-05-30 13:26:39.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1002/ : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"d4869ef31926261680982aa0bb2fb65e43a3e35a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-05-30 14:26:13.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b953446c1fa04c4410c110bd0733e3dfa03e0df1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-05-30 14:26:30.000000000","message":"Patch Set 3:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1004/","accounts_in_message":[],"_revision_number":3},{"id":"cd08403ff62c95e6cfec1d8e73b2b3267d1ec7c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 14:34:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d1a16070d72e489a9b93c0c52b892a73 : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":3},{"id":"9e0ab57cb7d82fab9c72791b539a32b129af06b7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-05-30 15:07:53.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1004/ : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"b6729bdb5fc27309f58510709d1403b740d9de9f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-05-30 16:57:54.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/8cac9bed1520420cb549fceb3c7b6f48 : FAILURE in 2h 15m 49s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/fc9b451841844a7cbdd6ac2e8ec93e23 : FAILURE in 2h 24m 46s","accounts_in_message":[],"_revision_number":3},{"id":"0f8a37980e2fb562ba4e2da6f2f891070d8f4f39","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-05-31 04:34:17.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a5a100bf6d1553a9ef0307efc15ee3a97ca904cb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-05-31 04:34:43.000000000","message":"Patch Set 4:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1008/","accounts_in_message":[],"_revision_number":4},{"id":"c32d4de21f662b6f8bfff20b874e4a9a38e1d523","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 04:44:58.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/abcd0ab94b3a453e98ee6de669905807 : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":4},{"id":"edb8556ae276b85f632b464ca479c957187c3916","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-05-31 05:45:06.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1008/ : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"64206ae4bd1b2c12a4bdc7d151b65a1d31fb08fb","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-05-31 06:31:48.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/d5a748d7e7fd4b50a3404b6f6028a10b : FAILURE in 1h 56m 28s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/a2c9c836a5814effa483740425cdf990 : SUCCESS in 1h 47m 47s","accounts_in_message":[],"_revision_number":4},{"id":"9b712158ec44892e8eb77685887c6752f6e10c85","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-05-31 15:02:02.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"cb5dbdc3ec53d7a198a66e6713ecd738ca863ffc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-05-31 15:02:17.000000000","message":"Patch Set 5:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1011/","accounts_in_message":[],"_revision_number":5},{"id":"a5bb7e5b91b2789f5ed8de91e5495844b46c3522","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 15:11:05.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/860fc3f3fb47437eae503aebe4702034 : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":5},{"id":"171276f64eb222168c127fca1b3dda3d18419489","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-05-31 16:12:24.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1011/ : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"8956d7350acb3da7ac46990c2ef51fbe6bd784aa","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-05-31 17:15:49.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/b163a207a1ee430ba0981eaa4eba18b7 : SUCCESS in 2h 08m 04s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/f0d216504c534c76b409ad349b251358 : SUCCESS in 1h 59m 04s","accounts_in_message":[],"_revision_number":5},{"id":"9d7094a70fb04679f8e34b21d46d13810e205ecc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-06-01 11:56:40.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"39c335cc20d0e17817e3f307b31fa25928ad1f7e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-01 11:57:02.000000000","message":"Patch Set 6:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1013/","accounts_in_message":[],"_revision_number":6},{"id":"f23dfc1ef7c82365319a92b2f3919a87cf2bc5e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-01 12:08:06.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e522cddceb634a4f9d88913aa925e735 : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":6},{"id":"1c89f8e035fdba067f85677e1268c92b289d6b75","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-01 12:42:02.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1013/ : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"5b683faf25c2c12463d19a7e6f2831fcc7672998","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-06-01 13:49:05.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/f9cac700d90841a2acf0c1bfd333930f : FAILURE in 1h 50m 10s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/137aeade0de141b4ab1c49b7b6a47803 : FAILURE in 1h 40m 11s","accounts_in_message":[],"_revision_number":6},{"id":"fc76bca1fb7d0a18f5ad9ebe11bae1807adda27e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-06-02 03:28:36.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"878fc5b2bc05ee8310d2d09091b95c3c176788fb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-02 03:28:58.000000000","message":"Patch Set 7:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1017/","accounts_in_message":[],"_revision_number":7},{"id":"2e301ea2a5bb4d371079afae5144fb9f77c869b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-06-02 03:29:44.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"de4e51d0889c14741e8d9ba9d67a20c2c90ce4db","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-02 03:30:08.000000000","message":"Patch Set 8:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1018/","accounts_in_message":[],"_revision_number":8},{"id":"0cfa0123bb3956a2ec7ca7e6585544ca1f4e8763","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-06-02 03:30:19.000000000","message":"Removed reviewer Bogdan Dobrelya.","accounts_in_message":[],"_revision_number":8},{"id":"6bbe6c91ef57c623103a30298ae23af8bb067734","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-06-02 03:30:21.000000000","message":"Removed reviewer James Slagle.","accounts_in_message":[],"_revision_number":8},{"id":"0e5cd5529437ceeec4504f1226e07cff273827e1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-02 03:32:24.000000000","message":"Patch Set 7:\n\nBuild Aborted \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1017/ : ABORTED","accounts_in_message":[],"_revision_number":7},{"id":"0fc162b3cadd04396be0952cd2101858de980e8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 03:35:41.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/74fc5beaeefc4bf6a537c5e12bdf2b43 : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":8},{"id":"24cc7fc8edf2ecf2e4772d063ee124aa5aecb700","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-06-02 03:39:38.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"86d468357603dfd9d70ea99b9a2f8fdbecee96b2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-02 03:39:59.000000000","message":"Patch Set 9:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1019/","accounts_in_message":[],"_revision_number":9},{"id":"ade5c5c27de74cdaad2b55da8a8912c7452f15fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 03:45:45.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/966806c069f24747a7caa4b1a95125de : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":9},{"id":"d55c0356fd56ed494389351e8ae89ce17c6b601d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-02 03:45:59.000000000","message":"Patch Set 8:\n\nBuild Aborted \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1018/ : ABORTED","accounts_in_message":[],"_revision_number":8},{"id":"9c01dcd4fb3143924f6d8e4a0fe964f98f5a09d8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-02 04:53:06.000000000","message":"Patch Set 9:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1019/ : FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"f314da66139db1b29e1870203eed0d17275de499","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-06-02 05:00:38.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/92018e94b0744e4ea0166764dd1c9870 : FAILURE in 1h 19m 38s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/f497e02ca67d413daf82f4d98ec9ac28 : FAILURE in 1h 16m 59s","accounts_in_message":[],"_revision_number":9},{"id":"146104a675689cf7fa20dcc908c6e9a788ee48e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-06-20 11:05:46.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"bfcd1dcfd0cdfc7102c0b958b5c5f622854d3ab5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-20 11:06:08.000000000","message":"Patch Set 10:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1080/","accounts_in_message":[],"_revision_number":10},{"id":"e0b79b02abbdcf703748e2e042b3262096c6d458","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 11:15:53.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9e85bf61a5a34b39bae7bbadc1434ed6 : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":10},{"id":"842a43fe7bcd2329dad659111b40d869afa15425","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-20 12:16:42.000000000","message":"Patch Set 10:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1080/ : FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"60ec1551fff9be06386bbc22bae8cc6f1cdda7b5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-06-20 12:26:40.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/506beda3c1584d49bf3e00f13bbe9dce : FAILURE in 1h 19m 40s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/7be37554cf484deb916b2f9270f55019 : FAILURE in 1h 17m 32s","accounts_in_message":[],"_revision_number":10},{"id":"19dc46c30cbb64e57eaf8e65766238bd609c65dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-07-07 06:40:13.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"f4008e01b28800d7183b656bbe905e79f09ae303","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-07-07 06:40:39.000000000","message":"Patch Set 11:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1135/","accounts_in_message":[],"_revision_number":11},{"id":"d25563a2c1c010e1dbc54c7070d87c11da5b7e75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 06:46:45.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9ea026a33ee4ceaad1e27ff4096b45e\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/76f5322e11694fb396650bb2cd335c20 : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":11},{"id":"cb9efd1bcbb0949773f425ffe6f0ecf325086834","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-07-07 06:51:55.000000000","message":"Patch Set 11:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1135/ : FAILURE","accounts_in_message":[],"_revision_number":11},{"id":"be031c7e6fa5c1029e7cf122c42382718854dcf3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-07-07 07:34:49.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"160e50f4fc278bce04904d7740b1d6219a0cfd56","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-07-07 07:35:04.000000000","message":"Patch Set 12:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1136/","accounts_in_message":[],"_revision_number":12},{"id":"682758ba0cd7ec2e9a09d896f8fc22f17ca224db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 07:45:39.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1d7a81fc0a44f0bb00ca2d95d4d3d77\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a8b3a7f480ca48099574cb8d70689fc6 : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":12},{"id":"4c5ff249da4183578267434a1b3ac3908b5909de","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-07-07 07:46:31.000000000","message":"Patch Set 12:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1136/ : FAILURE","accounts_in_message":[],"_revision_number":12},{"id":"c23496f31f82614987bd1e5920ad18c1c722a30e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-07-07 08:35:30.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"54aaf8e86f64f93bfd7d247b282a36010bc5a62a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-07-07 08:35:45.000000000","message":"Patch Set 13:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1137/","accounts_in_message":[],"_revision_number":13},{"id":"d9b2e52ab1f963a961aa2ddf5c6f01f2f9e9cbe9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 08:40:11.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3883f2b509a44906bc29c265699778a4\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2807b4f00a89443abf94e6630c687488 : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":13},{"id":"c7cf1053911a16d06e619b7893bc9084da37ae29","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-07-07 08:47:11.000000000","message":"Patch Set 13:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1137/ : FAILURE","accounts_in_message":[],"_revision_number":13},{"id":"90ac8162b8757c3e8162a16ba654d8c9ab1031f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-07-07 09:55:58.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"5a1be0d4f39cda96a56ac325d9fd567a800e7e3b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-07-07 09:56:21.000000000","message":"Patch Set 14:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1138/","accounts_in_message":[],"_revision_number":14},{"id":"76b8d9f0ffae637986600ff972411a99da44af30","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-07-07 10:07:15.000000000","message":"Patch Set 14:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1138/ : FAILURE","accounts_in_message":[],"_revision_number":14},{"id":"c75e2cab9dffc429fee0af2524c8ac9839fd723b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 10:07:51.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7a5d922599248b685a2272de7af1fb0\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c3dcbf569d3449ed8f6d908cfdf3f186 : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":14},{"id":"a0d95d0128d22819effad34eefd7953762f39f03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-07-07 10:52:44.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"025a95af37e5af44652434ce7982f8868c95386c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-07-07 10:53:10.000000000","message":"Patch Set 15:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1139/","accounts_in_message":[],"_revision_number":15},{"id":"97f068b421e64260f758cd8129849072d314f767","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 10:58:19.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a19e345430847199c991cad3f84313d\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/428bc58779754cc19a638acb86772e61 : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":15},{"id":"18a33f5efa382c86179b2ec427a1e539b168d3db","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-07-07 11:04:37.000000000","message":"Patch Set 15:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1139/ : FAILURE","accounts_in_message":[],"_revision_number":15},{"id":"bcfb3f432a9f558045859afa5de0fc3b7170f230","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-07-07 12:43:19.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/1763dff389e8481f8f4f00618d3c5253 : FAILURE in 1h 33m 24s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/9e6c2f3003914dbd9ba0a1ce6a832284 : FAILURE in 1h 26m 08s","accounts_in_message":[],"_revision_number":15},{"id":"5a9e1433d0c0be9bc68c8e505694ef0bb91e49aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-07-07 12:51:21.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"3febbced1bb466ff9b6d5360b9573ad67afd62ba","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-07-07 12:51:42.000000000","message":"Patch Set 16:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1140/","accounts_in_message":[],"_revision_number":16},{"id":"1e95f8e0ba3da1afc581c4bc55ba89c5bea8f434","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 12:56:18.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d82a116ca9f41aba2539d87cf59cc1b\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c7dd2634d96e41e79636cce4d9bcde6f : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":16},{"id":"1890149a61ec59a5141ed8cc0fdae4d4e9e7a099","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-07-07 13:01:22.000000000","message":"Patch Set 16:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1140/ : FAILURE","accounts_in_message":[],"_revision_number":16},{"id":"e2147063d8b26e8c2cdb53b8b2bef76d1670da32","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-07-07 16:16:09.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/3ffa104be6d146569de64332579c12d2 : SUCCESS in 2h 30m 01s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/f7e5a2ebcdf146448d829a91e1ac6ac8 : FAILURE in 1h 24m 58s","accounts_in_message":[],"_revision_number":16},{"id":"9d96a04f543a33283ec9c2b4c2cce628657b49e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-07-11 05:25:17.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"3952bb751fd7e0e807af8effec9ce26ae18fc73c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-07-11 05:25:40.000000000","message":"Patch Set 17:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1142/","accounts_in_message":[],"_revision_number":17},{"id":"bf4fa37af41843283496972f04e632e5724829a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 05:31:35.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4d69c21c3084d08b019b98085236cdf\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b142bb4af8fa46b7a06b192f79160301 : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":17},{"id":"d5e3d007b6c83f77f9acf4d70b2216d48c1dcbdc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-07-11 05:36:46.000000000","message":"Patch Set 17:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1142/ : FAILURE","accounts_in_message":[],"_revision_number":17},{"id":"442c262cc9ac7a88f603800db8308bdb17c7f3f0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-07-11 07:36:01.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/99b5920bce1e4dfbae1f5514dcacbacb : FAILURE in 1h 48m 52s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/349f22de24fe49f2b2f061977e443185 : FAILURE in 1h 26m 18s","accounts_in_message":[],"_revision_number":17},{"id":"ecc856c282f270072b077496be3baa0f883728c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-07-11 13:49:51.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"c3977df5f7c487ac26dcd71c63b08447a3539a2b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-07-11 13:50:11.000000000","message":"Patch Set 18:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1143/","accounts_in_message":[],"_revision_number":18},{"id":"91230b30a18e919273137d7fa24e17c38dc52eb9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-07-11 14:00:31.000000000","message":"Patch Set 18:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1143/ : FAILURE","accounts_in_message":[],"_revision_number":18},{"id":"8c8b899afe1f78f5b22e163a1a49bd5a050e80b8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-07-11 15:53:10.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/ace680f29bcc4674a6ddab601a895d37 : FAILURE in 1h 52m 30s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/8d37a32929ee4903bbd063d5e5b48baa : FAILURE in 1h 22m 38s","accounts_in_message":[],"_revision_number":18},{"id":"1ed92f3551dfdbdfb778cfef9f22f736ee67bde6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 16:42:04.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/553ceaae49504cc3a9ad854a5a8ac768\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/323076001e7344eeb96bb04fbb50c175 : SUCCESS in 2h 44m 38s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/4f8d76516c264ea68cd6e07de09dd1b2 : SUCCESS in 36m 36s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/149dedf9fe754a5ba2daf126cbbd5e0c : SUCCESS in 1h 51m 22s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/b06598b53a374d42b7b715e5dcf448a6 : SUCCESS in 1h 50m 06s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4755e3f1b5e8464aa21950d5ac81e133 : SUCCESS in 4m 59s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/d59a4121a03342f18d6f9e963de4fa30 : SUCCESS in 2h 20m 20s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/7217998fa2cb4aeba7ecdb2d072fc030 : SUCCESS in 1h 29m 00s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/1e753bfe03e445e9baa238d42d353dba : SUCCESS in 1h 14m 52s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/c5c4c9851444482180b2130926907c5b : SUCCESS in 1h 26m 24s (non-voting)\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/e8ea2db27e8e484fa857c4e4f621ded3 : SUCCESS in 1h 24m 27s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/855cf615cfcf429ea95c7a66fa4ebcce : SUCCESS in 1h 03m 10s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/048e520483624890b9686bc019683daf : SUCCESS in 20m 40s","accounts_in_message":[],"_revision_number":18},{"id":"6571f37d117d6215a74cca63b373070127f05c4d","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-07-12 03:55:33.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"abd0c5cb421af32fe214a87a6e037e4396bb6b69","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-07-13 07:18:18.000000000","message":"Patch Set 18: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"a9811613c90d66c881756a6e07d7e0885fd15b8e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-07-13 10:12:23.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"14f576630c619a0cfcff5c65484c8a3f6e617511","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-07-13 10:12:52.000000000","message":"Patch Set 19:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1154/","accounts_in_message":[],"_revision_number":19},{"id":"bcfbc3c39dbc13dd81aec56d655756e008aa1534","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-07-13 10:24:19.000000000","message":"Patch Set 19:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1154/ : FAILURE","accounts_in_message":[],"_revision_number":19},{"id":"feb197ac4228d48723a225cdac939a9136c1156f","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-07-13 10:25:34.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"364c48b95cc715c5b82716919ed8dc494d123e03","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2022-07-13 11:25:53.000000000","message":"Patch Set 19: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"396fbb212a2e1d84b1e533db22fcd5b93bf04e0f","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-07-13 11:25:53.000000000","message":"Patch Set 19: Code-Review+2","accounts_in_message":[],"_revision_number":19},{"id":"6f319f2acf0d4316fb3d7931f68987fc5226e1dd","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-07-13 13:10:35.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/4e98cf1677b645a58228ab420b7fefb7 : SUCCESS in 2h 21m 37s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/280039a3c43c42429d67077183b8b07e : SUCCESS in 2h 09m 35s","accounts_in_message":[],"_revision_number":19},{"id":"378e0d400b26e5739f29bcc7a227f20675aa24cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 13:29:33.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c107716c5ff456bb1d05d569df033f3\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/0fc28041edf94f6889b561ceff49995c : SUCCESS in 2h 54m 20s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/8d730b52ad514097b4eb6a2bd3624e56 : SUCCESS in 53m 06s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/421e668389e749b7b431cd2dc1eba8fb : SUCCESS in 1h 45m 43s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/1052cebc78c743599cb6d9ed81d6f465 : SUCCESS in 1h 45m 15s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f7e9462a47b848a9ac0b12917e15240d : SUCCESS in 4m 08s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/f804dcdcd06a44d6be19804403d5cd8e : SUCCESS in 3h 03m 57s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/39eec72a75224b71a3bcfd2c528899ad : SUCCESS in 1h 52m 45s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/f2158f2585974cbc9348f43c03c55b21 : SUCCESS in 1h 34m 18s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/2174b76ca4ee4a82ae17e37ff7cba1af : SUCCESS in 1h 41m 36s (non-voting)\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/d11f90c08ca244cba5734f4c846729bc : SUCCESS in 1h 46m 16s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/64a92209d64a4fda8c81e19c57297a65 : SUCCESS in 1h 20m 29s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/575fdf369e63418d9cee3aa41a2ef40a : SUCCESS in 21m 39s","accounts_in_message":[],"_revision_number":19},{"id":"48cfc7e738c7510b2539ba365ccd623bbd549d74","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-08-24 08:49:14.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"57ec4e6a3777991102d83b6048b2476314f9db3e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-08-24 10:17:50.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/fcc6d40849ba4bcf9e63fbd0b8c89102 : FAILURE in 1h 27m 26s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/c5fd320e11884fcb870d71567b13e985 : FAILURE in 1h 26m 01s","accounts_in_message":[],"_revision_number":20},{"id":"10428b850373b0c4a3f1c7d4127a483d41247a14","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-08-24 12:06:14.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"e45c1a2979e01d048f5b51845c14c9d1054877fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-08-24 13:37:50.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"823e919e79d18836c50822588892e07a10246378","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-08-24 15:13:42.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/abc5cd84e6de442f83c334dc9cbdc4f0 : FAILURE in 1h 34m 53s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/e87a5205ecda4a97bea309e666840bdf : FAILURE in 1h 34m 14s","accounts_in_message":[],"_revision_number":22},{"id":"6b9d58d90de65a94998de355958f3404c48ab2fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-08-24 16:26:00.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"87b6f8db63709f8e0fa0d505a15b2604a3d25071","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-08-24 18:58:53.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/7779df8303a94e65bb8bef183ee2d734 : NODE_FAILURE in 0s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/0b34b4c7f11a4e8d9a01d36f46ab6c4a : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":23},{"id":"52f7dff6a8a16c2d68b4ef3753c900720385011b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 19:57:47.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60776d11024b43c19ef9600bb7a5ff6b\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/7452bb04f67d429c9a270e1e2b4b2aef : SUCCESS in 2h 35m 42s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/dd128cf6b51a44cf9ef0525db16a87f6 : SUCCESS in 42m 58s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/bf96bc8170644513a04156719c92c886 : SUCCESS in 1h 31m 40s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/8c6f19210393482d8446046c0ef037cc : SUCCESS in 1h 21m 42s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e2ccabbba9df420099d4186843586738 : SUCCESS in 5m 18s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/4321abd66c5244e0b3dc6d614418a2cf : SUCCESS in 3h 02m 56s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/6e6628fc2d4a4f25ab4842a0eaf76027 : SUCCESS in 1h 50m 02s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/f4d1e837acab4b4898a62d286a7b664c : SUCCESS in 1h 19m 53s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/3f2337a12f9142019c450b41bbfcda9f : SUCCESS in 1h 30m 33s (non-voting)\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/8e390394184a42c8be581a0b971d4600 : SUCCESS in 1h 22m 34s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/a191befb9acd41abb9a6e7d5a0998c65 : SUCCESS in 1h 22m 51s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/4acf5463ed374d0dbaf3b7166fb2326d : SUCCESS in 21m 54s","accounts_in_message":[],"_revision_number":23},{"id":"e66d66ee325f63a8b74878aa37bb159b117fa8de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-08-25 04:53:31.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"855c1728d531580bc1347fb6caafa7ea346539ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 05:23:59.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8b99cbc37df04e318e513ed904b02ea8\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/92240e39f0104c26869cb5e88b80a016 : RETRY_LIMIT in 11s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9fa82b003b194830b00d108d49b8e44a : SUCCESS in 4m 39s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/351833d0842445f3a57f35e1579ba3b8 : RETRY_LIMIT in 10s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/b06cf434cdd2422d9aa616df020b397a : SUCCESS in 23m 48s\nSkipped 8 jobs","accounts_in_message":[],"_revision_number":24},{"id":"d56019c4f4edc35e6ca067d53eb4a5390c2893fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-08-25 06:18:58.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"cdb4548bb6c3d6d756032078ed926173617773f8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-08-25 07:48:05.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/2e5d5228948a4433aaf39d2d137c6231 : FAILURE in 1h 27m 47s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/a2a7a17e0f474ef0b979d159b580fa63 : FAILURE in 1h 25m 52s","accounts_in_message":[],"_revision_number":25},{"id":"a3b5c8ce162ffcee69ecf932b1a5613f23edd29d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-08-25 08:07:18.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"df0c56e528f5b52f83397d12bdc20cbf793392a0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-08-25 09:45:49.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/73a085fe92d44261a4e44bd29452cebd : FAILURE in 1h 35m 05s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/9d7d33d1531645d2971f2344c3dcf71c : FAILURE in 1h 36m 02s","accounts_in_message":[],"_revision_number":26},{"id":"a7437bdca631da83a0a01082292f8afb57913448","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 11:51:15.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb7801cb9f0c4be59c798a5398eff800\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/af4f1c930dd44a24a03fafe6fd47e5da : SUCCESS in 3h 06m 44s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/48a5e102558a447e9ede27c5f7b279e3 : SUCCESS in 54m 29s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/0c50da1a98354ee196f279f725b4e2f8 : SUCCESS in 1h 49m 00s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/4785e95701884299877ec90df1e3d4eb : SUCCESS in 1h 38m 53s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ead4dd52c3c54752b436011cfeec0027 : SUCCESS in 5m 57s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/91d52ca4aa3e4acfa0d6ba1aaca4d66b : SUCCESS in 2h 34m 38s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/6b4c0bd03a16470892a1ed86b9b81997 : SUCCESS in 1h 36m 00s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/4310efe183c44d76aca47d64a85df37b : SUCCESS in 1h 22m 35s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/1baff3c8da4b4af6be5d098081438c30 : FAILURE in 1h 53m 25s (non-voting)\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/99f4067024464dd0b03860e12840ab56 : SUCCESS in 1h 34m 20s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/9695e6a2688b4fbb8710b2e7c204d5ed : SUCCESS in 1h 09m 29s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/6875c323cee84d778134858703853f9a : SUCCESS in 22m 17s","accounts_in_message":[],"_revision_number":26},{"id":"1ad8cab741f4adb4d1d45a348989006fdf8894a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-08-29 13:41:39.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"9f8eaa6bcb62bdec54947b2527f23f0c2a67c378","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-08-29 15:05:28.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/1bc229344acd46f9bc7e33965a496426 : POST_FAILURE in 1h 22m 39s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/05628a4e3f6643de98b17d4b4b17b3cb : POST_FAILURE in 1h 21m 21s","accounts_in_message":[],"_revision_number":27},{"id":"85a05ff1841d66057d1847ae007c0e31a6facd54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 19:52:38.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e18d7e495a0046c99a546d07d0b2fd06\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/7136e49aa177471599717a2c9b28e7f0 : SUCCESS in 5h 59m 38s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/e09890073e844ee3a63b8057a9b486eb : SUCCESS in 52m 34s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/fbc99f75e90a4d629456269a7abae390 : SUCCESS in 1h 44m 59s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/edb93231d1eb40b5a293a79c2a7e5044 : SUCCESS in 1h 38m 48s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5a8c2bd42e374a509c3d7b25c08f4d3e : SUCCESS in 4m 32s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/30bf8cf26cc8488ab277f998e3e2c738 : SUCCESS in 2h 58m 22s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/2df22767dbaa43b8890e3ed1eabe75db : SUCCESS in 1h 48m 11s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/b585675485af4c4e98fcc10a1ff6de74 : SUCCESS in 1h 27m 39s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/e0eb64b3d2b14ddca54e54394e149a0c : SUCCESS in 2h 01m 52s (non-voting)\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/bc3b0e16800843738aad733096d1945f : SUCCESS in 1h 48m 34s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/0469ca603508447fb1b1bc895f64e7da : SUCCESS in 1h 14m 57s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/3e38b1a572d645499cf9441a9fdb8385 : SUCCESS in 21m 45s","accounts_in_message":[],"_revision_number":27},{"id":"3169b61bfc19ded7841696cd966329bef22cf2aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-08-30 04:46:38.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"69810003fc1b37dbe73e8324af6ba1711008fff4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-08-30 06:34:03.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/3d1396e0c87f44fbba03b93ad69badbf : POST_FAILURE in 1h 38m 52s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/a8b164c589454d20b094c1c8dbe15330 : FAILURE in 1h 45m 57s","accounts_in_message":[],"_revision_number":28},{"id":"f88d5d295a186383c8258a867a7e1870cedd2db3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 07:30:25.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34dd3e266b2d41668e7c33954f561b11\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/468eb7a163554289935d02d1bea869bc : SUCCESS in 2h 36m 11s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/97bbaa194baa4b51a77c8f7d1898fcc5 : SUCCESS in 41m 56s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/091e54ba4a1343a3b1bedc2ca1bc73f1 : SUCCESS in 1h 36m 07s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/6fbec25e2afc4b63982ccdd8561af3ee : SUCCESS in 1h 29m 29s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/777586f4dbd54e05bff6f30785432c84 : SUCCESS in 5m 31s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/47a150520d6845b3a0bc95bda53a9b2f : SUCCESS in 2h 18m 47s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/85da3da39e1845bba3716ef8f0203441 : SUCCESS in 1h 26m 56s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/76e1e1e698c24ae1ac38dd24311a90a3 : SUCCESS in 1h 15m 15s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/57e908f7c69b425cb7b3559b8fe773d3 : SUCCESS in 1h 35m 14s (non-voting)\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/b570b694446a4af58519ccee63c7bcc6 : SUCCESS in 1h 29m 16s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/0a87d38e84f94c388248a1a18d4627c0 : SUCCESS in 1h 08m 18s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/2b556cbe098c4cbc92629ab719f421b5 : SUCCESS in 15m 19s","accounts_in_message":[],"_revision_number":28},{"id":"e8702a64874fb7321759b2bbd581d915544bb2c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-08-30 12:50:45.000000000","message":"Uploaded patch set 29.","accounts_in_message":[],"_revision_number":29},{"id":"9e5790297793acb60b9f91b5381ec5e83575206f","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-08-30 12:51:31.000000000","message":"Patch Set 29:\n\n(1 comment)","accounts_in_message":[],"_revision_number":29},{"id":"bab0ae06c1283d8072c30ecca7c10313f9a60852","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-08-30 14:59:29.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/c1858180a2364e86a35ea0944bf7263b : SUCCESS in 2h 04m 28s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/deb86a771fb24eb6a3e0cd5199965b82 : SUCCESS in 2h 07m 21s","accounts_in_message":[],"_revision_number":29},{"id":"3f0f75afcb17a8fa3379e822932ec0dd47ac0701","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-08-30 15:04:29.000000000","message":"Uploaded patch set 30.","accounts_in_message":[],"_revision_number":30},{"id":"ec9e9a8936a286429ccfd48206f281d81ef3e680","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-08-30 17:49:04.000000000","message":"Patch Set 30: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/d8b21f2e820c435cbc9dfcd5527b8887 : FAILURE in 1h 40m 45s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/628a6262019149b7b37826d7985908f0 : FAILURE in 2h 43m 35s","accounts_in_message":[],"_revision_number":30},{"id":"7751e70315494668924e43a58abdc6a80cdab093","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 18:35:07.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e240f7909584c4291417966ca725785\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/f24de80fa31c489c8078063d64180f29 : SUCCESS in 2h 36m 59s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/8093bdf34d59453ca3d7f07d183fa2d1 : SUCCESS in 42m 40s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/61ab19d119af41f691efbeba6cb09da7 : SUCCESS in 1h 35m 49s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/a16af7de2df04fb392ff7ee28f4782b2 : SUCCESS in 1h 28m 17s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fe12027920d7465aae7480818ee686b8 : SUCCESS in 4m 52s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/d037cb70408f47bd827808860ed90a34 : SUCCESS in 3h 18m 18s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/9a8ec15bf3fd413f93dd51572fa2027a : SUCCESS in 1h 56m 35s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/8fa8fc2e9a9b461c83f81e040eb300e1 : SUCCESS in 1h 17m 14s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/c1f3fea8cd0d404fa29b9bf350672a33 : SUCCESS in 1h 36m 20s (non-voting)\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/12f28c1c912e4f4584dd4e155fff6ffb : SUCCESS in 1h 33m 39s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/abf9e8d32977491a8e38985589a1fcb0 : SUCCESS in 1h 13m 41s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/a1e958c18fc1485892eabe9a78bc818b : SUCCESS in 20m 23s","accounts_in_message":[],"_revision_number":30},{"id":"9097995c26c90bf7e981da21885f8fd4b69c8755","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-08-31 08:26:42.000000000","message":"Uploaded patch set 31.","accounts_in_message":[],"_revision_number":31},{"id":"323bc8ea435e98bc2dcd483934029722e1f2a243","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 08:52:08.000000000","message":"Patch Set 31: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fa52005cf6584ea89419564baa62949a\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/bc570cae50ca4e36abc0089ceb47340e : FAILURE in 6m 48s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/7e53815b075e498285bcf42c0e84577a : SUCCESS in 21m 58s\nSkipped 10 jobs","accounts_in_message":[],"_revision_number":31},{"id":"e540926b50fad5511436a2608b951c95be06d4c0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-08-31 11:40:04.000000000","message":"Patch Set 31: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/b8a6e8d63ae545ae98d1ed0fcb150525 : FAILURE in 3h 12m 18s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/2fb1f3962d9946278e6c071eaeb1a05b : FAILURE in 2h 04m 36s","accounts_in_message":[],"_revision_number":31},{"id":"4df409f933a7823b1fcb423f6fd666a5f2d9b9f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-09-01 12:36:22.000000000","message":"Uploaded patch set 32.","accounts_in_message":[],"_revision_number":32},{"id":"96aae61d582e38a489b0338e6a5dc53ddc437081","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-09-01 12:39:25.000000000","message":"Uploaded patch set 33.","accounts_in_message":[],"_revision_number":33},{"id":"3e39958d53d483d42a2d128e12845cce9f1ad4d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-09-01 12:54:21.000000000","message":"Uploaded patch set 34.","accounts_in_message":[],"_revision_number":34},{"id":"65afcd412e65b0a784320f0ad4d6081fb0225c17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-01 13:20:41.000000000","message":"Patch Set 34: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/755d99cff2e948f78e0434eaed62cfc8\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b456c084b7c9483bb62af567b525bc97 : FAILURE in 3m 45s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/d9d901928aa149e89040a0957aa77e1f : SUCCESS in 21m 12s\nSkipped 10 jobs","accounts_in_message":[],"_revision_number":34},{"id":"474bda11248dd254eef4a1e43762497f38befeb7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-09-01 14:46:40.000000000","message":"Patch Set 34: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/7b49056e9bcf426f8b67b9a73a40336f : FAILURE in 1h 50m 34s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/4b230b9f6d204b22a5fd0634ed886ecd : FAILURE in 1h 47m 59s","accounts_in_message":[],"_revision_number":34},{"id":"4fe8bc759f645e680654657db97b4b6df760f7a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-09-02 13:28:50.000000000","message":"Uploaded patch set 35.","accounts_in_message":[],"_revision_number":35},{"id":"4813c0ca0ed9d75e5f29e069cb84f4f523b9d4ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 13:55:37.000000000","message":"Patch Set 35: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e5b84049440e47f2a3525af89c14393f\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f2478bf71f5042ef94b11fef16c2a166 : FAILURE in 4m 38s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/2112462abd4b450089464311bcc79067 : SUCCESS in 20m 53s\nSkipped 10 jobs","accounts_in_message":[],"_revision_number":35},{"id":"a519cfb79c81534536321340f2d96aadfdf2f60d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-09-02 14:13:25.000000000","message":"Patch Set 35: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/9ccd39ff22f541e0a862c2c56a39e0b6 : FAILURE in 24m 35s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/d5726728c8c749bc8d119dc49f7a17ab : FAILURE in 22m 28s","accounts_in_message":[],"_revision_number":35},{"id":"f425ef343a6108388287a726248da5fba8b5efb6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-09-05 06:28:03.000000000","message":"Uploaded patch set 36.","accounts_in_message":[],"_revision_number":36},{"id":"9dde9bde9d6f0bbe6f7ac9e8bfed62825efbee3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-05 06:58:28.000000000","message":"Patch Set 36: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/844ea5c5474c4ea3b1e3f57929afdfb1\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/557b3218cdc341a78b092348ab5a3374 : FAILURE in 5m 32s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/b92aacd565464bcfaddd64845bea3819 : SUCCESS in 23m 07s\nSkipped 10 jobs","accounts_in_message":[],"_revision_number":36},{"id":"b7d8cdd358c80e3b7856625a67d8c9c501336353","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-09-05 08:11:25.000000000","message":"Uploaded patch set 37.","accounts_in_message":[],"_revision_number":37},{"id":"8020bdd2bd909e03e4ede79e912352385c936fab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-05 08:33:48.000000000","message":"Patch Set 37: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0068bdf9982d40a8b9bf76072c1d6604\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2b78ff21507a414d9bfe84dc97c2087c : FAILURE in 4m 38s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/212829659b1f49b386bd0d26233e7765 : SUCCESS in 20m 45s\nSkipped 10 jobs","accounts_in_message":[],"_revision_number":37},{"id":"ac3732bbe019e8169fcd4cea95a3fc5e8b5ad866","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-09-05 08:56:04.000000000","message":"Patch Set 37: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/0b82deaa663345cfb0f3a9eccf98ca7f : FAILURE in 23m 55s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/6b6d7925cfea4ec294a4c890c2d12299 : FAILURE in 24m 09s","accounts_in_message":[],"_revision_number":37},{"id":"5ec0465b842eb45fe8d171988b4af6d446e11c03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-09-05 10:53:25.000000000","message":"Uploaded patch set 38.","accounts_in_message":[],"_revision_number":38},{"id":"c49a4e942da62dd9744d8fd5d461e2191d03570c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-05 11:16:29.000000000","message":"Patch Set 38: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d6e52e3e56fd4e29af7b6f4f854543d3\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0e7ba0ece3c54acf8e919f20749ff64e : FAILURE in 6m 55s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/5f44a11db4dc44928ad50f69caf985a0 : SUCCESS in 22m 13s\nSkipped 10 jobs","accounts_in_message":[],"_revision_number":38},{"id":"0ce9c3eabe76a245a09d58716c6511a70d6aed2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-09-05 14:00:14.000000000","message":"Uploaded patch set 39.","accounts_in_message":[],"_revision_number":39},{"id":"6e63330e56d7f6fc271c0dfed056ae9f832419a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-05 14:26:19.000000000","message":"Patch Set 39: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c279fbcd6bf748beaf332db0de4e63e6\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/14ccf0e70b03493ba4d27e5fe3efd0af : FAILURE in 4m 32s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/d971f16dc39e4f74b7654c27c3d80d4e : SUCCESS in 21m 00s\nSkipped 10 jobs","accounts_in_message":[],"_revision_number":39},{"id":"cb5f6f77f17001daf54e5ad659c6f87e87d97d0b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-09-05 18:14:53.000000000","message":"Patch Set 39: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/0ff6a39855964cc8914d391c3d4456fa : TIMED_OUT in 4h 09m 12s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/661d2cd899f740489e552a4cf9861d73 : SUCCESS in 1h 52m 13s","accounts_in_message":[],"_revision_number":39},{"id":"294e28c144fc41df7018a5d044685f78a2ae7934","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-09-06 06:42:29.000000000","message":"Uploaded patch set 40.","accounts_in_message":[],"_revision_number":40},{"id":"e40acc4eb8f095c223e85cedcb3267b13ab6edb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 07:00:39.000000000","message":"Patch Set 40: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b06efde651e74f8785eff12e3f5a7ff2\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7e1e41e20d94404982150c6209e640cb : FAILURE in 4m 31s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/5eb2c4455a8a437988081483044d18dd : SUCCESS in 15m 56s\nSkipped 10 jobs","accounts_in_message":[],"_revision_number":40},{"id":"12ee7d0754659bef909df3df2a553427c4ea4ee3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-09-06 09:00:50.000000000","message":"Uploaded patch set 41.","accounts_in_message":[],"_revision_number":41},{"id":"6f3d11c9f00d0b0fd2e44b5ae2874f8b2d11107c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 09:23:09.000000000","message":"Patch Set 41: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a9f150e2938546f08df30d34fefff71d\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/cba30a4bfb944eb58de069f8fa1af0a1 : FAILURE in 5m 03s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/4050143fa297427bb38c0cadd3da7858 : SUCCESS in 20m 13s\nSkipped 10 jobs","accounts_in_message":[],"_revision_number":41},{"id":"ab693ec2cbfa7ae34a925d392ffb7939d60b7a0a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-09-06 13:57:02.000000000","message":"Patch Set 41: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/46d4d68b8e5a45afb1a63061c1cdf3a1 : TIMED_OUT in 4h 08m 50s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/177168efab1045fa839366245d132d19 : FAILURE in 1h 41m 24s","accounts_in_message":[],"_revision_number":41},{"id":"ae5ab7455a4b4d2da6459cf7c2b488ce3cd9c5f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-09-13 10:18:57.000000000","message":"Uploaded patch set 42.","accounts_in_message":[],"_revision_number":42},{"id":"042b33b9727268a62de7213ce6f14ae64e87d356","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-13 10:40:00.000000000","message":"Patch Set 42: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/140de98b194249f0ac926484e3e0a2d5\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9c2da1ff49ce4d59bac3d5d7bbb1ba47 : FAILURE in 4m 58s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/8bcb54650599421d83a24285fc89a411 : SUCCESS in 20m 09s\nSkipped 15 jobs","accounts_in_message":[],"_revision_number":42},{"id":"26ad459265a4929bc2659e87bb3668c625632b22","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-09-13 13:01:16.000000000","message":"Patch Set 42: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/9f4ba9c506a148e5b0e364e6530645b7 : FAILURE in 1h 49m 08s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/99d8ec11a0d348f39b0df224c6adbe5d : FAILURE in 1h 57m 15s","accounts_in_message":[],"_revision_number":42},{"id":"476cb2cbbc75b972d800bd30ab2dc16d8b19f962","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-09-14 04:37:50.000000000","message":"Uploaded patch set 43.","accounts_in_message":[],"_revision_number":43},{"id":"b81d86af10900b76ed6d1cfe4b8a03d9b0416725","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 05:07:18.000000000","message":"Patch Set 43: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/43eea772af94448a9ea6afa86d598742\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f245bc7c3a88438090771a26d8377e1f : FAILURE in 4m 17s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/8ad7bf1cc135407a9d05b5ba49e20ad0 : SUCCESS in 22m 11s\nSkipped 15 jobs","accounts_in_message":[],"_revision_number":43},{"id":"743389f7ccfe2cc42e4cef280eb1553d38326edb","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-09-14 06:48:35.000000000","message":"Patch Set 43: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/c5e527802b1146e0980938d72141ab20 : SUCCESS in 2h 09m 33s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/59111eaa3ab54abea5ac939bfb63190b : SUCCESS in 2h 08m 36s","accounts_in_message":[],"_revision_number":43},{"id":"e62c9c9b7935959d217034f1f422d2f403f809d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-09-14 07:38:40.000000000","message":"Uploaded patch set 44.","accounts_in_message":[],"_revision_number":44},{"id":"9344dc4c499d0a60e9e180a2211ffc5708fcd259","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 07:54:17.000000000","message":"Patch Set 44: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/408ec0049f354a819e26f0c76e11b61c\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6d52f5b9e3ca440e8611593be798b8f4 : FAILURE in 4m 02s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/98bd85a9f6a14d2887fc3e074095a351 : SUCCESS in 14m 01s\nSkipped 15 jobs","accounts_in_message":[],"_revision_number":44},{"id":"5f1ab7c0f7df1aab2a9ecb43e294336e1518f38c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-09-14 10:13:55.000000000","message":"Patch Set 44: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/454249ebe1114faeb125a35ad287d3c2 : SUCCESS in 2h 33m 52s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/72a8aebaaf504b90a9f09c593199c53f : RETRY_LIMIT in 17m 18s","accounts_in_message":[],"_revision_number":44},{"id":"7ce26adc233df25a157371c6400c31072213b719","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-09-14 11:14:48.000000000","message":"Uploaded patch set 45.","accounts_in_message":[],"_revision_number":45},{"id":"5319e45226911ac5a8729a9d3cf6ae7d80bd7cf8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 13:27:24.000000000","message":"Patch Set 45: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6c6c7b046fa54806974ef67360f37578\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fda014b97153422ab8dd7203bc4cec01 : FAILURE in 4m 42s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/8121ad9d61ab4fa0b3ab4d96cac6ae97 : SUCCESS in 27m 38s\nSkipped 15 jobs","accounts_in_message":[],"_revision_number":45},{"id":"7be89e6f43f195d9661831c11d22aaa00c6f58f6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-09-14 15:11:33.000000000","message":"Patch Set 45: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/828df388b0d74795b9e079e82bb23dbe : FAILURE in 2h 14m 19s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/6860015f2da541b78d5f6e5d54de598b : SUCCESS in 2h 22m 40s","accounts_in_message":[],"_revision_number":45},{"id":"b329e0114f9ec3556e23a5e76ae75e835de9859b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-09-14 15:44:54.000000000","message":"Uploaded patch set 46.","accounts_in_message":[],"_revision_number":46},{"id":"11bb518286d7170acf471d57950320be580ff889","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 16:11:55.000000000","message":"Patch Set 46: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/549d0eda3f3c4807af43beedd476868a\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a49e02cd201d4a6988b6c5fbfd8e5edd : FAILURE in 4m 08s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/c58831bb069e4abc9040a4efca46612b : SUCCESS in 21m 26s\nSkipped 15 jobs","accounts_in_message":[],"_revision_number":46},{"id":"8dfe902bf9428705d907207a40907b1fd04c66ed","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-09-14 19:16:33.000000000","message":"Patch Set 46: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/59334b48b19c49fdbfcfcfb5a8a65ff8 : SUCCESS in 2h 31m 17s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/707644013d194ae991b366fc5be3bb48 : SUCCESS in 2h 07m 46s","accounts_in_message":[],"_revision_number":46},{"id":"20d5a647963b55dfa5ef3afb5e19753795cfde15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-09-15 08:33:35.000000000","message":"Uploaded patch set 47.","accounts_in_message":[],"_revision_number":47},{"id":"4dbd679a06601d2a3ae2b8b8a037f1697297fba1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-15 09:01:42.000000000","message":"Patch Set 47: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b61efe974fc841cbbfb00f18bd4bbc04\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/61cb72b1aa6a40528b083c1a11f86411 : FAILURE in 5m 16s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/bfd5b286f59346b2bdf00f7ec373be12 : SUCCESS in 22m 46s\nSkipped 15 jobs","accounts_in_message":[],"_revision_number":47},{"id":"87775c62606d1ffd722292b7e6fd6a7eb0cc62ed","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-09-15 10:59:54.000000000","message":"Patch Set 47: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/d831a5c74d264b71a4a57b1fe40f39ad : SUCCESS in 2h 18m 20s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/1b8a5bf7328a4442b7b75bd0d63be72b : FAILURE in 2h 25m 13s","accounts_in_message":[],"_revision_number":47},{"id":"b113e529196fd8f74fd4b89b07180ad10bfd6599","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-09-15 11:55:02.000000000","message":"Patch Set 48: Published edit on patch set 47.","accounts_in_message":[],"_revision_number":48},{"id":"64b13d57362c77762f2e819b541df27984979d1c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-09-15 11:56:52.000000000","message":"Patch Set 49: Published edit on patch set 48.","accounts_in_message":[],"_revision_number":49},{"id":"6292c43e678e42568ada322fa22710398b1b60c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-15 12:23:11.000000000","message":"Patch Set 49: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/17c15056b9434e188fd3a537d14c234c\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/43b594ed7e96492ab3fd5aa62dc43fee : FAILURE in 4m 51s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/75591afc746e4d10b25f840afe6bd8fd : SUCCESS in 20m 57s\nSkipped 15 jobs","accounts_in_message":[],"_revision_number":49},{"id":"86ac15c197e3d7d98bc0025047766320c1deea83","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-09-15 14:14:01.000000000","message":"Patch Set 49: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/90780b7e6a0241ffb1ec2aac5ebd9481 : FAILURE in 36m 17s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/7b653f1f9f0e474396284e11a88f83cd : FAILURE in 1h 51m 15s","accounts_in_message":[],"_revision_number":49},{"id":"e198bd9fc4a0c9e80a4c4277ed3a6a892d5c45e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-09-16 05:10:20.000000000","message":"Patch Set 50: Published edit on patch set 49.","accounts_in_message":[],"_revision_number":50},{"id":"bd0a6b409a8eeb4f1faee969a7469010f290af65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-16 05:26:08.000000000","message":"Patch Set 50: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fa066e1966ca476eb98fc365dd62f44b\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/cc674cbaeb3f45bf9b0c3c78532883f3 : FAILURE in 4m 18s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/56cdcf97c790432383f553a381f8c49a : SUCCESS in 13m 49s\nSkipped 15 jobs","accounts_in_message":[],"_revision_number":50},{"id":"3db2bd4a27cc00384eeeab10cda6778264b7c878","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-09-16 07:16:18.000000000","message":"Patch Set 50: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/c21a9233ea474273ad33808acfd67c6d : FAILURE in 1h 54m 53s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/7440812ea4e94ef5b03c3ab53484256f : SUCCESS in 2h 05m 05s","accounts_in_message":[],"_revision_number":50},{"id":"95354dff99ad5d2d7c1de44e5d8488f663269059","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-09-20 06:43:40.000000000","message":"Uploaded patch set 51.","accounts_in_message":[],"_revision_number":51},{"id":"365b48145eb7828c6ebc49455ff2e4adc1ee4bbc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 07:00:15.000000000","message":"Patch Set 51: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f73ae103f54c4d39aabc7c62cc6accc7\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/cc292655ed0c4fc0b9921efc35c5d162 : FAILURE in 4m 20s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/02b1912651884462a4b9793656d31b28 : SUCCESS in 15m 07s\nSkipped 15 jobs","accounts_in_message":[],"_revision_number":51},{"id":"e80f2bb7e8730d84e7145879d51e7a269229d5e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-09-20 07:04:39.000000000","message":"Uploaded patch set 52.","accounts_in_message":[],"_revision_number":52},{"id":"925bab72340658f84328fa72b999bdc42337ae87","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-09-20 09:35:42.000000000","message":"Patch Set 52: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/1780a7da3148490ba45d41b92e37b571 : SUCCESS in 2h 20m 19s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/23416158b8774fc5ae4855662e8e1f3d : SUCCESS in 2h 29m 49s","accounts_in_message":[],"_revision_number":52},{"id":"15445b6959fa9fc415c181f9f2e47d66a17f164e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 10:33:48.000000000","message":"Patch Set 52: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ced63e22fc4549999ecddd908bd01214\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/1cce1cec539c41a29a34f91a74569178 : SUCCESS in 3h 19m 50s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/e5a22464e7d744d9ba2b91efb7901619 : SUCCESS in 40m 42s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/b30ec2252ac642e5affc2505570cf3c8 : SUCCESS in 2h 03m 14s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/a126f60936394c7f9fdceb219cb16545 : SUCCESS in 1h 40m 49s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1a3406db6dab43e2949ae17836046630 : SUCCESS in 4m 45s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/1c7144ce595049bb8d5c5eca6e6d9ad9 : SUCCESS in 2h 32m 32s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/c96139f025994557a847f890d5490730 : SUCCESS in 1h 35m 27s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/1b9b399b631e4b3d9179b7cede1b623a : SUCCESS in 1h 32m 40s\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/92e30dbac19c41758e42753e77b9bf64 : SUCCESS in 2h 00m 16s\n- tripleo-ci-centos-9-scenario004-standalone https://zuul.opendev.org/t/openstack/build/fc4cb8a3fcd64f5cb2d47b0df9e975b9 : SUCCESS in 1h 40m 18s\n- tripleo-ci-centos-9-scenario010-standalone https://zuul.opendev.org/t/openstack/build/2d9374a4c301409ebbfb75d7abc73445 : NODE_FAILURE in 0s\n- tripleo-ci-centos-9-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/c5bc344da6a24f4ca732ddc07e5a5f59 : SUCCESS in 1h 33m 39s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/cfe3e6d718dd4f22a23533fe233c1b6a : SUCCESS in 2h 14m 27s (non-voting)\n- tripleo-ci-centos-9-standalone-on-multinode-ipa https://zuul.opendev.org/t/openstack/build/c383f872ec1346128a692e71b3ed5208 : SUCCESS in 2h 31m 54s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/1215f0b4b36f4a10a7876e7307f5af44 : SUCCESS in 1h 33m 20s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/f78ba6a03b4c483eb336c1b5e9e54a20 : SUCCESS in 1h 11m 25s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/932400df5e30419fb63d098771f3eb39 : SUCCESS in 20m 07s","accounts_in_message":[],"_revision_number":52},{"id":"51fd16da677a80ff5031b66c70c28aba16263cdd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-09-20 12:56:04.000000000","message":"Uploaded patch set 53.","accounts_in_message":[],"_revision_number":53},{"id":"abe19d43d8f094057b7467ae0396609ecf6b4308","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 13:48:17.000000000","message":"Patch Set 53: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f0b2d24e3dc0420e9a319ce4e6f04d81\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/117c93c45bd84d0d8834d622efeeb045 : FAILURE in 28m 40s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b42203e455db457ab5f51f92b3cfe105 : SUCCESS in 4m 45s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/364490221c4343a9ab5b3ba396e8ebb7 : FAILURE in 26m 53s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/72d5bb2dad0a4e60a7f4b9c2101c569c : SUCCESS in 20m 46s\nSkipped 13 jobs","accounts_in_message":[],"_revision_number":53},{"id":"1e9198531f4485b7123643edf0aa93ae851c6375","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-09-20 15:18:48.000000000","message":"Patch Set 53: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/33f0629975d641a9b9a7803a931cc65d : SUCCESS in 2h 21m 36s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/fdf540e549924a838342f6eeff216ad3 : FAILURE in 1h 19m 04s","accounts_in_message":[],"_revision_number":53},{"id":"a5b944b841e260366535e6018623e90e5bc12110","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-09-21 07:22:23.000000000","message":"Patch Set 53: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":53},{"id":"9166f624442c2da35b41f42574811eaa7cf5b764","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-04 06:59:43.000000000","message":"Uploaded patch set 54.","accounts_in_message":[],"_revision_number":54},{"id":"aaf10b4584d03109c5edd322d7724cd733ad296a","author":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"date":"2022-10-04 07:13:22.000000000","message":"Patch Set 54:\n\n(1 comment)","accounts_in_message":[],"_revision_number":54},{"id":"164409446aa9115b4915a3c7cbe8ed5789ece37b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-04 07:27:14.000000000","message":"Uploaded patch set 55.","accounts_in_message":[],"_revision_number":55},{"id":"4f29816ae886f997d70906d718494eadf649d76c","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-04 07:27:48.000000000","message":"Patch Set 55:\n\n(1 comment)","accounts_in_message":[],"_revision_number":55},{"id":"d23047d4e5d20d85a33ce13662173679b0944c4e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-10-04 09:36:33.000000000","message":"Patch Set 55: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/8f355d5329b44ed4b4f737dbd60d878f : SUCCESS in 2h 02m 12s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/7aa5a12bf3654dc99d40f4c99aa12356 : SUCCESS in 2h 08m 25s","accounts_in_message":[],"_revision_number":55},{"id":"be7aa4eafdc09a8bfef55519de9e4ede804e8c91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-04 10:22:04.000000000","message":"Patch Set 55: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bc83f9beb5394d72a88dc9337af8030c\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/f6ddada7cefe4e90809ec8b2072d8f53 : SUCCESS in 2h 48m 52s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/5d89c6c2bec24335abaff4944a569850 : SUCCESS in 39m 51s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/eb3549217d7d4a5e8c7d3885c95ce42b : SUCCESS in 1h 40m 58s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/53904f2ad01c4635a497ff9118c23746 : SUCCESS in 1h 27m 29s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3dd968fdc67d4295860371cf1ba10840 : SUCCESS in 4m 02s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/cb58c4c2eaa3458ba9cddec8a3ff7c41 : SUCCESS in 2h 03m 18s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/3fa82f8d927642d5b26c53bf8558b6b6 : FAILURE in 54m 51s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/9b6b2815ebba41d7ac5a033f902b560c : SUCCESS in 1h 38m 20s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/50303e0c35c947b4a3c955d35e2e3d0b : SUCCESS in 1h 48m 02s (non-voting)\n- tripleo-ci-centos-9-standalone-on-multinode-ipa https://zuul.opendev.org/t/openstack/build/27b9b526b1494cfc8fc21819fd0cb89c : SUCCESS in 1h 53m 45s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/4325d7eccf63410196482e7276ad4642 : SUCCESS in 1h 20m 19s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/0a0d6694aaf94afaafccb49cc3a65841 : FAILURE in 31m 34s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/198880ecc3ad45c0a51d9e19533e407f : SUCCESS in 18m 48s","accounts_in_message":[],"_revision_number":55},{"id":"9b17a5a2d8aa0438d599aa2185515802e165afb4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-04 10:42:51.000000000","message":"Uploaded patch set 56.","accounts_in_message":[],"_revision_number":56},{"id":"3abf139e1463b8a6af80b35630d29e1370257fc4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-04 13:55:25.000000000","message":"Patch Set 56: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1493424b01f349c68fda4a8c29c92f5b\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/d0a8502d5efe407782aac74a938e7b49 : SUCCESS in 3h 04m 51s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/47fd3e3dfab44d4ea999d5def156af5c : SUCCESS in 48m 38s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/100c06c3fa504d2f9f3ca2d50b219908 : SUCCESS in 1h 47m 42s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/1e265963323f44cabd89ac9571890fd7 : SUCCESS in 1h 38m 19s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/efe1f4aade7d45d08e8d24d314d09f88 : SUCCESS in 4m 42s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/fd33e59d4b544b7fb4e260e853152c97 : SUCCESS in 2h 54m 09s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/b5425759248e4223834ac9bca9b014ac : SUCCESS in 1h 52m 05s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/69506426806a4dabbe6aaf638e0f30e4 : SUCCESS in 1h 20m 21s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/5fd99bf8e4db44f59fd20a1c7162cee7 : SUCCESS in 1h 53m 03s (non-voting)\n- tripleo-ci-centos-9-standalone-on-multinode-ipa https://zuul.opendev.org/t/openstack/build/f4280e00d4004ba29ed82f7f3ef945eb : SUCCESS in 1h 58m 21s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/eb02812232784b73bc5e216033bac7de : SUCCESS in 1h 32m 39s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/6d4bc77e3048450dbeea3af5803a4c88 : SUCCESS in 1h 10m 41s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/58c98cba151e4e349fa3de70a10e6783 : SUCCESS in 18m 04s","accounts_in_message":[],"_revision_number":56},{"id":"cd066987880ace13ef01c51e052765e507e2309c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-10-04 14:21:05.000000000","message":"Patch Set 56: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/ff86bea3987e4453b36ffa01d41ea2d7 : SUCCESS in 2h 15m 58s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/fcde7b1687644d1d810067ccfec7725e : FAILURE in 1h 56m 16s","accounts_in_message":[],"_revision_number":56},{"id":"6c6c5661131083b4853e8412698137d38ca86a5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-06 09:49:11.000000000","message":"Uploaded patch set 57.","accounts_in_message":[],"_revision_number":57},{"id":"8014f8456d9dcc9802d6c2c463757f66f374addd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-06 11:46:53.000000000","message":"Uploaded patch set 58.","accounts_in_message":[],"_revision_number":58},{"id":"4177818896392b009a5d54822502af5413e3c9e8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-10-06 14:01:53.000000000","message":"Patch Set 58: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/675b7f2f90464f79babfd09e2fc5a3d5 : SUCCESS in 2h 11m 25s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/50bc4e2cbe4c485298ca032127ee2956 : FAILURE in 1h 10m 52s","accounts_in_message":[],"_revision_number":58},{"id":"a18b766daec31dfee4cc7b21bba0866b0113bb52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 14:49:56.000000000","message":"Patch Set 58: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7fa2ed50752b4a6d93bf65688530332f\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/bd7d358192fa41e7af8a91c47c0edb38 : SUCCESS in 2h 55m 56s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/5927fa19c6e54620934b3aa3d6d3921b : SUCCESS in 40m 20s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/5e1b1b2f7fd14641b3c0b2e6acf4f70e : SUCCESS in 1h 47m 39s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/ac33155c56c84b6f9ccfe595058f8fbd : SUCCESS in 1h 41m 25s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/59fe94a924c54bd8ada9abe5bde2653f : SUCCESS in 4m 32s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/5a981c46001347dc90842649812411ac : SUCCESS in 2h 42m 27s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/0ccd29f85d04452cbe045cded83798a2 : SUCCESS in 1h 43m 12s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/46a2da7b1dc5418499ecff3c050487b5 : SUCCESS in 1h 24m 25s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/ab0e7b2fc9444136b79dcb45d51ed2f2 : SUCCESS in 1h 54m 00s (non-voting)\n- tripleo-ci-centos-9-standalone-on-multinode-ipa https://zuul.opendev.org/t/openstack/build/37ff32f85c564f199f5e98009faacaa1 : SUCCESS in 1h 52m 45s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/0e2a53070cb2436c9620447dee5c25d3 : FAILURE in 1h 15m 36s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/291e2a29d51c4b95a3799d5f3b898fbf : SUCCESS in 1h 12m 03s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/a66562cd2fcf4e02a4e073acd47b844b : SUCCESS in 21m 17s","accounts_in_message":[],"_revision_number":58},{"id":"f107e4ffda887bbe6d8a635073421737644aad48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-07 08:08:56.000000000","message":"Uploaded patch set 59.","accounts_in_message":[],"_revision_number":59},{"id":"5dae2c2a9898e03dce3f1d47c904edc69de381e1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-10-07 10:28:16.000000000","message":"Patch Set 59: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/1bb20b108ad84c0bb1586ea7457cb42e : SUCCESS in 2h 13m 01s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/a54824a0bfb64158a2abc8a20bd4e492 : FAILURE in 2h 10m 59s","accounts_in_message":[],"_revision_number":59},{"id":"5c1d42efbf5760dbd9788a23c0590d11bb8fbcc1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-07 11:12:16.000000000","message":"Patch Set 59: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f4534637f0fb4951a66ccf05dd82c249\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/af819b4ab79242ff9433278930aab1c3 : SUCCESS in 2h 54m 13s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/b3836d5bc9024b04acce2cc6c4a232d9 : SUCCESS in 41m 04s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/fd297e8668ad4ac5b75f3da8b8882aaf : SUCCESS in 1h 41m 03s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/c6067a10a6034ceda2f91ac2c08e47ca : SUCCESS in 1h 41m 44s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e5431ee8cf644bb59c1bb049e88353dc : SUCCESS in 5m 32s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/adb1cbb849704e168cf49263538391be : SUCCESS in 2h 50m 20s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/a81b1fe46c574c3ca9a047bdc35a9f72 : SUCCESS in 1h 48m 54s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/7e8d69b26d004eb0b0a82e1f610ace34 : SUCCESS in 1h 19m 42s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/5e3d2728f98f4729a7f8441929ad9761 : SUCCESS in 1h 47m 05s (non-voting)\n- tripleo-ci-centos-9-standalone-on-multinode-ipa https://zuul.opendev.org/t/openstack/build/aa8e2575f15d4fcfb582a88cd24b21ea : SUCCESS in 1h 57m 43s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/924a376a568f4032ae434bc34e8ca80b : SUCCESS in 1h 30m 26s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/b0cd154cac05487398c9ca1d462e95d8 : FAILURE in 52m 19s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/1793962e4d1d4c19ab74ebdbec35961b : SUCCESS in 19m 43s","accounts_in_message":[],"_revision_number":59},{"id":"eafa69802904982ef2991b68c43c6176f5f1be7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-07 12:59:25.000000000","message":"Uploaded patch set 60.","accounts_in_message":[],"_revision_number":60},{"id":"d7582138840057a9c5b02dcd2471eaa1b4731937","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-07 14:47:09.000000000","message":"Uploaded patch set 61.","accounts_in_message":[],"_revision_number":61},{"id":"47ca99da849d1a8b4a9bcf8022f66162dab6cbed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-07 17:48:10.000000000","message":"Patch Set 61: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4bb037db4189423a8a25e905dfe6be6b\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/25742da17a264b3a9066926602ea4f31 : SUCCESS in 2h 50m 38s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/d8347719d7be46058d57228bbb4ff72b : SUCCESS in 44m 36s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/1a69729a78f74f38bb838f6c3dfa3349 : SUCCESS in 1h 39m 55s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/ab16543eb580458c92b6cfc9118c4ef6 : SUCCESS in 1h 36m 21s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a19c0a7fdaca4b4db91eaab9b2fdd607 : SUCCESS in 3m 46s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/201bd2349efe492e909acc77f5472579 : SUCCESS in 2h 41m 13s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/5f9e2128c5e54d9785751c64637b4edc : SUCCESS in 1h 45m 46s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/72122446d6ec408687693eff9bdf861b : SUCCESS in 1h 20m 24s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/c2c42d560c224043b08c867dacba689e : FAILURE in 1h 34m 02s (non-voting)\n- tripleo-ci-centos-9-standalone-on-multinode-ipa https://zuul.opendev.org/t/openstack/build/ccca734299e742c98ebbf6230b495b77 : SUCCESS in 1h 47m 00s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/58d24174e6c54424b11772df17158e34 : SUCCESS in 1h 26m 51s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/465d305f774b4d989c572e22bafaa2f1 : FAILURE in 54m 41s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/b2bd8463699c476885699bf076844dd9 : SUCCESS in 20m 51s","accounts_in_message":[],"_revision_number":61},{"id":"ada1c838cf8e1bcfe55dc0116c7543df36d62e77","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-10-07 17:56:09.000000000","message":"Patch Set 61: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/b51a4360669a4fc1b54f779b8528b802 : FAILURE in 2h 00m 13s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/3cbca09488aa4a0e98c97619d2488643 : FAILURE in 2h 16m 16s","accounts_in_message":[],"_revision_number":61},{"id":"3d432e16bbc8b67021a78834fc2b1da7b0cc3d2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-09 10:37:56.000000000","message":"Uploaded patch set 62.","accounts_in_message":[],"_revision_number":62},{"id":"d7e1eee3f4826732046f88e3344779b55ac59abe","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-10-09 10:56:49.000000000","message":"Patch Set 62: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/1bb8a3e2bb6a46898662a101a71c544f : FAILURE in 17m 29s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/57d4efec489f46629add5f71c5b2bceb : FAILURE in 16m 25s","accounts_in_message":[],"_revision_number":62},{"id":"b82926bf97c3bb7f09139a50b7573705d4150b80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-09 11:16:43.000000000","message":"Uploaded patch set 63.","accounts_in_message":[],"_revision_number":63},{"id":"e95333752fcdd6a5b8a0d102de46aad78efe0178","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-09 11:46:50.000000000","message":"Uploaded patch set 64.","accounts_in_message":[],"_revision_number":64},{"id":"929d92fe15eb17fb318abfaab093b6771f3ba032","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-10-09 14:05:49.000000000","message":"Patch Set 64: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/bff53d75ab3841518e095f86f2043606 : FAILURE in 1h 43m 29s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/80ef6ca7bf39473c8c601d38548f718b : SUCCESS in 2h 18m 03s","accounts_in_message":[],"_revision_number":64},{"id":"90828093dc589121399afea044d872093bbbaa82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-09 14:47:42.000000000","message":"Patch Set 64: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/67a5c805d3794c1ebd37de274e7bf59a\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/4bf7143a596f4e448f5536ada721e4df : SUCCESS in 2h 47m 57s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/2afc6df518094d2398fbca7f8e686c70 : SUCCESS in 42m 14s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/2f9d198ecdaa42889a3f1bcde7d21738 : SUCCESS in 1h 28m 43s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/b71ecc26951c4a158a0738f98c84fa33 : SUCCESS in 1h 29m 43s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/37c03cabd26f4325a24a6d23bd060d37 : SUCCESS in 3m 22s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/26897da4ab114f85bffa09e89a1333b9 : SUCCESS in 2h 33m 52s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/e1f8407f690c4501901bac59fdc530f7 : SUCCESS in 1h 36m 51s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/df551a04079c46fe820e7da03a3a3d2e : SUCCESS in 1h 13m 58s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/8b2475fde3f442c68b42b004c8352dd2 : SUCCESS in 1h 48m 24s (non-voting)\n- tripleo-ci-centos-9-standalone-on-multinode-ipa https://zuul.opendev.org/t/openstack/build/a2d1d162c76d4b4daaaa80f3299847c2 : SUCCESS in 1h 42m 32s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/2c13c92c1ec54a099638f22b1d72eff4 : SUCCESS in 1h 20m 17s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/697691b7835349ee9e35d36c42b7ee6b : FAILURE in 50m 31s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/e2e19e37b96045319b687275d103400c : SUCCESS in 15m 47s","accounts_in_message":[],"_revision_number":64},{"id":"80be08dde843ee828014b394307751bcdbbc05b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-10 06:46:13.000000000","message":"Uploaded patch set 65.","accounts_in_message":[],"_revision_number":65},{"id":"7d2d48113c61d15d835615d210042277108ca076","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-10 06:48:09.000000000","message":"Uploaded patch set 66.","accounts_in_message":[],"_revision_number":66},{"id":"d2a3938a63525400f401612c20821ab24053a36f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-10 07:09:20.000000000","message":"Patch Set 66: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b099b8a1d89e4594bd19d3ce6c2378d4\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c155618c9bc443d398a77394e89b5706 : FAILURE in 3m 48s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/7715cf6f68ff4e5f9dc628c1e56da97f : SUCCESS in 19m 05s\nSkipped 11 jobs","accounts_in_message":[],"_revision_number":66},{"id":"320012e25510d35245ce27d42c3b57d357573e34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-10 07:09:55.000000000","message":"Uploaded patch set 67.","accounts_in_message":[],"_revision_number":67},{"id":"c4d7a8a80c76a80f35f71c930ab453e745edb49b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-10 07:21:22.000000000","message":"Uploaded patch set 68.","accounts_in_message":[],"_revision_number":68},{"id":"b4c8146881cacbf3bfe7bfbd659b792b83a68d14","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-10 07:55:33.000000000","message":"Uploaded patch set 69.","accounts_in_message":[],"_revision_number":69},{"id":"9ff8ffc78f8bef9727fad525b1b3cd4de2e81a9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-10 08:17:06.000000000","message":"Uploaded patch set 70.","accounts_in_message":[],"_revision_number":70},{"id":"c769a0e7e01da19941f03a29f1556e9bb17f61ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-10 08:57:39.000000000","message":"Uploaded patch set 71.","accounts_in_message":[],"_revision_number":71},{"id":"fcafdb8315c67e87b639198b0e10007d84f0bb13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-10 09:22:21.000000000","message":"Uploaded patch set 72.","accounts_in_message":[],"_revision_number":72},{"id":"87792ee4976801f6b4a374d3a660fd85ec7cc89e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-10 11:31:55.000000000","message":"Uploaded patch set 73.","accounts_in_message":[],"_revision_number":73},{"id":"be46de207166de8a330d69e447060bec7f6e65e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-10 12:49:14.000000000","message":"Uploaded patch set 74.","accounts_in_message":[],"_revision_number":74},{"id":"7826bb22f72d2156103e81eff4802dfc3c0ec295","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-10-10 14:50:21.000000000","message":"Patch Set 74: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/aa83f600ddb3474fbdc088c7304c01f7 : FAILURE in 1h 48m 41s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/07079cfc3e8043a49085130b20edbfb7 : SUCCESS in 1h 59m 31s","accounts_in_message":[],"_revision_number":74},{"id":"9f847bb922205822576801e4abb4b7756d308f87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-10 15:57:55.000000000","message":"Patch Set 74: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e2cdafe801614315bc5f212aa15c5780\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/b53017946fa44c0eb11f909a3ebfe5da : SUCCESS in 2h 59m 57s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/bbd913e88d0d4f28b965a75e4c2bbd24 : SUCCESS in 35m 26s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/152dec6d00dc473ab53177ffb9dc1496 : SUCCESS in 1h 19m 31s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/c923627683994edfa948b8ff3c4b4c7d : SUCCESS in 1h 34m 29s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f5e23764084d4462baf9a2b470670fe2 : SUCCESS in 6m 47s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/9db21714911d4186b13a418c6edb2f3d : SUCCESS in 2h 05m 50s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/876ef13d03954e76a1a0774d5110fd7c : SUCCESS in 1h 15m 32s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/32db7f95118c439a9db25521c61d8ddf : SUCCESS in 1h 50m 27s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/1e2630fd2d004840ba1d148b5d6b7896 : SUCCESS in 1h 51m 41s (non-voting)\n- tripleo-ci-centos-9-standalone-on-multinode-ipa https://zuul.opendev.org/t/openstack/build/f538aac7e9c64195b1f543ecc6d92246 : SUCCESS in 2h 08m 27s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/038805074a974ae1b37d2cb793a5306e : SUCCESS in 1h 37m 36s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/724411d835284fca9dc662af9b43c266 : FAILURE in 48m 31s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/53963270f8484748bc86a61c29992699 : SUCCESS in 19m 46s","accounts_in_message":[],"_revision_number":74},{"id":"2de34e3f8afd68d6abbeaca0ad6cb253e1f18f0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-11 04:12:31.000000000","message":"Uploaded patch set 75.","accounts_in_message":[],"_revision_number":75},{"id":"b2727a893db1644b099af507f6094664d4f5e20e","author":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"date":"2022-10-11 04:15:10.000000000","message":"Patch Set 75:\n\n(1 comment)","accounts_in_message":[],"_revision_number":75},{"id":"6232c313ed7daec92c7388b9db95154830a5f662","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-10-11 06:34:48.000000000","message":"Patch Set 75: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/580ac311cc7e471792d411dd3809de9a : SUCCESS in 1h 51m 27s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/4bd39d9edfff43cab4660f0829fa80f4 : SUCCESS in 1h 54m 11s","accounts_in_message":[],"_revision_number":75},{"id":"1b415f663faa05f3c2cc7828ba59b095c8440bcb","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2022-10-11 07:11:41.000000000","message":"Patch Set 75:\n\n(4 comments)","accounts_in_message":[],"_revision_number":75},{"id":"bc51c1706729bca0bc919892fe1d6c4ba489fff8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-11 08:34:50.000000000","message":"Uploaded patch set 76.","accounts_in_message":[],"_revision_number":76},{"id":"dcb55a4ae292796e2ada93886ab2004c303ad9df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-11 10:13:34.000000000","message":"Uploaded patch set 77.","accounts_in_message":[],"_revision_number":77},{"id":"b03b96a36754fc582079461b9e677ad85fcd31c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 10:30:55.000000000","message":"Patch Set 77: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1fd4a30e23cd4dacac6cbc1a43bbf0f6\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7fbb68e606914fb1a809ad1419ca4f95 : FAILURE in 4m 04s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/ab25ba3c7d1c4a28b2f32d9ee9f7ffcc : SUCCESS in 15m 44s\nSkipped 11 jobs","accounts_in_message":[],"_revision_number":77},{"id":"5522e42787b0177d1f5d4ac2daa950b9e059ad10","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-11 11:13:24.000000000","message":"Uploaded patch set 78.","accounts_in_message":[],"_revision_number":78},{"id":"19efd20b0708fb1a4b2bf42797284837e68bb8bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-11 13:09:07.000000000","message":"Uploaded patch set 79.","accounts_in_message":[],"_revision_number":79},{"id":"e0a8683963f7cb3506be292cb4a7e44611788f29","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2022-10-11 15:18:49.000000000","message":"Patch Set 79: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":79},{"id":"c89b9ad654d81e68f467e11227e9dda604ffb183","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2022-10-11 15:20:25.000000000","message":"Patch Set 79:\n\n(1 comment)","accounts_in_message":[],"_revision_number":79},{"id":"3572fdfcc76dc53e069c391ba69c42e2617d459f","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2022-10-11 15:26:06.000000000","message":"Patch Set 79:\n\n(1 comment)","accounts_in_message":[],"_revision_number":79},{"id":"9c2f0b3909e56593b30431decd33a528b37e1051","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 16:01:44.000000000","message":"Patch Set 79: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ced61608bc84701a085321b577d7662\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/d8ef0555586f41d3be2728ef68644731 : SUCCESS in 2h 42m 50s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/63b20110fa88465c89be04a95ae733b8 : SUCCESS in 36m 38s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/62f1a2defb044a18be65c822bd5e9073 : SUCCESS in 1h 24m 25s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/5834776246b34e99aa943c069034095b : SUCCESS in 1h 44m 54s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2ccc1647bef84b11bd77a52448037cfa : SUCCESS in 3m 54s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/3f6170d30a9f47f08bf1fd40d28ea14d : SUCCESS in 2h 10m 40s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/e464d3ab0c1a4cde9a78283c8007cc55 : SUCCESS in 1h 30m 25s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/11938ddd56b4478e98c1e741658101c2 : SUCCESS in 1h 26m 12s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/28d378cb4a27435eb5e17c31d434afc6 : SUCCESS in 1h 59m 01s (non-voting)\n- tripleo-ci-centos-9-standalone-on-multinode-ipa https://zuul.opendev.org/t/openstack/build/07007c07e0c4487780eca2b0d28cb0bc : SUCCESS in 1h 55m 21s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/e23b92b0513e417dbf35701800734024 : SUCCESS in 1h 34m 21s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/adc55f6dbbff4ca5934d8b2f350d9baa : SUCCESS in 1h 10m 18s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/eaee290299ac43d087f58cf62003207c : SUCCESS in 22m 06s","accounts_in_message":[],"_revision_number":79},{"id":"8ec76ccb7a42808fc9710df4d587b79457a13b3c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-10-11 17:05:20.000000000","message":"Patch Set 79: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/616a43b0415e4cbd9649c72aa70b8f69 : SUCCESS in 2h 28m 32s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/30c7ac7087814de0bf773efdab8ba3e1 : FAILURE in 1h 53m 29s","accounts_in_message":[],"_revision_number":79},{"id":"36c9f858e57b3d57107c8a66cdbfd381827010cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-12 01:02:14.000000000","message":"Uploaded patch set 80.","accounts_in_message":[],"_revision_number":80},{"id":"2872175cb0bcc7bf92ca0a55277fb352cb7a14cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-12 02:10:00.000000000","message":"Uploaded patch set 81.","accounts_in_message":[],"_revision_number":81},{"id":"268068f4af1304df07919295fdc3f6c1eec8fa07","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-10-12 04:27:22.000000000","message":"Patch Set 81: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/4cbce5aaa7f54d8896c163f0c1651f85 : FAILURE in 1h 46m 04s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/83db25cfff194df491b9ca23719e4e34 : SUCCESS in 2h 08m 46s","accounts_in_message":[],"_revision_number":81},{"id":"e9c33027b0aadd5c9e69044392171dfb3214dd27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 05:43:51.000000000","message":"Patch Set 81: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c5c7693670b14715a7d4cec2f5bf2ecf\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/a9d45ead63f841ffb453dd06df6e6c3d : SUCCESS in 2h 25m 34s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/fdaac0c19d87459bb066687cb919e7aa : SUCCESS in 36m 15s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/3d168ac7be3e482a9cdd4adece095676 : SUCCESS in 1h 19m 37s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/a0838460a2db4299ae89ee35d448553e : SUCCESS in 1h 25m 32s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1542fe862ec04fc19c81c4e48ce41e0c : SUCCESS in 5m 07s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/855f8a37419e4f098ce810b99e717753 : SUCCESS in 2h 50m 28s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/46ac7decb350446db0f1fc78f5575904 : SUCCESS in 1h 43m 18s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/77d03bb115ca4a009c334355adcfec52 : SUCCESS in 1h 11m 34s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/32b26c018fc34f3c8ed2af03eed00f07 : FAILURE in 1h 37m 35s (non-voting)\n- tripleo-ci-centos-9-standalone-on-multinode-ipa https://zuul.opendev.org/t/openstack/build/8941d106667c4fed8984216ea3d7a8a9 : SUCCESS in 1h 32m 24s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/20d81d2f787041f2b51966cd23ecc16c : SUCCESS in 1h 13m 23s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/696f7321928e420296b02acda8ddb574 : FAILURE in 27m 16s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/02191e7c9e2b425f88285412be8b367f : SUCCESS in 19m 54s","accounts_in_message":[],"_revision_number":81},{"id":"6044cb80dd259c7e27abfaef1274b4978a2b31cf","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-12 06:35:36.000000000","message":"Patch Set 81:\n\n(1 comment)","accounts_in_message":[],"_revision_number":81},{"id":"233b19046b46cd387452edf856b4530616d951a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-12 06:38:39.000000000","message":"Uploaded patch set 82.","accounts_in_message":[],"_revision_number":82},{"id":"efd8435aafa76a457973aca234174f78c74564f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-12 06:39:36.000000000","message":"Uploaded patch set 83.","accounts_in_message":[],"_revision_number":83},{"id":"fa2f870dc2e71a3921acdbe5b82d92ddbf1c3f43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-12 06:46:08.000000000","message":"Uploaded patch set 84.","accounts_in_message":[],"_revision_number":84},{"id":"0ef26c9c0a14e2e6e718f3d87c79758e7b7de816","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-12 06:48:46.000000000","message":"Uploaded patch set 85.","accounts_in_message":[],"_revision_number":85},{"id":"b4a088bb33e5dd9dda32efa52661420d468f99d6","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-12 07:56:10.000000000","message":"Patch Set 85: Workflow-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":85},{"id":"1dfff8f72158330eb43a584447a3b73944fb6dce","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-12 07:58:02.000000000","message":"Patch Set 85:\n\n(1 comment)","accounts_in_message":[],"_revision_number":85},{"id":"fd64772624c6ff429971452c31dd0bc4b3d46d72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-12 08:29:32.000000000","message":"Uploaded patch set 86.","accounts_in_message":[],"_revision_number":86},{"id":"10e8e3a5f5cc07f267a80807b2cdf2c40cf793ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-12 08:30:32.000000000","message":"Uploaded patch set 87.","accounts_in_message":[],"_revision_number":87},{"id":"077c37fb2bbcaedfd8db6b51d55f387d073eeeb3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-12 08:53:03.000000000","message":"Uploaded patch set 88.","accounts_in_message":[],"_revision_number":88},{"id":"734b4173bef3de38dd107aa7e31d0c5e64c3e2b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 11:11:28.000000000","message":"Patch Set 88: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/54d2e763d05844778885b165799ffa7c\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/65efc77d53c54fd3a19c06a66f04e3f1 : SUCCESS in 2h 05m 23s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/9a4a6e12c50643df8fd4ff23b5fc688e : SUCCESS in 34m 31s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/02da173f784c4de7b21eb52bd6464447 : SUCCESS in 1h 14m 30s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/8dd57d9b7da34047ab81d0a70fac112f : SUCCESS in 1h 23m 26s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/de91d08051b3464d8db9595c884a840f : SUCCESS in 3m 45s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/1b70769c4332481b97cbe9d8df5f3c5e : SUCCESS in 2h 12m 41s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/901bfd2adcce4d6dbc2058b230e58331 : SUCCESS in 1h 24m 50s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/26198b177c5d46769ef30e5d3f94131d : SUCCESS in 1h 05m 20s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/4d3e51a9f30e48d285b7c6f1e6d0c8f1 : RETRY_LIMIT in 11s (non-voting)\n- tripleo-ci-centos-9-standalone-on-multinode-ipa https://zuul.opendev.org/t/openstack/build/5600357c121c42fdb8c5a82bb0eeb290 : RETRY_LIMIT in 13s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/4df16d1979a94b0baf7893b8323b25a0 : SUCCESS in 1h 09m 40s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/323d1485ed89401e9cc44b6f298c52ac : RETRY_LIMIT in 9s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/5fa67076e26040d1b95dbc4f38d81f95 : SUCCESS in 18m 42s","accounts_in_message":[],"_revision_number":88},{"id":"a4bc909b7a1577489670cfcb6da8f50d4e40d756","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-10-12 11:50:43.000000000","message":"Patch Set 88: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/0de61c7adacd47d7a32dc5e7281eda7e : FAILURE in 20m 37s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/8365b9b1f39a400f8a40464537e6efa2 : FAILURE in 2h 37m 53s","accounts_in_message":[],"_revision_number":88},{"id":"37cb270625f1f5def7e6359bd3e3e1c1cbb2b418","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2022-10-12 13:42:45.000000000","message":"Patch Set 88:\n\n(1 comment)","accounts_in_message":[],"_revision_number":88},{"id":"8937540301b97b0863b5bfe67dad4af345993653","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2022-10-12 13:47:08.000000000","message":"Patch Set 88:\n\n(1 comment)","accounts_in_message":[],"_revision_number":88},{"id":"9ffb98c82904f69bbf1827ad7875afe7e688c93d","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2022-10-12 13:50:00.000000000","message":"Patch Set 88: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":88},{"id":"91c7ee920bc6601e96cd6a1d35e52ceee590f933","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-12 14:58:22.000000000","message":"Uploaded patch set 89.","accounts_in_message":[],"_revision_number":89},{"id":"c116771e87832be119ee5669f28d088dbd981ab3","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-12 14:58:39.000000000","message":"Patch Set 88:\n\n(2 comments)","accounts_in_message":[],"_revision_number":88},{"id":"9c71e81f8f8dd8a6486002ce32cbdf16ea042346","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-12 16:30:45.000000000","message":"Uploaded patch set 90.","accounts_in_message":[],"_revision_number":90},{"id":"7862449beaad92c4cd8fafca36315734fef34fb7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-12 16:32:21.000000000","message":"Uploaded patch set 91.","accounts_in_message":[],"_revision_number":91},{"id":"548fce8463cefbd5c9304e70a57bc5a3451a2635","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-10-12 18:52:56.000000000","message":"Patch Set 91: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/899e9dc1031a441dafbb0f648952699a : SUCCESS in 2h 19m 34s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/1e593ef02ba948c8be9748c7bc6c95fc : SUCCESS in 2h 15m 23s","accounts_in_message":[],"_revision_number":91},{"id":"fa7e509ec2d0ad8d6c064d26f76447e19b905661","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 20:01:32.000000000","message":"Patch Set 91: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cb183ff61ead4765ad06bba3a93130b5\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/b5c958516f014a1f9df00df250a24a4e : SUCCESS in 3h 15m 39s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/da175cf1cac642e987a16641af4d0fea : SUCCESS in 49m 19s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/f861fd51e08b4480a2ab01365ba7f587 : SUCCESS in 1h 50m 57s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/185da28e5b884b909959a6345fe602bd : SUCCESS in 1h 44m 19s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c7a84aa1ea174deaac1eb4f5385bcfc5 : SUCCESS in 4m 07s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/25d67d8a3c824493b948dadadd2890d9 : SUCCESS in 3h 04m 45s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/237641d6cfbe49399e636b54ca0320be : SUCCESS in 1h 48m 24s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/d86f6df6c35140229228312313350071 : SUCCESS in 1h 23m 21s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/6010700a91bd4122ade098468506b3dd : SUCCESS in 2h 02m 04s (non-voting)\n- tripleo-ci-centos-9-standalone-on-multinode-ipa https://zuul.opendev.org/t/openstack/build/e4fd3d677ac4498daee41cc415bd245a : SUCCESS in 1h 57m 32s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/5b1ed5a58c9a4ec0afc5ff3ec0f01025 : SUCCESS in 1h 31m 41s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/237eda0a0d4842c989afc961c8e29398 : FAILURE in 25m 40s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/72eaf3c2aea445298a0ec65e9cfdf268 : SUCCESS in 21m 27s","accounts_in_message":[],"_revision_number":91},{"id":"6ed6e79b6ef5a5d5fccd220cede8cc41ad612a67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-13 02:57:55.000000000","message":"Uploaded patch set 92.","accounts_in_message":[],"_revision_number":92},{"id":"32115cdfcb2889a12c6c26da817c76c8226be3dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-13 03:46:22.000000000","message":"Uploaded patch set 93.","accounts_in_message":[],"_revision_number":93},{"id":"89c29c3f0034f53fd217a68e94b261765c31bb6c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-10-13 06:05:10.000000000","message":"Patch Set 93: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/418f9a16fa774d1f954d17b7d1cc4f3f : SUCCESS in 2h 17m 21s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/f8e53a5e26f4469d85489201e40edc26 : FAILURE in 1h 24m 19s","accounts_in_message":[],"_revision_number":93},{"id":"2ca09ed692f36df2d2a32e7135cf7def36dc5810","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-13 06:47:47.000000000","message":"Patch Set 93: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59bb8c0a77d44698a6ae56688fc35956\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/adb6949b51674928afe2e53eb88f7194 : SUCCESS in 2h 55m 48s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/5cb25963621844b48d13d98336a5bba7 : SUCCESS in 32m 14s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/ab54f12d3cc14b4eb4755e5aab069d54 : SUCCESS in 1h 21m 33s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/a9823c10a7c84e408ffdc49380d84f26 : SUCCESS in 1h 18m 44s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/27d6e3fc0a3a400b82ff01da6126eb75 : SUCCESS in 3m 41s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/5abe9982733b4480899dcefae8ffe1c4 : SUCCESS in 2h 10m 50s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/f165152469ea4421954e6a8179e1fe9d : SUCCESS in 1h 26m 13s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/72f77e5d6f4b4a9aa985bc08a0bab54e : SUCCESS in 1h 07m 07s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/f7a1dc7113a84d47b80ec9d5cda03458 : SUCCESS in 1h 26m 44s (non-voting)\n- tripleo-ci-centos-9-standalone-on-multinode-ipa https://zuul.opendev.org/t/openstack/build/b3ce1b343fe04e6e9a54fce29574754d : SUCCESS in 2h 07m 06s\n- tripleo-ci-centos-9-content-provider-zed https://zuul.opendev.org/t/openstack/build/08ff9c0d351948b1ac487e3dd4b323b7 : SUCCESS in 1h 58m 44s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/3a524c278b464ad98fe8fe17d2057124 : SUCCESS in 1h 37m 50s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/df669c630dff464592eac68fa4b6d506 : SUCCESS in 1h 06m 50s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/3c8170f0158d444b8bcaad2c707fc263 : SUCCESS in 13m 07s","accounts_in_message":[],"_revision_number":93},{"id":"f3691780e01505004b2b061421c3006867d1da16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-13 07:52:11.000000000","message":"Uploaded patch set 94.","accounts_in_message":[],"_revision_number":94},{"id":"f35c924f2f9251417e24ac34889bc690386158db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-13 09:09:09.000000000","message":"Uploaded patch set 95.","accounts_in_message":[],"_revision_number":95},{"id":"09458c0dced03ba5e1b060187e141d17bdd04fba","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-10-13 11:31:48.000000000","message":"Patch Set 95: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/438c8a7552124f958c9a8701f6cf7939 : FAILURE in 2h 14m 05s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/8317bce4074540deb99763692872705b : SUCCESS in 2h 03m 40s","accounts_in_message":[],"_revision_number":95},{"id":"940bb115f6061e594809072dd98a56f1b5476135","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-13 11:57:49.000000000","message":"Uploaded patch set 96.","accounts_in_message":[],"_revision_number":96},{"id":"6eba2e9df408d42bb4d6f4ce4fc2b5f98c64cddf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-13 14:54:57.000000000","message":"Uploaded patch set 97.","accounts_in_message":[],"_revision_number":97},{"id":"088a1119374ab54743316f8d01b9c4b5990f8967","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-13 18:25:54.000000000","message":"Patch Set 97: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f16b774626940a3b293a99af3eee959\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/3d5168a729dd4b9485dbd888b40ae658 : SUCCESS in 3h 20m 29s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/8eca1abddec947648bad2a7021c156cb : SUCCESS in 56m 04s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/00c923d21c3c47e8bef84a941289cbbb : SUCCESS in 1h 51m 14s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/7d47dc183e0c4e41a7978404481029bb : SUCCESS in 1h 50m 05s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9ade934f23ff4a319928d2e8dd81a6e9 : SUCCESS in 3m 56s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/a2987011e5af42d8aa5c236aee38090d : SUCCESS in 2h 44m 31s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/446eb1cbbb084872a43edd2715a8de59 : SUCCESS in 2h 01m 26s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/ae2101b918ae4e1fb27d94217e04f97c : SUCCESS in 1h 30m 38s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/9cc610361a434f43a83debf54a23f8be : SUCCESS in 1h 57m 16s (non-voting)\n- tripleo-ci-centos-9-standalone-on-multinode-ipa https://zuul.opendev.org/t/openstack/build/ed8a0dc7660f404eb397359c6456a0d6 : SUCCESS in 2h 04m 08s\n- tripleo-ci-centos-9-content-provider-zed https://zuul.opendev.org/t/openstack/build/6f108093073f434591daabaa7bdcf789 : SUCCESS in 2h 05m 01s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/2d9fb39ff79a47af84c14b5cb55ede7b : SUCCESS in 1h 43m 22s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/95769a21705c463b93785ade769ad2e3 : SUCCESS in 50m 24s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/c34878ed88034bc39200ff0c0a5abdf6 : SUCCESS in 18m 11s","accounts_in_message":[],"_revision_number":97},{"id":"6dd80d3de2c477b0c4afad5a4a9640d4178085ce","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-10-13 19:11:43.000000000","message":"Patch Set 97: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/ebf4b344c78e47afae5b1f9e83071b27 : SUCCESS in 2h 15m 53s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/aa0c612176514afbb20b8969820edd3b : FAILURE in 2h 23m 49s","accounts_in_message":[],"_revision_number":97},{"id":"c7a547664ffa8469b4d37d11c07046b9447ae898","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-14 08:06:49.000000000","message":"Uploaded patch set 98.","accounts_in_message":[],"_revision_number":98},{"id":"c17035999c201b7ec83f297dfdd4834ae3ed5e76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-14 08:36:13.000000000","message":"Uploaded patch set 99.","accounts_in_message":[],"_revision_number":99},{"id":"a592303cdf4778b9d3ce7338ee57f52ef430c450","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-14 11:08:13.000000000","message":"Uploaded patch set 100.","accounts_in_message":[],"_revision_number":100},{"id":"ff4ba6a36d2ed5d2901807ef1dee796adb10ac6a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-14 11:51:58.000000000","message":"Uploaded patch set 101.","accounts_in_message":[],"_revision_number":101},{"id":"58ab0f295cda5d5dc9972c25446fedbe2cc71a3b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-10-14 14:04:55.000000000","message":"Patch Set 101: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/66d8e39dab8a40e4bd604ceea2c799d5 : FAILURE in 2h 10m 41s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/05bb1816b0634a7595325723b9304b90 : SUCCESS in 2h 11m 01s","accounts_in_message":[],"_revision_number":101},{"id":"d7c359055a57892ac2a0268386c76992440e7e67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-14 15:16:55.000000000","message":"Patch Set 101: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/580a967369b84669a61c6b98a295dd85\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/441a2e928e8d4d9f9b046e2a4b1ffdd5 : SUCCESS in 3h 19m 36s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/85c641eec09045adb03841e13105186e : SUCCESS in 46m 16s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/3988ddc8e1234d9eb442b743552d69f0 : SUCCESS in 1h 38m 53s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/c93417cf119d400d9b7f404363a38b06 : SUCCESS in 1h 42m 48s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/63b25f27726a4056a3276d1081c11c15 : SUCCESS in 3m 47s\n- tripleo-ci-centos-9-content-provider-zed https://zuul.opendev.org/t/openstack/build/0b2c301e1c6c44b7889f5306a6f9f830 : SUCCESS in 3h 12m 32s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/b0248b1dcc354b2e84716057d13e160d : SUCCESS in 2h 59m 29s\n- tripleo-ci-centos-9-containers-multinode-zed https://zuul.opendev.org/t/openstack/build/3cb049a34a6345518631f1244fdaffe1 : SUCCESS in 1h 57m 17s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/5f5b3af813d14af0bd4a41cf0b4a205b : SUCCESS in 1h 49m 56s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/bbdb8beeb8ef442f8e37976630b0124f : SUCCESS in 1h 33m 33s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/69ef1eb40b494c4e8a99d6f494dbb404 : SUCCESS in 1h 51m 17s (non-voting)\n- tripleo-ci-centos-9-standalone-on-multinode-ipa https://zuul.opendev.org/t/openstack/build/1f2419f502b74a01b13601ee1962936f : SUCCESS in 2h 04m 56s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/e285c95245804ba9b6bf2785656ffc3c : SUCCESS in 1h 33m 41s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/d38c8e849a3b4588ad40e23c01dc25f8 : SUCCESS in 1h 04m 06s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/4a899463e6ae4c00a6809b9182a98f02 : SUCCESS in 14m 06s","accounts_in_message":[],"_revision_number":101},{"id":"35dbddf3913d89d4de299c2283650318f0b2b0be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-14 16:13:30.000000000","message":"Uploaded patch set 102.","accounts_in_message":[],"_revision_number":102},{"id":"d86056668cbda8003f4e542d1ba8a6dbdd15ebf9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-14 17:36:02.000000000","message":"Uploaded patch set 103.","accounts_in_message":[],"_revision_number":103},{"id":"60ac1b6576ec3515a5cc5e7a798b14b14c446eae","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-10-14 19:55:41.000000000","message":"Patch Set 103: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/e2215cd2923942bca62b81018683b289 : SUCCESS in 2h 05m 34s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/cafe4223a733421f8527e19ec99fd86b : FAILURE in 1h 24m 13s","accounts_in_message":[],"_revision_number":103},{"id":"eb891a2c277dde8722de5b6f01ebcba19804e01f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-14 20:25:15.000000000","message":"Patch Set 103: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc91de51fb5b42eda0dc6844d3752c6d\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/9f24e888fa4d4dc1ad7f85374cd5d7a6 : SUCCESS in 2h 08m 16s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/a6298c6682e846dba6d1e8af82f53f8b : SUCCESS in 32m 58s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/54ad2a25fb3c46498456451ec14f26a8 : SUCCESS in 1h 23m 32s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/7c6f54c13f3c4de7833130103046f7d3 : SUCCESS in 1h 21m 05s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/dd47850e061d41f2baae12847a710ad5 : SUCCESS in 4m 03s\n- tripleo-ci-centos-9-content-provider-zed https://zuul.opendev.org/t/openstack/build/13141a692e5a499899cebdaec43e0006 : SUCCESS in 2h 39m 12s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/a31e0c992d4544b1b6c51a19fa38ba42 : SUCCESS in 2h 30m 14s\n- tripleo-ci-centos-9-containers-multinode-zed https://zuul.opendev.org/t/openstack/build/f0cb0c9707f948ee9ff93dd5e658b91a : SUCCESS in 1h 41m 43s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/21a56d26b4bb4a2db91f5a2acf0ebb7c : SUCCESS in 1h 34m 57s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/f4c34991963c44db809f683a43693f8d : SUCCESS in 1h 07m 36s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/47e84f4497dc4c089b16cf997c11e5eb : SUCCESS in 1h 27m 36s (non-voting)\n- tripleo-ci-centos-9-standalone-on-multinode-ipa https://zuul.opendev.org/t/openstack/build/f2074defce8b42c7a4bfe523676ca373 : SUCCESS in 1h 23m 55s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/79ad93585c354a8885ebec4e8fce11e3 : SUCCESS in 1h 05m 40s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/2a29550afb3d48688dafeab2dd5fbe57 : SUCCESS in 55m 27s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/dcfd5ee7daad47d187a535cda40bf19f : SUCCESS in 20m 40s","accounts_in_message":[],"_revision_number":103},{"id":"bcecc27fc4ff84b7a5e5544a4eb233f7ec59ebc1","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-10-17 07:02:41.000000000","message":"Patch Set 103: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":103},{"id":"d12a7019a2a23ec16a1ee229200fe152ab3506a1","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-17 11:26:50.000000000","message":"Patch Set 103:\n\n(2 comments)","accounts_in_message":[],"_revision_number":103},{"id":"24535f23eb61402da44adb66d67044f8b6740702","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-18 04:30:16.000000000","message":"Uploaded patch set 104.","accounts_in_message":[],"_revision_number":104},{"id":"31796e86fcb418f8851b4a2541d7ed765f0a2bb9","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-18 04:30:29.000000000","message":"Patch Set 103:\n\n(1 comment)","accounts_in_message":[],"_revision_number":103},{"id":"399bb96c41f3631e0451e0f199d19dc399d9bda7","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-18 04:31:47.000000000","message":"Patch Set 104:\n\n(1 comment)","accounts_in_message":[],"_revision_number":104},{"id":"89985cdc6abfc5f9c8d18841c9d553a68cf92e33","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-18 04:33:52.000000000","message":"Uploaded patch set 105.","accounts_in_message":[],"_revision_number":105},{"id":"f72497e0366c533f780214c0c17e80c8dac390b7","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-18 04:34:43.000000000","message":"Patch Set 104:\n\n(2 comments)","accounts_in_message":[],"_revision_number":104},{"id":"3660877df94c8b81aed521e54a66cb44943b7713","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-18 05:59:59.000000000","message":"Uploaded patch set 106.","accounts_in_message":[],"_revision_number":106},{"id":"f37aeff80b1a2122e646cbcfaea26d2ec374a0b8","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-10-18 06:22:25.000000000","message":"Patch Set 106: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":106},{"id":"c492d2d0ec2c1de9e56031b81a7e47a8133caa31","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-18 08:01:35.000000000","message":"Uploaded patch set 107.","accounts_in_message":[],"_revision_number":107},{"id":"388628a6ae613a6fd25980d159c959d86ce55db4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-10-18 08:16:10.000000000","message":"Patch Set 106: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/e2ec44a263ee480b871dae8dfcd1564d : SUCCESS in 2h 08m 02s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/e2abb32b2e5e4933ad87ab7d3f4b388f : SUCCESS in 1h 55m 26s","accounts_in_message":[],"_revision_number":106},{"id":"252d1ddde0996410b90482384a7ae231ec8eb462","author":{"_account_id":33080,"name":"Mikolaj Ciecierski","display_name":"Mikolaj Ciecierski","email":"mciecier@redhat.com","username":"mciecierski"},"date":"2022-10-18 09:35:12.000000000","message":"Patch Set 107:\n\n(1 comment)","accounts_in_message":[],"_revision_number":107},{"id":"70a6ddc8983536767a7c2f3d86f10b92cba08fe6","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-18 09:46:04.000000000","message":"Patch Set 107:\n\n(1 comment)","accounts_in_message":[],"_revision_number":107},{"id":"28b9a59a2b6ca6e83da1d1f0fef52524225ef0c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-18 09:47:30.000000000","message":"Uploaded patch set 108.","accounts_in_message":[],"_revision_number":108},{"id":"2bade23249bf44da5077df886b5842b74fdb6753","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-18 09:55:14.000000000","message":"Patch Set 108:\n\n(1 comment)","accounts_in_message":[],"_revision_number":108},{"id":"82df8cbc0411d236d248b5cad1b9d1a7098ff1ba","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-18 11:35:23.000000000","message":"Patch Set 108:\n\n(1 comment)","accounts_in_message":[],"_revision_number":108},{"id":"0c4ed5c6a67a3aafd5c97470e1504d2140de40d3","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2022-10-18 11:45:12.000000000","message":"Patch Set 108: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":108},{"id":"79d4e1dcf84b99d05051a3e87a6c05dd270178fe","author":{"_account_id":33080,"name":"Mikolaj Ciecierski","display_name":"Mikolaj Ciecierski","email":"mciecier@redhat.com","username":"mciecierski"},"date":"2022-10-18 12:04:16.000000000","message":"Patch Set 108:\n\n(2 comments)","accounts_in_message":[],"_revision_number":108},{"id":"e44fff2e8b804a72f544e3f4e9f9772758c1936c","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-18 12:13:54.000000000","message":"Patch Set 108:\n\n(2 comments)","accounts_in_message":[],"_revision_number":108},{"id":"0b0deb27ed6c1fe585586c5d406168dbc0dbba60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-18 12:32:41.000000000","message":"Uploaded patch set 109.","accounts_in_message":[],"_revision_number":109},{"id":"8d4f60c516ca2dd6e1162b3b2a12f8d597189ad9","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-18 12:33:34.000000000","message":"Patch Set 108:\n\n(1 comment)","accounts_in_message":[],"_revision_number":108},{"id":"04af36d8168fc0c16ca643e6d5674d3ec75aaf92","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-10-18 14:56:14.000000000","message":"Patch Set 109: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/b45e7e3778b14f7a9cda4879e4ae7d3a : FAILURE in 2h 02m 26s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/3a02d1f15c274b3eb02233f5a1e14bce : SUCCESS in 2h 09m 46s","accounts_in_message":[],"_revision_number":109},{"id":"9a4c6d0c1cbebdef3a86e8f13c64a23678d428aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-18 15:22:58.000000000","message":"Uploaded patch set 110.","accounts_in_message":[],"_revision_number":110},{"id":"a5d089f53f4906109e5c12e2159e81b2c01dda43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 15:51:49.000000000","message":"Patch Set 110: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c0e8fd6163e34c6d9cc3174b79e852b8\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/7a64bb7ad7f94bfa8492fcf142bceeed : FAILURE in 19m 34s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9ce828972b3547e896342ebe664e17fb : SUCCESS in 3m 59s\n- tripleo-ci-centos-9-content-provider-zed https://zuul.opendev.org/t/openstack/build/7d8850072512430a9620a156e97b4258 : FAILURE in 17m 48s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/d4005fa1e6ad41ba83b1081e055c3995 : FAILURE in 17m 10s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/e31de3892d384016a4b408150371dbfe : SUCCESS in 17m 41s\nSkipped 9 jobs","accounts_in_message":[],"_revision_number":110},{"id":"67bf93de254b2f1e59c6d07d9e721ba0da90eb45","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-18 15:57:07.000000000","message":"Patch Set 110:\n\n(1 comment)","accounts_in_message":[],"_revision_number":110},{"id":"33a23d2dba98128ddb6e0b7fbd68a5468a516051","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-10-18 16:02:05.000000000","message":"Patch Set 110: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/c57d8b57faf749a9b48f31149edff06c : FAILURE in 22m 44s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/8e0f6510cfe649e1bd9c17a2b48e3cd4 : FAILURE in 22m 30s","accounts_in_message":[],"_revision_number":110},{"id":"7cab7a493922be160a18d26d8f1d0b9afdeb501c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-19 03:42:10.000000000","message":"Uploaded patch set 111.","accounts_in_message":[],"_revision_number":111},{"id":"0376f5362377907d1806d102914737b6327d0a91","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-10-19 04:05:23.000000000","message":"Patch Set 111: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/ab900725a2db4c19a8bbc5addaec01df : FAILURE in 22m 20s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/4d1c5ce2e2d345d6b30fe180e058c07e : FAILURE in 21m 36s","accounts_in_message":[],"_revision_number":111},{"id":"d738ab6698410ff333c0aa5be1eed87a0f1d2f9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 04:10:25.000000000","message":"Patch Set 111: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3136fae513ff4000ac5f5a237e609852\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/973342c0531740a3803514363d5d1678 : FAILURE in 18m 13s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/cdce0abd053949f285cc9050100b1cf4 : SUCCESS in 4m 18s\n- tripleo-ci-centos-9-content-provider-zed https://zuul.opendev.org/t/openstack/build/e807aeb5514048838d8917bbacb4affa : FAILURE in 20m 16s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/8c1ce77a1a5b46998cd3d4b03b15ebdf : FAILURE in 17m 58s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/a7a5e57f22bb4806aaaf6ab79770df3f : SUCCESS in 18m 32s\nSkipped 9 jobs","accounts_in_message":[],"_revision_number":111},{"id":"54b63f1925f3928bf8e404ba330fd0b5d8354e73","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-10-19 07:45:47.000000000","message":"Patch Set 111: Code-Review+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":111},{"id":"c12426c70b5755c43b43e83a415444dae384d9e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-19 11:33:15.000000000","message":"Uploaded patch set 112.","accounts_in_message":[],"_revision_number":112},{"id":"849a131f879cc3e3e9383ff58e7b80b8d8586e2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-19 12:48:22.000000000","message":"Uploaded patch set 113.","accounts_in_message":[],"_revision_number":113},{"id":"48fc9e5a3eac64b951ddd889b46038734e1614d5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-10-19 15:03:01.000000000","message":"Patch Set 113: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/86f44612ce594aa88258d51e395d061e : SUCCESS in 2h 10m 18s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/dc69bff3dbcc4b2999f5d511c9c0cc6d : SUCCESS in 2h 13m 27s","accounts_in_message":[],"_revision_number":113},{"id":"d5e561c1c244bbb8883fac2b7a4125c4ee4ccef6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-19 15:13:10.000000000","message":"Uploaded patch set 114.","accounts_in_message":[],"_revision_number":114},{"id":"169b9533723547cd833d98304ee0a12616e2c051","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-10-19 17:53:21.000000000","message":"Patch Set 114: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/d4de43cfc81343cf8d564de929c3a522 : SUCCESS in 2h 11m 46s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/3032b71486ac42639e11f54ef560c1af : SUCCESS in 2h 15m 25s","accounts_in_message":[],"_revision_number":114},{"id":"ea5ea725591d72f6a5084824f809c2ff64879503","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 18:46:15.000000000","message":"Patch Set 114: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1b1bcf68fa245c7931ceb1b018c98a4\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/c49ea0fbbded4751af00e3839ce1dc97 : SUCCESS in 3h 19m 13s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/2d311eb44cab4787adbb5da222f5aefd : SUCCESS in 50m 15s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/22710f3714014f89bdedc77073f32d41 : SUCCESS in 1h 54m 37s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/39c9c9860e8e4361a729caba5f822ab3 : SUCCESS in 1h 40m 35s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/001ebd42e235447581f60a64f7eabdd7 : SUCCESS in 4m 02s\n- tripleo-ci-centos-9-content-provider-zed https://zuul.opendev.org/t/openstack/build/c0eb4b0fac244c06b110aef2529574a0 : SUCCESS in 3h 18m 53s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/ca4c8aad6caa41749cf543bd1140af1d : SUCCESS in 3h 20m 53s\n- tripleo-ci-centos-9-containers-multinode-zed https://zuul.opendev.org/t/openstack/build/ebfa96304c394a2ea51ca21806b2cc18 : SUCCESS in 1h 50m 58s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/d6ec6f571ac842f78fab60b2738157c3 : SUCCESS in 1h 55m 00s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/a7a7c8ce52ce4e68a832ffaf3e86ab35 : SUCCESS in 1h 25m 02s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/af12014e1537406b8956ff4d1df5bc13 : SUCCESS in 2h 04m 32s (non-voting)\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/6bff3b8809984f15a7237668a27945e6 : SUCCESS in 1h 40m 45s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/d0122bf57dd647d5a037ae2c9fd5999e : SUCCESS in 1h 14m 34s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/d6460e74786d426e80c6eadac9c0302d : SUCCESS in 18m 52s","accounts_in_message":[],"_revision_number":114},{"id":"29f185008608761e77b366d2dbc234a3e8cd845f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-20 01:54:52.000000000","message":"Uploaded patch set 115.","accounts_in_message":[],"_revision_number":115},{"id":"d1cd22b9d2f267e440c876c7735c782d26de5cee","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-10-20 04:49:38.000000000","message":"Patch Set 115: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/65b0e18198e54c38a292a4f0573074e9 : SUCCESS in 2h 27m 25s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/8c991816d9b34db19d0daa18874c7a79 : SUCCESS in 2h 04m 52s","accounts_in_message":[],"_revision_number":115},{"id":"4067881382fa28365f706217b3f5d7b0ff74d6ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 05:05:44.000000000","message":"Patch Set 115: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9e03e9fd4b14826b7ddc7ff2aa9b0ac\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/cb2c0a19c55f4297a65603c96f87cd60 : SUCCESS in 3h 01m 35s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/5d478df4b87f4503ba31048070572632 : SUCCESS in 58m 03s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/4c96b03335f24599bcb90ddd1cac45b3 : SUCCESS in 1h 59m 39s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/6f63a0cb0ecb440388ee14f0baa340b4 : SUCCESS in 1h 38m 29s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/49cb4eb56927493dbf0cb810af996f45 : SUCCESS in 3m 54s\n- tripleo-ci-centos-9-content-provider-zed https://zuul.opendev.org/t/openstack/build/e59f713da2c34b73840d89b35aba45e4 : SUCCESS in 3h 04m 14s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/e2f1e64cd838451da6c432f5adad02fa : SUCCESS in 2h 46m 19s\n- tripleo-ci-centos-9-containers-multinode-zed https://zuul.opendev.org/t/openstack/build/e85f93a98e1e42458e924441b90f8042 : SUCCESS in 2h 00m 02s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/933b35c1b3f14c6e8810b8b9d8b4d93a : SUCCESS in 1h 47m 49s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/0930a6d637ac45d5982f3f92c891e0ef : SUCCESS in 1h 32m 48s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/e52b1cba338e40739e6e3b54d9084f95 : SUCCESS in 1h 52m 55s (non-voting)\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/c62dccb57ff640f6a0750f6e3c097152 : SUCCESS in 1h 36m 50s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/694dd4216e1141b8b78d0b98b07f6be2 : SUCCESS in 1h 21m 30s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/d3de37a901ad48b0bb5b532ad3b769aa : SUCCESS in 19m 45s","accounts_in_message":[],"_revision_number":115},{"id":"89d90aa65b8672def2c5c53cf85bdcb0a0d62eb7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-20 05:30:20.000000000","message":"Uploaded patch set 116.","accounts_in_message":[],"_revision_number":116},{"id":"549966a078142c3f6ff67a7213132c71b5c34d03","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-20 07:57:13.000000000","message":"Patch Set 116:\n\n(1 comment)","accounts_in_message":[],"_revision_number":116},{"id":"88f3c8d3a2f7d434a908b02589432e524e627671","author":{"_account_id":33080,"name":"Mikolaj Ciecierski","display_name":"Mikolaj Ciecierski","email":"mciecier@redhat.com","username":"mciecierski"},"date":"2022-10-20 08:03:56.000000000","message":"Patch Set 116: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":116},{"id":"05e94e1badfb5fc12070e0e4405cf17832745353","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-10-20 08:20:16.000000000","message":"Patch Set 116: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/b43c26c1f7894f0482bfbb2d773466af : SUCCESS in 2h 15m 29s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/c578b9933fed4e559e134d7e84e36620 : SUCCESS in 2h 22m 51s","accounts_in_message":[],"_revision_number":116},{"id":"14cd1f79ffbc42c721b05a1b25d1f3d243b9a097","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 08:24:02.000000000","message":"Patch Set 116: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7bdd9f66222d41adb0d3d552d313f9d4\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/6358b1135bf240739afeeef2f99c606e : SUCCESS in 2h 28m 38s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/9c28b194fb93402697ebf489341978dc : SUCCESS in 36m 40s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/c42d9e01d5c94e408045830d00ddab1a : SUCCESS in 1h 40m 34s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/e8e300497bee408f92c8d456a63b01bc : SUCCESS in 1h 37m 59s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2d2b7256b2dc49a580d533de4837ccea : SUCCESS in 3m 40s\n- tripleo-ci-centos-9-content-provider-zed https://zuul.opendev.org/t/openstack/build/84a5d38d44454a66a97f606c94d3a28c : SUCCESS in 2h 40m 54s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/c6ac268eae4f4e17b25a89861e71e0f8 : SUCCESS in 2h 32m 38s\n- tripleo-ci-centos-9-containers-multinode-zed https://zuul.opendev.org/t/openstack/build/d8119669d3e84284bb2c7cacf16372e8 : FAILURE in 1h 49m 23s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/1d19f920522447f5862f9b75262e41b6 : SUCCESS in 1h 34m 40s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/8f43d102696f477ab123c2ebabbd0a4d : SUCCESS in 1h 08m 32s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/662b18f4a16040ae805b2ac4cdade83a : SUCCESS in 1h 38m 28s (non-voting)\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/2ac96c8a89c94a97875f0969d9d946b0 : SUCCESS in 1h 24m 11s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/6c7e65ad5b9340d89379207a8c6d961b : SUCCESS in 56m 58s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/a560bee52ec14b7d9ae1884192ce9f6e : SUCCESS in 22m 06s","accounts_in_message":[],"_revision_number":116},{"id":"61252ce4dbf8ae1840b3f0cb0b11928e0caf2616","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-20 08:26:28.000000000","message":"Patch Set 116:\n\n(1 comment)","accounts_in_message":[],"_revision_number":116},{"id":"f5b087f77f832a3e3153dd234c694faad9b33540","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-10-20 10:33:19.000000000","message":"Patch Set 116:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/70f9752696ec401b91cd1ad8d7f9ceca : SUCCESS in 2h 05m 16s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/cecb735eb3184b33956c8f7c0b12ac6d : SUCCESS in 2h 03m 24s","accounts_in_message":[],"_revision_number":116},{"id":"e833d72a72a8d55d2203ff653a990012c5fc60de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 10:49:08.000000000","message":"Patch Set 116: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b02a112acd1543388879e15219bc9770\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/8005a60a963b41bf8d30a1fc946b238e : SUCCESS in 2h 16m 19s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/07944574b35b47b1ae9ce381f696a7c9 : SUCCESS in 36m 21s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/1501d9a23816423abc9cb2068f0e1657 : SUCCESS in 1h 15m 23s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/e72398b0d08b40de91dad3fe33261ca4 : SUCCESS in 1h 11m 23s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b6adc7f7f239450cbdb8c0ed23b26fa4 : SUCCESS in 3m 44s\n- tripleo-ci-centos-9-content-provider-zed https://zuul.opendev.org/t/openstack/build/399a9e1e139f45e9abb2ef86e1407b0e : SUCCESS in 1h 59m 45s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/15b8294037c3427b9bb07f0518b8e2e3 : SUCCESS in 2h 00m 52s\n- tripleo-ci-centos-9-containers-multinode-zed https://zuul.opendev.org/t/openstack/build/6bb4e15b51f7407d83d6091f746b5b0a : SUCCESS in 1h 18m 30s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/2202ddc58cbe4f9ebb530e92527099c7 : SUCCESS in 1h 22m 54s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/ae52f9a2672f46ac8cd1e086195bd049 : SUCCESS in 58m 28s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/2fb06f1555ec4b5fab37438af42c7628 : SUCCESS in 1h 34m 24s (non-voting)\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/fc50d47042284f168974ee15f5319ed9 : SUCCESS in 1h 29m 50s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/6e49c83686424fd5978bebc0c5c7e4c7 : SUCCESS in 1h 03m 45s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/c1b74117f556451c83f8caa81bf91a34 : SUCCESS in 17m 36s","accounts_in_message":[],"_revision_number":116},{"id":"659090344a20719bd6fa2b1b9240c20ebfaf2907","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2022-10-20 11:15:52.000000000","message":"Patch Set 116: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":116},{"id":"180cb551583652aae2870af43fa4bfddf69f4fdb","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2022-10-20 11:47:25.000000000","message":"Patch Set 116: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":116},{"id":"0cac88f20514a02063a13e663cc38b1d13a37b13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 11:48:50.000000000","message":"Patch Set 116: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":116},{"id":"9b819b10eaa7a46b910c0ec9b9d322db20bb0d59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 15:10:16.000000000","message":"Patch Set 116: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36946ffe90574ea59fa8a7e84646d14c\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/479bc653314544d39012ae9ebc0709cd : SUCCESS in 2h 42m 52s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/53349f1a1c6646d49bdeb3726f133701 : SUCCESS in 40m 19s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/e05d1ab60a814577928269686802f8a8 : SUCCESS in 1h 37m 38s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/0918d267a7df489cb1336e520b5a24ae : SUCCESS in 1h 32m 50s\n- tripleo-ci-centos-9-content-provider-zed https://zuul.opendev.org/t/openstack/build/a80174d17f0f4b0581da5570e78cd94e : SUCCESS in 2h 38m 25s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/6e5f3b701fd6447b8edb5a427613b011 : SUCCESS in 2h 30m 40s\n- tripleo-ci-centos-9-containers-multinode-zed https://zuul.opendev.org/t/openstack/build/e17ee509e7c84fae9f222865f59b4089 : SUCCESS in 1h 34m 43s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/9914ee01bbec46a2be4b226b1fc5e6a7 : SUCCESS in 1h 31m 24s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/cd7f1fbad7a947babb53cc2a794e93da : SUCCESS in 1h 18m 18s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/8b9a147cf8f7445c93f87c1f67f1d053 : SUCCESS in 59m 45s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/cd956a81981045c6b78cdaf97b81cb29 : SUCCESS in 1h 27m 26s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/51af93bf4d3b4b918e339c5537cb5084 : SUCCESS in 5m 44s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/de95b43156ff48019ce8b8543da97f9d : SUCCESS in 19m 53s","accounts_in_message":[],"_revision_number":116},{"id":"854c987974d876b8424967ef3a5a4ff8d926aad9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 15:10:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":116}],"current_revision_number":116,"current_revision":"ccaabd8c1ef36a09e1077c809480cd2f7508b03e","revisions":{"84cc53b466b22fe0b6fca66bbc12f4563965fd1c":{"kind":"REWORK","_number":1,"created":"2022-05-30 10:33:53.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/1"}}},"commit":{"parents":[{"commit":"be5e48aeedf5ecd5ee4acdbed7c4fa1be2db0d50","subject":"Merge \"Push to local registry when building containers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/be5e48aeedf5ecd5ee4acdbed7c4fa1be2db0d50"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:33:49.000000000","tz":330},"subject":"[WIP] initial playbook to run deploy additional compute node","message":"[WIP] initial playbook to run deploy additional compute node\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/84cc53b466b22fe0b6fca66bbc12f4563965fd1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/84cc53b466b22fe0b6fca66bbc12f4563965fd1c"}]},"branch":"refs/heads/master"},"a8a6d4fd103e640a331113564553e9d2f9c38c32":{"kind":"REWORK","_number":2,"created":"2022-05-30 12:16:23.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/2"}}},"commit":{"parents":[{"commit":"be5e48aeedf5ecd5ee4acdbed7c4fa1be2db0d50","subject":"Merge \"Push to local registry when building containers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/be5e48aeedf5ecd5ee4acdbed7c4fa1be2db0d50"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 12:16:12.000000000","tz":330},"subject":"[WIP] initial playbook to run deploy additional compute node","message":"[WIP] initial playbook to run deploy additional compute node\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/a8a6d4fd103e640a331113564553e9d2f9c38c32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/a8a6d4fd103e640a331113564553e9d2f9c38c32"}]},"branch":"refs/heads/master"},"94c5d8d1f53b22b3ec882ccf8e39654f344ce961":{"kind":"REWORK","_number":3,"created":"2022-05-30 14:26:13.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/3"}}},"commit":{"parents":[{"commit":"be5e48aeedf5ecd5ee4acdbed7c4fa1be2db0d50","subject":"Merge \"Push to local registry when building containers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/be5e48aeedf5ecd5ee4acdbed7c4fa1be2db0d50"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 14:26:01.000000000","tz":330},"subject":"[WIP] initial playbook to run deploy additional compute node","message":"[WIP] initial playbook to run deploy additional compute node\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/94c5d8d1f53b22b3ec882ccf8e39654f344ce961"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/94c5d8d1f53b22b3ec882ccf8e39654f344ce961"}]},"branch":"refs/heads/master"},"548d0638e1a1ff3342ede5df05ac4e150b2936dd":{"kind":"REWORK","_number":4,"created":"2022-05-31 04:34:17.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/4"}}},"commit":{"parents":[{"commit":"be5e48aeedf5ecd5ee4acdbed7c4fa1be2db0d50","subject":"Merge \"Push to local registry when building containers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/be5e48aeedf5ecd5ee4acdbed7c4fa1be2db0d50"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-31 04:34:05.000000000","tz":330},"subject":"[WIP] initial playbook to run deploy additional compute node","message":"[WIP] initial playbook to run deploy additional compute node\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/548d0638e1a1ff3342ede5df05ac4e150b2936dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/548d0638e1a1ff3342ede5df05ac4e150b2936dd"}]},"branch":"refs/heads/master"},"19bfb30c322506aa805aab1c73ad85d45f6bcf64":{"kind":"REWORK","_number":5,"created":"2022-05-31 15:02:02.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/5"}}},"commit":{"parents":[{"commit":"be5e48aeedf5ecd5ee4acdbed7c4fa1be2db0d50","subject":"Merge \"Push to local registry when building containers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/be5e48aeedf5ecd5ee4acdbed7c4fa1be2db0d50"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-31 15:01:49.000000000","tz":330},"subject":"[WIP] initial playbook to run deploy additional compute node","message":"[WIP] initial playbook to run deploy additional compute node\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/19bfb30c322506aa805aab1c73ad85d45f6bcf64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/19bfb30c322506aa805aab1c73ad85d45f6bcf64"}]},"branch":"refs/heads/master"},"845d3eb1b1153e03d99753eac51f0986dfaa8163":{"kind":"REWORK","_number":6,"created":"2022-06-01 11:56:40.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/6"}}},"commit":{"parents":[{"commit":"be5e48aeedf5ecd5ee4acdbed7c4fa1be2db0d50","subject":"Merge \"Push to local registry when building containers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/be5e48aeedf5ecd5ee4acdbed7c4fa1be2db0d50"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-06-01 11:56:27.000000000","tz":330},"subject":"[WIP] initial playbook to run deploy additional compute node","message":"[WIP] initial playbook to run deploy additional compute node\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/845d3eb1b1153e03d99753eac51f0986dfaa8163"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/845d3eb1b1153e03d99753eac51f0986dfaa8163"}]},"branch":"refs/heads/master"},"da9c98b67608a56491ecc609238465e016a75c38":{"kind":"REWORK","_number":7,"created":"2022-06-02 03:28:36.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/7"}}},"commit":{"parents":[{"commit":"be5e48aeedf5ecd5ee4acdbed7c4fa1be2db0d50","subject":"Merge \"Push to local registry when building containers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/be5e48aeedf5ecd5ee4acdbed7c4fa1be2db0d50"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-06-02 03:28:22.000000000","tz":330},"subject":"[WIP] initial playbook to run deploy compute node","message":"[WIP] initial playbook to run deploy compute node\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/da9c98b67608a56491ecc609238465e016a75c38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/da9c98b67608a56491ecc609238465e016a75c38"}]},"branch":"refs/heads/master"},"150b9ba8b810d947a1dc856b6054743dc585a51d":{"kind":"REWORK","_number":8,"created":"2022-06-02 03:29:44.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/8"}}},"commit":{"parents":[{"commit":"be5e48aeedf5ecd5ee4acdbed7c4fa1be2db0d50","subject":"Merge \"Push to local registry when building containers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/be5e48aeedf5ecd5ee4acdbed7c4fa1be2db0d50"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-06-02 03:29:33.000000000","tz":330},"subject":"[WIP] initial playbook to run deploy compute node","message":"[WIP] initial playbook to run deploy compute node\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/150b9ba8b810d947a1dc856b6054743dc585a51d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/150b9ba8b810d947a1dc856b6054743dc585a51d"}]},"branch":"refs/heads/master"},"3e4ec781bb14f4ec48fdef45517a5c6c5cc009e5":{"kind":"REWORK","_number":9,"created":"2022-06-02 03:39:38.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/9"}}},"commit":{"parents":[{"commit":"be5e48aeedf5ecd5ee4acdbed7c4fa1be2db0d50","subject":"Merge \"Push to local registry when building containers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/be5e48aeedf5ecd5ee4acdbed7c4fa1be2db0d50"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-06-02 03:39:27.000000000","tz":330},"subject":"[WIP] initial playbook to run deploy compute node","message":"[WIP] initial playbook to run deploy compute node\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/3e4ec781bb14f4ec48fdef45517a5c6c5cc009e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/3e4ec781bb14f4ec48fdef45517a5c6c5cc009e5"}]},"branch":"refs/heads/master"},"ef3cf43e694987089c7df5690768d0942b7c34ea":{"kind":"REWORK","_number":10,"created":"2022-06-20 11:05:46.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/10"}}},"commit":{"parents":[{"commit":"be5e48aeedf5ecd5ee4acdbed7c4fa1be2db0d50","subject":"Merge \"Push to local registry when building containers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/be5e48aeedf5ecd5ee4acdbed7c4fa1be2db0d50"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-06-20 11:05:32.000000000","tz":330},"subject":"[WIP] initial playbook to run deploy compute node","message":"[WIP] initial playbook to run deploy compute node\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/ef3cf43e694987089c7df5690768d0942b7c34ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/ef3cf43e694987089c7df5690768d0942b7c34ea"}]},"branch":"refs/heads/master"},"6b8a7320b7e3099844f6342c7f3fc9a265061354":{"kind":"REWORK","_number":11,"created":"2022-07-07 06:40:13.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/11"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-07-07 06:39:53.000000000","tz":330},"subject":"[WIP] initial playbook to run deploy compute node","message":"[WIP] initial playbook to run deploy compute node\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/6b8a7320b7e3099844f6342c7f3fc9a265061354"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/6b8a7320b7e3099844f6342c7f3fc9a265061354"}]},"branch":"refs/heads/master"},"e53c1db0bfc7e53cbefd80a33ff625bbe85fd035":{"kind":"REWORK","_number":12,"created":"2022-07-07 07:34:49.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/12"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-07-07 07:34:39.000000000","tz":330},"subject":"[WIP] initial playbook to run deploy compute node","message":"[WIP] initial playbook to run deploy compute node\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/e53c1db0bfc7e53cbefd80a33ff625bbe85fd035"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/e53c1db0bfc7e53cbefd80a33ff625bbe85fd035"}]},"branch":"refs/heads/master"},"7fcdc3286ff130a727a3bcd82b30aa3af367838f":{"kind":"REWORK","_number":13,"created":"2022-07-07 08:35:30.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/13"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-07-07 08:35:21.000000000","tz":330},"subject":"[WIP] initial playbook to run deploy compute node","message":"[WIP] initial playbook to run deploy compute node\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7fcdc3286ff130a727a3bcd82b30aa3af367838f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7fcdc3286ff130a727a3bcd82b30aa3af367838f"}]},"branch":"refs/heads/master"},"559984d5180d4279b26af758abf1162c8c35a69f":{"kind":"REWORK","_number":14,"created":"2022-07-07 09:55:58.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/14"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-07-07 09:55:48.000000000","tz":330},"subject":"[WIP] initial playbook to run deploy compute node","message":"[WIP] initial playbook to run deploy compute node\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/559984d5180d4279b26af758abf1162c8c35a69f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/559984d5180d4279b26af758abf1162c8c35a69f"}]},"branch":"refs/heads/master"},"1db8119f37059ab82af134b190f30afb907ca404":{"kind":"REWORK","_number":15,"created":"2022-07-07 10:52:44.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/15"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-07-07 10:52:34.000000000","tz":330},"subject":"[WIP] initial playbook to run deploy compute node","message":"[WIP] initial playbook to run deploy compute node\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/1db8119f37059ab82af134b190f30afb907ca404"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/1db8119f37059ab82af134b190f30afb907ca404"}]},"branch":"refs/heads/master"},"e7fc4748a5b636042157c3156a5a97b481a5befd":{"kind":"REWORK","_number":16,"created":"2022-07-07 12:51:21.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/16"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-07-07 12:51:10.000000000","tz":330},"subject":"[WIP] initial playbook to run deploy compute node","message":"[WIP] initial playbook to run deploy compute node\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/e7fc4748a5b636042157c3156a5a97b481a5befd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/e7fc4748a5b636042157c3156a5a97b481a5befd"}]},"branch":"refs/heads/master"},"090d9ce80e100023a359ac64cfa8a74c9edd96db":{"kind":"REWORK","_number":17,"created":"2022-07-11 05:25:17.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/17"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-07-11 05:24:14.000000000","tz":330},"subject":"Initial playbook to Deploy deploy compute node","message":"Initial playbook to Deploy deploy compute node\n\nIt adds the playbook to deploy single compute node\nusing tripleo-ansible standalone playbooks on the host itself.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/090d9ce80e100023a359ac64cfa8a74c9edd96db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/090d9ce80e100023a359ac64cfa8a74c9edd96db"}]},"branch":"refs/heads/master"},"769108176ef37cbe5cdaaf480f396da7ffe954d0":{"kind":"REWORK","_number":18,"created":"2022-07-11 13:49:51.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/18"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-07-11 13:49:39.000000000","tz":330},"subject":"Initial playbook to Deploy deploy compute node","message":"Initial playbook to Deploy deploy compute node\n\nIt adds the playbook to deploy single compute node\nusing tripleo-ansible standalone playbooks on the host itself.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/769108176ef37cbe5cdaaf480f396da7ffe954d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/769108176ef37cbe5cdaaf480f396da7ffe954d0"}]},"branch":"refs/heads/master"},"9ad0d4a5d232eac1ab0e2ee02130e317047ff3e0":{"kind":"REWORK","_number":19,"created":"2022-07-13 10:12:23.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/19"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-07-13 10:11:31.000000000","tz":330},"subject":"Initial playbook to compute node using standalone ansible playbook","message":"Initial playbook to compute node using standalone ansible playbook\n\nIt adds the playbook to deploy single compute node\nusing tripleo-ansible standalone playbooks on the host itself.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/9ad0d4a5d232eac1ab0e2ee02130e317047ff3e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/9ad0d4a5d232eac1ab0e2ee02130e317047ff3e0"}]},"branch":"refs/heads/master"},"bc9b49c86230c961a91a315fe02a8a0330bb1eaf":{"kind":"REWORK","_number":20,"created":"2022-08-24 08:49:14.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/20"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-08-24 08:48:48.000000000","tz":330},"subject":"Initial playbook to compute node using standalone ansible playbook","message":"Initial playbook to compute node using standalone ansible playbook\n\nIt adds the playbook to deploy single compute node\nusing tripleo-ansible standalone playbooks on the host itself.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/bc9b49c86230c961a91a315fe02a8a0330bb1eaf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/bc9b49c86230c961a91a315fe02a8a0330bb1eaf"}]},"branch":"refs/heads/master"},"1a10d2dcc09293ec79c6c00209197dbe4fcc7d20":{"kind":"REWORK","_number":21,"created":"2022-08-24 12:06:14.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/21"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-08-24 12:05:22.000000000","tz":330},"subject":"Initial playbook to compute node using standalone ansible playbook","message":"Initial playbook to compute node using standalone ansible playbook\n\nIt adds the playbook to deploy single compute node\nusing tripleo-ansible standalone playbooks on the host itself.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/1a10d2dcc09293ec79c6c00209197dbe4fcc7d20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/1a10d2dcc09293ec79c6c00209197dbe4fcc7d20"}]},"branch":"refs/heads/master"},"a23718eed87b8b22733d9d9a28dcda872c31241e":{"kind":"REWORK","_number":22,"created":"2022-08-24 13:37:50.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/22"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-08-24 13:37:36.000000000","tz":330},"subject":"Initial playbook to compute node using standalone ansible playbook","message":"Initial playbook to compute node using standalone ansible playbook\n\nIt adds the playbook to deploy single compute node\nusing tripleo-ansible standalone playbooks on the host itself.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/a23718eed87b8b22733d9d9a28dcda872c31241e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/a23718eed87b8b22733d9d9a28dcda872c31241e"}]},"branch":"refs/heads/master"},"b248e9d84f7d9d39eec5430a4e72de7ed5fbca97":{"kind":"REWORK","_number":23,"created":"2022-08-24 16:26:00.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/23"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-08-24 16:25:49.000000000","tz":330},"subject":"Initial playbook to compute node using standalone ansible playbook","message":"Initial playbook to compute node using standalone ansible playbook\n\nIt adds the playbook to deploy single compute node\nusing tripleo-ansible standalone playbooks on the host itself.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/b248e9d84f7d9d39eec5430a4e72de7ed5fbca97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/b248e9d84f7d9d39eec5430a4e72de7ed5fbca97"}]},"branch":"refs/heads/master"},"cf4fff428892a95f99fa2b1cd3e28fa214dfa385":{"kind":"REWORK","_number":24,"created":"2022-08-25 04:53:31.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/24"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-08-25 04:53:19.000000000","tz":330},"subject":"Initial playbook to compute node using standalone ansible playbook","message":"Initial playbook to compute node using standalone ansible playbook\n\nIt adds the playbook to deploy single compute node\nusing tripleo-ansible standalone playbooks on the host itself.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/cf4fff428892a95f99fa2b1cd3e28fa214dfa385"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/cf4fff428892a95f99fa2b1cd3e28fa214dfa385"}]},"branch":"refs/heads/master"},"5e26801937ada41a8590d634e245cbed34484d76":{"kind":"REWORK","_number":25,"created":"2022-08-25 06:18:58.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/25"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-08-25 06:18:47.000000000","tz":330},"subject":"Initial playbook to compute node using standalone ansible playbook","message":"Initial playbook to compute node using standalone ansible playbook\n\nIt adds the playbook to deploy single compute node\nusing tripleo-ansible standalone playbooks on the host itself.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/5e26801937ada41a8590d634e245cbed34484d76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/5e26801937ada41a8590d634e245cbed34484d76"}]},"branch":"refs/heads/master"},"46fe8282486d49b499c6557023b602d57942e9ce":{"kind":"REWORK","_number":26,"created":"2022-08-25 08:07:18.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/26"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-08-25 08:07:07.000000000","tz":330},"subject":"Initial playbook to compute node using standalone ansible playbook","message":"Initial playbook to compute node using standalone ansible playbook\n\nIt adds the playbook to deploy single compute node\nusing tripleo-ansible standalone playbooks on the host itself.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/46fe8282486d49b499c6557023b602d57942e9ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/46fe8282486d49b499c6557023b602d57942e9ce"}]},"branch":"refs/heads/master"},"fb580f7c46e663748e7cd5c1f8c9157e85a7573d":{"kind":"REWORK","_number":27,"created":"2022-08-29 13:41:39.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/27"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-08-29 13:41:24.000000000","tz":330},"subject":"Initial playbook to compute node using standalone ansible playbook","message":"Initial playbook to compute node using standalone ansible playbook\n\nIt adds the playbook to deploy single compute node\nusing tripleo-ansible standalone playbooks on the host itself.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/fb580f7c46e663748e7cd5c1f8c9157e85a7573d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/fb580f7c46e663748e7cd5c1f8c9157e85a7573d"}]},"branch":"refs/heads/master"},"6160616d8fdcdea1e0cf35006bf7314f8bc88201":{"kind":"REWORK","_number":28,"created":"2022-08-30 04:46:38.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/28"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-08-30 04:46:24.000000000","tz":330},"subject":"Initial playbook to compute node using standalone ansible playbook","message":"Initial playbook to compute node using standalone ansible playbook\n\nIt adds the playbook to deploy single compute node\nusing tripleo-ansible standalone playbooks on the host itself.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/6160616d8fdcdea1e0cf35006bf7314f8bc88201"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/6160616d8fdcdea1e0cf35006bf7314f8bc88201"}]},"branch":"refs/heads/master"},"0bd625cdb3d696cb30eda644afbb40cee1dfc8eb":{"kind":"REWORK","_number":29,"created":"2022-08-30 12:50:45.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/29"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-08-30 12:50:26.000000000","tz":330},"subject":"Initial playbook to compute node using standalone ansible playbook","message":"Initial playbook to compute node using standalone ansible playbook\n\nIt adds the playbook to deploy single compute node\nusing tripleo-ansible standalone playbooks on the host itself.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/0bd625cdb3d696cb30eda644afbb40cee1dfc8eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/0bd625cdb3d696cb30eda644afbb40cee1dfc8eb"}]},"branch":"refs/heads/master"},"b281f516a3de12d63e2293ba2f8bbbcae7aade01":{"kind":"REWORK","_number":30,"created":"2022-08-30 15:04:29.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/30"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-08-30 15:04:17.000000000","tz":330},"subject":"Initial playbook to compute node using standalone ansible playbook","message":"Initial playbook to compute node using standalone ansible playbook\n\nIt adds the playbook to deploy single compute node\nusing tripleo-ansible standalone playbooks on the host itself.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/b281f516a3de12d63e2293ba2f8bbbcae7aade01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/b281f516a3de12d63e2293ba2f8bbbcae7aade01"}]},"branch":"refs/heads/master"},"c262928945b48dfa52d08f1443cac67257c619b3":{"kind":"REWORK","_number":31,"created":"2022-08-31 08:26:42.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/31"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-08-31 08:26:23.000000000","tz":330},"subject":"Initial playbook to compute node using standalone ansible playbook","message":"Initial playbook to compute node using standalone ansible playbook\n\nIt adds the playbook to deploy single compute node\nusing tripleo-ansible standalone playbooks on the host itself.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/c262928945b48dfa52d08f1443cac67257c619b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/c262928945b48dfa52d08f1443cac67257c619b3"}]},"branch":"refs/heads/master"},"32c5898aa57e4b671698343b475dc30922f6ac56":{"kind":"REWORK","_number":32,"created":"2022-09-01 12:36:22.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/32"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-09-01 12:36:04.000000000","tz":330},"subject":"Initial playbook to compute node using standalone ansible playbook","message":"Initial playbook to compute node using standalone ansible playbook\n\nIt adds the playbook to deploy single compute node\nusing tripleo-ansible standalone playbooks on the host itself.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/32c5898aa57e4b671698343b475dc30922f6ac56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/32c5898aa57e4b671698343b475dc30922f6ac56"}]},"branch":"refs/heads/master"},"75e2b03a1b7383f99b7e088d17dde491fc882208":{"kind":"REWORK","_number":33,"created":"2022-09-01 12:39:25.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/33"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-09-01 12:39:13.000000000","tz":330},"subject":"Initial playbook to compute node using standalone ansible playbook","message":"Initial playbook to compute node using standalone ansible playbook\n\nIt adds the playbook to deploy single compute node\nusing tripleo-ansible standalone playbooks on the host itself.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/75e2b03a1b7383f99b7e088d17dde491fc882208"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/75e2b03a1b7383f99b7e088d17dde491fc882208"}]},"branch":"refs/heads/master"},"15c8cada69af4dc203636bdc27312369fe573383":{"kind":"REWORK","_number":34,"created":"2022-09-01 12:54:21.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/34"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-09-01 12:54:09.000000000","tz":330},"subject":"Initial playbook to compute node using standalone ansible playbook","message":"Initial playbook to compute node using standalone ansible playbook\n\nIt adds the playbook to deploy single compute node\nusing tripleo-ansible standalone playbooks on the host itself.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/15c8cada69af4dc203636bdc27312369fe573383"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/15c8cada69af4dc203636bdc27312369fe573383"}]},"branch":"refs/heads/master"},"017563128268974a3e0b5f60339fa816ea1da1c4":{"kind":"REWORK","_number":35,"created":"2022-09-02 13:28:50.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/35"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-09-02 13:28:37.000000000","tz":330},"subject":"Initial playbook to compute node using standalone ansible playbook","message":"Initial playbook to compute node using standalone ansible playbook\n\nIt adds the playbook to deploy single compute node\nusing tripleo-ansible standalone playbooks on the host itself.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/017563128268974a3e0b5f60339fa816ea1da1c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/017563128268974a3e0b5f60339fa816ea1da1c4"}]},"branch":"refs/heads/master"},"32698ef5fa76d1c66e0a94db2033f15e314f4629":{"kind":"REWORK","_number":36,"created":"2022-09-05 06:28:03.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/36"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-09-05 06:27:49.000000000","tz":330},"subject":"Initial playbook to compute node using standalone ansible playbook","message":"Initial playbook to compute node using standalone ansible playbook\n\nIt adds the playbook to deploy single compute node\nusing tripleo-ansible standalone playbooks on the host itself.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/32698ef5fa76d1c66e0a94db2033f15e314f4629"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/32698ef5fa76d1c66e0a94db2033f15e314f4629"}]},"branch":"refs/heads/master"},"3a4abb454371d95d3c1a655e87b7503d18bcd0e0":{"kind":"REWORK","_number":37,"created":"2022-09-05 08:11:25.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/37"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-09-05 08:11:11.000000000","tz":330},"subject":"Initial playbook to compute node using standalone ansible playbook","message":"Initial playbook to compute node using standalone ansible playbook\n\nIt adds the playbook to deploy single compute node\nusing tripleo-ansible standalone playbooks on the host itself.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/3a4abb454371d95d3c1a655e87b7503d18bcd0e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/3a4abb454371d95d3c1a655e87b7503d18bcd0e0"}]},"branch":"refs/heads/master"},"e8e0f6c40e1fcf5ff6df46c91d0a2e9e145d304e":{"kind":"REWORK","_number":38,"created":"2022-09-05 10:53:25.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/38"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-09-05 10:53:13.000000000","tz":330},"subject":"Initial playbook to compute node using standalone ansible playbook","message":"Initial playbook to compute node using standalone ansible playbook\n\nIt adds the playbook to deploy single compute node\nusing tripleo-ansible standalone playbooks on the host itself.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/e8e0f6c40e1fcf5ff6df46c91d0a2e9e145d304e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/e8e0f6c40e1fcf5ff6df46c91d0a2e9e145d304e"}]},"branch":"refs/heads/master"},"cc585d8ce03fb29c210f90e3231ec98709a0fd1c":{"kind":"REWORK","_number":39,"created":"2022-09-05 14:00:14.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/39"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-09-05 14:00:01.000000000","tz":330},"subject":"Initial playbook to compute node using standalone ansible playbook","message":"Initial playbook to compute node using standalone ansible playbook\n\nIt adds the playbook to deploy single compute node\nusing tripleo-ansible standalone playbooks on the host itself.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/cc585d8ce03fb29c210f90e3231ec98709a0fd1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/cc585d8ce03fb29c210f90e3231ec98709a0fd1c"}]},"branch":"refs/heads/master"},"7f510adfc5108dda778a4859003d337777012ffb":{"kind":"REWORK","_number":40,"created":"2022-09-06 06:42:29.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/40"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-09-06 06:42:16.000000000","tz":330},"subject":"Initial playbook to compute node using standalone ansible playbook","message":"Initial playbook to compute node using standalone ansible playbook\n\nIt adds the playbook to deploy single compute node\nusing tripleo-ansible standalone playbooks on the host itself.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7f510adfc5108dda778a4859003d337777012ffb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7f510adfc5108dda778a4859003d337777012ffb"}]},"branch":"refs/heads/master"},"613e0a1fada213490c6e521ed2c5cfd70863fe4b":{"kind":"REWORK","_number":41,"created":"2022-09-06 09:00:50.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/41"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-09-06 09:00:33.000000000","tz":330},"subject":"Initial playbook to compute node using standalone ansible playbook","message":"Initial playbook to compute node using standalone ansible playbook\n\nIt adds the playbook to deploy single compute node\nusing tripleo-ansible standalone playbooks on the host itself.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/613e0a1fada213490c6e521ed2c5cfd70863fe4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/613e0a1fada213490c6e521ed2c5cfd70863fe4b"}]},"branch":"refs/heads/master"},"8357856f466b7b05228fa38da9c08b6718afb376":{"kind":"REWORK","_number":42,"created":"2022-09-13 10:18:57.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/42"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-09-13 10:18:40.000000000","tz":330},"subject":"Initial playbook to compute node using standalone ansible playbook","message":"Initial playbook to compute node using standalone ansible playbook\n\nIt adds the playbook to deploy single compute node\nusing tripleo-ansible standalone playbooks on the host itself.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/8357856f466b7b05228fa38da9c08b6718afb376"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/8357856f466b7b05228fa38da9c08b6718afb376"}]},"branch":"refs/heads/master"},"e9ba4859c29764d5cfcfd7746f75623b46fb114d":{"kind":"REWORK","_number":43,"created":"2022-09-14 04:37:50.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/43"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-09-14 04:37:35.000000000","tz":330},"subject":"Initial playbook to compute node using standalone ansible playbook","message":"Initial playbook to compute node using standalone ansible playbook\n\nIt adds the playbook to deploy single compute node\nusing tripleo-ansible standalone playbooks on the host itself.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/e9ba4859c29764d5cfcfd7746f75623b46fb114d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/e9ba4859c29764d5cfcfd7746f75623b46fb114d"}]},"branch":"refs/heads/master"},"5be4bef8637bf1d220fc976b73616423fbf8ce3d":{"kind":"REWORK","_number":44,"created":"2022-09-14 07:38:40.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/44"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-09-14 07:37:44.000000000","tz":330},"subject":"Initial playbook to compute node using standalone ansible playbook","message":"Initial playbook to compute node using standalone ansible playbook\n\nIt adds the playbook to deploy single compute node\nusing tripleo-ansible standalone playbooks on the host itself.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/5be4bef8637bf1d220fc976b73616423fbf8ce3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/5be4bef8637bf1d220fc976b73616423fbf8ce3d"}]},"branch":"refs/heads/master"},"d6e50ae0d43d1e1ea2aa5ec01952431179fe7a77":{"kind":"REWORK","_number":45,"created":"2022-09-14 11:14:48.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/45","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/45","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/45 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/45 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/45 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/45"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-09-14 11:14:35.000000000","tz":330},"subject":"Initial playbook to compute node using standalone ansible playbook","message":"Initial playbook to compute node using standalone ansible playbook\n\nIt adds the playbook to deploy single compute node\nusing tripleo-ansible standalone playbooks on the host itself.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/d6e50ae0d43d1e1ea2aa5ec01952431179fe7a77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/d6e50ae0d43d1e1ea2aa5ec01952431179fe7a77"}]},"branch":"refs/heads/master"},"daaf5d712928d397a9f33f81d261c82ad4202673":{"kind":"REWORK","_number":46,"created":"2022-09-14 15:44:54.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/46","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/46","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/46 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/46 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/46 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/46"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-09-14 15:44:37.000000000","tz":330},"subject":"Initial playbook to compute node using standalone ansible playbook","message":"Initial playbook to compute node using standalone ansible playbook\n\nIt adds the playbook to deploy single compute node\nusing tripleo-ansible standalone playbooks on the host itself.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/daaf5d712928d397a9f33f81d261c82ad4202673"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/daaf5d712928d397a9f33f81d261c82ad4202673"}]},"branch":"refs/heads/master"},"9ec6e6aaa9ebe1bfa52137e075238fe9b81cc658":{"kind":"REWORK","_number":47,"created":"2022-09-15 08:33:35.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/47","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/47","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/47 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/47 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/47 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/47"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-09-15 08:33:24.000000000","tz":330},"subject":"Initial playbook to compute node using standalone ansible playbook","message":"Initial playbook to compute node using standalone ansible playbook\n\nIt adds the playbook to deploy single compute node\nusing tripleo-ansible standalone playbooks on the host itself.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/9ec6e6aaa9ebe1bfa52137e075238fe9b81cc658"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/9ec6e6aaa9ebe1bfa52137e075238fe9b81cc658"}]},"branch":"refs/heads/master"},"aaa4d74cbf7abf25353d349d2530e90b736e50ef":{"kind":"REWORK","_number":48,"created":"2022-09-15 11:55:02.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/48","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/48","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/48 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/48 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/48 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/48"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"chandan kumar","email":"chkumar@redhat.com","date":"2022-09-15 11:55:02.000000000","tz":0},"subject":"Initial playbook to compute node using standalone ansible playbook","message":"Initial playbook to compute node using standalone ansible playbook\n\nIt adds the playbook to deploy single compute node\nusing tripleo-ansible standalone playbooks on the host itself.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/aaa4d74cbf7abf25353d349d2530e90b736e50ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/aaa4d74cbf7abf25353d349d2530e90b736e50ef"}]},"branch":"refs/heads/master"},"f568b9da7da89ba1dcd45937ae407bbe39a23052":{"kind":"REWORK","_number":49,"created":"2022-09-15 11:56:52.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/49","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/49","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/49 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/49 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/49 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/49"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"chandan kumar","email":"chkumar@redhat.com","date":"2022-09-15 11:56:52.000000000","tz":0},"subject":"Initial playbook to compute node using standalone ansible playbook","message":"Initial playbook to compute node using standalone ansible playbook\n\nIt adds the playbook to deploy single compute node\nusing tripleo-ansible standalone playbooks on the host itself.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/f568b9da7da89ba1dcd45937ae407bbe39a23052"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/f568b9da7da89ba1dcd45937ae407bbe39a23052"}]},"branch":"refs/heads/master"},"db0a1e3671fd444e6ef960ddd184171d1d558003":{"kind":"REWORK","_number":50,"created":"2022-09-16 05:10:20.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/50","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/50","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/50 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/50 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/50 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/50"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"chandan kumar","email":"chkumar@redhat.com","date":"2022-09-16 05:10:20.000000000","tz":0},"subject":"Initial playbook to compute node using standalone ansible playbook","message":"Initial playbook to compute node using standalone ansible playbook\n\nIt adds the playbook to deploy single compute node\nusing tripleo-ansible standalone playbooks on the host itself.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/db0a1e3671fd444e6ef960ddd184171d1d558003"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/db0a1e3671fd444e6ef960ddd184171d1d558003"}]},"branch":"refs/heads/master"},"dc91cf47d578fcc5f3b6cc7633bcc4a71785697e":{"kind":"REWORK","_number":51,"created":"2022-09-20 06:43:40.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/51","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/51","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/51 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/51 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/51 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/51"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-09-20 06:43:35.000000000","tz":330},"subject":"Add standalone-roles role","message":"Add standalone-roles role\n\nstandalone-roles is a new role to deploy standalone compute\nnode by using tripleo-ansible roles.\n\nThis role includes:\n- Generate the config from standalone deployment\n- Deploy the standalone compute on the standalone host itself.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/dc91cf47d578fcc5f3b6cc7633bcc4a71785697e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/dc91cf47d578fcc5f3b6cc7633bcc4a71785697e"}]},"branch":"refs/heads/master"},"392f4833710fd1e4af7ad6da2de62c135a3739e4":{"kind":"REWORK","_number":52,"created":"2022-09-20 07:04:39.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/52","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/52","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/52 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/52 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/52 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/52"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-09-20 07:04:25.000000000","tz":330},"subject":"Add standalone-roles role","message":"Add standalone-roles role\n\nstandalone-roles is a new role to deploy standalone compute\nnode by using tripleo-ansible roles.\n\nThis role includes:\n- Generate the config from standalone deployment\n- Deploy the standalone compute on the standalone host itself.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/392f4833710fd1e4af7ad6da2de62c135a3739e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/392f4833710fd1e4af7ad6da2de62c135a3739e4"}]},"branch":"refs/heads/master"},"a3df9a8e5fbb268cce7ce125caef31c76cc4c45a":{"kind":"REWORK","_number":53,"created":"2022-09-20 12:56:04.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/53","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/53","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/53 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/53 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/53 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/53"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-09-20 12:55:48.000000000","tz":330},"subject":"Add standalone-roles role","message":"Add standalone-roles role\n\nstandalone-roles is a new role to deploy standalone compute\nnode by using tripleo-ansible roles.\n\nThis role includes:\n- Generate the config from standalone deployment\n- Deploy the standalone compute on the standalone host itself.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/a3df9a8e5fbb268cce7ce125caef31c76cc4c45a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/a3df9a8e5fbb268cce7ce125caef31c76cc4c45a"}]},"branch":"refs/heads/master"},"f2d30efd4453a061a3dcc1b1545de6d8d6e1c25c":{"kind":"REWORK","_number":54,"created":"2022-10-04 06:59:43.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/54","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/54","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/54 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/54 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/54 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/54"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-04 06:57:00.000000000","tz":330},"subject":"Add standalone-roles role and its playbook","message":"Add standalone-roles role and its playbook\n\nstandalone-roles is a new role to deploy standalone compute\nnode by using tripleo-ansible roles.\n\nThis role includes:\n- Bootstrap the env for deployment\n- Deploy standalone when standalone_as_compute_node is set to true\n- Generate the config from standalone deployment\n- Deploy the compute services on the standalone host or on single node.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/f2d30efd4453a061a3dcc1b1545de6d8d6e1c25c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/f2d30efd4453a061a3dcc1b1545de6d8d6e1c25c"}]},"branch":"refs/heads/master"},"893583519c34a5a5edec25e8f9c9a0daca53bd3c":{"kind":"REWORK","_number":55,"created":"2022-10-04 07:27:14.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/55","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/55","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/55 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/55 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/55 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/55"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-04 07:27:03.000000000","tz":330},"subject":"Add standalone-roles role and its playbook","message":"Add standalone-roles role and its playbook\n\nstandalone-roles is a new role to deploy standalone compute\nnode by using tripleo-ansible roles.\n\nThis role includes:\n- Bootstrap the env for deployment\n- Deploy standalone when standalone_as_compute_node is set to true\n- Generate the config from standalone deployment\n- Deploy the compute services on the standalone host or on single node.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/893583519c34a5a5edec25e8f9c9a0daca53bd3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/893583519c34a5a5edec25e8f9c9a0daca53bd3c"}]},"branch":"refs/heads/master"},"c67752f552abd2c29ead20b2d9e7253f5bfde3d6":{"kind":"REWORK","_number":56,"created":"2022-10-04 10:42:51.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/56","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/56","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/56 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/56 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/56 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/56"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-04 10:42:40.000000000","tz":330},"subject":"Add standalone-roles role and its playbook","message":"Add standalone-roles role and its playbook\n\nstandalone-roles is a new role to deploy standalone compute\nnode by using tripleo-ansible roles.\n\nThis role includes:\n- Bootstrap the env for deployment\n- Deploy standalone when standalone_as_compute_node is set to true\n- Generate the config from standalone deployment\n- Deploy the compute services on the standalone host or on single node.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/c67752f552abd2c29ead20b2d9e7253f5bfde3d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/c67752f552abd2c29ead20b2d9e7253f5bfde3d6"}]},"branch":"refs/heads/master"},"6fb5bb1b263b9a1f60484a3d616d2c76cc30b9ee":{"kind":"REWORK","_number":57,"created":"2022-10-06 09:49:11.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/57","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/57","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/57 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/57 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/57 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/57"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-06 09:48:58.000000000","tz":330},"subject":"Add standalone-roles role and its playbook","message":"Add standalone-roles role and its playbook\n\nstandalone-roles is a new role to deploy standalone compute\nnode by using tripleo-ansible roles.\n\nThis role includes:\n- Bootstrap the env for deployment\n- Deploy standalone when standalone_as_compute_node is set to true\n- Generate the config from standalone deployment\n- Deploy the compute services on the standalone host or on single node.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/6fb5bb1b263b9a1f60484a3d616d2c76cc30b9ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/6fb5bb1b263b9a1f60484a3d616d2c76cc30b9ee"}]},"branch":"refs/heads/master"},"a2ade191ffce6e6a9124dee7153cc5f4135359c3":{"kind":"REWORK","_number":58,"created":"2022-10-06 11:46:53.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/58","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/58","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/58 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/58 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/58 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/58"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-06 11:46:41.000000000","tz":330},"subject":"Add standalone-roles role and its playbook","message":"Add standalone-roles role and its playbook\n\nstandalone-roles is a new role to deploy standalone compute\nnode by using tripleo-ansible roles.\n\nThis role includes:\n- Bootstrap the env for deployment\n- Deploy standalone when standalone_as_compute_node is set to true\n- Generate the config from standalone deployment\n- Deploy the compute services on the standalone host or on single node.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/a2ade191ffce6e6a9124dee7153cc5f4135359c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/a2ade191ffce6e6a9124dee7153cc5f4135359c3"}]},"branch":"refs/heads/master"},"f785be5e6229dd1dd60b45871519bc04bdde0ff5":{"kind":"REWORK","_number":59,"created":"2022-10-07 08:08:56.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/59","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/59","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/59 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/59 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/59 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/59"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-07 08:06:52.000000000","tz":330},"subject":"Add standalone-roles role and its playbook","message":"Add standalone-roles role and its playbook\n\nstandalone-roles is a new role to deploy standalone compute\nnode by using tripleo-ansible roles.\n\nThis role includes:\n- Bootstrap the env for deployment\n- Deploy standalone when standalone_as_compute_node is set to true\n- Generate the config from standalone deployment\n- Deploy the compute services on the standalone host or on single node.\n\nNote: container-prep is dropped as a dependency from standalone roles.\nAs it is not needed.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/f785be5e6229dd1dd60b45871519bc04bdde0ff5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/f785be5e6229dd1dd60b45871519bc04bdde0ff5"}]},"branch":"refs/heads/master"},"1b5017892a44dfcfe9ae7a9349b953ae2401aafa":{"kind":"REWORK","_number":60,"created":"2022-10-07 12:59:25.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/60","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/60","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/60 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/60 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/60 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/60"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-07 12:59:15.000000000","tz":330},"subject":"Add standalone-roles role and its playbook","message":"Add standalone-roles role and its playbook\n\nstandalone-roles is a new role to deploy standalone compute\nnode by using tripleo-ansible roles.\n\nThis role includes:\n- Bootstrap the env for deployment\n- Deploy standalone when standalone_as_compute_node is set to true\n- Generate the config from standalone deployment\n- Deploy the compute services on the standalone host or on single node.\n\nNote: container-prep is dropped as a dependency from standalone roles.\nAs it is not needed.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/1b5017892a44dfcfe9ae7a9349b953ae2401aafa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/1b5017892a44dfcfe9ae7a9349b953ae2401aafa"}]},"branch":"refs/heads/master"},"bb834dc9e7f086c6371325f31023244f4b943dcc":{"kind":"REWORK","_number":61,"created":"2022-10-07 14:47:09.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/61","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/61","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/61 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/61 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/61 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/61"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-07 14:47:00.000000000","tz":330},"subject":"Add standalone-roles role and its playbook","message":"Add standalone-roles role and its playbook\n\nstandalone-roles is a new role to deploy standalone compute\nnode by using tripleo-ansible roles.\n\nThis role includes:\n- Bootstrap the env for deployment\n- Deploy standalone when standalone_as_compute_node is set to true\n- Generate the config from standalone deployment\n- Deploy the compute services on the standalone host or on single node.\n\nNote: container-prep is dropped as a dependency from standalone roles.\nAs it is not needed.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/bb834dc9e7f086c6371325f31023244f4b943dcc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/bb834dc9e7f086c6371325f31023244f4b943dcc"}]},"branch":"refs/heads/master"},"eab7cc268aa9d2cf30e43859caad9e20237c2027":{"kind":"REWORK","_number":62,"created":"2022-10-09 10:37:56.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/62","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/62","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/62 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/62 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/62 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/62"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-09 10:37:52.000000000","tz":330},"subject":"Add standalone-roles role and its playbook","message":"Add standalone-roles role and its playbook\n\nstandalone-roles is a new role to deploy standalone compute\nnode by using tripleo-ansible roles.\n\nThis role includes:\n- Bootstrap the env for deployment\n- Deploy standalone when standalone_as_compute_node is set to true\n- Generate the config from standalone deployment\n- Deploy the compute services on the standalone host or on single node.\n\nNote: container-prep is dropped as a dependency from standalone roles.\nAs it is not needed.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/eab7cc268aa9d2cf30e43859caad9e20237c2027"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/eab7cc268aa9d2cf30e43859caad9e20237c2027"}]},"branch":"refs/heads/master"},"48d5449f4497cdb7e66509662bbdf94e347d68d8":{"kind":"REWORK","_number":63,"created":"2022-10-09 11:16:43.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/63","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/63","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/63 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/63 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/63 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/63"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-09 11:16:32.000000000","tz":330},"subject":"Add standalone-roles role and its playbook","message":"Add standalone-roles role and its playbook\n\nstandalone-roles is a new role to deploy standalone compute\nnode by using tripleo-ansible roles.\n\nThis role includes:\n- Bootstrap the env for deployment\n- Deploy standalone when standalone_as_compute_node is set to true\n- Generate the config from standalone deployment\n- Deploy the compute services on the standalone host or on single node.\n\nNote: container-prep is dropped as a dependency from standalone roles.\nAs it is not needed.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/48d5449f4497cdb7e66509662bbdf94e347d68d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/48d5449f4497cdb7e66509662bbdf94e347d68d8"}]},"branch":"refs/heads/master"},"31f66bea34156300527b0db5835efe3698b18f72":{"kind":"REWORK","_number":64,"created":"2022-10-09 11:46:50.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/64","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/64","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/64 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/64 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/64 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/64"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-09 11:46:37.000000000","tz":330},"subject":"Add standalone-roles role and its playbook","message":"Add standalone-roles role and its playbook\n\nstandalone-roles is a new role to deploy standalone compute\nnode by using tripleo-ansible roles.\n\nThis role includes:\n- Bootstrap the env for deployment\n- Deploy standalone when standalone_as_compute_node is set to true\n- Generate the config from standalone deployment\n- Deploy the compute services on the standalone host or on single node.\n\nNote: container-prep is dropped as a dependency from standalone roles.\nAs it is not needed.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/31f66bea34156300527b0db5835efe3698b18f72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/31f66bea34156300527b0db5835efe3698b18f72"}]},"branch":"refs/heads/master"},"afd3a66dffda11231b3351bbbb13f48edf2c121d":{"kind":"REWORK","_number":65,"created":"2022-10-10 06:46:13.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/65","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/65","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/65 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/65 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/65 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/65"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-10 06:45:59.000000000","tz":330},"subject":"Add standalone-roles role and its playbook","message":"Add standalone-roles role and its playbook\n\nstandalone-roles is a new role to deploy standalone compute\nnode by using tripleo-ansible roles.\n\nThis role includes:\n- Bootstrap the env for deployment\n- Deploy standalone when compute_on_same_host is set to true\n- Generate the config from standalone deployment\n- Deploy the compute services on the standalone host or on single node.\n- Generate the required inventory for target host when\n  compute_on_target_host is set to true\n- Deploy compute services on target host\n\nIt will be used in tripleo-external-compute-deployment-pipeline\njob template[1]\n\n[1]. https://review.opendev.org/c/openstack/tripleo-ci/+/843836\n\nNote: container-prep is dropped as a dependency from standalone roles.\nAs it is not needed.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/afd3a66dffda11231b3351bbbb13f48edf2c121d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/afd3a66dffda11231b3351bbbb13f48edf2c121d"}]},"branch":"refs/heads/master"},"6e35cac76d07521fd020750ec68ac8bfd5135975":{"kind":"REWORK","_number":66,"created":"2022-10-10 06:48:09.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/66","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/66","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/66 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/66 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/66 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/66"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-10 06:47:55.000000000","tz":330},"subject":"Add standalone-roles role and its playbook","message":"Add standalone-roles role and its playbook\n\nstandalone-roles is a new role to deploy standalone compute\nnode by using tripleo-ansible roles.\n\nThis role includes:\n- Bootstrap the env for deployment\n- Deploy standalone when compute_on_same_host is set to true\n- Generate the config from standalone deployment\n- Deploy the compute services on the standalone host or on single node.\n- Generate the required inventory for target host when\n  compute_on_target_host is set to true\n- Deploy compute services on target host\n\nIt will be used in tripleo-external-compute-deployment-pipeline\njob template[1]\n\n[1]. https://review.opendev.org/c/openstack/tripleo-ci/+/843836\n\nNote: container-prep is dropped as a dependency from standalone roles.\nAs it is not needed.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/6e35cac76d07521fd020750ec68ac8bfd5135975"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/6e35cac76d07521fd020750ec68ac8bfd5135975"}]},"branch":"refs/heads/master"},"bae3391e22fc90587c4f03ca11c536cf073a56a3":{"kind":"REWORK","_number":67,"created":"2022-10-10 07:09:55.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/67","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/67","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/67 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/67 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/67 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/67"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-10 07:09:20.000000000","tz":330},"subject":"Add standalone-roles role and its playbook","message":"Add standalone-roles role and its playbook\n\nstandalone-roles is a new role to deploy standalone compute\nnode by using tripleo-ansible roles.\n\nThis role includes:\n- Bootstrap the env for deployment\n- Deploy standalone when deploy_standalone is set to true\n- Generate the config from standalone deployment\n- Deploy the compute services on the standalone host or on single node.\n- Generate the required inventory for target host when\n  compute_on_target_host is set to true\n- Deploy compute services on target host\n\nIt will be used in tripleo-external-compute-deployment-pipeline\njob template[1]\n\n[1]. https://review.opendev.org/c/openstack/tripleo-ci/+/843836\n\nNote: container-prep is dropped as a dependency from standalone roles.\nAs it is not needed.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/bae3391e22fc90587c4f03ca11c536cf073a56a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/bae3391e22fc90587c4f03ca11c536cf073a56a3"}]},"branch":"refs/heads/master"},"aad8b0866f743c2e9e458688564dc88955fba69e":{"kind":"REWORK","_number":68,"created":"2022-10-10 07:21:22.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/68","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/68","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/68 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/68 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/68 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/68"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-10 07:20:48.000000000","tz":330},"subject":"Add standalone-roles role and its playbook","message":"Add standalone-roles role and its playbook\n\nstandalone-roles is a new role to deploy standalone compute\nnode by using tripleo-ansible roles.\n\nThis role includes:\n- Bootstrap the env for deployment\n- Deploy standalone when deploy_standalone is set to true\n- Generate the config from standalone deployment\n- Deploy the compute services on the standalone host or on single node.\n- Generate the required inventory for target host when\n  compute_on_target_host is set to true\n- Deploy compute services on target host\n\nIt will be used in tripleo-external-compute-deployment-pipeline\njob template[1]\n\n[1]. https://review.opendev.org/c/openstack/tripleo-ci/+/843836\n\nNote: container-prep is dropped as a dependency from standalone roles.\nAs it is not needed.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/aad8b0866f743c2e9e458688564dc88955fba69e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/aad8b0866f743c2e9e458688564dc88955fba69e"}]},"branch":"refs/heads/master"},"39fc50d0d516d26850562f8ec32ccacd075d7282":{"kind":"REWORK","_number":69,"created":"2022-10-10 07:55:33.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/69","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/69","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/69 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/69 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/69 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/69"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-10 07:55:22.000000000","tz":330},"subject":"Add standalone-roles role and its playbook","message":"Add standalone-roles role and its playbook\n\nstandalone-roles is a new role to deploy standalone compute\nnode by using tripleo-ansible roles.\n\nThis role includes:\n- Bootstrap the env for deployment\n- Deploy standalone when deploy_standalone is set to true\n- Generate the config from standalone deployment\n- Deploy the compute services on the standalone host or on single node.\n- Generate the required inventory for target host when\n  compute_on_target_host is set to true\n- Deploy compute services on target host\n\nIt will be used in tripleo-external-compute-deployment-pipeline\njob template[1]\n\n[1]. https://review.opendev.org/c/openstack/tripleo-ci/+/843836\n\nNote: container-prep is dropped as a dependency from standalone roles.\nAs it is not needed.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/39fc50d0d516d26850562f8ec32ccacd075d7282"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/39fc50d0d516d26850562f8ec32ccacd075d7282"}]},"branch":"refs/heads/master"},"c9c9188723e590637be116dd4b4cd6ce723be63c":{"kind":"REWORK","_number":70,"created":"2022-10-10 08:17:06.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/70","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/70","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/70 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/70 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/70 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/70"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-10 08:16:55.000000000","tz":330},"subject":"Add standalone-roles role and its playbook","message":"Add standalone-roles role and its playbook\n\nstandalone-roles is a new role to deploy standalone compute\nnode by using tripleo-ansible roles.\n\nThis role includes:\n- Bootstrap the env for deployment\n- Deploy standalone when deploy_standalone is set to true\n- Generate the config from standalone deployment\n- Deploy the compute services on the standalone host or on single node.\n- Generate the required inventory for target host when\n  compute_on_target_host is set to true\n- Deploy compute services on target host\n\nIt will be used in tripleo-external-compute-deployment-pipeline\njob template[1]\n\n[1]. https://review.opendev.org/c/openstack/tripleo-ci/+/843836\n\nNote: container-prep is dropped as a dependency from standalone roles.\nAs it is not needed.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/c9c9188723e590637be116dd4b4cd6ce723be63c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/c9c9188723e590637be116dd4b4cd6ce723be63c"}]},"branch":"refs/heads/master"},"f443b4eccafaac4278648482f62abff972fc76c5":{"kind":"REWORK","_number":71,"created":"2022-10-10 08:57:39.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/71","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/71","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/71 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/71 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/71 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/71"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-10 08:57:27.000000000","tz":330},"subject":"Add standalone-roles role and its playbook","message":"Add standalone-roles role and its playbook\n\nstandalone-roles is a new role to deploy standalone compute\nnode by using tripleo-ansible roles.\n\nThis role includes:\n- Bootstrap the env for deployment\n- Deploy standalone when deploy_standalone is set to true\n- Generate the config from standalone deployment\n- Deploy the compute services on the standalone host or on single node.\n- Generate the required inventory for target host when\n  compute_on_target_host is set to true\n- Deploy compute services on target host\n\nIt will be used in tripleo-external-compute-deployment-pipeline\njob template[1]\n\n[1]. https://review.opendev.org/c/openstack/tripleo-ci/+/843836\n\nNote: container-prep is dropped as a dependency from standalone roles.\nAs it is not needed.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/f443b4eccafaac4278648482f62abff972fc76c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/f443b4eccafaac4278648482f62abff972fc76c5"}]},"branch":"refs/heads/master"},"7be34813232c3dc0a58b20647d60705cf0d27f89":{"kind":"REWORK","_number":72,"created":"2022-10-10 09:22:21.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/72","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/72","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/72 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/72 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/72 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/72"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-10 09:22:10.000000000","tz":330},"subject":"Add standalone-roles role and its playbook","message":"Add standalone-roles role and its playbook\n\nstandalone-roles is a new role to deploy standalone compute\nnode by using tripleo-ansible roles.\n\nThis role includes:\n- Bootstrap the env for deployment\n- Deploy standalone when deploy_standalone is set to true\n- Generate the config from standalone deployment\n- Deploy the compute services on the standalone host or on single node.\n- Generate the required inventory for target host when\n  compute_on_target_host is set to true\n- Deploy compute services on target host\n\nIt will be used in tripleo-external-compute-deployment-pipeline\njob template[1]\n\n[1]. https://review.opendev.org/c/openstack/tripleo-ci/+/843836\n\nNote: container-prep is dropped as a dependency from standalone roles.\nAs it is not needed.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7be34813232c3dc0a58b20647d60705cf0d27f89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7be34813232c3dc0a58b20647d60705cf0d27f89"}]},"branch":"refs/heads/master"},"c6ae0f17a3000e0edcde46f5ec32976a0cbe37d8":{"kind":"REWORK","_number":73,"created":"2022-10-10 11:31:55.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/73","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/73","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/73 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/73 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/73 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/73"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-10 11:31:44.000000000","tz":330},"subject":"Add standalone-roles role and its playbook","message":"Add standalone-roles role and its playbook\n\nstandalone-roles is a new role to deploy standalone compute\nnode by using tripleo-ansible roles.\n\nThis role includes:\n- Bootstrap the env for deployment\n- Deploy standalone when deploy_standalone is set to true\n- Generate the config from standalone deployment\n- Deploy the compute services on the standalone host or on single node.\n- Generate the required inventory for target host when\n  compute_on_target_host is set to true\n- Deploy compute services on target host\n\nIt will be used in tripleo-external-compute-deployment-pipeline\njob template[1]\n\n[1]. https://review.opendev.org/c/openstack/tripleo-ci/+/843836\n\nNote: container-prep is dropped as a dependency from standalone roles.\nAs it is not needed.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/c6ae0f17a3000e0edcde46f5ec32976a0cbe37d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/c6ae0f17a3000e0edcde46f5ec32976a0cbe37d8"}]},"branch":"refs/heads/master"},"dbe4c9f15629bbbbfb043f278c0f5c5df3f3585f":{"kind":"REWORK","_number":74,"created":"2022-10-10 12:49:14.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/74","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/74","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/74 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/74 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/74 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/74"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-10 12:49:03.000000000","tz":330},"subject":"Add standalone-roles role and its playbook","message":"Add standalone-roles role and its playbook\n\nstandalone-roles is a new role to deploy standalone compute\nnode by using tripleo-ansible roles.\n\nThis role includes:\n- Bootstrap the env for deployment\n- Deploy standalone when deploy_standalone is set to true\n- Generate the config from standalone deployment\n- Deploy the compute services on the standalone host or on single node.\n- Generate the required inventory for target host when\n  compute_on_target_host is set to true\n- Deploy compute services on target host\n\nIt will be used in tripleo-external-compute-deployment-pipeline\njob template[1]\n\n[1]. https://review.opendev.org/c/openstack/tripleo-ci/+/843836\n\nNote: container-prep is dropped as a dependency from standalone roles.\nAs it is not needed.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/dbe4c9f15629bbbbfb043f278c0f5c5df3f3585f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/dbe4c9f15629bbbbfb043f278c0f5c5df3f3585f"}]},"branch":"refs/heads/master"},"7a57b06cf7607b99cc9520adbc7bb1fda0e11eb1":{"kind":"REWORK","_number":75,"created":"2022-10-11 04:12:31.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/75","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/75","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/75 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/75 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/75 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/75"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-11 04:12:20.000000000","tz":330},"subject":"Add standalone-roles role and its playbook","message":"Add standalone-roles role and its playbook\n\nstandalone-roles is a new role to deploy standalone compute\nnode by using tripleo-ansible roles.\n\nThis role includes:\n- Bootstrap the env for deployment\n- Deploy standalone when deploy_standalone is set to true\n- Generate the config from standalone deployment\n- Deploy the compute services on the standalone host or on single node.\n- Generate the required inventory for target host when\n  compute_on_target_host is set to true\n- Deploy compute services on target host\n\nIt will be used in tripleo-external-compute-deployment-pipeline\njob template[1]\n\n[1]. https://review.opendev.org/c/openstack/tripleo-ci/+/843836\n\nNote: container-prep is dropped as a dependency from standalone roles.\nAs it is not needed.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7a57b06cf7607b99cc9520adbc7bb1fda0e11eb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7a57b06cf7607b99cc9520adbc7bb1fda0e11eb1"}]},"branch":"refs/heads/master"},"a55c834dc1b8ba2251238021d56a143b9b428d14":{"kind":"REWORK","_number":76,"created":"2022-10-11 08:34:50.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/76","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/76","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/76 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/76 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/76 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/76"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-11 08:34:39.000000000","tz":330},"subject":"Add standalone-roles role and its playbook","message":"Add standalone-roles role and its playbook\n\nstandalone-roles is a new role to deploy standalone compute\nnode by using tripleo-ansible roles.\n\nThis role includes:\n- Bootstrap the env for deployment\n- Deploy standalone when deploy_standalone is set to true\n- Generate the config from standalone deployment\n- Deploy the compute services on the standalone host or on single node.\n- Generate the required inventory for target host when\n  compute_on_target_host is set to true\n- Deploy compute services on target host\n\nIt will be used in tripleo-external-compute-deployment-pipeline\njob template[1]\n\n[1]. https://review.opendev.org/c/openstack/tripleo-ci/+/843836\n\nNote: container-prep is dropped as a dependency from standalone roles.\nAs it is not needed.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/a55c834dc1b8ba2251238021d56a143b9b428d14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/a55c834dc1b8ba2251238021d56a143b9b428d14"}]},"branch":"refs/heads/master"},"32a329a1db5cd81c17dab5e494defa2b4c41c097":{"kind":"REWORK","_number":77,"created":"2022-10-11 10:13:34.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/77","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/77","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/77 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/77 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/77 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/77"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-11 10:13:19.000000000","tz":330},"subject":"Add standalone-roles role and its playbook","message":"Add standalone-roles role and its playbook\n\nstandalone-roles is a new role to deploy standalone compute\nnode by using tripleo-ansible roles.\n\nThis role includes:\n- Bootstrap the env for deployment\n- Deploy standalone when deploy_standalone is set to true\n- Generate the config from standalone deployment\n- Deploy the compute services on the standalone host or on single node.\n- Generate the required inventory for target host when\n  compute_on_target_host is set to true\n- Deploy compute services on target host\n\nIt will be used in tripleo-external-compute-deployment-pipeline\njob template[1]\n\n[1]. https://review.opendev.org/c/openstack/tripleo-ci/+/843836\n\nNote: container-prep is dropped as a dependency from standalone roles.\nAs it is not needed.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/32a329a1db5cd81c17dab5e494defa2b4c41c097"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/32a329a1db5cd81c17dab5e494defa2b4c41c097"}]},"branch":"refs/heads/master"},"eaed0ebfe96b0dc5822ab4611000c59436ccdbf1":{"kind":"REWORK","_number":78,"created":"2022-10-11 11:13:24.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/78","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/78","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/78 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/78 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/78 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/78"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-11 11:13:08.000000000","tz":330},"subject":"Add standalone-roles role and its playbook","message":"Add standalone-roles role and its playbook\n\nstandalone-roles is a new role to deploy standalone compute\nnode by using tripleo-ansible roles.\n\nThis role includes:\n- Bootstrap the env for deployment\n- Deploy standalone when deploy_standalone is set to true\n- Generate the config from standalone deployment\n- Deploy the compute services on the standalone host or on single node.\n- Generate the required inventory for target host when\n  compute_on_target_host is set to true\n- Deploy compute services on target host\n\nIt will be used in tripleo-external-compute-deployment-pipeline\njob template[1]\n\n[1]. https://review.opendev.org/c/openstack/tripleo-ci/+/843836\n\nNote: container-prep is dropped as a dependency from standalone roles.\nAs it is not needed.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/eaed0ebfe96b0dc5822ab4611000c59436ccdbf1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/eaed0ebfe96b0dc5822ab4611000c59436ccdbf1"}]},"branch":"refs/heads/master"},"5bca5b1d0e04e548f3542a2d39d8bfce2edb5464":{"kind":"REWORK","_number":79,"created":"2022-10-11 13:09:07.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/79","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/79","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/79 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/79 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/79 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/79"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-11 13:08:11.000000000","tz":330},"subject":"Add standalone-roles role and its playbook","message":"Add standalone-roles role and its playbook\n\nstandalone-roles is a new role to deploy standalone compute\nnode by using tripleo-ansible roles.\n\nThis role includes:\n- Bootstrap the env for deployment\n- Deploy standalone when deploy_standalone is set to true\n- Generate the config from standalone deployment\n- Deploy the compute services on the standalone host or on single node.\n- Generate the required inventory for target host when\n  compute_on_target_host is set to true\n- Deploy compute services on target host\n\nIt will be used in tripleo-external-compute-deployment-pipeline\njob template[1]\n\n[1]. https://review.opendev.org/c/openstack/tripleo-ci/+/843836\n\nNote: container-prep is dropped as a dependency from standalone roles.\nAs it is not needed.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/5bca5b1d0e04e548f3542a2d39d8bfce2edb5464"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/5bca5b1d0e04e548f3542a2d39d8bfce2edb5464"}]},"branch":"refs/heads/master"},"52bef5f62d2fa14c3f59d5a5d07a50a8730d280b":{"kind":"REWORK","_number":80,"created":"2022-10-12 01:02:14.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/80","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/80","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/80 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/80 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/80 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/80"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-12 01:02:03.000000000","tz":330},"subject":"Add standalone-roles role and its playbook","message":"Add standalone-roles role and its playbook\n\nstandalone-roles is a new role to deploy standalone compute\nnode by using tripleo-ansible roles.\n\nThis role includes:\n- Bootstrap the env for deployment\n- Deploy standalone when deploy_standalone is set to true\n- Generate the config from standalone deployment\n- Deploy the compute services on the standalone host or on single node.\n- Generate the required inventory for target host when\n  compute_on_target_host is set to true\n- Deploy compute services on target host\n\nIt will be used in tripleo-external-compute-deployment-pipeline\njob template[1]\n\n[1]. https://review.opendev.org/c/openstack/tripleo-ci/+/843836\n\nNote: container-prep is dropped as a dependency from standalone roles.\nAs it is not needed.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/52bef5f62d2fa14c3f59d5a5d07a50a8730d280b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/52bef5f62d2fa14c3f59d5a5d07a50a8730d280b"}]},"branch":"refs/heads/master"},"0ef2fac49bd23954b03b5893591d35e16562b3f6":{"kind":"REWORK","_number":81,"created":"2022-10-12 02:10:00.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/81","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/81","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/81 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/81 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/81 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/81"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-12 02:09:48.000000000","tz":330},"subject":"Add standalone-roles role and its playbook","message":"Add standalone-roles role and its playbook\n\nstandalone-roles is a new role to deploy standalone compute\nnode by using tripleo-ansible roles.\n\nThis role includes:\n- Bootstrap the env for deployment\n- Deploy standalone when deploy_standalone is set to true\n- Generate the config from standalone deployment\n- Deploy the compute services on the standalone host or on single node.\n- Generate the required inventory for target host when\n  compute_on_target_host is set to true\n- Deploy compute services on target host\n\nIt will be used in tripleo-external-compute-deployment-pipeline\njob template[1]\n\n[1]. https://review.opendev.org/c/openstack/tripleo-ci/+/843836\n\nNote: container-prep is dropped as a dependency from standalone roles.\nAs it is not needed.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/0ef2fac49bd23954b03b5893591d35e16562b3f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/0ef2fac49bd23954b03b5893591d35e16562b3f6"}]},"branch":"refs/heads/master"},"b0cc54fa451c5aed068cc80591f60bceaef05135":{"kind":"REWORK","_number":82,"created":"2022-10-12 06:38:39.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/82","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/82","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/82 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/82 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/82 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/82"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-12 06:38:27.000000000","tz":330},"subject":"Add standalone-roles role and its playbook","message":"Add standalone-roles role and its playbook\n\nstandalone-roles is a new role to deploy standalone compute\nnode by using tripleo-ansible roles.\n\nThis role includes:\n- Bootstrap the env for deployment\n- Deploy standalone when deploy_standalone is set to true\n- Generate the config from standalone deployment\n- Deploy the compute services on the standalone host or on single node.\n- Generate the required inventory for target host when\n  compute_on_target_host is set to true\n- Deploy compute services on target host\n\nIt will be used in tripleo-external-compute-deployment-pipeline\njob template[1]\n\n[1]. https://review.opendev.org/c/openstack/tripleo-ci/+/843836\n\nNote: container-prep is dropped as a dependency from standalone roles.\nAs it is not needed.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/b0cc54fa451c5aed068cc80591f60bceaef05135"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/b0cc54fa451c5aed068cc80591f60bceaef05135"}]},"branch":"refs/heads/master"},"551fd4666738298f82dcb9d7da8e344a4d814309":{"kind":"REWORK","_number":83,"created":"2022-10-12 06:39:36.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/83","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/83","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/83 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/83 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/83 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/83"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-12 06:39:08.000000000","tz":330},"subject":"Add standalone-roles role and its playbook","message":"Add standalone-roles role and its playbook\n\nstandalone-roles is a new role to deploy standalone compute\nnode by using tripleo-ansible roles.\n\nThis role includes:\n- Bootstrap the env for deployment\n- Deploy standalone when deploy_standalone is set to true\n- Generate the config from standalone deployment\n- Deploy the compute services on the standalone host or on single node.\n- Generate the required inventory for target host when\n  compute_on_target_host is set to true\n- Deploy compute services on target host\n\nIt will be used in tripleo-external-compute-deployment-pipeline\njob template[1]\n\n[1]. https://review.opendev.org/c/openstack/tripleo-ci/+/843836\n\nNote: container-prep is dropped as a dependency from standalone roles.\nAs it is not needed.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/551fd4666738298f82dcb9d7da8e344a4d814309"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/551fd4666738298f82dcb9d7da8e344a4d814309"}]},"branch":"refs/heads/master"},"0c2a7513cae4ea13c2b45896a99d59611fe387d9":{"kind":"REWORK","_number":84,"created":"2022-10-12 06:46:08.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/84","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/84","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/84 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/84 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/84 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/84"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-12 06:45:47.000000000","tz":330},"subject":"Add standalone-roles role and its playbook","message":"Add standalone-roles role and its playbook\n\nstandalone-roles is a new role to deploy standalone compute\nnode by using tripleo-ansible roles.\n\nThis role includes:\n- Bootstrap the env for deployment\n- Deploy standalone when deploy_standalone is set to true\n- Generate the config from standalone deployment\n- Deploy the compute services on the standalone host or on single node.\n- Generate the required inventory for target host when\n  compute_on_target_host is set to true\n- Deploy compute services on target host\n\nIt will be used in tripleo-external-compute-deployment-pipeline\njob template[1]\n\n[1]. https://review.opendev.org/c/openstack/tripleo-ci/+/843836\n\nNote: container-prep is dropped as a dependency from standalone roles.\nAs it is not needed.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/0c2a7513cae4ea13c2b45896a99d59611fe387d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/0c2a7513cae4ea13c2b45896a99d59611fe387d9"}]},"branch":"refs/heads/master"},"c337c3971e417daa8556e6b18e9015de109e23e6":{"kind":"REWORK","_number":85,"created":"2022-10-12 06:48:46.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/85","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/85","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/85 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/85 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/85 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/85"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-12 06:48:35.000000000","tz":330},"subject":"Add standalone-roles role and its playbook","message":"Add standalone-roles role and its playbook\n\nstandalone-roles is a new role to deploy standalone compute\nnode by using tripleo-ansible roles.\n\nThis role includes:\n- Bootstrap the env for deployment\n- Deploy standalone when deploy_standalone is set to true\n- Generate the config from standalone deployment\n- Deploy the compute services on the standalone host or on single node.\n- Generate the required inventory for target host when\n  compute_on_target_host is set to true\n- Deploy compute services on target host\n\nIt will be used in tripleo-external-compute-deployment-pipeline\njob template[1]\n\n[1]. https://review.opendev.org/c/openstack/tripleo-ci/+/843836\n\nNote: container-prep is dropped as a dependency from standalone roles.\nAs it is not needed.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/c337c3971e417daa8556e6b18e9015de109e23e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/c337c3971e417daa8556e6b18e9015de109e23e6"}]},"branch":"refs/heads/master"},"13c32a634f431da2be06349c5d43cf4f4c473661":{"kind":"REWORK","_number":86,"created":"2022-10-12 08:29:32.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/86","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/86","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/86 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/86 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/86 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/86"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-12 08:29:07.000000000","tz":330},"subject":"Add standalone-roles role and its playbook","message":"Add standalone-roles role and its playbook\n\nstandalone-roles is a new role to deploy standalone compute\nnode by using tripleo-ansible roles.\n\nThis role includes:\n- Bootstrap the env for deployment\n- Deploy standalone when deploy_standalone is set to true\n- Generate the config from standalone deployment\n- Deploy the compute services on the standalone host or on single node.\n- Generate the required inventory for target host when\n  compute_on_target_host is set to true\n- Deploy compute services on target host\n\nIt will be used in tripleo-external-compute-deployment-pipeline\njob template[1]\n\n[1]. https://review.opendev.org/c/openstack/tripleo-ci/+/843836\n\nNote: container-prep is dropped as a dependency from standalone roles.\nAs it is not needed.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/13c32a634f431da2be06349c5d43cf4f4c473661"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/13c32a634f431da2be06349c5d43cf4f4c473661"}]},"branch":"refs/heads/master"},"3308620ec6b1b7e4e731b6b1602b263c8e0bd40b":{"kind":"REWORK","_number":87,"created":"2022-10-12 08:30:32.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/87","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/87","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/87 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/87 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/87 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/87"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-12 08:30:21.000000000","tz":330},"subject":"Add standalone-roles role and its playbook","message":"Add standalone-roles role and its playbook\n\nstandalone-roles is a new role to deploy standalone compute\nnode by using tripleo-ansible roles.\n\nThis role includes:\n- Bootstrap the env for deployment\n- Deploy standalone when deploy_standalone is set to true\n- Generate the config from standalone deployment\n- Deploy the compute services on the standalone host or on single node.\n- Generate the required inventory for target host when\n  compute_on_target_host is set to true\n- Deploy compute services on target host\n\nIt will be used in tripleo-external-compute-deployment-pipeline\njob template[1]\n\n[1]. https://review.opendev.org/c/openstack/tripleo-ci/+/843836\n\nNote: container-prep is dropped as a dependency from standalone roles.\nAs it is not needed.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/3308620ec6b1b7e4e731b6b1602b263c8e0bd40b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/3308620ec6b1b7e4e731b6b1602b263c8e0bd40b"}]},"branch":"refs/heads/master"},"bccf2ff7bc08412fa97108072f0dcea23002d65d":{"kind":"REWORK","_number":88,"created":"2022-10-12 08:53:03.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/88","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/88","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/88 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/88 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/88 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/88"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-12 08:48:46.000000000","tz":330},"subject":"Add standalone-roles role and its playbook","message":"Add standalone-roles role and its playbook\n\nstandalone-roles is a new role to deploy standalone compute\nnode by using tripleo-ansible roles.\n\nThis role includes:-\n- On first node, Deploy the standalone\n- Generate 99 standalone vars\n- On Second node, Bootstrap the node\n- Copy 99 standalone vars from undercloud to subnode-1\n- Deploy Compute services\n\nIt will be used in tripleo-external-compute-deployment-pipeline\njob template[1]\n\n[1]. https://review.opendev.org/c/openstack/tripleo-ci/+/843836\n\nNote: container-prep is dropped as a dependency from standalone roles.\nAs it is not needed.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/bccf2ff7bc08412fa97108072f0dcea23002d65d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/bccf2ff7bc08412fa97108072f0dcea23002d65d"}]},"branch":"refs/heads/master"},"33f40dea182cc6f2c497ce83091c2c7a381dc386":{"kind":"REWORK","_number":89,"created":"2022-10-12 14:58:22.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/89","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/89","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/89 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/89 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/89 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/89"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-12 14:57:52.000000000","tz":330},"subject":"Add standalone-roles role and playbook","message":"Add standalone-roles role and playbook\n\nstandalone-roles is a new role to deploy standalone compute\nnode by using tripleo-ansible roles.\n\nThis role includes will run on two nodes and does following\nthings:\n- Bootstrap first node and deploy standalone\n- Generate 99 standalone vars and copy it to second node\n- Bootstrap second node and install packages\n- Copy 99 standalone vars to tripleo-inventory\n- Deploy compute serices.\n\nIt will be used in tripleo-external-compute-deployment-pipeline\njob template[1]\n\n[1]. https://review.opendev.org/c/openstack/tripleo-ci/+/843836\n\nNote: container-prep is dropped as a dependency from undercloud.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/33f40dea182cc6f2c497ce83091c2c7a381dc386"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/33f40dea182cc6f2c497ce83091c2c7a381dc386"}]},"branch":"refs/heads/master"},"2959643bb7139b15275c437848de0e69b472ed8a":{"kind":"REWORK","_number":90,"created":"2022-10-12 16:30:45.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/90","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/90","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/90 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/90 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/90 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/90"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-12 16:30:30.000000000","tz":330},"subject":"Add standalone-roles role and playbook","message":"Add standalone-roles role and playbook\n\nstandalone-roles is a new role to deploy standalone compute\nnode by using tripleo-ansible roles.\n\nThis role includes will run on two nodes and does following\nthings:\n- Bootstrap first node and deploy standalone\n- Generate 99 standalone vars and copy it to second node\n- Bootstrap second node and install packages\n- Copy 99 standalone vars to tripleo-inventory\n- Deploy compute serices.\n\nIt will be used in tripleo-external-compute-deployment-pipeline\njob template[1]\n\n[1]. https://review.opendev.org/c/openstack/tripleo-ci/+/843836\n\nNote: container-prep is dropped as a dependency from undercloud.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/2959643bb7139b15275c437848de0e69b472ed8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/2959643bb7139b15275c437848de0e69b472ed8a"}]},"branch":"refs/heads/master"},"e4170d3d0551567f2c93105fdf298769a744bad3":{"kind":"REWORK","_number":91,"created":"2022-10-12 16:32:21.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/91","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/91","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/91 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/91 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/91 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/91"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-12 16:32:08.000000000","tz":330},"subject":"Add standalone-roles role and playbook","message":"Add standalone-roles role and playbook\n\nstandalone-roles is a new role to deploy standalone compute\nnode by using tripleo-ansible roles.\n\nThis role includes will run on two nodes and does following\nthings:\n- Bootstrap first node and deploy standalone\n- Generate 99 standalone vars and copy it to second node\n- Bootstrap second node and install packages\n- Copy 99 standalone vars to tripleo-inventory\n- Deploy compute serices.\n\nIt will be used in tripleo-external-compute-deployment-pipeline\njob template[1]\n\n[1]. https://review.opendev.org/c/openstack/tripleo-ci/+/843836\n\nNote: container-prep is dropped as a dependency from undercloud.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/e4170d3d0551567f2c93105fdf298769a744bad3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/e4170d3d0551567f2c93105fdf298769a744bad3"}]},"branch":"refs/heads/master"},"c7d621d6bf9b5b8d603eb1a603f29fcd4a14916f":{"kind":"REWORK","_number":92,"created":"2022-10-13 02:57:55.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/92","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/92","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/92 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/92 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/92 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/92"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-13 02:57:35.000000000","tz":330},"subject":"Add standalone-roles role and playbook","message":"Add standalone-roles role and playbook\n\nstandalone-roles is a new role to deploy standalone compute\nnode by using tripleo-ansible roles.\n\nThis role includes will run on two nodes and does following\nthings:\n- Bootstrap first node and deploy standalone\n- Generate 99 standalone vars and copy it to second node\n- Bootstrap second node and install packages\n- Copy 99 standalone vars to tripleo-inventory\n- Deploy compute serices.\n\nIt will be used in tripleo-external-compute-deployment-pipeline\njob template[1]\n\n[1]. https://review.opendev.org/c/openstack/tripleo-ci/+/843836\n\nNote: container-prep is dropped as a dependency from undercloud.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/c7d621d6bf9b5b8d603eb1a603f29fcd4a14916f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/c7d621d6bf9b5b8d603eb1a603f29fcd4a14916f"}]},"branch":"refs/heads/master"},"b0ae6c9ad3284da816564ed97f968b1a534590ee":{"kind":"REWORK","_number":93,"created":"2022-10-13 03:46:22.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/93","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/93","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/93 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/93 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/93 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/93"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-13 03:46:05.000000000","tz":330},"subject":"Add standalone-roles role and playbook","message":"Add standalone-roles role and playbook\n\nstandalone-roles is a new role to deploy standalone compute\nnode by using tripleo-ansible roles.\n\nThis role includes will run on two nodes and does following\nthings:\n- Bootstrap first node and deploy standalone\n- Generate 99 standalone vars and copy it to second node\n- Bootstrap second node and install packages\n- Copy 99 standalone vars to tripleo-inventory\n- Deploy compute serices.\n\nIt will be used in tripleo-external-compute-deployment-pipeline\njob template[1]\n\n[1]. https://review.opendev.org/c/openstack/tripleo-ci/+/843836\n\nNote: container-prep is dropped as a dependency from undercloud.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/b0ae6c9ad3284da816564ed97f968b1a534590ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/b0ae6c9ad3284da816564ed97f968b1a534590ee"}]},"branch":"refs/heads/master"},"7d76532c74d7e916fa863ee02a0ef45b03425e88":{"kind":"REWORK","_number":94,"created":"2022-10-13 07:52:11.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/94","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/94","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/94 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/94 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/94 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/94"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-13 07:50:59.000000000","tz":330},"subject":"Add standalone-roles role and playbook","message":"Add standalone-roles role and playbook\n\nstandalone-roles is a new role to deploy standalone compute\nnode by using tripleo-ansible roles.\n\nThis role includes will run on two nodes and does following\nthings:\n- Bootstrap first node and deploy standalone\n- Generate 99 standalone vars and copy it to second node\n- Bootstrap second node and install packages\n- Copy 99 standalone vars to tripleo-inventory\n- Deploy compute serices.\n\nIt will be used in tripleo-external-compute-deployment-pipeline\njob template[1]\n\n[1]. https://review.opendev.org/c/openstack/tripleo-ci/+/843836\n\nNote: container-prep is dropped as a dependency from undercloud.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7d76532c74d7e916fa863ee02a0ef45b03425e88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7d76532c74d7e916fa863ee02a0ef45b03425e88"}]},"branch":"refs/heads/master"},"2ae737083730182a413c440cc7b8ba620763231c":{"kind":"REWORK","_number":95,"created":"2022-10-13 09:09:09.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/95","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/95","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/95 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/95 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/95 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/95"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-13 09:08:56.000000000","tz":330},"subject":"Add standalone-roles role and playbook","message":"Add standalone-roles role and playbook\n\nstandalone-roles is a new role to deploy standalone compute\nnode by using tripleo-ansible roles.\n\nThis role includes will run on two nodes and does following\nthings:\n- Bootstrap first node and deploy standalone\n- Generate 99 standalone vars and copy it to second node\n- Bootstrap second node and install packages\n- Copy 99 standalone vars to tripleo-inventory\n- Deploy compute serices.\n\nIt will be used in tripleo-external-compute-deployment-pipeline\njob template[1]\n\n[1]. https://review.opendev.org/c/openstack/tripleo-ci/+/843836\n\nNote: container-prep is dropped as a dependency from undercloud.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/2ae737083730182a413c440cc7b8ba620763231c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/2ae737083730182a413c440cc7b8ba620763231c"}]},"branch":"refs/heads/master"},"2a1cc084ea505970a5afd16b682329ffbeca8d4b":{"kind":"REWORK","_number":96,"created":"2022-10-13 11:57:49.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/96","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/96","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/96 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/96 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/96 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/96"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-13 11:57:37.000000000","tz":330},"subject":"Add standalone-roles role and playbook","message":"Add standalone-roles role and playbook\n\nstandalone-roles is a new role to deploy standalone compute\nnode by using tripleo-ansible roles.\n\nThis role includes will run on two nodes and does following\nthings:\n- Bootstrap first node and deploy standalone\n- Generate 99 standalone vars and copy it to second node\n- Bootstrap second node and install packages\n- Copy 99 standalone vars to tripleo-inventory\n- Deploy compute serices.\n\nIt will be used in tripleo-external-compute-deployment-pipeline\njob template[1]\n\n[1]. https://review.opendev.org/c/openstack/tripleo-ci/+/843836\n\nNote: container-prep is dropped as a dependency from undercloud.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/2a1cc084ea505970a5afd16b682329ffbeca8d4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/2a1cc084ea505970a5afd16b682329ffbeca8d4b"}]},"branch":"refs/heads/master"},"ed6209e45d8f8e966dbca040f3e8b8c5c26d144b":{"kind":"REWORK","_number":97,"created":"2022-10-13 14:54:57.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/97","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/97","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/97 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/97 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/97 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/97"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-13 14:54:42.000000000","tz":330},"subject":"Add standalone-roles role and playbook","message":"Add standalone-roles role and playbook\n\nstandalone-roles is a new role to deploy standalone compute\nnode by using tripleo-ansible roles.\n\nThis role includes will run on two nodes and does following\nthings:\n- Bootstrap first node and deploy standalone\n- Generate 99 standalone vars and copy it to second node\n- Bootstrap second node and install packages\n- Copy 99 standalone vars to tripleo-inventory\n- Deploy compute serices.\n\nIt will be used in tripleo-external-compute-deployment-pipeline\njob template[1]\n\n[1]. https://review.opendev.org/c/openstack/tripleo-ci/+/843836\n\nNote: container-prep is dropped as a dependency from undercloud.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/ed6209e45d8f8e966dbca040f3e8b8c5c26d144b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/ed6209e45d8f8e966dbca040f3e8b8c5c26d144b"}]},"branch":"refs/heads/master"},"5fbf7eafcfbb3f56af2bd19077f041230c78610d":{"kind":"REWORK","_number":98,"created":"2022-10-14 08:06:49.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/98","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/98","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/98 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/98 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/98 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/98"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-14 08:06:45.000000000","tz":330},"subject":"Add standalone-roles role and playbook","message":"Add standalone-roles role and playbook\n\nstandalone-roles is a new role to deploy standalone compute\nnode by using tripleo-ansible roles.\n\nThis role includes will run on two nodes and does following\nthings:\n- Bootstrap first node and deploy standalone\n- Generate 99 standalone vars and copy it to second node\n- Bootstrap second node and install packages\n- Copy 99 standalone vars to tripleo-inventory\n- Deploy compute serices.\n\nIt will be used in tripleo-external-compute-deployment-pipeline\njob template[1]\n\n[1]. https://review.opendev.org/c/openstack/tripleo-ci/+/843836\n\nNote: container-prep is dropped as a dependency from undercloud.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/5fbf7eafcfbb3f56af2bd19077f041230c78610d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/5fbf7eafcfbb3f56af2bd19077f041230c78610d"}]},"branch":"refs/heads/master"},"bedce2ace6973a45397104c6209450bf6eb1c548":{"kind":"REWORK","_number":99,"created":"2022-10-14 08:36:13.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/99","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/99","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/99 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/99 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/99 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/99"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-14 08:36:01.000000000","tz":330},"subject":"Add standalone-roles role and playbook","message":"Add standalone-roles role and playbook\n\nstandalone-roles is a new role to deploy standalone compute\nnode by using tripleo-ansible roles.\n\nThis role includes will run on two nodes and does following\nthings:\n- Bootstrap first node and deploy standalone\n- Generate 99 standalone vars and copy it to second node\n- Bootstrap second node and install packages\n- Copy 99 standalone vars to tripleo-inventory\n- Deploy compute serices.\n\nIt will be used in tripleo-external-compute-deployment-pipeline\njob template[1]\n\n[1]. https://review.opendev.org/c/openstack/tripleo-ci/+/843836\n\nNote: container-prep is dropped as a dependency from undercloud.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/bedce2ace6973a45397104c6209450bf6eb1c548"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/bedce2ace6973a45397104c6209450bf6eb1c548"}]},"branch":"refs/heads/master"},"94f51fb37c0d809afec347a7d2773f40492cefda":{"kind":"REWORK","_number":100,"created":"2022-10-14 11:08:13.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/100","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/100","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/100 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/100 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/100 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/100"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-14 11:08:02.000000000","tz":330},"subject":"Add standalone-roles role and playbook","message":"Add standalone-roles role and playbook\n\nstandalone-roles is a new role to deploy standalone compute\nnode by using tripleo-ansible roles.\n\nThis role includes will run on two nodes and does following\nthings:\n- Bootstrap first node and deploy standalone\n- Generate 99 standalone vars and copy it to second node\n- Bootstrap second node and install packages\n- Copy 99 standalone vars to tripleo-inventory\n- Deploy compute serices.\n\nIt will be used in tripleo-external-compute-deployment-pipeline\njob template[1]\n\n[1]. https://review.opendev.org/c/openstack/tripleo-ci/+/843836\n\nNote: container-prep is dropped as a dependency from undercloud.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/94f51fb37c0d809afec347a7d2773f40492cefda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/94f51fb37c0d809afec347a7d2773f40492cefda"}]},"branch":"refs/heads/master"},"4fd61a23f7e778f313bfe2bd55fdbac9868141ee":{"kind":"REWORK","_number":101,"created":"2022-10-14 11:51:58.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/101","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/101","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/101 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/101 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/101 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/101"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-14 11:51:45.000000000","tz":330},"subject":"Add standalone-roles role and playbook","message":"Add standalone-roles role and playbook\n\nstandalone-roles is a new role to deploy standalone compute\nnode by using tripleo-ansible roles.\n\nThis role includes will run on two nodes and does following\nthings:\n- Bootstrap first node and deploy standalone\n- Generate 99 standalone vars and copy it to second node\n- Bootstrap second node and install packages\n- Copy 99 standalone vars to tripleo-inventory\n- Deploy compute serices.\n\nIt will be used in tripleo-external-compute-deployment-pipeline\njob template[1]\n\n[1]. https://review.opendev.org/c/openstack/tripleo-ci/+/843836\n\nNote: container-prep is dropped as a dependency from undercloud.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/4fd61a23f7e778f313bfe2bd55fdbac9868141ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/4fd61a23f7e778f313bfe2bd55fdbac9868141ee"}]},"branch":"refs/heads/master"},"bf1175185deaa43e0bb5c57b4baafeadd60738b9":{"kind":"REWORK","_number":102,"created":"2022-10-14 16:13:30.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/102","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/102","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/102 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/102 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/102 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/102"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-14 16:13:19.000000000","tz":330},"subject":"Add standalone-roles role and playbook","message":"Add standalone-roles role and playbook\n\nstandalone-roles is a new role to deploy standalone compute\nnode by using tripleo-ansible roles.\n\nThis role includes will run on two nodes and does following\nthings:\n- Bootstrap first node and deploy standalone\n- Generate 99 standalone vars and copy it to second node\n- Bootstrap second node and install packages\n- Copy 99 standalone vars to tripleo-inventory\n- Deploy compute serices.\n\nIt will be used in tripleo-external-compute-deployment-pipeline\njob template[1]\n\n[1]. https://review.opendev.org/c/openstack/tripleo-ci/+/843836\n\nNote: container-prep is dropped as a dependency from undercloud.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/bf1175185deaa43e0bb5c57b4baafeadd60738b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/bf1175185deaa43e0bb5c57b4baafeadd60738b9"}]},"branch":"refs/heads/master"},"9806cf83140fc82972fc8c114b0686b3f2bc2d9b":{"kind":"REWORK","_number":103,"created":"2022-10-14 17:36:02.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/103","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/103","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/103 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/103 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/103 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/103"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-14 17:35:51.000000000","tz":330},"subject":"Add standalone-roles role and playbook","message":"Add standalone-roles role and playbook\n\nstandalone-roles is a new role to deploy standalone compute\nnode by using tripleo-ansible roles.\n\nThis role includes will run on two nodes and does following\nthings:\n- Bootstrap first node and deploy standalone\n- Generate 99 standalone vars and copy it to second node\n- Bootstrap second node and install packages\n- Copy 99 standalone vars to tripleo-inventory\n- Deploy compute serices.\n\nIt will be used in tripleo-external-compute-deployment-pipeline\njob template[1]\n\n[1]. https://review.opendev.org/c/openstack/tripleo-ci/+/843836\n\nNote: container-prep is dropped as a dependency from undercloud.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/9806cf83140fc82972fc8c114b0686b3f2bc2d9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/9806cf83140fc82972fc8c114b0686b3f2bc2d9b"}]},"branch":"refs/heads/master"},"a06b614207960b667318539d49dbf91ce79d910e":{"kind":"REWORK","_number":104,"created":"2022-10-18 04:30:16.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/104","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/104","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/104 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/104 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/104 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/104"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-18 04:28:31.000000000","tz":330},"subject":"Add EDPM role and playbook for standalone-compute deployment","message":"Add EDPM role and playbook for standalone-compute deployment\n\nexternal-data-plane-management role is a new role to deploy\nstandalone compute node by using tripleo-ansible roles.\n\nThis role includes will run on two nodes and does following\nthings:\n- Bootstrap first node and deploy standalone\n- Generate 99 standalone vars and copy it to second node\n- Bootstrap second node and install packages\n- Copy 99 standalone vars to tripleo-inventory\n- Deploy compute serices.\n\nIt will be used in tripleo-external-compute-deployment-pipeline\njob template[1]\n\n[1]. https://review.opendev.org/c/openstack/tripleo-ci/+/843836\n\nNote: container-prep is dropped as a dependency from undercloud.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/a06b614207960b667318539d49dbf91ce79d910e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/a06b614207960b667318539d49dbf91ce79d910e"}]},"branch":"refs/heads/master"},"f853f6860b91f302aa1bd793f79b760bb3d693b7":{"kind":"REWORK","_number":105,"created":"2022-10-18 04:33:52.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/105","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/105","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/105 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/105 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/105 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/105"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-18 04:33:40.000000000","tz":330},"subject":"Add EDPM role and playbook for standalone-compute deployment","message":"Add EDPM role and playbook for standalone-compute deployment\n\nexternal-data-plane-management role is a new role to deploy\nstandalone compute node by using tripleo-ansible roles.\n\nThis role includes will run on two nodes and does following\nthings:\n- Bootstrap first node and deploy standalone\n- Generate 99 standalone vars and copy it to second node\n- Bootstrap second node and install packages\n- Copy 99 standalone vars to tripleo-inventory\n- Deploy compute serices.\n\nIt will be used in tripleo-external-compute-deployment-pipeline\njob template[1]\n\n[1]. https://review.opendev.org/c/openstack/tripleo-ci/+/843836\n\nNote: container-prep is dropped as a dependency from undercloud.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/f853f6860b91f302aa1bd793f79b760bb3d693b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/f853f6860b91f302aa1bd793f79b760bb3d693b7"}]},"branch":"refs/heads/master"},"bc76df661e87bf7f15c63744b03f017d14af6864":{"kind":"REWORK","_number":106,"created":"2022-10-18 05:59:59.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/106","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/106","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/106 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/106 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/106 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/106"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-18 05:59:48.000000000","tz":330},"subject":"Add EDPM role and playbook for standalone-compute deployment","message":"Add EDPM role and playbook for standalone-compute deployment\n\nexternal-data-plane-management role is a new role to deploy\nstandalone compute node by using tripleo-ansible roles.\n\nThis role includes will run on two nodes and does following\nthings:\n- Bootstrap first node and deploy standalone\n- Generate 99 standalone vars and copy it to second node\n- Bootstrap second node and install packages\n- Copy 99 standalone vars to tripleo-inventory\n- Deploy compute serices.\n\nIt will be used in tripleo-external-compute-deployment-pipeline\njob template[1]\n\n[1]. https://review.opendev.org/c/openstack/tripleo-ci/+/843836\n\nNote: container-prep is dropped as a dependency from undercloud.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/bc76df661e87bf7f15c63744b03f017d14af6864"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/bc76df661e87bf7f15c63744b03f017d14af6864"}]},"branch":"refs/heads/master"},"194a5c994828e7732582316559fe1774566fee4d":{"kind":"REWORK","_number":107,"created":"2022-10-18 08:01:35.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/107","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/107","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/107 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/107 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/107 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/107"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-18 08:01:01.000000000","tz":330},"subject":"Add EDPM role and playbook for standalone-compute deployment","message":"Add EDPM role and playbook for standalone-compute deployment\n\nexternal-data-plane-management role is a new role to deploy\nstandalone compute node by using tripleo-ansible roles.\n\nThis role includes will run on two nodes and does following\nthings:\n- Bootstrap first node and deploy standalone\n- Generate 99 standalone vars and copy it to second node\n- Bootstrap second node and install packages\n- Copy 99 standalone vars to tripleo-inventory\n- Deploy compute serices.\n\nIt will be used in tripleo-external-compute-deployment-pipeline\njob template[1]\n\n[1]. https://review.opendev.org/c/openstack/tripleo-ci/+/843836\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/194a5c994828e7732582316559fe1774566fee4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/194a5c994828e7732582316559fe1774566fee4d"}]},"branch":"refs/heads/master"},"686dc038612b6ed71b48ec88c97c3183bf421f69":{"kind":"REWORK","_number":108,"created":"2022-10-18 09:47:30.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/108","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/108","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/108 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/108 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/108 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/108"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-18 09:47:19.000000000","tz":330},"subject":"Add EDPM role and playbook for standalone-compute deployment","message":"Add EDPM role and playbook for standalone-compute deployment\n\nexternal-data-plane-management role is a new role to deploy\nstandalone compute node by using tripleo-ansible roles.\n\nThis role includes will run on two nodes and does following\nthings:\n- Bootstrap first node and deploy standalone\n- Generate 99 standalone vars and copy it to second node\n- Bootstrap second node and install packages\n- Copy 99 standalone vars to tripleo-inventory\n- Deploy compute serices.\n\nIt will be used in tripleo-external-compute-deployment-pipeline\njob template[1]\n\n[1]. https://review.opendev.org/c/openstack/tripleo-ci/+/843836\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/686dc038612b6ed71b48ec88c97c3183bf421f69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/686dc038612b6ed71b48ec88c97c3183bf421f69"}]},"branch":"refs/heads/master"},"434c4042f649c3320096cd643e56da93c31f774f":{"kind":"REWORK","_number":109,"created":"2022-10-18 12:32:41.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/109","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/109","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/109 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/109 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/109 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/109"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-18 12:32:11.000000000","tz":330},"subject":"Add EDPM role and playbook for standalone-compute deployment","message":"Add EDPM role and playbook for standalone-compute deployment\n\nexternal-data-plane-management role is a new role to deploy\nstandalone compute node by using tripleo-ansible roles.\n\nThis role includes will run on two nodes and does following\nthings:\n- Bootstrap first node and deploy standalone\n- Generate 99 standalone vars and copy it to second node\n- Bootstrap second node and install packages\n- Copy 99 standalone vars to tripleo-inventory\n- Deploy compute serices.\n\nIt will be used in tripleo-external-compute-deployment-pipeline\njob template[1]\n\n[1]. https://review.opendev.org/c/openstack/tripleo-ci/+/843836\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/434c4042f649c3320096cd643e56da93c31f774f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/434c4042f649c3320096cd643e56da93c31f774f"}]},"branch":"refs/heads/master"},"19fd13a09b0bddf35d943fc5a606b40690ca6826":{"kind":"REWORK","_number":110,"created":"2022-10-18 15:22:58.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/110","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/110","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/110 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/110 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/110 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/110"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-18 15:22:46.000000000","tz":330},"subject":"Add EDPM role and playbook for standalone-compute deployment","message":"Add EDPM role and playbook for standalone-compute deployment\n\nexternal-data-plane-management role is a new role to deploy\nstandalone compute node by using tripleo-ansible roles.\n\nThis role includes will run on two nodes and does following\nthings:\n- Bootstrap first node and deploy standalone\n- Generate 99 standalone vars and copy it to second node\n- Bootstrap second node and install packages\n- Copy 99 standalone vars to tripleo-inventory\n- Deploy compute serices.\n\nIt will be used in tripleo-external-compute-deployment-pipeline\njob template[1]\n\n[1]. https://review.opendev.org/c/openstack/tripleo-ci/+/843836\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/19fd13a09b0bddf35d943fc5a606b40690ca6826"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/19fd13a09b0bddf35d943fc5a606b40690ca6826"}]},"branch":"refs/heads/master"},"8dd8fd018974554be1eccec57acd3fb205374afb":{"kind":"REWORK","_number":111,"created":"2022-10-19 03:42:10.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/111","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/111","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/111 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/111 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/111 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/111"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-19 03:41:59.000000000","tz":330},"subject":"Add EDPM role and playbook for standalone-compute deployment","message":"Add EDPM role and playbook for standalone-compute deployment\n\nexternal-data-plane-management role is a new role to deploy\nstandalone compute node by using tripleo-ansible roles.\n\nThis role includes will run on two nodes and does following\nthings:\n- Bootstrap first node and deploy standalone\n- Generate 99 standalone vars and copy it to second node\n- Bootstrap second node and install packages\n- Copy 99 standalone vars to tripleo-inventory\n- Deploy compute serices.\n\nIt will be used in tripleo-external-compute-deployment-pipeline\njob template[1]\n\n[1]. https://review.opendev.org/c/openstack/tripleo-ci/+/843836\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/8dd8fd018974554be1eccec57acd3fb205374afb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/8dd8fd018974554be1eccec57acd3fb205374afb"}]},"branch":"refs/heads/master"},"a3d56ef94f4300ac4b3b3f39c9afed6142f4bbe6":{"kind":"REWORK","_number":112,"created":"2022-10-19 11:33:15.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/112","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/112","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/112 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/112 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/112 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/112"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-19 11:33:04.000000000","tz":330},"subject":"Add EDPM role and playbook for standalone-compute deployment","message":"Add EDPM role and playbook for standalone-compute deployment\n\nexternal-data-plane-management role is a new role to deploy\nstandalone compute node by using tripleo-ansible roles.\n\nThis role includes will run on two nodes and does following\nthings:\n- Bootstrap first node and deploy standalone\n- Generate 99 standalone vars and copy it to second node\n- Bootstrap second node and install packages\n- Copy 99 standalone vars to tripleo-inventory\n- Deploy compute serices.\n\nIt will be used in tripleo-external-compute-deployment-pipeline\njob template[1]\n\n[1]. https://review.opendev.org/c/openstack/tripleo-ci/+/843836\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/a3d56ef94f4300ac4b3b3f39c9afed6142f4bbe6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/a3d56ef94f4300ac4b3b3f39c9afed6142f4bbe6"}]},"branch":"refs/heads/master"},"e188efe5a0f0dedd25449cad99240c656fbe1a91":{"kind":"REWORK","_number":113,"created":"2022-10-19 12:48:22.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/113","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/113","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/113 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/113 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/113 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/113"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-19 12:48:10.000000000","tz":330},"subject":"Add EDPM role and playbook for standalone-compute deployment","message":"Add EDPM role and playbook for standalone-compute deployment\n\nexternal-data-plane-management role is a new role to deploy\nstandalone compute node by using tripleo-ansible roles.\n\nThis role includes will run on two nodes and does following\nthings:\n- Bootstrap first node and deploy standalone\n- Generate 99 standalone vars and copy it to second node\n- Bootstrap second node and install packages\n- Copy 99 standalone vars to tripleo-inventory\n- Deploy compute serices.\n\nIt will be used in tripleo-external-compute-deployment-pipeline\njob template[1]\n\n[1]. https://review.opendev.org/c/openstack/tripleo-ci/+/843836\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/e188efe5a0f0dedd25449cad99240c656fbe1a91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/e188efe5a0f0dedd25449cad99240c656fbe1a91"}]},"branch":"refs/heads/master"},"b06af416faa37e028ec9b78b64b91bcf90a4335a":{"kind":"REWORK","_number":114,"created":"2022-10-19 15:13:10.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/114","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/114","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/114 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/114 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/114 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/114"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-19 15:12:56.000000000","tz":330},"subject":"Add EDPM role and playbook for standalone-compute deployment","message":"Add EDPM role and playbook for standalone-compute deployment\n\nexternal-data-plane-management role is a new role to deploy\nstandalone compute node by using tripleo-ansible roles.\n\nThis role includes will run on two nodes and does following\nthings:\n- Bootstrap first node and deploy standalone\n- Generate 99 standalone vars and copy it to second node\n- Bootstrap second node and install packages\n- Copy 99 standalone vars to tripleo-inventory\n- Deploy compute serices.\n\nIt will be used in tripleo-external-compute-deployment-pipeline\njob template[1]\n\n[1]. https://review.opendev.org/c/openstack/tripleo-ci/+/843836\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/b06af416faa37e028ec9b78b64b91bcf90a4335a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/b06af416faa37e028ec9b78b64b91bcf90a4335a"}]},"branch":"refs/heads/master"},"0bbc8971682349007cc3e59a56a05e08876680dd":{"kind":"REWORK","_number":115,"created":"2022-10-20 01:54:52.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/115","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/115","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/115 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/115 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/115 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/115"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-20 01:54:41.000000000","tz":330},"subject":"Add EDPM role and playbook for standalone-compute deployment","message":"Add EDPM role and playbook for standalone-compute deployment\n\nexternal-data-plane-management role is a new role to deploy\nstandalone compute node by using tripleo-ansible roles.\n\nThis role includes will run on two nodes and does following\nthings:\n- Bootstrap first node and deploy standalone\n- Generate 99 standalone vars and copy it to second node\n- Bootstrap second node and install packages\n- Copy 99 standalone vars to tripleo-inventory\n- Deploy compute serices.\n\nIt will be used in tripleo-external-compute-deployment-pipeline\njob template[1]\n\n[1]. https://review.opendev.org/c/openstack/tripleo-ci/+/843836\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/0bbc8971682349007cc3e59a56a05e08876680dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/0bbc8971682349007cc3e59a56a05e08876680dd"}]},"branch":"refs/heads/master"},"ccaabd8c1ef36a09e1077c809480cd2f7508b03e":{"kind":"REWORK","_number":116,"created":"2022-10-20 05:30:20.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/35/843835/116","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/35/843835/116","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/116 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/116 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/116 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/35/843835/116"}}},"commit":{"parents":[{"commit":"7142f40c51122eb2dad09161c231a96cedff6542","subject":"Merge \"Test Deployed Ceph in Wallaby standalone 001/004/010\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7142f40c51122eb2dad09161c231a96cedff6542"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-05-30 10:16:33.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-20 05:30:10.000000000","tz":330},"subject":"Add EDPM role and playbook for standalone-compute deployment","message":"Add EDPM role and playbook for standalone-compute deployment\n\nexternal-data-plane-management role is a new role to deploy\nstandalone compute node by using tripleo-ansible roles.\n\nThis role includes will run on two nodes and does following\nthings:\n- Bootstrap first node and deploy standalone\n- Generate 99 standalone vars and copy it to second node\n- Bootstrap second node and install packages\n- Copy 99 standalone vars to tripleo-inventory\n- Deploy compute serices.\n\nIt will be used in tripleo-external-compute-deployment-pipeline\njob template[1]\n\n[1]. https://review.opendev.org/c/openstack/tripleo-ci/+/843836\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I476c7299dec209c89db87e2182a35ec6a4924f92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/ccaabd8c1ef36a09e1077c809480cd2f7508b03e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/ccaabd8c1ef36a09e1077c809480cd2f7508b03e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}}]}],"submit_requirements":[]}
