)]}'
{"id":"openstack%2Ftripleo-quickstart-extras~843856","triplet_id":"openstack%2Ftripleo-quickstart-extras~master~I8c5e168a9fd18332c073223726ad3a3e338af04d","project":"openstack/tripleo-quickstart-extras","branch":"master","attention_set":{},"removed_from_attention_set":{"8449":{"account":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"last_update":"2022-05-30 21:57:15.000000000","reason":"Change was submitted"},"30002":{"account":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"last_update":"2022-05-30 16:17:44.000000000","reason":"Douglas Viroel replied on the change"},"9976":{"account":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"last_update":"2022-05-30 18:06:12.000000000","reason":"Ronelle Landy replied on the change"}},"hashtags":[],"change_id":"I8c5e168a9fd18332c073223726ad3a3e338af04d","subject":"Disable stackviz for tempest centos 8 to unblock gate","status":"MERGED","created":"2022-05-30 13:08:49.000000000","updated":"2022-05-30 21:57:15.000000000","submitted":"2022-05-30 21:57:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"843856","meta_rev_id":"94a53c7f0a2277b10fa38e2f3ae4a99ec430eafc","_number":843856,"virtual_id_number":843856,"owner":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-30 21:57:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2022-05-30 18:14:11.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":2,"date":"2022-05-30 16:17:44.000000000","_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":2,"date":"2022-05-30 18:06:12.000000000","_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":1,"date":"2022-05-30 19:01:35.000000000","_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-30 13:09:13.000000000","updated_by":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-30 15:43:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-30 15:45:29.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-30 16:12:38.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2022-05-30 16:12:38.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"REVIEWER"}],"messages":[{"id":"959a08f1b187f1f94823feba18e84da43aae7559","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-05-30 13:08:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7dfbc335ffa911cbabbe13ed5c8502d67e718333","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-05-30 13:09:13.000000000","message":"Patch Set 1:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1003/","accounts_in_message":[],"_revision_number":1},{"id":"4c751009d68a00e277e3c31b4328845cbb688050","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-05-30 13:49:19.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1003/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"8f0d100edff18129a59f2f9ed338ee7b8b5ce2a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 15:43:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/06ed48a3f80349cb92a201d0b21bd5db : SUCCESS in 2h 11m 07s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/38cd185b97b04ed0823da0efd916cec1 : SUCCESS in 41m 34s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/8a5924ed7f41408f979652155e26c939 : SUCCESS in 1h 19m 26s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/a02b66c293f9495d8411eaefb9de4211 : SUCCESS in 1h 16m 58s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8b3d92556f90479d89eff5cdf35d7afa : SUCCESS in 5m 38s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/a45e53a6634c4dd4b9c45df4ccc0723c : SUCCESS in 2h 25m 14s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/35aef5a1025b4ac6b22e6813f605d4ef : SUCCESS in 1h 31m 43s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/f28c9283472a440f80346894f38efeb8 : SUCCESS in 1h 07m 17s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/ba7decf805424a68b3489ea163b1a91c : SUCCESS in 1h 11m 30s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/dd80872f416c4370b3ba4cebd360b00f : SUCCESS in 58m 10s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/bf7d6dacc8d94543ae9ea406d0a2cf97 : SUCCESS in 21m 44s","accounts_in_message":[],"_revision_number":1},{"id":"414719b2b4fba73c22034661a8c201b411a15397","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-05-30 15:45:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/6f086c1a34fe4fe18b57e17eefd93658 : SUCCESS in 1h 49m 34s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/7831ef07b80746fab4b23b6f5114a0be : FAILURE in 2h 35m 26s","accounts_in_message":[],"_revision_number":1},{"id":"12b4897898cf219d39624f14ac293cde2110d6b6","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-05-30 16:11:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c681fdb012999aa6564fd43896c8be7dbdc4b271","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-05-30 16:12:21.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"ed68b4a539b2071f52c59321c285cde4f33cf82f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-05-30 16:12:42.000000000","message":"Patch Set 2:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1005/","accounts_in_message":[],"_revision_number":2},{"id":"e7768bdf0dd6afde5b63a7afbbb3945baccb731a","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2022-05-30 16:17:44.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3e5732d9335b8180e8b97f8b509d09f2ec5ccdf1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-05-30 16:54:12.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1005/ : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"10cee406d279136ba16275e8b1de40b734cd7644","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2022-05-30 18:06:12.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"64163625615983a438e877dad42a640bdd6a41ac","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-05-30 18:14:11.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/10c8540a520946e3a781d6818c1574f6 : FAILURE in 1h 02m 58s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/cf787b6149384df99f48b9a6d07f1442 : SUCCESS in 1h 58m 49s","accounts_in_message":[],"_revision_number":2},{"id":"c27d1d8b4be160cdc365aab6f1686db9e89b9379","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 18:57:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/fec7d56eda4546089f77dbe5ffea8847 : SUCCESS in 2h 21m 17s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/af6cce41b7a34fe79f66b9c40b2bec03 : SUCCESS in 37m 04s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/4cb045f0e73143238b340736b52c2d7d : SUCCESS in 1h 20m 20s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/e6c396b681a04bdbb95ec2b93892647a : SUCCESS in 1h 19m 33s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/95e2961bb17f4e8297e3e87ab5a8e0db : SUCCESS in 5m 01s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/17bb7f56c7df4b3c837ce740f3667c3b : SUCCESS in 2h 29m 59s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/c8f892a5064d4348af26a95a628dfa3a : SUCCESS in 1h 30m 46s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/bd36e30d9bb4477fa2daed553fcb6521 : SUCCESS in 1h 03m 59s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/aeacfad50d34474e8f9cd413b29b40ef : SUCCESS in 1h 07m 03s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/16abf31b91d54db1b3fae73cf40fda31 : SUCCESS in 59m 47s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/0e10618d50454bf09c1e673ca18d91fb : SUCCESS in 19m 46s","accounts_in_message":[],"_revision_number":2},{"id":"3245db9966e2a395a24fdec707cbf19e53a16d53","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2022-05-30 19:01:35.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c9fe9d3bc722a9d13a3fb84d6f6238119fc9f9d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 19:02:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3ae471d0091039bea69f32faffbda10f5fbb7d39","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 21:57:15.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/2bfb63383fed4a49aebd5133e9234e22 : SUCCESS in 2h 40m 18s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/e3dd111e5fbb4ec1956e199ebd5dd923 : SUCCESS in 45m 21s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/b53b6a2e81ee4b69a3d6d806d1d4b09b : SUCCESS in 1h 29m 45s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/1ec60856d88243719687417f45f9b9d9 : SUCCESS in 1h 32m 41s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/ec38d3274bad4c8bb483ecccc8082a33 : SUCCESS in 2h 07m 04s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/289282eceea447398c721c2e06b76545 : SUCCESS in 1h 19m 23s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/2282bd4f9bda47b58d2cee2f1ae84836 : SUCCESS in 1h 15m 02s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/b999ebedc8bb4458ab66cc365f5b3ea3 : SUCCESS in 52m 23s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/da8ae61f558341edbdcc9a88eaec8d3c : SUCCESS in 1h 16m 51s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1438d98918124356b053d97b34ede427 : SUCCESS in 5m 34s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/213f7e1a9e3d41c59d388ad5841077e7 : SUCCESS in 15m 48s","accounts_in_message":[],"_revision_number":2},{"id":"94a53c7f0a2277b10fa38e2f3ae4a99ec430eafc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 21:57:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8bcf4bfc0f2596776ca6abf284cce6337e097e3d","revisions":{"eff59a04b3939f4ecb4df9e8d3652a3a011c4372":{"kind":"REWORK","_number":1,"created":"2022-05-30 13:08:49.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/56/843856/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/56/843856/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/843856/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/843856/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/843856/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/843856/1"}}},"commit":{"parents":[{"commit":"be5e48aeedf5ecd5ee4acdbed7c4fa1be2db0d50","subject":"Merge \"Push to local registry when building containers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/be5e48aeedf5ecd5ee4acdbed7c4fa1be2db0d50"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2022-05-30 13:06:35.000000000","tz":180},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2022-05-30 13:06:35.000000000","tz":180},"subject":"WIP test disabling stackviz for tempest centos 8","message":"WIP test disabling stackviz for tempest centos 8\n\nAs commented in related bug (see comments/6) we need to disable\ninstallation of stackviz as there are issues with the constraints\nused. This is not a proper fix but if it works can unblock for now\n\nChange-Id: I8c5e168a9fd18332c073223726ad3a3e338af04d\nRelated-Bug: 1976247\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/eff59a04b3939f4ecb4df9e8d3652a3a011c4372"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/eff59a04b3939f4ecb4df9e8d3652a3a011c4372"}]},"branch":"refs/heads/master"},"8bcf4bfc0f2596776ca6abf284cce6337e097e3d":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-05-30 16:12:21.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/56/843856/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/56/843856/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/843856/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/843856/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/843856/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/56/843856/2"}}},"commit":{"parents":[{"commit":"be5e48aeedf5ecd5ee4acdbed7c4fa1be2db0d50","subject":"Merge \"Push to local registry when building containers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/be5e48aeedf5ecd5ee4acdbed7c4fa1be2db0d50"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2022-05-30 13:06:35.000000000","tz":180},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2022-05-30 16:12:21.000000000","tz":0},"subject":"Disable stackviz for tempest centos 8 to unblock gate","message":"Disable stackviz for tempest centos 8 to unblock gate\n\nAs commented in related bug (see comments/6) we need to disable\ninstallation of stackviz as there are issues with the constraints\nused. This is not a proper fix but if it works can unblock for now\n\nRelated-Bug: 1976247\n\nChange-Id: I8c5e168a9fd18332c073223726ad3a3e338af04d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/8bcf4bfc0f2596776ca6abf284cce6337e097e3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/8bcf4bfc0f2596776ca6abf284cce6337e097e3d"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
