)]}'
{"id":"openstack%2Ftripleo-quickstart-extras~846231","triplet_id":"openstack%2Ftripleo-quickstart-extras~master~I1abfbbd23ca93c01393d05057806ba9cc846fbed","project":"openstack/tripleo-quickstart-extras","branch":"master","attention_set":{},"removed_from_attention_set":{"8449":{"account":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"last_update":"2022-06-24 09:52:31.000000000","reason":"Change was submitted"},"18002":{"account":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"last_update":"2022-06-24 09:52:31.000000000","reason":"Change was submitted"},"30002":{"account":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"last_update":"2022-06-23 18:41:41.000000000","reason":"\u003cGERRIT_ACCOUNT_30002\u003e replied on the change","reason_account":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}},"31075":{"account":{"_account_id":31075,"name":"Pooja Jadhav","email":"pojadhav@redhat.com","username":"pojadhav"},"last_update":"2022-06-22 12:22:09.000000000","reason":"\u003cGERRIT_ACCOUNT_31075\u003e replied on the change","reason_account":{"_account_id":31075,"name":"Pooja Jadhav","email":"pojadhav@redhat.com","username":"pojadhav"}},"9976":{"account":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"last_update":"2022-06-24 09:52:31.000000000","reason":"Change was submitted"},"25402":{"account":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"last_update":"2022-06-24 05:50:20.000000000","reason":"\u003cGERRIT_ACCOUNT_25402\u003e replied on the change","reason_account":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"}},"8367":{"account":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"last_update":"2022-06-20 14:12:40.000000000","reason":"\u003cGERRIT_ACCOUNT_8367\u003e replied on the change","reason_account":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}}},"hashtags":[],"change_id":"I1abfbbd23ca93c01393d05057806ba9cc846fbed","subject":"Override Ceph --container-namespace for periodic jobs","status":"MERGED","created":"2022-06-16 22:08:08.000000000","updated":"2022-06-24 09:52:31.000000000","submitted":"2022-06-24 09:52:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":45,"unresolved_comment_count":6,"has_review_started":true,"submission_id":"846231","meta_rev_id":"663f44159645f92a7c8c0e52b47b10265605fc61","_number":846231,"virtual_id_number":846231,"owner":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-24 09:52:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2022-06-24 03:55:13.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":2,"date":"2022-06-24 01:38:45.000000000","_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":2,"date":"2022-06-24 01:38:49.000000000","_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":2,"date":"2022-06-24 05:50:20.000000000","_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":1,"date":"2022-06-24 01:38:49.000000000","_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31075,"name":"Pooja Jadhav","email":"pojadhav@redhat.com","username":"pojadhav"},{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"reviewers":{"REVIEWER":[{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":31075,"name":"Pooja Jadhav","email":"pojadhav@redhat.com","username":"pojadhav"}],"CC":[{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-16 22:08:27.000000000","updated_by":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-16 22:13:50.000000000","updated_by":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"reviewer":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"state":"CC"},{"updated":"2022-06-16 22:13:50.000000000","updated_by":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2022-06-16 22:13:50.000000000","updated_by":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"REVIEWER"},{"updated":"2022-06-16 23:57:29.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-17 01:16:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-17 10:41:21.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"state":"REVIEWER"},{"updated":"2022-06-17 10:41:21.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":31075,"name":"Pooja Jadhav","email":"pojadhav@redhat.com","username":"pojadhav"},"state":"REVIEWER"},{"updated":"2022-06-17 12:43:11.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"state":"REVIEWER"},{"updated":"2022-06-20 10:06:38.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2022-06-20 10:29:28.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"}],"messages":[{"id":"0e598045ea35535119755f90e1596bcbc8ade9f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"date":"2022-06-16 22:08:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"534d3fbc650d12a970979d8a27c684715f950e22","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-16 22:08:27.000000000","message":"Patch Set 1:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1062/","accounts_in_message":[],"_revision_number":1},{"id":"79fd0eb4532e4e610ec6a8bddc1c3b6be19f4821","author":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"date":"2022-06-16 22:13:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b7b1c3cc0ecd8342e8e34ec5679bfb368f1d686a","author":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"date":"2022-06-16 22:22:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a451ca280fd2d76084758c29db00a5e44b35e8b3","author":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"date":"2022-06-16 22:30:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2bad21386260a67b37643fc1d2275f878c0d8cc6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-16 23:17:57.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1062/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"19727421ab4a91c5c7a07831e93064025ca9f59c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-06-16 23:57:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/e5fe47e7694d452cb4aebf3039e5a6a6 : FAILURE in 1h 41m 20s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/a921c133e56349d8bd0d011889757a80 : FAILURE in 1h 20m 38s","accounts_in_message":[],"_revision_number":1},{"id":"67ac152a1f5b2cb46051e9833d58821a1fd7856e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 01:16:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/c388060ef1db46d1b2a49d1e83636d74 : FAILURE in 20m 15s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/62c974b82e3643658d12e24c9c6f4cd5 : SKIPPED\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/807fff6b32794f02b2bcd39269f5c0ae : SUCCESS in 4m 49s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/6caae47b7dc44c639340dafd71d26a3d : SUCCESS in 2h 52m 59s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/7e5ef9a022564141b6993e6aa569d85d : FAILURE in 1h 41m 07s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/5fa7bcae7ad646cf8e06561f426b4897 : SKIPPED\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/e15aa402458841fc8f99479a8d2ef950 : SKIPPED\n- tripleo-ci-centos-9-scenario004-standalone https://zuul.opendev.org/t/openstack/build/b404afcb7aaa4aab88831d7bd0194665 : SKIPPED\n- tripleo-ci-centos-9-scenario010-standalone https://zuul.opendev.org/t/openstack/build/ecc154b3daf947f1bf84ecb6c14bdc59 : SKIPPED\n- tripleo-ci-centos-9-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/ad5f2bbba6fd420589213c96e74179ef : SKIPPED\n- tripleo-ci-centos-9-standalone-on-multinode-ipa https://zuul.opendev.org/t/openstack/build/2d535481577c43dfad55bcf04f891e7d : SKIPPED\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/74f904fb52ef4d6b9262b99bac1935d3 : SKIPPED\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/2b878f67e4c74fddaea044e1fa29a38b : SUCCESS in 20m 23s","accounts_in_message":[],"_revision_number":1},{"id":"fa8e8dc69b6e3338b2d6d3a4bf39dd44ee7d3517","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-06-17 06:16:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f2400550d7f012ed3bb14a776ee3bf9c4aebf4af","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2022-06-17 10:41:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dc3821807f0ebac3348b622cab2406b5a0fdabd5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-17 10:41:41.000000000","message":"Patch Set 1:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1064/","accounts_in_message":[],"_revision_number":1},{"id":"bfca4b366ee5677dc53f41b615e316a4121ded76","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-17 11:50:38.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1064/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"00aa16605149c93bcb136155033c3104e49e3e7f","author":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"date":"2022-06-17 12:07:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"39fe340ab9b41a3b3bbec87739f9b7a4ed4ceec9","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-06-17 12:15:43.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e8da13d6dcc686f1ef3e76bc4eea33c1d952e60a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-06-17 12:19:37.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/af23ff42a6ff444cbe489f716b68b8fc : FAILURE in 1h 30m 54s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/bb7bcfa0d0d04229836f926aaa922551 : FAILURE in 1h 27m 55s","accounts_in_message":[],"_revision_number":1},{"id":"582d50844e2225c0b294a63fb745da7e1aaca4e2","author":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"date":"2022-06-17 12:25:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3ca02286a12f6ec4a0fcfca5851d44daabac8819","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-06-17 12:43:11.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"799950633f6759945bf2b5cee23ad150a4d1ac17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 14:04:38.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/c475dcfe6b474b7eb7c5d26bb13c9586 : SUCCESS in 3h 09m 43s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/d1127eda873f41dea8a2bc798fa8ccd7 : SUCCESS in 54m 58s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0dd7749fa40a4681b055e1ef9e50542f : SUCCESS in 5m 04s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/dd21a63ae58540b989278f8ed418e306 : SUCCESS in 2h 54m 11s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/67ace392e6ed47cd80afe42d3603e57c : SUCCESS in 1h 43m 43s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/f9f1341f6a2942fc8b428041c887e0ca : SUCCESS in 1h 35m 40s\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/ef2ef484478743c98567c15b670064f2 : FAILURE in 35m 02s\n- tripleo-ci-centos-9-scenario004-standalone https://zuul.opendev.org/t/openstack/build/571b0566870442bfae201cda7c101d46 : FAILURE in 36m 19s\n- tripleo-ci-centos-9-scenario010-standalone https://zuul.opendev.org/t/openstack/build/523defa14235464c8d955775438fd9a0 : FAILURE in 27m 12s\n- tripleo-ci-centos-9-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/c916b06a011845248d74d1934a046a0f : FAILURE in 33m 55s\n- tripleo-ci-centos-9-standalone-on-multinode-ipa https://zuul.opendev.org/t/openstack/build/5cd4c095a3e747cf808eae68a10ac2f0 : SUCCESS in 1h 56m 34s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/5c841585c2374feba99253ac5fca0cbc : SUCCESS in 1h 11m 52s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/650c626732144344adedacdc4ee15e81 : SUCCESS in 17m 53s","accounts_in_message":[],"_revision_number":1},{"id":"1050ce78556066be99b69c8ede17e2b72241ad7a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-06-17 14:22:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"eb3e7487d3abe1addc14e551db1dac39498bc34b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-17 14:22:39.000000000","message":"Patch Set 2:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1068/","accounts_in_message":[],"_revision_number":2},{"id":"0b4ebbd09d399b0e49b38a91d4d370b205ffe2b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-06-17 14:25:08.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0180e20aa02eec7b3817e1d7e93771be2b4dd798","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-17 14:25:47.000000000","message":"Patch Set 3:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1069/","accounts_in_message":[],"_revision_number":3},{"id":"24e165b448d1af76f5d4adcc979e5d0f412051a8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-17 14:26:46.000000000","message":"Patch Set 2:\n\nBuild Aborted \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1068/ : ABORTED","accounts_in_message":[],"_revision_number":2},{"id":"b99a3235fe2a7ed8387270383c0c00204a7e256c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-17 15:06:11.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1069/ : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"7c03945153972eba98e58f168e998f9131d68b06","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-06-17 15:07:27.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4d9181402646ec7b97a30ebd556b3237fefec5cc","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-06-17 15:09:30.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4c270677713e40896bd421ef97825e6b3d10ff78","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-06-17 15:15:58.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"90e8ed31fb8a206523d8c8fa89d3fd9e8930ae9d","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-06-17 15:27:47.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4872e7833ecd8decfa90a11b18a237ec8d9ad4b4","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-06-17 15:37:32.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1eaa2af658ce47a2ba004dd1e631cc2bf541a768","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-06-17 15:48:15.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/550d15665aac4ea387be96a40a42556b : FAILURE in 1h 22m 28s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/069b6f146d7c4af2b4b140825249d68a : FAILURE in 1h 19m 41s","accounts_in_message":[],"_revision_number":3},{"id":"0450d640717fc0a6e236a51e8406eea583062975","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-06-17 16:37:58.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3d4ec0423d5d6d5469045046b0d5f517cd599469","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-06-17 16:41:40.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"38391595c7cb3164e1f1c27a1268706028f4124e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-17 16:42:00.000000000","message":"Patch Set 4:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1070/","accounts_in_message":[],"_revision_number":4},{"id":"720dee4d512bf923f682dde8208be0afab672f8a","author":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"date":"2022-06-17 17:07:32.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ae67f9aeaa3594c24dcb81e21e1c8fbfc492384b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-17 17:22:38.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1070/ : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"cee0fd191963951a8669cc9ff561acf843052246","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-06-17 18:08:43.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/a1ec6d14c2a6442f85ba29c03e1dbf73 : FAILURE in 1h 25m 51s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/c32465de47f149849a50e3987a678936 : FAILURE in 1h 23m 39s","accounts_in_message":[],"_revision_number":4},{"id":"0bc9a010d36666090a7bc31ec194e57b6b5c0e75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 19:32:30.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/6dde38d7e15a42a19628815bbc303a91 : SUCCESS in 2h 42m 08s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/7f89ce6a765240a594aa89cf8f05a1d4 : SUCCESS in 40m 25s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/28d14a9a038a40ab94c7542d9d074b85 : SUCCESS in 5m 29s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/39f273aca94b4029b5ca74eae1114eb7 : SUCCESS in 2h 26m 34s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/12a5acf36b314e59b6ca09e5bd59c4c6 : SUCCESS in 1h 22m 16s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/03afe8844a4d4721a8f98ab9bdc574ab : SUCCESS in 1h 07m 51s\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/be210dda4fba4f95a3225fd24e088b56 : FAILURE in 25m 31s\n- tripleo-ci-centos-9-scenario004-standalone https://zuul.opendev.org/t/openstack/build/9987348956ae4b4dba20f6d21544082d : FAILURE in 36m 24s\n- tripleo-ci-centos-9-scenario010-standalone https://zuul.opendev.org/t/openstack/build/04d4cbdf94e643e5bd942d5d04d846b8 : FAILURE in 23m 56s\n- tripleo-ci-centos-9-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/39478642a9af492d9cb8df10b12f0578 : FAILURE in 25m 39s\n- tripleo-ci-centos-9-standalone-on-multinode-ipa https://zuul.opendev.org/t/openstack/build/778badcbf4a1482aa7c906687c0ce807 : SUCCESS in 1h 40m 40s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/af9c815a898e4ea48f18696d8a0fa4b0 : SUCCESS in 1h 04m 53s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/ef24fd3ee28c4bf580a9a3bf31ec66b9 : SUCCESS in 17m 21s","accounts_in_message":[],"_revision_number":4},{"id":"1c0499e9baa130041e979e72bc3b9a532eca2e58","author":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"date":"2022-06-17 21:02:01.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"48f11cb1ee9730f2046066a3ed8f16bcf8e50f39","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-06-17 21:52:24.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5287fe8a2071c8e7c91404c3b860fb9afeeeed57","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-06-17 21:56:47.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"1d9afdee82e18f805cca34673a92d31efa2faf53","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-17 21:57:09.000000000","message":"Patch Set 5:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1073/","accounts_in_message":[],"_revision_number":5},{"id":"3f666a707eb24d70b0dd32d9738249a66749ce66","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-17 23:06:46.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1073/ : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"fe999f03273be059a6e9301361a0dd1647af534d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-06-17 23:28:02.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/2b02e54470fa4a8b8e4e2a2582e0abfb : FAILURE in 1h 29m 55s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/dd050a0776014b34b3f89cb229a5958d : FAILURE in 1h 26m 49s","accounts_in_message":[],"_revision_number":5},{"id":"d279c76329ab3037f7f52b60306d1689318099e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-18 01:21:11.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/dc3dcc655ccb47508dcfc7b85dde5d37 : SUCCESS in 3h 13m 44s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/cb1c0c9571a541ceab2122dd236507ca : SUCCESS in 54m 59s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/cd81d27397454f98939487bba80acab7 : SUCCESS in 4m 52s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/609fc697fa764409a4acdb34e455da41 : SUCCESS in 2h 32m 28s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/b44ce35ac7a345ca8a797d421d7ea86b : SUCCESS in 1h 35m 53s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/2c4294f5dfd14423b70167841a3c1c51 : SUCCESS in 1h 28m 56s\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/c1c9e610c6ee42c2a84758b464f6b715 : FAILURE in 39m 18s\n- tripleo-ci-centos-9-scenario004-standalone https://zuul.opendev.org/t/openstack/build/4a159f2047164b8282b96f2fd412dd4e : FAILURE in 34m 54s\n- tripleo-ci-centos-9-scenario010-standalone https://zuul.opendev.org/t/openstack/build/5704705d80114d9dac4a5bbce7fade3c : FAILURE in 27m 05s\n- tripleo-ci-centos-9-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/84d5e578891947fc9e0508d66fec2f8a : FAILURE in 37m 28s\n- tripleo-ci-centos-9-standalone-on-multinode-ipa https://zuul.opendev.org/t/openstack/build/0a7657428d954f16b299a4f962d3032e : SUCCESS in 2h 04m 08s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/592637650350494c986c9d21d5408103 : SUCCESS in 1h 14m 41s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/daf1c551a711466cb4afccb3904b0327 : SUCCESS in 15m 50s","accounts_in_message":[],"_revision_number":5},{"id":"928e06795089157e2c310117ac84d6f0f98f4dc0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-06-18 06:59:00.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"e036bba1f9ee70f0a11efe0a628bddc471d3a3d9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-18 06:59:20.000000000","message":"Patch Set 6:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1074/","accounts_in_message":[],"_revision_number":6},{"id":"db6ac3db3080da6e6eb352fc8259c6bc5f4cdca2","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-06-18 07:01:17.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"9d2a8062fc83dc5fedf8b3842b3e412a3b2d6c02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-06-18 07:13:09.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"6453aa126e6f58b3e2db716cb76240810e28f0cb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-18 07:13:28.000000000","message":"Patch Set 7:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1075/","accounts_in_message":[],"_revision_number":7},{"id":"43d23ec8a6deaa68badef6efda8d2e8f7e01861e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-18 07:18:21.000000000","message":"Patch Set 6:\n\nBuild Aborted \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1074/ : ABORTED","accounts_in_message":[],"_revision_number":6},{"id":"04e484a58ffc3cd3492e90a55763f39a0d11c3fb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-18 07:53:47.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1075/ : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"764259ec93c8fa0b695904fed4d11d292043c14d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-06-18 08:36:49.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/7ecb0c8876504f3cbdb284208c2d1d5d : FAILURE in 1h 22m 37s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/957a43bb15bc4260a77569e4b7074a30 : FAILURE in 1h 18m 55s","accounts_in_message":[],"_revision_number":7},{"id":"ac1852f578cba526820fa535efad25e2e222af12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-18 10:36:33.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/231ccbf3d1504444a9b171e3ae0be0f0 : SUCCESS in 3h 09m 38s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/a16add0dbab24438bfb209786a119e0b : SUCCESS in 52m 13s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c9dfd5db024f482db52b2aa4847e4561 : SUCCESS in 4m 31s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/16e836793ca04809bf223ff6143d4217 : SUCCESS in 2h 14m 18s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/96fef7a0ef8f49eb8496d0530fd6407e : SUCCESS in 1h 08m 59s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/699b8c0a0873472da73afd1d616af1ca : SUCCESS in 1h 28m 54s\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/c0bfe6b8961b4734b1d1b50d570e25e9 : SUCCESS in 1h 41m 01s\n- tripleo-ci-centos-9-scenario004-standalone https://zuul.opendev.org/t/openstack/build/746494fc951a4d8dadac66f2af01b978 : SUCCESS in 1h 37m 44s\n- tripleo-ci-centos-9-scenario010-standalone https://zuul.opendev.org/t/openstack/build/3e57530eebcf4c3c9114c24a7a5691c5 : FAILURE in 26m 40s\n- tripleo-ci-centos-9-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/5eda77ddabc64e2a9ae4de7600099c82 : SUCCESS in 1h 29m 42s\n- tripleo-ci-centos-9-standalone-on-multinode-ipa https://zuul.opendev.org/t/openstack/build/96af6d9b46c44a768feaf69799499469 : SUCCESS in 1h 50m 09s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/79e57a03b4f34ff9b4f850d11c33ef41 : SUCCESS in 1h 00m 28s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/0f8ca157af5e4fbf90b61bd24c157958 : SUCCESS in 19m 57s","accounts_in_message":[],"_revision_number":7},{"id":"30190e3f25abc86cbff278a1c9c941723b43c3b8","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-06-18 10:37:41.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"2e9119f649cd7f14f451e7966345aeaf95f55ef8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-06-18 10:38:01.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"9665ea22c9f6b9d5ed2b379385a4f41082aaff57","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-18 10:38:21.000000000","message":"Patch Set 8:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1076/","accounts_in_message":[],"_revision_number":8},{"id":"f4380da4c8701fcba44622cea2390962dbc5b502","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-06-18 10:38:28.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"8fc39de491ff2ed67d7ae6c4a36df09a8a18ddcb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-18 11:17:16.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1076/ : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"81715e3d64dee7c618add1de9eafc54d1b129bff","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-06-18 12:01:32.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/e2044bda6e634e0b9714252433594133 : FAILURE in 1h 22m 27s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/d1b0b504ece148e3adc2ccbdec06a086 : FAILURE in 1h 21m 10s","accounts_in_message":[],"_revision_number":8},{"id":"5375723603ef85243374ae9b351aec45eeb58b02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-18 13:31:38.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/c13ac738f9524596971d12e2771dcb6d : SUCCESS in 2h 43m 52s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/2cc51f0d2e3a47fa8ad90d15099cfd70 : SUCCESS in 42m 26s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3dca085e44ea4d76ae679514f4f9537f : SUCCESS in 4m 18s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/10359b82f3cb4051a650e330594dfe62 : SUCCESS in 2h 06m 27s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/4268b0f81aa147aeaa4ae4e8583e12f3 : SUCCESS in 1h 22m 06s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/63132c3a9f6147b7999d9601677f422c : SUCCESS in 1h 05m 42s\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/2dbde3a3c4e3446892c7f9220c0dbe23 : SUCCESS in 1h 42m 24s\n- tripleo-ci-centos-9-scenario004-standalone https://zuul.opendev.org/t/openstack/build/834786f8936f46c78ffcbe519da70f67 : SUCCESS in 1h 22m 12s\n- tripleo-ci-centos-9-scenario010-standalone https://zuul.opendev.org/t/openstack/build/fe582603781043929f738667219b0303 : SUCCESS in 1h 55m 31s\n- tripleo-ci-centos-9-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/124a2c170a004032a82212b7ab170016 : SUCCESS in 1h 08m 26s\n- tripleo-ci-centos-9-standalone-on-multinode-ipa https://zuul.opendev.org/t/openstack/build/0aba5b62801b47bf8d8955df69481296 : SUCCESS in 1h 44m 13s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/56bf46f99b554359b8d9c6c11c68a796 : SUCCESS in 1h 05m 39s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/07798a6c2964442589b608d0388cd30d : SUCCESS in 20m 07s","accounts_in_message":[],"_revision_number":8},{"id":"12dec5edcc05429cfade7d4ca8bff28511c077b3","author":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"date":"2022-06-18 19:23:00.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"a34db5bfc98c0280f49fc5a3a0a3c170dce46f65","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-06-20 09:49:21.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"f0b806b6e5a7b2614950e1bfbb2a3a9f102c555e","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2022-06-20 10:02:50.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"f40a9b0798f38dd24fcb1bfeefaa5744cd41807b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 10:03:53.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"cd3288c12037d45c86cd26d7e515d77e99af557a","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-06-20 10:06:38.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"8684b77f0247761ebfc09fb3f9dfbf0aaca33eaa","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-06-20 10:07:49.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"9da9ba7ce86f18da05aae13f39a1043a10c493fb","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-06-20 10:29:28.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"4ef76304c802cebafc498438e9da18a26b9e2701","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 14:04:16.000000000","message":"Patch Set 8: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/635eea23e4534ea4b1a4e842e3b12356 : SUCCESS in 3h 56m 15s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/83036c5f648e4e12abf0d93586ca41a1 : SUCCESS in 50m 23s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/d0ca1b69b2424f5ea7bd4ec8deb64787 : SUCCESS in 3h 08m 20s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/230153707327407fbbd8f1fdbdc4c5ce : SUCCESS in 1h 42m 26s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/941656b9e0c448339114ec7677897925 : SUCCESS in 1h 39m 07s\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/7d1d4a9507f24eaaacc426e2bc23a5cf : SUCCESS in 2h 00m 46s\n- tripleo-ci-centos-9-scenario004-standalone https://zuul.opendev.org/t/openstack/build/1401a543ab744e338f7bbbb799fe53be : SUCCESS in 1h 51m 23s\n- tripleo-ci-centos-9-scenario010-standalone https://zuul.opendev.org/t/openstack/build/639a5bc35c0f44e59048178e97ab2b91 : FAILURE in 30m 04s\n- tripleo-ci-centos-9-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/00880d2d00f943a5a4b65ebe26cc018d : SUCCESS in 1h 24m 46s\n- tripleo-ci-centos-9-standalone-on-multinode-ipa https://zuul.opendev.org/t/openstack/build/4cb33079bec1410c97b12b555ea7f547 : SUCCESS in 2h 21m 45s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/7567a01c28fc43fc802c554c3a3a4b8a : SUCCESS in 1h 36m 29s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4930faf2a5014dc28cef7f8be39da9a9 : SUCCESS in 5m 12s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/feace0a840464d7bba272e5aa24c3c6f : SUCCESS in 22m 10s","accounts_in_message":[],"_revision_number":8},{"id":"b989ecddc63aea8f040805fcf1924948d9f9c5e5","author":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"date":"2022-06-20 14:12:28.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"f7754a927561937420f57022c614af35236726b1","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2022-06-20 14:12:40.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"e50fbc0675f63dcaa4e581580f505e05cd25c288","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-20 14:12:59.000000000","message":"Patch Set 8:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1081/","accounts_in_message":[],"_revision_number":8},{"id":"09522ffb8d0243ce405dd9e0dd7af277a256d2c0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-20 15:23:03.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1081/ : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"bf0a082d578a5c02d0130b22ca40d4133a3e4f9a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-06-20 15:44:34.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/d37b19c0d25e44f7937d93b07ce18966 : FAILURE in 1h 22m 07s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/f3a0d172f6f54b89bc2fb7fc8c63011b : FAILURE in 1h 20m 28s","accounts_in_message":[],"_revision_number":8},{"id":"c8df1232f9ab831e568e6b816137a7663650d8de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 17:12:19.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/2ef5ee2616c74c0497cb32e182efc913 : SUCCESS in 2h 52m 56s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/0147396f460c47efa7f467565b289f0d : SUCCESS in 45m 26s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/bb415faf4646487b8a2da91fb7bd4cf8 : SUCCESS in 4m 24s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/f2372f82b9a142cb8aae79718e267bae : SUCCESS in 2h 45m 33s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/7c27220cde3c461c92f16350bedb6895 : SUCCESS in 1h 41m 53s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/bba988798790493db193f410e8f19026 : SUCCESS in 1h 21m 52s\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/221ea57d1ccd4d7b886b4d39ede8eac4 : SUCCESS in 1h 35m 45s\n- tripleo-ci-centos-9-scenario004-standalone https://zuul.opendev.org/t/openstack/build/519cd402acd346f58ddbae5e1917b032 : SUCCESS in 1h 30m 08s\n- tripleo-ci-centos-9-scenario010-standalone https://zuul.opendev.org/t/openstack/build/892712a69798438d96299679a60cc328 : SUCCESS in 1h 36m 58s\n- tripleo-ci-centos-9-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/726c789ec7244a53a7f87ec4e3ed0e76 : SUCCESS in 1h 18m 42s\n- tripleo-ci-centos-9-standalone-on-multinode-ipa https://zuul.opendev.org/t/openstack/build/576935c7f2d242d9b035e1455db27be3 : SUCCESS in 1h 51m 37s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/753268baa78c4675bf6a98183a3238b7 : SUCCESS in 1h 15m 40s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/a9a2a095ddba4c58b1048b88c67d1a28 : SUCCESS in 19m 30s","accounts_in_message":[],"_revision_number":8},{"id":"e632f779a1427dfdbf9a07ec12957d0ab12d2bfc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 17:13:14.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"e07dab9b58585df491ee4419b27c820aeed91b1d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 20:53:16.000000000","message":"Patch Set 8: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/dd2d03b13a564c038c958d9897acf386 : SUCCESS in 2h 38m 07s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/8821b25626bc40dfaca19503d45216d9 : SUCCESS in 41m 34s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/29c845db4bf64628aad3ac851dfeabc2 : SUCCESS in 2h 28m 29s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/0564f0948e9747d3b441cd2cff8a619e : SUCCESS in 1h 30m 12s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/575b2b2ab3b840f6b1f2f7cc8d54c241 : SUCCESS in 1h 12m 48s\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/9d2a3e8106354909908fe040f9cf1991 : RETRY_LIMIT in 27s\n- tripleo-ci-centos-9-scenario004-standalone https://zuul.opendev.org/t/openstack/build/7eca7688681946f2915aaf6feeb0e503 : SUCCESS in 1h 33m 48s\n- tripleo-ci-centos-9-scenario010-standalone https://zuul.opendev.org/t/openstack/build/4a7a50f948a548e2ba92431cebb155c0 : SUCCESS in 1h 31m 33s\n- tripleo-ci-centos-9-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/78ea7ff391c84ceb9dba5327e026c661 : SUCCESS in 1h 26m 50s\n- tripleo-ci-centos-9-standalone-on-multinode-ipa https://zuul.opendev.org/t/openstack/build/26bd4d15837f4569a91dd91311d18edd : SUCCESS in 1h 41m 20s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/a822bb84de7d4f6ea320435685a80c2d : SUCCESS in 1h 05m 17s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e9739feb1ad14938adb675fd7b7ddb03 : SUCCESS in 5m 08s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/f264397876df435d999f9d482ee6c11d : SUCCESS in 17m 51s","accounts_in_message":[],"_revision_number":8},{"id":"82d8ed873cc67183007d5636025d6587e48a324c","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-06-20 21:04:08.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"163b6d97bb01a4f85d8ae1b990e47143e88a5492","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-20 21:04:31.000000000","message":"Patch Set 8:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1082/","accounts_in_message":[],"_revision_number":8},{"id":"e4ac4b7006e27d89c0058169dd2a3bf483803b18","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-20 21:45:09.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1082/ : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"fdf71f3d66969fe0de190f97b9e6f9acf0bdf18e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-06-20 22:23:40.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/0bcae7f7f9fd4b0cab3ba0f6fe369277 : FAILURE in 1h 18m 32s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/6adb0a0aac0745f482ef8428945553d3 : FAILURE in 1h 18m 20s","accounts_in_message":[],"_revision_number":8},{"id":"1d9624d8ad3b6fee12766cf23969dbb8f67a380b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 23:53:59.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/68fb85496dd841f6a28fa63e161e36d0 : SUCCESS in 2h 40m 05s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/c733aba296994db7819feb599c632c3a : SUCCESS in 38m 41s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/74a52b6e744f435898d3b4eb8e47f895 : SUCCESS in 4m 27s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/d85cbf57ecc84b62ae730267baf0e85d : SUCCESS in 2h 30m 38s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/4288ab101f9f44d8935860b70d141ea7 : SUCCESS in 1h 48m 05s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/80731093adf5497986571621b675645c : SUCCESS in 1h 03m 20s\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/2eabd83464cb428ea04b6d315a6bc04b : SUCCESS in 1h 15m 06s\n- tripleo-ci-centos-9-scenario004-standalone https://zuul.opendev.org/t/openstack/build/1b3cc1b2359b45aeaa8d9b7e75a698de : SUCCESS in 1h 21m 27s\n- tripleo-ci-centos-9-scenario010-standalone https://zuul.opendev.org/t/openstack/build/e5146d012afe447ea100dc48f9d91524 : SUCCESS in 1h 22m 37s\n- tripleo-ci-centos-9-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/7e28e913e0454dbab48a0fac20e0421e : SUCCESS in 1h 19m 53s\n- tripleo-ci-centos-9-standalone-on-multinode-ipa https://zuul.opendev.org/t/openstack/build/d3a484fb95304380a55e1c728862fd58 : SUCCESS in 1h 50m 57s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/f76a0c8b62bc438bb64a8329eebe4785 : SUCCESS in 1h 01m 28s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/c80961e4407b48a5ada6757f8e733136 : SUCCESS in 19m 39s","accounts_in_message":[],"_revision_number":8},{"id":"1c064572effe70b71866a7fb952ec79fbd889de9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 23:54:41.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"8016f48d8edc4ba3ee1106744c8def6726ae9bd7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 02:24:13.000000000","message":"Patch Set 8: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/53b84e76e32745bda442ddfb31e43e29 : SUCCESS in 2h 28m 50s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/f44b89e36cf045e3a5dc4b6c575ee23c : SUCCESS in 42m 31s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/f221cb908d7e4a9fa1c0f6715e40cd6a : SUCCESS in 2h 09m 20s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/7e6c347c8b0b4844b4c2e1dd9ea897a3 : SUCCESS in 1h 13m 58s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/60da961ecd5e4ac9a9521bb568f9b296 : FAILURE in 53m 46s\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/7fc5393181ef4d8cb316855f11b0d80d : FAILURE in 52m 44s\n- tripleo-ci-centos-9-scenario004-standalone https://zuul.opendev.org/t/openstack/build/97cfcc182449446d944bec9069e67277 : FAILURE in 1h 01m 25s\n- tripleo-ci-centos-9-scenario010-standalone https://zuul.opendev.org/t/openstack/build/c902afa1b81042b78bfdddcf7974aaf6 : SUCCESS in 1h 27m 06s\n- tripleo-ci-centos-9-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/3c8ca90e37944121a0d004fa8191263b : FAILURE in 58m 47s\n- tripleo-ci-centos-9-standalone-on-multinode-ipa https://zuul.opendev.org/t/openstack/build/8ac6b2e196bb4906bea04aeeedeefd28 : FAILURE in 1h 19m 33s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/dfc9b989fe7943c99e9648e4a3bff8e4 : SUCCESS in 1h 09m 35s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a59037dffd5945a4a8c2ecf35dc172fd : SUCCESS in 5m 12s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/b49d4ce534d84d7eb6d2951ad5ae0c03 : SUCCESS in 13m 56s","accounts_in_message":[],"_revision_number":8},{"id":"c0e8d93782277329febb8d84db1be2d6d9204e68","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2022-06-21 07:23:24.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"f554009e6c25b69e011f05241ab16104ecab2cc4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-21 07:23:51.000000000","message":"Patch Set 8:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1083/","accounts_in_message":[],"_revision_number":8},{"id":"5e557fd71712a264f888832073434fcdaa9fe58b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-21 07:27:16.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1083/ : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"3e2f16e8009dd887ed14f282473b8e035775582d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-06-21 08:48:13.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/b2cb5db5b3f04dce97333425663cda25 : FAILURE in 1h 22m 56s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/e36130ad2d95427987589678e2cd17e0 : FAILURE in 1h 23m 27s","accounts_in_message":[],"_revision_number":8},{"id":"1a4b79b62aba6d5ac57f8f62f129911416ee9adc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 11:10:51.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/82e688aa49fb421ab77bd12196af9835 : SUCCESS in 2h 56m 53s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/42faa8ab1db14becb7e02b7578eb6c9c : SUCCESS in 54m 28s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/184925e6c9f449e6a52f396041bf3fe5 : SUCCESS in 4m 21s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/dcb1666adcb6455791851ecdd3255732 : SUCCESS in 3h 36m 53s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/5e879b19a4fa405e83e8fa46a8d5b423 : FAILURE in 2h 13m 40s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/2b7cd5ffe88d4c76a4a835f221e3c909 : FAILURE in 1h 09m 00s\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/fd54af5a7eed4db68ebc40c94cfc2414 : FAILURE in 1h 03m 49s\n- tripleo-ci-centos-9-scenario004-standalone https://zuul.opendev.org/t/openstack/build/f4e204ff9e6041cfb664d81f2464ae60 : FAILURE in 1h 07m 04s\n- tripleo-ci-centos-9-scenario010-standalone https://zuul.opendev.org/t/openstack/build/6c789c034f6e45a8b939808194faf78d : SUCCESS in 1h 51m 39s\n- tripleo-ci-centos-9-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/a96928984c4948dc8cabfe846f06fde6 : FAILURE in 1h 27m 14s\n- tripleo-ci-centos-9-standalone-on-multinode-ipa https://zuul.opendev.org/t/openstack/build/39babdfb1a884ebcae7adc4afceecdf9 : FAILURE in 1h 25m 36s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/41c6435afe7f45fc9044304a69e3cc59 : SUCCESS in 1h 29m 09s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/e7ea7fee0ea440d4a92fb3234bed01cf : SUCCESS in 20m 12s","accounts_in_message":[],"_revision_number":8},{"id":"238302233f5c922abc46f89686a38ba126dd84f7","author":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"date":"2022-06-21 15:35:08.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"50f0818f379fc4c8f1dc12141570b0290e985895","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2022-06-21 18:42:29.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"5d8c55e99d4b961906cf4c38b7023d43fc884d19","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-21 18:42:52.000000000","message":"Patch Set 8:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1089/","accounts_in_message":[],"_revision_number":8},{"id":"576253934b62cee68ab3a871a747667f13c3eecc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-21 20:07:05.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1089/ : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"27f3642e6b846be4ce61d71e367273b4fb188214","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-06-21 20:15:16.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/492835d0c2e74397bdc54170ec340889 : FAILURE in 1h 31m 38s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/d5e1b22937a649678ae3e8e8334a32fa : FAILURE in 1h 22m 29s","accounts_in_message":[],"_revision_number":8},{"id":"5a91189b273059df5ef79e24c72e6d214b48b1af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 21:46:35.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/b7e977020bf6475bbc7ca1d2c63c2e3d : SUCCESS in 2h 56m 07s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/392085743e044a3f9616a1a2dfd4365d : SUCCESS in 39m 28s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/523dec4ed6f14febbba6215996b1137b : SUCCESS in 5m 35s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/ced48ce7ce174352a3abf17fa299f97f : SUCCESS in 2h 56m 28s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/ad1e2af111ac4031aa19b28bcc28f5bb : FAILURE in 2h 08m 05s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/fc58106e97da4892a5a4d85b7d690c0d : SUCCESS in 1h 08m 27s\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/f4c6833191f24475a8fefc44807693b8 : FAILURE in 51m 50s\n- tripleo-ci-centos-9-scenario004-standalone https://zuul.opendev.org/t/openstack/build/1004c4ada8f34290a09726f5f9d0a91e : FAILURE in 56m 43s\n- tripleo-ci-centos-9-scenario010-standalone https://zuul.opendev.org/t/openstack/build/bc230c3e7e8e47b5b37b201f72df686b : SUCCESS in 1h 25m 37s\n- tripleo-ci-centos-9-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/aa271b6996204544adf0dc7b0f9e121d : SUCCESS in 1h 09m 21s\n- tripleo-ci-centos-9-standalone-on-multinode-ipa https://zuul.opendev.org/t/openstack/build/91eb286e478d477592e49d412e49073d : SUCCESS in 2h 08m 50s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/9955686e797b43fcb30a5afef912435f : SUCCESS in 1h 19m 12s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/26555eec7c7b4d8281bbe9d5826427c5 : SUCCESS in 17m 28s","accounts_in_message":[],"_revision_number":8},{"id":"c2d58e78f9d7108cbb38c6779316369fa068bf13","author":{"_account_id":31075,"name":"Pooja Jadhav","email":"pojadhav@redhat.com","username":"pojadhav"},"date":"2022-06-22 12:22:09.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"927226610ca129ccac9e07b3bdd3b855e1b4d1de","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-22 12:22:27.000000000","message":"Patch Set 8:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1094/","accounts_in_message":[],"_revision_number":8},{"id":"3295cbb5185b8b3a81db281b10a3de5167bf9165","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-22 13:41:12.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1094/ : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"cde6039c73dc2375060911dba121c1d228f01088","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-06-22 14:45:24.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/32e9435ad43343bf9562ddb846b797fc : FAILURE in 1h 26m 12s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/bee12f65303940798bc767bea0084b49 : FAILURE in 1h 24m 10s","accounts_in_message":[],"_revision_number":8},{"id":"51354a26f23efe823c048297c4870967255901e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 15:15:12.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/6adb1fe4befc459eac0103d6dd8d4f3c : SUCCESS in 2h 41m 17s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/8fb603e2f32b4f9c84c7c21ec2a05c74 : SUCCESS in 40m 01s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/79e6b0161f32432180de8998adf63db0 : SUCCESS in 4m 54s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/04690013441943aeb2c22f4c45c1df8c : SUCCESS in 2h 26m 23s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/fa4742f11ea645f3a109fd9ea6ff0eb7 : SUCCESS in 1h 24m 15s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/461a7fb62dd74237883dbf9a929ed389 : SUCCESS in 1h 12m 02s\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/144278a0a2584670a70c8d0c4ba8422b : FAILURE in 52m 16s\n- tripleo-ci-centos-9-scenario004-standalone https://zuul.opendev.org/t/openstack/build/f2ff393ff4ad4584a40e3c0648f8121c : FAILURE in 53m 39s\n- tripleo-ci-centos-9-scenario010-standalone https://zuul.opendev.org/t/openstack/build/00b29a6339e34ee09140fbf757c2aca4 : SUCCESS in 1h 41m 53s\n- tripleo-ci-centos-9-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/b0ad77c0e81c4e4680ef211d1a9aeeff : SUCCESS in 1h 13m 50s\n- tripleo-ci-centos-9-standalone-on-multinode-ipa https://zuul.opendev.org/t/openstack/build/b0a8de4fca9f45ccb28d483b58076166 : FAILURE in 1h 41m 28s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/9348b1acd10543289c04f9b977455a86 : SUCCESS in 1h 03m 45s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/0aae13a3cf78482c8adfc3b8c841ad62 : SUCCESS in 18m 59s","accounts_in_message":[],"_revision_number":8},{"id":"e128fd4c137b97b1fb6c9e085144d295df040e40","author":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"date":"2022-06-22 18:00:22.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"f38a2efa7f4d70d3eaf94f9ea776958cd862910c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-22 18:00:42.000000000","message":"Patch Set 8:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1095/","accounts_in_message":[],"_revision_number":8},{"id":"36ef1a44a913bb92603f28bc71f11d3242cda836","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-22 19:19:37.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1095/ : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"c4299134ccecf81f7a8a5fefddd2b283e9e95a53","author":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"date":"2022-06-22 19:55:55.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"c7f3dcf2607f5c0968119e4cfd6f164026631eca","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-06-22 20:04:32.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/604fd9d1a0314acf85872b042e213764 : FAILURE in 1h 24m 29s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/8a586df8d7074605a6cb68a82d51c927 : FAILURE in 1h 21m 02s","accounts_in_message":[],"_revision_number":8},{"id":"00dd462837df622c22497384c438ed5726060895","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 21:17:28.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/040249b30392448297439299ac22a5d6 : SUCCESS in 2h 43m 25s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/e53a3c4cd02745f4ae7f875c943e1253 : SUCCESS in 44m 02s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ebf91674beec44a089a154f632bd7ef6 : SUCCESS in 4m 26s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/a13778cacee54684bed4bc85e015bad8 : SUCCESS in 3h 06m 44s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/df3beea7ea9c4abab61db8c75e9a2d8d : SUCCESS in 2h 07m 16s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/93bab5d222cd45f5bbff66bff50195d5 : SUCCESS in 1h 13m 28s\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/84a01c43cc76414eb7169cdcd0c53578 : FAILURE in 1h 24m 44s\n- tripleo-ci-centos-9-scenario004-standalone https://zuul.opendev.org/t/openstack/build/053dd69ae44546f3bcc3c8f1b558deba : SUCCESS in 1h 58m 47s\n- tripleo-ci-centos-9-scenario010-standalone https://zuul.opendev.org/t/openstack/build/32bea8396b7343028e9858d92669ea23 : FAILURE in 26m 21s\n- tripleo-ci-centos-9-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/81190a69c7f244b3ad846c3e6c84976b : SUCCESS in 1h 20m 02s\n- tripleo-ci-centos-9-standalone-on-multinode-ipa https://zuul.opendev.org/t/openstack/build/843726901817492dba9691bf95433b3a : SUCCESS in 1h 35m 03s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/4d3258787d3c4dfca8ecf8e0bbb8c181 : SUCCESS in 1h 05m 30s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/5c54f9c55ef240a9a55d3915dcb3cdc6 : SUCCESS in 17m 31s","accounts_in_message":[],"_revision_number":8},{"id":"ce0dcf8bcc862c1e6409001327434fb7f1a091f8","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2022-06-23 11:12:31.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"37e01a3194c29f20eb552dcd60365cfbbb37b49d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-23 11:13:19.000000000","message":"Patch Set 8:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1097/","accounts_in_message":[],"_revision_number":8},{"id":"6d203add62669353c77d2743c4f07a8311d86318","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-23 11:59:07.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1097/ : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"36f0138ce0a7b1f33357748433c148779d2d3aa3","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2022-06-23 12:03:32.000000000","message":"Patch Set 8: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"94591e2ee6917b843f2992b9b002c550901a40a5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-06-23 12:39:31.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/6daf338a3aa74d2aa0ed975e655b5180 : FAILURE in 44m 14s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/44f8917a2a234650a08a639525eccfcc : FAILURE in 1h 25m 55s","accounts_in_message":[],"_revision_number":8},{"id":"631f00a48e69493596fa2a2ce09c4e58d4155dc8","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-06-23 12:54:45.000000000","message":"Patch Set 8: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"7e77adf6e6a9d66412ae5ddfcc9a6cfe7cef2b38","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-06-23 12:55:57.000000000","message":"Patch Set 8: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"b5198b02c9769289d2f228ee4145e697659f3cc6","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2022-06-23 12:56:58.000000000","message":"Patch Set 8: -Workflow","accounts_in_message":[],"_revision_number":8},{"id":"72edc8210ae11ff4066ac030bd4967788eb99eed","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-06-23 13:50:04.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"0be5dd47220f70d3a2d35dec6118184ee8268ad9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-06-23 13:56:03.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"c870ce9172108818c95ada8f456598530e44aea4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-23 13:56:28.000000000","message":"Patch Set 9:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1100/","accounts_in_message":[],"_revision_number":9},{"id":"2b5cad2d75e6c9c3b9a595a09582e2f4610056a9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-23 13:57:20.000000000","message":"Patch Set 9:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1100/ : FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"2fea8e3abf48949886e1d0ea19e079957ba3374a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-06-23 17:12:42.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/27b4e1d9fc29436d927fd285e8f31218 : FAILURE in 1h 41m 55s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/edbf751646294f54925e103c9e4cb81d : FAILURE in 1h 47m 34s","accounts_in_message":[],"_revision_number":9},{"id":"2adb45e231c5bce63cc8f0e16807126ccd24ec70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-23 17:41:52.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/06d6375a91f94fd294d98e0edb558ec8 : SUCCESS in 3h 37m 42s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/b179d134d5d4417f903be45aca8c12e3 : SUCCESS in 42m 33s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d53cd7a5d44f42c69836c2d4e758e151 : SUCCESS in 4m 43s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/d3ef02a243c54bd994a3829149bf1238 : SUCCESS in 3h 12m 21s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/df4dbbab662a43c194c43789e15b83fe : SUCCESS in 2h 19m 27s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/993a80a619064a89b1e0eb1bf50efaa2 : SUCCESS in 1h 18m 32s\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/57ba6fcf80ac4f0cbb2c212edc02b8bb : SUCCESS in 1h 37m 48s\n- tripleo-ci-centos-9-scenario004-standalone https://zuul.opendev.org/t/openstack/build/836a6827dd3146c3b2759d5bd6928399 : SUCCESS in 1h 36m 31s\n- tripleo-ci-centos-9-scenario010-standalone https://zuul.opendev.org/t/openstack/build/900337cac77f47d1a4a2253893a2b19c : SUCCESS in 1h 43m 34s\n- tripleo-ci-centos-9-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/261a742751484686ab41126917016572 : SUCCESS in 1h 18m 11s\n- tripleo-ci-centos-9-standalone-on-multinode-ipa https://zuul.opendev.org/t/openstack/build/2b1de55ffc5143afaf63a369f8a7e3ba : SUCCESS in 2h 31m 10s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/58ae306a682d444da43484c871759b5c : SUCCESS in 1h 07m 47s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/722e846a4a604f7bbb7e15c2713c0f12 : SUCCESS in 19m 15s","accounts_in_message":[],"_revision_number":9},{"id":"7492e1681d93c981272660408ae07cf01eb3410f","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2022-06-23 18:41:41.000000000","message":"Patch Set 9: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"e413e445296e4de22c096da0892aafa104f7856e","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2022-06-23 18:46:43.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"bca8eb0205244fcc3fb262e8c25c725220fdbd24","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2022-06-23 18:46:55.000000000","message":"Patch Set 9: -Workflow","accounts_in_message":[],"_revision_number":9},{"id":"2e5841e2ee5bf5f873e2c9eacb91b2ed9702926a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-23 18:47:09.000000000","message":"Patch Set 9: Verified-2\n\nThis change is unable to merge due to a missing requirement.","accounts_in_message":[],"_revision_number":9},{"id":"586153a46ed80bb40957851b7185f59575f6f3e3","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2022-06-23 20:48:33.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"9c446020ec7a5a2df21356484225d536c41ed1a7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-23 20:48:54.000000000","message":"Patch Set 9:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1102/","accounts_in_message":[],"_revision_number":9},{"id":"f00ae157e67a1d7138dec8b40efa10bc5490a13a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2022-06-23 21:00:47.000000000","message":"Topic bug/1978998 removed","accounts_in_message":[],"_revision_number":9},{"id":"5d1f0012087d29e5430e3a40fc199ee560507de6","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2022-06-23 21:00:50.000000000","message":"Patch Set 9: Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"b0f884cd2780731e961b8615eb75d3555caaa7fa","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-23 21:33:33.000000000","message":"Patch Set 9:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1102/ : FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"63a1becc2529b374a07bd3d81148388541f358b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 00:15:44.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/5ec0359613904df8a95d8c831d63f09c : SUCCESS in 2h 59m 08s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/d590b4c9b4b94f38b0e9f220f6369833 : SUCCESS in 40m 46s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8f2d5c179d8c49f1b9c26b54cf8d26aa : SUCCESS in 6m 17s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/fedd21bd45fc4649a47f24954b01e11e : SUCCESS in 3h 14m 46s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/01e0728421a5438987157c153441af2e : POST_FAILURE in 2h 21m 18s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/8c8382d6ff624536b03467c7c8acbcc1 : SUCCESS in 1h 46m 14s\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/cfa5ca968a5048d7a4d46d7cd6b009b7 : SUCCESS in 1h 35m 55s\n- tripleo-ci-centos-9-scenario004-standalone https://zuul.opendev.org/t/openstack/build/e7aa2361d4ea4dfcac8c25254180f53c : SUCCESS in 1h 20m 30s\n- tripleo-ci-centos-9-scenario010-standalone https://zuul.opendev.org/t/openstack/build/caa65f7bf0df4f5bb4684c137b3e0ada : SUCCESS in 1h 52m 41s\n- tripleo-ci-centos-9-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/a20b00b1f6cb4b0db35fd8bece896f43 : SUCCESS in 1h 17m 29s\n- tripleo-ci-centos-9-standalone-on-multinode-ipa https://zuul.opendev.org/t/openstack/build/059d6967755344a7b66657a5d3c06ca2 : FAILURE in 1h 17m 48s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/dea51061e0cf484390218086b3eff4c8 : SUCCESS in 1h 32m 32s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/4f55424f2a9a4dfa93f7596a94ac714d : SUCCESS in 20m 08s","accounts_in_message":[],"_revision_number":9},{"id":"d208f6414fa5aac6f0e547d91b4a7cc6b66a105f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-06-24 00:34:29.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/2c210a49a3974dc2b9009ff7ba8ae491 : FAILURE in 1h 31m 23s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/4acb949c06984a2ba6300978201a240e : FAILURE in 1h 25m 14s","accounts_in_message":[],"_revision_number":9},{"id":"47d2b71b8122f7196a4fb9c13b7298e6e34425ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2022-06-24 01:38:45.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"7f6d19528aa9a9dea1f217f017d478b720f60ece","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2022-06-24 01:38:49.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"d6b4d7541ed5174e35c5a4a9e3156004d34d4186","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-24 01:39:08.000000000","message":"Patch Set 10:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1107/","accounts_in_message":[],"_revision_number":10},{"id":"2d91134af2eaf85c43aafbb48fd78b65fc8ccf9a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-24 02:24:15.000000000","message":"Patch Set 10:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1107/ : FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"427df4ed96f4f51731312f480eeeb7052284f10a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-06-24 03:55:13.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/c1d77aece6ab44e9976027c6ce189a25 : FAILURE in 1h 34m 43s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/54361c21b6fc4a7ebc96a8838f58e94b : FAILURE in 1h 37m 25s","accounts_in_message":[],"_revision_number":10},{"id":"a772e559ccca04f38a944fe8c623258649903268","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-06-24 05:50:20.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"7fd78b6ef798a71f5e734cad1943ca7a41c7d3ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 05:52:02.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/736d893935d2445fb62405f6c724296e : SUCCESS in 4h 03m 17s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/9824dc4d28914321b77a50184c9d072c : SUCCESS in 43m 40s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1ff93c2917924da28796e4bc1f75b373 : SUCCESS in 4m 17s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/7f4f84bf01ba48b39f96831cbe0d8a6f : SUCCESS in 2h 13m 23s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/36fd9e3745874c5b96a033ff9f50f845 : SUCCESS in 1h 15m 17s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/1f7f85e0f6344e1a8bed45b09826ad46 : SUCCESS in 1h 26m 30s\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/994f7f05bff34b76bd58839499eda3a4 : SUCCESS in 1h 37m 37s\n- tripleo-ci-centos-9-scenario004-standalone https://zuul.opendev.org/t/openstack/build/0f57a5f2efdf4f3f9fdba5adaff86d1f : SUCCESS in 1h 45m 17s\n- tripleo-ci-centos-9-scenario010-standalone https://zuul.opendev.org/t/openstack/build/89e45c2c88634098999c8c12c6ce645f : SUCCESS in 1h 35m 53s\n- tripleo-ci-centos-9-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/6fabf6b0c20d48b1af5d8d550d3a32d3 : SUCCESS in 1h 16m 35s\n- tripleo-ci-centos-9-standalone-on-multinode-ipa https://zuul.opendev.org/t/openstack/build/b86480e1bc374205817cf7044a95efce : SUCCESS in 2h 51m 18s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/ac3ded6b60524af48e95f31391d11a0b : SUCCESS in 1h 12m 43s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/5b441f1c214446a597e3858b9bdfeb32 : SUCCESS in 20m 13s","accounts_in_message":[],"_revision_number":10},{"id":"9092797382b505453eaf0eef0344c55837ddb419","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 05:52:51.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"ececefd1684cf96bc69cdd0d45ca4c957be6d7a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 09:52:31.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/6efa83cbde2540c7b2c67befdd65b718 : SUCCESS in 3h 54m 30s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/992e27113d7e4b5bad40cf9987700ee2 : SUCCESS in 59m 50s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/1c32aa4c1e364fa8b4f48d89c6ddc1b7 : SUCCESS in 2h 59m 42s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/fa46242ac903464fb7417e7a1ce95435 : SUCCESS in 1h 46m 10s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/219da0a75bd54048accd7c931deef9aa : SUCCESS in 1h 34m 56s\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/a5dcf51e6d39479daf93bf99bb8e7c75 : SUCCESS in 2h 37m 51s\n- tripleo-ci-centos-9-scenario004-standalone https://zuul.opendev.org/t/openstack/build/40ba1e3e176e4ed297e05d7fba7f4adb : SUCCESS in 1h 47m 11s\n- tripleo-ci-centos-9-scenario010-standalone https://zuul.opendev.org/t/openstack/build/6eb14fb887e84420ac04a69864d96a5b : SUCCESS in 1h 35m 21s\n- tripleo-ci-centos-9-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/2307aec1fd374bf0a930b3920be3fbd2 : SUCCESS in 2h 00m 05s\n- tripleo-ci-centos-9-standalone-on-multinode-ipa https://zuul.opendev.org/t/openstack/build/fe3ba5a1409a4ac1989de7afaea764ee : SUCCESS in 2h 08m 00s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/4d2b8429251a4ff6a7a5c77216624f25 : SUCCESS in 1h 24m 06s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/205db27553d041c0b6c2650c53596aeb : SUCCESS in 4m 35s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/f1921f6cebbf44d3a42fbd4e28b77577 : SUCCESS in 19m 57s","accounts_in_message":[],"_revision_number":10},{"id":"663f44159645f92a7c8c0e52b47b10265605fc61","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 09:52:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"0b586c6dcee71297fd875ec6b96fc0492bb09e7b","revisions":{"cd80d72ac4a875ed3696e59e0a1b6573264c746c":{"kind":"REWORK","_number":1,"created":"2022-06-16 22:08:08.000000000","uploader":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"ref":"refs/changes/31/846231/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/31/846231/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/846231/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/846231/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/846231/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/846231/1"}}},"commit":{"parents":[{"commit":"c5d5c0138f34ee7dda31a7fc0908045bb1c413ae","subject":"Merge \"Deploy Ceph before the Overcloud for standalone CI\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/c5d5c0138f34ee7dda31a7fc0908045bb1c413ae"}]}],"author":{"name":"John Fulton","email":"fulton@redhat.com","date":"2022-06-16 22:07:54.000000000","tz":-240},"committer":{"name":"John Fulton","email":"fulton@redhat.com","date":"2022-06-16 22:07:54.000000000","tz":-240},"subject":"Override Ceph --container-namespace for periodic jobs","message":"Override Ceph --container-namespace for periodic jobs\n\nIf the standalone job is periodic, use \u0027openstack overcloud\nceph deploy --container-namespace\u0027 to pull the container\ndirectly from quay.rdoproject.org.\n\nStandalone scenarios usually consume containers from a\ncontent-provider but periodic jobs pull them directly\nfrom quay.rdoproject.org. The periodic jobs then push\nthem to the undercloud as a container registry. Non-\nperiodic jobs do not push to the undercloud registry.\n\nWhen I982dedb53582fbd76391165c3ca72954c129b84a merged,\nperiodic standalone jobs broke because the undercloud\ncontainer registry was not configured when \u0027openstack\novercloud ceph deploy\u0027 was run. Because the container\nprepare file has a push_destination, deployed ceph\nassumes the containers were prepared in advance so it\nswaps out the container namespace as indicated by\npush_destination directive. Though this still happens\nwe override it again with --container-namespace.\n\nChange-Id: I1abfbbd23ca93c01393d05057806ba9cc846fbed\nCloses-bug: #1978998\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/cd80d72ac4a875ed3696e59e0a1b6573264c746c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/cd80d72ac4a875ed3696e59e0a1b6573264c746c"}]},"branch":"refs/heads/master"},"70ac7c11f7284bdf04218d40d182cd83f0a1b8f9":{"kind":"REWORK","_number":2,"created":"2022-06-17 14:22:21.000000000","uploader":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"ref":"refs/changes/31/846231/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/31/846231/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/846231/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/846231/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/846231/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/846231/2"}}},"commit":{"parents":[{"commit":"c5d5c0138f34ee7dda31a7fc0908045bb1c413ae","subject":"Merge \"Deploy Ceph before the Overcloud for standalone CI\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/c5d5c0138f34ee7dda31a7fc0908045bb1c413ae"}]}],"author":{"name":"John Fulton","email":"fulton@redhat.com","date":"2022-06-16 22:07:54.000000000","tz":-240},"committer":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2022-06-17 14:22:19.000000000","tz":120},"subject":"Override Ceph --container-namespace for periodic jobs","message":"Override Ceph --container-namespace for periodic jobs\n\nIf the standalone job is periodic, use \u0027openstack overcloud\nceph deploy --container-namespace\u0027 to pull the container\ndirectly from quay.rdoproject.org.\n\nStandalone scenarios usually consume containers from a\ncontent-provider but periodic jobs pull them directly\nfrom quay.rdoproject.org. The periodic jobs then push\nthem to the undercloud as a container registry. Non-\nperiodic jobs do not push to the undercloud registry.\n\nWhen I982dedb53582fbd76391165c3ca72954c129b84a merged,\nperiodic standalone jobs broke because the undercloud\ncontainer registry was not configured when \u0027openstack\novercloud ceph deploy\u0027 was run. Because the container\nprepare file has a push_destination, deployed ceph\nassumes the containers were prepared in advance so it\nswaps out the container namespace as indicated by\npush_destination directive. Though this still happens\nwe override it again with --container-namespace.\n\nChange-Id: I1abfbbd23ca93c01393d05057806ba9cc846fbed\nCloses-bug: #1978998\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/70ac7c11f7284bdf04218d40d182cd83f0a1b8f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/70ac7c11f7284bdf04218d40d182cd83f0a1b8f9"}]},"branch":"refs/heads/master"},"46ba5913371533fa1d2e22180d101e1e1e3d7fdd":{"kind":"REWORK","_number":3,"created":"2022-06-17 14:25:08.000000000","uploader":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"ref":"refs/changes/31/846231/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/31/846231/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/846231/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/846231/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/846231/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/846231/3"}}},"commit":{"parents":[{"commit":"c5d5c0138f34ee7dda31a7fc0908045bb1c413ae","subject":"Merge \"Deploy Ceph before the Overcloud for standalone CI\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/c5d5c0138f34ee7dda31a7fc0908045bb1c413ae"}]}],"author":{"name":"John Fulton","email":"fulton@redhat.com","date":"2022-06-16 22:07:54.000000000","tz":-240},"committer":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2022-06-17 14:25:04.000000000","tz":120},"subject":"Override Ceph --container-namespace for periodic jobs","message":"Override Ceph --container-namespace for periodic jobs\n\nIf the standalone job is periodic, use \u0027openstack overcloud\nceph deploy --container-namespace\u0027 to pull the container\ndirectly from quay.rdoproject.org.\n\nStandalone scenarios usually consume containers from a\ncontent-provider but periodic jobs pull them directly\nfrom quay.rdoproject.org. The periodic jobs then push\nthem to the undercloud as a container registry. Non-\nperiodic jobs do not push to the undercloud registry.\n\nWhen I982dedb53582fbd76391165c3ca72954c129b84a merged,\nperiodic standalone jobs broke because the undercloud\ncontainer registry was not configured when \u0027openstack\novercloud ceph deploy\u0027 was run. Because the container\nprepare file has a push_destination, deployed ceph\nassumes the containers were prepared in advance so it\nswaps out the container namespace as indicated by\npush_destination directive. Though this still happens\nwe override it again with --container-namespace.\n\nChange-Id: I1abfbbd23ca93c01393d05057806ba9cc846fbed\nCloses-bug: #1978998\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/46ba5913371533fa1d2e22180d101e1e1e3d7fdd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/46ba5913371533fa1d2e22180d101e1e1e3d7fdd"}]},"branch":"refs/heads/master"},"9ca35c4fcbcd312c29965d21c1dee830b4c3aa7c":{"kind":"REWORK","_number":4,"created":"2022-06-17 16:41:40.000000000","uploader":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"ref":"refs/changes/31/846231/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/31/846231/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/846231/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/846231/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/846231/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/846231/4"}}},"commit":{"parents":[{"commit":"c5d5c0138f34ee7dda31a7fc0908045bb1c413ae","subject":"Merge \"Deploy Ceph before the Overcloud for standalone CI\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/c5d5c0138f34ee7dda31a7fc0908045bb1c413ae"}]}],"author":{"name":"John Fulton","email":"fulton@redhat.com","date":"2022-06-16 22:07:54.000000000","tz":-240},"committer":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2022-06-17 16:41:35.000000000","tz":120},"subject":"Override Ceph --container-namespace for periodic jobs","message":"Override Ceph --container-namespace for periodic jobs\n\nIf the standalone job is periodic, use \u0027openstack overcloud\nceph deploy --container-namespace\u0027 to pull the container\ndirectly from quay.rdoproject.org.\n\nStandalone scenarios usually consume containers from a\ncontent-provider but periodic jobs pull them directly\nfrom quay.rdoproject.org. The periodic jobs then push\nthem to the undercloud as a container registry. Non-\nperiodic jobs do not push to the undercloud registry.\n\nWhen I982dedb53582fbd76391165c3ca72954c129b84a merged,\nperiodic standalone jobs broke because the undercloud\ncontainer registry was not configured when \u0027openstack\novercloud ceph deploy\u0027 was run. Because the container\nprepare file has a push_destination, deployed ceph\nassumes the containers were prepared in advance so it\nswaps out the container namespace as indicated by\npush_destination directive. Though this still happens\nwe override it again with --container-namespace.\n\nChange-Id: I1abfbbd23ca93c01393d05057806ba9cc846fbed\nCloses-bug: #1978998\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/9ca35c4fcbcd312c29965d21c1dee830b4c3aa7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/9ca35c4fcbcd312c29965d21c1dee830b4c3aa7c"}]},"branch":"refs/heads/master"},"a88c3e0b741c36da2939e65eb64a5f0987fb33bc":{"kind":"REWORK","_number":5,"created":"2022-06-17 21:56:47.000000000","uploader":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"ref":"refs/changes/31/846231/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/31/846231/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/846231/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/846231/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/846231/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/846231/5"}}},"commit":{"parents":[{"commit":"c5d5c0138f34ee7dda31a7fc0908045bb1c413ae","subject":"Merge \"Deploy Ceph before the Overcloud for standalone CI\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/c5d5c0138f34ee7dda31a7fc0908045bb1c413ae"}]}],"author":{"name":"John Fulton","email":"fulton@redhat.com","date":"2022-06-16 22:07:54.000000000","tz":-240},"committer":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2022-06-17 21:56:39.000000000","tz":120},"subject":"Override Ceph --container-namespace for periodic jobs","message":"Override Ceph --container-namespace for periodic jobs\n\nIf the standalone job is periodic, use \u0027openstack overcloud\nceph deploy --container-namespace\u0027 to pull the container\ndirectly from quay.rdoproject.org.\n\nStandalone scenarios usually consume containers from a\ncontent-provider but periodic jobs pull them directly\nfrom quay.rdoproject.org. The periodic jobs then push\nthem to the undercloud as a container registry. Non-\nperiodic jobs do not push to the undercloud registry.\n\nWhen I982dedb53582fbd76391165c3ca72954c129b84a merged,\nperiodic standalone jobs broke because the undercloud\ncontainer registry was not configured when \u0027openstack\novercloud ceph deploy\u0027 was run. Because the container\nprepare file has a push_destination, deployed ceph\nassumes the containers were prepared in advance so it\nswaps out the container namespace as indicated by\npush_destination directive. Though this still happens\nwe override it again with --container-namespace.\n\nChange-Id: I1abfbbd23ca93c01393d05057806ba9cc846fbed\nCloses-bug: #1978998\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/a88c3e0b741c36da2939e65eb64a5f0987fb33bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/a88c3e0b741c36da2939e65eb64a5f0987fb33bc"}]},"branch":"refs/heads/master"},"761799931aa12472d560c4c2a7b122599c41ab0f":{"kind":"REWORK","_number":6,"created":"2022-06-18 06:59:00.000000000","uploader":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"ref":"refs/changes/31/846231/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/31/846231/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/846231/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/846231/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/846231/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/846231/6"}}},"commit":{"parents":[{"commit":"c5d5c0138f34ee7dda31a7fc0908045bb1c413ae","subject":"Merge \"Deploy Ceph before the Overcloud for standalone CI\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/c5d5c0138f34ee7dda31a7fc0908045bb1c413ae"}]}],"author":{"name":"John Fulton","email":"fulton@redhat.com","date":"2022-06-16 22:07:54.000000000","tz":-240},"committer":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2022-06-18 06:58:47.000000000","tz":120},"subject":"Override Ceph --container-namespace for periodic jobs","message":"Override Ceph --container-namespace for periodic jobs\n\nIf the standalone job is periodic, use \u0027openstack overcloud\nceph deploy --container-namespace\u0027 to pull the container\ndirectly from quay.rdoproject.org.\n\nStandalone scenarios usually consume containers from a\ncontent-provider but periodic jobs pull them directly\nfrom quay.rdoproject.org. The periodic jobs then push\nthem to the undercloud as a container registry. Non-\nperiodic jobs do not push to the undercloud registry.\n\nWhen I982dedb53582fbd76391165c3ca72954c129b84a merged,\nperiodic standalone jobs broke because the undercloud\ncontainer registry was not configured when \u0027openstack\novercloud ceph deploy\u0027 was run. Because the container\nprepare file has a push_destination, deployed ceph\nassumes the containers were prepared in advance so it\nswaps out the container namespace as indicated by\npush_destination directive. Though this still happens\nwe override it again with --container-namespace.\n\nChange-Id: I1abfbbd23ca93c01393d05057806ba9cc846fbed\nCloses-bug: #1978998\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/761799931aa12472d560c4c2a7b122599c41ab0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/761799931aa12472d560c4c2a7b122599c41ab0f"}]},"branch":"refs/heads/master"},"d66f2aa18c00fd9e6b73529e5b853503c127a74d":{"kind":"REWORK","_number":7,"created":"2022-06-18 07:13:09.000000000","uploader":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"ref":"refs/changes/31/846231/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/31/846231/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/846231/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/846231/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/846231/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/846231/7"}}},"commit":{"parents":[{"commit":"c5d5c0138f34ee7dda31a7fc0908045bb1c413ae","subject":"Merge \"Deploy Ceph before the Overcloud for standalone CI\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/c5d5c0138f34ee7dda31a7fc0908045bb1c413ae"}]}],"author":{"name":"John Fulton","email":"fulton@redhat.com","date":"2022-06-16 22:07:54.000000000","tz":-240},"committer":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2022-06-18 07:13:04.000000000","tz":120},"subject":"Override Ceph --container-namespace for periodic jobs","message":"Override Ceph --container-namespace for periodic jobs\n\nIf the standalone job is periodic, use \u0027openstack overcloud\nceph deploy --container-namespace\u0027 to pull the container\ndirectly from quay.rdoproject.org.\n\nStandalone scenarios usually consume containers from a\ncontent-provider but periodic jobs pull them directly\nfrom quay.rdoproject.org. The periodic jobs then push\nthem to the undercloud as a container registry. Non-\nperiodic jobs do not push to the undercloud registry.\n\nWhen I982dedb53582fbd76391165c3ca72954c129b84a merged,\nperiodic standalone jobs broke because the undercloud\ncontainer registry was not configured when \u0027openstack\novercloud ceph deploy\u0027 was run. Because the container\nprepare file has a push_destination, deployed ceph\nassumes the containers were prepared in advance so it\nswaps out the container namespace as indicated by\npush_destination directive. Though this still happens\nwe override it again with --container-namespace.\n\nChange-Id: I1abfbbd23ca93c01393d05057806ba9cc846fbed\nCloses-bug: #1978998\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/d66f2aa18c00fd9e6b73529e5b853503c127a74d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/d66f2aa18c00fd9e6b73529e5b853503c127a74d"}]},"branch":"refs/heads/master"},"0e36183c7a91df51f262c3354a6d8121aa14a7c9":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2022-06-18 10:38:01.000000000","uploader":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"ref":"refs/changes/31/846231/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/31/846231/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/846231/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/846231/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/846231/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/846231/8"}}},"commit":{"parents":[{"commit":"d2008175420e839925bf0856272cdfeb3d280032","subject":"Only deploy ceph before the overcloud in master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/d2008175420e839925bf0856272cdfeb3d280032"}]}],"author":{"name":"John Fulton","email":"fulton@redhat.com","date":"2022-06-16 22:07:54.000000000","tz":-240},"committer":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2022-06-18 10:38:01.000000000","tz":0},"subject":"Override Ceph --container-namespace for periodic jobs","message":"Override Ceph --container-namespace for periodic jobs\n\nIf the standalone job is periodic, use \u0027openstack overcloud\nceph deploy --container-namespace\u0027 to pull the container\ndirectly from quay.rdoproject.org.\n\nStandalone scenarios usually consume containers from a\ncontent-provider but periodic jobs pull them directly\nfrom quay.rdoproject.org. The periodic jobs then push\nthem to the undercloud as a container registry. Non-\nperiodic jobs do not push to the undercloud registry.\n\nWhen I982dedb53582fbd76391165c3ca72954c129b84a merged,\nperiodic standalone jobs broke because the undercloud\ncontainer registry was not configured when \u0027openstack\novercloud ceph deploy\u0027 was run. Because the container\nprepare file has a push_destination, deployed ceph\nassumes the containers were prepared in advance so it\nswaps out the container namespace as indicated by\npush_destination directive. Though this still happens\nwe override it again with --container-namespace.\n\nChange-Id: I1abfbbd23ca93c01393d05057806ba9cc846fbed\nCloses-bug: #1978998\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/0e36183c7a91df51f262c3354a6d8121aa14a7c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/0e36183c7a91df51f262c3354a6d8121aa14a7c9"}]},"branch":"refs/heads/master","description":"Rebase"},"43a818138173195e00710bc48ba8d1fd1c7d34e8":{"kind":"REWORK","_number":9,"created":"2022-06-23 13:56:03.000000000","uploader":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"ref":"refs/changes/31/846231/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/31/846231/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/846231/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/846231/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/846231/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/846231/9"}}},"commit":{"parents":[{"commit":"d2008175420e839925bf0856272cdfeb3d280032","subject":"Only deploy ceph before the overcloud in master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/d2008175420e839925bf0856272cdfeb3d280032"}]}],"author":{"name":"John Fulton","email":"fulton@redhat.com","date":"2022-06-16 22:07:54.000000000","tz":-240},"committer":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2022-06-23 13:56:01.000000000","tz":120},"subject":"Override Ceph --container-namespace for periodic jobs","message":"Override Ceph --container-namespace for periodic jobs\n\nIf the standalone job is periodic, use \u0027openstack overcloud\nceph deploy --container-namespace\u0027 to pull the container\ndirectly from quay.rdoproject.org.\n\nStandalone scenarios usually consume containers from a\ncontent-provider but periodic jobs pull them directly\nfrom quay.rdoproject.org. The periodic jobs then push\nthem to the undercloud as a container registry. Non-\nperiodic jobs do not push to the undercloud registry.\n\nWhen I982dedb53582fbd76391165c3ca72954c129b84a merged,\nperiodic standalone jobs broke because the undercloud\ncontainer registry was not configured when \u0027openstack\novercloud ceph deploy\u0027 was run. Because the container\nprepare file has a push_destination, deployed ceph\nassumes the containers were prepared in advance so it\nswaps out the container namespace as indicated by\npush_destination directive. Though this still happens\nwe override it again with --container-namespace.\n\nChange-Id: I1abfbbd23ca93c01393d05057806ba9cc846fbed\nCloses-bug: #1978998\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/43a818138173195e00710bc48ba8d1fd1c7d34e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/43a818138173195e00710bc48ba8d1fd1c7d34e8"}]},"branch":"refs/heads/master"},"0b586c6dcee71297fd875ec6b96fc0492bb09e7b":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2022-06-24 01:38:45.000000000","uploader":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"ref":"refs/changes/31/846231/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/31/846231/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/846231/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/846231/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/846231/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/31/846231/10"}}},"commit":{"parents":[{"commit":"17d3f6af73cf9eca7e20c08e5c3c2cdae8ee9bc2","subject":"Merge \"Run dstat on overcloud nodes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/17d3f6af73cf9eca7e20c08e5c3c2cdae8ee9bc2"}]}],"author":{"name":"John Fulton","email":"fulton@redhat.com","date":"2022-06-16 22:07:54.000000000","tz":-240},"committer":{"name":"Ronelle Landy","email":"rlandy@redhat.com","date":"2022-06-24 01:38:45.000000000","tz":0},"subject":"Override Ceph --container-namespace for periodic jobs","message":"Override Ceph --container-namespace for periodic jobs\n\nIf the standalone job is periodic, use \u0027openstack overcloud\nceph deploy --container-namespace\u0027 to pull the container\ndirectly from quay.rdoproject.org.\n\nStandalone scenarios usually consume containers from a\ncontent-provider but periodic jobs pull them directly\nfrom quay.rdoproject.org. The periodic jobs then push\nthem to the undercloud as a container registry. Non-\nperiodic jobs do not push to the undercloud registry.\n\nWhen I982dedb53582fbd76391165c3ca72954c129b84a merged,\nperiodic standalone jobs broke because the undercloud\ncontainer registry was not configured when \u0027openstack\novercloud ceph deploy\u0027 was run. Because the container\nprepare file has a push_destination, deployed ceph\nassumes the containers were prepared in advance so it\nswaps out the container namespace as indicated by\npush_destination directive. Though this still happens\nwe override it again with --container-namespace.\n\nChange-Id: I1abfbbd23ca93c01393d05057806ba9cc846fbed\nCloses-bug: #1978998\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/0b586c6dcee71297fd875ec6b96fc0492bb09e7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/0b586c6dcee71297fd875ec6b96fc0492bb09e7b"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"}}]}],"submit_requirements":[]}
