)]}'
{"id":"openstack%2Ftripleo-quickstart-extras~847419","triplet_id":"openstack%2Ftripleo-quickstart-extras~master~I0d3966af5fc6b9236c5bc255282aadd3c48aaa63","project":"openstack/tripleo-quickstart-extras","branch":"master","topic":"build-test-packages","attention_set":{},"removed_from_attention_set":{"34208":{"account":{"_account_id":34208,"name":"Rafael Castillo","email":"rcastill@redhat.com","username":"rcastill"},"last_update":"2022-07-15 14:27:48.000000000","reason":"Change was abandoned"},"8449":{"account":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"last_update":"2022-07-15 14:27:48.000000000","reason":"Change was abandoned"},"30002":{"account":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"last_update":"2022-07-15 14:27:48.000000000","reason":"Change was abandoned"},"32962":{"account":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"last_update":"2022-07-15 14:25:32.000000000","reason":"\u003cGERRIT_ACCOUNT_32962\u003e replied on the change","reason_account":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"}},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2022-07-15 14:27:48.000000000","reason":"Change was abandoned"},"9976":{"account":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"last_update":"2022-07-15 14:27:48.000000000","reason":"Change was abandoned"},"12393":{"account":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"last_update":"2022-07-15 07:02:57.000000000","reason":"\u003cGERRIT_ACCOUNT_12393\u003e replied on the change","reason_account":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"}},"6926":{"account":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"last_update":"2022-07-14 15:44:18.000000000","reason":"\u003cGERRIT_ACCOUNT_6926\u003e replied on the change","reason_account":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}}},"hashtags":[],"change_id":"I0d3966af5fc6b9236c5bc255282aadd3c48aaa63","subject":"Allow passing changes to build-test-packages in zuul job vars","status":"ABANDONED","created":"2022-06-23 23:31:39.000000000","updated":"2023-05-16 11:09:35.000000000","total_comment_count":20,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"694aee2159ca202c7230f808cef943ca2321f52e","_number":847419,"virtual_id_number":847419,"owner":{"_account_id":34208,"name":"Rafael Castillo","email":"rcastill@redhat.com","username":"rcastill"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"all":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"tag":"autogenerated:zuul:gate","value":0,"date":"2022-07-15 14:16:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2022-07-14 13:42:05.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"recommended":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"all":[{"value":1,"date":"2022-07-15 06:42:17.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":1,"date":"2022-07-14 20:03:31.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":-1,"date":"2022-07-15 14:25:32.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2022-07-14 19:12:27.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":1,"date":"2022-07-14 16:15:15.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":1,"date":"2022-07-15 07:02:57.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":2,"date":"2022-07-14 19:24:01.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-14 15:44:18.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"all":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2022-07-15 14:15:57.000000000","_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"}],"CC":[{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-23 23:32:02.000000000","updated_by":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-24 04:47:18.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-24 10:20:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-24 12:01:58.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2022-07-01 15:15:20.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"CC"},{"updated":"2022-07-01 15:15:20.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2022-07-13 09:54:12.000000000","updated_by":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"reviewer":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"state":"CC"},{"updated":"2022-07-14 13:10:37.000000000","updated_by":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"reviewer":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"state":"REVIEWER"},{"updated":"2022-07-14 15:44:18.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"},{"updated":"2022-07-14 16:15:15.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2022-07-14 19:09:09.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"CC"},{"updated":"2022-07-14 19:09:09.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2022-07-14 19:12:27.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"REVIEWER"},{"updated":"2022-07-14 19:24:01.000000000","updated_by":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"reviewer":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"state":"REVIEWER"},{"updated":"2022-07-15 06:42:17.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"}],"messages":[{"id":"3cb49428061158d4dc4debe43187f664cef02c99","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34208,"name":"Rafael Castillo","email":"rcastill@redhat.com","username":"rcastill"},"date":"2022-06-23 23:31:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1d93601e6612ef27998db6565405ad02cbf04040","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-23 23:32:02.000000000","message":"Patch Set 1:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1103/","accounts_in_message":[],"_revision_number":1},{"id":"61b963dbc269be2aaafcd82c9ab0b792286253e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34208,"name":"Rafael Castillo","email":"rcastill@redhat.com","username":"rcastill"},"date":"2022-06-23 23:32:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"db5e6c857adead09b51a34a9be433b8f616157df","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-23 23:32:47.000000000","message":"Patch Set 2:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1104/","accounts_in_message":[],"_revision_number":2},{"id":"2244f9eb12a20264b38fdd397a9d0c86f96d640d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34208,"name":"Rafael Castillo","email":"rcastill@redhat.com","username":"rcastill"},"date":"2022-06-23 23:34:07.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ff6b4804578266ff2628b57b03fd1881d78af60e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-23 23:34:28.000000000","message":"Patch Set 3:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1105/","accounts_in_message":[],"_revision_number":3},{"id":"bbb5771abe99a374a6c40ec80111bb5448d70c94","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-23 23:35:07.000000000","message":"Patch Set 1:\n\nBuild Aborted \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1103/ : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"2faccb75f8f6da13c7e3903208bf2ab5d4ba8cb1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-23 23:36:07.000000000","message":"Patch Set 2:\n\nBuild Aborted \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1104/ : ABORTED","accounts_in_message":[],"_revision_number":2},{"id":"3a485f2a6665a886fb6b695c63960e27cdb6b6cb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-24 00:53:00.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1105/ : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"33d0603a1b5474686b1d4747ec3772bad0920713","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34208,"name":"Rafael Castillo","email":"rcastill@redhat.com","username":"rcastill"},"date":"2022-06-24 01:16:52.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"518f04e98bcaf3ecf0a268469526ea4f45b04232","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-24 01:17:08.000000000","message":"Patch Set 4:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1106/","accounts_in_message":[],"_revision_number":4},{"id":"2e6f2634d6db886f1227463ccc9cbddf4e6d49bb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-24 02:36:37.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1106/ : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"3cb3dbb1b50bc7e9fd5ebae91cebd65ed018348d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34208,"name":"Rafael Castillo","email":"rcastill@redhat.com","username":"rcastill"},"date":"2022-06-24 02:37:24.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a9b5fcad2adc9719413aa6c57229df143e48e245","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-24 02:37:37.000000000","message":"Patch Set 5:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1109/","accounts_in_message":[],"_revision_number":5},{"id":"0ed6b9a04d4d4b3d46df6c34b392e3654fdd9aea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34208,"name":"Rafael Castillo","email":"rcastill@redhat.com","username":"rcastill"},"date":"2022-06-24 03:08:55.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"773181c02b455c4d8657cb4356550c99969d96a2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-24 03:09:18.000000000","message":"Patch Set 6:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1110/","accounts_in_message":[],"_revision_number":6},{"id":"878fcde04ddb05721744ee42e140d01e71535a00","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-24 03:15:23.000000000","message":"Patch Set 5:\n\nBuild Aborted \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1109/ : ABORTED","accounts_in_message":[],"_revision_number":5},{"id":"40fa5f6177d7512de8e25af990293a3fa8b1acb8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-24 04:29:33.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1110/ : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"6e7357845340b90a59236f217f80d0f16d93bb03","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-06-24 04:47:18.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/616398795f6f45ae903474c08be40085 : FAILURE in 1h 22m 17s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/59f13360be5f4c00aaefaf33a4907104 : FAILURE in 1h 20m 31s","accounts_in_message":[],"_revision_number":6},{"id":"5efdcb7dc5be879310141e564dbf334ca1faec4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34208,"name":"Rafael Castillo","email":"rcastill@redhat.com","username":"rcastill"},"date":"2022-06-24 06:09:00.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"69129b4745c6b98eecd98ac608e1bef80d80f456","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-24 06:09:23.000000000","message":"Patch Set 7:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1111/","accounts_in_message":[],"_revision_number":7},{"id":"10f541f0b4d57e2b1f063fdc6ba43fc840172aae","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-24 07:28:17.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1111/ : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"4a6579eea377f98db6a4a9fcb953e10fba896f39","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-06-24 07:56:37.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/37acaaafc17c45aaac58c0933dacf4ad : FAILURE in 1h 30m 14s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/f415ae5a825245a48b40180386698dd3 : FAILURE in 1h 27m 06s","accounts_in_message":[],"_revision_number":7},{"id":"aff2ce38b1dbd30d2a4fe82003f6c81964ce4514","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 10:20:39.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/59dcf8b1dde440f0bff20d7100628836 : SUCCESS in 3h 52m 07s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/bebdc8b40d7d438e9cc9469287ae0ac5 : SUCCESS in 38m 34s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/8d475eb7cacc4d67bb61967fcfa90ea2 : POST_FAILURE in 2h 25m 28s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/03509170235e4e4592b0dbc27de2a8ab : SUCCESS in 1h 30m 02s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/028bf9fc763f40b7a31d89024e3a7ec1 : SUCCESS in 4m 58s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/49f0e31085b64ffeaeb6c915fbf47d00 : SUCCESS in 2h 15m 22s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/29370ae6483541a29527adb93e21a09d : SUCCESS in 1h 20m 37s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/baa4c03881b14cf581bf8407449a55fc : SUCCESS in 1h 47m 46s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/829cd90701d14c6fac81b3e215a662d1 : SUCCESS in 1h 12m 17s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/2ccf991686684134a7dc2783c8f496b3 : SUCCESS in 1h 08m 16s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/89cb9e870a5f4dd4ab4b6c78cb06cc1d : SUCCESS in 20m 18s","accounts_in_message":[],"_revision_number":7},{"id":"face989ef61d64be1652c9ce58c7d52ab1fe92bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2022-06-24 12:01:33.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"44e5d4b8e66d77e97d0beef90e14e1d63732329b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-24 12:01:55.000000000","message":"Patch Set 8:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1112/","accounts_in_message":[],"_revision_number":8},{"id":"90a079ba1f5ff407a811f512efff9b4dc551f944","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2022-06-24 12:01:58.000000000","message":"Patch Set 8: Workflow-1","accounts_in_message":[],"_revision_number":8},{"id":"72ac23b582d1f79ec9612f0990a8b93f93325604","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 12:20:08.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/014c6d766faf42b59fd8acfab355425b : SKIPPED\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/01ff09bfdbfd40e88ca652473e349782 : SKIPPED\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/f50b5408964a4b65b96b12ce4d84c70f : SKIPPED\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/af5e564a6cb84e489056a7ad2372c03f : SKIPPED\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1f31c777383841b39142983590f4f4ef : FAILURE in 4m 19s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/cf80f45e5e1b4338aa591e26e2e6eef2 : SKIPPED\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/b4bbbb7bd07a4f3db3dce1c8c0c9671d : SKIPPED\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/e902197817c24eaca1b780c190ef168e : SKIPPED\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/1193fbd0c9954d86aaf3460d54113307 : SKIPPED\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/6710fe6c699b4a1cac7d0ed36eda14cd : SKIPPED\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/7448887b7bd948e193682ed19fc3155d : SUCCESS in 16m 12s","accounts_in_message":[],"_revision_number":8},{"id":"382b0f37c8e236cde0911b5b63af250f456f86e0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-06-24 13:20:38.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1112/ : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"e1457e26392a077dc1c98a84cc39a900f36f6723","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-06-24 13:39:33.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/5c3e2dc9e2524d9199e3c716955f6359 : FAILURE in 1h 22m 15s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/82d689820d2245b79dd545e6777a3da1 : FAILURE in 1h 28m 01s","accounts_in_message":[],"_revision_number":8},{"id":"ff40b284b2fde216208b66a05e561650db8b29f3","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2022-07-01 15:15:20.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"2cd4348bc3f37be4bbe6d4a6538b8e9b064d5103","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2022-07-01 18:38:54.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"557679b67fb16ca5bc33b23ef9fe8d0cf63ed6b9","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2022-07-01 18:40:13.000000000","message":"Abandoned\n\nThis change was used to test artg_change_list as job variable. We will work on a different patch to propose an entension of this role.","accounts_in_message":[],"_revision_number":8},{"id":"291c6f2c039a3579b924cc40835266ae01ec7252","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-07-04 12:57:40.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"a24b564a0c9075e398eb332c2cd23cb5fa9ef6ca","tag":"autogenerated:gerrit:restore","author":{"_account_id":34208,"name":"Rafael Castillo","email":"rcastill@redhat.com","username":"rcastill"},"date":"2022-07-12 15:51:26.000000000","message":"Restored","accounts_in_message":[],"_revision_number":8},{"id":"f22e937b4871629afb9464e988e5c82fee167e08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 16:14:54.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9f8c5930580044a1a7e88a10f436a573\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/996859c687894cb08952df63b0f753b8 : SKIPPED\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/5c5acfb2d87243da9ab7b2dc113fd919 : SKIPPED\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/43644c284afd43929a814d380705accf : SKIPPED\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/bfd1e7442df543058b74026d37951fb5 : SKIPPED\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/566d5ce7ed414718b3d4c7574eb7ae4f : FAILURE in 5m 28s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/d866df7c6d774b329e7503245dd12115 : SKIPPED\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/c9cc1601d5cb4cae84d98d7014ea3d2c : SKIPPED\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/58b90a7cf6894c71907248d1acd538dc : SKIPPED\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/063e9dae9ba74fce88826f3a20c84a01 : SKIPPED (non-voting)\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/f0900f2d893144d9b94443df3ca51628 : SKIPPED\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/c59323d4398f4564b0a52001fbb00601 : SKIPPED\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/1f7b7c61401f4196a4e846f1f883bbe7 : SUCCESS in 21m 36s","accounts_in_message":[],"_revision_number":8},{"id":"86edb94e0713663fe32c15db1b0898742a1daf2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34208,"name":"Rafael Castillo","email":"rcastill@redhat.com","username":"rcastill"},"date":"2022-07-12 16:37:06.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"93d1e723551d981dea68de5bdd13c511ee10d04d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-07-12 16:37:29.000000000","message":"Patch Set 9:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1153/","accounts_in_message":[],"_revision_number":9},{"id":"f145cc080e41d729127e7381c9f1cf8ecb91ffa4","author":{"_account_id":34208,"name":"Rafael Castillo","email":"rcastill@redhat.com","username":"rcastill"},"date":"2022-07-12 16:38:20.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"1be134cee823ee732450f843d05b3bc03ac0487b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-07-12 16:48:27.000000000","message":"Patch Set 9:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1153/ : FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"4fe513fcf66bca2ca44c436006a0cc960bb67c44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 19:46:57.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8369ea77ca46484c8bd57ae543cd7e9d\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/28624acbac83497e93c222d3294e2603 : SUCCESS in 2h 42m 35s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/a895468b7bf94baa8c0485e5a5f5666e : SUCCESS in 47m 03s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/7ab0c08fb44247bb9e9f8bb69656613d : SUCCESS in 1h 38m 44s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/8ece0e49e0334c80984da1e7d35487d7 : SUCCESS in 1h 32m 52s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/889e29a0ee104215ab24c8802d1204da : SUCCESS in 5m 24s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/6bae527e60e041b797b748da64804511 : SUCCESS in 2h 59m 10s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/54909cd80e504a4990c7b64bcded8160 : SUCCESS in 1h 49m 47s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/de7b247c360141d385842a86d03a424f : SUCCESS in 1h 18m 43s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/e0d06cac59bf430192a992af91df9563 : FAILURE in 1h 29m 51s (non-voting)\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/84084e75f0404df7b79ced081cad4e1a : SUCCESS in 1h 27m 36s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/8addb46996fd46a5aafce200d6df852b : SUCCESS in 1h 19m 36s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/3da589fa8027415a8c8ee3c53aad3bcc : SUCCESS in 22m 24s","accounts_in_message":[],"_revision_number":9},{"id":"137dfdf246e59e63b6ffdaa61cc847b8ad8848f5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-07-12 20:12:53.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/533e4f677bee495bb56c905560cca78d : FAILURE in 2h 14m 04s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/f9ab721c4e734030a0c5085fca279d33 : SUCCESS in 2h 03m 46s","accounts_in_message":[],"_revision_number":9},{"id":"28287cdbce3338fd87e713b66f75f7dfdc68f656","author":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"date":"2022-07-13 09:54:12.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"d5d268a55e1a9790cfcc173ff2a78c462e26d219","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34208,"name":"Rafael Castillo","email":"rcastill@redhat.com","username":"rcastill"},"date":"2022-07-13 18:57:23.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"43ddf4ead7ebfb0c9410bf5656aa76812d860b53","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-07-13 18:57:40.000000000","message":"Patch Set 10:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1157/","accounts_in_message":[],"_revision_number":10},{"id":"6afb9c11d602276c030d7cb86d57efe4bd66f30c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-07-13 19:08:44.000000000","message":"Patch Set 10:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1157/ : FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"ce2a9422036ec595517fbc66b8431980dda1917e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-07-13 19:29:42.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/fa5bc880a2674983af4b531c68252d16 : FAILURE in 28m 25s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/8f6c8a77848a4d1da00f3552d39bc504 : FAILURE in 22m 15s","accounts_in_message":[],"_revision_number":10},{"id":"1b9fe05500baaa49cf29b9d39bd14cede3949dbe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 19:32:51.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/90caf40a4b4246548cc36f6d55892b05\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/840b84222f61455bb2b1fb924acd5c0c : FAILURE in 25m 45s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/5109ce0143024407b042dd64f63ae805 : SKIPPED\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/04fdfb507f5543cf9e16388e1f70564c : SKIPPED\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/57df9f96cf514a29ad85024e1abb195b : SKIPPED\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2678d1954fa04b998e88efe8d59326e2 : SUCCESS in 4m 51s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/3d9a87c092aa424f8b9c918afa56a994 : FAILURE in 27m 51s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/f6d269c34ee54c7d9845b62e21208032 : SKIPPED\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/6b176a363c6641b7a92e95819c557a36 : SKIPPED\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/f8bc196c46f24f3b8134f336f39d4650 : SKIPPED (non-voting)\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/019288164f5e4ff180f73303b0c363e0 : SKIPPED\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/cdc2216df186494596e1bfd08352d342 : SKIPPED\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/8109940ec413401eb02eaab7d3c3712b : SUCCESS in 17m 07s","accounts_in_message":[],"_revision_number":10},{"id":"8833e57cf9fb55fffb54a22c51644cc37a75e722","author":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"date":"2022-07-14 08:36:39.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"a46608f58f79a9f9ae0d99be178bc1d29aa80066","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"date":"2022-07-14 08:45:30.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"714d4b8660d84717a24280d043d2d1662fd9713c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-07-14 08:46:14.000000000","message":"Patch Set 11:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1159/","accounts_in_message":[],"_revision_number":11},{"id":"e913770a6ee7dcb9e2e30459225bf734fccb47ac","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-07-14 08:56:05.000000000","message":"Patch Set 11:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1159/ : FAILURE","accounts_in_message":[],"_revision_number":11},{"id":"5a6b22cf70d8689c8b55efc78e51ba7c97ace822","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"date":"2022-07-14 11:11:11.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"b4a75fce4eb087003e01eecc273e554c613c4901","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-07-14 11:11:39.000000000","message":"Patch Set 12:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1161/","accounts_in_message":[],"_revision_number":12},{"id":"133e305381aa380212cc69af6fe50f935d6ad0b4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-07-14 11:22:38.000000000","message":"Patch Set 12:\n\nBuild Failed \n\n* tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-extras-gate-master-tripleo-ci-delorean-full-featureset052/1161/ : FAILURE","accounts_in_message":[],"_revision_number":12},{"id":"41dcd9e8b6fbec2cbf4fba6b906d3070ed21584a","author":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"date":"2022-07-14 13:10:37.000000000","message":"Patch Set 12: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"6ecb6869e624f9bffb1226addefb9398b548f1cd","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-07-14 13:42:05.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/7e3a3ddaa39c4e3f8bdc969613433fcd : FAILURE in 2h 02m 02s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/1130de7109a546229b9bb1902175e459 : SUCCESS in 2h 29m 22s","accounts_in_message":[],"_revision_number":12},{"id":"7995d2eb79acf965e33979094ce2253a1540b610","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 13:50:56.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/318cc1157ee04fa2918a4d3ec6387e5b\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/5072d8051a0e42e282bb0f3de621b765 : SUCCESS in 2h 24m 30s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/c421aaeb40db47fe995665b4e4508194 : SUCCESS in 35m 21s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/8ef352aedf7745feb479786fc8f214d1 : SUCCESS in 1h 21m 10s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/84b95e363c4d40a5bf77a8bfe91981fc : SUCCESS in 1h 20m 02s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/04624390e0cd4101a13537ad5adbba08 : SUCCESS in 5m 11s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/6bd5631d004047fda23afbde37546eab : SUCCESS in 2h 27m 11s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/0c676bf35475427da690503757b7027d : SUCCESS in 1h 32m 16s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/8a48975ad99149e0ba93a127eae47f4c : SUCCESS in 1h 11m 49s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/440754a3d32a4ae18b88bf6825eeb2f1 : SUCCESS in 1h 24m 10s (non-voting)\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/f4446960dd754e84b5d54d07239c3d88 : SUCCESS in 1h 14m 30s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/befe63ad49d64c1fa53d6c8e9d26839b : SUCCESS in 57m 55s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/54ea589af0b24402899bb17a87cd44e9 : SUCCESS in 21m 14s","accounts_in_message":[],"_revision_number":12},{"id":"bfe2a8e96297189c68ec2c9fcf3bd2b77b032af7","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2022-07-14 15:44:18.000000000","message":"Patch Set 12: Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"c0449b957a30e0c06e8e0333581592255b8cca9d","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-07-14 16:15:15.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"cba07a87b7b910facf8dd88846dbfe71be55a258","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2022-07-14 19:12:27.000000000","message":"Patch Set 12: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"3d7a544b04dc8b7ce218a7ddc10f51e23087ca0e","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2022-07-14 19:24:01.000000000","message":"Patch Set 12: Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"8046e53bf2f8a98cba97041b61ae9ca977a9db37","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2022-07-14 20:03:31.000000000","message":"Patch Set 12: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":12},{"id":"a436743f68db2fb015339a62141054bc1d1007e9","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-07-15 06:42:17.000000000","message":"Patch Set 12: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"d3d7268ff3c855cf6d69ac477ef1176c80cabf73","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-07-15 07:02:57.000000000","message":"Patch Set 12: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"db1726e0aaf08c8340687bab05b5c835168c56f4","author":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"date":"2022-07-15 07:31:17.000000000","message":"Patch Set 12:\n\n(5 comments)","accounts_in_message":[],"_revision_number":12},{"id":"88a7205de0e3a360108085822053ff11e3215fb5","author":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"date":"2022-07-15 10:10:31.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"03cc4fdede9ccdcec4889382f2060fa155e249c1","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2022-07-15 14:15:57.000000000","message":"Patch Set 12: Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"620621beda41cb3f35f245952fcc4342c4804ac8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 14:16:47.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"4ea25d436d1ed2014cc4de36e727dacf1464a8ec","author":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"date":"2022-07-15 14:25:32.000000000","message":"Patch Set 12: Code-Review-1","accounts_in_message":[],"_revision_number":12},{"id":"9a1ae2c4db83db6723da894b4a6b37488bf4dff4","author":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"date":"2022-07-15 14:26:12.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"67443325fda36c60558931f9b013704262e33b33","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2022-07-15 14:27:48.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":12},{"id":"fdcf14389e26f0df575cec1f36008dc1b40b8268","tag":"autogenerated:gerrit:restore","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2023-05-16 10:58:59.000000000","message":"Restored","accounts_in_message":[],"_revision_number":12},{"id":"694aee2159ca202c7230f808cef943ca2321f52e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2023-05-16 11:09:35.000000000","message":"Abandoned\n\nwrong patch restored","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"9cfb099187630e31adbb5c736d56e1df1fea608e","revisions":{"434353d0b5a35ca0e47be24eedf4af988a220fba":{"kind":"REWORK","_number":1,"created":"2022-06-23 23:31:39.000000000","uploader":{"_account_id":34208,"name":"Rafael Castillo","email":"rcastill@redhat.com","username":"rcastill"},"ref":"refs/changes/19/847419/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/19/847419/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/19/847419/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/19/847419/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/19/847419/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/19/847419/1"}}},"commit":{"parents":[{"commit":"17d3f6af73cf9eca7e20c08e5c3c2cdae8ee9bc2","subject":"Merge \"Run dstat on overcloud nodes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/17d3f6af73cf9eca7e20c08e5c3c2cdae8ee9bc2"}]}],"author":{"name":"Rafael Castillo","email":"rcastill@redhat.com","date":"2022-06-23 23:27:26.000000000","tz":-420},"committer":{"name":"Rafael Castillo","email":"rcastill@redhat.com","date":"2022-06-23 23:31:38.000000000","tz":-420},"subject":"Allow passing changes to built-test-packages in zuul job vars","message":"Allow passing changes to built-test-packages in zuul job vars\n\nChange-Id: I0d3966af5fc6b9236c5bc255282aadd3c48aaa63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/434353d0b5a35ca0e47be24eedf4af988a220fba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/434353d0b5a35ca0e47be24eedf4af988a220fba"}]},"branch":"refs/heads/master"},"c678e4e130cceb55092c6069e9cef91a8895aeeb":{"kind":"REWORK","_number":2,"created":"2022-06-23 23:32:36.000000000","uploader":{"_account_id":34208,"name":"Rafael Castillo","email":"rcastill@redhat.com","username":"rcastill"},"ref":"refs/changes/19/847419/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/19/847419/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/19/847419/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/19/847419/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/19/847419/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/19/847419/2"}}},"commit":{"parents":[{"commit":"17d3f6af73cf9eca7e20c08e5c3c2cdae8ee9bc2","subject":"Merge \"Run dstat on overcloud nodes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/17d3f6af73cf9eca7e20c08e5c3c2cdae8ee9bc2"}]}],"author":{"name":"Rafael Castillo","email":"rcastill@redhat.com","date":"2022-06-23 23:27:26.000000000","tz":-420},"committer":{"name":"Rafael Castillo","email":"rcastill@redhat.com","date":"2022-06-23 23:32:30.000000000","tz":-420},"subject":"Allow passing changes to built-test-packages in zuul job vars","message":"Allow passing changes to built-test-packages in zuul job vars\n\nChange-Id: I0d3966af5fc6b9236c5bc255282aadd3c48aaa63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/c678e4e130cceb55092c6069e9cef91a8895aeeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/c678e4e130cceb55092c6069e9cef91a8895aeeb"}]},"branch":"refs/heads/master"},"3fb27aee3791c2421a5c9660a182d1cee359281e":{"kind":"REWORK","_number":3,"created":"2022-06-23 23:34:07.000000000","uploader":{"_account_id":34208,"name":"Rafael Castillo","email":"rcastill@redhat.com","username":"rcastill"},"ref":"refs/changes/19/847419/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/19/847419/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/19/847419/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/19/847419/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/19/847419/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/19/847419/3"}}},"commit":{"parents":[{"commit":"17d3f6af73cf9eca7e20c08e5c3c2cdae8ee9bc2","subject":"Merge \"Run dstat on overcloud nodes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/17d3f6af73cf9eca7e20c08e5c3c2cdae8ee9bc2"}]}],"author":{"name":"Rafael Castillo","email":"rcastill@redhat.com","date":"2022-06-23 23:27:26.000000000","tz":-420},"committer":{"name":"Rafael Castillo","email":"rcastill@redhat.com","date":"2022-06-23 23:34:02.000000000","tz":-420},"subject":"Allow passing changes to built-test-packages in zuul job vars","message":"Allow passing changes to built-test-packages in zuul job vars\n\nChange-Id: I0d3966af5fc6b9236c5bc255282aadd3c48aaa63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/3fb27aee3791c2421a5c9660a182d1cee359281e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/3fb27aee3791c2421a5c9660a182d1cee359281e"}]},"branch":"refs/heads/master"},"a727750d4dac62af3dfaf02f891e66086a608751":{"kind":"REWORK","_number":4,"created":"2022-06-24 01:16:52.000000000","uploader":{"_account_id":34208,"name":"Rafael Castillo","email":"rcastill@redhat.com","username":"rcastill"},"ref":"refs/changes/19/847419/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/19/847419/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/19/847419/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/19/847419/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/19/847419/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/19/847419/4"}}},"commit":{"parents":[{"commit":"17d3f6af73cf9eca7e20c08e5c3c2cdae8ee9bc2","subject":"Merge \"Run dstat on overcloud nodes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/17d3f6af73cf9eca7e20c08e5c3c2cdae8ee9bc2"}]}],"author":{"name":"Rafael Castillo","email":"rcastill@redhat.com","date":"2022-06-23 23:27:26.000000000","tz":-420},"committer":{"name":"Rafael Castillo","email":"rcastill@redhat.com","date":"2022-06-24 01:16:46.000000000","tz":-420},"subject":"Allow passing changes to built-test-packages in zuul job vars","message":"Allow passing changes to built-test-packages in zuul job vars\n\nChange-Id: I0d3966af5fc6b9236c5bc255282aadd3c48aaa63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/a727750d4dac62af3dfaf02f891e66086a608751"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/a727750d4dac62af3dfaf02f891e66086a608751"}]},"branch":"refs/heads/master"},"b318907767aeac8b8c53d1ffb16853f77185eda4":{"kind":"REWORK","_number":5,"created":"2022-06-24 02:37:24.000000000","uploader":{"_account_id":34208,"name":"Rafael Castillo","email":"rcastill@redhat.com","username":"rcastill"},"ref":"refs/changes/19/847419/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/19/847419/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/19/847419/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/19/847419/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/19/847419/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/19/847419/5"}}},"commit":{"parents":[{"commit":"17d3f6af73cf9eca7e20c08e5c3c2cdae8ee9bc2","subject":"Merge \"Run dstat on overcloud nodes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/17d3f6af73cf9eca7e20c08e5c3c2cdae8ee9bc2"}]}],"author":{"name":"Rafael Castillo","email":"rcastill@redhat.com","date":"2022-06-23 23:27:26.000000000","tz":-420},"committer":{"name":"Rafael Castillo","email":"rcastill@redhat.com","date":"2022-06-24 02:37:19.000000000","tz":-420},"subject":"Allow passing changes to built-test-packages in zuul job vars","message":"Allow passing changes to built-test-packages in zuul job vars\n\nChange-Id: I0d3966af5fc6b9236c5bc255282aadd3c48aaa63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/b318907767aeac8b8c53d1ffb16853f77185eda4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/b318907767aeac8b8c53d1ffb16853f77185eda4"}]},"branch":"refs/heads/master"},"11ce5f2ae2131727f54489f59ba039a3ab65a2f5":{"kind":"REWORK","_number":6,"created":"2022-06-24 03:08:55.000000000","uploader":{"_account_id":34208,"name":"Rafael Castillo","email":"rcastill@redhat.com","username":"rcastill"},"ref":"refs/changes/19/847419/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/19/847419/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/19/847419/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/19/847419/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/19/847419/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/19/847419/6"}}},"commit":{"parents":[{"commit":"17d3f6af73cf9eca7e20c08e5c3c2cdae8ee9bc2","subject":"Merge \"Run dstat on overcloud nodes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/17d3f6af73cf9eca7e20c08e5c3c2cdae8ee9bc2"}]}],"author":{"name":"Rafael Castillo","email":"rcastill@redhat.com","date":"2022-06-23 23:27:26.000000000","tz":-420},"committer":{"name":"Rafael","email":"rcastill@redhat.com","date":"2022-06-24 03:07:46.000000000","tz":0},"subject":"Allow passing changes to built-test-packages in zuul job vars","message":"Allow passing changes to built-test-packages in zuul job vars\n\nChange-Id: I0d3966af5fc6b9236c5bc255282aadd3c48aaa63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/11ce5f2ae2131727f54489f59ba039a3ab65a2f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/11ce5f2ae2131727f54489f59ba039a3ab65a2f5"}]},"branch":"refs/heads/master"},"1c6fb20c4e9a5ff23742f9eb2e4fd2bf2721ed16":{"kind":"REWORK","_number":7,"created":"2022-06-24 06:09:00.000000000","uploader":{"_account_id":34208,"name":"Rafael Castillo","email":"rcastill@redhat.com","username":"rcastill"},"ref":"refs/changes/19/847419/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/19/847419/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/19/847419/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/19/847419/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/19/847419/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/19/847419/7"}}},"commit":{"parents":[{"commit":"17d3f6af73cf9eca7e20c08e5c3c2cdae8ee9bc2","subject":"Merge \"Run dstat on overcloud nodes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/17d3f6af73cf9eca7e20c08e5c3c2cdae8ee9bc2"}]}],"author":{"name":"Rafael Castillo","email":"rcastill@redhat.com","date":"2022-06-23 23:27:26.000000000","tz":-420},"committer":{"name":"Rafael","email":"rcastill@redhat.com","date":"2022-06-24 06:07:53.000000000","tz":0},"subject":"Allow passing changes to built-test-packages in zuul job vars","message":"Allow passing changes to built-test-packages in zuul job vars\n\nChange-Id: I0d3966af5fc6b9236c5bc255282aadd3c48aaa63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/1c6fb20c4e9a5ff23742f9eb2e4fd2bf2721ed16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/1c6fb20c4e9a5ff23742f9eb2e4fd2bf2721ed16"}]},"branch":"refs/heads/master"},"72548cb1b0cbf9f71f3464d3405113b3303b4d93":{"kind":"REWORK","_number":8,"created":"2022-06-24 12:01:33.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/19/847419/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/19/847419/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/19/847419/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/19/847419/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/19/847419/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/19/847419/8"}}},"commit":{"parents":[{"commit":"17d3f6af73cf9eca7e20c08e5c3c2cdae8ee9bc2","subject":"Merge \"Run dstat on overcloud nodes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/17d3f6af73cf9eca7e20c08e5c3c2cdae8ee9bc2"}]}],"author":{"name":"Rafael Castillo","email":"rcastill@redhat.com","date":"2022-06-23 23:27:26.000000000","tz":-420},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2022-06-24 12:01:33.000000000","tz":0},"subject":"Allow passing changes to built-test-packages in zuul job vars","message":"Allow passing changes to built-test-packages in zuul job vars\n\nChange-Id: I0d3966af5fc6b9236c5bc255282aadd3c48aaa63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/72548cb1b0cbf9f71f3464d3405113b3303b4d93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/72548cb1b0cbf9f71f3464d3405113b3303b4d93"}]},"branch":"refs/heads/master"},"f23d425642745ec88c45996fcc7fb47f8b5a16e1":{"kind":"REWORK","_number":9,"created":"2022-07-12 16:37:06.000000000","uploader":{"_account_id":34208,"name":"Rafael Castillo","email":"rcastill@redhat.com","username":"rcastill"},"ref":"refs/changes/19/847419/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/19/847419/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/19/847419/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/19/847419/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/19/847419/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/19/847419/9"}}},"commit":{"parents":[{"commit":"17d3f6af73cf9eca7e20c08e5c3c2cdae8ee9bc2","subject":"Merge \"Run dstat on overcloud nodes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/17d3f6af73cf9eca7e20c08e5c3c2cdae8ee9bc2"}]}],"author":{"name":"Rafael Castillo","email":"rcastill@redhat.com","date":"2022-06-23 23:27:26.000000000","tz":-420},"committer":{"name":"Rafael Castillo","email":"rcastill@redhat.com","date":"2022-07-12 16:28:48.000000000","tz":-420},"subject":"Allow passing changes to build-test-packages in zuul job vars","message":"Allow passing changes to build-test-packages in zuul job vars\n\nThis patch adds a new variable to the build-test-packages role that\nallows specifying arbitrary extra changes to be built. The intended use\ncase is for ad-hoc testing a change against a version of another\npackage that isn\u0027t currently being built by DLRN.\n\nFor example, the tripleo jobs in the master branch of\nansible-collections-openstack should be run against\nopenstacksdk\u003e\u003d0.99.0, but currently the rdo u-c[1] have openstacksdk\npinned to 0.61.0. With this change, ansible-collections-openstack can\ntell build-test-packages to build the correct version[2].\n\n[1] https://github.com/redhat-openstack/rdoinfo/blob/f482a6ecc45910cb6ccae448bffecfac0a6999c8/tags/zed-uc.yml#L467\n[2] https://review.opendev.org/c/openstack/ansible-collections-openstack/+/844859\n\nChange-Id: I0d3966af5fc6b9236c5bc255282aadd3c48aaa63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/f23d425642745ec88c45996fcc7fb47f8b5a16e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/f23d425642745ec88c45996fcc7fb47f8b5a16e1"}]},"branch":"refs/heads/master"},"c2bc35700ceaa0dce1926ecb9d93dc6d9141d447":{"kind":"REWORK","_number":10,"created":"2022-07-13 18:57:23.000000000","uploader":{"_account_id":34208,"name":"Rafael Castillo","email":"rcastill@redhat.com","username":"rcastill"},"ref":"refs/changes/19/847419/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/19/847419/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/19/847419/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/19/847419/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/19/847419/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/19/847419/10"}}},"commit":{"parents":[{"commit":"17d3f6af73cf9eca7e20c08e5c3c2cdae8ee9bc2","subject":"Merge \"Run dstat on overcloud nodes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/17d3f6af73cf9eca7e20c08e5c3c2cdae8ee9bc2"}]}],"author":{"name":"Rafael Castillo","email":"rcastill@redhat.com","date":"2022-06-23 23:27:26.000000000","tz":-420},"committer":{"name":"Rafael Castillo","email":"rcastill@redhat.com","date":"2022-07-13 18:57:06.000000000","tz":-420},"subject":"Allow passing changes to build-test-packages in zuul job vars","message":"Allow passing changes to build-test-packages in zuul job vars\n\nThis patch adds a new variable to the build-test-packages role that\nallows specifying arbitrary extra changes to be built. The intended use\ncase is for ad-hoc testing a change against a version of another\npackage that isn\u0027t currently being built by DLRN.\n\nFor example, the tripleo jobs in the master branch of\nansible-collections-openstack should be run against\nopenstacksdk\u003e\u003d0.99.0, but currently the rdo u-c[1] have openstacksdk\npinned to 0.61.0. With this change, ansible-collections-openstack can\ntell build-test-packages to build the correct version[2].\n\n[1] https://github.com/redhat-openstack/rdoinfo/blob/f482a6ecc45910cb6ccae448bffecfac0a6999c8/tags/zed-uc.yml#L467\n[2] https://review.opendev.org/c/openstack/ansible-collections-openstack/+/844859\n\nChange-Id: I0d3966af5fc6b9236c5bc255282aadd3c48aaa63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/c2bc35700ceaa0dce1926ecb9d93dc6d9141d447"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/c2bc35700ceaa0dce1926ecb9d93dc6d9141d447"}]},"branch":"refs/heads/master"},"7489f900a6e8579bf42a7f1c7090230722aa3283":{"kind":"REWORK","_number":11,"created":"2022-07-14 08:45:30.000000000","uploader":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"ref":"refs/changes/19/847419/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/19/847419/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/19/847419/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/19/847419/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/19/847419/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/19/847419/11"}}},"commit":{"parents":[{"commit":"9f54c3129e49a1a4118db0abbab4a7a3285c490f","subject":"Fix tempest_allow logic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/9f54c3129e49a1a4118db0abbab4a7a3285c490f"}]}],"author":{"name":"Rafael Castillo","email":"rcastill@redhat.com","date":"2022-06-23 23:27:26.000000000","tz":-420},"committer":{"name":"Jakob Meng","email":"code@jakobmeng.de","date":"2022-07-14 08:45:03.000000000","tz":120},"subject":"Allow passing changes to build-test-packages in zuul job vars","message":"Allow passing changes to build-test-packages in zuul job vars\n\nThis patch adds a new variable to the build-test-packages role that\nallows specifying arbitrary extra changes to be built. The intended use\ncase is for ad-hoc testing a change against a version of another\npackage that isn\u0027t currently being built by DLRN.\n\nFor example, the tripleo jobs in the master branch of\nansible-collections-openstack should be run against\nopenstacksdk\u003e\u003d0.99.0, but currently the rdo u-c[1] have openstacksdk\npinned to 0.61.0. With this change, ansible-collections-openstack can\ntell build-test-packages to build the correct version[2].\n\n[1] https://github.com/redhat-openstack/rdoinfo/blob/f482a6ecc45910cb6ccae448bffecfac0a6999c8/tags/zed-uc.yml#L467\n[2] https://review.opendev.org/c/openstack/ansible-collections-openstack/+/844859\n\nChange-Id: I0d3966af5fc6b9236c5bc255282aadd3c48aaa63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7489f900a6e8579bf42a7f1c7090230722aa3283"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/7489f900a6e8579bf42a7f1c7090230722aa3283"}]},"branch":"refs/heads/master"},"9cfb099187630e31adbb5c736d56e1df1fea608e":{"kind":"REWORK","_number":12,"created":"2022-07-14 11:11:11.000000000","uploader":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"ref":"refs/changes/19/847419/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/19/847419/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/19/847419/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/19/847419/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/19/847419/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/19/847419/12"}}},"commit":{"parents":[{"commit":"9f54c3129e49a1a4118db0abbab4a7a3285c490f","subject":"Fix tempest_allow logic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/9f54c3129e49a1a4118db0abbab4a7a3285c490f"}]}],"author":{"name":"Rafael Castillo","email":"rcastill@redhat.com","date":"2022-06-23 23:27:26.000000000","tz":-420},"committer":{"name":"Jakob Meng","email":"code@jakobmeng.de","date":"2022-07-14 11:11:05.000000000","tz":120},"subject":"Allow passing changes to build-test-packages in zuul job vars","message":"Allow passing changes to build-test-packages in zuul job vars\n\nThis patch adds a new variable to the build-test-packages role that\nallows specifying arbitrary extra changes to be built. The intended use\ncase is for ad-hoc testing a change against a version of another\npackage that isn\u0027t currently being built by DLRN.\n\nFor example, the tripleo jobs in the master branch of\nansible-collections-openstack should be run against\nopenstacksdk\u003e\u003d0.99.0, but currently the rdo u-c[1] have openstacksdk\npinned to 0.61.0. With this change, ansible-collections-openstack can\ntell build-test-packages to build the correct version[2].\n\n[1] https://github.com/redhat-openstack/rdoinfo/blob/f482a6ecc45910cb6ccae448bffecfac0a6999c8/tags/zed-uc.yml#L467\n[2] https://review.opendev.org/c/openstack/ansible-collections-openstack/+/844859\n\nChange-Id: I0d3966af5fc6b9236c5bc255282aadd3c48aaa63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/9cfb099187630e31adbb5c736d56e1df1fea608e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/9cfb099187630e31adbb5c736d56e1df1fea608e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
