)]}'
{"id":"openstack%2Ftripleo-quickstart-extras~861809","triplet_id":"openstack%2Ftripleo-quickstart-extras~master~I3382ead6eafe8a77b9b874ca7a22d4266851e3ac","project":"openstack/tripleo-quickstart-extras","branch":"master","topic":"bug/1993343","attention_set":{},"removed_from_attention_set":{"8449":{"account":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"last_update":"2022-10-19 07:20:04.000000000","reason":"\u003cGERRIT_ACCOUNT_8449\u003e replied on the change","reason_account":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"}},"20182":{"account":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"last_update":"2022-10-19 10:03:03.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-10-19 06:48:32.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"12393":{"account":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"last_update":"2022-10-19 10:03:03.000000000","reason":"Change was submitted"},"32458":{"account":{"_account_id":32458,"name":"Ananya","display_name":"Ananya Banerjee","email":"anbanerj@redhat.com","username":"frenzyfriday"},"last_update":"2022-10-19 10:03:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3382ead6eafe8a77b9b874ca7a22d4266851e3ac","subject":"Pin rdopkg to 1.4.0","status":"MERGED","created":"2022-10-19 01:47:18.000000000","updated":"2022-10-19 15:34:41.000000000","submitted":"2022-10-19 10:03:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"861809-bug/1993343","meta_rev_id":"9561dfb6b5fe9ceab211c61e97d9ef0a05ca0ec5","_number":861809,"virtual_id_number":861809,"owner":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-19 10:03:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-10-19 05:39:28.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"all":[{"value":2,"date":"2022-10-19 06:34:19.000000000","_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":2,"date":"2022-10-19 07:20:04.000000000","_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":2,"date":"2022-10-19 06:48:32.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":1,"date":"2022-10-19 07:20:04.000000000","_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"date":"2022-10-19 06:48:47.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},{"_account_id":32458,"name":"Ananya","display_name":"Ananya Banerjee","email":"anbanerj@redhat.com","username":"frenzyfriday"}],"reviewers":{"REVIEWER":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":32458,"name":"Ananya","display_name":"Ananya Banerjee","email":"anbanerj@redhat.com","username":"frenzyfriday"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-19 01:47:43.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"state":"REVIEWER"},{"updated":"2022-10-19 01:47:43.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":32458,"name":"Ananya","display_name":"Ananya Banerjee","email":"anbanerj@redhat.com","username":"frenzyfriday"},"state":"REVIEWER"},{"updated":"2022-10-19 01:47:43.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2022-10-19 02:46:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-19 05:39:28.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-19 05:58:50.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2022-10-19 06:34:19.000000000","updated_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"REVIEWER"}],"messages":[{"id":"3d3e7bf112a389d122795ea1ddecb8e04066d484","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-19 01:47:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8155bb8e9030d9d6da01939d9edc912aa2577dfe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 02:46:00.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/41fd1b12a2014438812379022f62430a\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/03fb09925af1492e8437f11923de9b86 : FAILURE in 3m 54s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/be7109147d5e4291a240f181c06ab5ee : SUCCESS in 19m 50s\nSkipped 12 jobs","accounts_in_message":[],"_revision_number":1},{"id":"0a324673689c4cf1d627554badce69af2ee5011c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-10-19 03:39:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c5587371062658ad818937d7372fa7d4869464c5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-10-19 05:39:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/bfffa078a32749cf922b09ae736ecf73 : SUCCESS in 1h 58m 28s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/ff17b3fc080b4d64bc1f4da115c4fded : SUCCESS in 1h 57m 18s","accounts_in_message":[],"_revision_number":2},{"id":"0d149f0ed486bde983d1540f5efe1bd257cabbf6","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2022-10-19 06:34:19.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0708b1da101b7768bc142f2ab64cf31f37babf25","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-10-19 06:48:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"cb00391e4b155fc175217772a03f772b129648df","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-10-19 06:48:47.000000000","message":"Patch Set 2: -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"e7bbf589096da63737408ca5a8f4e777a22840dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 07:13:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc68ac48795241d4b9e210af22a2b876\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/5550066731764c558ef461dbf30b3e5b : SUCCESS in 3h 25m 24s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/61d7fb29cddd485080ae37740a06bed0 : SUCCESS in 53m 10s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/a309687a0bc1420c9b8de81a1439f46f : SUCCESS in 1h 58m 25s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/0fe55bd84f8e40f8b57b1aecb45b328e : SUCCESS in 1h 51m 48s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3b96a3a5a41e424eb5ea77d233a4b564 : SUCCESS in 3m 28s\n- tripleo-ci-centos-9-content-provider-zed https://zuul.opendev.org/t/openstack/build/e565b1c71dec4cc687f5c1a8c917f195 : SUCCESS in 2h 39m 54s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/6d5b87b76c354f078dad9cb4ec1755c7 : SUCCESS in 3h 11m 29s\n- tripleo-ci-centos-9-containers-multinode-zed https://zuul.opendev.org/t/openstack/build/04130b6fa914434ba0ce628c08207383 : SUCCESS in 1h 38m 23s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/41c34e5cdd3148beaa413cdfa2da681b : SUCCESS in 2h 00m 21s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/70393b524b164a968d607785e8504e9f : SUCCESS in 1h 35m 39s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/7446134fd59a4c35a8884e61bc47d3d1 : SUCCESS in 2h 15m 19s (non-voting)\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/c10176b5d5034639ab6aba6a287456cb : SUCCESS in 1h 42m 39s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/525ef6f8169e4ec497fb872024e70156 : SUCCESS in 1h 16m 01s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/012e87030baa45f0b7f4d67f75cefc65 : SUCCESS in 14m 32s","accounts_in_message":[],"_revision_number":2},{"id":"3d8040e66dcb419391dcf84f366e7d5fadb2005e","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-10-19 07:20:04.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"012b861d5a22a4781543d1ec80ca3461e8b3f9be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 07:20:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0be3e8e6435893939a1c1f7ff9572d9369e4c030","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-10-19 07:25:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6bd5e2d71e1aab676635cab71f33e28a93bb27a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 10:03:03.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/637d5a103d4340e887950c4213a64b1d\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/2da634fb35bd452fb46d04aeab975560 : SUCCESS in 2h 38m 36s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/6eda82e82ffb451b816c2129a1a47264 : SUCCESS in 43m 48s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/457dbc0c6df64497ba5383406a647421 : SUCCESS in 1h 35m 54s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/42e5ec3f184549149479e683cbe416a7 : SUCCESS in 1h 41m 00s\n- tripleo-ci-centos-9-content-provider-zed https://zuul.opendev.org/t/openstack/build/1b94a3c7ae3c45d7ba99d752e61a5427 : SUCCESS in 2h 38m 57s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/f263b3833af34763a63c39a96d1eb2fb : SUCCESS in 2h 41m 29s\n- tripleo-ci-centos-9-containers-multinode-zed https://zuul.opendev.org/t/openstack/build/b7d589505f6045f38c9da9dd7ca1bbc7 : SUCCESS in 1h 40m 58s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/4aa7eb7e18d24fc6999e3a878f3ce31a : SUCCESS in 1h 45m 08s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/ad504b3d1bce4755837fa7ba4b47cc1c : SUCCESS in 1h 18m 29s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/89813ca741fd4032905e5b67a4982e92 : SUCCESS in 1h 02m 26s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/13bf3787a6134acc9a17c844d3ecae2c : SUCCESS in 1h 25m 16s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1578843ce947426893ba746d2d839465 : SUCCESS in 3m 50s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/35aa4a4eaac54732bf678742cdf14267 : SUCCESS in 17m 26s","accounts_in_message":[],"_revision_number":2},{"id":"051e746e35200a65d8d2b095e2439584596a6f09","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 10:03:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9561dfb6b5fe9ceab211c61e97d9ef0a05ca0ec5","tag":"autogenerated:gerrit:revert","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-10-19 15:34:41.000000000","message":"Created a revert of this change as I3f4a7f90295e30bf32fd8bd6196d5135e44c34c1","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8226fa9a25f8d8071e48b605ca4af733b259bcc7","revisions":{"0a9ce0cbf9f5d622490aa2b5d6987d4da4bce172":{"kind":"REWORK","_number":1,"created":"2022-10-19 01:47:18.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/09/861809/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/09/861809/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/09/861809/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/09/861809/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/09/861809/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/09/861809/1"}}},"commit":{"parents":[{"commit":"52fe7a9397fe4705a3d5b6b3222e49bb5ff0e1ea","subject":"Merge \"Stop using os_stack module\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/52fe7a9397fe4705a3d5b6b3222e49bb5ff0e1ea"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-19 01:45:15.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-19 01:45:15.000000000","tz":330},"subject":"Pin rdopkg to 1.4.0","message":"Pin rdopkg to 1.4.0\n\nLatest release of rdopkg (1.5.0) with\nerror: metadata-generation-failed and is blocking the whole\nCI.\n\nWe are currently pinning rdopkg to older version till we fix\nthe rdopkg or install rdopkg from RPM.\n\nRelated-Bug: #1993343\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I3382ead6eafe8a77b9b874ca7a22d4266851e3ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/0a9ce0cbf9f5d622490aa2b5d6987d4da4bce172"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/0a9ce0cbf9f5d622490aa2b5d6987d4da4bce172"}]},"branch":"refs/heads/master"},"8226fa9a25f8d8071e48b605ca4af733b259bcc7":{"kind":"REWORK","_number":2,"created":"2022-10-19 03:39:45.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/09/861809/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart-extras","ref":"refs/changes/09/861809/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/09/861809/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/09/861809/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/09/861809/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart-extras refs/changes/09/861809/2"}}},"commit":{"parents":[{"commit":"52fe7a9397fe4705a3d5b6b3222e49bb5ff0e1ea","subject":"Merge \"Stop using os_stack module\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/52fe7a9397fe4705a3d5b6b3222e49bb5ff0e1ea"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-19 01:45:15.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-10-19 03:39:35.000000000","tz":330},"subject":"Pin rdopkg to 1.4.0","message":"Pin rdopkg to 1.4.0\n\nLatest release of rdopkg (1.5.0) with\nerror: metadata-generation-failed and is blocking the whole\nCI.\n\nWe are currently pinning rdopkg to older version till we fix\nthe rdopkg or install rdopkg from RPM.\n\nRelated-Bug: #1993343\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I3382ead6eafe8a77b9b874ca7a22d4266851e3ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/8226fa9a25f8d8071e48b605ca4af733b259bcc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart-extras/commit/8226fa9a25f8d8071e48b605ca4af733b259bcc7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"}}]}],"submit_requirements":[]}
